)]}'
{"id":"openstack%2Ftempest~496079","triplet_id":"openstack%2Ftempest~master~Iafbcc2b3048f5fcf1e453da0e4821268ffa0be59","project":"openstack/tempest","branch":"master","topic":"detail_check","hashtags":[],"change_id":"Iafbcc2b3048f5fcf1e453da0e4821268ffa0be59","subject":"Add detail specific fields check in list_backups","status":"MERGED","created":"2017-08-22 03:23:36.000000000","updated":"2017-08-30 09:47:45.000000000","submitted":"2017-08-30 09:47:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff31725f5d1c628d33a5a298ac361dff9e123912","_number":496079,"virtual_id_number":496079,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-08-30 09:47:45.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-08-22 04:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},{"value":2,"date":"2017-08-22 06:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2017-08-30 08:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2017-08-30 08:17:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-22 04:02:39.000000000","updated_by":{"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},"reviewer":{"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},"state":"REVIEWER"},{"updated":"2017-08-22 06:31:48.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2017-08-30 08:17:42.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2017-08-30 09:47:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"622a21675fc9fdab091057d5892bd9aa5c915c3b","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2017-08-22 03:23:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84d80f40aaa22ed41685dcb1b92fdee01e6a0100","author":{"_account_id":23081,"name":"Jeremy Zhang","email":"zhangjun_inspur@163.com","username":"jeremy.zhang"},"date":"2017-08-22 04:02:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"13c2c97929a0575949336081689ef077fc96f63d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-22 05:05:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/229edca/ : SUCCESS in 1h 32m 04s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/2c0e292/ : SUCCESS in 48m 57s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/430b84c/ : SUCCESS in 1h 14m 33s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/79/496079/1/check/gate-tempest-docs-ubuntu-xenial/aaafc15//doc/build/html/ : SUCCESS in 13m 53s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-tempest-pep8-ubuntu-xenial/a059525/ : SUCCESS in 2m 13s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-tempest-python27-ubuntu-xenial/e82a716/ : SUCCESS in 1m 52s\n- gate-tempest-python35 http://logs.openstack.org/79/496079/1/check/gate-tempest-python35/6664e69/ : SUCCESS in 3m 02s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/26b099d/ : SUCCESS in 57m 31s\n- gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-ironic-ipa-wholedisk-bios-agent_ipmitool-tinyipa-ubuntu-xenial-nv/d7a9f1a/ : SUCCESS in 45m 20s (non-voting)\n- gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-full-devstack-plugin-ceph-ubuntu-xenial/4e1290f/ : SUCCESS in 1h 11m 18s (non-voting)\n- gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv http://logs.openstack.org/79/496079/1/check/gate-puppet-openstack-integration-4-scenario001-tempest-centos-7-nv/8d8a686/ : SUCCESS in 59m 20s (non-voting)\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv http://logs.openstack.org/79/496079/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7-nv/182afed/ : SUCCESS in 1h 01m 28s (non-voting)\n- gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv http://logs.openstack.org/79/496079/1/check/gate-puppet-openstack-integration-4-scenario003-tempest-centos-7-nv/1652f1f/ : SUCCESS in 1h 17m 07s (non-voting)\n- gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv http://logs.openstack.org/79/496079/1/check/gate-puppet-openstack-integration-4-scenario004-tempest-centos-7-nv/e68f508/ : SUCCESS in 40m 28s (non-voting)\n- gate-tempest-dsvm-neutron-dvr-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-neutron-dvr-ubuntu-xenial/fb94650/ : SUCCESS in 1h 29m 14s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-ocata/7e4f06e/ : SUCCESS in 1h 38m 45s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial-newton/9537c0e/ : SUCCESS in 1h 15m 49s\n- gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv http://logs.openstack.org/79/496079/1/check/gate-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend-ubuntu-xenial-nv/2dff72e/ : SUCCESS in 1h 20m 27s (non-voting)\n- gate-tempest-tox-plugin-sanity-check-ubuntu-xenial http://logs.openstack.org/79/496079/1/check/gate-tempest-tox-plugin-sanity-check-ubuntu-xenial/4c791b8/ : FAILURE in 45m 22s (non-voting)\n- tempest-coverage-ubuntu-xenial-nv http://logs.openstack.org/79/496079/1/check/tempest-coverage-ubuntu-xenial-nv/5c7e730/ : SUCCESS in 2m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"749b726be9cd5c317570a3226dd9ed38bddf556f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-08-22 05:31:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/79/496079/1/check/dsvm-tempest-neutron-network/af30a80 : SUCCESS in 2h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"28811ea22a92d18db7122957f18107c8c86dc942","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-08-22 06:31:48.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027m fine with this change itself. But I guess there might be tons of like this thing in tempest...","accounts_in_message":[],"_revision_number":1},{"id":"0705a5495af611a8638bdda0cfb5d79c9fb997fa","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2017-08-22 06:41:32.000000000","message":"Patch Set 1:\n\n\u003e I\u0027m fine with this change itself. But I guess there might be tons\n \u003e of like this thing in tempest...\n\nactually not:) there are only a few similar issues:)","accounts_in_message":[],"_revision_number":1},{"id":"eacd002bba20032095845e7e9a4ff146d898af9f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2017-08-30 08:17:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07e73a162611a03c16f7cb90e03ae1f98ab458fd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 08:18:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"ad485026420dd26ebce3a617ebdcc55f46803e98","date":"2017-08-30 09:47:45.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"82ab9f06a4ebfcd9a0ede0868668190d935e5e64","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 09:47:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/79/496079/1/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a8e4908/ : SUCCESS in 1h 28m 56s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/79/496079/1/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/f35af29/ : SUCCESS in 50m 30s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/79/496079/1/gate/gate-tempest-dsvm-py35-ubuntu-xenial/b0c8a32/ : SUCCESS in 1h 22m 33s\n- gate-tempest-docs-ubuntu-xenial http://docs-draft.openstack.org/79/496079/1/gate/gate-tempest-docs-ubuntu-xenial/cffab6b//doc/build/html/ : SUCCESS in 4m 43s\n- gate-tempest-pep8-ubuntu-xenial http://logs.openstack.org/79/496079/1/gate/gate-tempest-pep8-ubuntu-xenial/4ccf5d3/ : SUCCESS in 2m 33s\n- gate-tempest-python27-ubuntu-xenial http://logs.openstack.org/79/496079/1/gate/gate-tempest-python27-ubuntu-xenial/eaea3c6/ : SUCCESS in 2m 56s\n- gate-tempest-python35 http://logs.openstack.org/79/496079/1/gate/gate-tempest-python35/de803f8/ : SUCCESS in 2m 10s\n- gate-grenade-dsvm-neutron-multinode-ubuntu-xenial http://logs.openstack.org/79/496079/1/gate/gate-grenade-dsvm-neutron-multinode-ubuntu-xenial/077c258/ : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0e8bebe2fc5b2132b267306952832962e4df975","revisions":{"f0e8bebe2fc5b2132b267306952832962e4df975":{"kind":"REWORK","_number":1,"created":"2017-08-22 03:23:36.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/79/496079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/79/496079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/496079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/496079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/496079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/79/496079/1"}}},"commit":{"parents":[{"commit":"e66994083158b60f00209a01c3ea6454333d1fc3","subject":"Merge \"[TrivialFix] Remove unused statements in compute tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e66994083158b60f00209a01c3ea6454333d1fc3"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2017-08-22 02:56:26.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2017-08-22 02:56:26.000000000","tz":480},"subject":"Add detail specific fields check in list_backups","message":"Add detail specific fields check in list_backups\n\nNow list_backups is called twice and both with\ndetail\u003dTrue and without specific fields check.\nThis is to:\n1. add detail specific fields check when detail\u003dTrue\n2. make one calling of list_backups to use detail\u003dFalse\n\nChange-Id: Iafbcc2b3048f5fcf1e453da0e4821268ffa0be59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f0e8bebe2fc5b2132b267306952832962e4df975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f0e8bebe2fc5b2132b267306952832962e4df975"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e66994083158b60f00209a01c3ea6454333d1fc3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
