)]}'
{"id":"openstack%2Ftempest~49819","triplet_id":"openstack%2Ftempest~master~I66813c900059dd4232de645c225032e58e9aecde","project":"openstack/tempest","branch":"master","topic":"bug/1235411","hashtags":[],"change_id":"I66813c900059dd4232de645c225032e58e9aecde","subject":"Cleanup test_list_server_filters setup/teardown","status":"MERGED","created":"2013-10-04 19:52:32.000000000","updated":"2013-10-07 22:05:32.000000000","submitted":"2013-10-07 22:05:31.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9cd9fb4ba44e7c96d74213fb4fcbae1539294fa","_number":49819,"virtual_id_number":49819,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-10-07 22:05:31.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-10-07 19:12:04.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"date":"2013-10-04 20:35:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2013-10-04 21:14:25.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-10-07 19:12:04.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2013-10-04 21:14:25.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-10-04 21:14:25.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2013-10-07 19:12:04.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2013-10-07 22:05:31.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ba80aa427aa107b2965d68a00f8479d0d6e033c9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2013-10-04 19:56:10.000000000","message":"Patch Set 1: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"3bbb80f5062a759d1c4ea403834be39e02d38e14","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-04 20:27:06.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/19/49819/1/check/gate-tempest-pep8/639001d : SUCCESS in 1m 53s\n- gate-tempest-docs http://logs.openstack.org/19/49819/1/check/gate-tempest-docs/7b9bb93 : SUCCESS in 3m 36s\n- check-tempest-devstack-vm-full http://logs.openstack.org/19/49819/1/check/check-tempest-devstack-vm-full/5be9dc5 : SUCCESS in 31m 51s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/19/49819/1/check/check-tempest-devstack-vm-postgres-full/293d572 : FAILURE in 32m 36s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/19/49819/1/check/check-tempest-devstack-vm-neutron/0a312b4 : SUCCESS in 23m 12s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/19/49819/1/check/gate-tempest-devstack-vm-large-ops/999e0ed : SUCCESS in 12m 04s\n- check-grenade-devstack-vm http://logs.openstack.org/19/49819/1/check/check-grenade-devstack-vm/e3b492b : SUCCESS in 34m 20s\n","accounts_in_message":[],"_revision_number":1},{"id":"9bd27cc3cb2336f3914ac586d0d5664bb335b4ca","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2013-10-04 20:30:44.000000000","message":"Patch Set 1: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"f78d845f704c10d9803ed388569917c1f7211815","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2013-10-04 20:34:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e651cdd9cd5dcfa7a5f2bfd7822324eadf6e3fe","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2013-10-04 20:35:25.000000000","message":"Patch Set 2: (1 inline comment)\n\nReplayed my note from patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0f36c1ac69b530dbc418ec6a012d8926c15422a8","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2013-10-04 21:14:25.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\nOh, I see what you\u0027re saying because we\u0027re filtering things for this test you want to make sure it\u0027s consistent for just the tests in the file. In that case I think manually generating the name up front is fine.","accounts_in_message":[],"_revision_number":2},{"id":"e6f8ae5396dec0de5a6af738639507e89b8f6cf7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-04 21:19:16.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/19/49819/2/check/gate-tempest-pep8/7675a79 : SUCCESS in 1m 25s\n- gate-tempest-docs http://logs.openstack.org/19/49819/2/check/gate-tempest-docs/27eb3ee : SUCCESS in 3m 41s\n- check-tempest-devstack-vm-full http://logs.openstack.org/19/49819/2/check/check-tempest-devstack-vm-full/0886bda : SUCCESS in 44m 21s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/19/49819/2/check/check-tempest-devstack-vm-postgres-full/85875e6 : SUCCESS in 39m 46s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/19/49819/2/check/check-tempest-devstack-vm-neutron/04595ba : SUCCESS in 22m 34s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/19/49819/2/check/gate-tempest-devstack-vm-large-ops/3585f92 : SUCCESS in 10m 43s\n- check-grenade-devstack-vm http://logs.openstack.org/19/49819/2/check/check-grenade-devstack-vm/bebc2af : SUCCESS in 27m 34s\n","accounts_in_message":[],"_revision_number":2},{"id":"fe54db8d2cb889adead5f479eaca286065d0e044","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2013-10-04 21:31:23.000000000","message":"Patch Set 2:\n\nMatt, yeah, the thing is we can let BaseComputeTest.create_server generate the name and then we can call client.get_server to get the server details and use that through the tests, which removes my need to explicitly get the server in the filtered_by_ip tests (and would remove the need for me to call rand_name in this code anymore).\n\nIt\u0027s really kind of six of one, half dozen the other. They both have pros and cons so I just wanted to point it out as something I was thinking about as I wrote this.","accounts_in_message":[],"_revision_number":2},{"id":"02cb5c212eaed4526dfcb14241564f52d315a651","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2013-10-07 19:12:04.000000000","message":"Patch Set 2: Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"a8da691f1ea0b2646033190ca703a2e7e68f727f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-07 19:12:05.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"3d7c2a1db81ce557a34e28959f4edd2f4fd3d430","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-07 19:41:56.000000000","message":"Patch Set 2: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/19/49819/2/gate/gate-tempest-pep8/7c6e345 : SUCCESS in 1m 40s\n- gate-tempest-docs http://logs.openstack.org/19/49819/2/gate/gate-tempest-docs/4424f60 : SUCCESS in 3m 55s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-full/ca32542 : FAILURE in 8m 12s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-postgres-full/34f72c2 : SUCCESS in 29m 37s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-neutron/f734c58 : SUCCESS in 19m 56s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-large-ops/7aa2855 : FAILURE in 8m 08s\n- gate-grenade-devstack-vm http://logs.openstack.org/19/49819/2/gate/gate-grenade-devstack-vm/baece5f : FAILURE in 8m 15s\n","accounts_in_message":[],"_revision_number":2},{"id":"865b9dac1dcc6f3dda1da6c5d85a7811d3d4a723","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-07 21:36:20.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"257c75324b3138747608181c7f6673c85e04c7cf","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2013-10-07 21:36:20.000000000","message":"Patch Set 2:\n\nreverify bug 1232283","accounts_in_message":[],"_revision_number":2},{"id":"6be0404e1366d4096acd9626495d1ec18686ef07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-07 22:05:31.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/19/49819/2/gate/gate-tempest-pep8/8a3dda5 : SUCCESS in 1m 53s\n- gate-tempest-docs http://logs.openstack.org/19/49819/2/gate/gate-tempest-docs/047980b : SUCCESS in 3m 59s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-full/80e8ee5 : SUCCESS in 28m 18s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-postgres-full/95f2be8 : SUCCESS in 28m 58s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-neutron/27babd7 : SUCCESS in 17m 22s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/19/49819/2/gate/gate-tempest-devstack-vm-large-ops/c816862 : SUCCESS in 12m 24s\n- gate-grenade-devstack-vm http://logs.openstack.org/19/49819/2/gate/gate-grenade-devstack-vm/2fd6e72 : SUCCESS in 22m 17s\n","accounts_in_message":[],"_revision_number":2},{"id":"bfb5d69bdee7b3c25480d59e53df805a8912795c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-07 22:05:32.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12c29e02920ec3d8d8fb4373bbed1d47da3a3cf3","revisions":{"77bb72c1219abaff34539359259533168b3c88eb":{"kind":"REWORK","_number":1,"created":"2013-10-04 19:52:32.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/19/49819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/49819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/49819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/49819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/49819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/49819/1"}}},"commit":{"parents":[{"commit":"385f0b116e8f02d24338e0f11f4ae3ccf2edd661","subject":"Merge \"Do not check for id in the keystone output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/385f0b116e8f02d24338e0f11f4ae3ccf2edd661"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2013-10-04 19:32:26.000000000","tz":-420},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2013-10-04 19:52:14.000000000","tz":-420},"subject":"Cleanup test_list_server_filters setup/teardown","message":"Cleanup test_list_server_filters setup/teardown\n\nThe test_list_server_filters test class was doing a lot of\nsetup/teardown work that is now handled in the parent class. We still\nhave to explicitly set the name since some tests use it for filtering\nbut otherwise we can rely on the parent class to do most of the heavy\nlifting here.\n\nThis change makes it so that the instances have a name based on the test\nclass rather than just a random integer which is hard to use for\ndebugging leaked instances.\n\nThis change also removes the get call from setup and moves it into the\ntest cases that actually need to use it (which are currently being\nskipped anyway).\n\nCloses-Bug: 1235411\n\nChange-Id: I66813c900059dd4232de645c225032e58e9aecde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/77bb72c1219abaff34539359259533168b3c88eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/77bb72c1219abaff34539359259533168b3c88eb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"385f0b116e8f02d24338e0f11f4ae3ccf2edd661","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"12c29e02920ec3d8d8fb4373bbed1d47da3a3cf3":{"kind":"REWORK","_number":2,"created":"2013-10-04 20:34:37.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/19/49819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/49819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/49819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/49819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/49819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/49819/2"}}},"commit":{"parents":[{"commit":"385f0b116e8f02d24338e0f11f4ae3ccf2edd661","subject":"Merge \"Do not check for id in the keystone output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/385f0b116e8f02d24338e0f11f4ae3ccf2edd661"}]}],"author":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2013-10-04 19:32:26.000000000","tz":-420},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2013-10-04 20:33:53.000000000","tz":-420},"subject":"Cleanup test_list_server_filters setup/teardown","message":"Cleanup test_list_server_filters setup/teardown\n\nThe test_list_server_filters test class was doing a lot of\nsetup/teardown work that is now handled in the parent class. We still\nhave to explicitly set the name since some tests use it for filtering\nbut otherwise we can rely on the parent class to do most of the heavy\nlifting here.\n\nThis change makes it so that the instances have a name based on the test\nclass rather than just a random integer which is hard to use for\ndebugging leaked instances.\n\nThis change also removes the get call from setup and moves it into the\ntest cases that actually need to use it (which are currently being\nskipped anyway).\n\nCloses-Bug: 1235411\n\nChange-Id: I66813c900059dd4232de645c225032e58e9aecde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/12c29e02920ec3d8d8fb4373bbed1d47da3a3cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/12c29e02920ec3d8d8fb4373bbed1d47da3a3cf3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"385f0b116e8f02d24338e0f11f4ae3ccf2edd661","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
