)]}'
{"id":"openstack%2Ftempest~527577","triplet_id":"openstack%2Ftempest~master~I82bb3c5bbeaa4de04f607266311e8452e25ce6b5","project":"openstack/tempest","branch":"master","topic":"update-version-in-subunit-describe-calls","hashtags":[],"change_id":"I82bb3c5bbeaa4de04f607266311e8452e25ce6b5","subject":"Update subunit_describe_calls about new ports","status":"MERGED","created":"2017-12-13 05:19:37.000000000","updated":"2021-03-04 21:31:50.000000000","submitted":"2021-03-04 21:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"527577-update-version-in-subunit-describe-calls","meta_rev_id":"12176154af2536743c64c2a7affdd811d40b5b77","_number":527577,"virtual_id_number":527577,"owner":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 21:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-04 13:13:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 16:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-04 16:41:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-13 06:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-13 07:15:04.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 11:32:07.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2021-03-02 16:57:02.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2021-03-02 20:45:50.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-03-04 15:04:56.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"212b71b919a4213a4a3b9b267cb72ee7f986a1e6","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-12-13 05:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"092da700859d2a2f61d1332bd493cd6bae1fba65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 06:56:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-py35/b6bd9d2/ : SUCCESS in 1h 19m 48s\n- openstack-tox-pep8 http://logs.openstack.org/77/527577/1/check/openstack-tox-pep8/545f241/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/77/527577/1/check/openstack-tox-py27/377994b/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/77/527577/1/check/openstack-tox-py35/61f0ac3/ : SUCCESS in 6m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/77/527577/1/check/build-openstack-sphinx-docs/85e757d/html/ : SUCCESS in 5m 36s\n- legacy-grenade-dsvm-neutron-multinode http://logs.openstack.org/77/527577/1/check/legacy-grenade-dsvm-neutron-multinode/3c5e3be/ : SUCCESS in 55m 20s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/77/527577/1/check/legacy-grenade-dsvm-neutron/4edf692/ : SUCCESS in 46m 22s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/1bcb204/ : SUCCESS in 1h 16m 26s (non-voting)\n- puppet-openstack-integration-4-scenario001-tempest-centos-7 http://logs.openstack.org/77/527577/1/check/puppet-openstack-integration-4-scenario001-tempest-centos-7/657aca5/ : SUCCESS in 1h 03m 17s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/77/527577/1/check/puppet-openstack-integration-4-scenario002-tempest-centos-7/6c21071/ : SUCCESS in 59m 07s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 http://logs.openstack.org/77/527577/1/check/puppet-openstack-integration-4-scenario003-tempest-centos-7/966179b/ : SUCCESS in 57m 46s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/77/527577/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/bb9339e/ : SUCCESS in 48m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-dvr http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-neutron-dvr/01f28b7/ : SUCCESS in 1h 17m 26s\n- legacy-tempest-dsvm-neutron-full-pike http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-neutron-full-pike/40c745d/ : SUCCESS in 1h 30m 22s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-neutron-full-ocata/ed82101/ : SUCCESS in 1h 23m 18s\n- legacy-tempest-dsvm-neutron-full-newton http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-neutron-full-newton/6a543ec/ : SUCCESS in 1h 20m 58s\n- legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-neutron-scenario-multinode-lvm-multibackend/fdb321b/ : FAILURE in 1h 23m 46s (non-voting)\n- legacy-tempest-dsvm-neutron-multinode-full http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-neutron-multinode-full/6cd4541/ : SUCCESS in 1h 30m 19s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/77/527577/1/check/openstack-tox-cover/ff2adec/cover/ : SUCCESS in 3m 05s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/527577/1/check/legacy-tempest-dsvm-neutron-full/a1b03fe/ : SUCCESS in 1h 14m 15s\n- tempest-full http://logs.openstack.org/77/527577/1/check/tempest-full/396b3cd/ : SUCCESS in 1h 16m 11s (non-voting)\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/77/527577/1/check/tempest-tox-plugin-sanity-check/6441d02/ : POST_FAILURE in 50m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab855c1a8058349415ae580bbe0d511bac53ecb6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2017-12-13 07:15:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/77/527577/1/check/dsvm-tempest-neutron-network/d2d37c4 : SUCCESS in 1h 54m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e11af32a5c85e89b47d56cc973977d2fd4a9a3ab","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-02-17 11:32:07.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for updating the docs. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"eebcc98f9b80a2d639ce2db17a3d1c14ca1f92d5","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-02 16:57:02.000000000","message":"Patch Set 1:\n\nrecheck to get fresh logs","accounts_in_message":[],"_revision_number":1},{"id":"271855242599c49b6bb357be84352d28f71b8114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 20:21:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c2f1fe8f167545f49b42af54765fb1b2 : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/208c2046fe514e9b9cf7c3efea3456b8 : SUCCESS in 1h 35m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85f260557621435b8f70f2696be41f8f : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe5cf6bc2284389a7d507f9cdacd6cf : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/550625f2afde4646be36d0eff858dc2a : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86453ea1e6844ca296abce466aed9174 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8596caa72aa840aab676032755952369 : SUCCESS in 20m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eb0ac74835164976a45893fbaed40e65 : FAILURE in 2h 00m 03s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/20bbc0392f894bf19a61f8e042cbc768 : SUCCESS in 1h 33m 46s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d5eb14d544242a6a23d2d7871e26145 : SUCCESS in 1h 27m 35s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/7d9c9830083a4eac937ede3c1fa6ae58 : SUCCESS in 1h 39m 54s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d024ff6436b24c3ca94b8da25bc67fe5 : SUCCESS in 1h 34m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/51a437219e464d7492964dc5b8bc7d42 : SUCCESS in 1h 42m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/39ad5cfbb83e40d68e72ebdc5d55e895 : SUCCESS in 1h 56m 40s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/68bba7cce6e442d7ac0dbccd42daf8cf : SUCCESS in 41m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b14a8827d3c4fa9ae24cfda53c4f755 : SUCCESS in 1h 07m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ecc6ea8cd4254c96806b0112e08d0c1f : SUCCESS in 2h 22m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/44b8aa72ce934cb99b0765765bb58839 : FAILURE in 1h 17m 31s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d77a4ec155f1409099f929483b5ca6e9 : POST_FAILURE in 1h 29m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9bd25a365d64e55a44402ca56b25c5a : SUCCESS in 1h 04m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d244ca1262a4437f9fff578c6cb38b6d : SUCCESS in 1h 03m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/61e363d05a084a68a7ffb5874a30da8d : SUCCESS in 4m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/25b8dd49c87249acaf43ab3595bf9c90 : FAILURE in 1h 16m 27s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1fe42f2bf04f451e9340e05d1b2b9028 : FAILURE in 1h 04m 01s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/59a9f1f41255407c81cd17d24a6da638 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e04de2995e245e5fc72af1e89f559f9829030e89","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-02 20:45:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nI\u0027m gonna update this as it\u0027s 3 years old. The main goal is ok, it nicely improves the doc.","accounts_in_message":[],"_revision_number":1},{"id":"c6cc1167b8d20f5fb532e9019985d1fde809ecb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-02 20:48:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b538cf64629ae5babe2b73a47ef958fb64e04339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:05:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f9382acf90440eda791c4bd207c6886 : SUCCESS in 57m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc80102828f8442985babf1f328773f2 : SUCCESS in 1h 30m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9237ed55aebf420f9d5ec9d78d0d2512 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fb531a7fe846ae8c18c3951a9cd9d1 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c9eddb112e49bba45183338c045786 : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bac487bab2342959ddf1872dc819898 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6fb8d224488442b8824bc3de99eb112 : FAILURE in 13m 57s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3362e029f35449b8a631a18cd713e502 : SUCCESS in 1h 19m 55s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/5421fd2c4c1b41499febe4414c34e7ad : SUCCESS in 1h 22m 22s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d633eb63699743e0b8bbc0a59bd8a471 : FAILURE in 1h 47m 58s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/5a2668cfcc5d488c878125c99eff644f : FAILURE in 1h 12m 45s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/81dc04f7917143218f8edfd10da1dcc1 : SUCCESS in 1h 29m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/d174de707dce40fcadf8c2170cb62a65 : SUCCESS in 1h 26m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/566004fc1c144ec4bc8be35e63b63825 : SUCCESS in 1h 51m 07s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a44bae037b0a498382e2c0af4bea0ffd : SUCCESS in 40m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0ff31bc0e844b34ba636923f2bce3f7 : SUCCESS in 1h 02m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1497d7e0b5f4ae7bf2198f017c55ea3 : SUCCESS in 2h 11m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c220c6bf387441a9bf517807fa56d94b : SUCCESS in 59m 37s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/16ed95e9646344de92daeb839a8d6813 : SUCCESS in 1h 32m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4d8767ba5f14c8db902cbe772bf9b9f : SUCCESS in 1h 06m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2f585ca68c0d4e508c1d314e297d9137 : SUCCESS in 1h 02m 50s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/24d7b98c84dd4340b13e3152befa8240 : SUCCESS in 4m 06s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9930ed77b8f1471e97e0edce84bcf7e3 : FAILURE in 1h 14m 04s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/cfc0ce80917a4e2aad7a48b5bb22e501 : FAILURE in 1h 11m 37s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8f8c684be68d4b82992557f05f1f8af8 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"08ca28cc6a8a331672bb64d787504b7a10af71dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-03 09:02:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1f3c545f98d2dae6eb93d1163aa605c676addbfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 11:31:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5b9be09fa0e648f3b0dc889eb7edf629 : SUCCESS in 1h 02m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e5babc9d6d7494ba033da01f7f793f1 : SUCCESS in 1h 26m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/009d1c1ae3ec460c85a9b1c0881e441c : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de932314136a4239bb2a0945b274a3df : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cefbaad8b493439184c4939e29cc1950 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c1dc35ec894c77833edae19490df61 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b08f4ad93a541c498fe3f09ad8a2aa3 : SUCCESS in 20m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/ab9d4bd575834316aa94762e631f612d : SUCCESS in 2h 19m 44s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/53bc37f6630b40ae8f3d385f5eb1f57f : SUCCESS in 1h 22m 44s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f5a0abd8dded41a4b2bb25fc8b2590a3 : FAILURE in 1h 14m 29s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/6715b4fd64fd45198b29d942ee79e2cd : TIMED_OUT in 2h 09m 53s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/48c1d7c240524737b1b62870def08802 : SUCCESS in 1h 28m 53s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/394e0ab7482d4b0e8a0c5c350c89eb39 : TIMED_OUT in 2h 07m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9716bd48bfd4cfd9e999ae19a3ac7f7 : SUCCESS in 1h 42m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/21dbd85a1c5a47e68cffe355dde3b37d : SUCCESS in 37m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7df988a6b7d4f07a46a460972bad61c : SUCCESS in 1h 02m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b530d899a2f841d7ade86751d82f65bc : SUCCESS in 2h 00m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b9aa449707b84be18c1e8e4903d3f180 : SUCCESS in 46m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/252d7d8567cf440da0a34e8939efbe98 : SUCCESS in 1h 09m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31dca29171b34b1f8ea0a4900b25e60d : SUCCESS in 1h 00m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/332a070dd6c54749a88864453bab8ee1 : SUCCESS in 1h 05m 49s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/baa911f9b4dd4cdda04ced0f63d05d0c : SUCCESS in 4m 33s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/07bb3ea6e6f74ee3af0dd72df39465e6 : FAILURE in 1h 12m 49s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/dbd3810edaa24b6492db616f7a5fec15 : FAILURE in 1h 04m 25s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/78424bb14a4e4f8f9aaf1038d41e2e52 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"242c2bb3e27d5adb9e098439a46af2e1e025bc53","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-03 13:06:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9ec4dbe518e87ea4b0f4077ae7adcfb014466305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:17:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/edb5950973254b3fb0910e25f3dfea2a : SUCCESS in 1h 02m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14fcd1a85ee84416a518b5a3388f70d1 : SUCCESS in 1h 29m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b996499fb7eb4705a9b507149cf23a36 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1b5154820a4d8da52353df526d001d : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bf6d2a579ae4bb2a9d8c343581c203d : FAILURE in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6db7fcc9e0904d36a57482cb65d8cb48 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caceb079936e428e9ca33f0a86624a38 : SUCCESS in 14m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b1ae90a2f1674d62a2f07073bfda1dcc : SUCCESS in 1h 38m 00s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3a3f8e8ac25d4171b5aca9fe62c15216 : SUCCESS in 1h 20m 24s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a02a97c772740ef9b139d0d03228f67 : SUCCESS in 1h 20m 55s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/a107b6907e534a6f825109519b263fc3 : SUCCESS in 1h 24m 26s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f227bb075aa64a419549a19c239dd509 : SUCCESS in 1h 26m 20s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/fbcfd7824f654340938f757792b3da79 : SUCCESS in 1h 24m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/88718e73568448d5997196a47e4808cd : SUCCESS in 1h 49m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/891765ed92664e2d8c3107d5df703377 : SUCCESS in 51m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6880533f616849f6bca4c5b17d2fd436 : SUCCESS in 1h 16m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/afeeeb570c2a43e1870e41497a0bea49 : SUCCESS in 2h 04m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e64d0c100f7427abe061453bd50753a : SUCCESS in 55m 21s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0254c85a07c54d1da1d60ffdc3c72e29 : FAILURE in 1h 37m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab67623cd6364c63a139332eeb2f3ac1 : SUCCESS in 57m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e0db827f81ac460bb851c7a05b133db8 : SUCCESS in 1h 01m 58s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/48c0a327c89443dc956149dd61abac0c : SUCCESS in 3m 53s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/022e10f230024c1d9574074923c079af : FAILURE in 1h 40m 56s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/26073d5d510244a2a9be1a67f2c3cfa3 : FAILURE in 1h 03m 17s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/784c239d567646d88003bdb0906dcf18 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"02f4c8cba4f8f45bdf9e343ed7dadc5cbb7e3c93","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-03 15:28:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3be95c6e68281091e151e86d5763ff22bab26a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:25:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b3797436d97443b38d95522f7158e0b7 : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8429186b1fab48cba936640acdedd634 : FAILURE in 1h 17m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b84616c5be084d9faf1e51240cf24f71 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e15640e438e414b8aac159004ef2629 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f58a1872b52c4709886890eee42c4b72 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aef0f9fea7145b49764513ca5e008e6 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a62a71f707745f3aa0a8fa12c32b263 : SUCCESS in 12m 50s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1f56362213c945278cab8148af03ea50 : SUCCESS in 1h 45m 39s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d71c2c3489044a24b11375a7c967f9e8 : SUCCESS in 1h 20m 51s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/64c6268b3b96474d81b5be10a83f9d40 : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/91af85eadf124b64be095c0a95b7e2a6 : SUCCESS in 1h 29m 39s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/a2c06f661a874cd188c8086824f037c4 : SUCCESS in 1h 37m 49s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/0535c1e8edad48928b3a969b38538bb6 : SUCCESS in 1h 31m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/55980f73d05a437e93f2c4853a1b0724 : SUCCESS in 1h 39m 09s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/44470f42525a497ba1d5678ce1524aaa : SUCCESS in 43m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fd23f2b191a4eaea8be99210d8bc679 : SUCCESS in 1h 02m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6e0ecb246d3436e97bc5214b5e83db9 : SUCCESS in 2h 18m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/530661c2bcb54f888cb9faec6e00bace : SUCCESS in 1h 00m 51s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c2756b77a07c4e889dfd7262f0af66f2 : SUCCESS in 1h 02m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a970d8fa21ca46969f90d2fb789c2e88 : SUCCESS in 1h 03m 35s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b4e19129e5cd4c00bdf6f0d2c657edcd : SUCCESS in 1h 08m 41s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c33bd526071341f4887b1beb9cd7bf5c : SUCCESS in 4m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6151b3d7d2ce4cca8d83068e53146e9d : FAILURE in 2h 00m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2c3ba38e9e5148c5905885a73eab932b : FAILURE in 1h 13m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/81ea1e4e1d4b4df0b14145c1fd9dd1d8 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"c17f5232ef7a6655f2d414c3a8ad29dac444a668","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-03 22:51:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"90b0e693522a55500e7ff464ac3ac313e594ae90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:30:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/97e07af848c5481ab61b1e0f04f85ccf : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ea5fd5634a74d8482db2015b56f46d8 : SUCCESS in 1h 25m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a60a867f4b44921ba6fa44847285dac : FAILURE in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c409134cf4a640849f5f3efbec4d6270 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3120a2a83114e2ca8f5ba59ed2930f5 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/167d5e97636444988d7de54b43984e74 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc01849133fd41ac9b8b60e746f1aef8 : SUCCESS in 16m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/9125af28a4f64cefb8b874ac8aca3b51 : FAILURE in 1h 45m 35s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a5c0eef318174544b5bfbacea9b3dc0a : SUCCESS in 1h 17m 51s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b14e035f0e484d4aa9e8f6d2130dd0fa : SUCCESS in 1h 32m 08s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/42cda2325fea4ca3b124c17cb0073eab : SUCCESS in 1h 37m 36s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/bf751025cb284ffba126e270cb7384d7 : SUCCESS in 1h 43m 51s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c238c687b4e34a06bb09d2496fda12e5 : SUCCESS in 1h 28m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7509e7378e6946f194a165d1e4814ccc : SUCCESS in 1h 31m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/fcce974a76c74a798f07029026a1d26a : SUCCESS in 31m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02a31e12cd0b4c42b6f34785f6107668 : SUCCESS in 1h 03m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d693e29736b47ffb98ee1ae7d21a7cb : SUCCESS in 2h 19m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f099c8d0d00e44b5bcf5700f428dbc3c : FAILURE in 52m 06s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/055126deece6431a90ebbe69262b421e : SUCCESS in 1h 17m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e497fe53ffca42e9a9a96f4c67d08bb4 : SUCCESS in 1h 04m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/df6f1efee19c4062935c00dd9ceff5bf : SUCCESS in 1h 15m 05s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/584b05b4adeb40fcbff7ff2ae435eb21 : SUCCESS in 3m 36s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/bdfd4c8f4f5142698503ed2d5c13fd31 : FAILURE in 1h 00m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/39a5c4af7ff747b4973fd62c5c4f1bc0 : FAILURE in 1h 01m 51s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/30217cb5d20e42ffb3ecb639437da537 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bd6edb20776eaa4e8fd5e515ab19ec0729d005ca","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-04 07:56:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cdf45f7b527bd97c11fb466bb2f399d1a33ab909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:13:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0a9b98b2ec244ae0b4e8aa1db01724a9 : SUCCESS in 1h 10m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0b80106436d4d96bb877d11d01053c0 : SUCCESS in 1h 36m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99d22121089f49dc8b30a7c838bcec8e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ddd015194f470ca1b606abdd65a5aa : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dee59d2532e54de4a14c43da59b430d9 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed837fcb2bd5409ba3082f7cc0b5d852 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca05fedbc04a4ff5b8d24a53d734f88a : SUCCESS in 11m 27s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/52edfcf147904531b3465a4b5a4a2786 : SUCCESS in 1h 34m 58s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/966eb07dd0364e96846d4af3febb27b6 : FAILURE in 1h 29m 40s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/08079599e3484f84917db4ce16f98f20 : SUCCESS in 1h 27m 39s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/ac2cf753c3794fe7a94f6156f01bae2d : SUCCESS in 1h 36m 20s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/349bce8307f54008b0ca6567ff09cadf : SUCCESS in 1h 29m 09s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/43ec2a846ebb4c3887bb8495324945a2 : SUCCESS in 1h 38m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/28a0699b378641858d843298861d795d : SUCCESS in 1h 43m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b498973cc3594095a9f276204e13a7c6 : SUCCESS in 37m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/deac737abb794fd5aaa0363b5dc67619 : SUCCESS in 1h 02m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c9d1e5b35444501babe0ca3336f3b68 : SUCCESS in 2h 15m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3a1ec64fbb1e4e95b05823de03397b71 : SUCCESS in 45m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/01f577a77e5c41cb884a03e8fd48ceba : FAILURE in 1h 30m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/16aefc9417764f1294bd5994cd1bcb98 : SUCCESS in 56m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5c833c8f7f234b9d8d793eb96dd2c1fc : SUCCESS in 54m 16s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9dd11fecc356445fac205d2f25b92ff6 : SUCCESS in 3m 15s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/30ce53107d294a17b6025ed19bacf3f6 : SUCCESS in 1h 17m 28s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/514c25ea182a4a799ea97c240635e048 : SUCCESS in 56m 43s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b8f32b556e5a4a0ab654a6949061e7c6 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1640d1d00f95e4160a33b136176fe3b39837ef1d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-04 10:18:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ced21e8674a2654a11bf81fd620e009c27a91592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 12:38:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/cac28d2e44b3433aaf035fdfe4405240 : SUCCESS in 1h 14m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/631a63bbd46c4a92b85d0b1c3077ad30 : SUCCESS in 1h 18m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f364e2d3e96b417eab32e489ce113edf : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a56ba98f7054f40b0a4751fa0677dc8 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a8598da4ebe4d9da72deb3227c754ad : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b98a7c5c10742cc91ad6a2fb323cbf6 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0599591ced164627b09ee1b9fb17fd42 : SUCCESS in 15m 28s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2d33dded6edf4105a5fcfaa94b5a3c50 : SUCCESS in 1h 21m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7ee40f24ce7a42d8918ac7a70b1f4d08 : SUCCESS in 1h 28m 29s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/98dbe1a0a29f4ef7812f7dd368f5afcc : SUCCESS in 1h 20m 48s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/d34ece3d71464e6eac06165ce47b1fd6 : SUCCESS in 1h 21m 31s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/403ef30c54f54df2993e85b5d8dff4fc : SUCCESS in 1h 24m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c2014fc832f14af290c7e8f887536293 : SUCCESS in 1h 20m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6f8345d64aba43eea98c44e26aa4788a : SUCCESS in 1h 40m 06s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/dd050ce053754ea884d755dd3a969910 : SUCCESS in 41m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61c433ffb7f44af88a7569b5575bef22 : SUCCESS in 57m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e68884ea98c4e54811944537fac0306 : SUCCESS in 2h 13m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f16cce0fa19b4dbba521006436da9b71 : SUCCESS in 53m 47s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/53912e3717d84d0bb324290a11972a1b : SUCCESS in 1h 11m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/645a0e0249c548c785357152de5bdb10 : SUCCESS in 1h 09m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d8c3ed6784ce43ff875c7775f6ea2f33 : FAILURE in 58m 07s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1f792f5e198141eaaaafbabc8b7a8895 : SUCCESS in 3m 50s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b000ca9c52c641868407c57105a2cbc0 : FAILURE in 1h 10m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6a086e946cbd43da961a5b1949dff371 : FAILURE in 1h 05m 07s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7f054a02ad1445f281b7bef3d153dcaf : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"4f35010d17d81be5ea7b5a0eb72fee9cac68f04a","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-03-04 13:13:46.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM! Thanks for updating the subunit_describe_calls() for new ports","accounts_in_message":[],"_revision_number":3},{"id":"8c6b78dac3bbb68dcfa0304459b7e7e30f16ca7f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-04 16:41:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nlgtm thanks","accounts_in_message":[],"_revision_number":3},{"id":"b71e128b5629fc122b27c3510505a7c434f6e8fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:43:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a18458fc655489aaf91d543b8963111f2c13e7bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:27:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7cedcfdf696844019845d720bd8a8e9b : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65ba59ab75ea47bba2f6a91befa0e6e2 : SUCCESS in 1h 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc92be146aea4fb5ac17a925f35d4ca1 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a72f623b53e44d09fa105042632092d : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d28b7e12373342e3a0d03fe45e2840a9 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf964fde8804809b348166094206a6a : SUCCESS in 13m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f70a72a8f8b40ab81692cfcac0489b4 : SUCCESS in 2h 08m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa11fcac45bb4419a586e1944a62e8ef : SUCCESS in 55m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87cc67286df74cceb57280f1a1e8b5af : SUCCESS in 1h 00m 04s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/76f588463cd847cd8843e5c8c9912930 : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":3},{"id":"f04f8340da3de4a5cf22f1c1e69a5b29140e191a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:27:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"12176154af2536743c64c2a7affdd811d40b5b77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:31:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e17c70d3704cc88c3f9228e180b5e9 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f7d53297ba58563ece8765be61cb1bc95e2fc8b4","revisions":{"7983ed8763e2203e77ab5aa5c3739107a159dcd8":{"kind":"REWORK","_number":1,"created":"2017-12-13 05:19:37.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/77/527577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/77/527577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/77/527577/1"}}},"commit":{"parents":[{"commit":"c536175d8d4d572278b4e592dadc548b34e4ea91","subject":"Merge \"Refactor volume groups and group snapshots tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c536175d8d4d572278b4e592dadc548b34e4ea91"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2017-12-13 05:13:28.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2017-12-13 05:13:28.000000000","tz":540},"subject":"Update version in subunit_describe_calls","message":"Update version in subunit_describe_calls\n\nThis commit updates the default version explanation in\nsubunit_describe_calls doc from Kilo to Newton. It seems like Newton[0]\nis the latest for that. From Ocata, there\u0027s no \"Firewalls and default\nports\" section in that doc, anymore.[1]\n\nAnd this commit also fixes some markups as inline literals.\n\n[0] https://docs.openstack.org/newton/config-reference/firewalls-default-ports.html\n[1] https://docs.openstack.org/ocata/config-reference/\n\nChange-Id: I82bb3c5bbeaa4de04f607266311e8452e25ce6b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7983ed8763e2203e77ab5aa5c3739107a159dcd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7983ed8763e2203e77ab5aa5c3739107a159dcd8"}]},"branch":"refs/heads/master"},"88e679f14223d4417c11ea8dafef36e4f1cfde2a":{"kind":"REWORK","_number":2,"created":"2021-03-02 20:48:35.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/77/527577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/77/527577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/77/527577/2"}}},"commit":{"parents":[{"commit":"c536175d8d4d572278b4e592dadc548b34e4ea91","subject":"Merge \"Refactor volume groups and group snapshots tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c536175d8d4d572278b4e592dadc548b34e4ea91"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2017-12-13 05:13:28.000000000","tz":540},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-03-02 20:40:28.000000000","tz":0},"subject":"Update subunit_describe_calls about new ports","message":"Update subunit_describe_calls about new ports\n\nThis commit updates the default version explanation in\nsubunit_describe_calls. It seems the \u0027Firewalls and default ports\u0027 is moved\nto a new location. In Newton it was here [0] but from Ocata it gives 404 [1].\nIt got moved to [2].\n\nAnd this commit also fixes some markups as inline literals.\n\n[0] https://docs.openstack.org/newton/config-reference/firewalls-default-ports.html\n[1] https://docs.openstack.org/ocata/config-reference/\n[2] https://docs.openstack.org/install-guide/firewalls-default-ports.html\n\nChange-Id: I82bb3c5bbeaa4de04f607266311e8452e25ce6b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/88e679f14223d4417c11ea8dafef36e4f1cfde2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/88e679f14223d4417c11ea8dafef36e4f1cfde2a"}]},"branch":"refs/heads/master"},"f7d53297ba58563ece8765be61cb1bc95e2fc8b4":{"kind":"REWORK","_number":3,"created":"2021-03-03 09:02:31.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/77/527577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/77/527577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/77/527577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/77/527577/3"}}},"commit":{"parents":[{"commit":"c536175d8d4d572278b4e592dadc548b34e4ea91","subject":"Merge \"Refactor volume groups and group snapshots tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c536175d8d4d572278b4e592dadc548b34e4ea91"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2017-12-13 05:13:28.000000000","tz":540},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-03-03 08:56:16.000000000","tz":0},"subject":"Update subunit_describe_calls about new ports","message":"Update subunit_describe_calls about new ports\n\nThis commit updates the default version explanation in\nsubunit_describe_calls. It seems the \u0027Firewalls and default ports\u0027\nis moved to a new location. In Newton it was here [0] but from Ocata\nit gives 404 [1] because it got moved to [2].\n\nAnd this commit also fixes some markups as inline literals.\n\n[0] https://docs.openstack.org/newton/config-reference/firewalls-default-ports.html\n[1] https://docs.openstack.org/ocata/config-reference/\n[2] https://docs.openstack.org/install-guide/firewalls-default-ports.html\n\nChange-Id: I82bb3c5bbeaa4de04f607266311e8452e25ce6b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f7d53297ba58563ece8765be61cb1bc95e2fc8b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f7d53297ba58563ece8765be61cb1bc95e2fc8b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
