)]}'
{"id":"openstack%2Ftempest~55101","triplet_id":"openstack%2Ftempest~master~I212873d46e20ff3161e376105af18020b1cf1c9d","project":"openstack/tempest","branch":"master","topic":"cross_tenant_connectivity_isolated","hashtags":[],"change_id":"I212873d46e20ff3161e376105af18020b1cf1c9d","subject":"scenario cross_tenant_connectivity","status":"MERGED","created":"2013-11-03 21:00:55.000000000","updated":"2013-12-18 23:30:52.000000000","submitted":"2013-12-18 23:30:51.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":45,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9979aa575cc20b78a2cd66efdb5fb72a94ccb774","_number":55101,"virtual_id_number":55101,"owner":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},{"value":0,"_account_id":6690,"name":"Rami Vaknin","email":"rvaknin@redhat.com","username":"rvaknin"},{"value":2,"date":"2013-12-18 23:30:51.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"value":0,"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},{"value":0,"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6846,"name":"ofer blaut","email":"oblaut@redhat.com","username":"oblaut"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"recommended":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"date":"2013-12-18 17:05:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},{"value":0,"_account_id":6690,"name":"Rami Vaknin","email":"rvaknin@redhat.com","username":"rvaknin"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"date":"2013-12-18 22:12:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"date":"2013-12-18 15:28:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"value":0,"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},{"value":0,"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":1,"date":"2013-12-18 14:36:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2013-12-18 14:36:39.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6846,"name":"ofer blaut","email":"oblaut@redhat.com","username":"oblaut"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"all":[{"value":0,"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},{"value":0,"_account_id":6690,"name":"Rami Vaknin","email":"rvaknin@redhat.com","username":"rvaknin"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":1,"date":"2013-12-18 22:12:20.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"date":"2013-12-18 15:28:52.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"value":0,"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},{"value":0,"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6846,"name":"ofer blaut","email":"oblaut@redhat.com","username":"oblaut"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},{"value":0,"_account_id":6690,"name":"Rami Vaknin","email":"rvaknin@redhat.com","username":"rvaknin"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":0,"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"value":0,"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},{"value":0,"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"value":0,"_account_id":6846,"name":"ofer blaut","email":"oblaut@redhat.com","username":"oblaut"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6690,"name":"Rami Vaknin","email":"rvaknin@redhat.com","username":"rvaknin"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":6846,"name":"ofer blaut","email":"oblaut@redhat.com","username":"oblaut"},{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},{"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-11-03 21:03:15.000000000","updated_by":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"reviewer":{"_account_id":2035,"name":"Maru Newby","email":"marun@redhat.com","username":"maru"},"state":"REVIEWER"},{"updated":"2013-11-03 21:03:22.000000000","updated_by":{"_account_id":6690,"name":"Rami Vaknin","email":"rvaknin@redhat.com","username":"rvaknin"},"reviewer":{"_account_id":6690,"name":"Rami Vaknin","email":"rvaknin@redhat.com","username":"rvaknin"},"state":"REVIEWER"},{"updated":"2013-11-03 21:03:31.000000000","updated_by":{"_account_id":6846,"name":"ofer blaut","email":"oblaut@redhat.com","username":"oblaut"},"reviewer":{"_account_id":6846,"name":"ofer blaut","email":"oblaut@redhat.com","username":"oblaut"},"state":"REVIEWER"},{"updated":"2013-11-03 21:03:49.000000000","updated_by":{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},"reviewer":{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},"state":"REVIEWER"},{"updated":"2013-11-21 09:59:56.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2013-12-03 18:17:11.000000000","updated_by":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"reviewer":{"_account_id":261,"name":"Salvatore Orlando","email":"salv.orlando@gmail.com","username":"salvatore-orlando"},"state":"REVIEWER"},{"updated":"2013-12-03 18:25:25.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2013-12-04 11:05:10.000000000","updated_by":{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},"reviewer":{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},"state":"REVIEWER"},{"updated":"2013-12-16 22:18:15.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2013-12-17 00:05:25.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2013-12-17 06:11:19.000000000","updated_by":{"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},"reviewer":{"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},"state":"REVIEWER"},{"updated":"2013-12-18 14:36:39.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-12-18 14:36:39.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2013-12-18 15:28:52.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-12-18 22:12:20.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2013-12-18 23:30:51.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f7229d322a1c922c830f2f1afee528bcab82b47d","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-11-03 21:31:34.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\nCan we get a more concise commit message? Just dumping the doc string in is a lot.\n\nAlso, any new network tests have to be set as smoke tag for not otherwise they aren\u0027t run in the gate. And we want to make sure this executes in the gate before merging.","accounts_in_message":[],"_revision_number":1},{"id":"f830c8f3896246d27c5bef0ceadf2e47b7e30494","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-03 21:32:24.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/1/check/gate-tempest-pep8/da6c52b : FAILURE in 1m 37s\n- gate-tempest-docs http://logs.openstack.org/01/55101/1/check/gate-tempest-docs/bcb0f85 : SUCCESS in 1m 15s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/1/check/gate-tempest-python27/3ed4be6 : SUCCESS in 1m 47s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/1/check/check-tempest-devstack-vm-full/4502e74 : SUCCESS in 28m 39s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/1/check/check-tempest-devstack-vm-postgres-full/28f4f17 : SUCCESS in 31m 17s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/1/check/check-tempest-devstack-vm-neutron/af5d44d : SUCCESS in 16m 46s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/1/check/gate-tempest-devstack-vm-large-ops/8f0157f : SUCCESS in 13m 37s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/1/check/gate-tempest-devstack-vm-neutron-large-ops/b9fdae4 : FAILURE in 18m 37s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/1/check/check-grenade-devstack-vm/735728c : SUCCESS in 27m 49s\n","accounts_in_message":[],"_revision_number":1},{"id":"29513c14e95e81b3916e8b030db2ad3626a3fd23","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-04 07:59:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d44809da4f25003bbcc8f29f9c0cce9ef8c3a98","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-04 08:31:13.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/2/check/gate-tempest-pep8/656f546 : SUCCESS in 1m 46s\n- gate-tempest-docs http://logs.openstack.org/01/55101/2/check/gate-tempest-docs/3ed1bb9 : SUCCESS in 1m 04s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/2/check/gate-tempest-python27/375e757 : SUCCESS in 1m 42s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/2/check/check-tempest-devstack-vm-full/a70ae41 : SUCCESS in 29m 37s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/2/check/check-tempest-devstack-vm-postgres-full/6e1131c : SUCCESS in 30m 41s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/2/check/check-tempest-devstack-vm-neutron/f10ead0 : FAILURE in 26m 15s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/2/check/gate-tempest-devstack-vm-large-ops/0022b64 : SUCCESS in 17m 04s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/2/check/gate-tempest-devstack-vm-neutron-large-ops/e0a2775 : FAILURE in 18m 35s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/2/check/check-grenade-devstack-vm/dd2fb3c : SUCCESS in 31m 48s\n","accounts_in_message":[],"_revision_number":2},{"id":"694f785429da42e8f1058cea4e7d5d619c8d91f6","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-04 09:15:27.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"cc3897d942726e448c885ee3c033b87d0418ff4b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-04 09:47:07.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/2/check/gate-tempest-pep8/224e145 : SUCCESS in 1m 38s\n- gate-tempest-docs http://logs.openstack.org/01/55101/2/check/gate-tempest-docs/f67571e : SUCCESS in 1m 18s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/2/check/gate-tempest-python27/1664bf6 : SUCCESS in 1m 19s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/2/check/check-tempest-devstack-vm-full/4df8fd1 : SUCCESS in 27m 35s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/2/check/check-tempest-devstack-vm-postgres-full/8a1f856 : SUCCESS in 29m 17s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/2/check/check-tempest-devstack-vm-neutron/64fd9c9 : FAILURE in 30m 25s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/2/check/gate-tempest-devstack-vm-large-ops/fb0ab7e : SUCCESS in 13m 44s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/2/check/gate-tempest-devstack-vm-neutron-large-ops/c272d26 : FAILURE in 18m 45s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/2/check/check-grenade-devstack-vm/44c0d3f : SUCCESS in 31m 34s\n","accounts_in_message":[],"_revision_number":2},{"id":"8c19630b325bd3e71b143182ccc47cdfb1dafb69","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-04 11:16:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd1848120ad5348e62b24ce9a588dd07431d5873","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-04 11:54:36.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/3/check/gate-tempest-pep8/648147a : SUCCESS in 1m 52s\n- gate-tempest-docs http://logs.openstack.org/01/55101/3/check/gate-tempest-docs/d3677fc : SUCCESS in 1m 01s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/3/check/gate-tempest-python27/10bf7e8 : SUCCESS in 1m 19s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/3/check/check-tempest-devstack-vm-full/23471ba : SUCCESS in 28m 25s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/3/check/check-tempest-devstack-vm-postgres-full/3a79c4b : SUCCESS in 30m 09s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/3/check/check-tempest-devstack-vm-neutron/9f6edb7 : FAILURE in 31m 05s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/3/check/gate-tempest-devstack-vm-large-ops/ff2e74f : SUCCESS in 13m 35s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/3/check/gate-tempest-devstack-vm-neutron-large-ops/a986699 : FAILURE in 20m 55s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/3/check/check-grenade-devstack-vm/1e6867a : SUCCESS in 37m 44s\n","accounts_in_message":[],"_revision_number":3},{"id":"55b0689347d818be764b7573306f26f2a2d68583","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-04 15:13:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bb4f357b7f4f4bafb2076487a5c9841a3f257b4b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-04 15:46:11.000000000","message":"Patch Set 4: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/4/check/gate-tempest-pep8/49aab6c : SUCCESS in 1m 40s\n- gate-tempest-docs http://logs.openstack.org/01/55101/4/check/gate-tempest-docs/6304b4f : SUCCESS in 1m 18s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/4/check/gate-tempest-python27/edf5dc6 : SUCCESS in 1m 19s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/4/check/check-tempest-devstack-vm-full/b69b776 : SUCCESS in 31m 46s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/4/check/check-tempest-devstack-vm-postgres-full/67614ee : SUCCESS in 30m 56s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/4/check/check-tempest-devstack-vm-neutron/0df109d : FAILURE in 26m 59s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/4/check/gate-tempest-devstack-vm-large-ops/eca2eb9 : SUCCESS in 13m 30s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/4/check/gate-tempest-devstack-vm-neutron-large-ops/3bafe56 : FAILURE in 19m 09s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/4/check/check-grenade-devstack-vm/a5396a3 : SUCCESS in 32m 46s\n","accounts_in_message":[],"_revision_number":4},{"id":"a3918ba8f5f235e37f3aba2967635c774de0ffa1","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-05 10:22:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4b058821b7c12323219f58a381d8ca7e96ccc742","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-05 11:00:30.000000000","message":"Patch Set 5: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/5/check/gate-tempest-pep8/92591a9 : SUCCESS in 1m 23s\n- gate-tempest-docs http://logs.openstack.org/01/55101/5/check/gate-tempest-docs/4c3b7f7 : SUCCESS in 3m 11s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/5/check/gate-tempest-python27/5fc6bbc : SUCCESS in 1m 50s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/5/check/check-tempest-devstack-vm-full/ebf69f1 : SUCCESS in 30m 45s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/5/check/check-tempest-devstack-vm-postgres-full/63e4122 : FAILURE in 35m 49s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/5/check/check-tempest-devstack-vm-neutron/b4ac2fa : FAILURE in 31m 33s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/5/check/gate-tempest-devstack-vm-large-ops/2d899a0 : SUCCESS in 14m 54s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/5/check/gate-tempest-devstack-vm-neutron-large-ops/e3db3ab : FAILURE in 20m 57s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/5/check/check-grenade-devstack-vm/e2f3e5e : SUCCESS in 37m 36s\n","accounts_in_message":[],"_revision_number":5},{"id":"c738adafe323d7c567337d53d925ba647a67ff09","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-05 11:11:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2c2ab5faafc46e1296809c2bf25826cd46dfcc45","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-05 11:44:33.000000000","message":"Patch Set 6: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/6/check/gate-tempest-pep8/8db11c5 : SUCCESS in 1m 48s\n- gate-tempest-docs http://logs.openstack.org/01/55101/6/check/gate-tempest-docs/36b90cf : SUCCESS in 1m 17s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/6/check/gate-tempest-python27/c7899b9 : SUCCESS in 1m 19s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/6/check/check-tempest-devstack-vm-full/def885a : SUCCESS in 28m 01s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/6/check/check-tempest-devstack-vm-postgres-full/089b1f2 : SUCCESS in 28m 20s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/6/check/check-tempest-devstack-vm-neutron/8b88b15 : FAILURE in 27m 17s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/6/check/gate-tempest-devstack-vm-large-ops/bbd359e : SUCCESS in 13m 57s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/6/check/gate-tempest-devstack-vm-neutron-large-ops/a2952e8 : FAILURE in 19m 03s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/6/check/check-grenade-devstack-vm/9bc080a : SUCCESS in 32m 50s\n","accounts_in_message":[],"_revision_number":6},{"id":"426cb65293c8cab1a2c9effcd6fbf7ff524378ff","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-05 12:02:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4e8ea227ef9a5caa49fdb82ce1ad1decb7492028","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-05 12:35:12.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/7/check/gate-tempest-pep8/334b785 : SUCCESS in 1m 39s\n- gate-tempest-docs http://logs.openstack.org/01/55101/7/check/gate-tempest-docs/200eccc : SUCCESS in 1m 16s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/7/check/gate-tempest-python27/99bf35c : SUCCESS in 1m 37s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/7/check/check-tempest-devstack-vm-full/dcd6175 : SUCCESS in 30m 15s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/7/check/check-tempest-devstack-vm-postgres-full/2bb7699 : SUCCESS in 30m 59s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/7/check/check-tempest-devstack-vm-neutron/f0746c1 : SUCCESS in 27m 04s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/7/check/gate-tempest-devstack-vm-large-ops/9c16a1e : SUCCESS in 13m 52s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/7/check/gate-tempest-devstack-vm-neutron-large-ops/06bcee4 : FAILURE in 20m 46s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/7/check/check-grenade-devstack-vm/f377695 : SUCCESS in 32m 05s\n","accounts_in_message":[],"_revision_number":7},{"id":"3f607b46918f583f73d279453985baa07851d6fd","author":{"_account_id":6683,"name":"Pavel Sedlák","email":"psedlak@redhat.com","username":"psedlak"},"date":"2013-11-05 18:27:33.000000000","message":"Patch Set 6: (3 inline comments)\n\nFew small things, as i wrote them for one patchset before latest, not sure if they are still valid so no +/- now.","accounts_in_message":[],"_revision_number":6},{"id":"23938651f2581fead7e34295f0b6070478abe79b","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-11 06:59:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"15f10177e1ee01f89c7026ecc36f4fe431083ab5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-11 07:39:25.000000000","message":"Patch Set 8: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/8/check/gate-tempest-pep8/5a728b3 : SUCCESS in 1m 45s\n- gate-tempest-docs http://logs.openstack.org/01/55101/8/check/gate-tempest-docs/004c391 : SUCCESS in 1m 48s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/8/check/gate-tempest-python27/71acd85 : SUCCESS in 1m 45s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/8/check/check-tempest-devstack-vm-full/2f4784a : SUCCESS in 34m 38s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/8/check/check-tempest-devstack-vm-postgres-full/d81f1f0 : SUCCESS in 33m 31s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/8/check/check-tempest-devstack-vm-neutron/67e6cdc : FAILURE in 32m 13s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/8/check/gate-tempest-devstack-vm-large-ops/669be85 : SUCCESS in 13m 56s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/8/check/gate-tempest-devstack-vm-neutron-large-ops/2abfbdf : FAILURE in 19m 48s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/8/check/check-grenade-devstack-vm/bed531a : SUCCESS in 39m 27s\n","accounts_in_message":[],"_revision_number":8},{"id":"cec1e2d53a62f7fcef24372fe2fd04d10e1846b7","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-11 07:56:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6c887d612c07acd4c7a1653221746eb3bcd63c46","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-11 08:18:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9800254419acce6072ca454372a1c56bd659abe2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-11 08:52:35.000000000","message":"Patch Set 10: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/10/check/gate-tempest-pep8/e3b1e1b : FAILURE in 1m 44s\n- gate-tempest-docs http://logs.openstack.org/01/55101/10/check/gate-tempest-docs/8020aa8 : SUCCESS in 1m 17s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/10/check/gate-tempest-python27/6650b56 : FAILURE in 1m 05s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/10/check/check-tempest-devstack-vm-full/9d7d064 : SUCCESS in 31m 26s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/10/check/check-tempest-devstack-vm-postgres-full/80d5920 : SUCCESS in 33m 59s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/10/check/check-tempest-devstack-vm-neutron/640ae60 : SUCCESS in 25m 39s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/10/check/gate-tempest-devstack-vm-large-ops/cdd1deb : SUCCESS in 13m 37s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/10/check/gate-tempest-devstack-vm-neutron-large-ops/7ae78c2 : FAILURE in 19m 16s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/10/check/check-grenade-devstack-vm/8967c9b : SUCCESS in 33m 47s\n","accounts_in_message":[],"_revision_number":10},{"id":"f02cf3834f7d90a1e6a0122ede43e038ee8870d6","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-11 11:17:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f38837a2eeb7f73814b167c9100f7ba3c62e38d6","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-11-11 11:18:00.000000000","message":"Patch Set 11:\n\nNew patchset patch-id matches previous patchset, but whitespace content has changed.","accounts_in_message":[],"_revision_number":11},{"id":"e7cc41da3084b71dc7fc10b7f27f2946c6415ba4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-11 11:53:10.000000000","message":"Patch Set 11: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/11/check/gate-tempest-pep8/3d59216 : SUCCESS in 1m 02s\n- gate-tempest-docs http://logs.openstack.org/01/55101/11/check/gate-tempest-docs/a751b61 : SUCCESS in 1m 20s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/11/check/gate-tempest-python27/2c20e09 : SUCCESS in 1m 05s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/11/check/check-tempest-devstack-vm-full/469a20e : SUCCESS in 34m 29s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/11/check/check-tempest-devstack-vm-postgres-full/f20c671 : SUCCESS in 35m 07s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/11/check/check-tempest-devstack-vm-neutron/1007532 : SUCCESS in 31m 48s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/11/check/gate-tempest-devstack-vm-large-ops/e72d360 : SUCCESS in 14m 21s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/11/check/gate-tempest-devstack-vm-neutron-large-ops/69093df : FAILURE in 19m 05s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/11/check/check-grenade-devstack-vm/61fd182 : SUCCESS in 33m 14s\n","accounts_in_message":[],"_revision_number":11},{"id":"2427d789b9f7ede1f2db77ffc1052888fc380bd4","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2013-11-11 12:46:26.000000000","message":"Patch Set 11: I would prefer that you didn\u0027t merge this\n\n(11 inline comments)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"e6636c24a7fcf0b9f14a6a92b3f56a5da61798c5","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-11 13:41:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"738a7549562eb1c9051bafa7d80ce34f40cd6477","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-11 14:14:44.000000000","message":"Patch Set 12: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/12/check/gate-tempest-pep8/7822d49 : SUCCESS in 1m 25s\n- gate-tempest-docs http://logs.openstack.org/01/55101/12/check/gate-tempest-docs/96f9414 : SUCCESS in 1m 17s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/12/check/gate-tempest-python27/465b1ae : SUCCESS in 1m 18s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/12/check/check-tempest-devstack-vm-full/f36ead6 : SUCCESS in 32m 51s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/12/check/check-tempest-devstack-vm-postgres-full/e895e5d : SUCCESS in 32m 28s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/12/check/check-tempest-devstack-vm-neutron/5609b2b : SUCCESS in 27m 10s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/12/check/gate-tempest-devstack-vm-large-ops/5dc1044 : SUCCESS in 13m 48s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/12/check/gate-tempest-devstack-vm-neutron-large-ops/aa2b4af : FAILURE in 20m 22s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/12/check/check-grenade-devstack-vm/9a96fd7 : SUCCESS in 33m 26s\n","accounts_in_message":[],"_revision_number":12},{"id":"8470ba5fd5bb3c6449308d9ab4ff0e754f469e69","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-11 15:43:18.000000000","message":"Patch Set 11: (7 inline comments)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"cfc0c9a3f9deeef790893fa7fe6c7b110507a58c","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-17 06:42:35.000000000","message":"Patch Set 12: Work In Progress\n\nping blocked checks fail on Jenkins, but works fine on my rhos setup. need to figure out this issue","accounts_in_message":[],"_revision_number":12},{"id":"706bb45c2d4a1a15b6014d3f17ccffe7a74ba79d","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-17 14:50:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b64e51f005204111ef896538741101224033ee6d","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-17 15:15:44.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d22bfde6032dface4edaec260bfb46b0f14d88f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-17 15:52:42.000000000","message":"Patch Set 14: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/14/check/gate-tempest-pep8/160a23e : SUCCESS in 1m 44s\n- gate-tempest-docs http://logs.openstack.org/01/55101/14/check/gate-tempest-docs/57c9427 : SUCCESS in 1m 50s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/14/check/gate-tempest-python27/a074a86 : SUCCESS in 1m 22s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/14/check/check-tempest-devstack-vm-full/c58de5a : SUCCESS in 33m 50s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/14/check/check-tempest-devstack-vm-postgres-full/355d3a5 : SUCCESS in 36m 42s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/14/check/check-tempest-devstack-vm-neutron/62c502a : SUCCESS in 28m 49s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/14/check/gate-tempest-devstack-vm-large-ops/a445257 : SUCCESS in 14m 32s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/14/check/gate-tempest-devstack-vm-neutron-large-ops/27fe562 : SUCCESS in 15m 46s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/14/check/check-grenade-devstack-vm/92513d9 : SUCCESS in 25m 05s\n","accounts_in_message":[],"_revision_number":14},{"id":"092d2694215ebda5ab0469ea0f79bac604db0f21","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2013-11-18 11:54:21.000000000","message":"Patch Set 14: I would prefer that you didn\u0027t merge this\n\n(4 inline comments)\n\nIt seems my several comments were ignored.\nCould you check them again?","accounts_in_message":[],"_revision_number":14},{"id":"cd63c6c515be80bfe82cebcfe3cf096c465e59d5","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-18 12:07:12.000000000","message":"Patch Set 14: Work In Progress\n\nno-connectivity (block) tests fail upstream (work on local setup) so they are commented","accounts_in_message":[],"_revision_number":14},{"id":"c7219e55055ef5ee66c9b96f2ac1dc03e1715de1","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-18 12:09:05.000000000","message":"Patch Set 14: (1 inline comment)\n\ncommented \"block tests\" are WIP. sorry for not tagging it","accounts_in_message":[],"_revision_number":14},{"id":"dbc29160320d7309a23b38c9a636e013a0a47803","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2013-11-18 12:12:04.000000000","message":"Patch Set 14: No score\n\nOK. change to No score.","accounts_in_message":[],"_revision_number":14},{"id":"4968faeeeb787dd851ab4253324edd2e35fea1ef","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-20 12:32:57.000000000","message":"Patch Set 11: (3 inline comments)\n\n","accounts_in_message":[],"_revision_number":11},{"id":"7a89ee060cde30705fb0e3f4a9affdc045645519","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-20 12:51:13.000000000","message":"Patch Set 14: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":14},{"id":"3535a865986f3214f7568e827cf9cd8b995daffb","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-20 13:27:42.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"7d0352488a5c33fc50df0552f592a05bae44dc21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-20 15:12:48.000000000","message":"Patch Set 15: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/15/check/gate-tempest-pep8/d2d0da9 : SUCCESS in 3m 23s\n- gate-tempest-docs http://logs.openstack.org/01/55101/15/check/gate-tempest-docs/dc048e0 : SUCCESS in 3m 36s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/15/check/gate-tempest-python27/661c4af : SUCCESS in 3m 56s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/15/check/check-tempest-devstack-vm-full/ff01431 : FAILURE in 1h 19m 41s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/15/check/check-tempest-devstack-vm-postgres-full/bc01563 : SUCCESS in 1h 01m 04s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/15/check/check-tempest-devstack-vm-neutron/7765e69 : SUCCESS in 1h 04m 07s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/15/check/gate-tempest-devstack-vm-large-ops/98ab7df : SUCCESS in 32m 46s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/15/check/gate-tempest-devstack-vm-neutron-large-ops/5dd0cfe : SUCCESS in 42m 34s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/15/check/check-grenade-devstack-vm/743eb12 : FAILURE in 1h 00m 20s\n","accounts_in_message":[],"_revision_number":15},{"id":"e92a0d4f988ef9aadd70d49fbcd66574e6d48310","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-21 09:51:16.000000000","message":"Patch Set 15:\n\nrecheck bug 1240175","accounts_in_message":[],"_revision_number":15},{"id":"2f4e27e25b2ad3d8042a8db974b0bcf2fc91bb6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-21 11:24:32.000000000","message":"Patch Set 15: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/15/check/gate-tempest-pep8/ec9b027 : SUCCESS in 3m 20s\n- gate-tempest-docs http://logs.openstack.org/01/55101/15/check/gate-tempest-docs/1ae8c0b : SUCCESS in 3m 13s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/15/check/gate-tempest-python27/914b51b : SUCCESS in 2m 46s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/15/check/check-tempest-devstack-vm-full/bd8c256 : SUCCESS in 1h 03m 48s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/15/check/check-tempest-devstack-vm-postgres-full/4e50235 : SUCCESS in 1h 02m 48s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/15/check/check-tempest-devstack-vm-neutron/6f7561a : SUCCESS in 1h 06m 30s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/15/check/gate-tempest-devstack-vm-large-ops/01f41c6 : SUCCESS in 32m 47s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/15/check/gate-tempest-devstack-vm-neutron-large-ops/e129aa6 : SUCCESS in 30m 24s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/15/check/check-grenade-devstack-vm/e4fe680 : SUCCESS in 1h 01m 13s\n","accounts_in_message":[],"_revision_number":15},{"id":"49fdbdeeae0f1549a08412a7a2d7a0a615e31a51","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-24 07:53:08.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"95c9c0e8175e98f5a4c037f599119b9f4717071e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-24 08:24:54.000000000","message":"Patch Set 16: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/16/check/gate-tempest-pep8/0441edf : SUCCESS in 1m 24s\n- gate-tempest-docs http://logs.openstack.org/01/55101/16/check/gate-tempest-docs/2962617 : SUCCESS in 1m 36s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/16/check/gate-tempest-python27/91209b2 : SUCCESS in 1m 27s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/16/check/check-tempest-devstack-vm-full/b562dec : FAILURE in 29m 11s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/16/check/check-tempest-devstack-vm-postgres-full/ecb3b03 : FAILURE in 31m 41s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/16/check/check-tempest-devstack-vm-neutron/08259ab : FAILURE in 28m 56s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/16/check/gate-tempest-devstack-vm-large-ops/ae46b03 : SUCCESS in 16m 34s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/16/check/gate-tempest-devstack-vm-neutron-large-ops/2512278 : SUCCESS in 15m 13s\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/16/check/check-grenade-devstack-vm/a231931 : FAILURE in 21m 29s\n","accounts_in_message":[],"_revision_number":16},{"id":"c0c279f633bf439d2930a6e89ff4749c58d1f1c2","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-24 09:02:21.000000000","message":"Patch Set 16: Rebased","accounts_in_message":[],"_revision_number":16},{"id":"045cb4482172ea8437aea274d3f145ef523d1f61","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-24 09:46:06.000000000","message":"Patch Set 17: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/17/check/gate-tempest-pep8/3effab2 : FAILURE in 1m 44s\n- gate-tempest-docs http://logs.openstack.org/01/55101/17/check/gate-tempest-docs/aa2c277 : SUCCESS in 1m 03s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/17/check/gate-tempest-python27/4ac57d2 : SUCCESS in 1m 25s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/17/check/check-tempest-devstack-vm-full/f1ca47e : SUCCESS in 43m 41s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/17/check/check-tempest-devstack-vm-postgres-full/dbf2052 : SUCCESS in 31m 08s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/17/check/check-tempest-devstack-vm-neutron/ddfe52c : SUCCESS in 25m 57s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/17/check/gate-tempest-devstack-vm-large-ops/95a362e : SUCCESS in 13m 23s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/17/check/gate-tempest-devstack-vm-neutron-large-ops/b8f25c3 : SUCCESS in 16m 11s\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/17/check/check-grenade-devstack-vm/7b81d0a : SUCCESS in 24m 48s\n","accounts_in_message":[],"_revision_number":17},{"id":"e07c5772a12c3a5092685bcae2c1929ee32d0687","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-24 09:50:44.000000000","message":"Patch Set 17: Rebased","accounts_in_message":[],"_revision_number":17},{"id":"f47e9eb99d8a7992a50cac5db71f710f64db934b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-24 10:26:54.000000000","message":"Patch Set 18: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/18/check/gate-tempest-pep8/cf5d94e : SUCCESS in 1m 25s\n- gate-tempest-docs http://logs.openstack.org/01/55101/18/check/gate-tempest-docs/799e6b2 : SUCCESS in 1m 18s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/18/check/gate-tempest-python27/84547e6 : SUCCESS in 1m 47s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/18/check/check-tempest-devstack-vm-full/ac76fd8 : SUCCESS in 26m 57s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/18/check/check-tempest-devstack-vm-postgres-full/9cdf7db : SUCCESS in 29m 51s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/18/check/check-tempest-devstack-vm-neutron/37c73ae : SUCCESS in 30m 55s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/18/check/gate-tempest-devstack-vm-large-ops/7ec6037 : SUCCESS in 14m 37s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/18/check/gate-tempest-devstack-vm-neutron-large-ops/cf8ac37 : SUCCESS in 15m 36s\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/18/check/check-grenade-devstack-vm/f8e3435 : SUCCESS in 36m 06s\n","accounts_in_message":[],"_revision_number":18},{"id":"45af724bf5de582ea3f0b119c53e5a413f091060","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-27 07:21:54.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5d329ecd4ab8479526e247a59b7ac5e75e4830a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-27 08:11:49.000000000","message":"Patch Set 19: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/19/check/gate-tempest-pep8/467ffcd : SUCCESS in 1m 31s\n- gate-tempest-docs http://logs.openstack.org/01/55101/19/check/gate-tempest-docs/78a1911 : SUCCESS in 1m 22s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/19/check/gate-tempest-python27/c51bdc6 : SUCCESS in 1m 26s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/19/check/check-tempest-devstack-vm-full/42ce504 : SUCCESS in 44m 42s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/19/check/check-tempest-devstack-vm-postgres-full/2aa8281 : FAILURE in 49m 46s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/19/check/check-tempest-devstack-vm-neutron/74bfa55 : SUCCESS in 39m 28s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/19/check/gate-tempest-devstack-vm-large-ops/91f5679 : SUCCESS in 16m 13s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/19/check/gate-tempest-devstack-vm-neutron-large-ops/88a2092 : SUCCESS in 16m 43s\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/19/check/check-grenade-devstack-vm/64b299e : FAILURE in 13m 23s\n","accounts_in_message":[],"_revision_number":19},{"id":"2466c48cbe583b5bc618e0bbc0e0e329b802bea7","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-27 09:37:29.000000000","message":"Patch Set 19:\n\nrecheck bug 1255292","accounts_in_message":[],"_revision_number":19},{"id":"32e81d529810bc8e9d71d77e3f693765f16c3b77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-27 10:32:52.000000000","message":"Patch Set 19: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/19/check/gate-tempest-pep8/acbf35d : SUCCESS in 1m 29s\n- gate-tempest-docs http://logs.openstack.org/01/55101/19/check/gate-tempest-docs/579e3ce : SUCCESS in 2m 13s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/19/check/gate-tempest-python27/40ff2e9 : SUCCESS in 1m 21s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/19/check/check-tempest-devstack-vm-full/1e96aa4 : SUCCESS in 55m 14s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/19/check/check-tempest-devstack-vm-postgres-full/b42d5a3 : SUCCESS in 31m 26s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/19/check/check-tempest-devstack-vm-neutron/dd149a9 : SUCCESS in 48m 11s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/19/check/gate-tempest-devstack-vm-large-ops/cdf38e4 : SUCCESS in 15m 23s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/19/check/gate-tempest-devstack-vm-neutron-large-ops/eb7c46a : SUCCESS in 15m 26s\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/19/check/check-grenade-devstack-vm/cce7670 : SUCCESS in 34m 14s\n","accounts_in_message":[],"_revision_number":19},{"id":"f10d4c075455a12528d985390df7d674497bca69","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-11-28 08:01:58.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ae846b52f3ac51e47d65d986c4b5d137491a2c6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-28 08:38:34.000000000","message":"Patch Set 20: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/20/check/gate-tempest-pep8/a153478 : SUCCESS in 2m 00s\n- gate-tempest-docs http://logs.openstack.org/01/55101/20/check/gate-tempest-docs/cfdbefe : SUCCESS in 1m 29s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/20/check/gate-tempest-python27/216cff8 : SUCCESS in 1m 58s\n- check-tempest-devstack-vm-full http://logs.openstack.org/01/55101/20/check/check-tempest-devstack-vm-full/9940680 : SUCCESS in 30m 53s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/01/55101/20/check/check-tempest-devstack-vm-postgres-full/70ff61b : SUCCESS in 36m 28s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/01/55101/20/check/check-tempest-devstack-vm-neutron/4114042 : SUCCESS in 28m 20s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/01/55101/20/check/gate-tempest-devstack-vm-large-ops/530dfc5 : SUCCESS in 12m 59s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/01/55101/20/check/gate-tempest-devstack-vm-neutron-large-ops/33f241b : SUCCESS in 15m 42s\n- check-grenade-devstack-vm http://logs.openstack.org/01/55101/20/check/check-grenade-devstack-vm/8cd0bb0 : SUCCESS in 22m 22s\n","accounts_in_message":[],"_revision_number":20},{"id":"440d3421efa8cfc576471b3cbc5bfa299c0c5d1f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2013-12-03 02:19:20.000000000","message":"Patch Set 20: (1 inline comment)\n\nAlmost good to me. I have one question in line.","accounts_in_message":[],"_revision_number":20},{"id":"d2133e9a56993dc297c38304122bd0773a75fce4","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-03 06:17:36.000000000","message":"Patch Set 20: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":20},{"id":"705a0c793a8d4be8aedc60b1d76f9ba703b64c66","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2013-12-03 18:25:25.000000000","message":"Patch Set 20: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":20},{"id":"bb286c5336ba90a498f718b399af7d674d99202f","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-03 18:42:51.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"992493294f284d3987c667277e523dce486ed79f","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-03 18:43:02.000000000","message":"Patch Set 20: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":20},{"id":"89cb99dbed9e40366e011598e962fd99f7bd24ce","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-03 18:44:16.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"8da585655d7add04ca765fc9177cfaa023abe5b0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-03 19:26:51.000000000","message":"Patch Set 22: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/22/check/gate-tempest-pep8/e16be27 : SUCCESS in 1m 31s\n- gate-tempest-docs http://logs.openstack.org/01/55101/22/check/gate-tempest-docs/9ae9fec : SUCCESS in 1m 20s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/22/check/gate-tempest-python27/aaa8de3 : SUCCESS in 1m 28s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/22/check/check-tempest-dsvm-full/47de622 : SUCCESS in 40m 34s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/22/check/check-tempest-dsvm-postgres-full/c4a7e9c : SUCCESS in 35m 38s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/22/check/check-tempest-dsvm-neutron/4bfca9e : SUCCESS in 27m 10s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/22/check/gate-tempest-dsvm-large-ops/2dd641c : SUCCESS in 15m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/22/check/gate-tempest-dsvm-neutron-large-ops/cdbdd76 : SUCCESS in 17m 23s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/22/check/check-grenade-dsvm/798e891 : SUCCESS in 29m 35s\n","accounts_in_message":[],"_revision_number":22},{"id":"ead3084372b27da46c8d68984927cfea60b04a31","author":{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},"date":"2013-12-04 11:05:10.000000000","message":"Patch Set 22: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nvery good works! only have a very soft -1. :-)","accounts_in_message":[],"_revision_number":22},{"id":"4d7edc3d7bf8c24cc7350066bf83ab18780df0bc","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-04 11:16:26.000000000","message":"Patch Set 22: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":22},{"id":"eca83bc7ab4979b1a91e29922e276b04b171a10a","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-04 11:17:22.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"ce19ef5a851356d271fd58881a50eb727e64050a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-04 12:31:27.000000000","message":"Patch Set 23: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/23/check/gate-tempest-pep8/997d2de : SUCCESS in 1m 13s\n- gate-tempest-docs http://logs.openstack.org/01/55101/23/check/gate-tempest-docs/86833dd : SUCCESS in 1m 32s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/23/check/gate-tempest-python27/7c39ae7 : SUCCESS in 1m 20s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/23/check/check-tempest-dsvm-full/1ccf95b : SUCCESS in 1h 13m 58s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/23/check/check-tempest-dsvm-postgres-full/318a584 : FAILURE in 34m 52s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/23/check/check-tempest-dsvm-neutron/d8edd1d : FAILURE in 34m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/23/check/gate-tempest-dsvm-large-ops/b4b044f : SUCCESS in 16m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/23/check/gate-tempest-dsvm-neutron-large-ops/e2c12a0 : SUCCESS in 15m 17s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/23/check/check-grenade-dsvm/9a85c2f : SUCCESS in 21m 58s\n","accounts_in_message":[],"_revision_number":23},{"id":"b4890464d15085ee96db1372b3fc59215ec03bce","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-04 13:13:43.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"4385d69efd21f290967c891778455798493c7d51","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-04 14:02:07.000000000","message":"Patch Set 24: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/24/check/gate-tempest-pep8/1d6742c : SUCCESS in 1m 44s\n- gate-tempest-docs http://logs.openstack.org/01/55101/24/check/gate-tempest-docs/b707ad0 : SUCCESS in 1m 21s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/24/check/gate-tempest-python27/1cfe483 : SUCCESS in 1m 20s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/24/check/check-tempest-dsvm-full/59e9570 : SUCCESS in 33m 49s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/24/check/check-tempest-dsvm-postgres-full/08e27d1 : SUCCESS in 48m 17s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/24/check/check-tempest-dsvm-neutron/051d9b3 : SUCCESS in 34m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/24/check/gate-tempest-dsvm-large-ops/baaacaa : SUCCESS in 17m 50s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/24/check/gate-tempest-dsvm-neutron-large-ops/c7b068a : SUCCESS in 15m 15s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/24/check/check-grenade-dsvm/3e6f0bc : SUCCESS in 32m 35s\n","accounts_in_message":[],"_revision_number":24},{"id":"94676280745c8b4d068fd7117ba7886414857400","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-12-07 13:25:39.000000000","message":"Patch Set 24: I would prefer that you didn\u0027t merge this\n\nthis is currently skipped in the gate. We need to figure out how to get it runnable in the gate before landing so we know it\u0027s a stable test.","accounts_in_message":[],"_revision_number":24},{"id":"b51d058f85f218229b594fe00541e53b9ff0255b","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-07 18:27:23.000000000","message":"Patch Set 24:\n\nWhy skipped? the \"negative\" tests are skipped untill bug is fixed. positive tests still work","accounts_in_message":[],"_revision_number":24},{"id":"d4f1d426e33916fdbf38a3d57e0b8aec5b32045d","author":{"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},"date":"2013-12-11 09:57:59.000000000","message":"Patch Set 24: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":24},{"id":"7edea5f02aefb9f9b8a5dcf1b9afa798f84f188c","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-11 18:13:00.000000000","message":"Patch Set 24: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":24},{"id":"43e3b3a9a60c0fd944223f6ac2bcff5dc1628bc6","author":{"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},"date":"2013-12-12 10:39:16.000000000","message":"Patch Set 24: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":24},{"id":"c50185609c202d1b59ff9392fc517709b376505d","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2013-12-16 06:22:02.000000000","message":"Patch Set 24:\n\n@Yair: I just want to clarify. I think this (positive scenario) test case is not skipped in the gate, right?","accounts_in_message":[],"_revision_number":24},{"id":"cfc88e4d2a929fc9ed7f59f2c59a4665c476e38f","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-16 07:30:03.000000000","message":"Patch Set 24:\n\n@Masayuki Igawa - Yes. Only the negative tests in this patch are skipped.","accounts_in_message":[],"_revision_number":24},{"id":"0f31c7a9d9d3eadb93035d694264ec72e53b8292","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2013-12-16 07:54:32.000000000","message":"Patch Set 24:\n\n@Yair: Thank you for your reply. But I can\u0027t find the negative test in this patch. Where is it?","accounts_in_message":[],"_revision_number":24},{"id":"4ac1afad56d219e2ab8e5f0f7ba66057846757ce","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-16 09:39:58.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"9be93db5ddd15ae4591151cc8fcc5ac6a077b96b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-16 10:17:49.000000000","message":"Patch Set 25: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/25/check/gate-tempest-pep8/db6162b : SUCCESS in 2m 06s\n- gate-tempest-docs http://logs.openstack.org/01/55101/25/check/gate-tempest-docs/7aa23f0 : SUCCESS in 1m 27s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/25/check/gate-tempest-python27/d87b146 : SUCCESS in 1m 24s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/25/check/check-tempest-dsvm-full/0123233 : SUCCESS in 37m 07s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/25/check/check-tempest-dsvm-postgres-full/3bacbfd : SUCCESS in 35m 55s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/25/check/check-tempest-dsvm-neutron/9a60ad1 : FAILURE in 29m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/25/check/gate-tempest-dsvm-large-ops/e6da293 : SUCCESS in 13m 41s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/25/check/gate-tempest-dsvm-neutron-large-ops/7bdb69b : SUCCESS in 14m 54s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/25/check/check-grenade-dsvm/dd17e0d : SUCCESS in 36m 41s\n","accounts_in_message":[],"_revision_number":25},{"id":"c4faf0657dd7ebc1455552704b7336984454d88c","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-16 11:22:53.000000000","message":"Patch Set 25: Rebased","accounts_in_message":[],"_revision_number":25},{"id":"d25fcdb5c3718376be1f44365b5865396ae8747c","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-16 11:54:09.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"9d1f9a11104bbb689f4be947fd98dd3917d180cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-16 12:40:01.000000000","message":"Patch Set 27: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/27/check/gate-tempest-pep8/dca400b : FAILURE in 1m 55s\n- gate-tempest-docs http://logs.openstack.org/01/55101/27/check/gate-tempest-docs/ed72689 : SUCCESS in 1m 44s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/27/check/gate-tempest-python27/b7a82be : SUCCESS in 1m 27s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/27/check/check-tempest-dsvm-full/a313fc9 : SUCCESS in 43m 47s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/27/check/check-tempest-dsvm-postgres-full/521c629 : SUCCESS in 45m 40s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/27/check/check-tempest-dsvm-neutron/20801d4 : FAILURE in 35m 32s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/27/check/gate-tempest-dsvm-large-ops/13093f8 : SUCCESS in 13m 37s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/27/check/gate-tempest-dsvm-neutron-large-ops/a788989 : SUCCESS in 15m 24s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/27/check/check-grenade-dsvm/9697ac3 : SUCCESS in 31m 23s\n","accounts_in_message":[],"_revision_number":27},{"id":"c8e6fcb850bbd469a39c556ab644fc96c1ba9743","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-16 13:36:00.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"1f4fcee358304f435a0c71e54c1833166cf9c679","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-16 14:45:46.000000000","message":"Patch Set 28: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/28/check/gate-tempest-pep8/6d19a91 : SUCCESS in 1m 38s\n- gate-tempest-docs http://logs.openstack.org/01/55101/28/check/gate-tempest-docs/5f9864e : SUCCESS in 1m 49s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/28/check/gate-tempest-python27/e0dc61e : SUCCESS in 1m 29s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/28/check/check-tempest-dsvm-full/a0e43ba : SUCCESS in 1h 09m 37s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/28/check/check-tempest-dsvm-postgres-full/5b13926 : SUCCESS in 1h 08m 49s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/28/check/check-tempest-dsvm-neutron/0541dff : FAILURE in 50m 56s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/28/check/gate-tempest-dsvm-large-ops/7509f8a : SUCCESS in 14m 15s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/28/check/gate-tempest-dsvm-neutron-large-ops/fcb9757 : SUCCESS in 15m 14s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/28/check/check-grenade-dsvm/fabf4ba : SUCCESS in 43m 45s\n","accounts_in_message":[],"_revision_number":28},{"id":"5e5b9acaa855ad20dbcc5103e2e799d8eb4f6557","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-16 18:17:27.000000000","message":"Patch Set 28:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":28},{"id":"a2a3dc012c6955931bf9b775946843c1a157dedb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-16 19:22:09.000000000","message":"Patch Set 28:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/28/check/gate-tempest-pep8/a218caf : SUCCESS in 1m 10s\n- gate-tempest-docs http://logs.openstack.org/01/55101/28/check/gate-tempest-docs/18b59ec : SUCCESS in 1m 27s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/28/check/gate-tempest-python27/c158986 : SUCCESS in 1m 16s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/28/check/check-tempest-dsvm-full/287bebb : SUCCESS in 53m 14s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/28/check/check-tempest-dsvm-postgres-full/e5b4b86 : SUCCESS in 56m 37s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/28/check/check-tempest-dsvm-neutron/f383f4b : FAILURE in 49m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/28/check/gate-tempest-dsvm-large-ops/9e59a8b : SUCCESS in 17m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/28/check/gate-tempest-dsvm-neutron-large-ops/ea50f2d : SUCCESS in 19m 34s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/28/check/check-grenade-dsvm/c6e5c45 : SUCCESS in 41m 13s\n","accounts_in_message":[],"_revision_number":28},{"id":"faf50da574714798a7617054842879bf7fac7af2","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2013-12-16 20:34:33.000000000","message":"Patch Set 28: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nIt would seem that map you are handing to the FloatingIPCheckTracker has the key/values reversed in their role.","accounts_in_message":[],"_revision_number":28},{"id":"90e86bb488ef9ded14e2d5e52f12f6cd52c70b49","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2013-12-16 22:19:10.000000000","message":"Patch Set 28: (1 inline comment)\n\nPlease disregard previous comments.","accounts_in_message":[],"_revision_number":28},{"id":"44ec854c149ac20373eb6bfdd848d13069c6d151","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2013-12-17 00:05:25.000000000","message":"Patch Set 28:\n\nI feel this patch may have a bug. Because tempest failed 3 times in the same place (_wait_for_floating_ip_association()).","accounts_in_message":[],"_revision_number":28},{"id":"32108d418f07b8fd01fdcf737852bb8f87858a44","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-17 06:04:29.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"2ae6d937368a876491b128a903c4ffc34f6ff266","author":{"_account_id":8622,"name":"Feng Ju","email":"jufeng@cn.ibm.com","username":"jufeng"},"date":"2013-12-17 06:11:19.000000000","message":"Patch Set 29: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":29},{"id":"51bbd1f6baad64f439e139b4bd2af2074fc1de55","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-17 07:09:24.000000000","message":"Patch Set 29: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/29/check/gate-tempest-pep8/37ed50c : SUCCESS in 2m 08s\n- gate-tempest-docs http://logs.openstack.org/01/55101/29/check/gate-tempest-docs/026febb : SUCCESS in 2m 08s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/29/check/gate-tempest-python27/b52466a : SUCCESS in 1m 19s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/29/check/check-tempest-dsvm-full/9c325e2 : SUCCESS in 43m 10s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/29/check/check-tempest-dsvm-postgres-full/87dbacd : SUCCESS in 44m 38s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/29/check/check-tempest-dsvm-neutron/ddb9bab : FAILURE in 1h 01m 44s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/29/check/gate-tempest-dsvm-large-ops/379cc20 : SUCCESS in 15m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/29/check/gate-tempest-dsvm-neutron-large-ops/102bcf2 : SUCCESS in 16m 44s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/29/check/check-grenade-dsvm/d5c830f : SUCCESS in 33m 10s\n","accounts_in_message":[],"_revision_number":29},{"id":"b3ce7e38c292fcb063c335617bf52ef7b51fa236","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-17 07:24:13.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"89c5aa4d7a628e9ca2a54470a489840540ca8b09","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-17 08:07:31.000000000","message":"Patch Set 30: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/30/check/gate-tempest-pep8/072f7ee : SUCCESS in 2m 03s\n- gate-tempest-docs http://logs.openstack.org/01/55101/30/check/gate-tempest-docs/fdea1e3 : SUCCESS in 1m 33s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/30/check/gate-tempest-python27/a03cf99 : SUCCESS in 1m 25s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/30/check/check-tempest-dsvm-full/6a91e75 : SUCCESS in 43m 07s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/30/check/check-tempest-dsvm-postgres-full/6d2cc0a : SUCCESS in 36m 39s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/30/check/check-tempest-dsvm-neutron/80b41cc : SUCCESS in 31m 33s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/30/check/gate-tempest-dsvm-large-ops/d41d9db : SUCCESS in 13m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/30/check/gate-tempest-dsvm-neutron-large-ops/cc0c69c : SUCCESS in 15m 50s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/30/check/check-grenade-dsvm/b177052 : SUCCESS in 30m 41s\n","accounts_in_message":[],"_revision_number":30},{"id":"50df06fd87cc100d77d881f33bbcfba5da187c7e","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-17 08:42:25.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"ecfdef3cb9cdd848af5a2daaa1e84815a5d899cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-17 09:19:59.000000000","message":"Patch Set 31: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/31/check/gate-tempest-pep8/3e914a7 : SUCCESS in 1m 37s\n- gate-tempest-docs http://logs.openstack.org/01/55101/31/check/gate-tempest-docs/3ec7aee : SUCCESS in 1m 51s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/31/check/gate-tempest-python27/dae06fe : SUCCESS in 1m 30s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/31/check/check-tempest-dsvm-full/2fcace9 : SUCCESS in 33m 34s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/31/check/check-tempest-dsvm-postgres-full/e825bb7 : FAILURE in 34m 07s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/31/check/check-tempest-dsvm-neutron/321413a : FAILURE in 37m 20s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/31/check/gate-tempest-dsvm-large-ops/d7e9511 : SUCCESS in 13m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/31/check/gate-tempest-dsvm-neutron-large-ops/5aab122 : SUCCESS in 23m 39s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/31/check/check-grenade-dsvm/b8ed359 : SUCCESS in 32m 49s\n","accounts_in_message":[],"_revision_number":31},{"id":"7dcb2d6bbd5b0959a13b594c0f985ed1797528a7","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-17 15:58:36.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"d11867c4aebadf05a1c66dd66fdaadf7fd03d139","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-17 16:14:51.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"b98dbb92d02cb72834c17753f6441e637ab33940","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-17 16:59:58.000000000","message":"Patch Set 33: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/33/check/gate-tempest-pep8/f49278f : SUCCESS in 2m 06s\n- gate-tempest-docs http://logs.openstack.org/01/55101/33/check/gate-tempest-docs/4f6685e : SUCCESS in 1m 49s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/33/check/gate-tempest-python27/e7fca47 : SUCCESS in 1m 48s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/33/check/check-tempest-dsvm-full/83e0a96 : SUCCESS in 44m 55s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/33/check/check-tempest-dsvm-postgres-full/11e12c2 : SUCCESS in 36m 47s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/33/check/check-tempest-dsvm-neutron/bc4eb2e : SUCCESS in 32m 22s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/33/check/gate-tempest-dsvm-large-ops/bd5487a : SUCCESS in 14m 37s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/33/check/gate-tempest-dsvm-neutron-large-ops/0f3fae9 : SUCCESS in 15m 37s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/33/check/check-grenade-dsvm/2b61c4d : SUCCESS in 29m 04s\n","accounts_in_message":[],"_revision_number":33},{"id":"c6711454c59ed0a3eb83328e1ec8bbeee7800430","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-12-17 17:28:14.000000000","message":"Patch Set 33: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":33},{"id":"23c1763e41a1b320815ae9cac7d17dabe73d9f12","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2013-12-17 19:14:18.000000000","message":"Patch Set 33: I would prefer that you didn\u0027t merge this\n\n(3 inline comments)\n\nPlease see in line comments","accounts_in_message":[],"_revision_number":33},{"id":"8ffe704dd5d73ea19aee997e232a33020fc173d1","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2013-12-17 21:10:50.000000000","message":"Patch Set 33: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":33},{"id":"13f2111f37d7f39b4fb4544d498b03ad8646c128","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-18 14:36:20.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"56e8af1e7433e136e21e88b2db489be69e30ba16","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-12-18 14:36:39.000000000","message":"Patch Set 34: Looks good to me (core reviewer)\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":34},{"id":"14f03caf9ce5c88a538bf4204c43c2bdd15c52ae","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2013-12-18 14:36:39.000000000","message":"Patch Set 34: Looks good to me, but someone else must approve\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":34},{"id":"58cf87667345931c14410bea02c796bb71512202","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-18 15:20:00.000000000","message":"Patch Set 34: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/34/check/gate-tempest-pep8/f38050a : SUCCESS in 1m 44s\n- gate-tempest-docs http://logs.openstack.org/01/55101/34/check/gate-tempest-docs/9419b8d : SUCCESS in 1m 26s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/34/check/gate-tempest-python27/6af428a : SUCCESS in 1m 12s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-full/eee5c5d : FAILURE in 41m 39s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-postgres-full/cfdcae3 : SUCCESS in 38m 25s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-neutron/feb1e58 : SUCCESS in 31m 35s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/34/check/gate-tempest-dsvm-large-ops/15dc9e4 : SUCCESS in 20m 50s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/34/check/gate-tempest-dsvm-neutron-large-ops/b9d1cf8 : SUCCESS in 16m 41s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/34/check/check-grenade-dsvm/1bfdb68 : SUCCESS in 43m 03s\n","accounts_in_message":[],"_revision_number":34},{"id":"ea1a9b7d26acfa03667749ea9f8643d35d387930","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-12-18 15:28:52.000000000","message":"Patch Set 34:\n\nrecheck bug 1261433","accounts_in_message":[],"_revision_number":34},{"id":"b887a6d29475fd84999d719f90551735effbef59","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-18 16:07:32.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/34/check/gate-tempest-pep8/ade075f : SUCCESS in 2m 01s\n- gate-tempest-docs http://logs.openstack.org/01/55101/34/check/gate-tempest-docs/1d250c5 : SUCCESS in 1m 11s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/34/check/gate-tempest-python27/58ef06d : SUCCESS in 1m 07s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-full/74cc41d : SUCCESS in 38m 30s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-postgres-full/8e6d69c : SUCCESS in 36m 12s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-neutron/7bb7612 : FAILURE in 33m 38s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/34/check/gate-tempest-dsvm-large-ops/9578c03 : SUCCESS in 13m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/34/check/gate-tempest-dsvm-neutron-large-ops/379a6c7 : SUCCESS in 16m 42s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/34/check/check-grenade-dsvm/525c36b : SUCCESS in 30m 18s\n","accounts_in_message":[],"_revision_number":34},{"id":"7d75847c0e81b0704a71a081b6d57aa5107fac29","author":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"date":"2013-12-18 17:05:36.000000000","message":"Patch Set 34:\n\nrecheck bug 1261433","accounts_in_message":[],"_revision_number":34},{"id":"af55d4ead03e62e73c7d3407879d9e154436789b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-18 17:57:41.000000000","message":"Patch Set 34: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/34/check/gate-tempest-pep8/e3ef80b : SUCCESS in 2m 08s\n- gate-tempest-docs http://logs.openstack.org/01/55101/34/check/gate-tempest-docs/99258e8 : SUCCESS in 1m 45s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/34/check/gate-tempest-python27/a127995 : SUCCESS in 1m 30s\n- check-tempest-dsvm-full http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-full/01b3d91 : SUCCESS in 46m 51s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-postgres-full/7592554 : SUCCESS in 51m 34s\n- check-tempest-dsvm-neutron http://logs.openstack.org/01/55101/34/check/check-tempest-dsvm-neutron/3b528ee : SUCCESS in 35m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/34/check/gate-tempest-dsvm-large-ops/3c1415b : SUCCESS in 14m 48s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/34/check/gate-tempest-dsvm-neutron-large-ops/134fa3e : SUCCESS in 17m 32s\n- check-grenade-dsvm http://logs.openstack.org/01/55101/34/check/check-grenade-dsvm/cc9583a : SUCCESS in 36m 40s\n","accounts_in_message":[],"_revision_number":34},{"id":"43f757facd73394692270dd159cd464b72289155","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2013-12-18 22:12:20.000000000","message":"Patch Set 34: Approved\n\nThanks, Yair!","accounts_in_message":[],"_revision_number":34},{"id":"75490a6346cafc6d91517ad82dac9355e87eb868","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-18 22:12:22.000000000","message":"Patch Set 34: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":34},{"id":"8cd65268c99b5ef0062394d8cb55835ea902dfa5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-18 23:30:51.000000000","message":"Patch Set 34: Verified\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/01/55101/34/gate/gate-tempest-pep8/09309c3 : SUCCESS in 1m 40s\n- gate-tempest-docs http://logs.openstack.org/01/55101/34/gate/gate-tempest-docs/3a82329 : SUCCESS in 1m 26s\n- gate-tempest-python27 http://logs.openstack.org/01/55101/34/gate/gate-tempest-python27/cd96e21 : SUCCESS in 1m 31s\n- gate-tempest-dsvm-full http://logs.openstack.org/01/55101/34/gate/gate-tempest-dsvm-full/3babd7d : SUCCESS in 39m 05s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/01/55101/34/gate/gate-tempest-dsvm-postgres-full/0f8cc2c : SUCCESS in 48m 54s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/01/55101/34/gate/gate-tempest-dsvm-neutron/221e3ec : SUCCESS in 33m 00s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/01/55101/34/gate/gate-tempest-dsvm-large-ops/6aa4d66 : SUCCESS in 15m 56s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/01/55101/34/gate/gate-tempest-dsvm-neutron-large-ops/eb6d24f : SUCCESS in 16m 45s\n- gate-grenade-dsvm http://logs.openstack.org/01/55101/34/gate/gate-grenade-dsvm/4947abe : SUCCESS in 38m 11s\n","accounts_in_message":[],"_revision_number":34},{"id":"c8f17ee51cd2efe9691f865b97e85c99187eb606","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-18 23:30:52.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":34}],"current_revision_number":34,"current_revision":"4d7efa6d030bca335c7fb9f45e53650ef2ceb562","revisions":{"5dc003fa7693a0557cf9667743f8643b675ae0a1":{"kind":"REWORK","_number":1,"created":"2013-11-03 21:00:55.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/1"}}},"commit":{"parents":[{"commit":"964a4b8e1b163aa34003fef93186fb371aafdc5c","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/964a4b8e1b163aa34003fef93186fb371aafdc5c"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 21:00:22.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\nThis test suite assumes that Nova has been configured to boot VM\u0027s with\nNeutron-managed networking, and attempts to verify cross tenant connectivity\nas follows\n\nssh:\n    in order to overcome \"ip namespace\", each tenant has an \"access point\"  VM\n    with floating-ip open to incoming ssh connection allowing network commands\n    (ping/ssh) to be executed from within the tenant-network-namespace\n    Tempest host performs key-based authentication to the ssh server via\n    floating IP address\n\nsetup:\n    for each tenant (demo and alt):\n        1. create a network\u0026subnet\n        2. create a router (if public router isn\u0027t configured)\n        3. connect tenant network to public network via router\n        4. create an access point:\n            a. a security group open to incoming ssh connection\n            b. a VM with a floating ip\n        5. for demo tenant - create another server to test in-tenant connections\n\nconnectivity test is done by pinging destination server via source server ssh\nconnection.\nsuccess - ping returns\nfailure - ping_timeout reached\n\ntests:\n        1. _verify_network_details\n        2. _verify_mac_addr: for each access point verify that\n        (subnet, fix_ip, mac address) are as defined in the port list\n        3. _test_in_tenant_block: test that in-tenant traffic is disabled\n        without rules allowing it\n        4. _test_in_tenant_allow: test that in-tenant traffic is enabled once an\n        appropriate rule has been created\n        5. _test_cross_tenant_block: test that cross-tenant traffic is disabled\n        without a rule allowing it on destination tenant\n        6. _test_cross_tenant_allow:\n            * test that cross-tenant traffic is enabled once an\n            * appropriate rule\n            has been created on destination tenant.\n            * test that reverse traffic is still blocked\n            * test than revesre traffic is enabled once an\n            * appropriate\n            * rule has been created on source tenant\n\nassumptions:\n        1. alt_tenant/user existed and is different from demo_tenant/user\n        2. Public network is defined and reachable from the Tempest host\n        3. Public router can either be:\n            * defined, in which case all tenants networks can\n            * connect directly\n            to it, and cross tenant check will be done on the private IP of the\n            destination tenant\n            or\n            * not defined (empty string), in which case each tanant\n            * will have\n            its own router connected to the public network\n\nhelper changes:\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5dc003fa7693a0557cf9667743f8643b675ae0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5dc003fa7693a0557cf9667743f8643b675ae0a1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"964a4b8e1b163aa34003fef93186fb371aafdc5c","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"cb06c3e4849f0b04c9d84b6ad65300f29b52fd3a":{"kind":"REWORK","_number":2,"created":"2013-11-04 07:59:18.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/2"}}},"commit":{"parents":[{"commit":"964a4b8e1b163aa34003fef93186fb371aafdc5c","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/964a4b8e1b163aa34003fef93186fb371aafdc5c"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-04 07:53:35.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cb06c3e4849f0b04c9d84b6ad65300f29b52fd3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cb06c3e4849f0b04c9d84b6ad65300f29b52fd3a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"964a4b8e1b163aa34003fef93186fb371aafdc5c","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"6b5645815bd0f6e540f14749eee0b427d13b98df":{"kind":"REWORK","_number":3,"created":"2013-11-04 11:16:44.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/3"}}},"commit":{"parents":[{"commit":"964a4b8e1b163aa34003fef93186fb371aafdc5c","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/964a4b8e1b163aa34003fef93186fb371aafdc5c"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-04 11:16:07.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6b5645815bd0f6e540f14749eee0b427d13b98df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6b5645815bd0f6e540f14749eee0b427d13b98df"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"964a4b8e1b163aa34003fef93186fb371aafdc5c","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"a774cedc1eb575b7c978c7ece57547b3796d94a1":{"kind":"REWORK","_number":4,"created":"2013-11-04 15:13:18.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/4"}}},"commit":{"parents":[{"commit":"964a4b8e1b163aa34003fef93186fb371aafdc5c","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/964a4b8e1b163aa34003fef93186fb371aafdc5c"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-04 15:12:42.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a774cedc1eb575b7c978c7ece57547b3796d94a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a774cedc1eb575b7c978c7ece57547b3796d94a1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"964a4b8e1b163aa34003fef93186fb371aafdc5c","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"095caa7768ac6df8f15efa523e9d753d25c1fd21":{"kind":"REWORK","_number":5,"created":"2013-11-05 10:22:42.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/5"}}},"commit":{"parents":[{"commit":"964a4b8e1b163aa34003fef93186fb371aafdc5c","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/964a4b8e1b163aa34003fef93186fb371aafdc5c"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-05 10:22:16.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\n***disabled in-tenant check for this patch only***\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/095caa7768ac6df8f15efa523e9d753d25c1fd21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/095caa7768ac6df8f15efa523e9d753d25c1fd21"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"964a4b8e1b163aa34003fef93186fb371aafdc5c","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"d50181d75dbc24b16968e0470ca609f6acdc5c89":{"kind":"REWORK","_number":6,"created":"2013-11-05 11:11:30.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/6"}}},"commit":{"parents":[{"commit":"964a4b8e1b163aa34003fef93186fb371aafdc5c","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/964a4b8e1b163aa34003fef93186fb371aafdc5c"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-05 11:10:44.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\n***disabled block check for this patch only***\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d50181d75dbc24b16968e0470ca609f6acdc5c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d50181d75dbc24b16968e0470ca609f6acdc5c89"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"964a4b8e1b163aa34003fef93186fb371aafdc5c","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"4b9b3f63627234fd805d870d5532efde1cedd910":{"kind":"REWORK","_number":7,"created":"2013-11-05 12:02:59.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/7"}}},"commit":{"parents":[{"commit":"964a4b8e1b163aa34003fef93186fb371aafdc5c","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/964a4b8e1b163aa34003fef93186fb371aafdc5c"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-05 12:02:14.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\n***disabled block check for this patch only***\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4b9b3f63627234fd805d870d5532efde1cedd910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4b9b3f63627234fd805d870d5532efde1cedd910"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"964a4b8e1b163aa34003fef93186fb371aafdc5c","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":9,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"0249e5362a6b14b716bf82a5922bb97e111a5c07":{"kind":"REWORK","_number":8,"created":"2013-11-11 06:59:42.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/8"}}},"commit":{"parents":[{"commit":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b1fc7356434fd8ac011810db150e0fb51ddb0fd9"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-11 06:57:51.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0249e5362a6b14b716bf82a5922bb97e111a5c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0249e5362a6b14b716bf82a5922bb97e111a5c07"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":10,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"2306161f120a77e5e1a09a0b48c9fb7e89a8013e":{"kind":"REWORK","_number":9,"created":"2013-11-11 07:56:22.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/9"}}},"commit":{"parents":[{"commit":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b1fc7356434fd8ac011810db150e0fb51ddb0fd9"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-11 07:55:35.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2306161f120a77e5e1a09a0b48c9fb7e89a8013e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2306161f120a77e5e1a09a0b48c9fb7e89a8013e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":10,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"37633f5e9b6ae32cc46656c34b8f5dab2e3d07c7":{"kind":"REWORK","_number":10,"created":"2013-11-11 08:18:30.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/10"}}},"commit":{"parents":[{"commit":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b1fc7356434fd8ac011810db150e0fb51ddb0fd9"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-11 08:18:03.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/37633f5e9b6ae32cc46656c34b8f5dab2e3d07c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/37633f5e9b6ae32cc46656c34b8f5dab2e3d07c7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":10,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"fe5b9ed087ab8c8addebd17b45fd52f031d0c76e":{"kind":"REWORK","_number":11,"created":"2013-11-11 11:17:53.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/11"}}},"commit":{"parents":[{"commit":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b1fc7356434fd8ac011810db150e0fb51ddb0fd9"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-11 11:16:41.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe5b9ed087ab8c8addebd17b45fd52f031d0c76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe5b9ed087ab8c8addebd17b45fd52f031d0c76e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":10,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"e8fb5a16e7f97e3bd39ebf1561df15c33db132e2":{"kind":"REWORK","_number":12,"created":"2013-11-11 13:41:10.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/12"}}},"commit":{"parents":[{"commit":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b1fc7356434fd8ac011810db150e0fb51ddb0fd9"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-11 13:35:05.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e8fb5a16e7f97e3bd39ebf1561df15c33db132e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e8fb5a16e7f97e3bd39ebf1561df15c33db132e2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1fc7356434fd8ac011810db150e0fb51ddb0fd9","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":10,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"460e00027b2570dc6ac841d7da5dbc476034d936":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2013-11-17 14:50:26.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/13"}}},"commit":{"parents":[{"commit":"6fbca6016029b503aaa4c1b7878f32fe8fc3ee3b","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fbca6016029b503aaa4c1b7878f32fe8fc3ee3b"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-03 19:53:18.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 14:39:52.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can create a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted neutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/460e00027b2570dc6ac841d7da5dbc476034d936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/460e00027b2570dc6ac841d7da5dbc476034d936"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6fbca6016029b503aaa4c1b7878f32fe8fc3ee3b","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":15,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"6a4470263bf19aacd52774b88419ce67cbf61199":{"kind":"REWORK","_number":14,"created":"2013-11-17 15:15:44.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/14"}}},"commit":{"parents":[{"commit":"5baa67a9b88e78183b16fc91c9359f250a9349c3","subject":"add generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5baa67a9b88e78183b16fc91c9359f250a9349c3"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different tenants\nfor each case test that default state blocks traffic, that security\nrules allow only specified traffic\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6a4470263bf19aacd52774b88419ce67cbf61199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6a4470263bf19aacd52774b88419ce67cbf61199"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5baa67a9b88e78183b16fc91c9359f250a9349c3","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":17,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"0e080e5209d32908af1620d19cfd9463030981af":{"kind":"REWORK","_number":15,"created":"2013-11-20 13:27:42.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/15"}}},"commit":{"parents":[{"commit":"e169288070610600ccb38a238deacb80c407b69f","subject":"adds generic method create_secgroup \u0026 rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e169288070610600ccb38a238deacb80c407b69f"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-20 13:22:46.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e080e5209d32908af1620d19cfd9463030981af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e080e5209d32908af1620d19cfd9463030981af"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e169288070610600ccb38a238deacb80c407b69f","is_merged_in_target_branch":false,"change_id":"I91ede0359ce6a8962f8bc3829c756106e8c51624","change_number":53305,"patch_set_number":19,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"1d16f5fbbef42837ebd921946f39c622102a1da7":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2013-11-24 07:53:08.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/16"}}},"commit":{"parents":[{"commit":"1da9913740846ebc5f9881a70f3c6fd6963f9bda","subject":"Support Neutron security groups in scenario testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1da9913740846ebc5f9881a70f3c6fd6963f9bda"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-24 07:52:28.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1d16f5fbbef42837ebd921946f39c622102a1da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1d16f5fbbef42837ebd921946f39c622102a1da7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1da9913740846ebc5f9881a70f3c6fd6963f9bda","is_merged_in_target_branch":false,"change_id":"I6de4bf9f7deed215cf61c87c3ae3f4240982523e","change_number":50892,"patch_set_number":14,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5c17f13781481000b7c8ce8651c92879e81165eb":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2013-11-24 07:53:08.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/17"}}},"commit":{"parents":[{"commit":"4c765502f05baab290417624ae632a7a059baabe","subject":"Support Neutron security groups in scenario testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4c765502f05baab290417624ae632a7a059baabe"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2013-11-24 09:02:21.000000000","tz":0},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c17f13781481000b7c8ce8651c92879e81165eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c17f13781481000b7c8ce8651c92879e81165eb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"4c765502f05baab290417624ae632a7a059baabe","is_merged_in_target_branch":false,"change_id":"I6de4bf9f7deed215cf61c87c3ae3f4240982523e","change_number":50892,"patch_set_number":15,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a5eae3b009b25827b37794d7cb4971e89e1bf449":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2013-11-24 07:53:08.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/18"}}},"commit":{"parents":[{"commit":"e34de7aa59ac78b2610dba8b950a8539abf938e6","subject":"Support Neutron security groups in scenario testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e34de7aa59ac78b2610dba8b950a8539abf938e6"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2013-11-24 09:50:44.000000000","tz":0},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a5eae3b009b25827b37794d7cb4971e89e1bf449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a5eae3b009b25827b37794d7cb4971e89e1bf449"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e34de7aa59ac78b2610dba8b950a8539abf938e6","is_merged_in_target_branch":false,"change_id":"I6de4bf9f7deed215cf61c87c3ae3f4240982523e","change_number":50892,"patch_set_number":16,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fa1a6dc43b0c6c2393c138f15620907a68cb2baa":{"kind":"REWORK","_number":19,"created":"2013-11-27 07:21:54.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/19"}}},"commit":{"parents":[{"commit":"c1af3ad93634625799e921d2a8aabbf37d1c835e","subject":"Support Neutron security groups in scenario testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c1af3ad93634625799e921d2a8aabbf37d1c835e"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-27 07:20:47.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fa1a6dc43b0c6c2393c138f15620907a68cb2baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fa1a6dc43b0c6c2393c138f15620907a68cb2baa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c1af3ad93634625799e921d2a8aabbf37d1c835e","is_merged_in_target_branch":false,"change_id":"I6de4bf9f7deed215cf61c87c3ae3f4240982523e","change_number":50892,"patch_set_number":18,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0cb05151b4d025f70f4ddf2b8d776a3a5aae4344":{"kind":"REWORK","_number":20,"created":"2013-11-28 08:01:58.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/20"}}},"commit":{"parents":[{"commit":"eb69f3fc9e245e8a59845976861947c695bec314","subject":"Support Neutron security groups in scenario testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eb69f3fc9e245e8a59845976861947c695bec314"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-28 07:59:35.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0cb05151b4d025f70f4ddf2b8d776a3a5aae4344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0cb05151b4d025f70f4ddf2b8d776a3a5aae4344"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"eb69f3fc9e245e8a59845976861947c695bec314","is_merged_in_target_branch":true,"change_id":"I6de4bf9f7deed215cf61c87c3ae3f4240982523e","change_number":50892,"patch_set_number":19,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7cf6f76992c8785767e512da86f2e49482ff63dd":{"kind":"REWORK","_number":21,"created":"2013-12-03 18:42:51.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/21"}}},"commit":{"parents":[{"commit":"c38be6def1554606df54abb3e1fde39ec8ef6067","subject":"Merge \"add some test for force delete:\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c38be6def1554606df54abb3e1fde39ec8ef6067"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-03 18:42:33.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7cf6f76992c8785767e512da86f2e49482ff63dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7cf6f76992c8785767e512da86f2e49482ff63dd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c38be6def1554606df54abb3e1fde39ec8ef6067","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a701818f390cced47871f7a1688aab2170f619ed":{"kind":"REWORK","_number":22,"created":"2013-12-03 18:44:16.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/22"}}},"commit":{"parents":[{"commit":"c38be6def1554606df54abb3e1fde39ec8ef6067","subject":"Merge \"add some test for force delete:\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c38be6def1554606df54abb3e1fde39ec8ef6067"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-03 18:44:03.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a701818f390cced47871f7a1688aab2170f619ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a701818f390cced47871f7a1688aab2170f619ed"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c38be6def1554606df54abb3e1fde39ec8ef6067","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f6eeeb8adaa828dc92900de08f0f7d5995ab40b2":{"kind":"REWORK","_number":23,"created":"2013-12-04 11:17:22.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/23"}}},"commit":{"parents":[{"commit":"c38be6def1554606df54abb3e1fde39ec8ef6067","subject":"Merge \"add some test for force delete:\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c38be6def1554606df54abb3e1fde39ec8ef6067"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-04 11:16:51.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f6eeeb8adaa828dc92900de08f0f7d5995ab40b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f6eeeb8adaa828dc92900de08f0f7d5995ab40b2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c38be6def1554606df54abb3e1fde39ec8ef6067","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"01066152f3fdc1645941f5a8fb97a784beab568e":{"kind":"REWORK","_number":24,"created":"2013-12-04 13:13:43.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/24"}}},"commit":{"parents":[{"commit":"c38be6def1554606df54abb3e1fde39ec8ef6067","subject":"Merge \"add some test for force delete:\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c38be6def1554606df54abb3e1fde39ec8ef6067"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-04 13:13:12.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nhelper changes:\ntenant_context_switch is now done by only loading a pre-initialized\ncompute_client with tenant credentials\n\nset_compute_context(tenant) - sets self.compute_client to a\ntenant-relevant compute_client\n\nremote_client:\n    1. ping_host - ping host via ssh client\n    2. get_mac_address - retrieve ssh server mac\n\nmanager:\n    1. move network and router creation to mananger\n    (from scenario/test_network_basic_ops)\n    2. add set_resource\u003dTrue to _create_secgroup_rule() so you can\ncreate a rule\n    not deleted by tearDown this is workaround for bug 1247568 in\n    python-neutronclient (preventing tearDownClass from ignoring deleted\nneutron\n    resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/01066152f3fdc1645941f5a8fb97a784beab568e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/01066152f3fdc1645941f5a8fb97a784beab568e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c38be6def1554606df54abb3e1fde39ec8ef6067","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"774cb09e0c1249d07d903f32c3477cbc38f1b5d9":{"kind":"REWORK","_number":25,"created":"2013-12-16 09:39:58.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/25"}}},"commit":{"parents":[{"commit":"992d1a14c7fd7ec37f3b12d3a93a6cd3c11cf398","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/992d1a14c7fd7ec37f3b12d3a93a6cd3c11cf398"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-16 09:37:22.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/774cb09e0c1249d07d903f32c3477cbc38f1b5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/774cb09e0c1249d07d903f32c3477cbc38f1b5d9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"992d1a14c7fd7ec37f3b12d3a93a6cd3c11cf398","is_merged_in_target_branch":false,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"799f245fcd23be7e885375bd46b5944533a43266":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2013-12-16 09:39:58.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/26"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2013-12-16 11:22:52.000000000","tz":0},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/799f245fcd23be7e885375bd46b5944533a43266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/799f245fcd23be7e885375bd46b5944533a43266"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dfe5f404d5dfbea6a2165cffefe140b6766a530c":{"kind":"REWORK","_number":27,"created":"2013-12-16 11:54:09.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/27"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-16 11:51:33.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe5f404d5dfbea6a2165cffefe140b6766a530c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe5f404d5dfbea6a2165cffefe140b6766a530c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"877532dc2b525c426b1fcf159867589221e7ba71":{"kind":"REWORK","_number":28,"created":"2013-12-16 13:36:00.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/28"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-16 13:34:21.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/877532dc2b525c426b1fcf159867589221e7ba71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/877532dc2b525c426b1fcf159867589221e7ba71"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fe9650ee170c74e6636cd5194a1c97dc46a69517":{"kind":"REWORK","_number":29,"created":"2013-12-17 06:04:29.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/29"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-17 06:04:13.000000000","tz":120},"subject":"scenario cross_tenant_connectivity - WIP","message":"scenario cross_tenant_connectivity - WIP\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe9650ee170c74e6636cd5194a1c97dc46a69517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe9650ee170c74e6636cd5194a1c97dc46a69517"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"48b73477b88eb8449363601d00c4157ff6351c55":{"kind":"REWORK","_number":30,"created":"2013-12-17 07:24:13.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/30"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-17 07:23:50.000000000","tz":120},"subject":"scenario cross_tenant_connectivity - WIP","message":"scenario cross_tenant_connectivity - WIP\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/48b73477b88eb8449363601d00c4157ff6351c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/48b73477b88eb8449363601d00c4157ff6351c55"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8dcd8d3f4681893fd91654d3d00667c88a4e4261":{"kind":"REWORK","_number":31,"created":"2013-12-17 08:42:25.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/31"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-17 08:41:40.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8dcd8d3f4681893fd91654d3d00667c88a4e4261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8dcd8d3f4681893fd91654d3d00667c88a4e4261"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a764eaa323ac3ee9b426c210e2a82d112b4c636d":{"kind":"REWORK","_number":32,"created":"2013-12-17 15:58:36.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/32"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-17 15:54:25.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a764eaa323ac3ee9b426c210e2a82d112b4c636d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a764eaa323ac3ee9b426c210e2a82d112b4c636d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0b720f2964c4969658fdd5c808ed87c33fca01d9":{"kind":"REWORK","_number":33,"created":"2013-12-17 16:14:51.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/33"}}},"commit":{"parents":[{"commit":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","subject":"Adds ping method to remote client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5f670abe2ccbb850eda4d2a25748bd763f2815ae"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-17 16:13:03.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0b720f2964c4969658fdd5c808ed87c33fca01d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0b720f2964c4969658fdd5c808ed87c33fca01d9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f670abe2ccbb850eda4d2a25748bd763f2815ae","is_merged_in_target_branch":true,"change_id":"I48931e968e76a24658bcda298e98f3a40c8358d1","change_number":60757,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4d7efa6d030bca335c7fb9f45e53650ef2ceb562":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2013-12-18 14:36:20.000000000","uploader":{"_account_id":8576,"name":"Yair Fried","email":"yfried@redhat.com","username":"yfried"},"ref":"refs/changes/01/55101/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/55101/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/55101/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/55101/34"}}},"commit":{"parents":[{"commit":"95ab960967ca85feaae2960c693677fe386b63f0","subject":"Merge \"avoid resource leaks in keypairs tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/95ab960967ca85feaae2960c693677fe386b63f0"}]}],"author":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-11-17 15:12:29.000000000","tz":120},"committer":{"name":"Yair Fried","email":"yfried@redhat.com","date":"2013-12-18 14:28:18.000000000","tz":120},"subject":"scenario cross_tenant_connectivity","message":"scenario cross_tenant_connectivity\n\ntest conectivity between VMs in the same tenant and on different\ntenants.\nfor each case, test that default state blocks traffic(*), that security\nrules allow only specified traffic\n\n(*)negative connectivity check is skipped until bug 1252620 is fixed\n   TODO(yfried): enable them once bug is fixed\n\nAlso adds workaround for bug 1247568 in python-neutronclient (preventing\ntearDownClass from ignoring deleted neutron resources\n\nChange-Id: I212873d46e20ff3161e376105af18020b1cf1c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4d7efa6d030bca335c7fb9f45e53650ef2ceb562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4d7efa6d030bca335c7fb9f45e53650ef2ceb562"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"95ab960967ca85feaae2960c693677fe386b63f0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
