)]}'
{"id":"openstack%2Ftempest~56738","triplet_id":"openstack%2Ftempest~master~Iefe5f90b3886c58736bdb27e9019bc19a8e49179","project":"openstack/tempest","branch":"master","topic":"update-lockutils-20131116","hashtags":[],"change_id":"Iefe5f90b3886c58736bdb27e9019bc19a8e49179","subject":"Update openstack/common/lockutils","status":"MERGED","created":"2013-11-16 03:51:38.000000000","updated":"2013-12-03 01:39:25.000000000","submitted":"2013-12-03 01:39:25.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0a3d4c09b72c204cdf708ca67f3253dc940d788c","_number":56738,"virtual_id_number":56738,"owner":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-12-03 01:39:25.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-12-02 21:33:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":2,"date":"2013-12-02 16:16:33.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2013-12-02 16:36:23.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-12-02 21:33:31.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"date":"2013-12-02 16:16:33.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2013-12-02 16:36:23.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-12-02 16:16:33.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2013-12-02 16:36:23.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2013-12-02 21:33:31.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2013-12-03 01:39:25.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"7baee279971337f8506d305766e3430831150472","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-16 04:47:13.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/38/56738/1/check/gate-tempest-pep8/2f1ac21 : SUCCESS in 1m 47s\n- gate-tempest-docs http://logs.openstack.org/38/56738/1/check/gate-tempest-docs/b2d98e7 : SUCCESS in 1m 47s\n- gate-tempest-python27 http://logs.openstack.org/38/56738/1/check/gate-tempest-python27/7ff8e13 : SUCCESS in 1m 42s\n- check-tempest-devstack-vm-full http://logs.openstack.org/38/56738/1/check/check-tempest-devstack-vm-full/2d57c03 : FAILURE in 55m 28s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/38/56738/1/check/check-tempest-devstack-vm-postgres-full/1dd727f : SUCCESS in 31m 10s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/38/56738/1/check/check-tempest-devstack-vm-neutron/968b606 : SUCCESS in 34m 54s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/38/56738/1/check/gate-tempest-devstack-vm-large-ops/7e84d01 : SUCCESS in 13m 53s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/38/56738/1/check/gate-tempest-devstack-vm-neutron-large-ops/3bd0d74 : SUCCESS in 16m 45s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/38/56738/1/check/check-grenade-devstack-vm/1545dae : SUCCESS in 34m 06s\n","accounts_in_message":[],"_revision_number":1},{"id":"51cb7bfdebed8f02e646a2642abfc8b4abdb631f","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-11-23 06:03:02.000000000","message":"Patch Set 1: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":1},{"id":"11f89d67fa717fc4f795a16aa0b1999df0de9e4c","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-11-23 06:03:02.000000000","message":"Patch Set 1:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":1},{"id":"b9027b8e0b41a43c83e13c13d5661af6d0d4d3e2","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2013-11-25 02:23:25.000000000","message":"Patch Set 1: Restored\n\nSorry, I was distracted on gate blocking bugs last week.","accounts_in_message":[],"_revision_number":1},{"id":"d02f90f21226387daac67b0832879e7f2618621e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-25 02:55:01.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/38/56738/1/check/gate-tempest-pep8/beaa5e6 : SUCCESS in 1m 44s\n- gate-tempest-docs http://logs.openstack.org/38/56738/1/check/gate-tempest-docs/5add5bc : SUCCESS in 1m 14s\n- gate-tempest-python27 http://logs.openstack.org/38/56738/1/check/gate-tempest-python27/6525873 : SUCCESS in 1m 20s\n- check-tempest-devstack-vm-full http://logs.openstack.org/38/56738/1/check/check-tempest-devstack-vm-full/e35d194 : SUCCESS in 30m 31s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/38/56738/1/check/check-tempest-devstack-vm-postgres-full/8e17926 : SUCCESS in 30m 53s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/38/56738/1/check/check-tempest-devstack-vm-neutron/26cec26 : SUCCESS in 25m 14s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/38/56738/1/check/gate-tempest-devstack-vm-large-ops/1c14b33 : SUCCESS in 18m 27s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/38/56738/1/check/gate-tempest-devstack-vm-neutron-large-ops/edb25a1 : SUCCESS in 16m 18s\n- check-grenade-devstack-vm http://logs.openstack.org/38/56738/1/check/check-grenade-devstack-vm/199ce5d : SUCCESS in 31m 32s\n","accounts_in_message":[],"_revision_number":1},{"id":"0f85ab8b1a3e8adb0bf3146f280cd667fd666d7d","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2013-12-02 16:16:33.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"29d3ed087e6aa0d4f4299b680bd8003a23761eb3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-02 16:36:23.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"0d490fac6cc03c6ff3dd0b9a18b8700f1796be0d","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-12-02 16:36:23.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"6238596bcccd4453e8b388f75c83a3495b02f527","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-02 19:04:11.000000000","message":"Patch Set 1: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/38/56738/1/gate/gate-tempest-pep8/ac696c8 : SUCCESS in 1m 53s\n- gate-tempest-docs http://logs.openstack.org/38/56738/1/gate/gate-tempest-docs/254c6e4 : SUCCESS in 1m 37s\n- gate-tempest-python27 http://logs.openstack.org/38/56738/1/gate/gate-tempest-python27/a8430d4 : SUCCESS in 2m 33s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/38/56738/1/gate/gate-tempest-devstack-vm-full/6721f1b : SUCCESS in 36m 22s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/38/56738/1/gate/gate-tempest-devstack-vm-postgres-full/29c39dc : SUCCESS in 50m 11s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/38/56738/1/gate/gate-tempest-devstack-vm-neutron/eeac312 : SUCCESS in 29m 35s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/38/56738/1/gate/gate-tempest-devstack-vm-large-ops/94fefb1 : SUCCESS in 17m 39s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/38/56738/1/gate/gate-tempest-devstack-vm-neutron-large-ops/b970e4d : FAILURE in 22m 42s\n- gate-grenade-devstack-vm http://logs.openstack.org/38/56738/1/gate/gate-grenade-devstack-vm/1677158 : SUCCESS in 24m 42s\n","accounts_in_message":[],"_revision_number":1},{"id":"bf4120819dc8af0d1505318d1164464fcb542db0","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2013-12-02 21:33:31.000000000","message":"Patch Set 1:\n\nreverify bug 1254890","accounts_in_message":[],"_revision_number":1},{"id":"0b514930c0eea129753dd2b26d489985cb231e6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-02 21:34:17.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"56d83fb16c88e42218cd6ac82e2ad8589c509cd8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-03 01:39:25.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1},{"id":"862e6b64c9794faa49f398266885e6390b6c0ad4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-03 01:39:25.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/38/56738/1/gate/gate-tempest-pep8/4c07b08 : SUCCESS in 2m 27s\n- gate-tempest-docs http://logs.openstack.org/38/56738/1/gate/gate-tempest-docs/c268246 : SUCCESS in 1m 40s\n- gate-tempest-python27 http://logs.openstack.org/38/56738/1/gate/gate-tempest-python27/715e2c3 : SUCCESS in 1m 21s\n- gate-tempest-dsvm-full http://logs.openstack.org/38/56738/1/gate/gate-tempest-dsvm-full/920ed29 : SUCCESS in 32m 38s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/38/56738/1/gate/gate-tempest-dsvm-postgres-full/6c4f79f : SUCCESS in 37m 12s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/38/56738/1/gate/gate-tempest-dsvm-neutron/2dd5bff : SUCCESS in 30m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/38/56738/1/gate/gate-tempest-dsvm-large-ops/a64a07c : SUCCESS in 16m 39s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/38/56738/1/gate/gate-tempest-dsvm-neutron-large-ops/ce1b7df : SUCCESS in 20m 57s\n- gate-grenade-dsvm http://logs.openstack.org/38/56738/1/gate/gate-grenade-dsvm/149278b : SUCCESS in 28m 17s\n","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2b1dea92c006c85a5dca5d8288a50bbd1296903","revisions":{"a2b1dea92c006c85a5dca5d8288a50bbd1296903":{"kind":"REWORK","_number":1,"created":"2013-11-16 03:51:38.000000000","uploader":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"ref":"refs/changes/38/56738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/38/56738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/38/56738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/38/56738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/38/56738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/38/56738/1"}}},"commit":{"parents":[{"commit":"26df0ac542ee11681fe859b2ab4c2cff50bb1c04","subject":"Merge \"Make smoke tests parallel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26df0ac542ee11681fe859b2ab4c2cff50bb1c04"}]}],"author":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2013-11-16 03:40:44.000000000","tz":660},"committer":{"name":"Michael Still","email":"mikal@stillhq.com","date":"2013-11-16 03:49:24.000000000","tz":660},"subject":"Update openstack/common/lockutils","message":"Update openstack/common/lockutils\n\nThe following commits are in this update:\n\n79e6bc6 fix lockutils.lock() to make it thread-safe\nace5120 Add main() to lockutils that creates temp dir for locks\n537d8e2 Allow lockutils to get lock_path conf from envvar\n371fa42 Move LockFixture into a fixtures module\nd498c42 Fix to properly log when we release a semaphore\n\nChange-Id: Iefe5f90b3886c58736bdb27e9019bc19a8e49179\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a2b1dea92c006c85a5dca5d8288a50bbd1296903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a2b1dea92c006c85a5dca5d8288a50bbd1296903"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"26df0ac542ee11681fe859b2ab4c2cff50bb1c04","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
