)]}'
{"id":"openstack%2Ftempest~586945","triplet_id":"openstack%2Ftempest~master~I845767da592119c354fec2fbfc4300fdd1d42ea1","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I845767da592119c354fec2fbfc4300fdd1d42ea1","subject":"Add cleanup after resource creation.","status":"ABANDONED","created":"2018-07-30 06:36:45.000000000","updated":"2021-07-15 22:52:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e3542f18806ca9fa0d9307f5fe002d9b316a73f","_number":586945,"virtual_id_number":586945,"owner":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2018-07-30 09:34:18.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":-1,"date":"2018-07-30 08:25:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"date":"2018-09-21 02:52:54.000000000","_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-07-08 17:13:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2018-07-30 08:55:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2021-07-08 17:13:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-30 08:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-30 08:55:55.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2018-07-30 09:34:18.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-08-02 11:21:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"98e91c72e482fe12a7ca0b38683b1ee488c0756a","author":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"date":"2018-07-30 06:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ada4da46b2c294cf789c9c12e6dfe41dbfb781dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-30 08:25:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/586945/1/check/tempest-full/804e65c/ : FAILURE in 1h 11m 21s\n- neutron-grenade http://logs.openstack.org/45/586945/1/check/neutron-grenade/54e6561/ : SUCCESS in 48m 49s\n- tempest-full-py3 http://logs.openstack.org/45/586945/1/check/tempest-full-py3/2d7a9c5/ : FAILURE in 1h 01m 42s\n- openstack-tox-pep8 http://logs.openstack.org/45/586945/1/check/openstack-tox-pep8/87a24ba/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/45/586945/1/check/openstack-tox-py27/0d14b6f/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/45/586945/1/check/openstack-tox-py35/1ebfddd/ : SUCCESS in 3m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/45/586945/1/check/build-openstack-sphinx-docs/6cbb0b7/html/ : SUCCESS in 9m 15s\n- neutron-grenade-multinode http://logs.openstack.org/45/586945/1/check/neutron-grenade-multinode/1e464c4/ : SUCCESS in 55m 59s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/45/586945/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/317ba72/ : FAILURE in 1h 13m 32s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/45/586945/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/6bd9ea1/ : SUCCESS in 55m 08s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/45/586945/1/check/neutron-tempest-dvr/60446a5/ : FAILURE in 1h 22m 26s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/45/586945/1/check/legacy-tempest-dsvm-neutron-full-ocata/df48511/ : FAILURE in 1h 16m 35s\n- openstack-tox-cover http://logs.openstack.org/45/586945/1/check/openstack-tox-cover/8af56f6/cover/ : SUCCESS in 4m 16s (non-voting)\n- nova-multiattach http://logs.openstack.org/45/586945/1/check/nova-multiattach/cdc5c6c/ : FAILURE in 50m 52s\n- tempest-full-parallel http://logs.openstack.org/45/586945/1/check/tempest-full-parallel/bf9f6a5/ : FAILURE in 1h 47m 17s (non-voting)\n- tempest-full-py36 http://logs.openstack.org/45/586945/1/check/tempest-full-py36/e06f56f/ : FAILURE in 1h 00m 55s (non-voting)\n- tempest-full-queens http://logs.openstack.org/45/586945/1/check/tempest-full-queens/3cc9202/ : FAILURE in 1h 16m 36s\n- tempest-full-queens-py3 http://logs.openstack.org/45/586945/1/check/tempest-full-queens-py3/a9805ea/ : FAILURE in 1h 15m 27s\n- tempest-full-pike http://logs.openstack.org/45/586945/1/check/tempest-full-pike/f20a15f/ : FAILURE in 1h 22m 33s\n- tempest-multinode-full http://logs.openstack.org/45/586945/1/check/tempest-multinode-full/fd15080/ : FAILURE in 1h 26m 22s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/45/586945/1/check/tempest-tox-plugin-sanity-check/4e708b7/ : FAILURE in 51m 29s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/45/586945/1/check/nova-cells-v1/f497db2/ : FAILURE in 49m 26s\n- nova-live-migration http://logs.openstack.org/45/586945/1/check/nova-live-migration/fcb3f33/ : SUCCESS in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d5ac6a3f6708d4836a3559fcf6356544ffcc4b1","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2018-07-30 08:55:55.000000000","message":"Patch Set 1: Code-Review-1\n\ncleanups are already included in base\u0027s helpers of creating servers/volumes.","accounts_in_message":[],"_revision_number":1},{"id":"81d5ffe5679ee2fd76e22dd1fd82045b97fabed3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-07-30 09:34:18.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027xenserver: recheck\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/45/586945/1/check/dsvm-tempest-neutron-network/612dff6 : FAILURE in 2h 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"60353f097ab8b280c274ba7a5e46db62b47fdc11","author":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"date":"2018-08-07 05:39:35.000000000","message":"Patch Set 1:\n\n\u003e cleanups are already included in base\u0027s helpers of creating\n \u003e servers/volumes.\n\n \u003e cleanups are already included in base\u0027s helpers of creating\n \u003e servers/volumes.\n\nThanks for the review, I think there are already in the other resources which are currently using addCleranup() after resource creation.\n\nhttps://github.com/openstack/tempest/blob/master/tempest/api/network/test_networks.py#L163\n\nBase class File: https://github.com/openstack/tempest/blob/master/tempest/api/network/base.py#L110\n\nSo can we remove the same from there too.?","accounts_in_message":[],"_revision_number":1},{"id":"7a4a57ed288396e6cb896bb28129704acc677336","author":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"date":"2018-08-13 04:39:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d3d4522886a980cdf71ebf538c3c804bdeb4c94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-13 06:37:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/586945/1/check/tempest-full/486be76/ : FAILURE in 1h 17m 59s\n- neutron-grenade http://logs.openstack.org/45/586945/1/check/neutron-grenade/1856da2/ : SUCCESS in 48m 30s\n- tempest-full-py3 http://logs.openstack.org/45/586945/1/check/tempest-full-py3/01e3f3b/ : FAILURE in 1h 21m 24s\n- openstack-tox-pep8 http://logs.openstack.org/45/586945/1/check/openstack-tox-pep8/fca6129/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/45/586945/1/check/openstack-tox-py27/ab00040/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/45/586945/1/check/openstack-tox-py35/a574ccb/ : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/45/586945/1/check/build-openstack-sphinx-docs/9f9f7f1/html/ : SUCCESS in 7m 08s\n- neutron-grenade-multinode http://logs.openstack.org/45/586945/1/check/neutron-grenade-multinode/f19f9e9/ : SUCCESS in 1h 00m 27s\n- legacy-tempest-dsvm-full-devstack-plugin-ceph http://logs.openstack.org/45/586945/1/check/legacy-tempest-dsvm-full-devstack-plugin-ceph/0fd4efc/ : FAILURE in 1h 11m 25s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/45/586945/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/ee15c40/ : SUCCESS in 52m 22s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/45/586945/1/check/neutron-tempest-dvr/cca6cce/ : FAILURE in 1h 20m 27s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/45/586945/1/check/legacy-tempest-dsvm-neutron-full-ocata/8f31886/ : FAILURE in 1h 14m 07s\n- openstack-tox-cover http://logs.openstack.org/45/586945/1/check/openstack-tox-cover/bb4c6e4/cover/ : SUCCESS in 4m 33s (non-voting)\n- nova-multiattach http://logs.openstack.org/45/586945/1/check/nova-multiattach/20ec983/ : FAILURE in 54m 09s\n- tempest-full-parallel http://logs.openstack.org/45/586945/1/check/tempest-full-parallel/ccac288/ : FAILURE in 1h 52m 03s (non-voting)\n- tempest-full-py36 http://logs.openstack.org/45/586945/1/check/tempest-full-py36/ad93cdb/ : FAILURE in 1h 10m 52s (non-voting)\n- tempest-full-queens http://logs.openstack.org/45/586945/1/check/tempest-full-queens/930404d/ : FAILURE in 1h 15m 26s\n- tempest-full-queens-py3 http://logs.openstack.org/45/586945/1/check/tempest-full-queens-py3/c35bf5c/ : FAILURE in 59m 05s\n- tempest-full-pike http://logs.openstack.org/45/586945/1/check/tempest-full-pike/09909fc/ : FAILURE in 1h 13m 06s\n- tempest-multinode-full http://logs.openstack.org/45/586945/1/check/tempest-multinode-full/68f2e24/ : FAILURE in 1h 23m 39s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/45/586945/1/check/tempest-tox-plugin-sanity-check/7919a49/ : FAILURE in 49m 38s (non-voting)\n- tempest-slow http://logs.openstack.org/45/586945/1/check/tempest-slow/08db3e7/ : SUCCESS in 1h 36m 00s\n- nova-cells-v1 http://logs.openstack.org/45/586945/1/check/nova-cells-v1/26a3e26/ : FAILURE in 1h 11m 24s\n- nova-live-migration http://logs.openstack.org/45/586945/1/check/nova-live-migration/93d65be/ : SUCCESS in 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c94b1a39a58cf740c4e8994cb4de2bfff5529711","author":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"date":"2018-09-21 02:52:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ff74188faf043d21d672aca443745616619366b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 05:32:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/45/586945/1/check/tempest-full/112db47/ : FAILURE in 1h 18m 12s\n- neutron-grenade http://logs.openstack.org/45/586945/1/check/neutron-grenade/57da951/ : SUCCESS in 1h 00m 03s\n- tempest-full-py3 http://logs.openstack.org/45/586945/1/check/tempest-full-py3/e970bab/ : FAILURE in 1h 26m 10s\n- openstack-tox-cover http://logs.openstack.org/45/586945/1/check/openstack-tox-cover/aeebd4a/cover/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/45/586945/1/check/openstack-tox-pep8/e65dee8/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/45/586945/1/check/openstack-tox-py27/d5985ce/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/45/586945/1/check/openstack-tox-py35/30605bb/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/45/586945/1/check/openstack-tox-py36/fc9888d/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/45/586945/1/check/openstack-tox-docs/f6cab1f/html/ : SUCCESS in 7m 03s\n- nova-multiattach http://logs.openstack.org/45/586945/1/check/nova-multiattach/754d105/ : FAILURE in 56m 41s\n- tempest-full-parallel http://logs.openstack.org/45/586945/1/check/tempest-full-parallel/7d4df2c/ : FAILURE in 2h 05m 27s (non-voting)\n- tempest-full-py36 http://logs.openstack.org/45/586945/1/check/tempest-full-py36/004e19e/ : TIMED_OUT in 2h 35m 45s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/45/586945/1/check/tempest-full-rocky/92f3ac5/ : FAILURE in 1h 10m 29s\n- tempest-full-rocky-py3 http://logs.openstack.org/45/586945/1/check/tempest-full-rocky-py3/08197ee/ : FAILURE in 1h 10m 11s\n- tempest-full-queens http://logs.openstack.org/45/586945/1/check/tempest-full-queens/5af358d/ : FAILURE in 1h 09m 01s\n- tempest-full-queens-py3 http://logs.openstack.org/45/586945/1/check/tempest-full-queens-py3/72cf65e/ : FAILURE in 1h 20m 00s\n- tempest-full-pike http://logs.openstack.org/45/586945/1/check/tempest-full-pike/94d3206/ : FAILURE in 1h 28m 22s\n- tempest-multinode-full http://logs.openstack.org/45/586945/1/check/tempest-multinode-full/1e4bf88/ : FAILURE in 1h 32m 58s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/45/586945/1/check/tempest-tox-plugin-sanity-check/88149a3/ : FAILURE in 49m 23s (non-voting)\n- tempest-slow http://logs.openstack.org/45/586945/1/check/tempest-slow/265d245/ : SUCCESS in 1h 43m 04s\n- nova-cells-v1 http://logs.openstack.org/45/586945/1/check/nova-cells-v1/5fcb5fd/ : FAILURE in 1h 50m 53s\n- nova-live-migration http://logs.openstack.org/45/586945/1/check/nova-live-migration/4baa711/ : SUCCESS in 37m 15s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/45/586945/1/check/neutron-grenade-multinode/d4fd016/ : SUCCESS in 59m 50s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/45/586945/1/check/devstack-plugin-ceph-tempest/4214b4f/ : FAILURE in 1h 33m 50s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/45/586945/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/4cbc0da/ : SUCCESS in 1h 12m 56s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/45/586945/1/check/neutron-tempest-dvr/cea5185/ : FAILURE in 1h 58m 46s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/45/586945/1/check/legacy-tempest-dsvm-neutron-full-ocata/e7bc30b/ : FAILURE in 1h 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"769630f56968764646a195e2beee2289ab51b485","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-08 17:13:24.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease rebase and update the patch if this is still needed. If no activity then we will abandon this review. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"6e3542f18806ca9fa0d9307f5fe002d9b316a73f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-15 22:52:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c48c844b2b4587668667435df9e7f13f00dab8e","revisions":{"8c48c844b2b4587668667435df9e7f13f00dab8e":{"kind":"REWORK","_number":1,"created":"2018-07-30 06:36:45.000000000","uploader":{"_account_id":25695,"name":"Deepak Mourya","email":"deepakmoriya7@gmail.com","username":"deepak_mourya"},"ref":"refs/changes/45/586945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/45/586945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/45/586945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/45/586945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/45/586945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/45/586945/1"}}},"commit":{"parents":[{"commit":"def30e0d4a56360eb86b33c0c2f873470d3f7d20","subject":"Merge \"Create port with port_vnic_type and port_profile from config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/def30e0d4a56360eb86b33c0c2f873470d3f7d20"}]}],"author":{"name":"deepak_mourya","email":"deepak.mourya@india.nec.com","date":"2018-07-30 06:10:56.000000000","tz":330},"committer":{"name":"deepak_mourya","email":"deepak.mourya@india.nec.com","date":"2018-07-30 06:10:56.000000000","tz":330},"subject":"Add cleanup after resource creation.","message":"Add cleanup after resource creation.\n\nThis PS. will add some missing cleanup after creation of\nvolume and server.\n\nChange-Id: I845767da592119c354fec2fbfc4300fdd1d42ea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8c48c844b2b4587668667435df9e7f13f00dab8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8c48c844b2b4587668667435df9e7f13f00dab8e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"def30e0d4a56360eb86b33c0c2f873470d3f7d20","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
