)]}'
{"id":"openstack%2Ftempest~608878","triplet_id":"openstack%2Ftempest~master~I2533429662ddafe376744f18698cdbbe041376b5","project":"openstack/tempest","branch":"master","topic":"bp/tempest-cli-unit-test-coverage","attention_set":{},"removed_from_attention_set":{"27078":{"account":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"last_update":"2023-07-18 01:50:51.000000000","reason":"Change was abandoned"},"17887":{"account":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"last_update":"2023-07-18 01:50:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2533429662ddafe376744f18698cdbbe041376b5","subject":"Test coverage of tempest account-generator for config and account file","status":"ABANDONED","created":"2018-10-09 08:10:44.000000000","updated":"2023-07-18 01:50:51.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dcf311e47c6b0ed5075ab7dfb73570d5252e0baf","_number":608878,"virtual_id_number":608878,"owner":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-30 14:05:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"value":-1,"date":"2021-12-10 22:59:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-09 08:13:15.000000000","updated_by":{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},"reviewer":{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},"state":"REVIEWER"},{"updated":"2018-10-09 12:40:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2018-12-14 11:26:28.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-21 03:38:35.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-22 00:40:47.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-03-22 23:24:01.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-09-04 03:22:20.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2019-09-30 14:05:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a335393ac7034a0f68d3a7d9708fd9c14c297a04","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-10-09 08:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46a7ac317528574e924a57a41fe41c5759cc0a89","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-10-09 08:20:15.000000000","message":"Patch Set 1:\n\n@all: Need you opinion\nI have around 25 test cases of similar kind for temepst account generator.\n\nCurrently I have used hard coded value of os-username, os-password, os-project-name and os-tenant-name.\n\nCommand is accepting only proper credentials, not fake.\n\nI didn\u0027t get alternative way using mocking or other means for these credentials. So I have used fixed values.\nCould anybody suggest the better approach in this, so that I can work on further test cases","accounts_in_message":[],"_revision_number":1},{"id":"c89b0644ace01b48f049497edf6abd6d68cadc0d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-10-09 10:30:30.000000000","message":"Patch Set 1:\n\n\u003e @all: Need you opinion\n \u003e I have around 25 test cases of similar kind for temepst account\n \u003e generator.\n \u003e \n \u003e Currently I have used hard coded value of os-username, os-password,\n \u003e os-project-name and os-tenant-name.\n \u003e \n \u003e Command is accepting only proper credentials, not fake.\n \u003e \n \u003e I didn\u0027t get alternative way using mocking or other means for these\n \u003e credentials. So I have used fixed values.\n \u003e Could anybody suggest the better approach in this, so that I can\n \u003e work on further test cases\n\nI am suer this is hitting from get_credentials but need to check the traceback. \n\noverall you need more mocking here to stub the credential part. waiting for gate results.","accounts_in_message":[],"_revision_number":1},{"id":"e819120974dff1746a16559d1a6b0d9b91ddfd87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-09 10:32:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 http://logs.openstack.org/78/608878/1/check/tempest-full-py3/530abe9/ : SUCCESS in 1h 51m 45s\n- openstack-tox-cover http://logs.openstack.org/78/608878/1/check/openstack-tox-cover/68ca653/ : FAILURE in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/1/check/openstack-tox-pep8/5d25ff1/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/1/check/openstack-tox-py27/33ae0ea/ : FAILURE in 3m 01s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/1/check/openstack-tox-py35/39a52e8/ : FAILURE in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/1/check/openstack-tox-py36/b1eaa61/ : FAILURE in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/78/608878/1/check/openstack-tox-docs/89cf810/html/ : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5099a960936ded505743389ab3d658a9b0d60490","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-10-09 11:46:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/78/608878/1/check/dsvm-tempest-neutron-network/ac1c5c6 : SUCCESS in 3h 07m 24s","accounts_in_message":[],"_revision_number":1},{"id":"45d13c764f18dc689dbbf1bf64e9a0d863cb63b9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-10-09 12:40:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58958c1a5f33ba34b3f193901667948cfa9d0001","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-10-09 16:14:54.000000000","message":"Patch Set 1:\n\n@gmann, thanks! let me try!","accounts_in_message":[],"_revision_number":1},{"id":"a0f6dd81639ce48c39a0c5b0dc86a9316143c41b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-10-10 22:31:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/608878/1/EMC_XtremIO_FC/333 : SUCCESS in 1h 29m 05s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/608878/1/EMC_XtremIO_iSCSI/343 : FAILURE in 1h 48m 02s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":1},{"id":"2bfdcc3c69ee5c9b05fb13590a9580d55f3a1504","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-10-11 13:30:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99d00e46436932904e353d5e355f759cba8f6b1d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-10-11 17:04:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/78/608878/2/check/dsvm-tempest-neutron-network/0af0e5f : SUCCESS in 2h 49m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c3b656330fa9547fa9916c394713e5f454de25a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-11 17:04:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 http://logs.openstack.org/78/608878/2/check/tempest-full-py3/b4ab97b/ : SUCCESS in 1h 26m 48s\n- openstack-tox-cover http://logs.openstack.org/78/608878/2/check/openstack-tox-cover/dc40ada/ : FAILURE in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/2/check/openstack-tox-pep8/c2c02f8/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/2/check/openstack-tox-py27/d18014b/ : FAILURE in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/2/check/openstack-tox-py35/dd83378/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/2/check/openstack-tox-py36/ba4bc8a/ : FAILURE in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/78/608878/2/check/openstack-tox-docs/c057b50/html/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6e006887d2ee036cd4c1df03fe12e3297d88a741","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-10-31 10:39:40.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"67530ea60704933c0e33746b3c155557f25d7a4e","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-27 06:45:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e4a3299bb896c309df8a5106376a9f34027732a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 06:58:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/78/608878/3/check/openstack-tox-cover/7e248aa/cover/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/3/check/openstack-tox-pep8/74ec9e7/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/3/check/openstack-tox-py27/2db73fc/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/3/check/openstack-tox-py35/673879f/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/3/check/openstack-tox-py36/97bc0e9/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/78/608878/3/check/openstack-tox-docs/e138cb8/html/ : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":3},{"id":"2f51ef3db81afaddc1d4bbb73569b034c5162f7a","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-27 07:05:30.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ae1e65e7200abb61d75c57fa01961665f4529126","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-27 07:05:44.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"a0551654616b821cf0c63dbd8735f4d6eb55ad14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 07:16:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/78/608878/4/check/openstack-tox-cover/78056f2/cover/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/4/check/openstack-tox-pep8/e376b92/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/4/check/openstack-tox-py27/8681928/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/4/check/openstack-tox-py35/3fe2c24/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/4/check/openstack-tox-py36/7b2196b/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/78/608878/4/check/openstack-tox-docs/3640f97/html/ : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"617b4d42fb7cf26c474e12030bcca8992fd5962b","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-27 10:21:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/78/608878/4/check/dsvm-tempest-neutron-network/bf4616b : SUCCESS in 3h 15m 13s","accounts_in_message":[],"_revision_number":4},{"id":"32fad7f9759d54a6f262c38177a13a7c5c632497","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-11-27 16:53:41.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/608878/4/EMC_XtremIO_FC/910 : SUCCESS in 1h 31m 38s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/608878/4/EMC_XtremIO_iSCSI/933 : SUCCESS in 1h 36m 07s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":4},{"id":"1784c95618e2bfb85b08f27c96f8a089d974f4c0","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-30 07:43:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9d91c194e9ea89e99c399f4bf83c0c6f73d85ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 09:26:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/78/608878/5/check/grenade-py3/77bd05a/ : SUCCESS in 1h 06m 21s\n- openstack-tox-cover http://logs.openstack.org/78/608878/5/check/openstack-tox-cover/a873174/cover/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/5/check/openstack-tox-pep8/a321cf3/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/5/check/openstack-tox-py27/c3b8224/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/5/check/openstack-tox-py35/fe5ec73/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/5/check/openstack-tox-py36/984e911/ : FAILURE in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/78/608878/5/check/openstack-tox-docs/5c43d28/html/ : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":5},{"id":"553eaf875d5b1b3d8a84a2c92d341ec41f8b323c","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-30 09:38:52.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"60103b13e07b3a6479da35bc7ed2096084822149","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-30 10:10:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"21a5ed119cf66e8ab5db26d4a218a6ccfaa043c2","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-30 10:14:49.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"d6715eb6e3c673f4a635176588283258c7dc59dc","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-30 11:01:21.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"90b23df776f542e885025da4313d97ef9f44d41f","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-30 12:33:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a79f36ad4632ec6923d196627710d32f1d54f22a","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-11-30 13:09:02.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3e9ab2ec891f8027444a83156a55edfb1ceb4141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 14:53:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/78/608878/11/check/grenade-py3/8e85273/ : SUCCESS in 54m 17s\n- openstack-tox-cover http://logs.openstack.org/78/608878/11/check/openstack-tox-cover/1046a6e/cover/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/11/check/openstack-tox-pep8/be7dc44/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/11/check/openstack-tox-py27/15b9e15/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/11/check/openstack-tox-py35/6d7386d/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/11/check/openstack-tox-py36/9ef7c37/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/78/608878/11/check/openstack-tox-docs/08c1a21/html/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":11},{"id":"c34e952e5ee1e2fb93eb94dd8fb894eaa421c973","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-11-30 15:45:40.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/78/608878/11/check/dsvm-tempest-neutron-network/200ccec : SUCCESS in 2h 36m 16s","accounts_in_message":[],"_revision_number":11},{"id":"823a702162e754796d9838d3fbc2a1f16636d5f4","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-12-07 09:06:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3945c6fe1305233c717aa9849c8fba1537e55612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 10:04:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/78/608878/12/check/grenade-py3/3dcdac9/ : SUCCESS in 52m 06s\n- openstack-tox-cover http://logs.openstack.org/78/608878/12/check/openstack-tox-cover/d3ba3e8/cover/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/12/check/openstack-tox-pep8/247c339/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/12/check/openstack-tox-py27/eaac1d7/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/12/check/openstack-tox-py35/3128c58/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/12/check/openstack-tox-py36/3edf84f/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/78/608878/12/check/openstack-tox-docs/55f002d/html/ : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":12},{"id":"6781e0ff03de7447a7089ad32ddcce01ee33d059","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-07 20:10:08.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/78/608878/12/check/EMC_XtremIO_FC/c79d5d7/EMC_XtremIO_FC/951 : FAILURE in 1h 33m 56s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/78/608878/12/check/EMC_XtremIO_iSCSI/15391e0/EMC_XtremIO_iSCSI/972 : SUCCESS in 1h 33m 47s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":12},{"id":"03aa4f10fa35f8ae23ecc8003004c2f96af24b47","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-12-11 10:51:22.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3966e38f514ce545c063ecadeaf4f1186de9b918","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 11:56:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/78/608878/13/check/grenade-py3/2bf17c5/ : SUCCESS in 58m 30s\n- openstack-tox-cover http://logs.openstack.org/78/608878/13/check/openstack-tox-cover/600c8d0/cover/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/13/check/openstack-tox-pep8/e61ebfd/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/13/check/openstack-tox-py27/41c040f/ : SUCCESS in 3m 02s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/13/check/openstack-tox-py35/1a0e0d1/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/13/check/openstack-tox-py36/6501fac/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/78/608878/13/check/openstack-tox-docs/604c401/html/ : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":13},{"id":"41a714ca1a38d102723df6ad0137202205171a5d","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-12-11 13:38:34.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/78/608878/13/check/dsvm-tempest-neutron-network/abf5124 : SUCCESS in 2h 45m 54s","accounts_in_message":[],"_revision_number":13},{"id":"c11d01cb54a1afd6576504858df55b822c414dbf","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2018-12-14 11:26:08.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1efd60c567292644dd9cc5efc6f632720fa33478","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-12-14 11:26:28.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"68ded25e85eb8924cdb688185317e58bf0c84b49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-14 11:42:08.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/78/608878/14/check/openstack-tox-cover/3c67077/cover/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/14/check/openstack-tox-pep8/9d3fb5f/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/14/check/openstack-tox-py27/bb29565/ : FAILURE in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/14/check/openstack-tox-py35/127aa68/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/14/check/openstack-tox-py36/496323a/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/78/608878/14/check/openstack-tox-docs/a8debb5/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":14},{"id":"4d60aecb78978ff540ec5eb09087566c8ed95d21","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-12-16 05:05:24.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/78/608878/14/check/EMC_XtremIO_FC/b139e5b/EMC_XtremIO_FC/1044 : FAILURE in 1h 56m 07s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/78/608878/14/check/EMC_XtremIO_iSCSI/7dd4685/EMC_XtremIO_iSCSI/1067 : FAILURE in 1h 54m 08s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":14},{"id":"d5fc8eb4d772a2b6d09c81f495cf357eb567e72f","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2019-03-07 10:55:00.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"8e18ce8494c6073f9317aa53c5201c43619ca3a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 11:06:43.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/78/608878/14/check/openstack-tox-cover/df8af05/cover/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/14/check/openstack-tox-pep8/e2b428d/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/14/check/openstack-tox-py27/7757ce3/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/14/check/openstack-tox-py35/d280617/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/14/check/openstack-tox-py36/0f34593/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/78/608878/14/check/openstack-tox-py37/49391cd/ : FAILURE in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/78/608878/14/check/openstack-tox-docs/c236e4d/html/ : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":14},{"id":"5ffb21bbd0d19e4b699aaea647911c1e6edf1035","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-03-20 17:09:47.000000000","message":"Patch Set 14:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":14},{"id":"53e1f9ad016809342ef91befd1548fa506d9814d","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-03-20 17:09:55.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"bef39c90d9cf11591d5d7e11e99ef716cc1ff2d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-20 18:54:47.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/78/608878/14/check/openstack-tox-cover/9c7575e/cover/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/78/608878/14/check/openstack-tox-pep8/fb2fbf5/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/78/608878/14/check/openstack-tox-py27/4602d49/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/78/608878/14/check/openstack-tox-py35/199e62d/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/78/608878/14/check/openstack-tox-py36/cfbc76a/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/78/608878/14/check/openstack-tox-py37/2df4388/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/78/608878/14/check/openstack-tox-docs/79c277c/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":14},{"id":"fd7568a56e5947f970dd0d81831515de59b8ed19","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 03:38:35.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/78/608878/14/check/EMC_XtremIO_FC/497dee7/EMC_XtremIO_FC/1974 : SUCCESS in 1h 27m 39s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/78/608878/14/check/EMC_XtremIO_iSCSI/a646506/EMC_XtremIO_iSCSI/1953 : SUCCESS in 1h 27m 41s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":14},{"id":"205700d445a5983bb297369667ebbe3a0ce658bc","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-03-22 00:40:47.000000000","message":"Patch Set 14: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"6d898b34b2ae17c4b58130a77bc5c6753460a5bc","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-03-22 23:24:01.000000000","message":"Patch Set 14: -Code-Review","accounts_in_message":[],"_revision_number":14},{"id":"7fd98cb9af5bf29648ebaf4c39e5cef306217f49","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-09-30 02:17:10.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"27e8a646c26437629ecbde7b80fc1719929ad217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 04:49:41.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1012f483de0a4d08b4fa11e958051478 : SUCCESS in 1h 44m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5834db33835242dea81b5f66741c3e40 : SUCCESS in 58m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3cfab75e90d54a15b7a6c5e6bf6ef68c : SUCCESS in 55m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eaef7eb4f1c447be8d157053f01836e1 : SUCCESS in 1h 26m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e32980ac57e49949bd467b11f34ccf7 : FAILURE in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd84a919658549fab8c91996e8c494ba : FAILURE in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1f4c348db334723b74798b3cc20faff : FAILURE in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b12f2ae265754aaf981eeba552074ef7 : FAILURE in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bcaf8de17894acaa1875070768a8fff : FAILURE in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9dfa3e9f9b2c400da4d3efbb7a6ad8f9 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be19b1eda1084374bdd09bf1cea8c330 : SUCCESS in 13m 08s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eb7ea82a3c714bc8b0f2235ad7a79193 : SUCCESS in 2h 03m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b2daefe5364b4d8e9ab2cfa6a5b5c211 : SUCCESS in 1h 19m 57s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/baf5d7a9d48546699730733a639fea0a : SUCCESS in 1h 35m 36s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/0463370b68334f7bb88ad02720d131a5 : SUCCESS in 1h 21m 19s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/d1fbf5946e3a48b8b49e7b032e44a262 : SUCCESS in 1h 18m 24s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/c948389b78704e7fa17f8508bd46b57f : SUCCESS in 1h 24m 41s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/c996c1795ded4c23a9aef25e2d3b4bbb : SUCCESS in 1h 19m 27s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/c5b0db20e1814f7cac903af0bb7950db : SUCCESS in 1h 19m 05s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8a237dd00b554a23930e6b6009a628aa : SUCCESS in 1h 54m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7fcaa1d61d764b9d9b5fe8d98297c40f : SUCCESS in 1h 37m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a5aed022276a49dd9b726fdbd7690dd2 : SUCCESS in 42m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d30211d990ab4dd48fc0809854161623 : SUCCESS in 1h 02m 07s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/39f541a30eca4bef9b8701841cb04ae4 : SUCCESS in 2h 26m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92d62deeb1384c4587537672bca6de73 : SUCCESS in 2h 22m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a77cb4e949d54d849eaa9d95015a8767 : SUCCESS in 48m 07s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/189c603c2851428b9cd48f6958e02f1e : SUCCESS in 1h 03m 24s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fba8203f5773430cb75650d41027a44e : SUCCESS in 1h 14m 26s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b9b35cc515c84312aaa4d430b91a0600 : FAILURE in 24m 02s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8f2d10d4a079424698830f2c5da7f334 : SUCCESS in 1h 14m 35s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/fb7c40f6ac284db6a96bb4ea9794ee03 : SUCCESS in 2m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0c675d25f488466f9def1c222220996e : SUCCESS in 1h 22m 28s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/76ea0bba6c8b495793375eb43af8624a : SUCCESS in 1h 10m 20s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0d4817a786354937aa2a20738141c98f : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":15},{"id":"81d566a8afbe3e5914a81ab55b8fad5375b45b88","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-09-30 11:37:18.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c5fb34ff4698fbb11b7f055bd46f669e95defaf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:05:43.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2fb8b40201554a96aac6253a2b478ea8 : SUCCESS in 1h 50m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/94552c0688254feeaa00fc09b5e734f0 : SUCCESS in 59m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7e10edb51cb34783b922e8b8e544de50 : SUCCESS in 1h 03m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc0a031a3aef4d2cbe0c2d3c071895cc : SUCCESS in 1h 23m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c89249344e7f4456b2e1b20030ac929f : FAILURE in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1a37a9314a4410ada8ada752a2b72c : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31609bdf63774602b407909bc4f417e5 : FAILURE in 3m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/72f2190469484e39bc2517bf6759bb59 : FAILURE in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cb1a656abb44a4f92716207a43be71f : FAILURE in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9076dbfd502143b982b60fe022a87799 : FAILURE in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a109f104c84004b09ea20340839d3e : SUCCESS in 7m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/5d536901cec04957a052f0fb571730e5 : TIMED_OUT in 2h 07m 26s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3631873cff63400998567e3946dd7d42 : SUCCESS in 1h 33m 57s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/66d3071d04794986bcc0e2df6955c034 : SUCCESS in 1h 37m 12s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/577b452155194c5bb26f382e78df3065 : SUCCESS in 1h 40m 36s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/cde6984c93dd44f0b2e12a22e885ef31 : SUCCESS in 1h 30m 13s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/ff22ca4f055741678fb1372838b67c5c : SUCCESS in 1h 16m 29s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/b8de5c2df75e42d982ba07ce796f07dc : SUCCESS in 1h 36m 44s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/f16dacda30194941a5aa7d9cbe2cfd78 : SUCCESS in 1h 34m 18s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ff07b17bd19a4961b382734cb8b67877 : SUCCESS in 1h 51m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/148e82600620443eb60f0b730058b9ab : SUCCESS in 1h 35m 25s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/34b221e3943f4ef082ae5f1b829a555c : SUCCESS in 57m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d910d416bccf476596f2af31ec2421f0 : SUCCESS in 1h 07m 01s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/a58c7838b61841fbb19bc16b0d4af2af : SUCCESS in 2h 20m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa87bce486384d5c9725d6bac541e46a : SUCCESS in 2h 22m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/33265463ce7f40d59c43e9ce6cb02577 : SUCCESS in 53m 51s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/038011f3950047768c741a6efef000d0 : SUCCESS in 1h 09m 40s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/461d50139c404e8f9e779e3f75ec85e7 : SUCCESS in 1h 17m 12s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5f42b7f217e643a799fe2e29b7f4253a : FAILURE in 22m 32s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/481a6999ce3f4be99fb0aac393ad8cc3 : SUCCESS in 1h 24m 35s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/88c17cbe96114e8ca3f58adaeb10c60c : SUCCESS in 2m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/bebebb96709d49f79996c7ac74a6a121 : SUCCESS in 1h 19m 54s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ce27de4df0e843d5b4159e66979d592e : SUCCESS in 1h 17m 21s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c1901e05744143559a4833cfa9778ade : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":16},{"id":"a4654e0b9b7cb5dbf42e06e7e516b502b01b8ee1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-10 22:59:53.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"dcf311e47c6b0ed5075ab7dfb73570d5252e0baf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-18 01:50:51.000000000","message":"Abandoned\n\nThanks. No activity on this since long, feel free to restore it if it is still needed.","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"29c6d255286d6f4674a01928db253122a088f5ff","revisions":{"1601f5382707bdd1bc45fa00a75f8433f79c4ebd":{"kind":"REWORK","_number":1,"created":"2018-10-09 08:10:44.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/1"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"subject":"[WIP] Test coverage for tempest account-generator","message":"[WIP] Test coverage for tempest account-generator\n\nAdding test cases of tempest account-generator on basis of different\nParameters used for command.\nThis patch will cover different perspective for different parameters to ensure\nThe coverage of tempest account generator\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1601f5382707bdd1bc45fa00a75f8433f79c4ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1601f5382707bdd1bc45fa00a75f8433f79c4ebd"}]},"branch":"refs/heads/master"},"814ef4100aa807ce7078086e31899c388b5c9f79":{"kind":"REWORK","_number":2,"created":"2018-10-11 13:30:50.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/2"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-11 13:30:18.000000000","tz":0},"subject":"[WIP] Test coverage for tempest account-generator","message":"[WIP] Test coverage for tempest account-generator\n\nAdding test cases of tempest account-generator on basis of different\nParameters used for command.\nThis patch will cover different perspective for different parameters to ensure\nThe coverage of tempest account generator\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/814ef4100aa807ce7078086e31899c388b5c9f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/814ef4100aa807ce7078086e31899c388b5c9f79"}]},"branch":"refs/heads/master"},"95c33e13b2608b1b9ec2cc947c1048c40fcc4e64":{"kind":"REWORK","_number":3,"created":"2018-11-27 06:45:09.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/3"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-27 06:44:34.000000000","tz":0},"subject":"[WIP] Test coverage for tempest account-generator","message":"[WIP] Test coverage for tempest account-generator\n\nAdding test cases of tempest account-generator on basis of different\nParameters used for command.\nThis patch will cover different perspective for different parameters to ensure\nThe coverage of tempest account generator\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/95c33e13b2608b1b9ec2cc947c1048c40fcc4e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/95c33e13b2608b1b9ec2cc947c1048c40fcc4e64"}]},"branch":"refs/heads/master"},"e612613cbbfe148b3d908cb8b52a3366bba8e67f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2018-11-27 07:05:30.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/4"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-27 07:05:07.000000000","tz":0},"subject":"Test coverage for tempest account-generator","message":"Test coverage for tempest account-generator\n\nAdding test cases of tempest account-generator on basis of different\nparameters used for command.\nThis patch will cover different perspective for different parameters\nto ensure the coverage of tempest account generator\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e612613cbbfe148b3d908cb8b52a3366bba8e67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e612613cbbfe148b3d908cb8b52a3366bba8e67f"}]},"branch":"refs/heads/master"},"a563b9d2a3c30a3208c6dd198299e51915b34dac":{"kind":"REWORK","_number":5,"created":"2018-11-30 07:43:58.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/5"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-30 07:43:36.000000000","tz":0},"subject":"Test coverage for tempest account-generator","message":"Test coverage for tempest account-generator\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used for command.\nThis patch will cover different perspective for different parameters\nto ensure the coverage of tempest account generator\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a563b9d2a3c30a3208c6dd198299e51915b34dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a563b9d2a3c30a3208c6dd198299e51915b34dac"}]},"branch":"refs/heads/master"},"7e5dda7a3393b35e0447f854cbcf16f9404660f6":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2018-11-30 09:38:52.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/6"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-30 09:38:28.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7e5dda7a3393b35e0447f854cbcf16f9404660f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7e5dda7a3393b35e0447f854cbcf16f9404660f6"}]},"branch":"refs/heads/master"},"ddd52d17cc909aec317f7996153542eb08168347":{"kind":"REWORK","_number":7,"created":"2018-11-30 10:10:13.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/7"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-30 10:09:46.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ddd52d17cc909aec317f7996153542eb08168347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ddd52d17cc909aec317f7996153542eb08168347"}]},"branch":"refs/heads/master"},"ff36b80a5722a758d1bd8840b87b956780e42bb2":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2018-11-30 10:14:49.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/8"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-30 10:14:31.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff36b80a5722a758d1bd8840b87b956780e42bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff36b80a5722a758d1bd8840b87b956780e42bb2"}]},"branch":"refs/heads/master"},"af9bd3bf1de4163ab3055baa42713dce61ef05e3":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2018-11-30 11:01:21.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/9"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-30 11:01:21.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/af9bd3bf1de4163ab3055baa42713dce61ef05e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/af9bd3bf1de4163ab3055baa42713dce61ef05e3"}]},"branch":"refs/heads/master"},"cbbf073a06172ae7af3e03be51d05f2728061c89":{"kind":"REWORK","_number":10,"created":"2018-11-30 12:33:19.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/10"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-30 12:32:25.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n*test_account_generator_no_extension_account_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cbbf073a06172ae7af3e03be51d05f2728061c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cbbf073a06172ae7af3e03be51d05f2728061c89"}]},"branch":"refs/heads/master"},"db330a4c1d290c4da73443c6c1b0c10a3266511c":{"kind":"REWORK","_number":11,"created":"2018-11-30 13:09:02.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/11"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-11-30 13:08:05.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n*test_account_generator_no_extension_account_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/db330a4c1d290c4da73443c6c1b0c10a3266511c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/db330a4c1d290c4da73443c6c1b0c10a3266511c"}]},"branch":"refs/heads/master"},"dc0e43d51a3992aa29d788fd1849194e82f99bc9":{"kind":"REWORK","_number":12,"created":"2018-12-07 09:06:43.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/12"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-12-07 09:05:58.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n*test_account_generator_no_extension_account_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dc0e43d51a3992aa29d788fd1849194e82f99bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dc0e43d51a3992aa29d788fd1849194e82f99bc9"}]},"branch":"refs/heads/master"},"ca2f1dde8d68ac405f99550e8d546d41da75fe95":{"kind":"REWORK","_number":13,"created":"2018-12-11 10:51:22.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/13"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-12-11 10:50:29.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n*test_account_generator_no_extension_account_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ca2f1dde8d68ac405f99550e8d546d41da75fe95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ca2f1dde8d68ac405f99550e8d546d41da75fe95"}]},"branch":"refs/heads/master"},"2f15b8436c75b76615a95c6a80b19a3d976b9817":{"kind":"REWORK","_number":14,"created":"2018-12-14 11:26:08.000000000","uploader":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"ref":"refs/changes/78/608878/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/14"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-12-14 11:25:45.000000000","tz":0},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n*test_account_generator_no_extension_account_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2f15b8436c75b76615a95c6a80b19a3d976b9817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2f15b8436c75b76615a95c6a80b19a3d976b9817"}]},"branch":"refs/heads/master"},"b38370b62f636c2277b4202d1e22422defaa6618":{"kind":"REWORK","_number":15,"created":"2019-09-30 02:17:10.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/608878/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/15"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2019-09-30 02:15:05.000000000","tz":-300},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n*test_account_generator_no_extension_account_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nAlso removed traceback.print_exc() from account_generator since\nLOG.exception(...) call will print the stack trace.\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b38370b62f636c2277b4202d1e22422defaa6618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b38370b62f636c2277b4202d1e22422defaa6618"}]},"branch":"refs/heads/master"},"29c6d255286d6f4674a01928db253122a088f5ff":{"kind":"REWORK","_number":16,"created":"2019-09-30 11:37:18.000000000","uploader":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"ref":"refs/changes/78/608878/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/608878/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/608878/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/608878/16"}}},"commit":{"parents":[{"commit":"eac094a8cf834d035316a900107f601adcc42ff5","subject":"docs: Fix incorrect docstrings in reStructuredText format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eac094a8cf834d035316a900107f601adcc42ff5"}]}],"author":{"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","date":"2018-10-09 08:09:48.000000000","tz":0},"committer":{"name":"Doug Schveninger","email":"ds6901@att.com","date":"2019-09-30 11:37:04.000000000","tz":-300},"subject":"Test coverage of tempest account-generator for config and account file","message":"Test coverage of tempest account-generator for config and account file\n\nAdding test cases of tempest account-generator on basis of config\nand account parameters used in command.\nThis patch will cover different perspective for config and account file\nparameters to ensure the coverage of tempest account generator.\n\nBelow test cases are added:\n*test_account_generator_no_config_file*\n*test_account_generator_no_exist_accounts_file*\n*test_account_generator_no_extension_account_file*\n\nSome of tests for config are added\nin \"https://review.openstack.org/#/c/620920/\" with the improvement\nin development code\n\nAlso removed traceback.print_exc() from account_generator since\nLOG.exception(...) call will print the stack trace.\n\nPartially-Implements: blueprint tempest-cli-unit-test-coverage\n\nChange-Id: I2533429662ddafe376744f18698cdbbe041376b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/29c6d255286d6f4674a01928db253122a088f5ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/29c6d255286d6f4674a01928db253122a088f5ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
