)]}'
{"id":"openstack%2Ftempest~608898","triplet_id":"openstack%2Ftempest~master~If31470dac38d7f08768f43fa77bd018a932f0313","project":"openstack/tempest","branch":"master","topic":"bp/volume-response-schema-validation","hashtags":[],"change_id":"If31470dac38d7f08768f43fa77bd018a932f0313","subject":"Add response schema validation for volume transfer","status":"MERGED","created":"2018-10-09 09:28:45.000000000","updated":"2019-09-09 05:16:35.000000000","submitted":"2019-09-09 05:15:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"608898-1568006123886-72f0986f","meta_rev_id":"c3e91a8760f6ba1acb1e7e38ceea269badcabd9d","_number":608898,"virtual_id_number":608898,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2019-09-09 05:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-09 02:53:20.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2019-09-06 18:12:02.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":1,"date":"2019-09-06 23:36:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-09 02:53:20.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-09 12:34:28.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-10 22:44:35.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 00:27:41.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-09-06 18:12:02.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-09-06 23:36:13.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2019-09-09 02:53:20.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-09-09 05:15:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02301569c405d5b231a140dde74ccd6e9e99e0c8","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2018-10-09 09:28:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"846656e9bc5a1091bd98260d57882c55607177cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-09 11:37:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/98/608898/1/check/tempest-full/b784a75/ : SUCCESS in 1h 41m 41s\n- neutron-grenade http://logs.openstack.org/98/608898/1/check/neutron-grenade/6c33283/ : SUCCESS in 50m 26s\n- tempest-full-py3 http://logs.openstack.org/98/608898/1/check/tempest-full-py3/4b86329/ : SUCCESS in 1h 23m 15s\n- openstack-tox-cover http://logs.openstack.org/98/608898/1/check/openstack-tox-cover/73eeb02/cover/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/98/608898/1/check/openstack-tox-pep8/ee0480f/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/98/608898/1/check/openstack-tox-py27/86c9d8a/ : SUCCESS in 3m 30s\n- openstack-tox-py35 http://logs.openstack.org/98/608898/1/check/openstack-tox-py35/7fc6a06/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/98/608898/1/check/openstack-tox-py36/7308525/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/98/608898/1/check/openstack-tox-docs/62a34b6/html/ : SUCCESS in 6m 23s\n- nova-multiattach http://logs.openstack.org/98/608898/1/check/nova-multiattach/36d9406/ : SUCCESS in 59m 43s\n- tempest-full-parallel http://logs.openstack.org/98/608898/1/check/tempest-full-parallel/06c9a7c/ : SUCCESS in 1h 48m 19s (non-voting)\n- tempest-full-py36 http://logs.openstack.org/98/608898/1/check/tempest-full-py36/ca805ee/ : SUCCESS in 1h 09m 18s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/98/608898/1/check/tempest-full-rocky/c49afd6/ : SUCCESS in 1h 39m 50s\n- tempest-full-rocky-py3 http://logs.openstack.org/98/608898/1/check/tempest-full-rocky-py3/cc41cce/ : SUCCESS in 1h 14m 18s\n- tempest-full-queens http://logs.openstack.org/98/608898/1/check/tempest-full-queens/842d1dc/ : SUCCESS in 1h 21m 48s\n- tempest-full-queens-py3 http://logs.openstack.org/98/608898/1/check/tempest-full-queens-py3/d74fb10/ : SUCCESS in 1h 23m 37s\n- tempest-full-pike http://logs.openstack.org/98/608898/1/check/tempest-full-pike/e3f451f/ : SUCCESS in 1h 44m 01s\n- tempest-multinode-full http://logs.openstack.org/98/608898/1/check/tempest-multinode-full/7e472ff/ : SUCCESS in 1h 36m 09s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/98/608898/1/check/tempest-tox-plugin-sanity-check/14bdee6/ : FAILURE in 55m 32s (non-voting)\n- tempest-slow http://logs.openstack.org/98/608898/1/check/tempest-slow/1d73882/ : SUCCESS in 1h 50m 17s\n- nova-cells-v1 http://logs.openstack.org/98/608898/1/check/nova-cells-v1/ad4953a/ : SUCCESS in 1h 04m 08s\n- nova-live-migration http://logs.openstack.org/98/608898/1/check/nova-live-migration/1f29b33/ : SUCCESS in 44m 38s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/98/608898/1/check/neutron-grenade-multinode/22c43d2/ : SUCCESS in 58m 27s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/98/608898/1/check/devstack-plugin-ceph-tempest/80e8d2c/ : SUCCESS in 1h 45m 35s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/98/608898/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/01049ea/ : SUCCESS in 47m 51s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/98/608898/1/check/neutron-tempest-dvr/a9f00f1/ : SUCCESS in 1h 40m 48s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/98/608898/1/check/legacy-tempest-dsvm-neutron-full-ocata/46544e3/ : SUCCESS in 1h 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"62cfc8243b574ce86b329f23cfb8540e20782d71","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2018-10-09 12:34:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-neutron-network http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/98/608898/1/check/dsvm-tempest-neutron-network/a4f4cc3 : SUCCESS in 2h 37m 37s","accounts_in_message":[],"_revision_number":1},{"id":"66f2acb14f1685cfbb9edc52426d3c237527f43f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-10-10 22:44:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/608898/1/EMC_XtremIO_FC/334 : ABORTED in 3h 00m 02s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/608898/1/EMC_XtremIO_iSCSI/344 : FAILURE in 6m 33s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":1},{"id":"71ba9bd89f4682a661729b8768c05e9c28327043","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2018-11-05 07:12:48.000000000","message":"Patch Set 1:\n\nCinder patch I52a81454a995920c1ba5534fea3d50f6c9986881 is merged","accounts_in_message":[],"_revision_number":1},{"id":"9689290bc16d9187a403a18baf271751a34e1f72","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-11-30 02:57:58.000000000","message":"Patch Set 1:\n\nHolding till we hear back from interop-wg.\nStarted the ML especially for interop-wg to judge the changes to avoid any failure in current guidelines. - http://lists.openstack.org/pipermail/openstack-discuss/2018-November/000358.html","accounts_in_message":[],"_revision_number":1},{"id":"9e05b35e677aa86c1c63cc187614d91be0aab58c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-12-05 12:21:29.000000000","message":"Patch Set 1: Code-Review-1\n\n-1 to hold these till we conclude on ML.","accounts_in_message":[],"_revision_number":1},{"id":"54fddc95a82de0f963ff8256f8c6e53286937c79","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-15 00:27:41.000000000","message":"Patch Set 1: -Code-Review\n\nAs discussed over ML, we are good to proceed on volume strict JSON validation.","accounts_in_message":[],"_revision_number":1},{"id":"bc4c3bbc6315e7c68abac6f3ace1712169f94457","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-08-21 00:23:50.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"3508a07fa1ca8cbf55d54f6a32fcba0ebdb52379","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 02:39:11.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- tempest-cinder-v2-api https://zuul.opendev.org/t/openstack/build/863c4e85c9da43b5957bc699d6b9a195 : SUCCESS in 47m 12s\n- tempest-all https://zuul.opendev.org/t/openstack/build/a602830db4a74eeeb1c29cb118e4a20d : FAILURE in 2h 02m 48s\n- legacy-tempest-dsvm-neutron-dvr-multinode-full https://zuul.opendev.org/t/openstack/build/93deafea9d1d4477a560a0f73a5bd2e4 : SUCCESS in 1h 49m 31s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d2e609a1b0d49178ef0480fd8bef16b : SUCCESS in 2h 10m 06s (non-voting)\n- nova-tempest-v2-api https://zuul.opendev.org/t/openstack/build/6d60e64a55da419284f3630623a8acd4 : SUCCESS in 1h 04m 25s\n- legacy-tempest-dsvm-lvm-multibackend https://zuul.opendev.org/t/openstack/build/149302ad1e264f29bd05f35e53cf54e5 : SUCCESS in 1h 35m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/360044a82a7f4c8eb9cb1b97e1afd33e : SUCCESS in 1h 05m 25s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/fda5b15cc0dc43428deaaac046bc7e27 : SUCCESS in 1h 56m 22s\n- tempest-full-py3-opensuse150 https://zuul.opendev.org/t/openstack/build/8735d72c66e94a2fa5ac9f0f8e86cf88 : SUCCESS in 1h 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b686c8e2f327060307f40d1c06f6628f82abaa16","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-08-23 01:22:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e61d9c9d033133f9f549d422d849187c04393269","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-08-23 02:26:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f880c7a4697ced18cf214d5dfccd78f37b0813ed","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-08-23 10:48:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLgtm","accounts_in_message":[],"_revision_number":1},{"id":"f469c7a8b248a6fba206977ca16f3b4bb80b0d87","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-08-30 17:14:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b68caad870603b2bc3c1f43a28a550fdce017ade","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-04 19:28:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nThis LGTM, even though as Doug pointed out the 3.55 microversion update adds additional params in the response body, but I don\u0027t believe the Tempest currently supports microversioned repsonse schemas for Cinder, only Nova, so I think this is good to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f1966aacd5b7d8a1e5e6f932396ff109269b6706","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-09-06 18:11:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d9e8d3008b10406d78982ebeb4eadab53b39a39c","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-09-06 18:12:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7cb83382d5ddc9a8adce3a49e5c0a1e5e77d76ea","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-09-06 23:36:13.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm https://developer.openstack.org/api-ref/block-storage/v3/index.html#create-a-volume-transfer","accounts_in_message":[],"_revision_number":2},{"id":"0e0e639bf5df73f49ea44ca5ebd309d4be638728","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 02:42:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2af4b034a2a461d8542fa33120624a9 : SUCCESS in 1h 41m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ff36784f1b0c492685218412f9a54585 : SUCCESS in 1h 01m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3d225604b83f476e930b3e7f0c019095 : SUCCESS in 1h 05m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03b940592f5840d5b4c52547c3264de6 : SUCCESS in 1h 29m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08488b70547648eda91bc8285ccd5d9e : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edecdbd477eb4797b39eb6e83df06a31 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa76b304eecd4584bab745872cf4dcc3 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/734aa7e326c0409981e157f0153829a6 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4209d9642f744ceb8a7ba1eba5205655 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9963a464e994446986e68df063b1b77 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c1498fb18249c19a81ee613f6b4caf : SUCCESS in 7m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c1fe2cc8ca4c4130a3d1e2c2ebbb2eb4 : TIMED_OUT in 2h 08m 53s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d7ab2c6031d144f6a74ad4199ac45668 : SUCCESS in 1h 31m 57s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/1a9a8045c2664711ac804f541412e2f3 : SUCCESS in 1h 26m 03s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/50280b3cdaa545ed836e50782e72b51e : SUCCESS in 1h 21m 55s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/95d9dabd9c364d9695808984af26bb78 : SUCCESS in 1h 23m 35s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/e8a2d4e920754d97a97ec4c031f31ce0 : SUCCESS in 1h 16m 10s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/6618d4d5dadb43848f46b8ca968c082d : SUCCESS in 1h 22m 40s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/a42521e6179848b5b59d00c28199796a : SUCCESS in 1h 14m 26s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c458afe44454b72a1acc8e1658a3623 : SUCCESS in 2h 10m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dfb6414263494f7982ffb0e6385c407a : SUCCESS in 1h 34m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6e638c5bdf604a32bd4a7cee26495d0d : SUCCESS in 45m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d706bd268f4c40508d7720c5c6f167b1 : SUCCESS in 1h 09m 45s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/b9c48f07902347efae5a0a01b0f83028 : SUCCESS in 2h 33m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3799e9790b1646d1bb3d66b377b08976 : SUCCESS in 2h 30m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e10047f18b0470e91df6be0d2de958b : SUCCESS in 58m 01s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eaa6e78003bb435d9d21c3239e41cc2e : SUCCESS in 1h 07m 38s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/23416abc531f402a84d0c110f4c4be0d : SUCCESS in 1h 11m 57s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/eea37569b21c4528bfb6b57dc25f697e : FAILURE in 24m 17s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f463338ad8124f32869c7ad05bd0d15c : SUCCESS in 1h 21m 57s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/64496d4dfecf4b7cba847b83aa36c419 : SUCCESS in 2m 14s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0bcdd18facb340bda74c08eff0ee9dce : SUCCESS in 1h 14m 44s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/dfc872dd68db4cf898ffabf30896be52 : SUCCESS in 1h 01m 08s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/66c68a90140046c186f0d9af5aca1adc : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b5d56174dbdcb1c901c50788585009d7bcd42bb2","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-09 02:53:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0771dc02f2bd1f381675d86d5e4a5790e4a2afeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 02:53:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"180ae9ae238f10577a5996bea2e30f499ca7505f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 05:15:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e5b18d2b2601505a7037f6aae0836a73bfee6044","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 05:15:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d6e56ccf4d54ebab9f391cd3dea3c73 : SUCCESS in 1h 41m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/65a6c55f10ab4dcd9fde5c20a4b7f5f5 : SUCCESS in 1h 05m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/af23c3d27d494fe6b881962ff89fbcf6 : SUCCESS in 1h 02m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2408cd3c1d6c4858ac87fd817882bad6 : SUCCESS in 1h 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2771e673bbf4843a82a37e941a1f817 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a40877675ae643a997e3ee6d2e654f88 : SUCCESS in 3m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/721958c0fa7c4a28ae8973a366fc6163 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20ae1d0e24d64da19f29550c1391c788 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4466914273244ae8761a06cd81a72e4 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3220f69c26f42219465c5d667d85bbb : SUCCESS in 8m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/453c3f452187401382c72bcacb2ad8d7 : SUCCESS in 2h 18m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5dccec24a02e454fbf572e3a8514a3a0 : SUCCESS in 1h 23m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bdbc7e7b862404f906ef48b6d8e884e : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c76db225d7fb7f081275d020e887e9c82ff1afb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 05:16:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81496ac8c5d44089ba8d3bd6b22958e4 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"48a76dbcc77f8276c92e8be1da0348caff4b357e","revisions":{"a177cf6a6e31a9f26fc2f19a32a5083282bed4e8":{"kind":"REWORK","_number":1,"created":"2018-10-09 09:28:45.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/98/608898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/98/608898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/608898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/608898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/608898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/98/608898/1"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2018-10-09 09:34:19.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2018-10-09 09:34:19.000000000","tz":480},"subject":"Add response schema validation for volume transfer","message":"Add response schema validation for volume transfer\n\nThis is to add response schema validation for volume transfer.\nBesides, created_at should be removed from the response example of\n\"List volume transfers for a project\" in api ref,\nI52a81454a995920c1ba5534fea3d50f6c9986881 will fix them.\n\nChange-Id: If31470dac38d7f08768f43fa77bd018a932f0313\npartially-implements: blueprint volume-response-schema-validation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a177cf6a6e31a9f26fc2f19a32a5083282bed4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a177cf6a6e31a9f26fc2f19a32a5083282bed4e8"}]},"branch":"refs/heads/master"},"48a76dbcc77f8276c92e8be1da0348caff4b357e":{"kind":"REWORK","_number":2,"created":"2019-09-06 18:11:58.000000000","uploader":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"ref":"refs/changes/98/608898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/98/608898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/608898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/608898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/608898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/98/608898/2"}}},"commit":{"parents":[{"commit":"dfe511eef288dc36d8de18f0ccd081dfc6c534f6","subject":"Merge \"Adding unit test of setting config path for \"tempest run\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/dfe511eef288dc36d8de18f0ccd081dfc6c534f6"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2018-10-09 09:34:19.000000000","tz":480},"committer":{"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","date":"2019-09-06 18:11:51.000000000","tz":0},"subject":"Add response schema validation for volume transfer","message":"Add response schema validation for volume transfer\n\nThis is to add response schema validation for volume transfer.\nBesides, created_at should be removed from the response example of\n\"List volume transfers for a project\" in api ref,\nI52a81454a995920c1ba5534fea3d50f6c9986881 will fix them.\n\nChange-Id: If31470dac38d7f08768f43fa77bd018a932f0313\npartially-implements: blueprint volume-response-schema-validation\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/48a76dbcc77f8276c92e8be1da0348caff4b357e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/48a76dbcc77f8276c92e8be1da0348caff4b357e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
