)]}'
{"id":"openstack%2Ftempest~612396","triplet_id":"openstack%2Ftempest~master~I0cd0e7c0cfcbd9942736b6a7cbd0929e3878b2e0","project":"openstack/tempest","branch":"master","topic":"fix-run-tempest-job","hashtags":[],"change_id":"I0cd0e7c0cfcbd9942736b6a7cbd0929e3878b2e0","subject":"Fix parameters in run-tempest role","status":"ABANDONED","created":"2018-10-22 13:57:20.000000000","updated":"2018-10-23 16:13:48.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"93a726b58b32c5b730acbd73d91d61dac4a9bd7e","_number":612396,"virtual_id_number":612396,"owner":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2018-10-22 19:58:39.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":-1,"date":"2018-10-22 18:35:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2018-10-23 09:26:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-22 18:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-22 19:58:39.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-23 09:26:37.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"2a783f96d3da285806339f9967bd68778fd6ee05","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2018-10-22 13:57:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16735f0647d20d2e9494d14bebc1a1afd54bec9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-22 18:35:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/96/612396/1/check/tempest-full/0529b24/ : TIMED_OUT in 2h 08m 56s\n- neutron-grenade http://logs.openstack.org/96/612396/1/check/neutron-grenade/f7065d7/ : SUCCESS in 55m 15s\n- tempest-full-py3 http://logs.openstack.org/96/612396/1/check/tempest-full-py3/1842e26/ : TIMED_OUT in 2h 09m 15s\n- openstack-tox-cover http://logs.openstack.org/96/612396/1/check/openstack-tox-cover/db7aaca/cover/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/96/612396/1/check/openstack-tox-pep8/6a0f6d8/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/96/612396/1/check/openstack-tox-py27/2fad89c/ : SUCCESS in 3m 31s\n- openstack-tox-py35 http://logs.openstack.org/96/612396/1/check/openstack-tox-py35/2fd7dad/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/96/612396/1/check/openstack-tox-py36/5c66ede/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/96/612396/1/check/openstack-tox-docs/ad9f1a5/html/ : SUCCESS in 8m 16s\n- devstack-tempest http://logs.openstack.org/96/612396/1/check/devstack-tempest/72c451d/ : SUCCESS in 1h 52m 23s\n- nova-multiattach http://logs.openstack.org/96/612396/1/check/nova-multiattach/b9b4942/ : SUCCESS in 1h 01m 23s\n- tempest-full-parallel http://logs.openstack.org/96/612396/1/check/tempest-full-parallel/01b726b/ : TIMED_OUT in 2h 08m 20s (non-voting)\n- tempest-full-py36 http://logs.openstack.org/96/612396/1/check/tempest-full-py36/3f78237/ : TIMED_OUT in 2h 05m 26s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/96/612396/1/check/tempest-full-rocky/7c2fec4/ : TIMED_OUT in 2h 10m 03s\n- tempest-full-rocky-py3 http://logs.openstack.org/96/612396/1/check/tempest-full-rocky-py3/cf2a2df/ : TIMED_OUT in 2h 09m 10s\n- tempest-full-queens http://logs.openstack.org/96/612396/1/check/tempest-full-queens/fd6f8cc/ : TIMED_OUT in 2h 10m 27s\n- tempest-full-queens-py3 http://logs.openstack.org/96/612396/1/check/tempest-full-queens-py3/d6a6808/ : TIMED_OUT in 2h 09m 18s\n- tempest-full-pike http://logs.openstack.org/96/612396/1/check/tempest-full-pike/69834a0/ : TIMED_OUT in 2h 10m 57s\n- tempest-multinode-full http://logs.openstack.org/96/612396/1/check/tempest-multinode-full/52018da/ : TIMED_OUT in 3h 10m 37s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/96/612396/1/check/tempest-tox-plugin-sanity-check/fcd1fc2/ : FAILURE in 43m 29s (non-voting)\n- tempest-slow http://logs.openstack.org/96/612396/1/check/tempest-slow/ccaa239/ : TIMED_OUT in 3h 06m 58s\n- nova-cells-v1 http://logs.openstack.org/96/612396/1/check/nova-cells-v1/b43b23a/ : SUCCESS in 1h 01m 38s\n- nova-live-migration http://logs.openstack.org/96/612396/1/check/nova-live-migration/caaf24e/ : SUCCESS in 46m 16s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/96/612396/1/check/neutron-grenade-multinode/1c3e885/ : SUCCESS in 1h 06m 39s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/96/612396/1/check/devstack-plugin-ceph-tempest/26acc12/ : FAILURE in 30m 19s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/96/612396/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/529dbe0/ : SUCCESS in 49m 57s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/96/612396/1/check/neutron-tempest-dvr/2021740/ : SUCCESS in 1h 18m 21s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/96/612396/1/check/legacy-tempest-dsvm-neutron-full-ocata/5930623/ : SUCCESS in 1h 33m 57s\n- interop-tempest-consistency http://logs.openstack.org/96/612396/1/check/interop-tempest-consistency/d451492/ : SUCCESS in 3m 39s (non-voting)\n- tempest-full-test-account-py3 http://logs.openstack.org/96/612396/1/check/tempest-full-test-account-py3/1d3f3b5/ : FAILURE in 1h 41m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/96/612396/1/check/tempest-full-test-account-no-admin-py3/04bc1a7/ : TIMED_OUT in 2h 07m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64ac0ca711ac88506a594909269f9b4ae3d17728","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2018-10-22 19:58:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/612396/1/EMC_XtremIO_FC/397 : SUCCESS in 1h 27m 59s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/612396/1/EMC_XtremIO_iSCSI/453 : FAILURE in 1h 49m 28s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":1},{"id":"61f641eade8e5d5f54ee6e68de6245de2b9f0885","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-10-23 09:26:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a5425d4729fc9cd2b2dc7727003d1da69a86777","author":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"date":"2018-10-23 16:13:48.000000000","message":"Abandoned\n\nThe solution is to use `all` tox environment instead of `full`","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e8d27f2b2cce303f26c14361724bad57a23bdbb","revisions":{"0e8d27f2b2cce303f26c14361724bad57a23bdbb":{"kind":"REWORK","_number":1,"created":"2018-10-22 13:57:20.000000000","uploader":{"_account_id":5950,"name":"Ilya Shakhat","email":"shakhat@gmail.com","username":"shakhat"},"ref":"refs/changes/96/612396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/96/612396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/612396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/612396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/612396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/96/612396/1"}}},"commit":{"parents":[{"commit":"f46eebed9d0aaeb9e31b7afb1bf8f552ca65ed19","subject":"Merge \"Fix help message of user_unique_last_password_count\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f46eebed9d0aaeb9e31b7afb1bf8f552ca65ed19"}]}],"author":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-10-22 13:51:00.000000000","tz":120},"committer":{"name":"Ilya Shakhat","email":"shakhat@gmail.com","date":"2018-10-22 13:51:00.000000000","tz":120},"subject":"Fix parameters in run-tempest role","message":"Fix parameters in run-tempest role\n\nReadme of run-tempest job declares variable `tempest_test_regex`\nto be a regular expression used to select the tests. But in\norder to work this variable needs to be passed to tox with\n--regex keyword. Otherwise it is passed as positional parameter\nleading to error: \"tempest run: error: unrecognized arguments\" [1]\n\nThe similar issue is with `tempest_test_blacklist`, which is declared\nas a blacklist file to skip tests that are not needed, so should be\npassed with --blacklist-file parameter.\n\n[1] http://logs.openstack.org/59/609959/2/check/tempest-full-profiler-enabled/3054075/ara-report/result/1e906e1a-aa43-42a0-b5fa-2926916dc294/\n\nChange-Id: I0cd0e7c0cfcbd9942736b6a7cbd0929e3878b2e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e8d27f2b2cce303f26c14361724bad57a23bdbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e8d27f2b2cce303f26c14361724bad57a23bdbb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f46eebed9d0aaeb9e31b7afb1bf8f552ca65ed19","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
