)]}'
{"id":"openstack%2Ftempest~63435","triplet_id":"openstack%2Ftempest~master~I9b1bbfe231c85c01938bd68be4e5974bd24130d6","project":"openstack/tempest","branch":"master","topic":"config_proxy","hashtags":[],"change_id":"I9b1bbfe231c85c01938bd68be4e5974bd24130d6","subject":"moving to global lazy loaded config","status":"MERGED","created":"2013-12-20 15:11:02.000000000","updated":"2013-12-20 21:53:07.000000000","submitted":"2013-12-20 21:53:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ff7a3cb08d3dc972cb943b1bbe8a2afa7c2457b","_number":63435,"virtual_id_number":63435,"owner":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-12-20 21:53:06.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-12-20 20:05:02.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":2,"date":"2013-12-20 20:06:54.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2013-12-20 20:05:02.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":1,"date":"2013-12-20 20:06:54.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-12-20 20:05:02.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2013-12-20 21:53:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"32e596fe1fd5a4038ecb93def6b1c4cc155f96ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-20 15:37:16.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/35/63435/1/check/gate-tempest-pep8/3e7bbf2 : FAILURE in 2m 01s\n- gate-tempest-docs http://logs.openstack.org/35/63435/1/check/gate-tempest-docs/d72b449 : SUCCESS in 1m 27s\n- gate-tempest-python27 http://logs.openstack.org/35/63435/1/check/gate-tempest-python27/f1e5f1b : SUCCESS in 1m 42s\n- check-tempest-dsvm-full http://logs.openstack.org/35/63435/1/check/check-tempest-dsvm-full/8f8fd17 : FAILURE in 11m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/63435/1/check/check-tempest-dsvm-postgres-full/52dc57d : FAILURE in 9m 58s\n- check-tempest-dsvm-neutron http://logs.openstack.org/35/63435/1/check/check-tempest-dsvm-neutron/0a9af4e : FAILURE in 11m 02s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/63435/1/check/gate-tempest-dsvm-large-ops/23e9bcd : FAILURE in 12m 19s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/63435/1/check/gate-tempest-dsvm-neutron-large-ops/32d3ea1 : FAILURE in 10m 48s\n- check-grenade-dsvm http://logs.openstack.org/35/63435/1/check/check-grenade-dsvm/4425dc5 : FAILURE in 26m 08s\n","accounts_in_message":[],"_revision_number":1},{"id":"05f1bb6d70e68b728c3a7946051ffa7abd72d54a","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-12-20 16:46:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0f8bcbfa7d1ef9e5a01877c417b029ec2002c94f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-20 17:08:22.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/35/63435/2/check/gate-tempest-pep8/60b42d5 : SUCCESS in 2m 20s\n- gate-tempest-docs http://logs.openstack.org/35/63435/2/check/gate-tempest-docs/3702379 : SUCCESS in 1m 33s\n- gate-tempest-python27 http://logs.openstack.org/35/63435/2/check/gate-tempest-python27/592158a : SUCCESS in 1m 21s\n- check-tempest-dsvm-full http://logs.openstack.org/35/63435/2/check/check-tempest-dsvm-full/c2ab7a8 : FAILURE in 17m 13s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/63435/2/check/check-tempest-dsvm-postgres-full/a034683 : FAILURE in 10m 39s\n- check-tempest-dsvm-neutron http://logs.openstack.org/35/63435/2/check/check-tempest-dsvm-neutron/7d74074 : FAILURE in 11m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/63435/2/check/gate-tempest-dsvm-large-ops/5cd1f0c : FAILURE in 9m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/63435/2/check/gate-tempest-dsvm-neutron-large-ops/e931107 : FAILURE in 14m 56s\n- check-grenade-dsvm http://logs.openstack.org/35/63435/2/check/check-grenade-dsvm/e9db8c4 : FAILURE in 21m 32s\n","accounts_in_message":[],"_revision_number":2},{"id":"b10eb5507876c93647389ecaa5e3aed9f67a10d7","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-12-20 19:01:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"28b7359b37edfbb4a3131389d6c334e9691fd54b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-20 19:54:48.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-pep8 http://logs.openstack.org/35/63435/3/check/gate-tempest-pep8/e136af6 : SUCCESS in 2m 15s\n- gate-tempest-docs http://logs.openstack.org/35/63435/3/check/gate-tempest-docs/613ab8a : SUCCESS in 1m 26s\n- gate-tempest-python27 http://logs.openstack.org/35/63435/3/check/gate-tempest-python27/b5b7048 : SUCCESS in 1m 22s\n- check-tempest-dsvm-full http://logs.openstack.org/35/63435/3/check/check-tempest-dsvm-full/a3e986e : SUCCESS in 43m 07s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/35/63435/3/check/check-tempest-dsvm-postgres-full/7962648 : SUCCESS in 53m 27s\n- check-tempest-dsvm-neutron http://logs.openstack.org/35/63435/3/check/check-tempest-dsvm-neutron/ad1459f : SUCCESS in 35m 06s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/63435/3/check/gate-tempest-dsvm-large-ops/9a57a4f : SUCCESS in 13m 20s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/63435/3/check/gate-tempest-dsvm-neutron-large-ops/251d8b6 : FAILURE in 10m 45s\n- check-grenade-dsvm http://logs.openstack.org/35/63435/3/check/check-grenade-dsvm/771ee99 : SUCCESS in 28m 57s\n","accounts_in_message":[],"_revision_number":3},{"id":"3059a9ee9c4401f5c328694f3073da0962d6e8b4","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2013-12-20 19:55:07.000000000","message":"Patch Set 3:\n\nI noticed tempest failed, I think you hit bug(s):\n\n- https://bugs.launchpad.net/bugs/1262906\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1262906\n\nFor a code review which has been approved but failed to merge,\nyou can reverify by leaving a comment like this:\n\n    reverify bug 1262906","accounts_in_message":[],"_revision_number":3},{"id":"c0f1863d91e2cb8155b59f60637e1fac258ffec9","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2013-12-20 20:05:02.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\nTempestConfig should be renamed so no one tries to use it in the future but that can be in a follow-on patch.","accounts_in_message":[],"_revision_number":3},{"id":"00497ff938f587e336ada59e884dbd8b8fcffd76","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2013-12-20 20:06:54.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\nthe fail on this is due to a race that\u0027s being worked, so known issue. Rushing it through because this is going to be a crazy rebase issue otherwise.\n\nCleanups coming that will handle things like ensuring people don\u0027t call the base class directly.","accounts_in_message":[],"_revision_number":3},{"id":"1162c15ff3653df2c318b288007150aad160c04b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-20 20:06:55.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"ff6a1d95df7a789c0a28478594a7d87133cfd2f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-20 21:53:06.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- gate-tempest-pep8 http://logs.openstack.org/35/63435/3/gate/gate-tempest-pep8/33aea52 : SUCCESS in 1m 19s\n- gate-tempest-docs http://logs.openstack.org/35/63435/3/gate/gate-tempest-docs/218d29d : SUCCESS in 1m 13s\n- gate-tempest-python27 http://logs.openstack.org/35/63435/3/gate/gate-tempest-python27/ffbcf82 : SUCCESS in 1m 23s\n- gate-tempest-dsvm-full http://logs.openstack.org/35/63435/3/gate/gate-tempest-dsvm-full/952bd5d : SUCCESS in 44m 38s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/35/63435/3/gate/gate-tempest-dsvm-postgres-full/d1a89d7 : SUCCESS in 53m 46s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/35/63435/3/gate/gate-tempest-dsvm-neutron/7e04d97 : SUCCESS in 37m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/35/63435/3/gate/gate-tempest-dsvm-large-ops/5195519 : SUCCESS in 14m 27s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/35/63435/3/gate/gate-tempest-dsvm-neutron-large-ops/3317cf2 : SUCCESS in 19m 17s\n- gate-grenade-dsvm http://logs.openstack.org/35/63435/3/gate/gate-grenade-dsvm/1195614 : SUCCESS in 29m 20s\n","accounts_in_message":[],"_revision_number":3},{"id":"43abd546dff77b241d87167e9b5dc645fc81bc54","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-12-20 21:53:07.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86bd8427e36d31346a74cac4e5115b0fc2734255","revisions":{"b8262597647fe3ac02fc5b5c908064ea1b46f5cc":{"kind":"REWORK","_number":1,"created":"2013-12-20 15:11:02.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/35/63435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/35/63435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/35/63435/1"}}},"commit":{"parents":[{"commit":"9ac0da7fb84d5aab3793e426277dbd559ef0e89a","subject":"Merge \"scenario/network_basic_ops: reassociate floating-ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ac0da7fb84d5aab3793e426277dbd559ef0e89a"}]}],"author":{"name":"Sean Dague","email":"sean.dague@samsung.com","date":"2013-12-20 14:56:44.000000000","tz":-300},"committer":{"name":"Sean Dague","email":"sean.dague@samsung.com","date":"2013-12-20 14:56:44.000000000","tz":-300},"subject":"wip: moving to global lazy loaded config","message":"wip: moving to global lazy loaded config\n\none of the problems we\u0027ve got with tempest is the fact that config\nloading is tied into the class hierarchy. However there is no\nreason why it should be. If we instead create a config proxy\nobject we can lazy load the actual config when we are executing,\nand not do it at import time.\n\nThis is going to take some iteration to get correct, but it\nstarts us with a clean pep8 run for the first time ever.\n\nChange-Id: I9b1bbfe231c85c01938bd68be4e5974bd24130d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8262597647fe3ac02fc5b5c908064ea1b46f5cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8262597647fe3ac02fc5b5c908064ea1b46f5cc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9ac0da7fb84d5aab3793e426277dbd559ef0e89a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"03665084fe79f7d91e02fe1eaee33b6d723dcb33":{"kind":"REWORK","_number":2,"created":"2013-12-20 16:46:42.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/35/63435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/35/63435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/35/63435/2"}}},"commit":{"parents":[{"commit":"9ac0da7fb84d5aab3793e426277dbd559ef0e89a","subject":"Merge \"scenario/network_basic_ops: reassociate floating-ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ac0da7fb84d5aab3793e426277dbd559ef0e89a"}]}],"author":{"name":"Sean Dague","email":"sean.dague@samsung.com","date":"2013-12-20 14:56:44.000000000","tz":-300},"committer":{"name":"Sean Dague","email":"sean.dague@samsung.com","date":"2013-12-20 16:46:33.000000000","tz":-300},"subject":"wip: moving to global lazy loaded config","message":"wip: moving to global lazy loaded config\n\none of the problems we\u0027ve got with tempest is the fact that config\nloading is tied into the class hierarchy. However there is no\nreason why it should be. If we instead create a config proxy\nobject we can lazy load the actual config when we are executing,\nand not do it at import time.\n\nThis is going to take some iteration to get correct, but it\nstarts us with a clean pep8 run for the first time ever.\n\nChange-Id: I9b1bbfe231c85c01938bd68be4e5974bd24130d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/03665084fe79f7d91e02fe1eaee33b6d723dcb33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/03665084fe79f7d91e02fe1eaee33b6d723dcb33"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9ac0da7fb84d5aab3793e426277dbd559ef0e89a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"86bd8427e36d31346a74cac4e5115b0fc2734255":{"kind":"REWORK","_number":3,"created":"2013-12-20 19:01:15.000000000","uploader":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"ref":"refs/changes/35/63435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/35/63435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/35/63435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/35/63435/3"}}},"commit":{"parents":[{"commit":"9ac0da7fb84d5aab3793e426277dbd559ef0e89a","subject":"Merge \"scenario/network_basic_ops: reassociate floating-ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ac0da7fb84d5aab3793e426277dbd559ef0e89a"}]}],"author":{"name":"Sean Dague","email":"sean.dague@samsung.com","date":"2013-12-20 14:56:44.000000000","tz":-300},"committer":{"name":"Sean Dague","email":"sean.dague@samsung.com","date":"2013-12-20 19:00:14.000000000","tz":-300},"subject":"moving to global lazy loaded config","message":"moving to global lazy loaded config\n\none of the problems we\u0027ve got with tempest is the fact that config\nloading is tied into the class hierarchy. However there is no\nreason why it should be. If we instead create a config proxy\nobject we can lazy load the actual config when we are executing,\nand not do it at import time.\n\nThis could use future iteration, but it does a huge job in\nremoving config from the object inheritance tree which massively\nsimplifies our ability to use config variables throughout the code.\n\nChange-Id: I9b1bbfe231c85c01938bd68be4e5974bd24130d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86bd8427e36d31346a74cac4e5115b0fc2734255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86bd8427e36d31346a74cac4e5115b0fc2734255"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9ac0da7fb84d5aab3793e426277dbd559ef0e89a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
