)]}'
{"id":"openstack%2Ftempest~639319","triplet_id":"openstack%2Ftempest~master~I9deeb0e527a8537c71657507cabd81a63a421d9c","project":"openstack/tempest","branch":"master","topic":"bug/1817701","hashtags":[],"change_id":"I9deeb0e527a8537c71657507cabd81a63a421d9c","subject":"Fix use of get_server_ip with fixed networks","status":"MERGED","created":"2019-02-26 12:01:58.000000000","updated":"2019-09-03 20:49:45.000000000","submitted":"2019-09-03 20:47:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"639319-1567543661592-a02f6a61","meta_rev_id":"0a99f5b939d02e5f6bfc2a0781a608c44bf50972","_number":639319,"virtual_id_number":639319,"owner":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"value":2,"date":"2019-09-03 20:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":2,"date":"2019-09-03 18:06:36.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2019-09-03 07:49:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":1,"date":"2019-09-03 18:06:36.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-27 05:39:29.000000000","updated_by":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"reviewer":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"state":"REVIEWER"},{"updated":"2019-03-07 02:15:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-09-02 13:44:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-03 07:49:27.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-09-03 18:06:36.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-09-03 20:47:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e01c97dec0f27ed21a7e53814415e20db56e3188","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2019-02-26 12:01:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddc226d9133008ce509924497c434a0813f74f77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 14:31:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/19/639319/1/check/tempest-full/ffb1deb/ : SUCCESS in 1h 58m 24s\n- neutron-grenade http://logs.openstack.org/19/639319/1/check/neutron-grenade/d85728b/ : SUCCESS in 54m 59s\n- grenade-py3 http://logs.openstack.org/19/639319/1/check/grenade-py3/f0839e3/ : SUCCESS in 56m 33s\n- tempest-full-py3 http://logs.openstack.org/19/639319/1/check/tempest-full-py3/33d0c83/ : SUCCESS in 1h 25m 54s\n- openstack-tox-cover http://logs.openstack.org/19/639319/1/check/openstack-tox-cover/dc0502c/cover/ : SUCCESS in 6m 28s\n- openstack-tox-pep8 http://logs.openstack.org/19/639319/1/check/openstack-tox-pep8/b06c91c/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/19/639319/1/check/openstack-tox-py27/081e75c/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/19/639319/1/check/openstack-tox-py35/2ba076a/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/19/639319/1/check/openstack-tox-py36/57a97fc/ : SUCCESS in 4m 00s\n- openstack-tox-py37 http://logs.openstack.org/19/639319/1/check/openstack-tox-py37/d6957cd/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/19/639319/1/check/openstack-tox-docs/48030f5/html/ : SUCCESS in 6m 40s\n- tempest-full-parallel http://logs.openstack.org/19/639319/1/check/tempest-full-parallel/0291b20/ : TIMED_OUT in 2h 05m 19s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/19/639319/1/check/tempest-full-py3-ipv6/143a4af/ : SUCCESS in 1h 20m 19s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/19/639319/1/check/tempest-full-rocky/ad8abdc/ : SUCCESS in 1h 46m 20s\n- tempest-full-rocky-py3 http://logs.openstack.org/19/639319/1/check/tempest-full-rocky-py3/3e745c7/ : SUCCESS in 1h 14m 39s\n- tempest-full-queens http://logs.openstack.org/19/639319/1/check/tempest-full-queens/782cad6/ : SUCCESS in 1h 37m 12s\n- tempest-full-queens-py3 http://logs.openstack.org/19/639319/1/check/tempest-full-queens-py3/9a72cfa/ : SUCCESS in 1h 13m 58s\n- tempest-full-pike http://logs.openstack.org/19/639319/1/check/tempest-full-pike/6021c77/ : SUCCESS in 1h 32m 00s\n- tempest-multinode-full http://logs.openstack.org/19/639319/1/check/tempest-multinode-full/00e9055/ : SUCCESS in 1h 50m 37s\n- tempest-multinode-full-py3 http://logs.openstack.org/19/639319/1/check/tempest-multinode-full-py3/64c4b4e/ : SUCCESS in 1h 43m 41s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/19/639319/1/check/tempest-tox-plugin-sanity-check/ed6eee9/ : FAILURE in 51m 15s (non-voting)\n- tempest-slow http://logs.openstack.org/19/639319/1/check/tempest-slow/69449c0/ : SUCCESS in 2h 13m 26s\n- tempest-slow-py3 http://logs.openstack.org/19/639319/1/check/tempest-slow-py3/514c911/ : SUCCESS in 2h 10m 16s\n- nova-live-migration http://logs.openstack.org/19/639319/1/check/nova-live-migration/bee5a4f/ : SUCCESS in 49m 45s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/19/639319/1/check/neutron-grenade-multinode/6e0b44c/ : SUCCESS in 1h 11m 42s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/19/639319/1/check/devstack-plugin-ceph-tempest/69f390e/ : SUCCESS in 1h 57m 01s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/639319/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/a296ad5/ : SUCCESS in 48m 10s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/19/639319/1/check/neutron-tempest-dvr/fdd873f/ : SUCCESS in 1h 18m 22s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/19/639319/1/check/legacy-tempest-dsvm-neutron-full-ocata/b8813e9/ : SUCCESS in 1h 18m 31s\n- interop-tempest-consistency http://logs.openstack.org/19/639319/1/check/interop-tempest-consistency/f2a0fab/ : SUCCESS in 5m 12s\n- tempest-full-test-account-py3 http://logs.openstack.org/19/639319/1/check/tempest-full-test-account-py3/fa4a637/ : SUCCESS in 1h 19m 45s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/19/639319/1/check/tempest-full-test-account-no-admin-py3/63c4b6f/ : SUCCESS in 1h 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75ba991a6831f87f09be396ee48faac15227c4b5","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2019-02-27 05:39:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7344c4939a4013e899312f46ed14a8512ce9be9f","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2019-02-27 06:13:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2bf58c5abcc9262e2cf97c5ed224519ecb766516","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2019-02-27 06:13:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1387e552d6e8d78214cccbe862b9c9adbaac509","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 09:41:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/19/639319/2/check/tempest-full/2ae88c2/ : SUCCESS in 1h 38m 35s\n- neutron-grenade http://logs.openstack.org/19/639319/2/check/neutron-grenade/228cd31/ : SUCCESS in 51m 37s\n- grenade-py3 http://logs.openstack.org/19/639319/2/check/grenade-py3/914dd57/ : SUCCESS in 1h 03m 09s\n- tempest-full-py3 http://logs.openstack.org/19/639319/2/check/tempest-full-py3/c114e42/ : SUCCESS in 1h 34m 30s\n- openstack-tox-cover http://logs.openstack.org/19/639319/2/check/openstack-tox-cover/0b90cb5/cover/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/19/639319/2/check/openstack-tox-pep8/8229890/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/19/639319/2/check/openstack-tox-py27/075cea1/ : SUCCESS in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/19/639319/2/check/openstack-tox-py35/ca211d6/ : SUCCESS in 6m 05s\n- openstack-tox-py36 http://logs.openstack.org/19/639319/2/check/openstack-tox-py36/a3caf5f/ : SUCCESS in 4m 53s\n- openstack-tox-py37 http://logs.openstack.org/19/639319/2/check/openstack-tox-py37/e66b0c2/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/19/639319/2/check/openstack-tox-docs/4c61708/html/ : SUCCESS in 4m 11s\n- tempest-full-parallel http://logs.openstack.org/19/639319/2/check/tempest-full-parallel/0446ccd/ : SUCCESS in 1h 57m 03s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/19/639319/2/check/tempest-full-py3-ipv6/166424d/ : SUCCESS in 1h 24m 20s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/19/639319/2/check/tempest-full-rocky/98a4568/ : SUCCESS in 1h 46m 06s\n- tempest-full-rocky-py3 http://logs.openstack.org/19/639319/2/check/tempest-full-rocky-py3/9ec8d82/ : SUCCESS in 1h 27m 48s\n- tempest-full-queens http://logs.openstack.org/19/639319/2/check/tempest-full-queens/4b66eba/ : SUCCESS in 1h 29m 49s\n- tempest-full-queens-py3 http://logs.openstack.org/19/639319/2/check/tempest-full-queens-py3/c9122a8/ : SUCCESS in 1h 17m 43s\n- tempest-full-pike http://logs.openstack.org/19/639319/2/check/tempest-full-pike/c1314d9/ : SUCCESS in 1h 38m 46s\n- tempest-multinode-full http://logs.openstack.org/19/639319/2/check/tempest-multinode-full/0213e7e/ : SUCCESS in 1h 42m 22s\n- tempest-multinode-full-py3 http://logs.openstack.org/19/639319/2/check/tempest-multinode-full-py3/fab7d70/ : SUCCESS in 1h 48m 09s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/19/639319/2/check/tempest-tox-plugin-sanity-check/230b3c9/ : FAILURE in 1h 04m 23s (non-voting)\n- tempest-slow http://logs.openstack.org/19/639319/2/check/tempest-slow/d0c2d8e/ : SUCCESS in 2h 34m 18s\n- tempest-slow-py3 http://logs.openstack.org/19/639319/2/check/tempest-slow-py3/0ef73c5/ : SUCCESS in 2h 15m 02s\n- nova-live-migration http://logs.openstack.org/19/639319/2/check/nova-live-migration/b6826db/ : SUCCESS in 45m 58s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/19/639319/2/check/neutron-grenade-multinode/b41ca44/ : SUCCESS in 1h 06m 57s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/19/639319/2/check/devstack-plugin-ceph-tempest/8b87b1d/ : SUCCESS in 2h 02m 46s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/19/639319/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/5c2397f/ : SUCCESS in 49m 35s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/19/639319/2/check/neutron-tempest-dvr/449ef5e/ : SUCCESS in 1h 54m 40s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/19/639319/2/check/legacy-tempest-dsvm-neutron-full-ocata/b4f176d/ : SUCCESS in 1h 17m 46s\n- interop-tempest-consistency http://logs.openstack.org/19/639319/2/check/interop-tempest-consistency/5deb1cc/ : SUCCESS in 4m 29s\n- tempest-full-test-account-py3 http://logs.openstack.org/19/639319/2/check/tempest-full-test-account-py3/e360e37/ : SUCCESS in 1h 21m 00s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/19/639319/2/check/tempest-full-test-account-no-admin-py3/23bee4d/ : SUCCESS in 1h 19m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e8e00270f1cfb462d57cb606ac313f7861752ea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-07 02:15:44.000000000","message":"Patch Set 2: Code-Review-1\n\nYeah, there are issue in few cases. I thought we always get complete server response and then pass it to get_server_ip.  \n\nNone of the gate job run test with fixed_ip as connect_method where get_server_ip method use the addresses from server response. While grep into whole code, along with test_server_actions.py, I found few more place which need similar fix\n\n- https://github.com/openstack/tempest/blob/91d92424bc49350ee6eff8dce0c6eb3fd18d8fb0/tempest/api/compute/servers/test_attach_interfaces.py\n\n-https://github.com/openstack/tempest/blob/1b43d266102735617e226152b66b8dbdecf1c759/tempest/api/compute/servers/test_server_actions.py#L104\n\n- https://github.com/openstack/tempest/blob/1b43d266102735617e226152b66b8dbdecf1c759/tempest/api/compute/servers/test_device_tagging.py#L370","accounts_in_message":[],"_revision_number":2},{"id":"18fe0825c6b757ac250172f25f2a19138a7d763d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 13:44:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a01ab51fffef3499856515816f075786b2d7208","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 13:56:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d8875ad8327a42e4b20365c1165fc0839a34ec26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 16:50:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f51b13914f894926adb2828351474125 : SUCCESS in 1h 57m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/93bd94ffae534719aaf0a0a1383c51f0 : SUCCESS in 51m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4dc97907d2dd4f6e8ab1ffce689c730c : SUCCESS in 1h 01m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0465aede8c848f3a0fa6e375ff8b980 : SUCCESS in 1h 31m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee5b34de98114b47afc4d3e686a0f695 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d2af37fd19438babf6a572a57553f6 : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff56d4fd39d44c9d9dd297d3a839de44 : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fdec1c75f3654f90b243b70c20b50988 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61a5e93f328741b68c62fc15509c96af : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8b8dc5c39a04d7186645eef8593fe84 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae24ea17536e4f0389c2ceb55715935d : SUCCESS in 9m 37s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f01ee8d284db48fbbf9404a084378cff : SUCCESS in 2h 06m 30s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c9a5757febc4407daf7b6cee49eb81c7 : SUCCESS in 1h 50m 55s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/72676235ad654c0fa12e9259fb3d6544 : SUCCESS in 2h 01m 11s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/356ae30724cb4eb2b2a6e341c272d92c : SUCCESS in 1h 30m 19s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/23bf0c82bda24829af0af2a2f57b0de1 : SUCCESS in 1h 32m 42s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/c4d3d84bf4ce487fa90c51caa06a8cee : SUCCESS in 1h 25m 52s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/15a61e080d9e46a5ae5d664a6ea6dde9 : SUCCESS in 1h 29m 09s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/58664db20e304f16a9c72aab96a5444a : SUCCESS in 1h 23m 16s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0b3ccf744a844f9bb9af04e8cfa72702 : SUCCESS in 1h 49m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1c7a5567827443e599d77c6d808be549 : SUCCESS in 1h 50m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/13d5be0788dc49cab56a9a9710a38dad : SUCCESS in 55m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b88cfc62da64c029eb45e8371310c2b : SUCCESS in 1h 09m 20s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/d4ffa2fc17174f2cbef7053217ff4731 : SUCCESS in 2h 40m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd491302248d489c84f9b5c1e519f4b1 : SUCCESS in 2h 24m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8d1be569ef954293b1b10ea34cdb0b37 : SUCCESS in 49m 35s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8ed5c7454364a53a1de5287e76a8a98 : SUCCESS in 1h 13m 58s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32b9c0fb41d44b2db67e86108dd4e9e6 : SUCCESS in 1h 24m 47s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/22a4e605c438486fb97fd89e108bc6d4 : FAILURE in 11m 05s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/61127486e7dd45afb540682d3987d845 : SUCCESS in 1h 24m 05s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b13212c0b03742ed939455d5a11a4aa6 : SUCCESS in 3m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c6023078e5d24451bb932ae77103df83 : SUCCESS in 1h 20m 12s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6320be72bc3942449bf78207244cb2df : SUCCESS in 1h 17m 06s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/75e4231888764a37a42cf516902ac113 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":4},{"id":"5377c54d4ab04f4fbb9ecf756c40fa8ff613aef6","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-03 07:49:27.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM. I think another possible solution is that calling show_server() in create_test_server(). But I think it\u0027s too much and confusing. So, this is better.","accounts_in_message":[],"_revision_number":4},{"id":"eaec36b6fd78bb6e2ad3266b94829e9866826c93","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-09-03 18:06:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"10e5cc5a6c5ad1e0e38ff36dbac0f854abb49e0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:07:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1c0beabfca583921ae9855004322845c0c0b815e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:47:41.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9822db133fe046ec960907c7e315fec8 : SUCCESS in 1h 37m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ed1bc4a996df41688d2f0094e4733f3e : SUCCESS in 51m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9bb81dd4af3a43a48f2a68964b2448c4 : SUCCESS in 1h 01m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b87eda2416f4b9484268d4cf9cd6c79 : SUCCESS in 1h 27m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0827171f543441a8c641fdf7d907edf : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e074d5b667648ec8bc10cd886650bc5 : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1ae3cee27c294ace8745a932fbb3196c : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af78fedd23504375be01468a1ced29e3 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e177e91f3a54bbd86f5355916a46f09 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e77f5de249a4c6bbde444eb21cba215 : SUCCESS in 7m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/616dcd0821eb468ab5867b17f60009ed : SUCCESS in 2h 29m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/224db395f1094da3993c783350de87e5 : SUCCESS in 1h 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64ddc5d915b947b883b5c71bd84c77e1 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":4},{"id":"6f50d7a39120decb8ca1af0b2198f7f2088cf2a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:47:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9a6716c1c73071dfc5198534b4cb41d326fa58b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:49:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b6072b985e423e964ff8c743133a68 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fa30d2fb51d40cb7ef527fa11ce5ab9dd81d6d94","revisions":{"3a5bb868d52a293c9a3b802d9c8d2ce67c4ca4db":{"kind":"REWORK","_number":1,"created":"2019-02-26 12:01:58.000000000","uploader":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"ref":"refs/changes/19/639319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/639319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/639319/1"}}},"commit":{"parents":[{"commit":"1a4b61911073437f6a28aeef1cc931403131019b","subject":"Merge \"Conditionally mark test_attach_detach_volume as slow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a4b61911073437f6a28aeef1cc931403131019b"}]}],"author":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-02-26 11:54:53.000000000","tz":540},"committer":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-02-26 11:59:14.000000000","tz":540},"subject":"Fix validaation test_resize_volume_backed_server_confirm()","message":"Fix validaation test_resize_volume_backed_server_confirm()\n\ntest_resize_volume_backed_server_confirm() in test_server_actions\nis trying to validate created server, but there was missing parameters\nto server be validated.\n\nAdd missing paramters when server created, and pass server object\nfrom show_server() to get_server_ip() since it internally use\n\u0027addresses\u0027 field when \u0027fixed\u0027 connection method used for validation.\n\nChange-Id: I9deeb0e527a8537c71657507cabd81a63a421d9c\nCloses-Bug: #1817701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3a5bb868d52a293c9a3b802d9c8d2ce67c4ca4db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3a5bb868d52a293c9a3b802d9c8d2ce67c4ca4db"}]},"branch":"refs/heads/master"},"8646de7addfdcf7200807a6792673b63d1018457":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-02-27 06:13:18.000000000","uploader":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"ref":"refs/changes/19/639319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/639319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/639319/2"}}},"commit":{"parents":[{"commit":"1a4b61911073437f6a28aeef1cc931403131019b","subject":"Merge \"Conditionally mark test_attach_detach_volume as slow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a4b61911073437f6a28aeef1cc931403131019b"}]}],"author":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-02-26 11:54:53.000000000","tz":540},"committer":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-02-27 06:12:51.000000000","tz":540},"subject":"Fix validation test_resize_volume_backed_server_confirm()","message":"Fix validation test_resize_volume_backed_server_confirm()\n\ntest_resize_volume_backed_server_confirm() in test_server_actions\nis trying to validate created server, but there was missing parameters\nto server be validated.\n\nAdd missing paramters when server created, and pass server object\nfrom show_server() to get_server_ip() since it internally use\n\u0027addresses\u0027 field when \u0027fixed\u0027 connection method used for validation.\n\nChange-Id: I9deeb0e527a8537c71657507cabd81a63a421d9c\nCloses-Bug: #1817701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8646de7addfdcf7200807a6792673b63d1018457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8646de7addfdcf7200807a6792673b63d1018457"}]},"branch":"refs/heads/master"},"5b97ab060c802db507da5f70d1055bcee498b398":{"kind":"REWORK","_number":3,"created":"2019-09-02 13:44:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/639319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/639319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/639319/3"}}},"commit":{"parents":[{"commit":"9e63a0323e48543c47d733ce45cafde45fc2ddcb","subject":"Merge \"Add group-based-policy to blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e63a0323e48543c47d733ce45cafde45fc2ddcb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-02 13:41:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-02 13:44:46.000000000","tz":60},"subject":"Fix use of get_server_ip with fixed networks","message":"Fix use of get_server_ip with fixed networks\n\nIf a fixed network is used, we see errors like the following in three\ncompute tests:\n\nKeyError: \u0027addresses\u0027\n\nThis is because get_server_ip expects to be called with a detailed\nserver response, but in these cases is called with a create server\nresponse, which does not contain the \u0027addresses\u0027 field.\n\nThis change fixes the issue by obtaining a detailed server response\nwhere necessary.\n\nChange-Id: I9deeb0e527a8537c71657507cabd81a63a421d9c\nCloses-Bug: #1817701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5b97ab060c802db507da5f70d1055bcee498b398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5b97ab060c802db507da5f70d1055bcee498b398"}]},"branch":"refs/heads/master"},"fa30d2fb51d40cb7ef527fa11ce5ab9dd81d6d94":{"kind":"REWORK","_number":4,"created":"2019-09-02 13:56:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/639319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/639319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/639319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/639319/4"}}},"commit":{"parents":[{"commit":"9e63a0323e48543c47d733ce45cafde45fc2ddcb","subject":"Merge \"Add group-based-policy to blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e63a0323e48543c47d733ce45cafde45fc2ddcb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-02 13:41:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-02 13:56:24.000000000","tz":60},"subject":"Fix use of get_server_ip with fixed networks","message":"Fix use of get_server_ip with fixed networks\n\nIf a fixed network is used, we see errors like the following in three\ncompute tests:\n\nKeyError: \u0027addresses\u0027\n\nThis is because get_server_ip expects to be called with a detailed\nserver response, but in these cases is called with a create server\nresponse, which does not contain the \u0027addresses\u0027 field.\n\nThis change fixes the issue by obtaining a detailed server response\nwhere necessary.\n\nChange-Id: I9deeb0e527a8537c71657507cabd81a63a421d9c\nCloses-Bug: #1817701\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fa30d2fb51d40cb7ef527fa11ce5ab9dd81d6d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fa30d2fb51d40cb7ef527fa11ce5ab9dd81d6d94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
