)]}'
{"id":"openstack%2Ftempest~639393","triplet_id":"openstack%2Ftempest~master~I275be9f4daddcfa5625757be3a0e9f4b0950239a","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I275be9f4daddcfa5625757be3a0e9f4b0950239a","subject":"DNM:Testing","status":"ABANDONED","created":"2019-02-26 17:24:21.000000000","updated":"2020-01-23 00:58:21.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d86dd7b7b20de62d3d7a925b3afd81552806a77","_number":639393,"virtual_id_number":639393,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-04-30 00:21:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-01 02:30:01.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 00:21:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aed55482fe896e60837a0ba097a333bff6603e48","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-02-26 17:24:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e486753137ca4f1b8d14c7ea3270150e244b4d99","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-02-26 17:24:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b146ad288be173c8e419a5bf1ea5abada5c20169","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 18:40:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/93/639393/1/check/requirements-check/37b2cc3/ : FAILURE in 2m 53s\n- tempest-full http://logs.openstack.org/93/639393/1/check/tempest-full/a498d58/ : FAILURE in 29m 23s\n- neutron-grenade http://logs.openstack.org/93/639393/1/check/neutron-grenade/d0f9fbc/ : FAILURE in 27m 10s\n- grenade-py3 http://logs.openstack.org/93/639393/1/check/grenade-py3/d05e0e3/ : FAILURE in 38m 17s\n- tempest-full-py3 http://logs.openstack.org/93/639393/1/check/tempest-full-py3/419419c/ : FAILURE in 28m 19s\n- openstack-tox-cover http://logs.openstack.org/93/639393/1/check/openstack-tox-cover/7ab81b6/ : FAILURE in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/93/639393/1/check/openstack-tox-pep8/d9a9f1d/ : FAILURE in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/93/639393/1/check/openstack-tox-py27/a58cd1a/ : FAILURE in 3m 22s\n- openstack-tox-py35 http://logs.openstack.org/93/639393/1/check/openstack-tox-py35/966e29c/ : FAILURE in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/93/639393/1/check/openstack-tox-py36/90d558e/ : FAILURE in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/93/639393/1/check/openstack-tox-py37/1e86e26/ : FAILURE in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/93/639393/1/check/openstack-tox-docs/da14106/ : FAILURE in 3m 23s\n- devstack-tempest http://logs.openstack.org/93/639393/1/check/devstack-tempest/9cc33c0/ : FAILURE in 31m 02s\n- devstack-tempest-ipv6 http://logs.openstack.org/93/639393/1/check/devstack-tempest-ipv6/54fdf46/ : FAILURE in 31m 48s (non-voting)\n- tempest-full-parallel http://logs.openstack.org/93/639393/1/check/tempest-full-parallel/385239f/ : FAILURE in 40m 06s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/93/639393/1/check/tempest-full-py3-ipv6/8282b03/ : FAILURE in 29m 05s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/93/639393/1/check/tempest-full-rocky/0dbcf74/ : FAILURE in 28m 25s\n- tempest-full-rocky-py3 http://logs.openstack.org/93/639393/1/check/tempest-full-rocky-py3/0130f7f/ : FAILURE in 27m 03s\n- tempest-full-queens http://logs.openstack.org/93/639393/1/check/tempest-full-queens/fa9483d/ : FAILURE in 28m 00s\n- tempest-full-queens-py3 http://logs.openstack.org/93/639393/1/check/tempest-full-queens-py3/a8e9ee4/ : FAILURE in 23m 11s\n- tempest-full-pike http://logs.openstack.org/93/639393/1/check/tempest-full-pike/bb167b8/ : FAILURE in 31m 47s\n- tempest-multinode-full http://logs.openstack.org/93/639393/1/check/tempest-multinode-full/e64a4a5/ : FAILURE in 31m 51s\n- tempest-multinode-full-py3 http://logs.openstack.org/93/639393/1/check/tempest-multinode-full-py3/3345e81/ : FAILURE in 32m 28s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/93/639393/1/check/tempest-tox-plugin-sanity-check/6648f0c/ : FAILURE in 1h 04m 28s (non-voting)\n- tempest-slow http://logs.openstack.org/93/639393/1/check/tempest-slow/6bb1230/ : FAILURE in 28m 29s\n- tempest-slow-py3 http://logs.openstack.org/93/639393/1/check/tempest-slow-py3/41ca257/ : FAILURE in 26m 39s\n- nova-live-migration http://logs.openstack.org/93/639393/1/check/nova-live-migration/b471094/ : FAILURE in 35m 21s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/93/639393/1/check/neutron-grenade-multinode/b7d8c3e/ : FAILURE in 32m 02s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/93/639393/1/check/devstack-plugin-ceph-tempest/2d0f8ab/ : FAILURE in 33m 29s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/93/639393/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/f337184/ : SUCCESS in 47m 13s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/93/639393/1/check/neutron-tempest-dvr/ec20fc6/ : FAILURE in 24m 13s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/93/639393/1/check/legacy-tempest-dsvm-neutron-full-ocata/6baebf2/ : FAILURE in 32m 06s\n- interop-tempest-consistency http://logs.openstack.org/93/639393/1/check/interop-tempest-consistency/013dc34/ : SUCCESS in 3m 58s\n- tempest-full-test-account-py3 http://logs.openstack.org/93/639393/1/check/tempest-full-test-account-py3/a514daa/ : FAILURE in 34m 05s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/93/639393/1/check/tempest-full-test-account-no-admin-py3/98d544c/ : FAILURE in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db8589fdc8d64b8f065f11b6d4e21708cabbfa37","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-02-27 18:16:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"84478da91198f3702d2afd2c1e9a3595286bfca2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-02-28 18:54:12.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"41815edca0cfd4a9327ef2c2a73ddb72fc6373eb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-02-28 18:54:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7c701e91216a71cc295f0d4f77e08833479e48b7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-02-28 18:54:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ff4195f580f680ced2046e8c73eb8beca77a6730","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 23:39:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/93/639393/3/check/tempest-full/5b3f042/ : SUCCESS in 1h 49m 21s\n- neutron-grenade http://logs.openstack.org/93/639393/3/check/neutron-grenade/5054e89/ : SUCCESS in 1h 02m 18s\n- grenade-py3 http://logs.openstack.org/93/639393/3/check/grenade-py3/ff6efb0/ : SUCCESS in 57m 29s\n- tempest-full-py3 http://logs.openstack.org/93/639393/3/check/tempest-full-py3/a530bc3/ : SUCCESS in 1h 24m 03s\n- openstack-tox-cover http://logs.openstack.org/93/639393/3/check/openstack-tox-cover/d497fef/cover/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/93/639393/3/check/openstack-tox-pep8/923b6e7/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/93/639393/3/check/openstack-tox-py27/6b70eca/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/93/639393/3/check/openstack-tox-py35/b9b2131/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/93/639393/3/check/openstack-tox-py36/3ced6e9/ : SUCCESS in 5m 12s\n- openstack-tox-py37 http://logs.openstack.org/93/639393/3/check/openstack-tox-py37/4a90549/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/93/639393/3/check/openstack-tox-docs/94e1472/html/ : SUCCESS in 7m 48s\n- devstack-tempest http://logs.openstack.org/93/639393/3/check/devstack-tempest/df961de/ : SUCCESS in 46m 42s\n- devstack-tempest-ipv6 http://logs.openstack.org/93/639393/3/check/devstack-tempest-ipv6/9b4f230/ : SUCCESS in 38m 00s (non-voting)\n- tempest-full-parallel http://logs.openstack.org/93/639393/3/check/tempest-full-parallel/7246440/ : TIMED_OUT in 2h 06m 52s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/93/639393/3/check/tempest-full-py3-ipv6/9964ca9/ : SUCCESS in 1h 42m 01s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/93/639393/3/check/tempest-full-rocky/2968e90/ : SUCCESS in 1h 36m 24s\n- tempest-full-rocky-py3 http://logs.openstack.org/93/639393/3/check/tempest-full-rocky-py3/e997976/ : SUCCESS in 1h 40m 16s\n- tempest-full-queens http://logs.openstack.org/93/639393/3/check/tempest-full-queens/a70c8a2/ : SUCCESS in 1h 38m 38s\n- tempest-full-queens-py3 http://logs.openstack.org/93/639393/3/check/tempest-full-queens-py3/6c28938/ : SUCCESS in 1h 18m 36s\n- tempest-full-pike http://logs.openstack.org/93/639393/3/check/tempest-full-pike/2710174/ : FAILURE in 1h 40m 15s\n- tempest-multinode-full http://logs.openstack.org/93/639393/3/check/tempest-multinode-full/75b81ce/ : SUCCESS in 2h 02m 37s\n- tempest-multinode-full-py3 http://logs.openstack.org/93/639393/3/check/tempest-multinode-full-py3/230528a/ : SUCCESS in 1h 47m 24s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/93/639393/3/check/tempest-tox-plugin-sanity-check/1c483e9/ : FAILURE in 1h 03m 29s (non-voting)\n- tempest-slow http://logs.openstack.org/93/639393/3/check/tempest-slow/46c68e1/ : SUCCESS in 2h 15m 13s\n- tempest-slow-py3 http://logs.openstack.org/93/639393/3/check/tempest-slow-py3/ca210ba/ : FAILURE in 2h 23m 47s\n- nova-live-migration http://logs.openstack.org/93/639393/3/check/nova-live-migration/21788e9/ : SUCCESS in 48m 24s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/93/639393/3/check/neutron-grenade-multinode/9c289d5/ : SUCCESS in 1h 01m 45s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/93/639393/3/check/devstack-plugin-ceph-tempest/e5abc6d/ : SUCCESS in 1h 31m 06s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/93/639393/3/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/d3080f3/ : SUCCESS in 59m 05s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/93/639393/3/check/neutron-tempest-dvr/fd303a8/ : SUCCESS in 1h 31m 02s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/93/639393/3/check/legacy-tempest-dsvm-neutron-full-ocata/598e096/ : SUCCESS in 1h 31m 42s\n- interop-tempest-consistency http://logs.openstack.org/93/639393/3/check/interop-tempest-consistency/1426c35/ : SUCCESS in 5m 27s\n- tempest-full-test-account-py3 http://logs.openstack.org/93/639393/3/check/tempest-full-test-account-py3/85eabd8/ : SUCCESS in 1h 24m 29s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/93/639393/3/check/tempest-full-test-account-no-admin-py3/b7e0ac1/ : SUCCESS in 1h 05m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c714c002c1db20f5c2868d6009263255b2c270cc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-03-01 02:30:01.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/93/639393/3/check/EMC_XtremIO_FC/0a0919e/EMC_XtremIO_FC/1774 : SUCCESS in 1h 27m 56s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/93/639393/3/check/EMC_XtremIO_iSCSI/a7f8944/EMC_XtremIO_iSCSI/1753 : SUCCESS in 1h 25m 42s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":3},{"id":"8fa4330d08506db1239c275669257ea886e9279f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-29 21:23:35.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3d59a4bcb42ba54bfbd51cb3531febdab945eb22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 21:23:51.000000000","message":"Patch Set 4: Verified-1\n\n(9 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  Unknown project git.openstack.org/openstack/oslo.cache\n\nThe error appears in the following job stanza:\n\n  job:\n      name: tempest-full\n      parent: devstack-tempest\n      # This currently works from stable/pike on.\n      # Before stable/pike, legacy version of tempest-full\n      # \u0027legacy-tempest-dsvm-neutron-full\u0027 run.\n      branches: ^(?!stable/ocata).*$\n      description: |\n        Base integration test with Neutron networking and py27.\n        Former names for this job where:\n          * legacy-tempest-dsvm-neutron-full\n          * gate-tempest-dsvm-neutron-full-ubuntu-xenial\n      required-projects:\n        - git.openstack.org/openstack/oslo.cache\n        - git.openstack.org/openstack/oslo.concurrency\n        - git.openstack.org/openstack/oslo.config\n      vars:\n        tox_venvlist: all\n        tempest_test_regex: |\n          (?x)\n          tempest\\.api\\.compute # Run only compute tests\n        devstack_localrc:\n          ENABLE_FILE_INJECTION: true\n          ENABLE_VOLUME_MULTIATTACH: true\n\n  in \"openstack/tempest/.zuul.yaml@master\", line 86, column 3","accounts_in_message":[],"_revision_number":4},{"id":"3ff00a8a27be3939fb40d13128d29d09ab80482c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-29 21:31:49.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"1a0e0b9a597e005622e8b66cb78729306e600840","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-29 21:32:59.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f0afb60816dac7095d487049312f83a77d0e39cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 00:21:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/93/639393/6/check/tempest-full/d09af02/ : TIMED_OUT in 2h 06m 02s\n- neutron-grenade http://logs.openstack.org/93/639393/6/check/neutron-grenade/7b6e9de/ : FAILURE in 23m 58s\n- grenade-py3 http://logs.openstack.org/93/639393/6/check/grenade-py3/b92ac4d/ : FAILURE in 28m 08s\n- tempest-full-py3 http://logs.openstack.org/93/639393/6/check/tempest-full-py3/43e1cac/ : SUCCESS in 1h 23m 41s\n- openstack-tox-cover http://logs.openstack.org/93/639393/6/check/openstack-tox-cover/cc0bb59/cover/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/93/639393/6/check/openstack-tox-pep8/f48a3f2/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/93/639393/6/check/openstack-tox-py27/59fef0e/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/93/639393/6/check/openstack-tox-py35/5e1f616/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/93/639393/6/check/openstack-tox-py36/d1787d1/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/93/639393/6/check/openstack-tox-py37/005655f/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/93/639393/6/check/openstack-tox-docs/f648640/html/ : SUCCESS in 8m 15s\n- devstack-tempest http://logs.openstack.org/93/639393/6/check/devstack-tempest/aab14e1/ : SUCCESS in 35m 22s\n- devstack-tempest-ipv6 http://logs.openstack.org/93/639393/6/check/devstack-tempest-ipv6/ecadf14/ : SUCCESS in 40m 21s (non-voting)\n- tempest-full-parallel http://logs.openstack.org/93/639393/6/check/tempest-full-parallel/9675c3b/ : FAILURE in 33m 25s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/93/639393/6/check/tempest-full-py3-ipv6/43e2d24/ : SUCCESS in 1h 32m 03s (non-voting)\n- tempest-full-stein http://logs.openstack.org/93/639393/6/check/tempest-full-stein/38683cd/ : TIMED_OUT in 2h 06m 50s\n- tempest-full-stein-py3 http://logs.openstack.org/93/639393/6/check/tempest-full-stein-py3/4feed49/ : SUCCESS in 1h 25m 32s\n- tempest-full-rocky http://logs.openstack.org/93/639393/6/check/tempest-full-rocky/3efb8f8/ : SUCCESS in 2h 03m 02s\n- tempest-full-rocky-py3 http://logs.openstack.org/93/639393/6/check/tempest-full-rocky-py3/444b291/ : SUCCESS in 1h 25m 44s\n- tempest-full-queens http://logs.openstack.org/93/639393/6/check/tempest-full-queens/78d385c/ : SUCCESS in 2h 07m 29s\n- tempest-full-queens-py3 http://logs.openstack.org/93/639393/6/check/tempest-full-queens-py3/4567766/ : SUCCESS in 1h 24m 29s\n- tempest-full-pike http://logs.openstack.org/93/639393/6/check/tempest-full-pike/c1e50a9/ : TIMED_OUT in 2h 11m 20s\n- tempest-multinode-full http://logs.openstack.org/93/639393/6/check/tempest-multinode-full/89314c6/ : SUCCESS in 1h 42m 12s\n- tempest-multinode-full-py3 http://logs.openstack.org/93/639393/6/check/tempest-multinode-full-py3/05f7261/ : SUCCESS in 2h 03m 32s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/93/639393/6/check/tempest-tox-plugin-sanity-check/796240d/ : FAILURE in 42m 31s (non-voting)\n- tempest-slow http://logs.openstack.org/93/639393/6/check/tempest-slow/29a51cd/ : SUCCESS in 2h 32m 40s\n- tempest-slow-py3 http://logs.openstack.org/93/639393/6/check/tempest-slow-py3/a7da267/ : SUCCESS in 2h 09m 58s\n- nova-live-migration http://logs.openstack.org/93/639393/6/check/nova-live-migration/04c3bd6/ : FAILURE in 28m 08s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/93/639393/6/check/neutron-grenade-multinode/e167371/ : FAILURE in 27m 50s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/93/639393/6/check/devstack-plugin-ceph-tempest/7d76da8/ : FAILURE in 1h 31m 15s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/93/639393/6/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/1c8fc9b/ : FAILURE in 21m 40s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/93/639393/6/check/neutron-tempest-dvr/8194164/ : SUCCESS in 1h 23m 01s\n- interop-tempest-consistency http://logs.openstack.org/93/639393/6/check/interop-tempest-consistency/56a9e22/ : SUCCESS in 3m 58s\n- tempest-full-test-account-py3 http://logs.openstack.org/93/639393/6/check/tempest-full-test-account-py3/c3bd77b/ : SUCCESS in 1h 36m 12s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/93/639393/6/check/tempest-full-test-account-no-admin-py3/1dd112d/ : SUCCESS in 1h 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9d86dd7b7b20de62d3d7a925b3afd81552806a77","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 00:58:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6b749b048b43a70668024a36c93880f663058897","revisions":{"1a68b9fe8ddbafbb6a170ebb38eaf8906d96649e":{"kind":"REWORK","_number":1,"created":"2019-02-26 17:24:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/639393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/639393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/639393/1"}}},"commit":{"parents":[{"commit":"6fc72aaa257d998c30d4fae687167992b9059547","subject":"Merge \"add python 3.7 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fc72aaa257d998c30d4fae687167992b9059547"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-02-26 17:24:05.000000000","tz":0},"committer":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-02-26 17:24:05.000000000","tz":0},"subject":"DNM:Testing","message":"DNM:Testing\n\nChange-Id: I275be9f4daddcfa5625757be3a0e9f4b0950239a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a68b9fe8ddbafbb6a170ebb38eaf8906d96649e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a68b9fe8ddbafbb6a170ebb38eaf8906d96649e"}]},"branch":"refs/heads/master"},"c7e518016f20f42b7945a256129c7945a3999cd5":{"kind":"REWORK","_number":2,"created":"2019-02-28 18:54:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/639393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/639393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/639393/2"}}},"commit":{"parents":[{"commit":"6fc72aaa257d998c30d4fae687167992b9059547","subject":"Merge \"add python 3.7 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fc72aaa257d998c30d4fae687167992b9059547"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-02-26 17:24:05.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-02-28 18:54:23.000000000","tz":0},"subject":"DNM:Testing","message":"DNM:Testing\n\nChange-Id: I275be9f4daddcfa5625757be3a0e9f4b0950239a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c7e518016f20f42b7945a256129c7945a3999cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c7e518016f20f42b7945a256129c7945a3999cd5"}]},"branch":"refs/heads/master"},"2ede2f3d8ed1fa7820dfea6f79feb79d30715044":{"kind":"REWORK","_number":3,"created":"2019-02-28 18:54:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/639393/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/639393/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/639393/3"}}},"commit":{"parents":[{"commit":"6fc72aaa257d998c30d4fae687167992b9059547","subject":"Merge \"add python 3.7 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fc72aaa257d998c30d4fae687167992b9059547"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-02-26 17:24:05.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-02-28 18:54:37.000000000","tz":0},"subject":"DNM:Testing","message":"DNM:Testing\n\nChange-Id: I275be9f4daddcfa5625757be3a0e9f4b0950239a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2ede2f3d8ed1fa7820dfea6f79feb79d30715044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2ede2f3d8ed1fa7820dfea6f79feb79d30715044"}]},"branch":"refs/heads/master"},"48b729c7be9e77eb66d6494f4d1b4236fcf5b5db":{"kind":"REWORK","_number":4,"created":"2019-04-29 21:23:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/639393/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/639393/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/639393/4"}}},"commit":{"parents":[{"commit":"6fc72aaa257d998c30d4fae687167992b9059547","subject":"Merge \"add python 3.7 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fc72aaa257d998c30d4fae687167992b9059547"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-02-26 17:24:05.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-29 21:23:32.000000000","tz":0},"subject":"DNM:Testing","message":"DNM:Testing\n\nChange-Id: I275be9f4daddcfa5625757be3a0e9f4b0950239a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/48b729c7be9e77eb66d6494f4d1b4236fcf5b5db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/48b729c7be9e77eb66d6494f4d1b4236fcf5b5db"}]},"branch":"refs/heads/master"},"3cf33fdfca28be5c3489c39cd56f711f88f29018":{"kind":"REWORK","_number":5,"created":"2019-04-29 21:31:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/639393/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/639393/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/639393/5"}}},"commit":{"parents":[{"commit":"6fc72aaa257d998c30d4fae687167992b9059547","subject":"Merge \"add python 3.7 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fc72aaa257d998c30d4fae687167992b9059547"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-02-26 17:24:05.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-29 21:31:45.000000000","tz":0},"subject":"DNM:Testing","message":"DNM:Testing\n\nChange-Id: I275be9f4daddcfa5625757be3a0e9f4b0950239a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3cf33fdfca28be5c3489c39cd56f711f88f29018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3cf33fdfca28be5c3489c39cd56f711f88f29018"}]},"branch":"refs/heads/master"},"6b749b048b43a70668024a36c93880f663058897":{"kind":"REWORK","_number":6,"created":"2019-04-29 21:32:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/93/639393/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/639393/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/639393/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/639393/6"}}},"commit":{"parents":[{"commit":"6fc72aaa257d998c30d4fae687167992b9059547","subject":"Merge \"add python 3.7 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fc72aaa257d998c30d4fae687167992b9059547"}]}],"author":{"name":"ghanshyam","email":"gmann@ghanshyammann.com","date":"2019-02-26 17:24:05.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-29 21:32:56.000000000","tz":0},"subject":"DNM:Testing","message":"DNM:Testing\n\nChange-Id: I275be9f4daddcfa5625757be3a0e9f4b0950239a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6b749b048b43a70668024a36c93880f663058897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6b749b048b43a70668024a36c93880f663058897"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
