)]}'
{"id":"openstack%2Ftempest~642401","triplet_id":"openstack%2Ftempest~master~I5c1e3b675d770bbeaeece676880c56d840646284","project":"openstack/tempest","branch":"master","topic":"useless_log","hashtags":[],"change_id":"I5c1e3b675d770bbeaeece676880c56d840646284","subject":"Remove LOG in misc.py","status":"MERGED","created":"2019-03-11 10:08:27.000000000","updated":"2019-03-15 02:49:24.000000000","submitted":"2019-03-15 02:49:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"642401-1552618164267-f6588e1f","meta_rev_id":"b61a59d298714f19badff1ee2275775bd143d687","_number":642401,"virtual_id_number":642401,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-03-13 06:01:33.000000000","_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2019-03-15 02:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2019-03-13 05:31:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":2,"date":"2019-03-14 02:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-03-15 00:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-03-15 00:38:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-12 00:27:24.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-03-13 05:31:52.000000000","updated_by":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"reviewer":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"state":"REVIEWER"},{"updated":"2019-03-13 06:01:33.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-14 02:07:53.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-03-15 00:38:00.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-03-15 02:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6304f06bb09827ac69e88782f8faee1545f4ff5a","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-03-11 10:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57470ee56aa9a1bada2eeab0013b763e474646f2","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2019-03-11 10:37:05.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks fine","accounts_in_message":[],"_revision_number":1},{"id":"b4f83afc997569a716d666ad21fe637c4bb33830","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-11 12:39:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/01/642401/1/check/tempest-full/71d1a06/ : SUCCESS in 1h 39m 07s\n- neutron-grenade http://logs.openstack.org/01/642401/1/check/neutron-grenade/f40f234/ : SUCCESS in 58m 47s\n- grenade-py3 http://logs.openstack.org/01/642401/1/check/grenade-py3/81c25c3/ : SUCCESS in 1h 01m 18s\n- tempest-full-py3 http://logs.openstack.org/01/642401/1/check/tempest-full-py3/01cedb6/ : SUCCESS in 1h 28m 40s\n- openstack-tox-cover http://logs.openstack.org/01/642401/1/check/openstack-tox-cover/6d3cabe/cover/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/01/642401/1/check/openstack-tox-pep8/f0d5338/ : FAILURE in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/01/642401/1/check/openstack-tox-py27/e126401/ : SUCCESS in 6m 01s\n- openstack-tox-py35 http://logs.openstack.org/01/642401/1/check/openstack-tox-py35/a9c48a3/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/01/642401/1/check/openstack-tox-py36/ff743dd/ : SUCCESS in 6m 00s\n- openstack-tox-py37 http://logs.openstack.org/01/642401/1/check/openstack-tox-py37/9701401/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/01/642401/1/check/openstack-tox-docs/920acd1/html/ : SUCCESS in 8m 17s\n- tempest-full-parallel http://logs.openstack.org/01/642401/1/check/tempest-full-parallel/138d4bf/ : SUCCESS in 1h 58m 48s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/01/642401/1/check/tempest-full-py3-ipv6/2756222/ : SUCCESS in 1h 30m 54s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/01/642401/1/check/tempest-full-rocky/eb28afc/ : SUCCESS in 1h 35m 30s\n- tempest-full-rocky-py3 http://logs.openstack.org/01/642401/1/check/tempest-full-rocky-py3/2819000/ : SUCCESS in 1h 13m 22s\n- tempest-full-queens http://logs.openstack.org/01/642401/1/check/tempest-full-queens/76ca806/ : SUCCESS in 1h 32m 05s\n- tempest-full-queens-py3 http://logs.openstack.org/01/642401/1/check/tempest-full-queens-py3/3c00ac9/ : SUCCESS in 1h 13m 42s\n- tempest-full-pike http://logs.openstack.org/01/642401/1/check/tempest-full-pike/aa27f1d/ : SUCCESS in 1h 35m 12s\n- tempest-multinode-full http://logs.openstack.org/01/642401/1/check/tempest-multinode-full/40af480/ : SUCCESS in 1h 42m 43s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/642401/1/check/tempest-multinode-full-py3/725485f/ : FAILURE in 1h 32m 47s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/01/642401/1/check/tempest-tox-plugin-sanity-check/6297f7d/ : SUCCESS in 52m 52s (non-voting)\n- tempest-slow http://logs.openstack.org/01/642401/1/check/tempest-slow/62e8588/ : SUCCESS in 2h 23m 50s\n- tempest-slow-py3 http://logs.openstack.org/01/642401/1/check/tempest-slow-py3/b007241/ : SUCCESS in 2h 03m 46s\n- nova-live-migration http://logs.openstack.org/01/642401/1/check/nova-live-migration/fe4cc77/ : SUCCESS in 49m 12s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/01/642401/1/check/neutron-grenade-multinode/b7cfe8e/ : SUCCESS in 1h 12m 54s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/01/642401/1/check/devstack-plugin-ceph-tempest/2993d44/ : SUCCESS in 1h 44m 38s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/01/642401/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/bebf14a/ : SUCCESS in 51m 05s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/01/642401/1/check/neutron-tempest-dvr/6f4ff41/ : SUCCESS in 1h 38m 44s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/01/642401/1/check/legacy-tempest-dsvm-neutron-full-ocata/5af36b1/ : SUCCESS in 1h 28m 03s\n- interop-tempest-consistency http://logs.openstack.org/01/642401/1/check/interop-tempest-consistency/4ef1006/ : SUCCESS in 6m 34s\n- tempest-full-test-account-py3 http://logs.openstack.org/01/642401/1/check/tempest-full-test-account-py3/d00b945/ : SUCCESS in 1h 28m 44s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/01/642401/1/check/tempest-full-test-account-no-admin-py3/7edaeec/ : SUCCESS in 1h 07m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f43bc3b049191eb65422df71b54b10f895f8c40a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-03-11 22:22:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/01/642401/1/check/EMC_XtremIO_FC/ae68d89/EMC_XtremIO_FC/1913 : SUCCESS in 1h 36m 53s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/01/642401/1/check/EMC_XtremIO_iSCSI/ec15d3e/EMC_XtremIO_iSCSI/1892 : SUCCESS in 1h 27m 21s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":1},{"id":"6c5010eea2c8cc622f751a433e92e9f696366242","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-03-12 00:27:24.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is pretty trivial and I would fast track but don\u0027t want to break https://docs.openstack.org/tempest/latest/REVIEWING.html#when-to-approve","accounts_in_message":[],"_revision_number":1},{"id":"1d1d77cb27d86c45e5f10209bc42284b22baae5c","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-03-13 01:00:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e3ec007ebe0d5a0d84a06018d4f40462d0a7c5a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 04:00:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/01/642401/2/check/tempest-full/9e632b5/ : SUCCESS in 1h 38m 45s\n- neutron-grenade http://logs.openstack.org/01/642401/2/check/neutron-grenade/a05afd7/ : SUCCESS in 57m 19s\n- grenade-py3 http://logs.openstack.org/01/642401/2/check/grenade-py3/2b2ab8e/ : SUCCESS in 58m 24s\n- tempest-full-py3 http://logs.openstack.org/01/642401/2/check/tempest-full-py3/5bf2d76/ : SUCCESS in 1h 32m 47s\n- openstack-tox-cover http://logs.openstack.org/01/642401/2/check/openstack-tox-cover/a7196e5/cover/ : SUCCESS in 5m 52s\n- openstack-tox-pep8 http://logs.openstack.org/01/642401/2/check/openstack-tox-pep8/1f52934/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/01/642401/2/check/openstack-tox-py27/43bf918/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/01/642401/2/check/openstack-tox-py35/5b70563/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/01/642401/2/check/openstack-tox-py36/a861087/ : SUCCESS in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/01/642401/2/check/openstack-tox-py37/4653279/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/01/642401/2/check/openstack-tox-docs/7869c88/html/ : SUCCESS in 7m 55s\n- tempest-full-parallel http://logs.openstack.org/01/642401/2/check/tempest-full-parallel/235a27c/ : SUCCESS in 1h 49m 00s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/01/642401/2/check/tempest-full-py3-ipv6/130d190/ : SUCCESS in 1h 20m 46s (non-voting)\n- tempest-full-rocky http://logs.openstack.org/01/642401/2/check/tempest-full-rocky/3f9bdab/ : SUCCESS in 1h 31m 33s\n- tempest-full-rocky-py3 http://logs.openstack.org/01/642401/2/check/tempest-full-rocky-py3/736b6e5/ : SUCCESS in 1h 24m 34s\n- tempest-full-queens http://logs.openstack.org/01/642401/2/check/tempest-full-queens/1dfc9de/ : SUCCESS in 1h 16m 39s\n- tempest-full-queens-py3 http://logs.openstack.org/01/642401/2/check/tempest-full-queens-py3/aa86807/ : SUCCESS in 1h 06m 53s\n- tempest-full-pike http://logs.openstack.org/01/642401/2/check/tempest-full-pike/7a976f9/ : SUCCESS in 1h 27m 09s\n- tempest-multinode-full http://logs.openstack.org/01/642401/2/check/tempest-multinode-full/ac0f741/ : SUCCESS in 1h 51m 49s\n- tempest-multinode-full-py3 http://logs.openstack.org/01/642401/2/check/tempest-multinode-full-py3/c9441e8/ : SUCCESS in 1h 47m 03s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/01/642401/2/check/tempest-tox-plugin-sanity-check/e422a53/ : SUCCESS in 55m 02s (non-voting)\n- tempest-slow http://logs.openstack.org/01/642401/2/check/tempest-slow/97671ad/ : SUCCESS in 2h 20m 06s\n- tempest-slow-py3 http://logs.openstack.org/01/642401/2/check/tempest-slow-py3/ca67540/ : SUCCESS in 2h 09m 32s\n- nova-live-migration http://logs.openstack.org/01/642401/2/check/nova-live-migration/909dadc/ : SUCCESS in 51m 38s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/01/642401/2/check/neutron-grenade-multinode/3d7441c/ : SUCCESS in 1h 14m 39s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/01/642401/2/check/devstack-plugin-ceph-tempest/8eed1bc/ : SUCCESS in 1h 36m 25s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/01/642401/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/c4121a3/ : SUCCESS in 48m 02s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/01/642401/2/check/neutron-tempest-dvr/556636e/ : SUCCESS in 1h 20m 58s\n- legacy-tempest-dsvm-neutron-full-ocata http://logs.openstack.org/01/642401/2/check/legacy-tempest-dsvm-neutron-full-ocata/456b17e/ : SUCCESS in 1h 18m 12s\n- interop-tempest-consistency http://logs.openstack.org/01/642401/2/check/interop-tempest-consistency/b17865d/ : SUCCESS in 4m 19s\n- tempest-full-test-account-py3 http://logs.openstack.org/01/642401/2/check/tempest-full-test-account-py3/cdb1789/ : SUCCESS in 1h 20m 46s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/01/642401/2/check/tempest-full-test-account-no-admin-py3/a8b20d6/ : SUCCESS in 1h 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22f3bab68638fc4be798867a3f3d7431e6093d1b","author":{"_account_id":27078,"name":"Manik Bindlish","email":"manikbindlish19@gmail.com","username":"manik"},"date":"2019-03-13 05:31:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"76afe9f9f544468d337571755de09617493d9cc3","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2019-03-13 06:01:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- EMC_XtremIO_FC http://publiclogs.emc.com/01/642401/2/check/EMC_XtremIO_FC/3e1d872/EMC_XtremIO_FC/1931 : SUCCESS in 1h 31m 56s\n- EMC_XtremIO_iSCSI http://publiclogs.emc.com/01/642401/2/check/EMC_XtremIO_iSCSI/a24f3b3/EMC_XtremIO_iSCSI/1911 : SUCCESS in 1h 26m 44s\n\nLeave a comment with \u0027run-dell-emc-xtremio\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-xio-ci","accounts_in_message":[],"_revision_number":2},{"id":"cb057465d0c093629f60803018742a494207b312","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-03-14 02:07:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a5c3d28f482621b968ab04195f125daded5a277a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-03-15 00:38:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"86696723ea9ea15e6d80cd151cd24013e7a0812c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 00:38:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57a7e9ac971733a6fc2e8d9e19851ca690942a84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 02:49:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/01/642401/2/gate/tempest-full/c39d86d/ : SUCCESS in 1h 43m 21s\n- neutron-grenade http://logs.openstack.org/01/642401/2/gate/neutron-grenade/8f2f828/ : SUCCESS in 53m 16s\n- grenade-py3 http://logs.openstack.org/01/642401/2/gate/grenade-py3/6324a88/ : SUCCESS in 58m 28s\n- tempest-full-py3 http://logs.openstack.org/01/642401/2/gate/tempest-full-py3/43a3f1b/ : SUCCESS in 1h 30m 50s\n- openstack-tox-pep8 http://logs.openstack.org/01/642401/2/gate/openstack-tox-pep8/e045c6e/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/01/642401/2/gate/openstack-tox-py27/1d98957/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/01/642401/2/gate/openstack-tox-py35/5c8e532/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/01/642401/2/gate/openstack-tox-py36/64cfe5d/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/01/642401/2/gate/openstack-tox-py37/57bfce7/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/01/642401/2/gate/openstack-tox-docs/5544c3c/html/ : SUCCESS in 6m 09s\n- tempest-slow-py3 http://logs.openstack.org/01/642401/2/gate/tempest-slow-py3/2bf39bd/ : SUCCESS in 2h 08m 03s\n- neutron-grenade-multinode http://logs.openstack.org/01/642401/2/gate/neutron-grenade-multinode/506192d/ : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":2},{"id":"30c6bf552cfe678e937c456ba1d7c8def65c4892","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-15 02:49:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19cb7836dc8c2174afc19ee07b9b15b6c0f07b5a","revisions":{"a3aab0620d84812fc775e97bd60beaffd0408bff":{"kind":"REWORK","_number":1,"created":"2019-03-11 10:08:27.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/01/642401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/642401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/642401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/642401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/642401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/642401/1"}}},"commit":{"parents":[{"commit":"868855466058ed4c18975340899b4f6b3945c6af","subject":"Merge \"Add unit tests for tempest cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/868855466058ed4c18975340899b4f6b3945c6af"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-03-11 10:07:16.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-03-11 10:07:16.000000000","tz":480},"subject":"Remove LOG in misc.py","message":"Remove LOG in misc.py\n\nLOG is not used in misc.py, so this is to remove it.\n\nChange-Id: I5c1e3b675d770bbeaeece676880c56d840646284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a3aab0620d84812fc775e97bd60beaffd0408bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a3aab0620d84812fc775e97bd60beaffd0408bff"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"868855466058ed4c18975340899b4f6b3945c6af","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"19cb7836dc8c2174afc19ee07b9b15b6c0f07b5a":{"kind":"REWORK","_number":2,"created":"2019-03-13 01:00:11.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/01/642401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/01/642401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/642401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/642401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/01/642401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/01/642401/2"}}},"commit":{"parents":[{"commit":"868855466058ed4c18975340899b4f6b3945c6af","subject":"Merge \"Add unit tests for tempest cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/868855466058ed4c18975340899b4f6b3945c6af"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-03-11 10:07:16.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-03-13 01:00:05.000000000","tz":0},"subject":"Remove LOG in misc.py","message":"Remove LOG in misc.py\n\nLOG is not used in misc.py, so this is to remove it.\n\nChange-Id: I5c1e3b675d770bbeaeece676880c56d840646284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/19cb7836dc8c2174afc19ee07b9b15b6c0f07b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/19cb7836dc8c2174afc19ee07b9b15b6c0f07b5a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"868855466058ed4c18975340899b4f6b3945c6af","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
