)]}'
{"id":"openstack%2Ftempest~656437","triplet_id":"openstack%2Ftempest~master~I05e10af75eb5235d5130f1dec9f0c2b13e418bb9","project":"openstack/tempest","branch":"master","topic":"tempest","hashtags":[],"change_id":"I05e10af75eb5235d5130f1dec9f0c2b13e418bb9","subject":"Move to opendev","status":"ABANDONED","created":"2019-04-30 03:36:30.000000000","updated":"2019-05-29 00:01:18.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7aa33865bc3e42e46b99f220e3a145fb6c57653d","_number":656437,"virtual_id_number":656437,"owner":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"date":"2019-05-21 12:36:13.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2019-05-21 12:28:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"disliked":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":-1,"date":"2019-05-23 07:49:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-21 09:15:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 18:56:07.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-05-08 14:47:23.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-21 09:15:27.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2019-05-21 12:28:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-21 12:36:13.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2019-05-23 07:49:59.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"}],"messages":[{"id":"203d03c775c3bfd4cd1795956e9f532cda7ef017","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-04-30 03:36:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29e93624c418ad2335b3ca8302e19dc982264c1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 06:09:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/37/656437/1/check/tempest-full/53b3d8f/ : SUCCESS in 1h 33m 24s\n- neutron-grenade http://logs.openstack.org/37/656437/1/check/neutron-grenade/60f72d7/ : SUCCESS in 52m 24s\n- grenade-py3 http://logs.openstack.org/37/656437/1/check/grenade-py3/efae953/ : SUCCESS in 56m 31s\n- tempest-full-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-py3/54b13cc/ : FAILURE in 1h 22m 16s\n- openstack-tox-cover http://logs.openstack.org/37/656437/1/check/openstack-tox-cover/5a5a604/cover/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/37/656437/1/check/openstack-tox-pep8/281101c/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/37/656437/1/check/openstack-tox-py27/2be8688/ : SUCCESS in 2m 57s\n- openstack-tox-py35 http://logs.openstack.org/37/656437/1/check/openstack-tox-py35/08ffd1e/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/37/656437/1/check/openstack-tox-py36/88e38e9/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/37/656437/1/check/openstack-tox-py37/3d1140a/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/37/656437/1/check/openstack-tox-docs/98a3b05/html/ : SUCCESS in 8m 17s\n- build-openstack-releasenotes http://logs.openstack.org/37/656437/1/check/build-openstack-releasenotes/71a937d/html/ : SUCCESS in 6m 39s\n- tempest-full-parallel http://logs.openstack.org/37/656437/1/check/tempest-full-parallel/ce50f7f/ : SUCCESS in 2h 00m 58s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/37/656437/1/check/tempest-full-py3-ipv6/13f9671/ : SUCCESS in 1h 37m 09s (non-voting)\n- tempest-full-stein http://logs.openstack.org/37/656437/1/check/tempest-full-stein/bd89981/ : SUCCESS in 1h 30m 51s\n- tempest-full-stein-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-stein-py3/94b6598/ : SUCCESS in 1h 39m 16s\n- tempest-full-rocky http://logs.openstack.org/37/656437/1/check/tempest-full-rocky/6c745e5/ : SUCCESS in 1h 19m 44s\n- tempest-full-rocky-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-rocky-py3/c80db19/ : SUCCESS in 1h 24m 51s\n- tempest-full-queens http://logs.openstack.org/37/656437/1/check/tempest-full-queens/adc102c/ : SUCCESS in 1h 25m 03s\n- tempest-full-queens-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-queens-py3/e1da1e7/ : SUCCESS in 1h 20m 36s\n- tempest-full-pike http://logs.openstack.org/37/656437/1/check/tempest-full-pike/7273798/ : SUCCESS in 1h 33m 16s\n- tempest-multinode-full http://logs.openstack.org/37/656437/1/check/tempest-multinode-full/4870f62/ : SUCCESS in 1h 58m 46s\n- tempest-multinode-full-py3 http://logs.openstack.org/37/656437/1/check/tempest-multinode-full-py3/dfb5cad/ : SUCCESS in 1h 59m 19s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/37/656437/1/check/tempest-tox-plugin-sanity-check/37e77a6/ : FAILURE in 44m 28s (non-voting)\n- tempest-slow http://logs.openstack.org/37/656437/1/check/tempest-slow/1dd0827/ : SUCCESS in 2h 28m 10s\n- tempest-slow-py3 http://logs.openstack.org/37/656437/1/check/tempest-slow-py3/7714d8b/ : SUCCESS in 2h 10m 13s\n- nova-live-migration http://logs.openstack.org/37/656437/1/check/nova-live-migration/4825067/ : SUCCESS in 46m 40s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/37/656437/1/check/neutron-grenade-multinode/3acdbe3/ : SUCCESS in 1h 10m 53s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/37/656437/1/check/devstack-plugin-ceph-tempest/4f5f3bf/ : SUCCESS in 1h 32m 40s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/37/656437/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/e0807dd/ : FAILURE in 22m 39s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/37/656437/1/check/neutron-tempest-dvr/82dab76/ : SUCCESS in 1h 36m 25s\n- interop-tempest-consistency http://logs.openstack.org/37/656437/1/check/interop-tempest-consistency/8ca901f/ : SUCCESS in 3m 27s\n- tempest-full-test-account-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-test-account-py3/435d886/ : SUCCESS in 1h 38m 13s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-test-account-no-admin-py3/869a63b/ : SUCCESS in 1h 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a449fa6ce9e2c4e6f818caafa9843a3acc849b37","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-05 02:32:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"131b4fde3ed8f26d26c1080095b4ddcc872f4a94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 05:04:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/37/656437/1/check/tempest-full/6519e39/ : SUCCESS in 1h 56m 45s\n- neutron-grenade http://logs.openstack.org/37/656437/1/check/neutron-grenade/8dbfe64/ : SUCCESS in 53m 49s\n- grenade-py3 http://logs.openstack.org/37/656437/1/check/grenade-py3/6050e99/ : SUCCESS in 56m 43s\n- tempest-full-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-py3/cb23012/ : SUCCESS in 1h 29m 44s\n- openstack-tox-cover http://logs.openstack.org/37/656437/1/check/openstack-tox-cover/0b5f68a/cover/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/37/656437/1/check/openstack-tox-pep8/025a366/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/37/656437/1/check/openstack-tox-py27/ea31744/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/37/656437/1/check/openstack-tox-py35/09073cd/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/37/656437/1/check/openstack-tox-py36/0362d84/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/37/656437/1/check/openstack-tox-py37/0d56a32/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/37/656437/1/check/openstack-tox-docs/a0c1cec/html/ : SUCCESS in 7m 48s\n- build-openstack-releasenotes http://logs.openstack.org/37/656437/1/check/build-openstack-releasenotes/8053fb4/html/ : SUCCESS in 6m 19s\n- tempest-full-parallel http://logs.openstack.org/37/656437/1/check/tempest-full-parallel/7886fa1/ : TIMED_OUT in 2h 05m 53s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/37/656437/1/check/tempest-full-py3-ipv6/4c8d5a2/ : SUCCESS in 1h 21m 23s (non-voting)\n- tempest-full-stein http://logs.openstack.org/37/656437/1/check/tempest-full-stein/1ae3e65/ : SUCCESS in 1h 54m 48s\n- tempest-full-stein-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-stein-py3/4a3a625/ : SUCCESS in 1h 31m 36s\n- tempest-full-rocky http://logs.openstack.org/37/656437/1/check/tempest-full-rocky/3f54aa0/ : SUCCESS in 1h 28m 38s\n- tempest-full-rocky-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-rocky-py3/4aa4e14/ : SUCCESS in 1h 36m 35s\n- tempest-full-queens http://logs.openstack.org/37/656437/1/check/tempest-full-queens/b4f68bb/ : SUCCESS in 1h 28m 18s\n- tempest-full-queens-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-queens-py3/44a901b/ : SUCCESS in 1h 14m 35s\n- tempest-full-pike http://logs.openstack.org/37/656437/1/check/tempest-full-pike/9c2fb7e/ : SUCCESS in 1h 32m 27s\n- tempest-multinode-full http://logs.openstack.org/37/656437/1/check/tempest-multinode-full/a10200d/ : SUCCESS in 1h 53m 21s\n- tempest-multinode-full-py3 http://logs.openstack.org/37/656437/1/check/tempest-multinode-full-py3/db4c3a9/ : SUCCESS in 1h 56m 22s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/37/656437/1/check/tempest-tox-plugin-sanity-check/78663ac/ : FAILURE in 40m 54s (non-voting)\n- tempest-slow http://logs.openstack.org/37/656437/1/check/tempest-slow/cd8975c/ : SUCCESS in 2h 17m 51s\n- tempest-slow-py3 http://logs.openstack.org/37/656437/1/check/tempest-slow-py3/a04b74e/ : SUCCESS in 2h 18m 21s\n- nova-live-migration http://logs.openstack.org/37/656437/1/check/nova-live-migration/5427356/ : SUCCESS in 49m 19s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/37/656437/1/check/neutron-grenade-multinode/e204b14/ : SUCCESS in 1h 05m 32s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/37/656437/1/check/devstack-plugin-ceph-tempest/182bb3e/ : SUCCESS in 1h 38m 19s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/37/656437/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/3ed2b70/ : FAILURE in 21m 57s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/37/656437/1/check/neutron-tempest-dvr/6f4b2a1/ : SUCCESS in 1h 27m 19s\n- interop-tempest-consistency http://logs.openstack.org/37/656437/1/check/interop-tempest-consistency/54eeab9/ : SUCCESS in 4m 50s\n- tempest-full-test-account-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-test-account-py3/59949dd/ : SUCCESS in 1h 43m 43s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/37/656437/1/check/tempest-full-test-account-no-admin-py3/c11236e/ : SUCCESS in 1h 22m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f9492c80b182ba54e968348597aaf0c3171f4b4","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-05 06:18:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fac58ad7d2e7e5849a1c56bd71d5e84fd912d8de","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-05-06 18:56:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6ec6f2bb80ff1480bad6928b60ed2bf5af691aa5","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-07 07:50:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"de51e2379046eae123b99213ea82b03f558448a9","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-07 07:53:03.000000000","message":"Patch Set 2:\n\n\u003e (5 comments)\n\n\nThanks. the requirement url has changed to opendev.org","accounts_in_message":[],"_revision_number":2},{"id":"fa355acd00ad26f8573b7f0f605b16f933980ac3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-05-07 08:13:57.000000000","message":"Patch Set 2: Code-Review-1\n\nThis is a subset of https://review.opendev.org/#/c/655081/","accounts_in_message":[],"_revision_number":2},{"id":"6b01a83447c368bac8a1f834af6c0e3cf95ac85a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 10:10:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/37/656437/2/check/tempest-full/e4a79e5/ : SUCCESS in 1h 51m 18s\n- neutron-grenade http://logs.openstack.org/37/656437/2/check/neutron-grenade/5a88aad/ : SUCCESS in 1h 03m 22s\n- grenade-py3 http://logs.openstack.org/37/656437/2/check/grenade-py3/4d9450f/ : SUCCESS in 1h 02m 37s\n- tempest-full-py3 http://logs.openstack.org/37/656437/2/check/tempest-full-py3/ecc14e6/ : SUCCESS in 1h 34m 05s\n- openstack-tox-cover http://logs.openstack.org/37/656437/2/check/openstack-tox-cover/42d30b0/cover/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/37/656437/2/check/openstack-tox-pep8/1a10349/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/37/656437/2/check/openstack-tox-py27/e9c48a9/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/37/656437/2/check/openstack-tox-py35/79ce1e2/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/37/656437/2/check/openstack-tox-py36/acbff15/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/37/656437/2/check/openstack-tox-py37/668201e/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/37/656437/2/check/openstack-tox-docs/e877e11/html/ : SUCCESS in 9m 21s\n- build-openstack-releasenotes http://logs.openstack.org/37/656437/2/check/build-openstack-releasenotes/0301d74/html/ : SUCCESS in 6m 00s\n- tempest-full-parallel http://logs.openstack.org/37/656437/2/check/tempest-full-parallel/5a5c82d/ : SUCCESS in 1h 43m 44s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/37/656437/2/check/tempest-full-py3-ipv6/abaff2a/ : SUCCESS in 1h 40m 10s (non-voting)\n- tempest-full-stein http://logs.openstack.org/37/656437/2/check/tempest-full-stein/85ca989/ : SUCCESS in 1h 38m 06s\n- tempest-full-stein-py3 http://logs.openstack.org/37/656437/2/check/tempest-full-stein-py3/5ddff7d/ : SUCCESS in 1h 29m 18s\n- tempest-full-rocky http://logs.openstack.org/37/656437/2/check/tempest-full-rocky/da6c5fb/ : SUCCESS in 1h 35m 53s\n- tempest-full-rocky-py3 http://logs.openstack.org/37/656437/2/check/tempest-full-rocky-py3/6fdaf71/ : SUCCESS in 1h 23m 53s\n- tempest-full-queens http://logs.openstack.org/37/656437/2/check/tempest-full-queens/4312370/ : SUCCESS in 1h 30m 20s\n- tempest-full-queens-py3 http://logs.openstack.org/37/656437/2/check/tempest-full-queens-py3/7aae47c/ : SUCCESS in 1h 20m 16s\n- tempest-full-pike http://logs.openstack.org/37/656437/2/check/tempest-full-pike/35dae6a/ : SUCCESS in 1h 44m 06s\n- tempest-multinode-full http://logs.openstack.org/37/656437/2/check/tempest-multinode-full/08a99a2/ : SUCCESS in 1h 46m 59s\n- tempest-multinode-full-py3 http://logs.openstack.org/37/656437/2/check/tempest-multinode-full-py3/e2ec4bb/ : SUCCESS in 1h 40m 27s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/37/656437/2/check/tempest-tox-plugin-sanity-check/3dc35fe/ : FAILURE in 34m 27s (non-voting)\n- tempest-slow http://logs.openstack.org/37/656437/2/check/tempest-slow/c48aa2e/ : SUCCESS in 2h 16m 33s\n- tempest-slow-py3 http://logs.openstack.org/37/656437/2/check/tempest-slow-py3/8ae5e34/ : SUCCESS in 2h 17m 47s\n- nova-live-migration http://logs.openstack.org/37/656437/2/check/nova-live-migration/131d701/ : SUCCESS in 48m 31s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/37/656437/2/check/neutron-grenade-multinode/506ee2c/ : SUCCESS in 1h 07m 47s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/37/656437/2/check/devstack-plugin-ceph-tempest/b220665/ : SUCCESS in 1h 25m 11s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/37/656437/2/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/a887e04/ : FAILURE in 21m 55s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/37/656437/2/check/neutron-tempest-dvr/35e5f2c/ : SUCCESS in 1h 40m 10s\n- interop-tempest-consistency http://logs.openstack.org/37/656437/2/check/interop-tempest-consistency/6f438f1/ : SUCCESS in 4m 04s\n- tempest-full-test-account-py3 http://logs.openstack.org/37/656437/2/check/tempest-full-test-account-py3/a30437c/ : SUCCESS in 1h 44m 09s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/37/656437/2/check/tempest-full-test-account-no-admin-py3/0621966/ : SUCCESS in 1h 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d65c55940b4e65cb59a5f51bfd03986a17ee5cf","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-08 14:47:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13c46e35f700697a2c33f7c2d13ba9bd4f45f85d","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-21 08:43:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dab87f68e78ffa0e0b018f1495338aeac3e0a299","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-05-21 09:15:27.000000000","message":"Patch Set 3: Code-Review+1\n\nIs the community discussion using this style? can you give a reference link? thanks.","accounts_in_message":[],"_revision_number":3},{"id":"203b3fa7e08374804c84406cc40494b3c6f1e8a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 12:28:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/37/656437/3/check/tempest-full/71137bb/ : SUCCESS in 1h 56m 23s\n- neutron-grenade http://logs.openstack.org/37/656437/3/check/neutron-grenade/e6ebbfd/ : SUCCESS in 54m 48s\n- grenade-py3 http://logs.openstack.org/37/656437/3/check/grenade-py3/a62bec5/ : SUCCESS in 59m 27s\n- tempest-full-py3 http://logs.openstack.org/37/656437/3/check/tempest-full-py3/234c776/ : SUCCESS in 1h 35m 12s\n- openstack-tox-cover http://logs.openstack.org/37/656437/3/check/openstack-tox-cover/8737396/cover/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/37/656437/3/check/openstack-tox-pep8/8d973c5/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/37/656437/3/check/openstack-tox-py27/6204a6a/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/37/656437/3/check/openstack-tox-py35/d539de5/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/37/656437/3/check/openstack-tox-py36/70cd66d/ : SUCCESS in 3m 38s\n- openstack-tox-py37 http://logs.openstack.org/37/656437/3/check/openstack-tox-py37/3d8d841/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/37/656437/3/check/openstack-tox-docs/929f0e5/html/ : SUCCESS in 8m 08s\n- build-openstack-releasenotes http://logs.openstack.org/37/656437/3/check/build-openstack-releasenotes/1280841/html/ : SUCCESS in 6m 54s\n- tempest-full-parallel http://logs.openstack.org/37/656437/3/check/tempest-full-parallel/06de781/ : SUCCESS in 2h 10m 36s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/37/656437/3/check/tempest-full-py3-ipv6/6b93251/ : SUCCESS in 1h 34m 54s (non-voting)\n- tempest-full-stein http://logs.openstack.org/37/656437/3/check/tempest-full-stein/cc14876/ : SUCCESS in 1h 22m 31s\n- tempest-full-stein-py3 http://logs.openstack.org/37/656437/3/check/tempest-full-stein-py3/8f653b1/ : SUCCESS in 1h 41m 25s\n- tempest-full-rocky http://logs.openstack.org/37/656437/3/check/tempest-full-rocky/c4260cb/ : SUCCESS in 1h 25m 55s\n- tempest-full-rocky-py3 http://logs.openstack.org/37/656437/3/check/tempest-full-rocky-py3/b5a3bf9/ : SUCCESS in 1h 21m 47s\n- tempest-full-queens http://logs.openstack.org/37/656437/3/check/tempest-full-queens/f92a2af/ : SUCCESS in 1h 27m 40s\n- tempest-full-queens-py3 http://logs.openstack.org/37/656437/3/check/tempest-full-queens-py3/051ec03/ : SUCCESS in 1h 19m 01s\n- tempest-full-pike http://logs.openstack.org/37/656437/3/check/tempest-full-pike/e2248fe/ : SUCCESS in 1h 50m 23s\n- tempest-multinode-full http://logs.openstack.org/37/656437/3/check/tempest-multinode-full/da558f1/ : SUCCESS in 1h 58m 34s\n- tempest-multinode-full-py3 http://logs.openstack.org/37/656437/3/check/tempest-multinode-full-py3/865ca55/ : SUCCESS in 1h 54m 26s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/37/656437/3/check/tempest-tox-plugin-sanity-check/c626e76/ : SUCCESS in 47m 31s (non-voting)\n- tempest-slow http://logs.openstack.org/37/656437/3/check/tempest-slow/f22622e/ : SUCCESS in 2h 27m 36s\n- tempest-slow-py3 http://logs.openstack.org/37/656437/3/check/tempest-slow-py3/2352b45/ : SUCCESS in 2h 11m 28s\n- nova-live-migration http://logs.openstack.org/37/656437/3/check/nova-live-migration/2c0319c/ : SUCCESS in 44m 27s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/37/656437/3/check/neutron-grenade-multinode/31aed67/ : SUCCESS in 1h 05m 03s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/37/656437/3/check/devstack-plugin-ceph-tempest/574bc45/ : SUCCESS in 1h 36m 24s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/37/656437/3/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/2012110/ : FAILURE in 9m 21s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/37/656437/3/check/neutron-tempest-dvr/3757749/ : SUCCESS in 1h 27m 59s\n- interop-tempest-consistency http://logs.openstack.org/37/656437/3/check/interop-tempest-consistency/70ffce9/ : SUCCESS in 3m 57s\n- tempest-full-test-account-py3 http://logs.openstack.org/37/656437/3/check/tempest-full-test-account-py3/0a59b57/ : SUCCESS in 1h 28m 55s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/37/656437/3/check/tempest-full-test-account-no-admin-py3/baa9996/ : SUCCESS in 1h 09m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93ef45008325994a4d54d3f1e47c226c828156b6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-05-21 12:36:13.000000000","message":"Patch Set 3:\n\nThis change is a subset of https://review.opendev.org/#/c/655081/, and should be thus discarded.","accounts_in_message":[],"_revision_number":3},{"id":"01cf64916f4cd7daeab7a2b9d8c2046c9df392ed","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-23 07:49:59.000000000","message":"Patch Set 3: Code-Review-1\n\nYeah, I agree with Luigi\u0027s comment.","accounts_in_message":[],"_revision_number":3},{"id":"0f167daae4258f5386a205bb6264111e18ff4b47","author":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"date":"2019-05-29 00:01:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a74f52ea06442e5a05831c5908fb8088c4f0a0a9","revisions":{"8b318a4896efe0a2f2135adf2ab8b4c843fb7b2b":{"kind":"REWORK","_number":1,"created":"2019-04-30 03:36:30.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/37/656437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/37/656437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/37/656437/1"}}},"commit":{"parents":[{"commit":"661a52ed4523ef3efab966e0986ecea918476c50","subject":"Merge \"Fix the bug of deleting security group after testing changing password.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/661a52ed4523ef3efab966e0986ecea918476c50"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-04-30 03:33:38.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-04-30 03:36:30.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I05e10af75eb5235d5130f1dec9f0c2b13e418bb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8b318a4896efe0a2f2135adf2ab8b4c843fb7b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8b318a4896efe0a2f2135adf2ab8b4c843fb7b2b"}]},"branch":"refs/heads/master"},"519f1ebe58fa7865bc8a90823f28ccab95c0d481":{"kind":"REWORK","_number":2,"created":"2019-05-07 07:50:10.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/37/656437/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/37/656437/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/37/656437/2"}}},"commit":{"parents":[{"commit":"661a52ed4523ef3efab966e0986ecea918476c50","subject":"Merge \"Fix the bug of deleting security group after testing changing password.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/661a52ed4523ef3efab966e0986ecea918476c50"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-04-30 03:33:38.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-07 07:50:00.000000000","tz":0},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I05e10af75eb5235d5130f1dec9f0c2b13e418bb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/519f1ebe58fa7865bc8a90823f28ccab95c0d481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/519f1ebe58fa7865bc8a90823f28ccab95c0d481"}]},"branch":"refs/heads/master"},"a74f52ea06442e5a05831c5908fb8088c4f0a0a9":{"kind":"REWORK","_number":3,"created":"2019-05-21 08:43:50.000000000","uploader":{"_account_id":28613,"name":"tony_zhang","email":"zhangyuan1@chinaunicom.cn","username":"tonybrad"},"ref":"refs/changes/37/656437/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/37/656437/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/37/656437/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/37/656437/3"}}},"commit":{"parents":[{"commit":"661a52ed4523ef3efab966e0986ecea918476c50","subject":"Merge \"Fix the bug of deleting security group after testing changing password.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/661a52ed4523ef3efab966e0986ecea918476c50"}]}],"author":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-04-30 03:33:38.000000000","tz":480},"committer":{"name":"tonybrad","email":"zhangyuan1@chinaunicom.cn","date":"2019-05-21 08:43:46.000000000","tz":0},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I05e10af75eb5235d5130f1dec9f0c2b13e418bb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a74f52ea06442e5a05831c5908fb8088c4f0a0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a74f52ea06442e5a05831c5908fb8088c4f0a0a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
