)]}'
{"id":"openstack%2Ftempest~670725","triplet_id":"openstack%2Ftempest~master~I61a9a106242b195ae499f182c3907a558b5ba032","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I61a9a106242b195ae499f182c3907a558b5ba032","subject":"DNM: extra logging for test_server_connectivity_cold_migration_revert","status":"ABANDONED","created":"2019-07-14 14:23:13.000000000","updated":"2019-07-22 20:45:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c3fdb04aec9647efc4aafe006e6d333435f089c","_number":670725,"virtual_id_number":670725,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-07-14 17:42:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-14 17:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e3fdbd30615929ff0eaf555d4df607f9a45432c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-07-14 14:23:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a86fd178a1722a3972857df96444e2b4bcc61589","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 17:42:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/25/670725/1/check/tempest-full/07ce486/ : TIMED_OUT in 2h 09m 21s\n- neutron-grenade http://logs.openstack.org/25/670725/1/check/neutron-grenade/d4fcf35/ : SUCCESS in 1h 06m 32s\n- grenade-py3 http://logs.openstack.org/25/670725/1/check/grenade-py3/45dfd5a/ : SUCCESS in 1h 02m 05s\n- tempest-full-py3 http://logs.openstack.org/25/670725/1/check/tempest-full-py3/9998d39/ : SUCCESS in 1h 40m 09s\n- openstack-tox-cover http://logs.openstack.org/25/670725/1/check/openstack-tox-cover/20f7c9f/cover/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/25/670725/1/check/openstack-tox-pep8/8d2175f/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/25/670725/1/check/openstack-tox-py27/6cf5a6b/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/25/670725/1/check/openstack-tox-py35/f3cb379/ : SUCCESS in 7m 24s\n- openstack-tox-py36 http://logs.openstack.org/25/670725/1/check/openstack-tox-py36/97114b4/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/25/670725/1/check/openstack-tox-py37/193e0a0/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/25/670725/1/check/openstack-tox-docs/f794b5f/html/ : SUCCESS in 15m 59s\n- tempest-full-parallel http://logs.openstack.org/25/670725/1/check/tempest-full-parallel/6baa6c3/ : SUCCESS in 2h 04m 11s (non-voting)\n- tempest-full-py3-ipv6 http://logs.openstack.org/25/670725/1/check/tempest-full-py3-ipv6/a094c55/ : SUCCESS in 1h 25m 09s (non-voting)\n- tempest-full-stein http://logs.openstack.org/25/670725/1/check/tempest-full-stein/6520757/ : SUCCESS in 1h 46m 38s\n- tempest-full-stein-py3 http://logs.openstack.org/25/670725/1/check/tempest-full-stein-py3/ff82740/ : SUCCESS in 1h 48m 05s\n- tempest-full-rocky http://logs.openstack.org/25/670725/1/check/tempest-full-rocky/9260d08/ : SUCCESS in 1h 29m 46s\n- tempest-full-rocky-py3 http://logs.openstack.org/25/670725/1/check/tempest-full-rocky-py3/99ca780/ : SUCCESS in 1h 24m 35s\n- tempest-full-queens http://logs.openstack.org/25/670725/1/check/tempest-full-queens/c1c3723/ : SUCCESS in 1h 28m 06s\n- tempest-full-queens-py3 http://logs.openstack.org/25/670725/1/check/tempest-full-queens-py3/353a217/ : SUCCESS in 1h 27m 10s\n- tempest-full-pike http://logs.openstack.org/25/670725/1/check/tempest-full-pike/781b9e9/ : SUCCESS in 1h 48m 58s\n- tempest-multinode-full http://logs.openstack.org/25/670725/1/check/tempest-multinode-full/6d0c2d8/ : SUCCESS in 1h 55m 15s\n- tempest-multinode-full-py3 http://logs.openstack.org/25/670725/1/check/tempest-multinode-full-py3/9efe69e/ : SUCCESS in 1h 53m 12s\n- tempest-tox-plugin-sanity-check http://logs.openstack.org/25/670725/1/check/tempest-tox-plugin-sanity-check/651e01e/ : SUCCESS in 59m 08s (non-voting)\n- tempest-slow http://logs.openstack.org/25/670725/1/check/tempest-slow/fabe598/ : SUCCESS in 2h 51m 56s\n- tempest-slow-py3 http://logs.openstack.org/25/670725/1/check/tempest-slow-py3/5d0c129/ : SUCCESS in 2h 26m 57s\n- nova-live-migration http://logs.openstack.org/25/670725/1/check/nova-live-migration/59e09e9/ : SUCCESS in 1h 08m 14s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/25/670725/1/check/neutron-grenade-multinode/27e6bf4/ : SUCCESS in 1h 15m 34s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/25/670725/1/check/devstack-plugin-ceph-tempest/4f3f067/ : SUCCESS in 1h 33m 03s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 http://logs.openstack.org/25/670725/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/52a7847/ : FAILURE in 29m 41s (non-voting)\n- neutron-tempest-dvr http://logs.openstack.org/25/670725/1/check/neutron-tempest-dvr/5e1723d/ : FAILURE in 1h 51m 17s\n- interop-tempest-consistency http://logs.openstack.org/25/670725/1/check/interop-tempest-consistency/ad686b8/ : SUCCESS in 3m 16s\n- tempest-full-test-account-py3 http://logs.openstack.org/25/670725/1/check/tempest-full-test-account-py3/e313f76/ : SUCCESS in 1h 25m 49s (non-voting)\n- tempest-full-test-account-no-admin-py3 http://logs.openstack.org/25/670725/1/check/tempest-full-test-account-no-admin-py3/072f137/ : SUCCESS in 1h 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c3fdb04aec9647efc4aafe006e6d333435f089c","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-07-22 20:45:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72af77f9337d99397a6482308e0f4fe6d4fb0082","revisions":{"72af77f9337d99397a6482308e0f4fe6d4fb0082":{"kind":"REWORK","_number":1,"created":"2019-07-14 14:23:13.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/25/670725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/25/670725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/25/670725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/25/670725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/25/670725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/25/670725/1"}}},"commit":{"parents":[{"commit":"d605315e7f7e1f71cc20437049daca6430f3cfb4","subject":"Merge \"Revert \"Skip test_server_connectivity_cold_migration_revert until fixed\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d605315e7f7e1f71cc20437049daca6430f3cfb4"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-07-14 14:22:30.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-07-14 14:22:30.000000000","tz":-240},"subject":"DNM: extra logging for test_server_connectivity_cold_migration_revert","message":"DNM: extra logging for test_server_connectivity_cold_migration_revert\n\nChange-Id: I61a9a106242b195ae499f182c3907a558b5ba032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/72af77f9337d99397a6482308e0f4fe6d4fb0082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/72af77f9337d99397a6482308e0f4fe6d4fb0082"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
