)]}'
{"id":"openstack%2Ftempest~675371","triplet_id":"openstack%2Ftempest~master~I8a3e11c606c21490462b6f3e4bad71c39a5e63bc","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I8a3e11c606c21490462b6f3e4bad71c39a5e63bc","subject":"Add a single resize test to the smoke set","status":"ABANDONED","created":"2019-08-08 15:00:50.000000000","updated":"2019-08-27 13:41:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"068d3d0167f2a9456573f9e6aa42368c1a876296","_number":675371,"virtual_id_number":675371,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"date":"2019-08-16 02:41:22.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"date":"2019-08-15 07:31:54.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-08-08 22:42:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-09 01:12:47.000000000","_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-08-14 14:26:27.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"date":"2019-08-16 01:29:54.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 22:42:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 01:12:47.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2019-08-14 14:26:27.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-08-16 01:29:54.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-08-16 02:41:22.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"f2a4a91f0638bd14370f663448a539a407d333fb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-08 15:00:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac49a925cce0f3a1fbf983c7da9f231082c71ad6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 22:42:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://logs.opendev.org/71/675371/1/check/tempest-full/89d0e6c/ : SUCCESS in 1h 37m 15s\n- neutron-grenade https://logs.opendev.org/71/675371/1/check/neutron-grenade/2f4e0d9/ : SUCCESS in 53m 42s\n- grenade-py3 https://logs.opendev.org/71/675371/1/check/grenade-py3/ab0a26a/ : SUCCESS in 57m 34s\n- tempest-full-py3 https://logs.opendev.org/71/675371/1/check/tempest-full-py3/fcaa936/ : SUCCESS in 1h 19m 03s\n- openstack-tox-cover https://logs.opendev.org/71/675371/1/check/openstack-tox-cover/1ef4e6a/cover/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://logs.opendev.org/71/675371/1/check/openstack-tox-pep8/68a08c7/ : SUCCESS in 7m 45s\n- openstack-tox-py27 https://logs.opendev.org/71/675371/1/check/openstack-tox-py27/4844459/ : SUCCESS in 4m 15s\n- openstack-tox-py35 https://logs.opendev.org/71/675371/1/check/openstack-tox-py35/664cc0e/ : SUCCESS in 4m 04s\n- openstack-tox-py36 https://logs.opendev.org/71/675371/1/check/openstack-tox-py36/e59c59c/ : SUCCESS in 5m 20s\n- openstack-tox-py37 https://logs.opendev.org/71/675371/1/check/openstack-tox-py37/f2a3133/ : SUCCESS in 4m 20s\n- openstack-tox-docs https://logs.opendev.org/71/675371/1/check/openstack-tox-docs/26c8573/html/ : SUCCESS in 5m 27s\n- tempest-full-parallel https://logs.opendev.org/71/675371/1/check/tempest-full-parallel/54a9daf/ : SUCCESS in 2h 01m 56s (non-voting)\n- tempest-full-py3-ipv6 https://logs.opendev.org/71/675371/1/check/tempest-full-py3-ipv6/9589900/ : SUCCESS in 1h 16m 13s (non-voting)\n- tempest-full-stein https://logs.opendev.org/71/675371/1/check/tempest-full-stein/c63bde7/ : SUCCESS in 1h 55m 55s\n- tempest-full-stein-py3 https://logs.opendev.org/71/675371/1/check/tempest-full-stein-py3/8500b4a/ : SUCCESS in 1h 38m 29s\n- tempest-full-rocky https://logs.opendev.org/71/675371/1/check/tempest-full-rocky/00de9bf/ : SUCCESS in 1h 37m 09s\n- tempest-full-rocky-py3 https://logs.opendev.org/71/675371/1/check/tempest-full-rocky-py3/8ae6607/ : SUCCESS in 1h 33m 01s\n- tempest-full-queens https://logs.opendev.org/71/675371/1/check/tempest-full-queens/bfd103f/ : SUCCESS in 1h 46m 18s\n- tempest-full-queens-py3 https://logs.opendev.org/71/675371/1/check/tempest-full-queens-py3/b480b3a/ : SUCCESS in 1h 27m 51s\n- tempest-multinode-full https://logs.opendev.org/71/675371/1/check/tempest-multinode-full/34e096b/ : SUCCESS in 1h 42m 34s\n- tempest-multinode-full-py3 https://logs.opendev.org/71/675371/1/check/tempest-multinode-full-py3/6db4efc/ : SUCCESS in 1h 40m 02s\n- tempest-tox-plugin-sanity-check https://logs.opendev.org/71/675371/1/check/tempest-tox-plugin-sanity-check/5396aa0/ : SUCCESS in 51m 43s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/71/675371/1/check/tempest-ipv6-only/8250dcf/ : SUCCESS in 1h 09m 05s\n- tempest-slow https://logs.opendev.org/71/675371/1/check/tempest-slow/a4aef9b/ : SUCCESS in 2h 58m 59s\n- tempest-slow-py3 https://logs.opendev.org/71/675371/1/check/tempest-slow-py3/e01e001/ : SUCCESS in 2h 14m 57s\n- nova-live-migration https://logs.opendev.org/71/675371/1/check/nova-live-migration/4480b87/ : SUCCESS in 47m 40s (non-voting)\n- neutron-grenade-multinode https://logs.opendev.org/71/675371/1/check/neutron-grenade-multinode/70de8b4/ : SUCCESS in 1h 15m 34s\n- devstack-plugin-ceph-tempest https://logs.opendev.org/71/675371/1/check/devstack-plugin-ceph-tempest/290ceee/ : SUCCESS in 1h 20m 29s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://logs.opendev.org/71/675371/1/check/puppet-openstack-integration-4-scenario004-tempest-centos-7/c05f7a9/ : FAILURE in 23m 43s (non-voting)\n- neutron-tempest-dvr https://logs.opendev.org/71/675371/1/check/neutron-tempest-dvr/a4d67af/ : SUCCESS in 1h 15m 11s\n- interop-tempest-consistency https://logs.opendev.org/71/675371/1/check/interop-tempest-consistency/7979db9/ : SUCCESS in 3m 57s\n- tempest-full-test-account-py3 https://logs.opendev.org/71/675371/1/check/tempest-full-test-account-py3/dd6b808/ : SUCCESS in 1h 22m 25s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://logs.opendev.org/71/675371/1/check/tempest-full-test-account-no-admin-py3/e8b2292/ : SUCCESS in 1h 12m 33s (non-voting)\n- openstack-tox-bashate https://logs.opendev.org/71/675371/1/check/openstack-tox-bashate/59c0cd7/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7fad16748d487b18f7bcfb4778583203120a70e5","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-08-09 01:12:47.000000000","message":"Patch Set 1:\n\nDoes anyone have a good definition of what makes a smoke test?  IMO I thought it would be the minimal amount of tests that make sure that the micro services and resources like  rabbitmq and database are configured and talking to everyone.","accounts_in_message":[],"_revision_number":1},{"id":"5f4d5bd32bc6ca0ffe0d161b426e8329b2d1a205","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-09 08:43:19.000000000","message":"Patch Set 1:\n\n\u003e Does anyone have a good definition of what makes a smoke test?  IMO\n \u003e I thought it would be the minimal amount of tests that make sure\n \u003e that the micro services and resources like  rabbitmq and database\n \u003e are configured and talking to everyone.\nI found this definition in [1]:\nThe type\u003d\u0027smoke\u0027 attribute is used to signify that a test is a so-called smoke test, which is a type of test that tests the most vital OpenStack functionality, like listing servers or flavors or creating volumes. The attribute should be sparingly applied to only the tests that sanity-check the most essential functionality of an OpenStack cloud.\n\n\nAlthough it does not really help to decide if resize / migrate is an essential operation or not. In a cloud that needs to be ever upgraded some form of move operation is needed. I could enable a migrate test instead but resize and migrate is basically the same in the background except the fact that resize not just move the server but also changes the flavor of it. So resize give us a tiny bit of more coverage for the same price as migrate.\n\n[1] https://docs.openstack.org/tempest/latest/HACKING.html#smoke-attribute","accounts_in_message":[],"_revision_number":1},{"id":"35cd3b24e4b11b32e4e87888040810aa12699cf4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-14 14:26:27.000000000","message":"Patch Set 1:\n\nI agree with gibi that move operations like cold migration are important for upgrades. Live migration is more important if you support it, and we have the nova-grenade-live-migration job to test live migration across mixed service version computes, but that doesn\u0027t test the resize/cold migrate flows at all, which is why this came up in the referenced nova change in the commit message (which would have otherwise broke resize/cold migration flows with a Stein level compute).\n\nSmoke tests are also usually supposed to be pretty quick, and the one tagged here is relatively fast (results from 3 different jobs):\n\n2019-08-08 20:33:53.253320 | primary | {1} tempest.api.compute.servers.test_server_actions.ServerActionsTestJSON.test_resize_server_confirm [34.904545s] ... ok\n\n2019-08-08 20:34:03.091064 | primary | {1} tempest.api.compute.servers.test_server_actions.ServerActionsTestJSON.test_resize_server_confirm [23.687867s] ... ok\n\n2019-08-08 20:56:23.292393 | primary | {1} tempest.api.compute.servers.test_server_actions.ServerActionsTestJSON.test_resize_server_confirm [31.236877s] ... ok\n\nAlternatives to this would be if we can somehow run both smoke and a test_resize_server_confirm in a specific grenade job, like grenade-py3. That is maybe possible using the tempest_test_regex attribute in the zuul job config.\n\nAnother alternative might be to run test_resize_server_confirm explicitly in the nova-grenade-live-migration job, but then that job is no longer only doing live migration (though the nova-live-migration job has been doing evacuate testing for awhile as well, so maybe it\u0027s time we rename those jobs to nova-grenade-multinode and nova-multinode, respectively).\n\nNote that this change won\u0027t cover reverting a resize/cold migration which was also potentially broken in the referenced nova change, so if we want resize+revert testing in a grenade job and want to avoid marking that as smoke as well, we should probably consider one of the alternatives above.","accounts_in_message":[],"_revision_number":1},{"id":"e2772508bedf4b1b9c898a14e0f63f0ad57daa19","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-15 07:31:54.000000000","message":"Patch Set 1:\n\n\u003e I agree with gibi that move operations like cold migration are\n \u003e important for upgrades. Live migration is more important if you\n \u003e support it, and we have the nova-grenade-live-migration job to test\n \u003e live migration across mixed service version computes, but that\n \u003e doesn\u0027t test the resize/cold migrate flows at all, which is why\n \u003e this came up in the referenced nova change in the commit message\n \u003e (which would have otherwise broke resize/cold migration flows with\n \u003e a Stein level compute).\n \u003e \n \u003e Smoke tests are also usually supposed to be pretty quick, and the\n \u003e one tagged here is relatively fast (results from 3 different jobs):\n \u003e \n \u003e 2019-08-08 20:33:53.253320 | primary | {1} tempest.api.compute.servers.test_server_actions.ServerActionsTestJSON.test_resize_server_confirm\n \u003e [34.904545s] ... ok\n \u003e \n \u003e 2019-08-08 20:34:03.091064 | primary | {1} tempest.api.compute.servers.test_server_actions.ServerActionsTestJSON.test_resize_server_confirm\n \u003e [23.687867s] ... ok\n \u003e \n \u003e 2019-08-08 20:56:23.292393 | primary | {1} tempest.api.compute.servers.test_server_actions.ServerActionsTestJSON.test_resize_server_confirm\n \u003e [31.236877s] ... ok\n \u003e \n \u003e Alternatives to this would be if we can somehow run both smoke and\n \u003e a test_resize_server_confirm in a specific grenade job, like\n \u003e grenade-py3. That is maybe possible using the tempest_test_regex\n \u003e attribute in the zuul job config.\n\ngrenade-py3 isn\u0027t multinode. :/ So I guess test_resize_server_confirm would be skipped.\n\n \u003e \n \u003e Another alternative might be to run test_resize_server_confirm\n \u003e explicitly in the nova-grenade-live-migration job, but then that\n \u003e job is no longer only doing live migration (though the\n \u003e nova-live-migration job has been doing evacuate testing for awhile\n \u003e as well, so maybe it\u0027s time we rename those jobs to\n \u003e nova-grenade-multinode and nova-multinode, respectively).\n\nIf the currently proposed patch is a no-go, then I for me the renaming the nova-grenade-live-migration job to nova-grenade-multinode and adding these test explicitly there would the preferred choice. \n\n \u003e \n \u003e Note that this change won\u0027t cover reverting a resize/cold migration\n \u003e which was also potentially broken in the referenced nova change, so\n \u003e if we want resize+revert testing in a grenade job and want to avoid\n \u003e marking that as smoke as well, we should probably consider one of\n \u003e the alternatives above.","accounts_in_message":[],"_revision_number":1},{"id":"95ec5ac1e18d5d7897bf01df9c7c92fbffebed21","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-16 02:41:22.000000000","message":"Patch Set 1:\n\nI agree that resize operation is not so basic operation and we should not make it smoke. many jobs (recent ipv6 job) run the smoke tests and many more might be running in gate as well as in production. This can affect them.\n\nI will suggest to go for the solution of running required tests on job side via tox env or tempest regex.","accounts_in_message":[],"_revision_number":1},{"id":"068d3d0167f2a9456573f9e6aa42368c1a876296","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-27 13:41:23.000000000","message":"Abandoned\n\nAbandon in favor of https://review.opendev.org/678841","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1823a487ecaa48c7f5540d556046d09e1130439c","revisions":{"1823a487ecaa48c7f5540d556046d09e1130439c":{"kind":"REWORK","_number":1,"created":"2019-08-08 15:00:50.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/71/675371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/675371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/675371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/675371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/675371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/675371/1"}}},"commit":{"parents":[{"commit":"ba2b8ed37e58c881ea13953f4e4165b8a3db100a","subject":"Merge \"add show detail for volume Api V3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ba2b8ed37e58c881ea13953f4e4165b8a3db100a"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-08-08 14:52:02.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-08-08 14:52:02.000000000","tz":120},"subject":"Add a single resize test to the smoke set","message":"Add a single resize test to the smoke set\n\nTo cover resize and migrate RPC changes in nova with grenade this patch\ntags test_resize_server_confirm with \"smoke\" as grenade runs the smoke\nset. This come up with in the nova review [1].\n\n[1]https://review.opendev.org/#/c/655721/10\n\nChange-Id: I8a3e11c606c21490462b6f3e4bad71c39a5e63bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1823a487ecaa48c7f5540d556046d09e1130439c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1823a487ecaa48c7f5540d556046d09e1130439c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
