)]}'
{"id":"openstack%2Ftempest~675523","triplet_id":"openstack%2Ftempest~master~I44ac1ae576740e310d9b26beb436ac9c31082729","project":"openstack/tempest","branch":"master","topic":"parallel-sanity","hashtags":[],"change_id":"I44ac1ae576740e310d9b26beb436ac9c31082729","subject":"[WIP] Make tempest sanity faster","status":"ABANDONED","created":"2019-08-09 05:42:28.000000000","updated":"2022-12-24 05:46:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"da8cfc8cab3c22c63cd181d3de1d0879ece5150d","_number":675523,"virtual_id_number":675523,"owner":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2019-08-09 06:46:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":-1,"date":"2019-08-09 05:59:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 06:46:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"410772bb85d77bf43669327ee9eac099338b35de","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-08-09 05:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fbfcc83bc187ec1ff2efbe58b1713551057d2e2","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-08-09 05:59:15.000000000","message":"Patch Set 1: Workflow-1\n\nThis is not good.. I should think about this more carefully.","accounts_in_message":[],"_revision_number":1},{"id":"5389f8386adecab2aeef1da609971545e2996db8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 06:46:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/23/675523/1/check/openstack-tox-cover/8368011/cover/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://logs.opendev.org/23/675523/1/check/openstack-tox-pep8/59419cc/ : SUCCESS in 5m 53s\n- openstack-tox-py27 https://logs.opendev.org/23/675523/1/check/openstack-tox-py27/8d104b2/ : SUCCESS in 3m 31s\n- openstack-tox-py35 https://logs.opendev.org/23/675523/1/check/openstack-tox-py35/86d1d58/ : SUCCESS in 3m 58s\n- openstack-tox-py36 https://logs.opendev.org/23/675523/1/check/openstack-tox-py36/c6d6f3d/ : SUCCESS in 4m 30s\n- openstack-tox-py37 https://logs.opendev.org/23/675523/1/check/openstack-tox-py37/517f4bb/ : SUCCESS in 5m 09s\n- openstack-tox-docs https://logs.opendev.org/23/675523/1/check/openstack-tox-docs/06ef849/html/ : SUCCESS in 3m 57s\n- tempest-tox-plugin-sanity-check https://logs.opendev.org/23/675523/1/check/tempest-tox-plugin-sanity-check/e50eef2/ : SUCCESS in 12m 00s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/23/675523/1/check/tempest-ipv6-only/933d534/ : SUCCESS in 1h 02m 28s\n- openstack-tox-bashate https://logs.opendev.org/23/675523/1/check/openstack-tox-bashate/cfecff6/ : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"da8cfc8cab3c22c63cd181d3de1d0879ece5150d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-24 05:46:23.000000000","message":"Abandoned\n\nNo activities on this since long time, feel free to restore it if needed.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee283b6afc47a924ab6f639c8176639fd3dd69ee","revisions":{"ee283b6afc47a924ab6f639c8176639fd3dd69ee":{"kind":"REWORK","_number":1,"created":"2019-08-09 05:42:28.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/23/675523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/23/675523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/675523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/675523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/675523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/23/675523/1"}}},"commit":{"parents":[{"commit":"411f4bbc5c1a52acebf1e5ec13d5f98943e99b19","subject":"Retry http requests to generate plugins list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/411f4bbc5c1a52acebf1e5ec13d5f98943e99b19"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-08-09 05:38:34.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-08-09 05:38:34.000000000","tz":540},"subject":"[WIP] Make tempest sanity faster","message":"[WIP] Make tempest sanity faster\n\nThis commit makes the tempest sanity job faster with parallel execution.\nThis change might increase about 50 processes compared to the original.\nBut it should be much faster than the original.\n\nChange-Id: I44ac1ae576740e310d9b26beb436ac9c31082729\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ee283b6afc47a924ab6f639c8176639fd3dd69ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ee283b6afc47a924ab6f639c8176639fd3dd69ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
