)]}'
{"id":"openstack%2Ftempest~678198","triplet_id":"openstack%2Ftempest~master~I9f88d5f70711a50f74b960353d583b10f434c9fc","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I9f88d5f70711a50f74b960353d583b10f434c9fc","subject":"DNM:testing account jobs","status":"ABANDONED","created":"2019-08-23 10:49:15.000000000","updated":"2020-12-29 21:35:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"098cdd70d550fe7366f5a919a7b7c5924df58386","_number":678198,"virtual_id_number":678198,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-01 08:15:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-01 08:15:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"636bb988788d2bf47c661b140e8a9addd3a12086","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-23 10:49:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c2364ad42eda77a3bd33113a36911978404b7ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/13ce779b088e4663ad4052a9d9220973 : SUCCESS in 1h 39m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a330fa6823fb41bca3e61be15f6e0749 : SUCCESS in 1h 08m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d77578b8256a4843bc3e5b91a7f080a2 : SUCCESS in 1h 10m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4836e29f02347f1adf45e7da2d6bf74 : SUCCESS in 1h 21m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc8f8b7b4beb456bb51ac8bcd6130a1b : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e5b1435aa9d47e897f7d1c913f23e5d : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb106acd6a6d45b4b38476cfd2d1aa86 : SUCCESS in 7m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f59ffe9ab84e4a5d89934612c81ac630 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4880061e48c644d0bc95a163ddd2937e : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7ee27605d7945f88b3ec52df038e8ce : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00677e2b7f5b4e79be3b6395fbfa000e : SUCCESS in 8m 58s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/911df86f35dd49519c51000ffaee62e6 : SUCCESS in 45m 50s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c73db24ab09e460f9c715d26e26f3006 : SUCCESS in 47m 23s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f3c61811c88c4717aafe8da0d74d1dc8 : SUCCESS in 2h 04m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/154f86782aa848a8a77d5ebd70c95a9e : SUCCESS in 1h 24m 32s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/01f841537d2d4ef492533964b91d7290 : SUCCESS in 1h 42m 25s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/b40cfa05edd347a8a29c8bff8e6d7b5a : SUCCESS in 1h 37m 05s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/cda5d8f5f51e4a02850682e10e7f1ba6 : SUCCESS in 1h 38m 22s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/c1e271be31394ba39960262c7b7ff571 : SUCCESS in 1h 28m 07s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/eb542a1fb0fa4c06aa11f75f7b5c2c9d : SUCCESS in 1h 30m 38s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/e32c6e2459c7413586f92e7a5f3fccd0 : SUCCESS in 1h 25m 57s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/efa49911662f4630bf6e6fa3a32e294c : SUCCESS in 1h 41m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6f73bf3d0e3c4406bd0e35fcaf8a53c1 : SUCCESS in 1h 46m 23s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b419bd62a24547cbb71ab6236f799ddc : FAILURE in 52m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f94539f7058c4efead16edd2d78e9947 : SUCCESS in 59m 52s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/1b140e5276724e69abfc99d60a8aa96f : SUCCESS in 2h 42m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/05bb1af4bfcf4ef6952b76f639d2cf5c : SUCCESS in 2h 25m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/93ff28809dda4e0b8f9bd76857b30e9c : SUCCESS in 52m 52s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04fc018fa4a34b23aaa83e2e8f8f189d : SUCCESS in 1h 20m 35s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7792efbd3df04dffa8b8b4ec4bd3ddaa : SUCCESS in 1h 12m 09s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bb82e35626414b0b878a6dae8aca2125 : FAILURE in 24m 12s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7a4b0e6d0433435192dd2abfa56c7546 : SUCCESS in 1h 35m 16s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7bbd53dbc23d40848ca1be27f41044d0 : SUCCESS in 2m 29s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/46333f1638dd4de1adebda62e4cb5cc2 : FAILURE in 1h 09m 37s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/713cfa98899d4fd9a70e0be95b60e8ae : SUCCESS in 1h 04m 41s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/aadc0b49684044e1ad4e1a88824fa39c : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d62fca842d67dc78d87484efddc391c34466d62c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-01 05:11:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d559f089c2003fceb39699116f88fe90f0e5f8fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-01 08:15:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/921851944bce4a689ae672f7a59b5c92 : SUCCESS in 1h 42m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e4305aa3e99a4523b759308d0cc6a426 : SUCCESS in 1h 00m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9cb1413db024f4b99007d7e42b50f69 : SUCCESS in 1h 05m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d39ca03598b4ffe9965bfa57e32779d : SUCCESS in 1h 22m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c323c08c39c542d0b666a898227e4407 : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4fbc9504d1d4a0881caa4dc9fc7b331 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb839020f3fa4c4dad4e8d6a896c90fa : FAILURE in 3m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/422fdea3fafc4374b0eb9029ff230238 : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dcb9588e8194f8e8c05e048db0f62dd : FAILURE in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e98d6fffab8438089f051bbc1e82eab : FAILURE in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b7464c21df4a2f95058b70bfe5d0d7 : SUCCESS in 7m 06s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/5de65b2508f64ce5ae488038cf0638eb : SUCCESS in 38m 13s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/0b60f523fcc347a7beb428f23210a656 : SUCCESS in 39m 03s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d0449eea46cd440f836dc89443922f70 : SUCCESS in 2h 01m 59s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e5c9fd4a87d0471099350b5960baed85 : SUCCESS in 1h 35m 32s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/9a27d0bcdb824272b74e2c44f3862410 : SUCCESS in 1h 39m 54s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/c9523f24d7e949aea9a51983435073ff : SUCCESS in 1h 36m 16s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/e4290271ee8b4d36a22f93fcb4010288 : SUCCESS in 1h 33m 09s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/6adbedbe3c304c28944d449e8a71638f : SUCCESS in 1h 22m 26s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/02a8925e35934d439d21c502ec259d50 : SUCCESS in 1h 28m 15s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/64a6c3e8e9e343868adbaaba64849028 : SUCCESS in 1h 21m 09s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/73d0ebb719bc47d89442476e554171bc : SUCCESS in 1h 54m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7664b23b5a1744cebfa5ebd7f0f779aa : SUCCESS in 1h 32m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2b2eeff4da6f45009bc6982c0bc0ef90 : SUCCESS in 51m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06880f2a86454be2a3ffd9c11650916e : SUCCESS in 1h 07m 17s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/a3d4aaa344c24f079d81ec41359a841a : SUCCESS in 2h 39m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a35b521a698540b1b1c4a549facfa5bd : SUCCESS in 2h 23m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3d479dbe70454e6ebb5e2fa07d1f933a : SUCCESS in 48m 12s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ddbfe1e898b4616a0ce7ef1ade74baf : SUCCESS in 1h 08m 29s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b5ee9cc4ddf94faaada39bf3d1ac96dd : SUCCESS in 1h 26m 33s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/adf7add50f0b4260aa18be7e8059effd : FAILURE in 9m 19s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/46f29509385442fdafd9158dfa46cbdd : SUCCESS in 1h 18m 15s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/623c62c9a9124973aa0dd6390d772186 : SUCCESS in 2m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/899b84c4da014f7693fd79ef94fd4b91 : FAILURE in 56m 02s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8afd26a7c59542138b999d596187388a : SUCCESS in 58m 25s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5fdb9ce805ec44c7bb28c08097e1ac49 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"098cdd70d550fe7366f5a919a7b7c5924df58386","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-29 21:35:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"571efc294b4d4161df46cc4b00dce08577441412","revisions":{"116aa55f30ea9d4c86e45959ed69549e815c8901":{"kind":"REWORK","_number":1,"created":"2019-08-23 10:49:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/98/678198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/98/678198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/678198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/678198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/678198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/98/678198/1"}}},"commit":{"parents":[{"commit":"519f390000c3459a765ef1d0448ec966591494cf","subject":"Merge \"Fix Upstream test cases to be case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/519f390000c3459a765ef1d0448ec966591494cf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-23 10:47:46.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-23 10:49:13.000000000","tz":0},"subject":"DNM:testing account jobs","message":"DNM:testing account jobs\n\nDepends-On: https://review.opendev.org/#/c/580356/\nChange-Id: I9f88d5f70711a50f74b960353d583b10f434c9fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/116aa55f30ea9d4c86e45959ed69549e815c8901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/116aa55f30ea9d4c86e45959ed69549e815c8901"}]},"branch":"refs/heads/master"},"571efc294b4d4161df46cc4b00dce08577441412":{"kind":"REWORK","_number":2,"created":"2019-09-01 05:11:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/98/678198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/98/678198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/678198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/678198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/98/678198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/98/678198/2"}}},"commit":{"parents":[{"commit":"519f390000c3459a765ef1d0448ec966591494cf","subject":"Merge \"Fix Upstream test cases to be case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/519f390000c3459a765ef1d0448ec966591494cf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-08-23 10:47:46.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-01 05:10:52.000000000","tz":0},"subject":"DNM:testing account jobs","message":"DNM:testing account jobs\n\nDepends-On: https://review.opendev.org/#/c/580356/\nChange-Id: I9f88d5f70711a50f74b960353d583b10f434c9fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/571efc294b4d4161df46cc4b00dce08577441412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/571efc294b4d4161df46cc4b00dce08577441412"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
