)]}'
{"id":"openstack%2Ftempest~684858","triplet_id":"openstack%2Ftempest~master~I096ea150bfdd1d700a15612a391142216f0fd0e8","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I096ea150bfdd1d700a15612a391142216f0fd0e8","subject":"Move logging from _request() to raw_request()","status":"MERGED","created":"2019-09-26 00:07:21.000000000","updated":"2019-09-30 07:10:17.000000000","submitted":"2019-09-30 07:08:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684858-1569827337895-af8bbaff","meta_rev_id":"33fd5f35fbf836cabb441029d7b5c5c4698dbf76","_number":684858,"virtual_id_number":684858,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-30 07:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-26 08:37:38.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2019-09-30 04:20:30.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2019-09-26 01:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":1,"date":"2019-09-27 03:25:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":1,"date":"2019-09-30 04:20:30.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 01:13:22.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-09-26 08:37:38.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2019-09-27 03:25:37.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2019-09-30 04:20:30.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-09-30 07:08:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97e19016e4d7d8d210050896cdef94a754c5746b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-26 00:07:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bfc2bf3dce2742bc3e09666d58ff5b72b2182c4","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-26 00:36:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nLGTM, but there are pep8 violations.","accounts_in_message":[],"_revision_number":1},{"id":"b8e3da05561de5c162dc882d9d30ef00f8eadcab","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-26 00:54:22.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3be94aabf4be92a5fd9ae1fbdfb38b2bed568be0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-09-26 01:07:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e42197ff335e1e267ec7bfb04d1668318c2045f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-09-26 01:13:22.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks, LGTM","accounts_in_message":[],"_revision_number":2},{"id":"02d987600f096ab4d813a0b1ed6e0d1fd00786f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 04:11:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0d16e25085da44c98c298cba3c7804f5 : SUCCESS in 1h 40m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0535f04a9960402eb9be422e4bab2c4c : SUCCESS in 57m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/648fc76f091a4fbb8ce19b5f4deb71bd : SUCCESS in 1h 15m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20057336b5b64abf92bc688878733496 : SUCCESS in 1h 18m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09ec9e141ce940adb13aaa3028531754 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b4992f1adc4d5395491b1b80c58c42 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b37197f16e240058e6cd1169bc3f177 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e87a87c6f30246629b701eaf70a4c2ea : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ee6ed9c3e164c3590e54091a2c7e2df : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/412fd3e81fe6453696244823126d8726 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0e939d7ed144ebaf4b7cdcd1ab70ed : SUCCESS in 10m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/10743224e4ef40f9b4626b28a44674e0 : TIMED_OUT in 2h 08m 14s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b95fb055a1824998afd707b6daa12696 : SUCCESS in 1h 40m 26s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/10b60c529c8b46b582a56ec4326230f5 : SUCCESS in 1h 31m 36s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/991b1385a33544b7ad6b71f77609c18f : SUCCESS in 1h 10m 25s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/3e8e80f3b08f41d59b6b6c36c261cd78 : SUCCESS in 1h 26m 53s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/91fc060154554342834bffdd61e32dd8 : SUCCESS in 1h 22m 10s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/2f791bee9bcc4049a9ba218852dedbeb : SUCCESS in 1h 26m 09s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/fe425f729ccd472f9897fc3a6e21dccb : SUCCESS in 1h 15m 58s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c1f8894ac167453e92eb8e4dd7a0fa78 : SUCCESS in 1h 34m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2ce21279b62846d18f7459f0c17fbbee : SUCCESS in 1h 41m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0335f3b148654070800405f867c70715 : SUCCESS in 1h 00m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/819d746668bd466a815d1cdb501c0036 : SUCCESS in 58m 53s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/6a1df7e4508b47b488cd16a2fbd8973e : SUCCESS in 2h 58m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a171f11a1144bb9be28d135050921aa : SUCCESS in 2h 12m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/94c2a34592e14a2db0c00e92ba71b54d : SUCCESS in 55m 48s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6e01676cb684fd0986b6836c8349493 : SUCCESS in 1h 03m 29s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fcf734ad50fe47b28e691a3ebedcb5c8 : SUCCESS in 1h 12m 26s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/dabc11cb4fd14cf3975002fa09680435 : FAILURE in 21m 58s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ec772065abd4484a9c4be5447ceb81fb : SUCCESS in 1h 23m 28s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7f31edaeb41b489e8688b3a14f4f39f0 : SUCCESS in 2m 53s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d345caadc1c44a2792f42d1a420680ac : SUCCESS in 1h 28m 54s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/31ad9815ac3446ae853ba42ce07dacf8 : SUCCESS in 1h 08m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4a8ae81710a84fd8bd22bbccbba628b4 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"27ad4ca0ba8411a12575e120e3678d0cafa98554","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-09-26 08:37:38.000000000","message":"Patch Set 2:\n\nlgtm.\nAnd then we can remove the log around calling raw_request, like\nhttps://github.com/openstack/tempest/blob/e227022df9728e7b04cb5b3482a9b3360706962d/tempest/lib/services/volume/v3/versions_client.py#L37","accounts_in_message":[],"_revision_number":2},{"id":"e87830731c6887a8385b8d6b869a6114a0ad2a52","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2019-09-27 03:25:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"529bb1d738820775d25671fc4415bca996f6f18c","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2019-09-30 04:20:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLGTM, although it would be nice if this module https://github.com/openstack/tempest/blob/master/tempest/tests/lib/common/test_rest_client.py included unit tests for both these methods.","accounts_in_message":[],"_revision_number":2},{"id":"c3092dac4feddb1f0f8c926f4a7ae67e1df46764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 04:20:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df3e2dbaa9ef55ab62e3a68743259ee7b1009e45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 07:08:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/38fb2f774bbb427b81a32444ac85e089 : SUCCESS in 1h 37m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a5671281ded241e98cf54f80cb9fd347 : SUCCESS in 1h 03m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a13f4ede18a40bdb57feef3eadd5a05 : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ea9158b94864dc08f4d196af0d8ba04 : SUCCESS in 1h 50m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3030ea03c0dc486fa4f324cb9db49083 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6f2977d5b834a9ea76ac21fac5ee908 : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/106a88f02b1149bab4f9b3328c524a41 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8728ba5eb964522a97ea96cf32b1fb9 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9518a7c2f9b4eb082405c664a838693 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4dde6b98b04b7782c2e7be5037eea2 : SUCCESS in 7m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/415f6926fc344edab7e30dd03f25d264 : SUCCESS in 2h 45m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c4c6ab5fb1848ffadb650fb308b3e92 : SUCCESS in 1h 09m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bc01d1987444aac9377825afc5d6e17 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3653e656ce182b686ff1a318ec25ff5496432199","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 07:08:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"33fd5f35fbf836cabb441029d7b5c5c4698dbf76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 07:10:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374a721b85f1419fad4a9f468cbaecd2 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cb3cf038ebbe6e561aeca10adffd575338bfca61","revisions":{"c50e7bcae67c0d4d1e312e700447bb6cf5290b2c":{"kind":"REWORK","_number":1,"created":"2019-09-26 00:07:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/684858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/58/684858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/684858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/684858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/684858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/58/684858/1"}}},"commit":{"parents":[{"commit":"56bd3285b5ed22ec62fe314e96efb0b2adce06b1","subject":"Merge \"Remove barbican-tempest-plugin from blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/56bd3285b5ed22ec62fe314e96efb0b2adce06b1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-26 00:02:54.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-26 00:02:54.000000000","tz":0},"subject":"Move logging from _request() to raw_request()","message":"Move logging from _request() to raw_request()\n\nrest_client log all request and response data when called via its get,\npost, put, delete method etc because logging is done in self._request().\n\nBut when any service client make API request via raw_request() then logging\nis not done.\n\n_request() method inte9rnally call the raw_request() method so moving the\nlogging from _request() to raw_request() will not effect existing behavior.\n\nChange-Id: I096ea150bfdd1d700a15612a391142216f0fd0e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c50e7bcae67c0d4d1e312e700447bb6cf5290b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c50e7bcae67c0d4d1e312e700447bb6cf5290b2c"}]},"branch":"refs/heads/master"},"cb3cf038ebbe6e561aeca10adffd575338bfca61":{"kind":"REWORK","_number":2,"created":"2019-09-26 01:07:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/684858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/58/684858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/684858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/684858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/684858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/58/684858/2"}}},"commit":{"parents":[{"commit":"56bd3285b5ed22ec62fe314e96efb0b2adce06b1","subject":"Merge \"Remove barbican-tempest-plugin from blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/56bd3285b5ed22ec62fe314e96efb0b2adce06b1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-26 00:02:54.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-09-26 01:06:28.000000000","tz":0},"subject":"Move logging from _request() to raw_request()","message":"Move logging from _request() to raw_request()\n\nrest_client log all request and response data when called via its get,\npost, put, delete method etc because logging is done in self._request().\n\nBut when any service client make API request via raw_request() then logging\nis not done.\n\n_request() method internally call the raw_request() method so moving the\nlogging from _request() to raw_request() will not effect existing behavior.\n\nChange-Id: I096ea150bfdd1d700a15612a391142216f0fd0e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cb3cf038ebbe6e561aeca10adffd575338bfca61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cb3cf038ebbe6e561aeca10adffd575338bfca61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
