)]}'
{"id":"openstack%2Ftempest~686073","triplet_id":"openstack%2Ftempest~master~I840b273c37ca7cc4592c43813abfb424337e2836","project":"openstack/tempest","branch":"master","topic":"admin-auth-system-scope","hashtags":[],"change_id":"I840b273c37ca7cc4592c43813abfb424337e2836","subject":"Add system scope for admin auth","status":"MERGED","created":"2019-10-01 23:26:57.000000000","updated":"2021-01-21 16:26:35.000000000","submitted":"2021-01-21 05:18:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"686073-admin-auth-system-scope","meta_rev_id":"6ff6caf62c2785a5d0310d0e1257d4ae7f94d917","_number":686073,"virtual_id_number":686073,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 05:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-21 16:26:35.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2021-01-21 16:26:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2021-01-20 19:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-21 16:26:35.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2021-01-20 19:40:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":1,"date":"2021-01-21 16:26:35.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 09:12:11.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-12-11 04:56:26.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-01 14:54:53.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-10-01 14:54:58.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-10-01 14:55:06.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-10-02 08:28:03.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-10-30 01:38:34.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-10-30 02:09:05.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-10-31 18:01:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-10-31 22:49:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 14:57:10.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-01-21 11:12:05.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"908b5587cc879dbe3687c57b04757339d1ee0e59","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-01 23:26:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcc234dcba94fc130108184861e162c7bff59524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 02:14:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ae2abad520c3431ab1fc601dbd21bdbc : SUCCESS in 1h 44m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1e3661666ef24c1e9ddf62715cf0d52c : SUCCESS in 57m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4205e897f0fe46cdad23efc3dde71178 : SUCCESS in 59m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4079c1564254590a6dfc5fe677ac7e3 : SUCCESS in 1h 20m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f556647f7eb4851af9d632f6b69ed21 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef61843d5ceb45c58ccfa54cadd9154a : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dee0bb02244043c5a1e9fbf8220046d4 : SUCCESS in 3m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d44d6cce2097411b8ff50556b4a882ba : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54dafe492f0b4e4baf4f9cca3214e7b2 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cf6395fae77489aabacb12ad6acdbf1 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c660594ade3466f84fd1ff398c6f162 : SUCCESS in 9m 43s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/fe0e840dab444b029c0d84fcf08d35d6 : SUCCESS in 1h 48m 51s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b6ee934c340e4206b33f12d7fc034f5b : SUCCESS in 1h 28m 55s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/fdc3b684cbb44279b52a2ae40f31e1a8 : SUCCESS in 1h 37m 07s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/1bc37b21adbf4e00a774d9caa92d7616 : SUCCESS in 1h 27m 49s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/f2068944322f42ff8b1f4aa6dd7e0f8b : FAILURE in 1h 13m 55s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/8107786d39644a3e9bf10765c5c7eda2 : FAILURE in 1h 00m 54s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/3e5960e0594442efb12368edc95c668a : FAILURE in 1h 11m 36s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/461e2092abc84e2da408e787b4127e99 : FAILURE in 1h 03m 17s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/53145b44bb8f4bacb19e90dfcecd17ba : SUCCESS in 1h 38m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e35e756030a546c893ce47445a776c5b : SUCCESS in 1h 30m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b54e90ef92104bb9b50ec424172bee70 : SUCCESS in 45m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abed1d4195f74b80af6c0a167cde2160 : SUCCESS in 52m 03s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/f1249e5dccc941dbbc44c0c0c58e546b : SUCCESS in 2h 35m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b59f3805a9148328d56694a945f4199 : SUCCESS in 2h 23m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d1afff8a875b441da3e88e40c124dbc6 : SUCCESS in 43m 56s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b15399664dd4722a8b85e24b6e1bcc4 : SUCCESS in 1h 04m 01s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51884eb5f2b2441882c3443f2e482996 : SUCCESS in 1h 11m 52s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a8efa922e27e481f9d239c38a0c815a5 : FAILURE in 26m 00s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f5dfd59028ac4f6cbd4d1e029a61fd02 : SUCCESS in 1h 07m 59s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6dc7459982bc4c4b81a7416527d01be4 : SUCCESS in 2m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/456d9ddacfe144b58da2f7c2e29a12e9 : FAILURE in 35m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/dbd217712480446b8cf14dcb513b478a : FAILURE in 31m 23s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d7e52b3b8a764cea8e3b39be3b388041 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b996ed94ad96ed4a1a33a507050c0ac146270edd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-02 15:41:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5df8cd564290bf9ad71a86eab34ef7dac6bbbc79","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-10-02 17:02:29.000000000","message":"Patch Set 2:\n\nthere is old patch for this from lance - https://review.opendev.org/#/c/604909/12\n\nI am ok to go with either one.","accounts_in_message":[],"_revision_number":2},{"id":"90fe9f5ec12cdf3fd2c2e347bc0491d7a9810526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 18:20:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3ad95c2924b74acabc455af45501cc09 : SUCCESS in 1h 52m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3151df47f20e452887b453cbbb2ac218 : SUCCESS in 1h 09m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5148d539bffe4c6194547d2895a394a2 : SUCCESS in 1h 02m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a2219b0e5734de7bbf99032b4dd9aae : SUCCESS in 1h 19m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e436b7fe22a49278aa1fe0dca2ea6a0 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497e62837ff449969537a8e1bae4a669 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d63649889db4fce9bc209e965d56e15 : SUCCESS in 3m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3db7e0227fff498d9a6a4a3fb7c04e82 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a144b8d8f494149b12ae50e793307ce : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3907ed0935a4fd29b3cacb17b427569 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558e847b5ae5447ba50f477e32956463 : SUCCESS in 10m 53s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b91b604d548b488397fce378fecbe0c9 : TIMED_OUT in 2h 09m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d17ea5eef60f41d09a135760006948f8 : SUCCESS in 1h 33m 50s (non-voting)\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/64fc9da2ebd74305b90d038b5b9458fa : SUCCESS in 1h 35m 49s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/692beb2206894ec784aa8d8e0a38a197 : SUCCESS in 1h 43m 36s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/ad5cf61ecc5d4aa39d9cae4f4b897686 : SUCCESS in 1h 50m 43s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/0dd30eedc96b42bd91c7598585b31e28 : SUCCESS in 1h 47m 49s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/f52e808d3aa54ee48883a77e7a20ad29 : SUCCESS in 1h 59m 07s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/fafc934dcfe148da8d65eef763196535 : SUCCESS in 1h 14m 41s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b6715f7c3bca4813a8b52e42f427fb1b : SUCCESS in 1h 56m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/93a4bd518ea54f62b74c39396e7d4f27 : SUCCESS in 1h 44m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/44cd93c658e648ff86fcbd6751fe67ac : SUCCESS in 49m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94b1956bff3c4d7b81087089d6f2aedd : SUCCESS in 1h 09m 56s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/f90f3adaf6b74643a1f4955f8e55f8aa : SUCCESS in 2h 18m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/410c2018a1b043b28853978e9f889fc5 : SUCCESS in 2h 20m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d5cb6e00a9e8499da786e9bbe89237e9 : SUCCESS in 1h 06m 36s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8b55678c8564faa8e399b0feb5a9818 : SUCCESS in 1h 14m 32s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff27b9ae4c3d478b9d9abf85bfbbd0bd : SUCCESS in 1h 15m 54s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8ba52c882be84751b7c7e8070ecc6160 : FAILURE in 22m 36s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a8d1ad8f751642ad822cdb25574340b8 : SUCCESS in 1h 25m 42s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b2a1d9343d6343d7a8148660f063e64a : SUCCESS in 2m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b315d60cef47499ab639d7f06a4046a9 : SUCCESS in 1h 27m 41s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/038028e48aa14c10a7952ef51420628f : SUCCESS in 1h 07m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c2029afbe774430a9e04149fc7be1b02 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"4f63e9a36f2bd1f65a88440b55445931d4d1066e","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-10-02 22:13:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- Tempest-Sriov http://13.74.249.42/73/686073/2/check-sriov-tempest/Tempest-Sriov/c672c24 : FAILURE in 1h 22m 42s (non-voting)\n\nTo re-run the job post \u0027recheck tempest-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":2},{"id":"733a374305394e43331f4e53739347e6c26ed61d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-11 04:05:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"61eb629abb173d4ed9c8b464cd549ef5b2158bea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-11 04:56:26.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Tempest-Sriov http://13.74.249.42/refs/changes/73/686073/3/Tempest-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"58a4fbc3dd549a457c714c7cd9f1a6f2326c6285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 07:13:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/763b7ec3279f4718a0e9df58d7be44ce : SUCCESS in 1h 29m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9c9e738433ac4dcab04b7b589b02e693 : FAILURE in 54m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cf0d8e5057e94b58ad028734158ad297 : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e481fc36d134c5483c5d73dc995e022 : SUCCESS in 1h 19m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c536fdb81f3e4531820e86023a0ea701 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85a8c5bdf8374d5e81e9f2e54a8f361a : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5738349b9baa4f3ea9f4ad111486978b : SUCCESS in 3m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/94803ef9ca174a12a1bda350d53cf7c2 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf233b8fbf504aa782ed19ce9e432e6a : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a0719669a4648d38eee37ef226f2f77 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf4692cfc0c4c91bdab5b2abf248d5d : SUCCESS in 23m 43s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1120bcda6f374275831ea77697753eea : SUCCESS in 2h 01m 16s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d3e80f9c6db04ac1af5678f52a010c8b : SUCCESS in 1h 32m 18s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/3c505c7e76a44b9db7dcec95cf5daa68 : SUCCESS in 1h 30m 28s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/cc304a17854c40e5ab0fe1dec3e90775 : SUCCESS in 1h 44m 23s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/6137d7a49b99418bba0359285e35bb8b : FAILURE in 1h 35m 31s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/ca0e63778e074dfdbd4fdf47b4ee4c4b : SUCCESS in 1h 19m 24s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/27998015980a4515b9f7594e4112d3ca : SUCCESS in 1h 55m 33s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/39c487b1c1c14023b50bedca430b6290 : SUCCESS in 1h 28m 17s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/f78463a2008e47919aa728ce9b4e1bc8 : SUCCESS in 1h 41m 41s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/0f60cd6a4d984cf2a7468054c407a1e4 : SUCCESS in 1h 33m 21s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9b9473d2613e47e6bd5e3e5183cf4530 : SUCCESS in 2h 06m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4efda0886ca34b929a5d413afa630a9e : FAILURE in 1h 58m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9187ac71d9474e91affd697ba7c995ae : SUCCESS in 48m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ba2abcec42a4e9b9e81005bdb3480cd : SUCCESS in 1h 24m 28s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/3ef274e7322646e89f0221d9cd92d15a : SUCCESS in 2h 51m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d1bcb543f813469e9ca212fcd8ebd8db : SUCCESS in 2h 23m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4c463bd208dd4cc9ba12499db0b24d38 : SUCCESS in 1h 11m 48s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/911ce4e84822489cbe91ba02222100f3 : SUCCESS in 1h 09m 59s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94ae033b7db8475d8f07266e4620a328 : SUCCESS in 1h 27m 53s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cd1a9113f02f4029b520b1ff65d2d0bf : FAILURE in 22m 07s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/184486b925a14434a512cd8a1a79cf71 : SUCCESS in 1h 12m 51s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0af5ee068e1f47bbb5f25ba2b5e36a8c : SUCCESS in 2m 15s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6b700dd2c128411d8ca49107f01a6f13 : SUCCESS in 1h 45m 53s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/55edec3f35ec4e4fbf1e9796cd22fb47 : SUCCESS in 1h 04m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/891d96e6bcd54f26b6e6c9f2f1f6198e : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e9413d696075b065d53c6f9b7d035b43969c4164","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-11 16:53:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a81930726df727b052f616661e5765407d3a2ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:53:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bbaf8a23de5c45b6a45ee99ae4c32855 : SUCCESS in 1h 24m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2aca629b59334d72aadf091d70860fa0 : FAILURE in 59m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b23ecf4ea5cf4c4e8067293c1e96f5ee : SUCCESS in 1h 19m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65823964115f4fed96d965e4f9952594 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ae218970464bdcbb603b6955755f19 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8972fcc61c8b456fa9bcceb64d834c44 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8512b7ed380e440a92a412917264eb60 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60ae6d1311184fb1bb6d49f70bc7df14 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e359087071c64db28ed6a5494c52cd32 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa39b251338429c82686072058b6188 : SUCCESS in 7m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1e1080300e034f878e3faff97ed72615 : SUCCESS in 1h 53m 05s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/20048f078e284f5e99856c3c21f795e7 : SUCCESS in 1h 36m 35s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/3f31962743a84d1684e9179f1c6cbdc3 : SUCCESS in 1h 32m 46s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/4415326c376f4e97b3c0898559cd111c : SUCCESS in 1h 13m 05s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/1e3b93b8cc0c4f37bb40259fc6db1812 : SUCCESS in 1h 26m 14s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/ff13a722faf74a3a9ea1a9f82a0b264e : SUCCESS in 1h 34m 28s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/64966b4ac3604aadb424a93e7163d0d7 : SUCCESS in 1h 19m 18s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/7ee55a9084f8497c83e0d005ced86631 : SUCCESS in 1h 47m 29s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/390195d354524646b710265c8d6fa9b1 : SUCCESS in 1h 34m 23s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/5a540f0def6f4a8fa0f6015d183ed204 : SUCCESS in 1h 24m 55s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc392ef2bd764f539755f084c933cd1b : SUCCESS in 1h 37m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3e0c862254d24955b23016b5eb59fbe0 : SUCCESS in 1h 55m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7f52eb1a90ef4f6cb7d0eec5f2e2a2ed : SUCCESS in 54m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62247faed5d740c1b2972899b7c99c3d : SUCCESS in 1h 01m 54s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/c9beef7bbc0e4af3ace579db723cd4f3 : SUCCESS in 2h 34m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5d350dc92f1040d49ea46470c8001b3d : SUCCESS in 2h 45m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0fb694c12cef4f3a9d2f6630969ce43a : SUCCESS in 1h 05m 40s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b998c7eddee4cd58b8bd5f2aea59f96 : SUCCESS in 1h 23m 43s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/717b73770e434170bf9126eeaac8f6bf : SUCCESS in 1h 12m 53s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ac0cb48628ea49d4bf572a13a96dd2f2 : FAILURE in 24m 01s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e6f44b960b9349c5931e21722d443c42 : SUCCESS in 1h 10m 30s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a50c2ea76b81461cb0929ed32703f525 : SUCCESS in 4m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/dfb7699227af42b1b8b961c8bf8c8e8c : SUCCESS in 1h 38m 24s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/867ecb9881e2443abcbb2b04170f39f4 : SUCCESS in 1h 03m 58s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6accc6a2178940bb84cc5c7cf6bf3b47 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"feaa7d54fa00ff2afb188231f200442036af006d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-12 18:23:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ca081f03390f22bb0864b415312ae91e3eef3d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 21:57:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9fe7b44c20064b008614b5d4a058806f : SUCCESS in 1h 47m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cc3d66abdb9843b9b355e1cfcb58e041 : SUCCESS in 1h 04m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f45e8aebeb0048c2b27196c564cea62c : SUCCESS in 1h 32m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c42daae7f5904d8383a59735958168f0 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9dba311d274c8586ec7a54aac70636 : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0ba40d5992f470090ee98d1ffef4282 : SUCCESS in 7m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9be7054f9618462db013b7a528edbf08 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01952de502fd4b98acc47aba287579e7 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1960ced6d1d249d7917831b9dea82029 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b33a29ea7034245a06672fb76aed8d6 : SUCCESS in 9m 36s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/45017b14bb5e4216b85ae65556999ea2 : SUCCESS in 1h 54m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/36dab2e1924d4dbbb79c07229c693275 : SUCCESS in 1h 26m 53s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/05182c29aa194644ac471d574e4bca51 : SUCCESS in 1h 43m 20s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/852bbb1c878c479fb3022f92319938e0 : SUCCESS in 1h 29m 01s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/7cf9c9acd9d74a0c9290d76a59fd6321 : SUCCESS in 1h 38m 31s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a165605daf454cfd8f2ced19b5637030 : SUCCESS in 1h 28m 11s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/890947b50e4f40a0922d6ba2908c9aa5 : SUCCESS in 1h 28m 04s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/4d7000af873047b3ada35c13a9b2a263 : SUCCESS in 1h 25m 58s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/f83dacb6454441b4a873623879f8284f : SUCCESS in 1h 23m 53s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/d62059c85c0d4245aecec251a848cd7b : SUCCESS in 1h 39m 30s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/82f78fd0af654fc485c0316fdb195174 : SUCCESS in 1h 42m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/29979eb9e47e460caf25fbd647fbbe16 : SUCCESS in 1h 53m 41s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f348989711544d48b7811d171dabfe1b : SUCCESS in 44m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad2fb2fffed04b4b95443198a77291d7 : SUCCESS in 1h 23m 58s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/7fd876d157664d9699461cb7cc6df287 : TIMED_OUT in 3h 08m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9723ecb902d41bba4e34180be360641 : SUCCESS in 2h 29m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/28c7309843b14662b8a020ec6ee0a435 : SUCCESS in 43m 56s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68e37d1841434e938745f896a7ea8cb7 : SUCCESS in 1h 10m 34s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2598aa3c63f24ecd8375c61e24fd30ad : SUCCESS in 1h 11m 23s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c70b5e550e14495a8c0bbb21ecc75ceb : FAILURE in 26m 14s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/008d2084f55642e787d4658dd162cd74 : SUCCESS in 1h 22m 45s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/404f000c03184824a2745d66ac5455c1 : SUCCESS in 5m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4ed1b6d35aab4a81a9a9e33f31b0bd32 : SUCCESS in 1h 49m 31s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c94776a10fbf44848c02ea290b7d31e8 : SUCCESS in 1h 12m 18s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2eb42de5e05247ffae891e4653cf976a : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4b6a3be7b4561f8218921842e1583a9bb38553ad","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-12 22:01:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"06121cd2ce666d70f9e3ea4f3de38af7b9a8e951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 00:56:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f338207a81c144fa86932405af812a71 : SUCCESS in 1h 58m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dd1251c7f9984501b482f7b338ee1416 : SUCCESS in 1h 13m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19dee69bd76345baba4da6baccbe73d4 : SUCCESS in 1h 49m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d49f274c5b6146718cefa72dd7dabd94 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f67e3850e8914d0e9f0c748dd1e48b8c : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a55f289686474af5a1ea1aa95b55fe66 : SUCCESS in 5m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b41e41a2372d4a61bf41c4463879a4fe : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ea9e47ea495400491f98ff6feca83aa : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca52d9fb384547a49622aab71e1c1c8c : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6debc3c5b2345f092f58ca3f25baf9f : SUCCESS in 9m 25s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2ad4f0dac4564710a5df69cf14b16ec5 : TIMED_OUT in 2h 06m 47s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/248219ea633a44b29bf03f8bdc1d530a : SUCCESS in 1h 36m 10s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/1e22578a837e44dea7865b2b2532a5b5 : SUCCESS in 1h 44m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/de71c13793f74556a5391c45c1720db3 : SUCCESS in 1h 29m 47s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/19c47e4a1f1b45ef988517c525f356f1 : SUCCESS in 1h 26m 09s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e41871e89775475bae0503c604054c80 : SUCCESS in 1h 27m 08s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/f62c5a5405294b018d03ce908cec3b4c : SUCCESS in 1h 34m 40s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/a10c7ffac9d141a99224c0102f1643db : SUCCESS in 1h 20m 40s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/839dda048571472ab7620862e9aa39b8 : SUCCESS in 1h 35m 45s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/837b8ed3f4284a0196e3b91a6227f751 : SUCCESS in 1h 22m 29s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9ec72d5220dc4e42bc9770eba7264dd0 : SUCCESS in 1h 47m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f08a37f141fd41f295a5ee5a6e929794 : SUCCESS in 1h 28m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7479a00b38324f76809f28f617ba7024 : SUCCESS in 47m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/730f13ca6fe942b6880099ee1b054918 : SUCCESS in 1h 07m 04s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/33c607f364c548838cdca99df90ff7c6 : SUCCESS in 2h 41m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e17658c0d81949e18f07cbd9bb1fe940 : SUCCESS in 2h 18m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a4673d5165e64c20bb795717659ec279 : SUCCESS in 1h 01m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/07117fe463fb4671bce627d21537464c : FAILURE in 1h 22m 03s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d8b813d350f1416cab24c1c93050b992 : SUCCESS in 1h 03m 56s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/abf3057f37554f8f9e02f319f729b05f : FAILURE in 26m 31s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a0615e9ecb4e46f5b0d82a59fc71c157 : SUCCESS in 1h 18m 48s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9804ce9d7c6b4c8bb7720f2c5ae0c530 : SUCCESS in 3m 26s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c3ec11bda1c24a8aa0a87bd3727645bf : SUCCESS in 1h 23m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8b5d3745f38a4c7c8193d34b4999cf31 : FAILURE in 1h 24m 20s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3586b326eb4347c8be056d2431c048dc : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8452a5bde4fefae3c7f67d380d2fbef8186d9651","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-13 01:14:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"644dfffb96c819dcab09b038df931993550b0abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 03:58:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/86a1ded8454745a487dc03c6d0efde39 : SUCCESS in 1h 53m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c1c5ab8b3d2144a2a42a9b5634843783 : SUCCESS in 1h 05m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ac65c6ffdc54e5abd2312f9081b7255 : SUCCESS in 1h 25m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96a86fe5f29e4f41b7838b598b2f90e1 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95148dc8b7642f09a388385f6bfac44 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bf2df99eb9b48ac8618effc35ad45f6 : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/811fed275c5a4f098e6649cf40de0a09 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e0d8282eb04469b04f25314bd8d0b6 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3f20f5e9e4d437899cdf7ac84579e23 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac0aae7ca234078a5b77383b9e398bb : SUCCESS in 11m 48s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/00459df33a7241439e7daa9299329117 : SUCCESS in 1h 46m 58s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e7d6317f5de14d6186ef2bbfe60f039b : SUCCESS in 1h 16m 02s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/9c02d4d709074587a36f7ae89ccf9857 : SUCCESS in 1h 27m 28s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/9035f3a310b447e2aafcc13abb27f2b2 : SUCCESS in 1h 29m 54s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/93f8351276d8460a88437ded3bd057aa : SUCCESS in 1h 32m 39s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/98aa5d5fdec94d4dae8fcf3df9480e71 : SUCCESS in 1h 14m 39s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/590a9565dd7246038d12956f335d2274 : SUCCESS in 1h 39m 01s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/448ccd247e54443987c3cd7953db550b : SUCCESS in 1h 16m 12s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/5feeac7f0ee944b8a15c7dd7e029727f : FAILURE in 1h 24m 05s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/75e3c1d85c5b4f88a19005eb5dc40508 : SUCCESS in 1h 42m 31s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/beb14596680d41cea46a19a435639b5a : SUCCESS in 1h 39m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eea04329fa0740809e719de78acb2207 : SUCCESS in 1h 45m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/72a8a1627dac4f8aaaeda5c044c183fb : SUCCESS in 47m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ca68f8d05bd4b29a8426677784c2ad2 : SUCCESS in 1h 26m 22s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/978c873ca57c417fb977cb7a2d6f8863 : SUCCESS in 2h 36m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2901f1dac1a400cab8abcbe1d32f56b : SUCCESS in 2h 15m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ea9aba0dd2f04c539812d8e3e500119d : SUCCESS in 51m 10s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d044b4b0d16e4b3d99cc63a09fbbee94 : SUCCESS in 1h 07m 27s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ffad333e77e24f28ab064114f97b5020 : SUCCESS in 1h 31m 50s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3287c96074e449af934b8e1a29232b24 : FAILURE in 23m 35s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0250b669893947ffa81f95e2cb4aae24 : SUCCESS in 1h 44m 01s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/767b02206cbb49a2b9145575da0c2928 : SUCCESS in 3m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/cfd166b403db48ee850d4e916f8c132d : FAILURE in 1h 33m 25s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bdda18e0353b44bfb37cdc2758895163 : SUCCESS in 1h 03m 10s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/696e0e83849a4758af570530a2038e7a : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":3},{"id":"62395c835a06dca1a1b52dd324c9de600f53f724","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-13 04:47:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6ed99171900e5a5596dcf912314cfb771b696d39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 07:28:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/18ba7be0ab9641c6833e33afbc312572 : SUCCESS in 1h 26m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7e6a5314307a4ebbad6f15cc2d07d3b4 : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bdba4377a6f54168a1baadb4c1feb12a : SUCCESS in 1h 33m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bbe85fde22746df968185bf8a7d063c : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce98ac2b7d2454c99b69df3f247c76e : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8ef6ff3ce8e41e08bda5cc0652b401d : SUCCESS in 3m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55782a77285c47ca966db397dd68a816 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bfd395995bc4762a880cc837f145ccb : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8eed6373a7774e718352c636ace17999 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dda201a174343d78050ea20283b46ee : SUCCESS in 8m 36s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/bf57cada20d344309a1bdafcd9a770a2 : SUCCESS in 2h 09m 37s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/0d917bd9cc064fc89aea571dc39fb475 : SUCCESS in 1h 19m 19s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/2632ed5671ef4f38aed8d122988f026d : SUCCESS in 1h 34m 52s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/0f96be3f872d4e2c8247b6dcae7d4643 : SUCCESS in 1h 17m 26s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/698e38e2d5c14b079a4cdee58cd1662f : SUCCESS in 1h 21m 43s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/1c330a971a00458e821ff8de9f57aa43 : SUCCESS in 1h 24m 00s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/2caccacc66f448c18153ff047fbff0cf : SUCCESS in 2h 11m 46s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/85c7b470070c4b5986171401352e1a54 : SUCCESS in 1h 17m 29s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/bf11e69934b84bf3a174dafb06921f53 : SUCCESS in 1h 24m 53s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/4a6a00019ad44afbb6948a217c6a11eb : SUCCESS in 1h 15m 25s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a19d0b5b37eb40829eb87c974176db96 : SUCCESS in 1h 47m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7f216d67445540ba932055f1149e4ec9 : SUCCESS in 1h 52m 09s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3560425a3416413e83cee92619c32dbe : SUCCESS in 52m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d96d6b9eb4940eea4562a571559d9ee : SUCCESS in 1h 02m 53s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/db42b2bb61554a4d9d8f0383abc2a266 : SUCCESS in 2h 37m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1782ca23803e490f94afa6dd3b92dd72 : SUCCESS in 2h 22m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/06194eee0b424fb88b73c464a735c8e3 : SUCCESS in 48m 18s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22b97977a17946228d34f25c34f33583 : SUCCESS in 1h 15m 00s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce4188acd8474e2b99e13cc4d5aacb75 : SUCCESS in 1h 05m 28s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/20fdeb368faf4679be3f592feb5a99c2 : FAILURE in 22m 41s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7478415a6f4f43ab92a9c7343104677b : SUCCESS in 1h 27m 34s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8b951a7f5bcd4ed59f4013803b000c12 : SUCCESS in 2m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0270623d3f5b4499a7ef76e8c99b48c0 : SUCCESS in 1h 20m 30s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/68f3982ad1cd477db30422c97585860c : SUCCESS in 1h 19m 03s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/87fe27dab6d24854841f8d50fd1fedba : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"16be640de21adc613910efcf53d77789e66fad1c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-06-20 06:59:13.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"9f5067379d3acc0183b06c9c1bda413da84a2cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 09:07:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/199c2e6c307a4439a0a08f9ee92253da : SUCCESS in 1h 10m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5895a5ebfdd4aa48fec1f1b62b1dff9 : SUCCESS in 1h 27m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a042a8643f454dd2b7c161cb62542b9a : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d948ae6e5d48aa907488aad088e7b4 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08cddb04fcf24478b0cbac6c88520d86 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b94caaaccd204e0084db43fff627dcde : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c3e049b5194eaaacb58ac2370ce2de : FAILURE in 12m 42s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c6eb6a95fbfa478ea287232ebb32bdc8 : SUCCESS in 1h 45m 59s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b21c043379cc4ee0af69287c632d7356 : SUCCESS in 1h 22m 34s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/a47f1a541ea8457c8be02544c5aad3dd : SUCCESS in 1h 35m 06s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7aece0a4bb4c484584a5282534897c82 : SUCCESS in 1h 21m 06s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/cb926e4e67ce429ba6aba8dfa7dbbc2d : SUCCESS in 1h 10m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/077754ea898542349f0f42bd63eda15b : SUCCESS in 1h 29m 13s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4538669398144bdba11c39ead5520ed7 : FAILURE in 7m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9dc837250764904ac9f7b7e9f252a58 : SUCCESS in 49m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fdd2c5e1f3847fd8cc46d12e66714c6 : SUCCESS in 2h 04m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/038e94d2061b401c886dbbf60abbc81b : SUCCESS in 54m 26s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6fd2ef28ae514440bb5904103b956383 : TIMED_OUT in 2h 05m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4d932649b2b41909cf86aeb0cd36dc1 : SUCCESS in 1h 07m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bae95f1fa50b4e87871ccfaf473bd94d : SUCCESS in 1h 17m 12s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/391b0d0879124c198084b524b25b2668 : SUCCESS in 3m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/893a8197adfe4ef88726928a28c7f205 : TIMED_OUT in 2h 05m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6f61f9f32c724c3eb4ca3092de2dc7a3 : TIMED_OUT in 2h 05m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0b94b5d64ab74f5fba01299a4fceddfc : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"dfde4901e29616fcfe3f9bc3e1ba578e7cf7de70","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-21 00:04:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a260748187d5c15af9cee37185d95c3e1fa0e762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-21 02:03:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/98ce10611c884fc4adf596ad4d3b2c1b : SUCCESS in 1h 09m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b44863f8281473fa45d1cc9dc268779 : SUCCESS in 1h 15m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d49fac96ee4983adcf1dca18c0b7b7 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e0ba2f3161400b9a35a4d5cf647c09 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32882e50aa8c4fc783839e7add1d7f92 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e56eb684e7b54f66bf7a35732ccd9452 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e65b71b1804274bb39b03f83778fd7 : FAILURE in 8m 05s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/69054d238d9645f79e9423d3cfaf23f1 : SUCCESS in 1h 43m 00s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ce2321c010d84fc79c89f6a2b9b20461 : SUCCESS in 1h 11m 18s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/9b288dce05a14f39afc9663a8a09ea11 : SUCCESS in 1h 25m 04s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6ad7d95c62e34f6796574a6b6fd39954 : SUCCESS in 1h 33m 22s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/6ea21cfbe2e14768a3c7f9ed6eb95500 : SUCCESS in 1h 26m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9fb091d0545b44418bc131c8e9bc3f36 : SUCCESS in 1h 27m 45s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/842d5c8d9b2e4f6983da0a5058b8ef94 : SUCCESS in 43m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5b28bbbd9564d6d8be930c2ad917c15 : SUCCESS in 50m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1f479f0efc44b899edc429ed782cbe7 : SUCCESS in 1h 53m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aaae5428b4f94e2781c4bc813f62e47b : SUCCESS in 51m 34s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c3c8d034745e4c53b77ef76869b1c9c4 : SUCCESS in 1h 18m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6102148ef75445e84207f7078875e8d : SUCCESS in 1h 10m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/efb75311f3f94340aefbff30b64675f5 : SUCCESS in 50m 57s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d7260b75ed2b433ebcefccb54ffd2f44 : SUCCESS in 4m 05s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/45459cb9d9e446f7af135631f814c8a9 : SUCCESS in 1h 19m 04s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/88d00a91e4694743975aaae48020ab98 : SUCCESS in 1h 02m 52s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/68d2b83a87974ffb91809b7482161dc3 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":4},{"id":"75428dfa742214b22778d83bff0030b60d492129","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-06-26 21:45:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a15f6891c9d667f3428f1c11362de3614214ce05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 23:49:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/be54957deb474da5a7371241cb67cb59 : SUCCESS in 59m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1168ef3ffcb4936bf02e1425f9bd65f : SUCCESS in 1h 13m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16ea8579db0447dfb7ae2bf9a4a978ae : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c655daae944290b93402880ee10eba : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22e65167e7d148dc842819b507952a7f : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2056bf79d1634295a340e99bd03137e0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d73ec987d04c05acc54b231d33067e : FAILURE in 9m 54s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/21c0b47fca0e43a8b0d600b0a0ac48ec : SUCCESS in 1h 51m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/cc16ad7f9ef0403983751110e307d73f : SUCCESS in 1h 16m 25s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7174098a14fe4b1abd8ee6736eb0407b : SUCCESS in 1h 18m 57s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/295c31af97104bf9bbb30f114bd6cc7e : SUCCESS in 1h 19m 56s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/f364b7d4dc78470ea88add487c3edcbf : SUCCESS in 1h 18m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8bee9794a05f4462bc5a16c9f473a7c3 : SUCCESS in 1h 39m 07s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2e9fb84481de4f56bf97918081525b76 : SUCCESS in 47m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65aed9282a964624bf9a6370b902fc88 : SUCCESS in 1h 10m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3735fd89ddc34eb69e9fe591aead5cf1 : SUCCESS in 1h 58m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8d3af55328704eefa30bc244dcd0bb4c : SUCCESS in 51m 26s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6d110f3c39544910b8813930698720b1 : SUCCESS in 45m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de4eac9ecffb4666b85efa8b2e6dcf1c : SUCCESS in 46m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/52976f65d7e74ed5a0cfbf75a4653981 : SUCCESS in 1h 10m 03s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/954434b5114543a898f3cdbcce5f432a : SUCCESS in 3m 22s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b14068369f384c70ad14694a5d642d10 : SUCCESS in 1h 15m 31s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a9586cf2f2bd41e488bc7697633d2717 : SUCCESS in 57m 08s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8b3355023f2d42dd9dc7a9d0f02b6dbf : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":4},{"id":"d3cd48ed6653213cc22c6aeae68b5c70647328dd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-06-30 20:38:51.000000000","message":"Patch Set 4:\n\nIt seems like pdflatex and sphinx are not happy about the number of new config options being added in this change, this build error seems to be the culprit:\n\nhttps://zuul.opendev.org/t/openstack/build/84d73ec987d04c05acc54b231d33067e/log/sphinx-build-pdf.log#1874\n\nThe only thing that seems to fix it is to clip all the help text for every option, including the options that existed before, to just a single line, which is not a good option.","accounts_in_message":[],"_revision_number":4},{"id":"81f90fd55d934df171986459312d6aa9e8bec17d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-07-01 14:21:20.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"89ee6b77ec269b605d68e1ccbd6b3ce19823a656","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-03 16:29:00.000000000","message":"Patch Set 4:\n\n\u003e It seems like pdflatex and sphinx are not happy about the number of\n \u003e new config options being added in this change, this build error\n \u003e seems to be the culprit:\n \u003e \n \u003e https://zuul.opendev.org/t/openstack/build/84d73ec987d04c05acc54b231d33067e/log/sphinx-build-pdf.log#1874\n \u003e \n \u003e The only thing that seems to fix it is to clip all the help text\n \u003e for every option, including the options that existed before, to\n \u003e just a single line, which is not a good option.\n\nYeah, sample config is too large as single code block, fixing here - https://review.opendev.org/#/c/739262/","accounts_in_message":[],"_revision_number":4},{"id":"930c402d500582bdbdaf90df661123dd9f00a90a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-03 17:44:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"109214c26d721ee3eddee08fe03249c0221edb36","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-03 17:45:47.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e \n \u003e Yeah, sample config is too large as single code block, fixing here\n \u003e - https://review.opendev.org/#/c/739262/\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"61a0389bed435063b73a78638ad0ceca351ce763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 18:43:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/02719c94e5b44fb78b62f236a5b0b9ec : FAILURE in 23m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7879dccd5ff34a3f9011e6b8bb9fe25f : FAILURE in 21m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3ef99dca2c44790afb950af148b3f22 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83570d6fd3f94d788d64ec020faa808e : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85402ae0c78d4092ac9ec231930f246b : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/464f13700a0444b494101daeb49607f6 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33a8f77438c48149de5d9b5a245cef3 : SUCCESS in 12m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/76129e85199141928f32c6d6d17c8136 : FAILURE in 24m 48s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6b3092e6a54c4bed98fedde7ff5ae071 : FAILURE in 21m 11s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/78f0c4e034234b8e8ba3b9ca4230267a : FAILURE in 23m 06s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a27858e4c65147299f7bf6c81f19ff32 : FAILURE in 26m 48s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/03985ef1386e4bdca45c43b702997a90 : FAILURE in 28m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ddd10235fd0a466caddb7e408173d8d1 : FAILURE in 27m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/284f0eca43c74c29849d719bfb9df71a : SUCCESS in 43m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb6616446865450a9459fedffcfe185e : FAILURE in 20m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/83947234fe3a496a87ea11a52f95f589 : FAILURE in 23m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f27312dc782c4174b50730cd819d661b : FAILURE in 33m 49s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9442280681f2493587c377f237742ec5 : FAILURE in 25m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9dbe4b48a98642f984d0142d5942e4ae : FAILURE in 30m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0d404fcef96c4e4f9c23f88e98adc884 : FAILURE in 24m 21s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/cec2da63b0784dc0b01fc90acc195259 : SUCCESS in 5m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/27f446530bbf48feb8398695740ad70d : FAILURE in 21m 18s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/abb2e13e6f6a41b095e1dab96dbdb931 : FAILURE in 19m 44s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1749790e193b4ae398c90b67a0c77cf2 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":5},{"id":"f7b4645c020f911a91242d7fa362749b3d4977b5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-07-06 13:11:37.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9ae9c0cafd75d59de052e49638074a18e099ad8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:29:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2582a3f8936d4d6c8b838a2640889845 : SUCCESS in 1h 02m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76236ef89c7d400a9664ae43cbe37463 : SUCCESS in 1h 16m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57b54c219f704a079c40224cfd3267e2 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd1c8588e874ee88fdb95cdebec42ed : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cbe57e029d74dcda8655b31fd9f18d8 : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cbc93899b154c42aa5d122fa8902150 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4abf64ddf0bb448999599cf51f8cd6ff : SUCCESS in 13m 34s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2e82548a64414f4abbdeeb207fbc2d0a : SUCCESS in 1h 37m 54s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/1e93c7ef7de642438f7183e6df3ab8a1 : SUCCESS in 1h 16m 27s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/8bbe7ca39ef441cb9424a74bc5e8cbd7 : SUCCESS in 1h 25m 07s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/63371f5f8851458fa5d104959cb146c6 : SUCCESS in 1h 42m 44s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9bf5a44afdf74cc5878f4ea1b474e68c : SUCCESS in 1h 21m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1bfe5e050673406b89506542f9ab2118 : SUCCESS in 2h 00m 17s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e6509217952d4dc6a80e503a6bbc0c49 : SUCCESS in 47m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6c07966bf7a4d128b46f6fb95fe983f : SUCCESS in 1h 02m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89320380b52843cf9b84cbfaf086f6f8 : SUCCESS in 2h 11m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3bbbf13ad0b540e6b635a368bdac365b : SUCCESS in 59m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9e87a7779c64440b964d4f9382fa0be8 : SUCCESS in 1h 10m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f8b053c74064e7591685fdfd36b0514 : SUCCESS in 1h 08m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6e0348bbb28e41dba899c60e2a85f772 : SUCCESS in 1h 12m 17s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6029ccf0866740f793acc87292e170e5 : SUCCESS in 6m 49s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3e278246b0d44c51a10b2f1d3fb80003 : SUCCESS in 1h 18m 50s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/993d315c73b645eeb205ac62798ca16f : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f1ad0dba74414a19a9b54d0da9ea9959 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":5},{"id":"ac46f6c77dd2c1b6f38fcb0ba662e8dc76d6d27c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-07-06 17:24:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d591553080139c5f9d4f06f85c9a23e25033174d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-06 19:03:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d497c86ac09ed07b098c28e49730cc6b7e5cb423","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-07-06 19:31:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc8400f6006fb1f9f173ec092a7f1a49054f848f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-08 20:06:18.000000000","message":"Patch Set 5: Code-Review+1 Review-Priority+1\n\n(2 comments)\n\nThanks for working on this and apology for late review. I have labeled this as Important change.\n\nReviewing the complete series before +2.","accounts_in_message":[],"_revision_number":5},{"id":"c26cca0d9f535c68098cd654fefec6f07f9eb1bf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-08 21:34:41.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8ae7750fb2d6536cc949b97d59bb51d9b249dadd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-13 22:25:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"be194907ec38875e01cb25e87c6ac3bc282ca456","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-13 22:48:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"531d9ed5d190b4c7d3a9cca5841ed4b42ad8606f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 02:00:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b8da0f2030e74e7dab6cab69647bdb70 : SUCCESS in 1h 12m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49e1ba4cb2874f7098275281c4563bdc : SUCCESS in 1h 44m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9ad9b26f16b46ad8e9f26a1ef95320c : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b9b551e5634242b61a7811df8ea07d : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc1a8ebb03c147a08c0a78ba3060ddb8 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e303b7ceb7044a3bf34a5dce89d73f3 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a870635d9143609fcabac606f1cc53 : SUCCESS in 26m 07s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/77a2fb0085004a8599afddc8480b91b6 : SUCCESS in 2h 07m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/38386d4944184b30bf8e482fe7b2c496 : TIMED_OUT in 2h 08m 17s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7f0d983df37a46a985a23943fc8372b1 : SUCCESS in 2h 02m 05s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/83b629652e954ca1b4e4ed27e50a61a8 : TIMED_OUT in 2h 08m 06s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/328ab266816748919bbf02352a6bb11d : TIMED_OUT in 2h 05m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5f4d598d3ba841868500eb09589e46dc : SUCCESS in 3h 09m 04s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/cf852e6a3415409b913a3975ddbfd5d9 : SUCCESS in 46m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68324feb784442439eb658b84922ccf1 : SUCCESS in 1h 39m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3297fc4a8d5a41b08eb493e1600fca3d : TIMED_OUT in 3h 06m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f743f7d756d540cb9cbd68d92c357601 : SUCCESS in 1h 27m 02s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c2d9063389864b0d946aad0a4943fed5 : TIMED_OUT in 2h 08m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ae76c50b44c4b1b88c1f4904bdc2393 : SUCCESS in 1h 51m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c4c7c85bde484c2f981365d632c0aa9d : SUCCESS in 2h 20m 26s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5055dac3492f404ea844451eccd29871 : SUCCESS in 5m 29s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0426f02927ec423b9c1e33a1f673bc97 : SUCCESS in 1h 56m 01s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4019e64ca6d04c8c9c709569e1cbd768 : TIMED_OUT in 2h 07m 58s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b717e123e7b74dbbb4c458c67c6208d7 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":6},{"id":"b4fad2cedcacd3b54ca0a6daefc76fd2f1bb1a01","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-14 14:48:43.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a8e3bf9910e986d6d0469bd23d49e296632a847c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 17:57:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/405899c0fa4841b1ad9375a7bd5a98eb : SUCCESS in 59m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a5c4a324ab749578e7f54927063da61 : SUCCESS in 1h 26m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8018ca27c6a473d821aa757e3024d5f : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be3c926193b41a4a9146dab5f6afdcb : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7abbfa9f8ddd4f57841c419b3dc3cc55 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/014985ed5a424d1685ba8ab868deedf2 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293092dd98ec4dc5a4df07d5d4461116 : SUCCESS in 8m 20s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3e635556011946d4ad76b8aa488db7d9 : SUCCESS in 1h 39m 35s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/cb87664a3b1f450cb2315b0302e16db2 : SUCCESS in 1h 36m 24s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6612242c59a64e1d9bba3a2cdf83ba9c : SUCCESS in 1h 34m 16s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ded322ef3c6a424ab9497abacc8c28c9 : SUCCESS in 1h 05m 25s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/f035ad478123478087b255312ca05e42 : SUCCESS in 1h 05m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4a0dac8b797f47e698cbb57724bc6dc2 : SUCCESS in 2h 07m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7d98b171921a4d86975776cea04a9118 : SUCCESS in 41m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28425b344399405dbd6c1ef343a03239 : SUCCESS in 58m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/672ea23e1be144e08c4f984cf545c926 : SUCCESS in 2h 16m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/898d3312f64247bfa7ff159c303f7ccf : SUCCESS in 54m 49s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1b03a0a407374ddca258935b6220f382 : SUCCESS in 1h 09m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/58604f2b9a654efd9624de21887b5053 : SUCCESS in 1h 08m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7189c45136934dc19530d08a3bde8bed : SUCCESS in 1h 16m 27s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2d4181b589dc41329a5e85f632151e10 : SUCCESS in 5m 03s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2cfd13ddbe6749e8b8934b9b4809645b : SUCCESS in 1h 27m 46s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6c4248dc580442c1a85adb66e03a6613 : SUCCESS in 1h 12m 43s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9e446690897340dbbd5a73844af50f22 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":6},{"id":"bc60771c13e35ad4351fe912993ecd2e25c6bb55","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-30 02:38:30.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b474460508914006ea96fda68f187dc32bb93c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 04:33:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fdbc959158a847d486d77d594b44d782 : SUCCESS in 57m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9430686754342beb7010660f6dce658 : SUCCESS in 1h 22m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bc90b2ca6a14f7f9579a5230c134852 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0e4f124b0345f8ac8d91f18241aba4 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c980479dba9489690ba51ea87b17278 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d54539e3c154dd48b3c200fa8f11dc6 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/108fa51ed0914b3b9946c3635ad7db45 : SUCCESS in 14m 06s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b10b9e2cefd442e7821baae31e5d4fb8 : SUCCESS in 1h 51m 54s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7d9cd277a3374beab6bcbeb3502e13b9 : SUCCESS in 1h 17m 03s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/a43cffa2009e49719a5681ff1dfaccf7 : SUCCESS in 1h 17m 16s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7cfa77034f1c49b38d1436a6c6c7c4e6 : SUCCESS in 1h 15m 05s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/d85cabece3ce4be78d0ae09bd75fb150 : SUCCESS in 50m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/acbfa920a91d4f16a49a936b03867f83 : SUCCESS in 1h 35m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/82dc5bcc7b8c4e41ad2ce4079267fc08 : SUCCESS in 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b07c26ed2f564fbb8deb5484d283cec5 : SUCCESS in 57m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23cac69abba649df87d57410c21969d4 : SUCCESS in 1h 26m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/16bdcc5b48c04b23bef64fddfe92ca80 : SUCCESS in 49m 02s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4058ea4052754e75b019a574750ea378 : SUCCESS in 1h 04m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50a895b2a660460096cd9af1464b29f8 : SUCCESS in 1h 07m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0ee49f18d2a146de903efab0109cc750 : SUCCESS in 1h 27m 11s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0c101f304b2d4fb5950efb33ad343c3f : SUCCESS in 2m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a2502eb9493341bd8d91cb7bbc59a46a : SUCCESS in 1h 23m 28s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ecea0695592f43aeb06da639430d28af : SUCCESS in 54m 31s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7a4eb00554eb40ab952313a1a61d6e88 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":7},{"id":"172d47d143e6f8fe7c2f66fa617ee94168e4e3e2","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-01 18:38:03.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)\n\nlooks good, I turned on enforce_scope in my devstack env, I successfully  verified that unless I set auth.admin_system \u003d all the tests are failing, I\u0027m getting forbidden error due to not being able to create dynamic creds.\n\nother cores, please review, due to the complexity of the change it\u0027d be great if more than usual number of cores verified this.","accounts_in_message":[],"_revision_number":7},{"id":"04ee27c8d5759d197bf1e6dbf0b2fb84651f34ff","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-10-01 20:50:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"97cad6d19bbf3e7dc288fbe39799936fd66774ec","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-10-01 20:51:12.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f03b91c07f9c346f02ac9b330d4a46f7e0c2c41b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 01:41:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/12fb6d824a914d0f843ada7cf0bb2109 : SUCCESS in 58m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c503eaf2402743dc86e629fee69cc778 : SUCCESS in 1h 20m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/968150b4e8234d79bfa17be44891be54 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7a360cb6824fbeae49f9b0dfdbf10d : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9108f514a924d68ab50def352f7fdcb : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0ed1fa4d9654f218e36065f4fdd0e02 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838d0c6363ca4ef09af0af0d9c5fcdd3 : SUCCESS in 11m 50s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c3aac4bab622425387d842fdcf469038 : SUCCESS in 1h 30m 00s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/10efc2c15fd44ea49a2c07864d7f2c08 : FAILURE in 1h 22m 31s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2b817f7137ee43aea484221183b4b6e5 : SUCCESS in 1h 28m 23s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6a91aece245a42c9abc1a7074abf6c7f : SUCCESS in 1h 20m 44s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/904c82d4cb5d45d981d7a6fc83a27a23 : SUCCESS in 1h 20m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e420f56697634bf2a93b8f163cef296b : SUCCESS in 1h 35m 43s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b61fedee4a0f4299bd65f71988989cda : SUCCESS in 38m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0792d50190e4aacb1fb5d3a3e5f911f : POST_FAILURE in 1h 07m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aec592339f4e4737a206478cd6c0c769 : SUCCESS in 2h 03m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b5499eb0abd04301bd88d1a721f9bac7 : SUCCESS in 47m 06s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e6016814986641429e07b83e4f13f948 : SUCCESS in 1h 01m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5446020635934c60bbc6f94dea50a486 : SUCCESS in 1h 05m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0bd07fa5f88c4f278f546c55a48a644b : FAILURE in 58m 06s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7179d1ad17d848c68df0c4df893e8cfc : SUCCESS in 4m 04s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/81058edf9f7141c09d24450562d8f457 : FAILURE in 1h 02m 51s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a44e8fbe3dcf4b828435b8754aa66a60 : FAILURE in 59m 51s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6b08d388398346b8bb0b37d22e7eee27 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":8},{"id":"15dca584e86a2fb511075e997ce98e960a6e1e7f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-02 08:28:03.000000000","message":"Patch Set 8: Code-Review+2\n\nrecheck .. thanks, it LGTM","accounts_in_message":[],"_revision_number":8},{"id":"59c8542d121d542721dacbae1f76e43d060a3840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 15:16:04.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bfedcf18b589403cadcd2cc546490c5a : SUCCESS in 1h 00m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24311ddea05f45b7bf584ab017b639d4 : SUCCESS in 1h 22m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c378017c48914a07af63f3a5d1867a04 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1166c2ec0a24b97bc0e2bc051a6ec52 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fe4e795f2f8467bb41b0a2d80c76539 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/889bf79dc50c45119ca9b2553f907cfd : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced1aadc0c5c4f76b5934bb51dd0c936 : SUCCESS in 12m 30s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c3db8ee6a15847cbaff59ff833d969a0 : SUCCESS in 1h 22m 48s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/770d07b7700a459eb22fb05a2cda59f3 : FAILURE in 1h 05m 34s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c65eaf9d7ff94e05b148169276dfe97b : SUCCESS in 1h 17m 55s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/853d02dfd7fc49c6b35f61ff90fe2bb6 : SUCCESS in 1h 25m 35s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/806b6c5170fb46b0abc607ab5d3b7c46 : SUCCESS in 1h 15m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2fb7924254b8437aaa804aa238585139 : SUCCESS in 1h 47m 38s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9f3e2d529a9f44a9a185c62259c48861 : SUCCESS in 48m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ba4692dcdbc44d6801a9980236cfdf6 : SUCCESS in 1h 03m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b270fed3ed544b2fabb42d81baeac9fc : SUCCESS in 2h 12m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c21fe9093f3642e2a0cdb74f049d0595 : SUCCESS in 50m 27s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/26245f6675224903b68b5d206e869ce7 : SUCCESS in 1h 02m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff03ab3e72f64d41a807cafad20f8e7a : SUCCESS in 1h 10m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7015a940c79141c786cbbfa6fa9fd6ea : FAILURE in 1h 01m 21s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9e22fe8937fe4d1da4fa929095d0450f : SUCCESS in 3m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/bb79419892774969ba123846890b1769 : FAILURE in 1h 24m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f8d8b873dc1e4d7e949b87ba9fac5a96 : SUCCESS in 58m 29s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6b9602ed0d7e4ba89231e432f91e23a5 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":8},{"id":"7781f9256f22f194538e28c6a8a874a63b85481d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-10-02 16:24:42.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"ba4c1959ff03d2caad16de24e09ec1b33c3ce0d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 22:00:11.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce40c6f780f14cc08b110c6fbcf941c9 : SUCCESS in 1h 05m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fa4993914f64b04969b82cd249d4324 : SUCCESS in 1h 22m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8c0b2dfb6334850b501b5bb64c626c0 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7c9357dcf344bcbf189e4d2069b8cc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/888f20d49d17497588a95857dd474321 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a56f038f157245a2bfa50a5d7ab67afc : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4493a804e19d4021b6ccd04f2cd6c671 : SUCCESS in 14m 30s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/cff16f91f56e41b9ba2f56fcf91e20f4 : SUCCESS in 1h 47m 31s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a1db37f488e740cf905be18493f42445 : SUCCESS in 1h 26m 04s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/fb3ebd1114b94b24aab83c69a3e7e7a2 : SUCCESS in 1h 07m 12s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3f4143f575f6437891c31e4da5e21278 : SUCCESS in 1h 14m 51s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/354d6f07d6dd42b3bff57ae9e3af5ae2 : SUCCESS in 1h 23m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4d1ceeccd519421da1c68de301b73f0b : SUCCESS in 1h 40m 33s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d76b046c14b24bbeb7aa5dcf90881d7a : SUCCESS in 44m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cf3570458ef44d3afcaec5f46481aa3 : SUCCESS in 1h 03m 56s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0fd5027a94e545c2a7393bd5337050b4 : SUCCESS in 2h 02m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0814275b27e24d979a67dcd0d68b6ef8 : SUCCESS in 53m 02s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fc4242bc450149e5938aac6cd7dd1867 : SUCCESS in 1h 08m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7127b4af382c4ff8a03822015cea4013 : SUCCESS in 1h 12m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5bbc97f26a5f458a89ef897012cc2156 : FAILURE in 1h 14m 05s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1c55fb9abaab474ba54b41746b8a76b4 : SUCCESS in 3m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/10b00ab52f5041dbacff7f2c85fd7205 : FAILURE in 1h 11m 23s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f5a100799ed840ba8cd7e113b86fe09f : FAILURE in 59m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/aaa7d854e9524ad0a159a6a2d411e97a : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":8},{"id":"94ec2d679f7226ce6fdff8dafb171f06e9b7ea1d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-10-09 17:04:29.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3d631dfad5d7803e6910c66dfec7198bb2a09509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-10 10:50:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/aac466b7320e4be8b47abaa38b9e1d61 : SUCCESS in 1h 03m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91015b35d4e64d878683bb468c5f4813 : SUCCESS in 1h 32m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b25abc228cd4118b3d1be65f7c7a43e : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0531ce8bfd44ff09bcd0c1583b365ad : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2867ded96f614aa9a2533af2417ad017 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/265f86517b184241a348b818c0bd8f72 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09889c3fae1d4796995cb5aa2263a39f : SUCCESS in 13m 16s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b1efa999aef24691897aae249361616c : SUCCESS in 1h 39m 08s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e51d8e0f967548b3985669281ee63e83 : SUCCESS in 1h 17m 04s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/fa0c252818444c5f894a4f580e7f443e : SUCCESS in 1h 17m 10s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/507ab3f167d94201a5690ae33c5010c4 : SUCCESS in 1h 20m 38s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/06d73cfedef74760b102219dfeec71cc : SUCCESS in 1h 15m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee02ff24a9d04b8aa740b9e8066c7177 : SUCCESS in 1h 35m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d64b3ad69c0e4fb7913e0a8f497a1ebc : SUCCESS in 41m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/565f0e12c47840e49fc706e8ddab369a : SUCCESS in 1h 13m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e06c0b3687e4f8aaab94ea83388455a : SUCCESS in 2h 13m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/27f2179201ef456c85943e2d99e26d73 : SUCCESS in 1h 19m 48s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0d0fb3138de04d649f5f97e7c26fdf0d : SUCCESS in 1h 05m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e976cee854214b54af16dc51c5196bd9 : SUCCESS in 1h 07m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e81569b9ee99460db0c7aaea008bf5dc : SUCCESS in 1h 11m 29s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2429a75a1be642c7a3063a68569f44ca : SUCCESS in 2m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4553dd52bcb24bd596278c774fe799ff : SUCCESS in 1h 10m 14s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f9334c61b1444482ae8249f94bd9c127 : SUCCESS in 54m 48s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0f258280042b4cf7b2857b195fcb7675 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":8},{"id":"f7ce9743ffa5d08776befe00595907da45f845e5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-29 18:39:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"41a319a02fed3136041d228c547ee0cc12bdbc5f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-10-29 21:19:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e2212235ed1950ae29f8dd8a6470a67730b6314d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-10-30 01:38:34.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"999d977aec7afad0e3a9850e1e5b2395fa269029","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-30 02:09:05.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"cdb75c8f9f7a66398b18785c6cde0fe3b1838ddb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-31 18:01:44.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\n(1 comment)\n\nlgtm, testing the complete series in https://review.opendev.org/#/c/613251/ before +w.","accounts_in_message":[],"_revision_number":8},{"id":"4134248fd77d0f30a9c6d6a5e6467c0ed84599b4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-31 20:25:37.000000000","message":"Patch Set 8:\n\nlet me rebase it for https://review.opendev.org/#/c/740122/4","accounts_in_message":[],"_revision_number":8},{"id":"5ded53dcfa76d533eaa907d4cb48b4e0a4cb4593","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-31 20:25:41.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"d298227be5f85a935a171c52474faad3d487a87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 22:49:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b7780b1d71c94a61a8a3241eb646832a : SUCCESS in 1h 03m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eca37e16cb534ca6b375d351db64a8ce : SUCCESS in 1h 19m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e68cdd72380347279203cd0be4b611ba : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296d8d3306b1440eab4d40b1ac1ff375 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9bd4c60efb04999a97642d0753d7ac8 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a20f7293d9af400890426a7251fdf48c : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fcdf66d6904a2c82b0171aa3cb798c : SUCCESS in 11m 54s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4f85b4c29d2a4eddac933e4eadb92841 : FAILURE in 1h 33m 26s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/37b88c80db124eb7af05c64d30f36920 : SUCCESS in 1h 37m 36s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/3da1840d19ee4dfa86e1e4334f8c2da7 : SUCCESS in 1h 17m 54s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/759a194d225f4a7694ca17e6ecef2907 : SUCCESS in 1h 37m 25s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/8d144c3d34494d73836405f3abb07744 : SUCCESS in 1h 23m 32s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/8ded0204e62248518b38182cdf2e51e1 : SUCCESS in 1h 27m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3172415033184330bae72bb2235a95a8 : SUCCESS in 1h 41m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/109fdc4a100f4d28bc4ae7b169d075a9 : SUCCESS in 29m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f052fd6d82b471880cb04ededf2b563 : SUCCESS in 2h 14m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/334879286b9644e48eaff573d9f8b7af : SUCCESS in 2h 06m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9b7a552b78a74e2f8cb7ea86e6f8419e : SUCCESS in 1h 05m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3f0cd59433f64a1d852797d97c2c08b4 : SUCCESS in 1h 05m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c72fa0c7c4c74881b93e167f0fd5a240 : SUCCESS in 1h 07m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/65640c9a6c4044f1b024b42ccddec58c : SUCCESS in 1h 16m 13s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f76cf610f1704ab1bc66d132171f6416 : SUCCESS in 2m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/98315bd0e58d4ec385eb32d9712234f6 : SUCCESS in 1h 14m 43s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/7dd07831ae2a4d49a5e12027d40ffd31 : SUCCESS in 1h 18m 44s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3d0dd8e0fef047d3b0a517f44d2f7b15 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":9},{"id":"0be25b8f9710313a27a375b0ed4717f67a977491","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-09 18:44:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"67134876eb6874e1b636c639906b5acebd1f8dcc","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-16 23:26:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a02d72d757660ec7f4cc0d90c49e690ad6112bc3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-17 14:57:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7362662bbc2532bf8092c1201d75d0334a8c1660","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-19 18:14:23.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"06457c5665b79c7cec936480e4f564aa62910361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-11 19:11:45.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"98db88c1ad9e5c78837d1eae34a6703273b16fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 22:26:22.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c9cd54706ce34ef78c53fd7b39b42e85 : SUCCESS in 56m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9d95b23177542deb2d580e2a4d1f59f : SUCCESS in 1h 16m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f91b77e3059417194e497df09a1d37b : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197607545b6e40b7b72e9349c58343fa : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cc77aec59814d87b92c58d0d413364c : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce7d9c22159d428db383f4640e9c8414 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156a623e73fc4d9cbd927fd7b883d48a : SUCCESS in 13m 15s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/601474692e8a46abbd12ef8611a0a14f : SUCCESS in 1h 38m 15s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c9dc5c15cdaa42ceb56bb258d21953ca : FAILURE in 1h 13m 31s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7e448087a4594d198140db8c1e4411ca : SUCCESS in 1h 30m 08s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/fd4b15a5c59e48f9af1873b8dc54586d : SUCCESS in 1h 14m 36s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/21e20e8ecc1745f39b1de2590d7b87cd : SUCCESS in 1h 25m 27s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/cc0c985e30654f5eab3bf52835492a89 : SUCCESS in 1h 05m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0229d84b2fd34aa6b43b637495daef8d : SUCCESS in 1h 44m 20s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0de8032c2ca24361954012f535369f64 : SUCCESS in 47m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/888cf70eda9147f6b0d25e012d35773d : SUCCESS in 1h 01m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3cb147efae9e49cc97f8506519abd1c8 : SUCCESS in 2h 16m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2bbcbaa0a85846b5ba894cffb5bbddb7 : SUCCESS in 47m 38s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b0d1a5f27129433cbcf246d602b29e21 : POST_FAILURE in 56m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef84af6fba754b739c6ff4328af0fac0 : SUCCESS in 1h 03m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/36322e7b8d104ed49a3948e3b344d35d : FAILURE in 1h 01m 55s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2354f68842894d248670cb1e61c64c46 : SUCCESS in 5m 22s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e75c543f32484fe0b9d48e9688e58dc4 : FAILURE in 1h 07m 17s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a855b933f2494912a18cf86cede37800 : SUCCESS in 1h 06m 08s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/572d4fa03cd84b708316705466c41634 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":10},{"id":"f1c7fcc8b65458f754e288f97b62239616d5c037","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-13 08:51:30.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"6d4e1a69fe3c7a443c1d90a764817b7dcd6c6feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 21:51:24.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/065b5aaa411b4b91b1e0072f9412e350 : SUCCESS in 1h 00m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3be01406bb0410d9bdd9c16039aa2a7 : SUCCESS in 1h 18m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a1bb06feabd4eccb9e7a9213f9dc4a3 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64285b0d843549a6be03b4692fc63970 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/254ee87ac321402bb8e26e034e99284d : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/979ee95daf2d4843ae24d9a9113b5323 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d4769f25684720b8154d31a7974e3d : SUCCESS in 13m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d9f27e4007f6468daab46148dde2d73f : SUCCESS in 1h 50m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/9cb512045c2844029662d822edbfaffb : SUCCESS in 1h 26m 44s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/72535f7b14cd41dcafff3cff3aef8ee1 : SUCCESS in 1h 16m 35s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/8e91eb33825b434cbec66059e0eaf188 : SUCCESS in 1h 44m 32s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2923d305692748be82c807bdee158c23 : SUCCESS in 2h 01m 34s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/4f8aa211ed8c455d9c244802bbfcaf71 : SUCCESS in 1h 21m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dace0735d5f949a7a936d71a004d8d31 : FAILURE in 1h 35m 33s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a5c767bcd4034f648178ae3c641b3dc0 : SUCCESS in 41m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5945785737874422b825d6b98c2c276a : SUCCESS in 1h 17m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/358b419abcc640d3801aa6d48c43c832 : FAILURE in 2h 04m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3aae9281f0454d77b897dc3daec4d4c7 : SUCCESS in 55m 51s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a30601545b854a8fbc77cbe928eea5e8 : SUCCESS in 1h 09m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/47e872eea8c64bbe9a1acf32f715b446 : SUCCESS in 57m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8551364ee5de4689bacecea7c7a9667f : FAILURE in 58m 42s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b19fa69cf09d4671a8f7596636dad605 : SUCCESS in 2m 56s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/dd8474e396014d689d6d0a137ee0dec6 : FAILURE in 1h 14m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e78db9eebed24ce7bb07c96d3a662632 : FAILURE in 1h 03m 32s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6feba406e39a478abbe5ae5f75cc3895 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":10},{"id":"2a6d9449fe6b08304403ee970d43d7e35f0fc802","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-19 22:49:31.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6199a62ae66eb26920d9ad6f3756206135b99936","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2021-01-19 23:05:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c195b3f96aa352ddcc92c5b004e2c0bbf0bb3d98","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-19 23:14:15.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nall good here, waiting for my last test scenario result- https://review.opendev.org/c/openstack/devstack/+/616416/12","accounts_in_message":[],"_revision_number":10},{"id":"67d1a04e9522fc37aa6e454b5a93cc9e74569fee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-20 19:39:23.000000000","message":"Patch Set 10:\n\n\u003e Patch Set 10:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e all good here, waiting for my last test scenario result- https://review.opendev.org/c/openstack/devstack/+/616416/12\n\nit passing as expected .","accounts_in_message":[],"_revision_number":10},{"id":"1839a90a27f48e72de002fcac204829f41b1cfa5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-20 19:40:03.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nlgtm, All scenario with new and old policy with new and old token work as expected - https://review.opendev.org/q/topic:%22system_scope_testing%22","accounts_in_message":[],"_revision_number":10},{"id":"eca550e1f4a1f341608da03c7e6af56c15a218a3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-20 19:40:08.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8088fc91971dc2722048447876a28a50388123c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 00:49:13.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0e7a9b703504c2b95d5e3c10a4705c6 : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b68794d60024de2aba4b6ed1c32b86e : SUCCESS in 1h 19m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/584c9932ea3d42baac51d1eef6a8a27b : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2580ad66c5b4a088b783a43fdd5992f : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6606d678b9fc4da183c1e6ee0d6479ed : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af5fe5fff1154d05989b84f42ede9d43 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3476ad33b8c448480359839d4f02e0c : SUCCESS in 13m 40s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/19a1d68883aa43d58797acc9e27e8fbc : SUCCESS in 1h 39m 03s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/612966e02fe44a3e8f63c58017ac7ba6 : SUCCESS in 1h 18m 37s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/975c8e0e6cc346aea1791983ab978707 : SUCCESS in 1h 07m 57s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/a7b54b1334a5479dbdcbfb8eeed1de43 : SUCCESS in 1h 26m 52s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/ec86a151195a465098a46a0d2c4b08ca : SUCCESS in 1h 13m 33s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/604097328e8843d98f2dd2733144987e : SUCCESS in 1h 11m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/33c77c03ed3244e7a302b3463ffeafc0 : POST_FAILURE in 2h 13m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/75c3aa41363849759322e50682d21953 : SUCCESS in 51m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2f746dfa2a74c61906d8a7ef0ffc89d : SUCCESS in 58m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71f81f1073854bd6bd2c69f0c772bd47 : SUCCESS in 2h 11m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/04a6e30fb6ae4dfa8abf6f51bfb81c65 : SUCCESS in 55m 17s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1fd024bc1fec4cd7b20b56bf27ad1d6a : SUCCESS in 1h 38m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/330d8c63ff27480e94fc5a92d8c5c0be : SUCCESS in 1h 18m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5e5c4814831145588da3f0adab21856a : FAILURE in 1h 00m 31s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3ba14079fa7c43a0b47a6b15fc203165 : SUCCESS in 2m 55s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/50b987a6ab254c1d8bad95c80a2be396 : SUCCESS in 1h 13m 02s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0e5959d3086d47e28398d8328f47dba2 : SUCCESS in 1h 04m 19s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d68305771f5a48faafcb445a9d515c08 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":10},{"id":"d6c473f7bcf4f1129528d3aaf596bf3ea37ae4e5","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2021-01-21 01:05:57.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"2478ac223d4455cf97dbc5be9e29d3039523bc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 03:06:48.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/de94c71c6c64494393812df085d94389 : SUCCESS in 1h 02m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1613dda28b74b5eb907bdd8e20e627e : SUCCESS in 1h 21m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df7b1f46d4794026ab91bae775c024f2 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c5aeb2675843459c92576ab5e111a6 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9c0956a061b48f68daae10a8d08b28a : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76991d89d985430383f128bebb1c0e3d : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ba6bc6e713472a9092db8b43978f56 : SUCCESS in 11m 43s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/97c5fd6c345d49b781ea9f61accef6ba : SUCCESS in 1h 39m 19s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/38c054c5a0544c3cbeea36e53446d408 : SUCCESS in 1h 17m 07s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7945b40d5ebd472482f9e55201967a37 : SUCCESS in 1h 15m 33s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/756d2b2a1fe1459cbd24a92bfb29c5c1 : SUCCESS in 1h 27m 02s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/ac3109cb1dbc4eee995da22dae75fbd8 : SUCCESS in 1h 15m 25s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c854297c9e334ba1b60a4f7ea8507e56 : SUCCESS in 1h 18m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6d94d3af37b84680be04d1e6cc5cf954 : SUCCESS in 1h 36m 32s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0d4cf68a26c74b79aed4d6c0d69bf84f : SUCCESS in 41m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e8cf7486db542b18bc4b232b8b0b69c : SUCCESS in 1h 01m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c2b6645da054a6184833ff945f73d2e : SUCCESS in 1h 57m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d9a5baef8ca44a89be382289b6e9f6b8 : SUCCESS in 51m 25s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9e3d3fef2c024fccae9ccf885190ff23 : SUCCESS in 1h 03m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf111285317144ca907b0fe57bc21a26 : SUCCESS in 1h 02m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1a26be35d7ec4457b3ba410cbc5e2ddc : SUCCESS in 1h 06m 14s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/64170f50302349c6a19ce751e86ad86d : SUCCESS in 2m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/865c43b125214d3a92c611aa3f900787 : FAILURE in 1h 22m 56s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ddc24201495f4e00819dc136085a5002 : SUCCESS in 1h 16m 32s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ef1a293df94d46d590a55f5d7c9535ea : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":10},{"id":"4a659339d5d9902dfe6f8c0f2c8f953d91ccd5ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 03:07:13.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0cd2ec79b9a4efd4ea08f8d51e1cf4848695787e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 05:18:23.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8d495df9033c43f98bf50eac39532eed : SUCCESS in 57m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c42ea819406f461f8f2d5fbcee3d970d : SUCCESS in 2h 03m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5147c5879a30433f8205f7c177a511d3 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fcfd92a29564b529f3e9fbc12317b0d : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81982b59f7084c8f8fece6696e01c9cb : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2e08ce90224b3fa7d67e43ff9576ee : SUCCESS in 12m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3e57234418e4cceb078082d420f5005 : SUCCESS in 1h 50m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dc99d29a0a4415aaafb748408f95681 : SUCCESS in 51m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7c8b37a2a0940e9b171d75c05a91218 : SUCCESS in 50m 51s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a54ec1d6f56b401786ae69d6e2a6d35b : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":10},{"id":"0d09f1034aa26d546c258bb69f947f2c6d49e040","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 05:18:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"f3ea921e8b946300a7cc1b04f630f0b41173e592","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 05:20:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/908cf59e1fdc4da19d9a787620589e12 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":10},{"id":"eba96a20542f57b6d2dd1bd041b3a6ddf7c76712","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-01-21 11:12:05.000000000","message":"Patch Set 10:\n\nWe have some of our jobs failing tempest tests with Unauthorized post this change, is this some regression which not caught in tempest ci?\n\nexample:- https://logserver.rdoproject.org/ci.centos.org/weirdo-generic-puppet-openstack-scenario001/18199/weirdo-project/logs/testr_results.html","accounts_in_message":[],"_revision_number":10},{"id":"83821f7ccdff6a79721f8b05c477899ffddfda45","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-01-21 13:49:11.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2b941d18b03674c798e0c81d845be6521318ef4c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-01-21 13:51:28.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"285ec5fa6b166669aa32211cfc130f741950abca","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-21 14:21:03.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4f96330cda11a6c82f284e31fbe38e9b826a3952","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-01-21 15:08:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3ce1ca742c8fa0e2a6def759498ffbcf42edb094","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-21 16:26:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"cd0bbbdad37a31248d479ef78df948da0a1e850e","revisions":{"088e03c351b4d59669a7b962eceb4fa3d8448375":{"kind":"REWORK","_number":1,"created":"2019-10-01 23:26:57.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/1"}}},"commit":{"parents":[{"commit":"b12f6e72952445d7ef1e7459942ca5f3754ce5d4","subject":"Merge \"Add workaround to handle the testtool skip exception issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b12f6e72952445d7ef1e7459942ca5f3754ce5d4"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/088e03c351b4d59669a7b962eceb4fa3d8448375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/088e03c351b4d59669a7b962eceb4fa3d8448375"}]},"branch":"refs/heads/master"},"10b9d8c479599fb46c34f5dbdee87b7204486429":{"kind":"REWORK","_number":2,"created":"2019-10-02 15:41:29.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/2"}}},"commit":{"parents":[{"commit":"b12f6e72952445d7ef1e7459942ca5f3754ce5d4","subject":"Merge \"Add workaround to handle the testtool skip exception issue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b12f6e72952445d7ef1e7459942ca5f3754ce5d4"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-02 15:41:23.000000000","tz":-420},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/10b9d8c479599fb46c34f5dbdee87b7204486429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/10b9d8c479599fb46c34f5dbdee87b7204486429"}]},"branch":"refs/heads/master"},"80e312aef4b1a7062a45377253e2877586bcf18d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-11 04:05:09.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/3"}}},"commit":{"parents":[{"commit":"74e3e7c738730d973b2e9aa4f75729f0837b8db7","subject":"Merge \"Replaced testr with stestr\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/74e3e7c738730d973b2e9aa4f75729f0837b8db7"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2019-12-11 00:52:39.000000000","tz":-480},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/80e312aef4b1a7062a45377253e2877586bcf18d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/80e312aef4b1a7062a45377253e2877586bcf18d"}]},"branch":"refs/heads/master"},"c356408cfaae050e621ea9d8add6630533efef6f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-20 06:59:13.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/4"}}},"commit":{"parents":[{"commit":"367e148d439042ab43ea3fb610d7f6d456b89b47","subject":"Merge \"Fix 404 when deleteing volume group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/367e148d439042ab43ea3fb610d7f6d456b89b47"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-06-20 06:56:25.000000000","tz":-420},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c356408cfaae050e621ea9d8add6630533efef6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c356408cfaae050e621ea9d8add6630533efef6f"}]},"branch":"refs/heads/master"},"76e4b398cb4aa8f0e173dd7c5f21f7fdb6ab9b3b":{"kind":"REWORK","_number":5,"created":"2020-07-03 17:44:46.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/5"}}},"commit":{"parents":[{"commit":"367e148d439042ab43ea3fb610d7f6d456b89b47","subject":"Merge \"Fix 404 when deleteing volume group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/367e148d439042ab43ea3fb610d7f6d456b89b47"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-07-03 17:42:07.000000000","tz":-420},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nDepends-on: https://review.opendev.org/739262\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/76e4b398cb4aa8f0e173dd7c5f21f7fdb6ab9b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/76e4b398cb4aa8f0e173dd7c5f21f7fdb6ab9b3b"}]},"branch":"refs/heads/master"},"f9254f7c48ddbab7d8fe96dcfa24af2487336d6d":{"kind":"REWORK","_number":6,"created":"2020-07-13 22:48:57.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/6"}}},"commit":{"parents":[{"commit":"367e148d439042ab43ea3fb610d7f6d456b89b47","subject":"Merge \"Fix 404 when deleteing volume group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/367e148d439042ab43ea3fb610d7f6d456b89b47"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-07-13 22:25:21.000000000","tz":-420},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nDepends-on: https://review.opendev.org/739262\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f9254f7c48ddbab7d8fe96dcfa24af2487336d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f9254f7c48ddbab7d8fe96dcfa24af2487336d6d"}]},"branch":"refs/heads/master"},"91bf5d6a6820ae8071253df5bb2485c9db16ecc3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-07-30 02:38:30.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/7"}}},"commit":{"parents":[{"commit":"3b9f03ec6dc14ca3b1603728ad076ac63a8a5e2f","subject":"Merge \"Adding description for testcases - network part1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3b9f03ec6dc14ca3b1603728ad076ac63a8a5e2f"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-07-30 02:11:08.000000000","tz":-420},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nDepends-on: https://review.opendev.org/739262\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91bf5d6a6820ae8071253df5bb2485c9db16ecc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91bf5d6a6820ae8071253df5bb2485c9db16ecc3"}]},"branch":"refs/heads/master"},"13219f6dc930ceb8306c51a7daa9d454d9c47ecb":{"kind":"REWORK","_number":8,"created":"2020-10-01 20:50:51.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/73/686073/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/8"}}},"commit":{"parents":[{"commit":"3b9f03ec6dc14ca3b1603728ad076ac63a8a5e2f","subject":"Merge \"Adding description for testcases - network part1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3b9f03ec6dc14ca3b1603728ad076ac63a8a5e2f"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-10-01 20:50:33.000000000","tz":-420},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nDepends-on: https://review.opendev.org/739262\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/13219f6dc930ceb8306c51a7daa9d454d9c47ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/13219f6dc930ceb8306c51a7daa9d454d9c47ecb"}]},"branch":"refs/heads/master"},"5c0af10d1d11f9bc9883d46913bead5f642b951c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-10-31 20:25:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/686073/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/9"}}},"commit":{"parents":[{"commit":"b439c974faab1d593ce14ecb3cbfd5467b0005cf","subject":"Merge \"Add Tempest gate job for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b439c974faab1d593ce14ecb3cbfd5467b0005cf"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-10-31 20:25:41.000000000","tz":0},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nDepends-on: https://review.opendev.org/739262\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c0af10d1d11f9bc9883d46913bead5f642b951c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c0af10d1d11f9bc9883d46913bead5f642b951c"}]},"branch":"refs/heads/master"},"cd0bbbdad37a31248d479ef78df948da0a1e850e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-01-11 19:11:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/686073/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/686073/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/686073/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/686073/10"}}},"commit":{"parents":[{"commit":"fec2c93cdcc14ad08d0a35136ee287525e7a4879","subject":"Merge \"Remove stable/stein testing jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fec2c93cdcc14ad08d0a35136ee287525e7a4879"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-01 23:18:36.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-01-11 19:11:45.000000000","tz":0},"subject":"Add system scope for admin auth","message":"Add system scope for admin auth\n\nKeystone is moving away from using either project-scope or domain-scope\nfor the main cloud administrator user, and instead moving toward the\nadmin user having a role assignment on the \"system\" scope[1]. This will\nmean that no particular project or domain is special, and instead the\ncloud administrator scopes to the system in order to make\ndeployment-wide changes. Keystone has now migrated all of its policies\nto understand system scope[2], and if a deployment sets\n[oslo_policy]/enforce_scope\u003dtrue in keystone.conf and uses the new\npolicies, an admin user scoped to the admin project will not be able to\ncreate dynamic credentials for tempest.\n\nThis patch adds a new parameter ``[auth]/admin_system`` to indicate that\nneither the ``admin_project`` or ``admin_domain`` parameters apply to\nthe admin user and that the user should instead authenticate with the\nsystem scope. This also adds ``admin_user_domain_name`` so that the\nadmin user can be found in its domain (namespace) without setting\n``domain_name``, and for completeness also adds\n``admin_project_domain_name`` so that ``domain_name`` could be omitted\neven if using project scope.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n[2] https://bugs.launchpad.net/keystone/+bugs?field.status%3Alist\u003dFIXRELEASED\u0026field.tag\u003dsystem-scope\n\nDepends-on: https://review.opendev.org/739262\n\nChange-Id: I840b273c37ca7cc4592c43813abfb424337e2836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cd0bbbdad37a31248d479ef78df948da0a1e850e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cd0bbbdad37a31248d479ef78df948da0a1e850e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
