)]}'
{"id":"openstack%2Ftempest~695389","triplet_id":"openstack%2Ftempest~master~I70565411851439551e0f2073df2e1b7fdf73d123","project":"openstack/tempest","branch":"master","topic":"ipv6-doc-prefix","hashtags":[],"change_id":"I70565411851439551e0f2073df2e1b7fdf73d123","subject":"Revert \"Use the IPv6 documentation prefix as the default cidr\"","status":"ABANDONED","created":"2019-11-21 10:30:49.000000000","updated":"2019-11-22 01:06:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c842db179c6c577374fc74caaf17cba877987f74","_number":695389,"virtual_id_number":695389,"owner":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2019-11-21 10:35:11.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"date":"2019-11-22 01:06:04.000000000","_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-21 13:19:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2019-11-21 10:40:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"date":"2019-11-21 10:30:49.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2019-11-21 23:46:58.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-21 10:30:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"date":"2019-11-21 10:30:49.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 10:30:49.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-11-21 10:30:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-11-21 10:30:49.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2019-11-21 10:35:11.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-21 10:40:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-11-21 13:19:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-21 23:46:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"28399550badf64f23e658b6372dfe64d1361ecae","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-11-21 10:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2a5c5609906d5de14400faa85ef81076747429b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-11-21 10:35:11.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Tempest-Sriov http://13.74.249.42/refs/changes/89/695389/1/Tempest-Sriov : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5378a8c29f1aeae7a44bbc44209ace4258ab2eff","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-21 10:40:13.000000000","message":"Patch Set 1: Code-Review-1\n\nWhy do you want to revert this? Seems a good fix to me.","accounts_in_message":[],"_revision_number":1},{"id":"f600293d7c304b4cbf395638ca75897d7be7c120","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-21 10:43:58.000000000","message":"Patch Set 1:\n\nI saw your comment on the original patch now. Seems that this would rather need to get fixed in neutron-vpnaas? Please create a bug report otherwise.","accounts_in_message":[],"_revision_number":1},{"id":"60c67cd1e2c2bca7090b397a6a799235ae1cb0aa","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-11-21 11:20:14.000000000","message":"Patch Set 1:\n\n\u003e I saw your comment on the original patch now. Seems that this would\n \u003e rather need to get fixed in neutron-vpnaas? Please create a bug\n \u003e report otherwise.\n\nThanks for review, I will see if this change makes neutron-vpnaas gate passed[1]. Then I can fix in neutron-vpnaas side.\n\n[1] https://review.opendev.org/#/c/693965/\nBug Link: https://bugs.launchpad.net/neutron/+bug/1852516","accounts_in_message":[],"_revision_number":1},{"id":"ba0fd6240473ea2a7b68bf3dbcab87ba7caf3b78","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-11-21 11:43:11.000000000","message":"Patch Set 1:\n\nneutron-vpnaas gate green now.[1]\nSo let\u0027s wait for Brian and other reviewers\u0027 reply.\n\n[1] https://review.opendev.org/#/c/693965/","accounts_in_message":[],"_revision_number":1},{"id":"ac9db26ba99e49e2293951b804f69485785ed5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 13:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9149d79188c244e499944bded7685b1a : SUCCESS in 1h 21m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b72deef6e62f460f98883ca735952412 : SUCCESS in 52m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0f225594034e46199a999768415aa539 : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7760faeb0e644cbbb8f7b465541ed64c : SUCCESS in 1h 12m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/016342f901ae44229bddb4961e28f61e : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ebee3274ea4815818f0a12fce5e766 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/739754f6cc2e444d9dc90713597adc0e : SUCCESS in 3m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/94dfdf637e664e74b6ba84a7f8188ebc : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3064957a275746c38f56726367f7cb1a : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd2ee60258024eae83ed287ed1050709 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293c05e74b9f4cf692d313afd2f53710 : SUCCESS in 10m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/6d07b8f2edd7429dae09abf9f3e1e0a0 : SUCCESS in 1h 57m 23s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c468a79862174696a929dd4711909061 : FAILURE in 1h 08m 59s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/2d5e94fdb50f442581051ba2cd3c5422 : SUCCESS in 1h 37m 53s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3ba3e6b656704ab6adb582aa47799662 : SUCCESS in 1h 16m 01s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/68345c22f78c473ca87af6e9c15eab51 : SUCCESS in 1h 18m 03s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9a249f2783f2459b96738efdadee6750 : SUCCESS in 1h 22m 20s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/17843550f9024dcbb9f21b410ae7a2be : SUCCESS in 1h 23m 52s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/46bf3a851e824da2b0caea5ccf90bb5a : SUCCESS in 1h 13m 45s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/8e6a9611528e41c3ab07f2031dc33a58 : SUCCESS in 1h 23m 23s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/58bde864402d4ad889cf806bce864326 : SUCCESS in 1h 28m 53s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9b02f444653d4ff386853a1c0ccd703e : SUCCESS in 1h 52m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cd04a36c45424eb4abeca4b23c91eb29 : SUCCESS in 1h 49m 23s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c23bbfe0f4254a168dc677445e62f632 : SUCCESS in 42m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7d1481b03c34e659f73c3ae580b80d9 : SUCCESS in 59m 22s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/297398cd3f424a1c978d165512a604b6 : SUCCESS in 2h 30m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4772e94e2df4722a401cc45e0e28121 : SUCCESS in 2h 29m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/44cd2e94b2aa4fde8f5a437ef2c6e7f4 : SUCCESS in 55m 37s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/524f5689b40740bdae552c9194cc58e8 : SUCCESS in 1h 19m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/46139046a7ba4af0b0d39b017723cf7f : SUCCESS in 1h 00m 36s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a5a249fd73c0481d919d580288714d3e : FAILURE in 31m 40s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/88f90c87d3384bf4a35d7d8e6252295b : SUCCESS in 1h 33m 55s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b0b6d13bbacf4282b74f7fa037eac232 : SUCCESS in 3m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b0e66b6eff4e4fafb6d28d6c3888a6f7 : SUCCESS in 1h 27m 56s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3f4e02cbc229415c8a5135ed6d161f8e : SUCCESS in 58m 32s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/77319b6998c34e1ab5b02201a5728e14 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"37bada8bb5846f1166c38da55a27ef290010ac84","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-21 15:11:36.000000000","message":"Patch Set 1:\n\nhttps://opendev.org/openstack/neutron-vpnaas/src/branch/master/neutron_vpnaas/tests/tempest/scenario/test_vpnaas.py#L151 is the location that implicitly assumes the tempest default. likely you need to override the tempest default in your test if you depend on a specific value","accounts_in_message":[],"_revision_number":1},{"id":"bf3347ebb739b57c9c37c6f23ca65731abd51be7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-21 23:46:58.000000000","message":"Patch Set 1:\n\nSorry I broke vpnaas.\n\nCan you change the test Jens mentioned to use this value instead?\n\nCONF.network.project_network_v6_cidr","accounts_in_message":[],"_revision_number":1},{"id":"79de1de266f76ec1a524f4ab2436eda1e3f2b521","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-11-22 01:06:04.000000000","message":"Patch Set 1:\n\nThanks Jens and Brian, I will have it changed in vpnaas side.","accounts_in_message":[],"_revision_number":1},{"id":"355755e2bd83949c6005996dafeda839ec1048e7","author":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"date":"2019-11-22 01:06:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee92d90a2ba94af24c610ed3ab59acc82f931b02","revisions":{"ee92d90a2ba94af24c610ed3ab59acc82f931b02":{"kind":"REWORK","_number":1,"created":"2019-11-21 10:30:49.000000000","uploader":{"_account_id":12860,"name":"Dongcan Ye","email":"yedongcan@yeah.net","username":"yedongcan"},"ref":"refs/changes/89/695389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/89/695389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/89/695389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/89/695389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/89/695389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/89/695389/1"}}},"commit":{"parents":[{"commit":"5689355cf19a8b2e07bfc6d4c26696af8b4cc6b1","subject":"Use the IPv6 documentation prefix as the default cidr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5689355cf19a8b2e07bfc6d4c26696af8b4cc6b1"}]}],"author":{"name":"Dongcan Ye","email":"yedongcan@yeah.net","date":"2019-11-21 10:30:49.000000000","tz":0},"committer":{"name":"Dongcan Ye","email":"yedongcan@yeah.net","date":"2019-11-21 10:30:49.000000000","tz":0},"subject":"Revert \"Use the IPv6 documentation prefix as the default cidr\"","message":"Revert \"Use the IPv6 documentation prefix as the default cidr\"\n\nThis reverts commit 5689355cf19a8b2e07bfc6d4c26696af8b4cc6b1.\n\nChange-Id: I70565411851439551e0f2073df2e1b7fdf73d123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ee92d90a2ba94af24c610ed3ab59acc82f931b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ee92d90a2ba94af24c610ed3ab59acc82f931b02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
