)]}'
{"id":"openstack%2Ftempest~696226","triplet_id":"openstack%2Ftempest~master~I4095cfeff213317bb20caec74c1887c968a52624","project":"openstack/tempest","branch":"master","topic":"master2","hashtags":[],"change_id":"I4095cfeff213317bb20caec74c1887c968a52624","subject":"modify status code according to the latest API documentation","status":"MERGED","created":"2019-11-27 03:14:56.000000000","updated":"2020-03-17 13:15:28.000000000","submitted":"2020-03-17 13:13:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696226-1584450804785-74ea99eb","meta_rev_id":"01066be7ced072fbe77561fd0e97de2b5989a69a","_number":696226,"virtual_id_number":696226,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-05 08:51:49.000000000","_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-17 13:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},{"value":1,"date":"2020-03-12 07:05:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-03-17 09:29:07.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2020-03-15 23:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-03-17 09:29:07.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-12 07:05:35.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-03-15 23:47:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-17 09:29:07.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-03-17 13:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ee7f150299d9dd2973912934288b8a18f798f47","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2019-11-27 03:14:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29a8bb7d5c769db9aa485d2561b8139fa34d2f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 05:45:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c9be64f9ce3e40d98e1c4e37f2790258 : SUCCESS in 1h 30m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6b6748ea668c461f874324de607759f1 : SUCCESS in 1h 00m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d82fb831a9ab46a388a729b632b1e246 : SUCCESS in 1h 02m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9f1104697b443d8b4bd64e797c91e1e : SUCCESS in 1h 21m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af8064eda500454cbca93c3d6a16c904 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201b2791b2b04e53a7dd916744a82805 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d702fea4ff843dab804c96c27460faf : SUCCESS in 3m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/683882961e5d41fe80d3ad7a727a638e : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968ace0967aa4eebb096b953c3f3a48d : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17e0726db6f54bae94c9b833434c4935 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/086f6edb49f54f368584e7374629d35c : SUCCESS in 8m 02s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/002f9fa556e2452e91abcb5d87f254c5 : SUCCESS in 1h 51m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/22fba67784b442329a0199e69bdd5cd2 : SUCCESS in 1h 19m 57s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/6dcdb58328b746caac2f06b68a46e959 : SUCCESS in 1h 30m 02s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/d6790f2ad3234ac5b45bf81246e0560e : SUCCESS in 1h 22m 30s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/a0d7b5d32cf24be08fd3375106a429aa : SUCCESS in 1h 31m 23s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/1152bf9acfce4b35b9c7338b4eff970b : SUCCESS in 1h 14m 36s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/c625ec28a3a840a7b061235f8d4f4cea : SUCCESS in 1h 21m 48s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/15bfe6c648244aa6b542337cec57b2a5 : SUCCESS in 1h 36m 54s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/0ed4f4617ca34268acef698eb9b49a78 : SUCCESS in 1h 19m 45s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/394fbfbe35b24605a6b479697954860d : SUCCESS in 1h 15m 07s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2ed755cf729547cbac3822ee732e1c9a : SUCCESS in 1h 59m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/20d0468414b349f79f5a30efee82f78e : SUCCESS in 1h 36m 54s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/8ce37f8557b94c378e790eb7f00a35a2 : SUCCESS in 42m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2e74705dddc4774b04ae8a25bc74d0a : SUCCESS in 1h 24m 22s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/4920d441c5574b2e8ffd09c77975d131 : SUCCESS in 2h 25m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/454e218cdbf541a88cb61452fa74cc80 : SUCCESS in 2h 02m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4e5ef7dee7fc49fea6a1dc51bb1562ae : SUCCESS in 49m 42s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/02d0b414b61d4e28abbf5793241e5b05 : SUCCESS in 1h 06m 37s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3844f24d5f9a451194e62bb964f7360e : SUCCESS in 1h 24m 32s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d4f39ecf1df84ac5bcd9b4b633d7add5 : FAILURE in 23m 05s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/95efc2c2d62644ef93c319c2bcd62213 : SUCCESS in 1h 15m 04s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/be8f85d3854c4a31bf1372cf792be676 : SUCCESS in 2m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8e70ea943edc4e0d9f55028064fd9a4e : SUCCESS in 1h 09m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ac338064dc7843f9a96c87483c2534e5 : SUCCESS in 1h 05m 41s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/20a0e34aa2794c3b81aa8a4d6d7dc1d4 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6d24649de7c04bdd585ac2a21ab5a840ade03e4e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2019-12-03 15:33:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a5f383b8014a91b903cb393b6ff0234d87ef2169","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-26 18:16:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91088b08241d2c20cb7f6dea2110a6a1ae7bd8ba","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-02-27 06:28:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ecb8ee944aa8dcb188b1d210d32d87b628ae2b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:26:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8534e62237bd48c9a69ebb0d107dbcc8 : FAILURE in 57m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26f9ddfb4a5c4ef487e131f519902b8c : RETRY_LIMIT in 1h 22m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b657f4d29e471fbb1ccceec73bd5f6 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ef8f70e9dd4af39048dafbcc692e21 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9df12e0375dd420ba597953e98cc5b6c : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c70accaf18e4adf9ac9c12f090f82b0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d60afd7bb184114ade760d332dac29d : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5eee12b16924bdfbba401e6745962eb : SUCCESS in 19m 21s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/6e9edbbfc5f744108fea2838fcac72c4 : RETRY_LIMIT in 32m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/669345d2b02f404ea63744e1caecec06 : RETRY_LIMIT in 1h 20m 02s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f4fb288fc27045e3b4315f97b147792f : RETRY_LIMIT in 42m 47s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/eebf7cbc094348058d16c6a9cebfe381 : RETRY_LIMIT in 35m 44s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/0b82cf2ab719467dbcd1a63101383845 : RETRY_LIMIT in 41m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f0ed8850e330451f8b51495a5ecf0998 : RETRY_LIMIT in 35m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/bde9fb2aac7a407b9fd8ec178e3779bd : RETRY_LIMIT in 43m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/891843c7626b4851ae10472440dfd48b : SUCCESS in 1h 06m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4543d4fe4da45599bd5e17115713fe4 : RETRY_LIMIT in 1h 20m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/556e2e0449474f46bdd680ae462a5242 : RETRY_LIMIT in 1h 19m 14s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1355e8a1826e411fba8cdf1da59e3150 : RETRY_LIMIT in 35m 10s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e63c08105b44ead8d33d96e9f2d657b : FAILURE in 18m 43s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1cb2c3b7e9a64e719ef32c0fb514013b : FAILURE in 24m 14s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a058b869e5894b46aaec8d3d2607b6cc : RETRY_LIMIT in 35m 24s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0520df93ca204aca8ad7753d1da2fbd1 : SUCCESS in 5m 18s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2a9c52d168014a869d755e71b5742a4c : SUCCESS in 1h 28m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/42be967de9fe46b08101be8e18fbd5c2 : RETRY_LIMIT in 35m 08s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0f31ca4dbec94abdbfa95ce707984734 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b07b85272110879865a5609bcaca8806c6ce9ca4","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-02-28 08:19:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"984cb804a2ae28bcc90b1e1a88dcc6e11ed96a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:15:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d725b958d4244ceabcd4e694ebfa369b : SUCCESS in 1h 03m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/831b6b0c010140b089b4eccef375ec07 : FAILURE in 1h 20m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/094d61c272e54d7aa3c212968450ddd4 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c969d8872edd4281a18a04d061f80e8d : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/759fc62111b34d4aa9b7366816a7a6cc : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e17dec44e26f41919253785e32f5bcd8 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3df18a9992564f6f8030beb9b08962d8 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9549585d2a0649ed9409a456f499d415 : SUCCESS in 13m 28s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b7beb51d1a314a6ab39f06ca9e87ccb5 : TIMED_OUT in 2h 10m 11s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/dc81947617544b36aaa8fa63bdcfcce2 : SUCCESS in 1h 30m 53s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e4585501f7a542189a8b4cd1771aac53 : SUCCESS in 1h 26m 56s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/1948a01f67cb4718a3840bd91643b3bd : SUCCESS in 1h 24m 38s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/921259e5deab4ce09dc4cd6924726346 : SUCCESS in 1h 17m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b3ba9598f76c4841a15efa132d26d82e : SUCCESS in 1h 35m 50s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c7fb58826dff4cd4a4f8299cec0537fb : SUCCESS in 57m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fc06c25dba24a58b800077853637d07 : SUCCESS in 1h 03m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ded2500536f64c42904eb866c1ae4dc1 : SUCCESS in 2h 43m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3dc57aac4d98497693fbf93f0c6d8848 : SUCCESS in 53m 49s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9b29a742ee442cb89b046065698186b : SUCCESS in 1h 19m 17s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6aa97c0dd744e31a7028d779813d412 : FAILURE in 18m 47s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/60a7cf2c3ff54d6898d304f497b4c63d : FAILURE in 29m 39s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d21fbd227ae94baf853eae723e9c2e54 : SUCCESS in 1h 30m 17s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/de695686419c4198b7ea543eceaa9b33 : SUCCESS in 5m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/870fa3b764f941128dba156344051927 : SUCCESS in 1h 25m 57s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1fdb3a75f649495f8dd6ebdd846735f5 : SUCCESS in 1h 08m 08s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7f14c99515fb4fcda84b1f7a3d8feb32 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"3e8ea153c726a276e4af3fc88a78d6ac97378537","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-03-04 06:46:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a4395bc3171057c42b5755fecd274810f38922a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:13:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e89817f4e21a4de8b10b4f3fc54c8de1 : SUCCESS in 1h 08m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca6625f2d035401ba42cd232932b2553 : SUCCESS in 1h 21m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8f671df00ff4976af908b83f8c06a26 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4217668863884ce58f974a213da7e6bb : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/767f447c26434250a406ad3e417fcc63 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c972bc0a8ded4e6c8f33a19a324ecea0 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70a679bf76f848fd9a4da634af2cceb1 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8af00851324b75bf1058aa9964fb2d : SUCCESS in 9m 18s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f50a00a00ef7425c809396c941d45385 : SUCCESS in 1h 28m 18s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/5d996786fef84e849a228fbe248cb814 : SUCCESS in 1h 23m 08s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/23249305824e478a86ec023f0d23bf94 : SUCCESS in 1h 25m 48s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/956c52fac08a4ec4a18b5bb22e5f3deb : SUCCESS in 1h 16m 18s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/aa8edc3eca244d479ba644de6a9fafcf : SUCCESS in 1h 31m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f4a0156726d54933b2d9621bd8574e92 : SUCCESS in 2h 00m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6eb672b2aea44a76b4168f9edff54c2d : SUCCESS in 52m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23714a30c15a4d1c93717273e867e026 : SUCCESS in 58m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3abc53568f984058a69e6c8205f01371 : SUCCESS in 2h 16m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7c71148d09cf483cb0c97c47bba4d71d : SUCCESS in 48m 32s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b1b4c678308a48f4a8aa11d9b4247fd1 : SUCCESS in 1h 08m 13s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b09dfd7e23845ada580d37de87f07f7 : SUCCESS in 1h 14m 17s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ffba8642b9494be68a02e515b53c52a6 : FAILURE in 23m 15s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/87226e4f00c94d8ca4c2fd48c1fbb182 : FAILURE in 1h 00m 42s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b3f986462b05430995a14a5db85f32c9 : SUCCESS in 2m 59s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f9226a7a2d72452c874624b5e5be4b46 : SUCCESS in 1h 36m 45s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/10166c5e3b9d474f940c216643c3ad8d : SUCCESS in 59m 57s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/dd22d9f09b9349efa39570697a6e7da4 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0417daf8477bb9a4e2e8ca6f924dded0000066cf","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-03-04 09:40:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a7ce1343afae19fd37a6f2284f7bfe6dd0fe7da7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 13:14:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9ebf011a6b4a495f95e581cb6aaadd45 : SUCCESS in 1h 10m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf8b799beb6d42e181e5325a476006e1 : SUCCESS in 1h 28m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a93e0161f344caaab849e8d52f122bd : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30bfb1e6e1a8418ca1876c28f3adbf4b : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a285329ac1134cce852e2a1bd25ac7c2 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcc26f5079074d298eca06623563c591 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ec5af616a5a4f3a92bcc8ab6782ab44 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ead4a59e0314109b580d4efb36a8937 : SUCCESS in 13m 36s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/43ab923cf0e742bd956170ebb0b72bc4 : SUCCESS in 1h 53m 26s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ae2a839200ac40df84faa20121db7ec0 : SUCCESS in 1h 24m 54s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f83d4946e3c1473caf25724422d2f443 : SUCCESS in 1h 19m 56s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/6b9d50b0ca664e3b8e076c6a433f5fc9 : SUCCESS in 1h 33m 54s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/e09827a3a1274bf7b4b1e7404b2c6d1f : SUCCESS in 1h 17m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0478c40ad4a34acaaca62282062089f2 : SUCCESS in 1h 38m 40s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ddd31c59e3b54b5cb8fcae169d9dab64 : SUCCESS in 50m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40e2909ea47941238cd22795702f06f0 : SUCCESS in 1h 03m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/572ce77f538e4e4eb19f40a5df800a42 : SUCCESS in 2h 28m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/679d8d293bfc480990700c12bbc8292d : SUCCESS in 1h 04m 02s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/deb6d4d652d1449b81fcd19e1d289ff9 : SUCCESS in 1h 31m 07s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55ae7ae6686948cd8b293b4381ec1cb1 : SUCCESS in 1h 16m 02s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7e5afeec5e5b4527a9ac0d08ff5d063f : FAILURE in 28m 54s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0c76977d12d0498c9186d3beb81ec653 : FAILURE in 1h 06m 54s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c8dd1bd83f5a4376899bca550dda9308 : SUCCESS in 3m 47s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/216c3c358d9b436ba8d4309eb264b118 : SUCCESS in 1h 22m 23s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3db6920617d5498a9c8a8a4f50d29182 : SUCCESS in 1h 10m 11s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/58b009c1ac454a498a1927b337664dc8 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f966b3fc905dfcd651cee8c450f15513763178ed","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2020-03-05 08:51:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29e3af9c23908cd743af06f001bf238027f76d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 11:41:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/44ca2660c2e3437fa6fcaed4572fb72d : SUCCESS in 1h 31m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf040f50f6dc4bcf8e9095e0a8f27925 : SUCCESS in 1h 19m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e469e70310494a4aa5d9e57c4be6a31f : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1e34f5a394a4a8eb3527fa894ee3770 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b01e30acbf194ea2b80e2e1d79aecb2c : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cdd65e9a3734017a6dfe2c67bc5c32f : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fb55b6d7e954484bc1c16bd82e1236d : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9455c01a5f1d4b99916634067b12e8d5 : SUCCESS in 10m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/62170d7b993349ceae5036d853a720b8 : TIMED_OUT in 2h 07m 20s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/2e38c345d2ed4d4fbce24780b9335603 : SUCCESS in 1h 34m 26s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/fd9657830e1842aaa15accb18223beca : SUCCESS in 1h 28m 37s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/8d4640868f2d4189b27012856428af01 : SUCCESS in 1h 08m 41s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/dd3ad5e0176e45dcb37419fa2754fb41 : SUCCESS in 1h 37m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/33c3daab932e400899966c2576173f5d : SUCCESS in 2h 03m 54s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/595b6ff1a83e4e26a46f4d3b6512f083 : SUCCESS in 37m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b33de95ee29c42a48025ee1987b4c2f8 : SUCCESS in 1h 11m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad172ca0296845f1b3f6c0b5a49fed42 : SUCCESS in 2h 42m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1333f7a7c872414691ef5a9aa140d57f : SUCCESS in 1h 06m 00s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/267dc541fe7f45aa8b3402b91436ab0c : SUCCESS in 1h 14m 01s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d18d8e0ac8245fca84945d36aa77340 : SUCCESS in 1h 09m 03s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/94dd24af1f534e1199d26ce7a4ed2e8a : FAILURE in 25m 41s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/70c92808c8444700a3a1a0b54a7b1613 : SUCCESS in 1h 20m 42s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a46f4a3458824c4c8f7a72288dc334df : SUCCESS in 3m 25s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f0279f888b75416b9b86315032d7bf67 : SUCCESS in 1h 16m 31s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e283eae1f51546449d4cd61e21ec5204 : SUCCESS in 1h 09m 14s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/968ece7243844993a42f0104c492d750 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"f9a8dd77871a3a7b2cfacd1e02cf06c1e56e385b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-03-12 07:05:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a2d05897088b5638f3a9d0c166551efe06ac0a86","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-15 23:47:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c85a31ac7cfdbbf3cd6aa20841b432ef104b9817","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2020-03-17 09:29:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks like old enough thing ..","accounts_in_message":[],"_revision_number":2},{"id":"cc0d9ed85e079ee6c9567cbc4cb7b0df5c946b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 09:29:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76b45b3f065888cbe1a299127031eefe07d21f8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 13:13:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4a42f0e27e1545e5a79e7e6521555948 : SUCCESS in 1h 05m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59e2292728704b7ba6aaebdd42592256 : SUCCESS in 1h 25m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413ef849b24949679fc2151fe8a4a808 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef75435f18014ba3abf9591e944b7a61 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/025c2f41796d481486fb89169df3e17b : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c82d13b795b482c864dfbc59ae59838 : SUCCESS in 10m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/69e7aa8b74c843f38d7849a243b1fad6 : SUCCESS in 2h 36m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e698a2fca2f246ef98798cf8c46a24a7 : SUCCESS in 1h 18m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b5257597dbd4d9095cdc4fa8b772ae0 : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b40f7f7e2db8b2088512dee5b5842c582527a034","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 13:13:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"20c94d98e283d1aca804797ebd0c04496c79030c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 13:15:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db939dfaa85b4390877e74e788a8794d : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7eb1d2dc9992ae1efe8e1303d4fa0058d1a0538f","revisions":{"8efb6c78039e1b1c7378db30f0e37336a1eb2872":{"kind":"REWORK","_number":1,"created":"2019-11-27 03:14:56.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/26/696226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/26/696226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/26/696226/1"}}},"commit":{"parents":[{"commit":"9ef43160912f5ed7a71640b370d0dde2d84f6249","subject":"Merge \"Remove unused opensuse150 job definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ef43160912f5ed7a71640b370d0dde2d84f6249"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2019-11-27 03:11:36.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2019-11-27 03:12:14.000000000","tz":480},"subject":"modify status code according to the latest API documentation","message":"modify status code according to the latest API documentation\n\nmodify delete all tag definitions interface status code according\nto the latest API document\n\nChange-Id: I4095cfeff213317bb20caec74c1887c968a52624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8efb6c78039e1b1c7378db30f0e37336a1eb2872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8efb6c78039e1b1c7378db30f0e37336a1eb2872"}]},"branch":"refs/heads/master"},"7eb1d2dc9992ae1efe8e1303d4fa0058d1a0538f":{"kind":"REWORK","_number":2,"created":"2020-02-27 06:28:51.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/26/696226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/26/696226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/26/696226/2"}}},"commit":{"parents":[{"commit":"9ef43160912f5ed7a71640b370d0dde2d84f6249","subject":"Merge \"Remove unused opensuse150 job definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ef43160912f5ed7a71640b370d0dde2d84f6249"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2019-11-27 03:11:36.000000000","tz":480},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-02-27 06:28:32.000000000","tz":0},"subject":"modify status code according to the latest API documentation","message":"modify status code according to the latest API documentation\n\nmodify delete all tag definitions interface status code according\nto the latest API document\n\nChange-Id: I4095cfeff213317bb20caec74c1887c968a52624\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7eb1d2dc9992ae1efe8e1303d4fa0058d1a0538f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7eb1d2dc9992ae1efe8e1303d4fa0058d1a0538f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
