)]}'
{"id":"openstack%2Ftempest~696826","triplet_id":"openstack%2Ftempest~master~I834a3150c096397b66caf2a4078e1260d22e0fbf","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I834a3150c096397b66caf2a4078e1260d22e0fbf","subject":"Modify the miswritten comment of API reference with volumes_client in v3","status":"ABANDONED","created":"2019-12-02 07:51:08.000000000","updated":"2019-12-03 02:37:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"feab0b1172fbcfb236852657f8a861812e467ac2","_number":696826,"virtual_id_number":696826,"owner":{"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-03 00:56:14.000000000","_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-02 10:30:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-03 02:06:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 10:30:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 02:06:13.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"}],"messages":[{"id":"d32752e8d8d1155f7ef2c004ad5277af31d71d60","author":{"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},"date":"2019-12-02 07:51:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97d064aaa3fe97ff56374071ba9f9e553615d507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 10:30:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1b31b5e85f4241c8a77e0015f5817587 : SUCCESS in 1h 22m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c19495fcae44400db17c9e8859f4764d : SUCCESS in 1h 05m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/35bbac7b6bbc4755a8a905529b80e5cb : SUCCESS in 1h 14m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e1c007f002a489d99b93c941ea612e4 : SUCCESS in 1h 23m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9413eb17f74c4de19c9760e53d397fe0 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80de245b518c484188733b903f276697 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3d68d0fd6344020bfe9d99976faf319 : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/976de1f4de1246f8ba2630d1c7737939 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b32acb0ae0044810bbc0237e8c6ba62a : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50d31c906fcc4f85a2f8a610d3fce874 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a8f154e2fc486390333bac7311f6c0 : SUCCESS in 9m 24s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/fe7e2a443eae4cb0ab5edac3dcde58c6 : SUCCESS in 2h 00m 35s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/15b93fbd9d4447d5a30cb2b5855c7f75 : SUCCESS in 1h 23m 45s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/2ff4de8ee392434a83a220baae64c56a : SUCCESS in 1h 31m 04s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6922b62f613d4280a8987de515ff472e : FAILURE in 1h 36m 38s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/c04deef365384b0bad78168bf638494d : SUCCESS in 1h 19m 40s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/631645a7348846bfac795f6e70728f72 : SUCCESS in 1h 57m 30s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/aa90ca4d498543098d957ef78497ecb2 : SUCCESS in 1h 25m 10s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/661654f7484745fdb99e1f790a81ee73 : SUCCESS in 1h 22m 12s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/32cb0a6cdf764d3eacbda53b3dcf9095 : SUCCESS in 1h 14m 29s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/c9aa2ec355de4c59b5eb0c2245d20084 : SUCCESS in 1h 17m 58s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bceab5ac7abd41158ce8ace2b4917390 : SUCCESS in 1h 44m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f118f1a36a774ebab341239fe4f386d8 : SUCCESS in 1h 40m 25s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d65d054c9bf44052ab716ed17f14a5fe : SUCCESS in 50m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1da3b102a7d34fe984fbe74750c232d7 : SUCCESS in 59m 26s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/a72072105e174edb916a29c70c733083 : SUCCESS in 2h 22m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5c13cfee56d482985687023b1064d1e : SUCCESS in 2h 18m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/45b85a33cab9474c814d8da6528df2a5 : SUCCESS in 52m 49s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/332fcfa2e0ed4972a8cdd25deb3e25e8 : SUCCESS in 1h 04m 49s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8cab4975fada420591ef7066f66c638d : SUCCESS in 1h 02m 15s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0f1b0bc1c1674de39e850df4e8cb1913 : FAILURE in 24m 55s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/63509181d0ee4985b28f64d119e082cb : SUCCESS in 1h 10m 50s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5985a50084d64310976bc2bc28aa25e4 : SUCCESS in 2m 19s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5062fbed38a040b5861fd14594e7ad02 : TIMED_OUT in 2h 06m 10s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2e5585f4a630423581dd307f41941a2c : SUCCESS in 1h 04m 12s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/247df9f5be244fe6af6372fc77880080 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9d3a752f8e2239e26da70f3ed58a6cb90daf102e","author":{"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},"date":"2019-12-03 00:56:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8681d3da32ca3071239e6de5a6413bfcf5cf9723","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-12-03 02:06:13.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for noticing this, and https://review.opendev.org/#/c/694893/ does the same thing:)","accounts_in_message":[],"_revision_number":1},{"id":"feab0b1172fbcfb236852657f8a861812e467ac2","author":{"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},"date":"2019-12-03 02:37:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f3e72b066cced15b21b00395cd6886092c83373","revisions":{"2f3e72b066cced15b21b00395cd6886092c83373":{"kind":"REWORK","_number":1,"created":"2019-12-02 07:51:08.000000000","uploader":{"_account_id":30410,"name":"linanbj","email":"linanbj@inspur.com","username":"linanbj"},"ref":"refs/changes/26/696826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/26/696826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/696826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/26/696826/1"}}},"commit":{"parents":[{"commit":"5d9229b5216037586d4e8f6920d6e21bf11f41ae","subject":"Merge \"Use python 3 on all nodes in multinode jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5d9229b5216037586d4e8f6920d6e21bf11f41ae"}]}],"author":{"name":"linanbj","email":"linanbj@inspur.com","date":"2019-12-02 07:41:49.000000000","tz":480},"committer":{"name":"linanbj","email":"linanbj@inspur.com","date":"2019-12-02 07:42:17.000000000","tz":480},"subject":"Modify the miswritten comment of API reference with volumes_client in v3","message":"Modify the miswritten comment of API reference with volumes_client in v3\n\nModify the miswritten comment of API reference with volumes_client in v3\n\nChange-Id: I834a3150c096397b66caf2a4078e1260d22e0fbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2f3e72b066cced15b21b00395cd6886092c83373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2f3e72b066cced15b21b00395cd6886092c83373"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
