)]}'
{"id":"openstack%2Ftempest~702493","triplet_id":"openstack%2Ftempest~master~Ib5ebbc6d5538a6179eca27c203faee6805af0093","project":"openstack/tempest","branch":"master","topic":"702493","hashtags":[],"change_id":"Ib5ebbc6d5538a6179eca27c203faee6805af0093","subject":"Refactor methods to be used by other tests","status":"MERGED","created":"2020-01-14 17:47:11.000000000","updated":"2020-02-10 13:03:03.000000000","submitted":"2020-02-10 13:00:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702493-1581339633346-ff1353d7","meta_rev_id":"1d31d9ae4b25535d533c752431a81c4715d16610","_number":702493,"virtual_id_number":702493,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-02-10 07:50:43.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"date":"2020-02-10 07:34:29.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 13:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"date":"2020-02-06 05:32:48.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-23 07:08:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":2,"date":"2020-01-28 06:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-01-27 13:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-21 11:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":1,"date":"2020-01-28 06:44:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 11:33:31.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-01-23 07:08:01.000000000","updated_by":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"REVIEWER"},{"updated":"2020-01-27 13:52:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-28 06:44:07.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-02-10 07:34:29.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-02-10 07:50:43.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-10 13:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3cfe715b1be9473f12f24cda7ca0edb0b9ac0332","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-14 17:47:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06faa6d7f5cd590d6035e73a3543b01e84842260","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-14 18:30:25.000000000","message":"Patch Set 1:\n\nwhere you are using those ?","accounts_in_message":[],"_revision_number":1},{"id":"5be7f6abd65bf9393d5d6e1e66d0a9fa5f3f0038","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-14 19:39:10.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/1/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"48c028d95a90745d9ef0bfa231f55b5be9813da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 04:16:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a001083d5fb94d2cbd33049c5fd05f1e : SUCCESS in 1h 48m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/56506964339f4be897388467aeff79d9 : SUCCESS in 1h 00m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4afd98eae55542739dd1578ee7cb7767 : SUCCESS in 1h 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d689b6d00484e67a92027bbbeccad67 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9553b479810e4d7b9ad897c79b756357 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9dc1a94f4af4a65b43024a03ea5783a : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3fc33b341768423186ddc500986ff6ba : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52bb811b70de4abbad99e943edaf88ed : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb4728c3403e4599a306c7472f415f96 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e27be7f295145078e4581f69d717203 : SUCCESS in 6m 58s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eaa5462a26034d90b6ef3a5981f99ac2 : SUCCESS in 2h 01m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/2afdadc0887746a28843851bcc511412 : SUCCESS in 1h 19m 12s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/0bbd2f8747114a619a4823a8f18b2bd7 : SUCCESS in 1h 39m 08s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a29aee80ba5348649c6da56d8850fc64 : SUCCESS in 1h 34m 56s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/0a183b67ab3041c182fe6250256fd106 : SUCCESS in 1h 37m 18s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/7945bd4db1c54b40857b7f58572daca8 : SUCCESS in 1h 30m 12s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/be78d5de933743309659c8e0898f1412 : SUCCESS in 1h 36m 35s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/80b5eed2ac404c3cac04ffcdb3932102 : SUCCESS in 1h 26m 26s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/07b6d19152d04433ad9e7ef829dd8e99 : SUCCESS in 1h 35m 10s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/1057668b048a4e8da428ff2f04420809 : SUCCESS in 1h 48m 00s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c2858b2294cc4333ac385458e4831694 : FAILURE in 1h 44m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e94c2a19f04f47eca2bfdb0b451db56e : SUCCESS in 1h 59m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/37e67064df724c6c9ab97e7510db1125 : SUCCESS in 56m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2871bb32395e4f83909fb3ab910d9b59 : SUCCESS in 59m 51s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/9c39a58c1a6c41f3b238af3a69e0e7c9 : SUCCESS in 2h 31m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3af21d3dcc9c4f34a7b862c57d64f799 : SUCCESS in 2h 27m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8db300e2243240ca9832115b2ca823ce : SUCCESS in 58m 02s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1d394378e8646fa9a5b83e830290220 : FAILURE in 1h 14m 38s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4889b611419645aea98d15817ace6e70 : SUCCESS in 1h 07m 45s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/07d7a14e0f1749929a93d2ff702b56e1 : FAILURE in 27m 16s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3937d7c1950c4bfab73a807f01a3bcc0 : SUCCESS in 1h 19m 53s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7ccdf751774040adbc8d3d40d2daac61 : SUCCESS in 3m 14s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8bfed12c3aec4f0e8f2cbc508833e29e : SUCCESS in 1h 26m 12s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a2eacbc18f0745e68ddbeae2ec0dc085 : SUCCESS in 57m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/43b07677692a4782b1fc24dc331cd636 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"19767b369cfcc1ca0e835f3e18bb0d8b8941ebe9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-15 09:39:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b88912d1413ac6f55082f13c5ee96cac4c68a8ac","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-15 09:41:43.000000000","message":"Patch Set 1:\n\n\u003e where you are using those ?\n\ncurrently here[1] but i think \u0027_boot_instance_from_resource\u0027 and \u0027_create_volume_from_image\u0027 will be useful in general for future scenario tests.\n\n[1] https://review.opendev.org/#/c/702495/","accounts_in_message":[],"_revision_number":1},{"id":"5f742c5eb0515edd61c37c6e4378a4e0447eb047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 12:15:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8305fee954204468948c7a787ce0ac4d : SUCCESS in 1h 54m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/91d1df8007664257b0eddbd6bc96320f : FAILURE in 1h 05m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ff19e573e9e48e8895b0ba041694827 : SUCCESS in 1h 47m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55fce0cf4bab40b581a8a27a900c76d2 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3202de2ca1ce40ab9b2d50a4cbd5d58d : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff88bb547ceb496aab30b84ddef31ab9 : SUCCESS in 6m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/412afd464aec40988168dd9c874b3623 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f425b4813d249e8bf43b3585b201418 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fb2523af69748fc9497d94dbf227278 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f39858f93af4c78b62f91a5dbb743a2 : SUCCESS in 10m 58s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/be137c785dc2479cb4c90bd25fcc6f9d : TIMED_OUT in 2h 09m 08s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3bf6633cefcf45baaab4ff22a2fa5246 : SUCCESS in 1h 44m 37s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/767383fceac94086b390f6474e991935 : TIMED_OUT in 2h 09m 29s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a19d8af32e714cf88d50e0d1c61f9361 : SUCCESS in 1h 21m 46s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/f964bdd9d04c4650b4c09e921a6c07c7 : SUCCESS in 1h 48m 26s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/0da5ca5f22dc4ef9b725c13be25989a9 : SUCCESS in 1h 29m 09s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/c18e905e7f6f4fdb9ab2969ddad2f728 : SUCCESS in 1h 43m 17s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/3597d0dfa31e4ac88c97667f6fc645d6 : SUCCESS in 1h 32m 57s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/70d0ce0340304a59975525bae5fc9029 : SUCCESS in 1h 28m 24s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/be3ce2c279de4fbcbc48061988961131 : SUCCESS in 1h 21m 25s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9620268b926f468b8a5f888710c5af16 : SUCCESS in 1h 49m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1ecf8df07a7f443f83683a8e534ca81e : SUCCESS in 1h 53m 13s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1536426b87af498d8292cf15825f8856 : SUCCESS in 1h 04m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bd49bfa93b04437b3f2b4526c3d381b : SUCCESS in 1h 18m 26s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/6c2fc816753a47a8be6c4619d06d5c59 : SUCCESS in 2h 30m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71662033fbf942bba7c541b8f7556637 : SUCCESS in 2h 16m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/474e24df3ffd4f03a55f273bf52cfbc4 : SUCCESS in 58m 41s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d42846e6f8c34271aa0dadc3d4ed8d75 : SUCCESS in 1h 14m 15s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7faee8feacf94a2187281c67ad53bdc1 : SUCCESS in 1h 05m 21s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9c5a2594d3a34a3da10b473410027669 : FAILURE in 26m 20s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3b5becaf1d88494a903912ca53697bd6 : SUCCESS in 1h 20m 08s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f8ba6af946b446dda0db02768b32dd98 : SUCCESS in 4m 01s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/77a00b4680104ffda36c28243f175945 : SUCCESS in 1h 23m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/341e7be017fe4bb9b308de73735ac475 : SUCCESS in 59m 49s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/298faa134138494e8a2d233fc42b5b89 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1d745deb8ad7383e451f92fb56f864778372139b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-15 23:37:15.000000000","message":"Patch Set 1:\n\n\u003e \u003e where you are using those ?\n \u003e \n \u003e currently here[1] but i think \u0027_boot_instance_from_resource\u0027 and\n \u003e \u0027_create_volume_from_image\u0027 will be useful in general for future\n \u003e scenario tests.\n \u003e \n \u003e [1] https://review.opendev.org/#/c/702495/\n\nI was suspecting that. actually scenario manager is not stable interface yet so we do not recommend to use it outside of Tempest. If using this unstable version then there is chance of breaking the plugin. It happened in past and we updated all plugin to maintain their own copy of scenario manager. \n\nOnce we modify this and make it stable interface then plugin can replace it with Tempest version.","accounts_in_message":[],"_revision_number":1},{"id":"e30c67623ff1428a7a2573e93aec00b69e4ff5ab","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-15 23:39:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nthough this change lgtm to define more common method in scenario manager and make it available to plugin once stable.\n\nOnly comment to make it public method.","accounts_in_message":[],"_revision_number":1},{"id":"f65a299f8cf14cb5586fc94715ed3288491c3955","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-16 12:08:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37090951d4a2307bf898724afa550187546e3bd4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-16 12:08:54.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nThanks for the review gmann.","accounts_in_message":[],"_revision_number":2},{"id":"3052a127d3ea7c9e513be8dc36b620c194e5e45a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-16 12:59:18.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/2/SRIOV-tempest-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"65d787e9004625849a8e84258fe8e602c69e70b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 15:14:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/72501a802ea5404f8f8b20bddf4872e3 : SUCCESS in 1h 34m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77a4736b1ccb4e9db5b3cb03eec2d786 : SUCCESS in 55m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d084721c67348d8972c159720fec00c : SUCCESS in 1h 16m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5e7cbb0b45f4d5ba7edde66e2420029 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d49082c0c2a4f028adfe195431f022d : FAILURE in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93e3dfb17dbd40cd990a52684ef9ec21 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/68b7455ba6b641e38a8b707584884d46 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da45db60e6a34ad985eb5e1ee8dca1ad : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5dc2f1ddad44e9ab15ecdffb664c71f : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6af3e1031a46f883800ae2c72e80a6 : SUCCESS in 13m 17s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c1076f956b1a490f907403b44a33acef : SUCCESS in 2h 01m 55s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/820024cf3d794574a8ca59e60972041c : SUCCESS in 1h 33m 44s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/7015eeb0421d4498b2879021c0e46155 : SUCCESS in 1h 52m 43s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/14ea739f4dca4a489ec4ac9f22efb727 : SUCCESS in 1h 12m 21s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/d7eac9b5b9fa46829ba037f09bb3c406 : SUCCESS in 1h 41m 10s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/d11b7c450c6f4a199e5135bf1a8dce32 : SUCCESS in 1h 28m 27s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/29afc9c889b74449a5892998ff8160bd : SUCCESS in 1h 33m 20s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/50bdb31917854816b83d3e68bbdbcd11 : SUCCESS in 1h 10m 25s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/ad9d9bdc1afc467c9337ae41be490d9a : SUCCESS in 1h 19m 54s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/1ac2887d3100412cb52c6d0a28cc92c2 : SUCCESS in 1h 19m 10s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ed77b8edc7fe40d6a578a299cf434513 : SUCCESS in 1h 43m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/44f035d9b9d14afaa7ca4188db6bbdc6 : SUCCESS in 1h 41m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c754005e527a414c9ef9b9a529763cd9 : SUCCESS in 46m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99f96d0d4725446f824677b40d0d8e20 : SUCCESS in 1h 19m 33s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/cd8c5cbfb97848b78a8c15c55d056da5 : SUCCESS in 2h 59m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8267fd384c894cd18ff8eb03b69196b3 : SUCCESS in 2h 20m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c9479b620bfd4226a830056735e131a7 : FAILURE in 37m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1df7f931aaaf46b5bba167a76206b977 : SUCCESS in 1h 28m 07s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e44fcac587f45d6b844bd5ac3de3e39 : SUCCESS in 1h 15m 11s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8e89a14086274defae684c74fa579e37 : FAILURE in 26m 31s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4e6bf5c4d57046258c76ab828493e395 : SUCCESS in 1h 10m 43s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2775015ac05c43cca007930d53d53969 : SUCCESS in 5m 48s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/995c4512699441a1bc09e7e7ef5bb497 : SUCCESS in 1h 35m 53s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/047d774b30a94fa98de471c3c415e315 : SUCCESS in 56m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e2194d74138b46b1949f5b1bb5bff16c : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d01bb609409d2ccb93012fa1b7514db0caf8e54a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-16 15:37:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03cc315fc20def9397dcf20770d071d37087cbf7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-17 16:54:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"423120e8f2a6942f299309f4aae5c4fb4586a9b7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-17 16:55:52.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"2ffad8f0f59776d076be8d4353518200104f85b5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-17 18:07:17.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"4d9fd9fef6845ecf1aafa519da8a94a4dfce885a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 00:40:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/52ef829db2424425a95d0d637b680df1 : SUCCESS in 2h 01m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/efd9adf6c8fb416794f9ede63297d5c4 : SUCCESS in 51m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a82b990e30d41309b9b497863038992 : SUCCESS in 1h 39m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8875c38c8e314a18b46d977936b0a64f : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74732dfc3260476dafad73488db7b3a8 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9a2caa8de794c0cbf97f103a8633411 : SUCCESS in 7m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0e9e38ba9b14ca09d7e6c4ba81726ef : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9b4b3788c5a45b39b8453372527a944 : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/050837b47b2b4df9b2345ac3b54ca150 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8431c488dd3540bdb822e87ce0e2d1a2 : SUCCESS in 9m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/555ba87a38364b728c8fd880160f21f3 : TIMED_OUT in 2h 06m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/59c43fed871e4722baab19affaf386cb : SUCCESS in 1h 29m 19s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/521df51f31554da498919bf0d1d7bf4d : SUCCESS in 1h 40m 21s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/39716327cbc84d57b8e9683ad658df91 : SUCCESS in 1h 22m 54s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/6ad689e48d2a4e09b24d7a165367a15a : SUCCESS in 1h 39m 06s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e9166c1448834a0baf5500c7d0432dc6 : SUCCESS in 1h 28m 37s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/9a226dd7b608483ca645c5d33ae5a822 : SUCCESS in 1h 44m 04s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/e6410d449d93478bb0786ad33e36517b : SUCCESS in 1h 37m 25s\n- tempest-full-queens https://zuul.opendev.org/t/openstack/build/83f383646adb454f8ff848a3a614d668 : SUCCESS in 1h 26m 28s\n- tempest-full-queens-py3 https://zuul.opendev.org/t/openstack/build/881d5a41654d449cbea47a95e720d7c1 : SUCCESS in 1h 33m 42s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/85a269120cfa42a5a4972fa0b9b3edbb : SUCCESS in 1h 39m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c348a5e69409432bbc61752243670764 : SUCCESS in 1h 36m 03s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f123c2b2c01c4b4db5a443a87e2c2896 : SUCCESS in 1h 07m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e249170ca81d4e25bbc7555870c5145d : SUCCESS in 1h 13m 41s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/51b97b22a04f42c39aa575f499eed3c9 : SUCCESS in 2h 29m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0694b3c66c2b460184802c0af4a2f621 : SUCCESS in 2h 33m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d69c63327f734bcc8257c42f50eb734e : FAILURE in 1h 05m 10s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6211b8db921f418fafe0400476868e48 : SUCCESS in 1h 34m 57s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9747e75e92d64bb49700d290eff808e9 : SUCCESS in 1h 18m 32s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/44404d29c923462cb91be644ee78d8ad : FAILURE in 27m 02s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5ab9541ed0cb43f7989ac62d5c3cd3d7 : SUCCESS in 1h 42m 52s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/125cdfcdf6fc49e7b25420b59b4138df : SUCCESS in 5m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8723fba532d1410c94a2186dcfdd2d86 : SUCCESS in 1h 18m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c560cc83375f4a2f9dbf011b690844f5 : SUCCESS in 1h 01m 02s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b9124c641c52400791949c590ae9ca04 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"8b708e5b65dbb6c0867e4049a9726816cd7fbed8","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-01-21 11:33:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5c8b23381aa9e3dd4ac54cb72583431e943c6cea","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2020-01-23 07:08:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a1c4570cb1c73a91ab8147a0d1f90d34eff9c50f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-27 13:52:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"447869eefc7735a3a52150a58fb5774538d2beef","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-01-28 06:44:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"c558606c7f9df865c0a5ccc82b147484f5184806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 06:44:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c83f06e393ecd5c9fa2a43c71c5da8a0a1e0590a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 09:09:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/dca6a7421c0642f4bf55c49bbeebed2a : SUCCESS in 1h 46m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b957fd6ea830414284e21718b273d977 : SUCCESS in 53m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e218c86a883b4cc2bb1085f0a8d00104 : SUCCESS in 1h 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db6929912ec484c955d6376bd46f227 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f6823d5b59a4b2495aa4d343b3c8740 : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90ef5f17462c49398b9523e9d5c8432d : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e2b6fefc7554e9887419338aab36451 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2b79c220c44486788e7c1c063dd16f8 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b6da72938f44e798b9fe249fde84a76 : SUCCESS in 8m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/83d57378a502495ea31db2beb8b6c108 : SUCCESS in 2h 18m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e25bad2282ca4d35909782d3ac82bd65 : FAILURE in 49m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/516b3e289d5a48f0abefa0803203b327 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9ecc4b9862e62a716224a092068ca3c63ef2b6fa","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-01-28 12:40:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"43e28574a7a8c6e9e91e7c31820ff081ab462561","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-28 14:31:50.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"77064b16a5ac968b58adf9f316906c7301ae6909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:51:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/174d732fc86d42b8a45e8b72a246ec21 : SUCCESS in 1h 39m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f42e235d5f9444d49bb05e415735b4db : SUCCESS in 1h 15m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a48d77ad3e1448cb0db4be80721b07e : SUCCESS in 1h 32m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82c4cb24698246cc99d7bad22e0a6576 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2cf9a5e7e84b16b73af0d6fcd1fc79 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/deda28bba53f43778846c351934db1bf : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2dff33f69736404ab33a1de9832e59b4 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/220ecc13ac3341938ec5fc55d1ab65f9 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf0614ff881c493a89a75c21f8f60637 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439eb80586384d568e4ab67e08617283 : SUCCESS in 15m 13s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0cc8d8e59a08418f86dea0367a793f70 : SUCCESS in 1h 57m 05s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/09aebb49e3e649b8966e9a7e3cbcfc0c : SUCCESS in 1h 22m 10s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/b9e40dd79d43456aab21474346191ff2 : SUCCESS in 1h 27m 53s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c060ae053d744972ab49000dcb1f395e : SUCCESS in 1h 46m 26s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/ca724f92a4de48c0977036046913b632 : SUCCESS in 1h 25m 53s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/ed73ad22c2f84131ae555cc671b6dfb1 : SUCCESS in 1h 19m 53s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/b630a25c0ba44707a655bf2555e6f7e8 : SUCCESS in 1h 25m 13s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/6653b1cf55be41b5bff47d124a0c19f6 : SUCCESS in 1h 22m 08s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b4548b967376426aa43fc2189f60016b : SUCCESS in 1h 32m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/62a91920c08b427c8620a1ff8a3dcd21 : SUCCESS in 1h 58m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6d8a0e1e18d7457391788972b6c7018b : SUCCESS in 1h 02m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e7039a295b54ff5bd4983825590bbcc : SUCCESS in 1h 01m 46s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/a6dd409e66c744a8a2b1a9ea62ad6e41 : SUCCESS in 3h 04m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f1e7702963054a01b60ca95f5ae7d9a2 : SUCCESS in 2h 24m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c67714955aa34f729480f2b08b1d8446 : SUCCESS in 53m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/14899765d76a4334873eaf93595cdcc6 : SUCCESS in 1h 12m 37s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/959722c41d71497384d4d936ed3b9f17 : SUCCESS in 1h 21m 59s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/22b3d2b76c81431daf9b0a551a00bbf5 : FAILURE in 28m 44s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/890ff5690f8642cf882c1f2f1930ab96 : SUCCESS in 1h 26m 52s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f9f4326a08b54608a806a44781121c23 : SUCCESS in 5m 44s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/02f53ad9e24a45d4add45d0ae6720f10 : SUCCESS in 1h 19m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a60b1ff0dda64f0d8e9308575c5a1212 : SUCCESS in 57m 56s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/22e3ff5966db4381a9dac4a6f20e02df : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"c8b376036df9dc0bc42f1a7331b66f650a3d18df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:52:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5459c6faaf8453c62bc3dedc4107f2811e7b76dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:23:04.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1437bea7adf243efbcd7de202cb3fe88 : TIMED_OUT in 2h 07m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/755674253760435b819fdddf81d8e536 : SUCCESS in 1h 02m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0efa7a55f6ea4774bc4993c744d885f5 : SUCCESS in 1h 34m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b6d6564d6b47ee97423a69d4d0d6af : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46084a8449054568abde42ae2b10d801 : SUCCESS in 8m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/981a1505ce494b3fbae7fdc54be43c2b : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48b62d2c280a478fb6c66b905e353468 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cef8b10379e241faab6adf8e7ac420c4 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698b317ec9eb451eb8d55507d3a14785 : SUCCESS in 9m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4db0d3d08b3c45679a7bcc3ba22c22bb : SUCCESS in 2h 23m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2573596a631045cab22cda216d6c47fe : FAILURE in 1h 40m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2928a9f246bd46cfb5991966f8a10cc8 : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e497a3bd6c3840db8f272f4edb468b8583e07b98","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-28 18:24:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"46f703acf73bc3ed4f0359d3da03e112b3c37b1d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-28 20:26:47.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"31280203b2caaa40e2e676a2778c7056e09d3e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 23:36:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/beed8a27acf24222b3525249160b6951 : SUCCESS in 1h 30m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d26f9acbaa7e49c8a2290a2e7729193d : SUCCESS in 1h 03m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9c5c3aea46d48c28e95eb66f130cca2 : SUCCESS in 1h 17m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4251da3c4ce4430daea10e3a2a30b8c8 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa946f05ba7412aa481fea7da6e3e25 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e53d1e2d6574ca4929a8698dd01a713 : SUCCESS in 5m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5b377966be16436a836cfa899dcbeb36 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbaff1aae17b4bf6b9fadd9c8d4f3259 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24f325c68325403eaa93bb9dc2bb0fbf : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3c68f77d834984bfb5bc85216a807a : SUCCESS in 10m 04s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/756e365e2ea545a88c451e6930a734d8 : SUCCESS in 1h 45m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a3f18504e6ed47c0b66131a8c39e156a : SUCCESS in 1h 55m 59s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/88908cd074284999ab87df8126eb6e33 : SUCCESS in 1h 44m 16s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6218f470343d4471aedb4bb25a6acae7 : SUCCESS in 1h 35m 02s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/3b0b396ea8024e229354ff13903b3353 : SUCCESS in 1h 32m 22s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/3f4bb84d8d7c4148ba8c082c0d36871f : SUCCESS in 1h 24m 41s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/35006cb4a22d49e79d6df94d043a00f3 : FAILURE in 1h 24m 17s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/3a7043fae1f3406abaee7b37556e06ea : SUCCESS in 1h 32m 20s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1988951cda3f48b4be6baa13991eec84 : SUCCESS in 2h 03m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b5ed8e3c32e442749c72e99fa6d65416 : SUCCESS in 1h 55m 06s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/11e4dec2e7a5464989019d217b0c8615 : SUCCESS in 1h 01m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b51b5a835864225862bb42990ceb5a7 : SUCCESS in 1h 05m 52s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/95eee1e6a486481a930b5e42ec8c8cdf : SUCCESS in 2h 31m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/062ed6a7c6f246639f5d477ca7775776 : SUCCESS in 2h 24m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/113b8555995749dc8add69aaae21b158 : SUCCESS in 51m 29s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2038712cd9444703a750e9be5aa32ce7 : SUCCESS in 59m 52s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e50754ca264c442092fbba5d645686ff : FAILURE in 1h 00m 28s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b37a916017844a67a77320f34d96117a : FAILURE in 26m 44s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f5ee3f8496494661b38dc34c8bf73d92 : SUCCESS in 1h 07m 33s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/982ea745d50a46909de48d1a6cc354cd : SUCCESS in 3m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9d06374085504047b16b5e1468ecf519 : SUCCESS in 1h 13m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/04ed7da487dd4de0b5572e88a9ae7bdd : SUCCESS in 1h 08m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/913a7e9ecf4e4e5c86451c0a0ed8c4ab : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"397e5456eb24d2846921236469f8811996df6163","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-28 23:41:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3384d90032bc07f0fc651d67c21b95e314c632ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-29 01:10:20.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6d97ba4673d818c15f9a60621b04edda95176e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 02:18:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a56cc58b41454ca881fa68ad5a41ace0 : SUCCESS in 1h 37m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d350813705d04a059ea5aced810f1190 : SUCCESS in 1h 03m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2edd1298dfd54e7bae214cbb94616492 : SUCCESS in 1h 24m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0213e997891f4faebdfdfe22b0eba7af : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbe47a143d8d4a69afaec5f2051252b4 : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eddd4e45895947be8edc0bf54be70193 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9d449987658e419b947d71448b181bd2 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b868d58911644d8989e139395017222 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa52c8a2bd394e9bbbdcc93d5ab7465e : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e9f69ce7ae4f7fad846f52435392ef : SUCCESS in 10m 21s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/867c7867bbb4417f8c4eacdcdc9f4c7f : TIMED_OUT in 2h 09m 30s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/39194a189f5c45d7a18d42f7cfb0687b : SUCCESS in 1h 25m 50s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/6b4a6366882a44b5a32a06a1ac639c4f : SUCCESS in 1h 45m 26s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/0174ee7580604a18805d7b83af2e4286 : SUCCESS in 1h 31m 14s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/7fafc8a3fa8043fb9297d1580d4f2123 : SUCCESS in 1h 27m 53s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/5bb3f9f49cc94908930693efc2c597c2 : SUCCESS in 1h 21m 42s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/642aa844a39149709890980e0537491a : SUCCESS in 1h 21m 25s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/75781dc80716487abea9a70460fc456f : SUCCESS in 1h 22m 10s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e635dd144e10449d9453eb9dccf7991c : SUCCESS in 1h 41m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/41a5de1a4aec48ffb8d1a5a5d5eaa63d : SUCCESS in 1h 39m 33s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/90610948c282418a999a6a2961369b70 : SUCCESS in 54m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9578a4e5494248b2a6c4adea78903998 : SUCCESS in 1h 02m 47s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/9d1788666f3643d7846561efca75cc2b : SUCCESS in 2h 31m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94e086c5e53a49229bf93f85508d90d1 : SUCCESS in 2h 10m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bc114bfeb7a644e5869f441e2d7c86fc : SUCCESS in 47m 52s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0248f5f5d6bb4cf38f944bd3148debbe : SUCCESS in 1h 09m 25s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5f90e6cafc4d45f5a68fc7d82d7c1d10 : SUCCESS in 59m 48s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/863705fde0bc42eeac794afa3f8895c5 : FAILURE in 23m 54s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5beb13edf8c34ade837f27b55c7ccf77 : SUCCESS in 1h 12m 28s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ccbd87411b7a4376b0cf89859e902ebc : SUCCESS in 2m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e1f539c3cfd34bf5b3b24e50ae2052a3 : SUCCESS in 1h 11m 17s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f7c6fe76f7b747cb946ed0d4eaf18496 : SUCCESS in 53m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/96a5e1d478c34505a0776faa6e2755d5 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"a4771d904acce8960dcbb4ef90432ac6a72f38a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 02:19:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a126b11d94ecf3d9208db325a044df99ed624bd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 06:10:16.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7792f5961846486cbf3b184f91c216d7 : SUCCESS in 1h 25m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6a9604be2b9a442d9ff0dab9f2a03fa2 : SUCCESS in 1h 02m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66fea7b768c14eb2a8753a9b54cea5e3 : SUCCESS in 1h 51m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8dbe0be9e24739820eb7eee3b927cc : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac9dc81839b941a2a50d794d953092ff : SUCCESS in 4m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7739df9640e54c51afe931da2de3233d : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73cc0a6198614a7ebe8d518d8e7a19f5 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcdcbb0ac2c645caa51285d1cece92da : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8170fdf377e8461bb953168e070f2218 : SUCCESS in 8m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d9d864d16d7f41739dd5b2d30f1580bf : FAILURE in 2h 49m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/973f465fbdcd413f96690241d15915d7 : SUCCESS in 1h 03m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a342381da4ad4fd3ac5c611eed27d218 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":3},{"id":"de6e9686a2c4525faf2c2abb56964ff4ef1a49d2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-29 06:45:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8b8f42d9fb1364ae957cd54652b1566a22ae7c2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-29 07:31:01.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"07233b04dd142a795cf111a8df18c08e2c43dc3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:30:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2f7fdb73ae946799b176bd28585f9a6 : SUCCESS in 1h 31m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d71f93b3b2d64acf9661beb3df7b1615 : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de3fbf5ee5de49be90fc63611365103c : SUCCESS in 1h 25m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5a2bdbc48bf4fd48490b1d7b5687a50 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8be3e6d83f438fbc878e88352eea38 : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7428a4c296341f6bd53d94aa700ae46 : SUCCESS in 6m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5485def562cf4eb9b7dc9b808320a9ae : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00a810e2f7584fc3963ce298d24f27c7 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/376e07db5bb04c82a6359b155a407090 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36c20b42c584d37ad065c6d68ad88a1 : SUCCESS in 19m 22s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/97849f4245bc4c698bfde2270cc9f05d : SUCCESS in 2h 01m 06s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ef8bd460ecba42e1a28cf61f12db869e : SUCCESS in 1h 22m 05s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/f550b156a4fa466b8ae7eab624133dac : SUCCESS in 1h 36m 45s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7b888531a9034c188edbae8e03027758 : SUCCESS in 1h 41m 11s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/3c9c3c9504f749f28ecd43ec2f2d215b : SUCCESS in 1h 35m 07s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/925be5d8841e4c62810bbdaa79f8bb92 : SUCCESS in 1h 23m 03s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/2f0b8c56b17d4ee79e43f2b03c7786f8 : SUCCESS in 1h 38m 10s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/52dbe2251fec4725b4865be30e43e8d8 : SUCCESS in 1h 33m 23s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/39d48ec3fd724c32a5afe6c30945da04 : SUCCESS in 1h 53m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/57a9d33135e14e60886eac30b270783b : FAILURE in 1h 37m 54s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2ebb2f383217462f964fbe2f935f9d2f : SUCCESS in 1h 06m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7db154605074abb944c1d04249aa984 : SUCCESS in 1h 17m 45s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/cd3ebf5d62e44a30875e8d1166e16c73 : SUCCESS in 2h 37m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03da128abb68431bbeaf0fefc65155c0 : SUCCESS in 2h 25m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8a899bf2e20745f9a1ade1f99ec5b967 : SUCCESS in 51m 05s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0f9ab1e46044eb792f55fa182d7c1fb : SUCCESS in 1h 08m 10s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/190557f9092741389002e0f7f9d7c89d : SUCCESS in 1h 43m 18s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7844f58625b84dffbc0c75e370728945 : FAILURE in 24m 04s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/dfd4d3e16abd4771b86c889e702af1e2 : SUCCESS in 1h 27m 18s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1fee83738b46461babc24ec484f2484b : SUCCESS in 7m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/50212b52138942f68492c87125767730 : SUCCESS in 1h 25m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/95713bf420c74b7abf3fd13d43eadae1 : SUCCESS in 1h 22m 05s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/395bcf10034e433fb6c94ed3764a7bd8 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":3},{"id":"2eb23d77f5a973a86a3223dfbbc24c6f85f81959","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-29 17:15:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"44a36c64055178ec901f987410e008f4c8bb096d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-29 18:42:20.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6352c96676b19e33c55aa568167c247be9f6d7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 21:06:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a1e522aa62e7423bbb265ba9ec66efc5 : SUCCESS in 1h 39m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2c6f55780519483ca87757357802a388 : FAILURE in 1h 15m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f46b85d8d2ac4650b132a60a1104f2ca : SUCCESS in 1h 13m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ea90815ec3c4844967d7b657c5be696 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f541124087c4441a5604fc6cc0ec5bc : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50d7e9cc06e346d6b30aef8c7956a720 : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/39740748c3d94a6fbf510024dacf38f1 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e30ec79abbc4d5fa38efe7a1a78f5c4 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceab00728d194a109434f6c4d7e8c392 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095ac7d41b0c44a6bbff736789faf108 : SUCCESS in 12m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e32ab13d92394ac484419f4974be60c8 : TIMED_OUT in 2h 09m 39s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a636b7155267469b808d3ba67f6565ab : SUCCESS in 1h 21m 49s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/9141e3b0766e498fa8b7e5fdb277b24a : SUCCESS in 1h 51m 13s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e93d9bca73c14562ba53399c361085b4 : SUCCESS in 1h 55m 02s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/b7a91748e76e4ef68ed8ba257c5bb839 : SUCCESS in 1h 24m 40s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/3f077580b61e49d5b04f5a1822cf8b0a : SUCCESS in 1h 26m 59s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/05e59f2b82a04e43970e7f4539311409 : SUCCESS in 1h 39m 13s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/795d7354031f4534ab0224bb5ceac27b : SUCCESS in 1h 17m 31s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/49f1ebe9ef864d668839174b6247ea78 : SUCCESS in 1h 42m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/650e6df8b07445a397a2259150af82d1 : SUCCESS in 1h 36m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1b16286d2cde4b8a908d315c4acb3c14 : SUCCESS in 50m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92fff3fa2d7c4e47ade4ec9fb1bb2327 : SUCCESS in 1h 07m 11s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/18c2fb9b77024aad884d902b40bafd16 : SUCCESS in 2h 31m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f377348ec604afa9929ccd7bc6e35a9 : SUCCESS in 2h 18m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b97c307455f141d6b9557034a25ea06f : SUCCESS in 58m 07s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca648641598f4844b36409ec85c27b83 : SUCCESS in 1h 11m 47s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6218d438cfb42fab9b79c18f9f6c62c : SUCCESS in 1h 12m 35s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/86ebcf1281e34fe2be676b7aabb2418b : FAILURE in 26m 37s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/21db592b075e4acaab36cf9f41917bd4 : SUCCESS in 1h 14m 53s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/57c53af41a1a49cba5dfc7ebf597f39c : SUCCESS in 4m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7e92473b387542d5af045b3677cd3ba6 : SUCCESS in 1h 57m 39s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ec96882077af46ee8d927195ca8799ba : SUCCESS in 1h 24m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2bf11def82db4b579614eae7b765eea2 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"812b3fbd3f85d4ac41e30f0a4067699f8c3d315a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-01-31 10:02:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fbd1412301c4f6dc4941c5d8c9f0302065af236d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-31 10:20:54.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"dab3699c36c6686285e335794142b1a9ecff2bf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 12:59:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1cd29cd34b73472ea7457a7ca0949d72 : SUCCESS in 1h 20m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c390c9b481464e56a2919dc8bbf85ee6 : SUCCESS in 1h 05m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4a9473d5d3b4161bdb56a93fb721074 : SUCCESS in 1h 19m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8cd9aa4b0f24c6a974493a88c15871c : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7489758064647928a7f4ecef17d5e0c : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/242e2f0031c0451196601d3ac08ea572 : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f515ed66e12947458a1059fead82f18b : FAILURE in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d65d144cc074c07af3ee0e3c151bfea : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd745d626c674fc8a5d9816c19bfef53 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9aa515f65a4b6da77c8ce749206c25 : SUCCESS in 11m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1dec200801e24079b5cbf54651a8fae9 : SUCCESS in 1h 47m 25s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/0ad7db74f1b14456b06fc54c5bbeb6db : SUCCESS in 1h 24m 27s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/62dab6a3ecf2409a9131919d46902d73 : SUCCESS in 1h 36m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2b41d2aad808455e84a9786d848ee92f : SUCCESS in 1h 13m 18s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/23d92fbc8f594a578850acec3198c686 : SUCCESS in 1h 47m 21s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/ade72f0bd8e94cb49db04a5e6cb44d53 : SUCCESS in 1h 29m 56s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/f46d6a81b3cf402091221df42ecd0451 : FAILURE in 37m 06s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/2923bd2f06064d709d5e970c619c852c : FAILURE in 35m 28s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/66d147a0993b4e67ad55a661535cf344 : SUCCESS in 1h 52m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fa27cdf4b83944a883da9ea1ae799794 : SUCCESS in 1h 52m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/655f0aa32aff4645b68a6bf1f3ec08bf : SUCCESS in 47m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec8314b610f54d7f8375ab1acdecda90 : SUCCESS in 59m 34s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/3e1aebb34c9046c4b2bdf85b4668eb06 : SUCCESS in 2h 26m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff3a3a9b8ed1445397a438be001d57e1 : SUCCESS in 2h 12m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/83e8db9b33a24e86a72deb557ffcc3c2 : SUCCESS in 51m 21s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0b281ebb3a44f94983e4078d32790a4 : SUCCESS in 1h 27m 12s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94242f6ae89040a9808d888551dbb2e1 : SUCCESS in 1h 15m 01s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3a05cb054d584a349f19fb10fc3f385c : FAILURE in 50m 02s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d5f17eeca8ad4733af6397f567151024 : SUCCESS in 1h 13m 56s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d68631578b6c49a5917fcb3173ef4878 : SUCCESS in 2m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c462506bb026413f9fcf6e8a65cd88fa : SUCCESS in 1h 16m 26s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/28b42db7525947ec8f5c125edc7db2df : SUCCESS in 1h 00m 14s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/42625e4001d0417cbf736b6008c73ed7 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3},{"id":"3e4c08476a8f11c4025a336689d6a821dc725f4c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-02-03 17:08:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"93d5772320a351105be5bd5ee22cbf922871ccf1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-03 17:46:19.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"d43abecc6e9e2bdbc69067e122247c26598397d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 22:55:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5f339b1562054e1c91ec87893e5f2b42 : SUCCESS in 1h 38m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aaa9a6eb618b401e97cf8d02b9aabe89 : SUCCESS in 58m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1301cfb0ea649f8952531bca30ebe79 : SUCCESS in 1h 19m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e9e14c6f81f4902ab32becc08b9c4b6 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d9ee36a9794aeab348abe46504eaaf : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/304e64029f414723aeb0c5a1e377c3a8 : SUCCESS in 6m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/99361f9506e343dd9bc313fc5dc3f332 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ccd5cba2b294e93b4d9a7a79c308955 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10d10af3981543c38d185bf752d1324d : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/494724c7c8184f7fb926028c79a6f3f2 : SUCCESS in 9m 17s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c6d4d0a5125a4f379fec488044a9a47c : SUCCESS in 1h 45m 16s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/26c2e7145d0b4329962e7bbd7be4f696 : SUCCESS in 1h 43m 39s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/4ad6612db2ed40f1880828015831763b : SUCCESS in 1h 27m 09s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/702f0b1ed82c4557a8e9c73014c594dd : SUCCESS in 1h 23m 20s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/5efef84a23ed41be981c672cf0fb01bf : SUCCESS in 1h 30m 29s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e47462e9946e416eb71480a50effc36f : SUCCESS in 1h 17m 08s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/4fde9a7813b54bd994b49384f61fc152 : FAILURE in 1h 14m 20s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/c2d42b4118174403926b4c38fb714621 : SUCCESS in 1h 46m 39s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bce05850d92f47ed9359e02377057f97 : SUCCESS in 1h 50m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5787e9a4de444b458524617236ae240a : SUCCESS in 1h 39m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/09a9de9bff134d32902fac83fcc55f96 : SUCCESS in 54m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cfc66f888be44d9a4814365e37f9fbf : SUCCESS in 1h 03m 19s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/52857b98814248298430313f928387a0 : SUCCESS in 2h 28m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87c92dee1ffc4af0adbb0e972efd6157 : SUCCESS in 2h 25m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/63d165dde4b54906bcf9d9340e4a2565 : SUCCESS in 56m 01s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/759251c3740b4dfeadf834fb7f234434 : SUCCESS in 1h 42m 50s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7703943199824745b4cfd38eb17bfdc4 : SUCCESS in 1h 11m 46s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/915c16647fdb43a790d0fd778d47cc74 : FAILURE in 26m 05s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f10743ff02244b6db243001d1fde4c8a : SUCCESS in 1h 23m 42s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d1bb3df2e9d5446392eced3e5ff8aafc : SUCCESS in 3m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/97de7a0ab5bf47bebe7bb45e1cfe3be8 : SUCCESS in 1h 21m 06s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/cb8b56560dcd4d91a0ac055847292cae : SUCCESS in 1h 10m 15s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/89fdb5c6f83247cab94e5d519e41c7ab : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"667d6f62911a2e4e1f74720308129e6a0de87d87","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-02-06 05:32:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"18792ff9b644d2fae877f727509cccb9085f55ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-06 05:50:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"108a78c8ab2eb2c35c61ce34d721199a659ed173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 08:19:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/49a9ba239492494a8100a9bdc8550342 : SUCCESS in 1h 35m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/714dd1eab8ff433cbd93c8b22643a184 : SUCCESS in 1h 00m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/986b28c123184de59edc5186c78a71fe : SUCCESS in 1h 29m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa23f40442e044689e4671f8e3060667 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf99a7d1389a4711b388ea1d0985be0d : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c13d40f5a34478f8ed957f8ab76d89a : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc0503bbe184f2d92345355a4e7c6a4 : SUCCESS in 8m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/ab0884756a0849cd935c44a212c6587d : TIMED_OUT in 2h 07m 26s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/57923fa5141c4f08bd50c331c747136f : SUCCESS in 1h 24m 58s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/77b9f09455a04134b11c28f5def2563f : SUCCESS in 1h 31m 28s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/506cfb9f649d46b48a33fb9fee304b8c : SUCCESS in 1h 18m 46s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/547754c1f97d44eab4374ca64a478a04 : SUCCESS in 1h 30m 10s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/0b33e8c0395c4bea8bab88458833b2c1 : SUCCESS in 1h 21m 50s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/00da0f531be74904883366a45dede566 : FAILURE in 19m 20s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/fe7d7bdf09d04afcbd62e720318e25f4 : FAILURE in 20m 08s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b999cbcd66ef49f1a0794739871891d5 : SUCCESS in 1h 38m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0970dd0fd4e146959c738c58f5754cbd : SUCCESS in 1h 46m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3d01d22da0304f16aa6545b9ac021862 : SUCCESS in 1h 01m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57fdc48ec6ac43c88443ef00079272de : SUCCESS in 1h 08m 14s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/5bbdc3b613254c32be57c472bc4bfe8d : SUCCESS in 2h 35m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cef3a74d241462cb93e0e97caa4064b : SUCCESS in 2h 40m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/97e6acf6791b441ba1bd19cd8e79f0b1 : SUCCESS in 1h 00m 04s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9752b0dadcd47a390e55e069e832be9 : SUCCESS in 1h 10m 11s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0e4a353644940848021896095057a9b : SUCCESS in 1h 07m 33s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b328a50834c44ba7a0c7635cee8803aa : FAILURE in 22m 54s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/93c77c4ee8c2408795539edc37a2697d : FAILURE in 1h 04m 05s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f3b77658705b4f89b7951c3dab8cd2e9 : SUCCESS in 3m 17s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f6d9770b7ffd4e19afa0edf2ba7ccd70 : SUCCESS in 1h 17m 54s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/43eb363fc6554280b8b4037a5b845dde : SUCCESS in 1h 13m 57s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b09ac668736649fda14dd1a9d0e482f4 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"b24fea250a02e92ddb3bbe7512366eb5c610cd7f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-02-10 07:34:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8d2b9f2bddf0c2a6bdbc8391383c71bb87324099","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-10 07:50:43.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/93/702493/3/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"962eb2285be7afb074837bc5d2244a89374d62df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:14:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c2608b6f0a3e4764984c4d4c537ef4bf : SUCCESS in 1h 24m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4b78f5e4d5414d0fb3c8ba6c4b051f45 : SUCCESS in 1h 10m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c00ba68e7a074e5aa00745ee243a5f12 : SUCCESS in 1h 24m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/844b0ee13f9441fea1ba8cdb89c0595c : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b646d9773b3342e0a50d5f9efde9d4aa : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73d81617f442418fab6cdac5bcaa3977 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc6fa20809b4bbdb01226c042920b2c : SUCCESS in 9m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/7d31c1874bd7465cb61929108dbcbe85 : SUCCESS in 1h 45m 06s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c6255f09a4a2481eb0bda91ae2cdd63c : SUCCESS in 1h 40m 44s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/e04d9e21bf384e75b9265107b9156ca1 : SUCCESS in 1h 48m 39s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/9b618c3210e546efa4d7e76d392e8cd0 : SUCCESS in 1h 28m 01s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/32616739f14a4ea98ebb808e61a6dff6 : SUCCESS in 1h 29m 14s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/1e503707c63640ada3b398926eafebff : SUCCESS in 1h 15m 21s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/5fe1c63e01984c40b139eb6713bfba38 : SUCCESS in 1h 28m 46s (non-voting)\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/ef94918bd61e437da7ace41c99b8471b : SUCCESS in 1h 25m 29s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad0991f38cf74467a2f332e97300f2e2 : SUCCESS in 1h 57m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2b01c6dba72c4b069a832e1616568747 : SUCCESS in 1h 44m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/70b1aa8f18524e38962b3dd70f327718 : SUCCESS in 52m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d32dc3005c6c4ba4bec29ac40317a92b : SUCCESS in 56m 53s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/0df3fc176a4944a6887ca5759132dd70 : SUCCESS in 2h 37m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb79d00d259d4c0a9f6796a0dee90c1b : SUCCESS in 2h 11m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/49793ae969194139a6b1b6b9378c7c52 : SUCCESS in 52m 36s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b11eb59c6134a0fbe3ee442d84dad07 : SUCCESS in 1h 12m 05s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f44a411afa3240208448bc196f48ea6e : SUCCESS in 1h 01m 01s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8752f6f8920244f7a3b97707d770f995 : FAILURE in 25m 11s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4b031eedd74f4868802b44c6705f47e8 : SUCCESS in 1h 11m 13s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/db0ca1540dd14d78b61cb43976b9dcc8 : SUCCESS in 3m 43s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0de088a9c2be4115ac278bd97aab5503 : SUCCESS in 1h 20m 12s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fb45468a4a3240aab077eedf5e67837b : SUCCESS in 1h 09m 36s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/442b730af48e45658a760274b0a6ddab : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ca299b1cd4e6c524e2aaaac89c10552b1ac50561","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:15:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1281cf44346e9ebf3667cb655d77b8e636821440","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:00:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1094ced0e25cbd483028bda4f3f62e4cc47e4e0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:00:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8626d1c051204323911c293c469619b4 : SUCCESS in 1h 24m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/205de8c24af94243b3592db4a5562dcb : SUCCESS in 1h 04m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cda34196ea7b4eb0b79e3cbeff161883 : SUCCESS in 1h 17m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/453144113c534cf7a241d1d2a450970b : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07c4b152e0ae4babae6f6ec22dd9046e : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3790ec4f3142470a8abf0dbef2681b33 : SUCCESS in 12m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86dbef7b8e014c808641173ec6bb4ef7 : SUCCESS in 2h 40m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9aab06eb1c844a96b0f4988f2c6020ad : SUCCESS in 1h 20m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b781e9b3e6f4506af96119bde7c1a9e : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":3},{"id":"e8a75099779baea58d91b0ca88a006fe337ba21f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:03:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ba00b649c4429b9cedabe1cb969138 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a8bfa172c19e96ba56a99b9c2a108d55df20b7ba","revisions":{"363697e62f8c4a69425a51c3e27da42c44195077":{"kind":"REWORK","_number":1,"created":"2020-01-14 17:47:11.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/93/702493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/702493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/702493/1"}}},"commit":{"parents":[{"commit":"b63725301488a910d2ad5e1a3f8a11b45b187845","subject":"Merge \"Blacklist compute.volumes and compute.images API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b63725301488a910d2ad5e1a3f8a11b45b187845"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-01-14 17:34:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-01-14 17:34:44.000000000","tz":0},"subject":"Refactor methods to be used by other tests","message":"Refactor methods to be used by other tests\n\nThese methods are useful for a new snapshot test so moving them into\nmanager to be reused.\n\nChange-Id: Ib5ebbc6d5538a6179eca27c203faee6805af0093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/363697e62f8c4a69425a51c3e27da42c44195077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/363697e62f8c4a69425a51c3e27da42c44195077"}]},"branch":"refs/heads/master"},"6fd509f83d1574965db9ee2a38f59334cb048237":{"kind":"REWORK","_number":2,"created":"2020-01-16 12:08:00.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/93/702493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/702493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/702493/2"}}},"commit":{"parents":[{"commit":"b63725301488a910d2ad5e1a3f8a11b45b187845","subject":"Merge \"Blacklist compute.volumes and compute.images API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b63725301488a910d2ad5e1a3f8a11b45b187845"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-01-14 17:34:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-01-16 11:56:03.000000000","tz":0},"subject":"Refactor methods to be used by other tests","message":"Refactor methods to be used by other tests\n\nThese methods are useful for a new snapshot test so moving them into\nmanager to be reused.\n\nChange-Id: Ib5ebbc6d5538a6179eca27c203faee6805af0093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fd509f83d1574965db9ee2a38f59334cb048237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6fd509f83d1574965db9ee2a38f59334cb048237"}]},"branch":"refs/heads/master"},"a8bfa172c19e96ba56a99b9c2a108d55df20b7ba":{"kind":"REWORK","_number":3,"created":"2020-01-17 16:54:50.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/93/702493/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/93/702493/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/93/702493/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/93/702493/3"}}},"commit":{"parents":[{"commit":"b63725301488a910d2ad5e1a3f8a11b45b187845","subject":"Merge \"Blacklist compute.volumes and compute.images API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b63725301488a910d2ad5e1a3f8a11b45b187845"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-01-14 17:34:44.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-01-17 16:49:50.000000000","tz":0},"subject":"Refactor methods to be used by other tests","message":"Refactor methods to be used by other tests\n\nThese methods are useful for a new snapshot test so moving them into\nmanager to be reused.\n\nChange-Id: Ib5ebbc6d5538a6179eca27c203faee6805af0093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a8bfa172c19e96ba56a99b9c2a108d55df20b7ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a8bfa172c19e96ba56a99b9c2a108d55df20b7ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
