)]}'
{"id":"openstack%2Ftempest~703646","triplet_id":"openstack%2Ftempest~master~I2ae25eb7aa88258580a4163f617548df35e74bdb","project":"openstack/tempest","branch":"master","topic":"queen-eos-qa","hashtags":[],"change_id":"I2ae25eb7aa88258580a4163f617548df35e74bdb","subject":"Remove the Queens stable branch jobs from master gate","status":"MERGED","created":"2020-01-21 15:48:30.000000000","updated":"2020-01-27 23:24:48.000000000","submitted":"2020-01-27 23:21:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703646-1580167305534-bbef5f2b","meta_rev_id":"143842a33fee3c43b420e56f07fc3578e0aa7f05","_number":703646,"virtual_id_number":703646,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"date":"2020-01-26 02:58:44.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 23:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2020-01-26 00:18:53.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":2,"date":"2020-01-24 01:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2020-01-27 17:47:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":1,"date":"2020-01-27 08:47:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":1,"date":"2020-01-26 00:18:53.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"date":"2020-01-24 00:55:52.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 00:50:40.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 00:50:40.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-01-24 00:55:52.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-01-24 01:21:25.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-01-26 00:18:53.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-01-27 08:47:49.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-01-27 17:47:34.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-01-27 23:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba94790eb978122efd9250c4cb5b236f1883d35c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-21 15:48:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf6e3d1897bcb66abcf1e2cb1a6edce773864a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 21:15:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/824c3311c7ec493984eb2c35b98dbfea : SUCCESS in 1h 29m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cdb435bdfc8e4fe49000600a44f99afa : SUCCESS in 1h 18m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4d7545d18ee4e8b8cc053d03d0e4018 : SUCCESS in 1h 29m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c5c9738ea1e4f92a0486db62ff2b28a : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2397c875106f40869408f503d2ee21b9 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/661efa658af6444db99a90c8791bf71c : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bd6c9f7f593e49e0a0b9ce25578d5d0f : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f9a2f59abb74c26803b671138593710 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d541a76a0e04e83976d08e2af35761b : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775e7a9905fa4e42967199f1b36da02c : SUCCESS in 9m 55s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/237b8e3735c64aa199b321301ee7b0ba : SUCCESS in 53m 18s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/289e13514e8a436fb65c2a2758b28b87 : SUCCESS in 42m 35s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4d1894cc3b8a4b11b16bc83aacd2aac2 : SUCCESS in 1h 44m 20s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/1baeded43df94772a28342e2909c8a0f : SUCCESS in 1h 14m 54s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/7313f238a19941b592c1f307d395c921 : SUCCESS in 1h 31m 29s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/bb3b69f7963743f8bad829156fd6cfd9 : SUCCESS in 1h 12m 43s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/2b4bf918fba140eeaaae281d6d857fc9 : SUCCESS in 1h 23m 03s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/455144abb7284c5aa94132d5c872b7de : SUCCESS in 1h 15m 27s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/aa4e05484dac4d06a5ee9a6309843b63 : SUCCESS in 1h 42m 46s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/2d4143836b9847de8ccde8dced8f8897 : SUCCESS in 1h 41m 58s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/57a36c8c0f4d478d9935cb9b163cf79e : SUCCESS in 1h 55m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8b4f82e8977747a58c54e6c56e6149e1 : SUCCESS in 2h 08m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/eeb1f277123c422ab7f6d0fdc2254b10 : SUCCESS in 59m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b5908e603304f9b85c12ce2e80b1d4f : SUCCESS in 1h 04m 15s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/f7142e93460d4076bdf6e1020f59b9af : SUCCESS in 3h 00m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c9005676a074d98952eb256ea9d78e4 : SUCCESS in 2h 42m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9e20da8bcb714b19b8f535af38fc5086 : SUCCESS in 54m 45s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a7f0f5d97a94d3994d988b7439abf85 : SUCCESS in 1h 04m 04s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0c109ffa9494adaa85deea1cec02ebc : SUCCESS in 1h 11m 04s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/af7d5ff901b949709a0abd78d97cd33a : FAILURE in 27m 23s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a55d1ef78cc644f1a354090163f6adc7 : SUCCESS in 1h 09m 50s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/bda16b31074a403693d11f6b2ea4c961 : SUCCESS in 2m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/cb13d6fce9694a2f9f1611864c7846f3 : SUCCESS in 1h 19m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9123f9d4952d492b85a96498e71aa5e1 : SUCCESS in 1h 22m 32s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2c2d81a8a3b347879252ae08f7cd1aed : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"927d08645d0358a2bd2d2efac96aa9aeb72cf2b0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-22 21:01:15.000000000","message":"Topic set to queen-eos-qa","accounts_in_message":[],"_revision_number":1},{"id":"ba95f5e7e2a8eb0250d4cde99f17c1ba08b04fd4","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-01-24 01:21:25.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"d1ec94cd89787a7c5ea63cfb3177d1f8de233135","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-01-26 00:18:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79d15d9c37d8700cc6c41ff694bd539d89818640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 00:19:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c4b75512231707f7d4292e857172a749616316e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 02:40:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/539fe7798678420cab50901fc804b8ab : SUCCESS in 1h 13m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b789e92ff9744ab0a6764ffd45405d0c : SUCCESS in 1h 10m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c819cd4136c04931a8e4c2d76c652b0e : SUCCESS in 1h 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f706e17d7b6643e4a80614c5190d34f0 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/866c3032ed034dc4bd24a6341607c827 : SUCCESS in 3m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/20a009df88204f5e889da5219c862f3a : FAILURE in 2m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9ce6a1a11f547fa86c90f12c3cc5af5 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/888031c3d1c24c7185bfd2edd80e164b : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f0a385c9df46739d6dd6787fd13d76 : SUCCESS in 8m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ed2fd4eb4cd4259b5b65c99ee980b63 : SUCCESS in 2h 17m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f5589406ba74040bfcfa53bc740d98a : SUCCESS in 1h 10m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ce27cba47844efd8f369631ced14bac : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8e332ba37435ea7eebcf021491aefa408e911560","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-26 02:58:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40dd889ca941d81e45d793675e9cf124423cab2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 05:28:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/91bb204b706c4c628bfeb7b8bf3869ee : SUCCESS in 1h 44m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f579d7d7f0b14c1485d1d1ab570bb561 : SUCCESS in 55m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e20d04af91004343a9f0e2ae61e3872f : SUCCESS in 1h 19m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcf702d327374e5caa10761c5b8db592 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a711f0c2ee3348958d1265b4b8e198bb : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/783b0f49a8a545de84d15315b0ececc2 : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/19b8c9d72878445f822c51e33b5f9750 : FAILURE in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ede776198fd46ccaf1184e67254aa25 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65479aed27f44937a7e12b919f8e804b : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d1dbe982fa46d4ae5713b763a60e87 : SUCCESS in 10m 27s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/ea6aa1a0042640139f18a4e7705277be : SUCCESS in 37m 47s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/9b8d7c08f5f94d2d96f197a4cf52923a : SUCCESS in 36m 16s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3983c8d4352548ac9a36f6ca6727cb17 : SUCCESS in 1h 50m 15s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3b7e1f9def0544b4a943e830dda0dcdd : SUCCESS in 1h 13m 17s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/978ffbb6a1174748bbde124061f036fe : SUCCESS in 1h 24m 51s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/62e9227ad99a40299524dd9bbc1d8b70 : SUCCESS in 1h 15m 36s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/3a38fc970b144052a8777f3461f17850 : SUCCESS in 1h 12m 50s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/62c73f98adc6407e8ab57fbf4166d49e : SUCCESS in 1h 13m 25s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/523a851848c440f289af08ce2fe102a8 : SUCCESS in 1h 23m 19s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/550f92d9673d48d8ba617a0357720734 : FAILURE in 18m 32s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f2e141ef90049d1bc93e3f4f199e2cd : SUCCESS in 1h 32m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a0edab0e759d4b4197cb491c607e95a2 : SUCCESS in 1h 33m 08s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a720ff5afd06452c96fb5595e003a0f6 : SUCCESS in 50m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d279704d5cf42da8fdcc5a687ecd05e : SUCCESS in 1h 02m 55s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/e6d7c3c211224410b44feddd6f1d0c91 : SUCCESS in 2h 24m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2f88600554d949b3918479b9bc31232f : SUCCESS in 2h 12m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/95f7ead4c3d74455ad1c32c991fdbefc : SUCCESS in 48m 42s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/15d261eda6f84eee9f84b137e48f30d3 : FAILURE in 1h 06m 41s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68af4a64628d440489ea8f4a0ac22e85 : SUCCESS in 1h 00m 58s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/66d87dd0fdc446ba93a5cb20200b3e99 : FAILURE in 27m 46s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8a275fe7e46f4c90bac23bc63f987f1a : SUCCESS in 1h 08m 33s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c7b75273ddc54cbd922bb90ae26a1ae8 : SUCCESS in 3m 04s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8c6ccb906c95464688093718dfd33cac : SUCCESS in 1h 16m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b63b8e685ad84eb7a994dc3116b13d42 : FAILURE in 1h 15m 11s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ccbc5f874f2b44deb4f2aebaef15077b : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"50f9b81108d4f2897c1f54b369bcc316eabb5c7d","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-01-27 08:47:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b437d38586fa63da49b0665af5f9656c9a1381ec","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-01-27 17:47:34.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck  ... py35 fix has been merged","accounts_in_message":[],"_revision_number":1},{"id":"c20fbb640755d087a5534ad0d5937e8a8de07030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 20:25:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7adeef280ec4436f9384d7f8a6eaf99b : SUCCESS in 1h 30m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65f916c240644f32ab17bdd0a6da6958 : SUCCESS in 59m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8841cdcbef8a4ad2b80238744f6cbf09 : SUCCESS in 1h 26m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f532acac971465d873b227f51bcd5d5 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/667befd508ff42a48c941ae57782978f : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58dba6f04f354df1a144e2c14d437516 : SUCCESS in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31d417e7de414c7bb914a48f6a2a24e4 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/087b2c10ab694e4a97bb786f01c2dae7 : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e566d085050c40b1bf71793b54f8c128 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70411f89ef2544c9b1a6fd1cbdfd297d : SUCCESS in 15m 45s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/3cf5c90415514d869392d02854ea9643 : SUCCESS in 39m 09s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a0e03a4bb61f4cad8abb5eababd681aa : SUCCESS in 39m 37s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/00572d7279c04bcc895e4b88ef8272df : SUCCESS in 1h 56m 52s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ccf830467d90454daa3d3d3578d3029b : SUCCESS in 1h 16m 21s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/5e4e13e2e3d54c6d889e48ef4fa50e8b : SUCCESS in 1h 35m 58s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/9c75abeb7bc3472ab36afd4dd495c40e : SUCCESS in 1h 17m 57s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/1bf5e162c455455fb4f8caa412096088 : SUCCESS in 1h 27m 11s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a8dcc31af2e144c681b33d9715c14ac6 : SUCCESS in 1h 20m 26s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/ab71ff6839ab4b3cbf5762fa1e1131ec : SUCCESS in 1h 24m 50s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/f0de7828a1164a4c8afc311638ce2f42 : SUCCESS in 1h 46m 33s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5ad02e2f93ea489f99a750f72694092a : SUCCESS in 1h 45m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2219e37857ae4d319fec4add2a15248b : SUCCESS in 1h 32m 14s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4b20d11b95284004a7fbb9b12b94f780 : SUCCESS in 51m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17c7dd0c63ce484aa1a7a6813901565c : SUCCESS in 57m 35s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/eadcd4ba3fad4422a03b7d25d6544c26 : SUCCESS in 2h 30m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e825dd24954e47ca8ab81df7b963bc97 : SUCCESS in 2h 13m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6b7ecd70ed94482c9665ef6b0dd9065c : SUCCESS in 56m 27s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b8f5cfe04b2424d944d1c0865f76de8 : SUCCESS in 1h 08m 10s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d32f3a3989f407b87bca2b303f2cee4 : SUCCESS in 1h 09m 17s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2c37fbfcaf854c6d87b4e43ef8feffbd : FAILURE in 26m 29s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/3408155abb074f2cbd790191319fe42f : SUCCESS in 1h 12m 32s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9b3b029e421c44e8bf9a445025b6af27 : SUCCESS in 5m 21s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/934c48737fed461a9c764fad268d170a : SUCCESS in 1h 16m 41s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/271ab41875304ede8bf6f74930756633 : SUCCESS in 1h 23m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d58484bb0d164e8d8bd7bd5040c50042 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0483dfd88cd0ec966ee956c31b2bcd21fa30f43f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 20:26:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcba1a6eadce3ea8cb22c63ce3eb2886c432dbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 23:21:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/80af6692437241068c4bc5a12c75bc6c : SUCCESS in 1h 52m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d58344f14b314214892ee98a36361a21 : SUCCESS in 1h 00m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80d56ac44c334b4ab23cb33d62872b93 : SUCCESS in 1h 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cab746b895746cf9213119e49553543 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7147596c1524bf5a2a16841f303e342 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/315b06958f0e4c568bbec487d0fbac3e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a063e9c854a4470ab44bfb97da9cf7c8 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/683c1b1e3e2a459192e5241f0459f5a1 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa452e76cab84aad9a174ffd8178ea4f : SUCCESS in 17m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e20c0affd9a447cbbf96a5748394889 : SUCCESS in 2h 48m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f4f8ad776644dff83237d376b08f12f : SUCCESS in 1h 27m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/494253b2b14d4774b2e0fd9b0c86d736 : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"bde7469046717373e2b624cad4e6711243eb3891","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 23:21:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"143842a33fee3c43b420e56f07fc3578e0aa7f05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 23:24:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055ee9507e2543528f587941bb5f03e4 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64da1a1403ae87226b3f4ac33c43a256fbaffbdd","revisions":{"64da1a1403ae87226b3f4ac33c43a256fbaffbdd":{"kind":"REWORK","_number":1,"created":"2020-01-21 15:48:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/703646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/46/703646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/46/703646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/46/703646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/46/703646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/46/703646/1"}}},"commit":{"parents":[{"commit":"d43223773d75d2e82fb33a1281038e611c41d0f3","subject":"Merge \"Add releasenote to tag the end of support for Queens\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d43223773d75d2e82fb33a1281038e611c41d0f3"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-21 15:45:39.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-21 15:45:39.000000000","tz":-360},"subject":"Remove the Queens stable branch jobs from master gate","message":"Remove the Queens stable branch jobs from master gate\n\nTempest 23.0.0 is released now, which marks the end of support\nfor Queens release - https://review.opendev.org/#/c/703521/\n\nWe do not need to run the Queens stable jobs on master gate now.\n\nDepends-On: https://review.opendev.org/#/c/703521/\nChange-Id: I2ae25eb7aa88258580a4163f617548df35e74bdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/64da1a1403ae87226b3f4ac33c43a256fbaffbdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/64da1a1403ae87226b3f4ac33c43a256fbaffbdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
