)]}'
{"id":"openstack%2Ftempest~707223","triplet_id":"openstack%2Ftempest~master~I95a1050d6a5e2e40793ae7f18ff0d326455b4b36","project":"openstack/tempest","branch":"master","topic":"bp/flavor-extra-spec-validators","hashtags":[],"change_id":"I95a1050d6a5e2e40793ae7f18ff0d326455b4b36","subject":"compute: Use valid extra specs to test \u0027os-extra_specs\u0027 API","status":"MERGED","created":"2020-02-11 18:10:58.000000000","updated":"2020-04-08 21:19:20.000000000","submitted":"2020-04-08 21:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707223-1586380542770-43e60a6a","meta_rev_id":"caea9f7743e57a14e1e33b362094c6cc93117838","_number":707223,"virtual_id_number":707223,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-02-12 10:22:24.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-18 15:13:00.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 21:15:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":2,"date":"2020-04-08 14:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-17 09:15:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":1,"date":"2020-04-07 16:57:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2020-04-08 09:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"date":"2020-04-08 09:55:35.000000000","_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-04-08 09:55:35.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"date":"2020-04-08 09:55:35.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":1,"date":"2020-04-08 14:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"date":"2020-04-08 09:55:35.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"date":"2020-04-08 09:55:35.000000000","_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-04-08 09:55:35.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"date":"2020-04-08 09:55:35.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"date":"2020-04-08 09:55:35.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 10:22:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-17 09:15:18.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-04-07 16:57:34.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-04-08 09:43:55.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"reviewer":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-04-08 09:55:35.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-04-08 14:30:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-08 21:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3808cd4dc4db277a9b05d67c8f79bd4df181b1b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-11 18:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cf18c0c842f561c33946b8e264845be1dd1fed3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-11 18:31:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/23/707223/1/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9e84a0a5198460864abd8efd58118e1ff12751fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 02:09:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b889f7e180640d79c3f50d501f35765 : FAILURE in 1h 21m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a48acaff0d0147ecbd595c61c499c93b : SUCCESS in 1h 05m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/364c74c1754d4d91b9c7cc2b09eb02e6 : FAILURE in 1h 18m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdbf5018d0ec4a38853adcb4fdca49c2 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e25569ca76f847759744d14593904582 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/882ba9b254b249b1bd8678ccd2eda95f : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587df00830b0407b8f8a7dd74fd0222e : SUCCESS in 12m 30s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0b7008abe1794a93adcc660c13323367 : TIMED_OUT in 2h 07m 57s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d603b7d2f6964a4b8dafc1a6d04c72f3 : FAILURE in 1h 19m 54s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/fc6e618a24b84d4eb1e89bcf09b54154 : FAILURE in 1h 24m 07s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/13bebe9f7abf4867a4e48d021d6c76bf : FAILURE in 1h 19m 05s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/e4f7aa365d624cbb8725e2ac638115ac : FAILURE in 1h 10m 00s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a0fab0ec14784c379d5e2f041023edd9 : FAILURE in 1h 05m 07s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/47e824b1ae7844b6a02dd50e08b525d8 : FAILURE in 12m 12s (non-voting)\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/65bb0f13b57f4304b9867eb058b5cda9 : SUCCESS in 1h 27m 51s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb8835da41104211b27b06a67bdd3371 : FAILURE in 1h 21m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/135d41ba53844993a4877f35fcb70cf4 : FAILURE in 2h 13m 08s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1e949e0fecf74ba59b22b799df6628f6 : SUCCESS in 54m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6400939f1334576963221b0667fb7fd : SUCCESS in 1h 14m 19s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/9569e2ce9e1c46729d0f9a8d83d48e28 : SUCCESS in 2h 30m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb9004edc55841598d7088774629283f : SUCCESS in 2h 36m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/65f7cd4ae5a24abcbff044ae5632450e : FAILURE in 1h 01m 34s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a53d2cc7a0384c8cbcac6ceb78bffddf : SUCCESS in 1h 16m 00s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6664a90b4264111bbf6bd47c0b71d30 : FAILURE in 1h 03m 43s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ef2feab032164501aba4c2a7907154cf : FAILURE in 28m 03s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/76e4f19f30d04284aa44e73ea5e75a06 : FAILURE in 1h 01m 34s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e91586dbeff84eab85ab55619a2f17ab : SUCCESS in 5m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b035edf39f174e8198929033fe902c4e : FAILURE in 58m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c6260c91c0b8454cb2e6e233d658eb13 : SUCCESS in 1h 16m 52s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e23dfed2da2943eb82c113c4f72fd546 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9dc2fb5823d3fa6d20c6df5ee1261bf99596d4b3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-12 10:01:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d7ca8ed676e588d941243118446d769871822de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-02-12 10:22:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/23/707223/2/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3ae859c382404beec91feb0a9ff0fbf00adc1dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:40:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e56e3c5268b6404095964d68d96caf7d : SUCCESS in 1h 18m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3592a6c86226419cabcfb06906649889 : SUCCESS in 1h 00m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61b59e2023c5481680a538b881a33100 : SUCCESS in 1h 22m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1265b7f91f214ac998e18cb50db20d5b : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c4b6ee4212446abfd53f3ecee4ef3b : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47f738139c1540779438e278c8d06689 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bffb2a454f34ebd8cd80abf075c7a07 : SUCCESS in 11m 24s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/279044666d494fc18f47452973d61735 : SUCCESS in 1h 51m 16s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/319e904d70bc48a9a9fc4bd3882b4ee3 : SUCCESS in 1h 19m 37s (non-voting)\n- tempest-full-train https://zuul.opendev.org/t/openstack/build/014448b18cb549738f03a0e548eb8bd7 : TIMED_OUT in 2h 08m 12s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/60bc1bbde5a74285a48d8a5db30cbb3a : FAILURE in 1h 12m 02s\n- tempest-full-stein https://zuul.opendev.org/t/openstack/build/af2a42a1da49466a90b70da55cdfbd2e : SUCCESS in 1h 42m 59s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/04902215dd664d449cfeab9132f7f23f : SUCCESS in 1h 20m 54s\n- tempest-full-rocky https://zuul.opendev.org/t/openstack/build/6958281ae92e4ce5a2e35d12dd3628ed : SUCCESS in 1h 37m 20s (non-voting)\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/2b4a0ab74a2f40988082eb59d5a14340 : SUCCESS in 1h 32m 37s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b8cecc240dcf40a094687e916f6ad830 : SUCCESS in 1h 38m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/20b24905c65d4b67b35ff4c8025d84fb : SUCCESS in 1h 47m 17s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f8537892ed884c69926610dff124d179 : SUCCESS in 48m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee06514a426f4697a1e77caff65d8462 : SUCCESS in 1h 00m 39s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/0051b1c29348498884fdaee2b14a3300 : SUCCESS in 2h 28m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0544e5557bf94202a3dfb7b63dde72ee : SUCCESS in 2h 14m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/10e722db9ed543b98de7746a2733f1d8 : SUCCESS in 50m 45s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/632f512840634a07b88d4ebc8de55c21 : SUCCESS in 1h 04m 34s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58b946c415924ad2afa20d4ca8494761 : SUCCESS in 1h 03m 07s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f0739f5cd341440bb4032ba40217975a : FAILURE in 26m 32s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c02a818f7f4b4a7bb23af0e7b14ebf6b : SUCCESS in 1h 07m 29s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/896056295ee54330aa585e0ff0998d8c : SUCCESS in 4m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f04ae52cb5294895b82ba583e1f7b4ad : FAILURE in 59m 24s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/70f8c5de8329414ca62151e004a8db10 : SUCCESS in 1h 07m 17s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5c44baf619db4f52964050f10ddd8c7e : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"64916ddb1d111e7f584aaf102059e928ff34ec62","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-02-17 09:15:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"01b73f43262f158962e97bf2c123bb214e16aad8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-18 15:13:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"33fa79346925d4a9f6d38f4fabd57a45c4c7c93f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 23:37:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/70c8f06279434fda8b89a644f510b1c1 : SUCCESS in 54m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b1010d671154104846823c9ba59f058 : SUCCESS in 1h 20m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62b739992c4f49968a3ed778410877f9 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ef022ae93a24c129aa1195c1ae490cf : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12e3edc25da943b78eff57a69e828e2b : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b3a6fca21004fd2b15dc0b3d149fbcd : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbb03da0d35c4864bc99a209b28eaef0 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e39f5f688134ea68e9806db243848cc : SUCCESS in 12m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d3e2c51f798e4bb68234dd44fca38b5d : SUCCESS in 1h 46m 03s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b3004adf25454b73beb7d414dea129af : SUCCESS in 1h 26m 22s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/99ca8bf888f248adbd82f65f5ff8a0f2 : SUCCESS in 1h 18m 15s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/3c3bbe4043cf4cf4accef0a7cc09d4d6 : SUCCESS in 1h 09m 06s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/612968f46a194062a2796d7f2b092d06 : SUCCESS in 1h 43m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e6c8c0f251f04c08850138d53daade1e : SUCCESS in 1h 55m 13s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f4fd526fe0ea456eb0fb1baa55149e81 : SUCCESS in 51m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd01a3ed807b4f25bdf5753cf1f0feb7 : SUCCESS in 1h 07m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b911776189bd452fbd74f2d8bfd0382b : SUCCESS in 2h 24m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1e157675b5c24c42b586e2b5e5492262 : SUCCESS in 56m 07s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc4eed909e9242e9aad665c2f428457f : SUCCESS in 1h 15m 03s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40641ddf64a643d684c318fd5f2de158 : FAILURE in 20m 56s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9340a5b8cac5447785224d68713cd1e4 : FAILURE in 29m 16s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bf11d068875348cfb6bcb4b47a0a5660 : SUCCESS in 1h 18m 27s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1c16a21679004c82aa803af0ec96af62 : SUCCESS in 5m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1408d6ab7628484ba15f8d2df205eef1 : SUCCESS in 1h 22m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/68d7fedf3a18422e889ff624b7ef9f7b : SUCCESS in 1h 07m 37s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d0f7cbd9d71d46a98aba0e6383192315 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4faa3094cda1b46a8f83dee97bbdd3076b76339c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-04-07 16:57:34.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"4ae3bc8b2eac9a00f7f4fbb7c770ece511fe4d86","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-04-08 09:43:55.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, from tempest point of view nothing has changed, only fake specs were replaced by valid ones","accounts_in_message":[],"_revision_number":2},{"id":"99966cdf4b9e2783b15ff29e64a8f870c9cd8a29","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-08 14:30:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis lgtm to move to right things. old acceptable values and what tempest pass are not something we want to test in Tempest for old release/microversion. If that is a valid feature or sometinng then we could test old and new both.","accounts_in_message":[],"_revision_number":2},{"id":"e0ddc81ddf9f34cdc1a845e26183f787964a7651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:02:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be255f899015709bad228aa11b04bca772d73ccc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:15:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"284eb522e532dc591886b34fbd1b55d61f8f6cd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:15:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a475c114094b498bb15bb2b7fa97cd0a : SUCCESS in 1h 09m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94a5927e5b2a43dbb4b2c0bb99eb2e39 : SUCCESS in 1h 27m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffca095829046939579354b08a0a26c : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5da61b458f56406d991022204e6e193a : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f0122ba045742dc9c93334812afb96e : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1df8f8c9804b24a7e3360da5f71f00 : SUCCESS in 12m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/436b1c27341a4801afdbcab2faead1a1 : SUCCESS in 2h 18m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98d42be6adef4d6486982efd8767b3c3 : SUCCESS in 1h 22m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11a38baa95e84d2190391e08860015a5 : SUCCESS in 1h 06m 30s\n\nWarning:\n  Comments left for invalid file tempest/lib/services/clients.py","accounts_in_message":[],"_revision_number":2},{"id":"cb1ae9141496df20b15b8558510395c4a589afe8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:19:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54107c8a96ec4ae68ca98bd690d21974 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e30375e174ed29ee614a91b6aa6d0341067dcae2","revisions":{"7b4b345adf9a6c40b347f7489cb4b3a8f35751d5":{"kind":"REWORK","_number":1,"created":"2020-02-11 18:10:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/707223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/23/707223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/707223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/707223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/707223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/23/707223/1"}}},"commit":{"parents":[{"commit":"fbde5a394842746cb4e78504cfb0a14252af38d0","subject":"Merge \"waiters: Add wait_for_volume_attachment_remove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fbde5a394842746cb4e78504cfb0a14252af38d0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-11 18:06:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-11 18:10:39.000000000","tz":0},"subject":"compute: Use valid extra specs to test \u0027os-extra_specs\u0027 API","message":"compute: Use valid extra specs to test \u0027os-extra_specs\u0027 API\n\nIn Ussuri, nova is hoping to implement support for extra spec\nvalidation. If or when this is completed, only extra specs that are\nrecognised will be permitted by default. We\u0027re currently using a number\nof arbitrary extra specs to validate the \u0027os-extra_specs\u0027 API meaning we\nwill fall afoul of this validation. Head the issue off now by using\nrecognized extra specs, namely \u0027hw:numa_nodes\u0027 and \u0027hw:cpu_policy\u0027,\ninstead.\n\nChange-Id: I95a1050d6a5e2e40793ae7f18ff0d326455b4b36\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7b4b345adf9a6c40b347f7489cb4b3a8f35751d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7b4b345adf9a6c40b347f7489cb4b3a8f35751d5"}]},"branch":"refs/heads/master"},"e30375e174ed29ee614a91b6aa6d0341067dcae2":{"kind":"REWORK","_number":2,"created":"2020-02-12 10:01:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/707223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/23/707223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/707223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/707223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/707223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/23/707223/2"}}},"commit":{"parents":[{"commit":"fbde5a394842746cb4e78504cfb0a14252af38d0","subject":"Merge \"waiters: Add wait_for_volume_attachment_remove\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fbde5a394842746cb4e78504cfb0a14252af38d0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-11 18:06:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-12 10:00:43.000000000","tz":0},"subject":"compute: Use valid extra specs to test \u0027os-extra_specs\u0027 API","message":"compute: Use valid extra specs to test \u0027os-extra_specs\u0027 API\n\nIn Ussuri, nova is hoping to implement support for extra spec\nvalidation. If or when this is completed, only extra specs that are\nrecognised will be permitted by default. We\u0027re currently using a number\nof arbitrary extra specs to validate the \u0027os-extra_specs\u0027 API meaning we\nwill fall afoul of this validation. Head the issue off now by using\nrecognized extra specs, namely \u0027hw:numa_nodes\u0027 and \u0027hw:cpu_policy\u0027,\ninstead.\n\nChange-Id: I95a1050d6a5e2e40793ae7f18ff0d326455b4b36\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e30375e174ed29ee614a91b6aa6d0341067dcae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e30375e174ed29ee614a91b6aa6d0341067dcae2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
