)]}'
{"id":"openstack%2Ftempest~709961","triplet_id":"openstack%2Ftempest~master~I486049a2994d2b16e7ffdcf9afb3b759e685465c","project":"openstack/tempest","branch":"master","topic":"fix14","hashtags":[],"change_id":"I486049a2994d2b16e7ffdcf9afb3b759e685465c","subject":"Add image task api","status":"ABANDONED","created":"2020-02-26 07:39:05.000000000","updated":"2022-12-24 05:46:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d960c257ddc9bba85f907dbca79d9a4de66f57db","_number":709961,"virtual_id_number":709961,"owner":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-12 05:21:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 05:21:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d618c1c12178718f06daa52593dfbdc8c001d79e","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-02-26 07:39:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22cc966e4c4525aed9cea07cac241c543bca548e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:06:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/941dc481c53d415e863ec192ed4f632a : SUCCESS in 1h 01m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fa03df5c346481191b70e67dcfacf61 : SUCCESS in 1h 37m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbf9e8ba8892436dad74230bd4fbadc1 : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d0760a2dd4f4d49b05510d511593d04 : FAILURE in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0de922bef9f45b9bc0fb845cd08dc1a : FAILURE in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c8994c4824c41ff8def6665ae068623 : FAILURE in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb14f97efe2a4cf4add18dc14875b612 : FAILURE in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750d56d63e674b0b814cc8c22835830e : SUCCESS in 14m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/515f3cdddbd6430e86ae211b5c942fca : SUCCESS in 10m 04s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/838f03299d9f441da493004b53a67b8c : TIMED_OUT in 2h 07m 34s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/108a866d24e547f7a4ac21bf33536d4f : SUCCESS in 1h 30m 45s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3053424e458e45669050c4f065cdfcf2 : SUCCESS in 1h 33m 41s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/2ad77edbb4e645e097a37a091b7ed8d5 : SUCCESS in 1h 45m 31s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/26808d9095e04f50a8532e2f0ade0dde : SUCCESS in 1h 34m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9d777808bb4e453080d1b451af435d0b : SUCCESS in 1h 49m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b476c9816f7842648626d1906181c87e : SUCCESS in 43m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/799283e91e74405e9d368cf19860f8b2 : SUCCESS in 1h 10m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be37e643cb5b4fbaa5efea9b63b0f81f : SUCCESS in 2h 24m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/077bd7ede28747c6b5c24ca4130c838e : SUCCESS in 1h 01m 34s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ce4b93df6d44ea28390cc6678f92e7e : FAILURE in 1h 00m 37s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/235d077b5f084989b2982719f6bf65da : FAILURE in 13m 09s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d2f57bbc3d8d4755adc28b2f7ce411ca : FAILURE in 39m 22s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d21858c788514a5195db2db97ade203e : SUCCESS in 1h 06m 33s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/dfcb276dc1424addaf200044765b0b10 : SUCCESS in 4m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/da21d17c91094b778c433968beca53b9 : SUCCESS in 1h 26m 28s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/676f62b65c8a4ef389a1158be22a5fb1 : SUCCESS in 1h 13m 48s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/03231e3c5e8a46ec898a3119a30436ec : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"56e7f6bd7aedac3cb4dfab1f4fc37be720846f34","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-02-27 01:19:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e2f6ba908b8b231c8e7852b26c4057b569fad50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 03:54:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cbfb19e912054c15b546f0dc5e16a32f : SUCCESS in 1h 00m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/230df68ea6f940ecbc6f63b79ae29f05 : SUCCESS in 1h 17m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6fec9cd51ae4e60be842ab31fe471ed : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5decd533ec4e4df8908ff8c9fa8e7455 : FAILURE in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7440582ca36c4f53abf7d1128b26b009 : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/039e460ef33547c8ba5c4da7dadead2b : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19f6bba1d56a45d38b3eef1849acc700 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dba88fc320d41a58eded4b474653e6c : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/246893d664f2478a87aa56604edcfc46 : SUCCESS in 12m 04s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/953d8511a057434ea9343175810b9f8e : TIMED_OUT in 2h 05m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e23f42cfb42b4340a2d035e42cca4212 : SUCCESS in 1h 17m 55s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/222fbcd03ab44cdea4f82c9d87277e6b : SUCCESS in 1h 37m 38s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/bbede7a2d95a4bdfaf2f9138bd4f9a07 : SUCCESS in 1h 23m 09s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/0c165cee8d9a4b50b57ded9cbe3e449f : SUCCESS in 1h 26m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/860671f52ed740b6a9a6fc1e128586fb : SUCCESS in 1h 34m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a801036bdae2407a804753aa7275d310 : SUCCESS in 50m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00a5259587b24f99896d01a94ecf7591 : SUCCESS in 1h 04m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3d3c363df1b48a9b372fe13a082da4c : SUCCESS in 2h 32m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c7bdb7ca33f54b3684502eff66c3da96 : SUCCESS in 53m 57s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f8fde648cc204e24bf9a0eaf47d1f5d2 : SUCCESS in 1h 08m 22s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6df6d354a6cd460eba9086204103d2a1 : FAILURE in 15m 08s (non-voting)\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/89f8a58da10a46e5a0b7885af7a699ac : FAILURE in 25m 20s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/096b6cf7bdbd49bb81babd1aad592541 : SUCCESS in 1h 07m 45s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a882473940894cec838883f450ffe8b2 : SUCCESS in 6m 05s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0e3737ed6a2443f4a83211e25f09eb7a : SUCCESS in 1h 34m 34s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/dc823d2efc654582a677c7fc1fe0ead1 : SUCCESS in 1h 11m 24s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/78ea138daffa410197d3a731de082ae7 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8675b737ff93eaa4d71aa750f62685d366852981","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-03-19 06:46:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6f4a009cea639d46aa7cd845febc00e9fdc9c992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 09:03:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5da55d2935df46b29793f34c3d5a1041 : SUCCESS in 1h 03m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bacd11a3f4fe4c87ab919b469d90ffc6 : SUCCESS in 1h 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5f3e2c97bca4bd49f3cecd12ff2b19a : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6050651f0bf4eff8b6e93b4bcdf4d3f : FAILURE in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79f2c655eb5b4714a641bd02c2396b3b : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16ab60785c564add996b8303e5a0650c : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b3994867201479e8a7c5836c7b37408 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14df3b3266a14dbea8195b06e667fe80 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222154986a63466eab357b38048396ba : SUCCESS in 10m 50s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/121484be84fb4c6fb10667fef8afe967 : SUCCESS in 1h 57m 58s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/72f9a8c2997247f1b99600c742308eda : SUCCESS in 1h 23m 10s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a9a765916745473fbf7ffca8ce498d78 : SUCCESS in 1h 24m 03s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/625740155a034dc49b1fada2c8650714 : SUCCESS in 1h 32m 14s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/52f3655287c5462988f2c60dcb24588d : SUCCESS in 1h 28m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/90229e7b16e74ab0a36dbfc970dcfe14 : SUCCESS in 1h 39m 32s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/aa1e774b09ab49efaee978e4b60b70a6 : SUCCESS in 46m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/770c858d13944fc5b8c863d8f5faf4ec : SUCCESS in 1h 06m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/072dc25090164750950fb1427c0c1708 : FAILURE in 2h 14m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/676bc361e03e44309cf942ebb733ee4c : SUCCESS in 51m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c9c08a1038344a9580d562c6322f3f9c : SUCCESS in 1h 03m 47s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1340671f54b7488ba80bff816bed0820 : SUCCESS in 1h 06m 36s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5fc5e306d61a41bfabfebb7aad46f4ed : FAILURE in 28m 19s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/92699c5dcddb41b2a7f9144c9f37ba28 : SUCCESS in 1h 11m 19s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/341287d60cea413dab5901c0bd9876c4 : SUCCESS in 3m 08s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0225471be852494fbec6acd5454d926a : SUCCESS in 1h 20m 19s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/297d3c5efae44ed794e426ff42b073d1 : SUCCESS in 1h 15m 38s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bc54117ba6224434985f931170ffa296 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b8b5d48a51ab70be1de9d7e4337531d10bdcd33c","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-03-19 09:23:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"297f801df141c2bdcf151bf830d85cf33a22a7f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:10:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/119799ebb2314737baece02584eb406f : SUCCESS in 1h 13m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c748b0c98e41494f90965a5d3044c9d0 : SUCCESS in 1h 38m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2163356c7364de29e4ee6eb263cdd1f : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ecfdbb88c984a95a8661d0ee02b6c90 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23fc55ddffa744e08851de6d3def4b1f : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1b424da0ee64e9892c67dd3e7ac7436 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdffdab683b144a387791c637df00c06 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c3096dcb1cf4ee2b37a35cd9b90e08f : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57ef0095a7c742479fb41c8ea4494da0 : SUCCESS in 11m 56s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/fc48ee54708248d495012879a77e63c8 : SUCCESS in 1h 57m 24s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7960207e628e41099a45198f557581bf : SUCCESS in 1h 22m 15s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/994565d6cde946b897abe9b6ed60264b : SUCCESS in 1h 32m 21s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/720937a9fd2c4fa8aa4cb1a6abef05e6 : SUCCESS in 1h 31m 28s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/61411f8bf5b44eca80e5644d340c6fcb : SUCCESS in 1h 16m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5795263635e74de9ab1d9b0349fbe126 : FAILURE in 1h 38m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/eef8060750e04d138b347b3c2be0e7b7 : SUCCESS in 58m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f3d88635d84606bd5dbafc7d68c889 : SUCCESS in 1h 13m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3438144e13934263a044624c8fdcce48 : SUCCESS in 2h 43m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2d6458273848452ba58ae7ad59a97c73 : SUCCESS in 48m 42s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/42facd07a03c4d6f9198fee6d867c868 : SUCCESS in 1h 02m 47s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/199fd5044017498e9dd96eef731f16e1 : FAILURE in 16m 52s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7e95c66faff746ad89e073f9f6c64a78 : FAILURE in 7m 06s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5da932c04d6d4e73be60eaa16e347f88 : FAILURE in 1h 10m 42s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8115ed1bb16e49868f7e5123f24fd922 : SUCCESS in 3m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7c8e52b2f18045559b92624261cb0820 : SUCCESS in 1h 38m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fb43dc5957b34b31a8cf4236be822ddb : SUCCESS in 58m 58s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/efffcaa7f7c344598f29cf46cb1e5d9b : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":4},{"id":"19fa234a6929f5d7f204ee7815eb625f12563282","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-10-12 02:59:40.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"98829df30cf50e983524334a1ed59fd605a68538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 02:59:53.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"f6c8e925f187d3ca078143485373dcd574c41cac","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-10-12 03:18:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"87208fc779eab2ebbd55ece61b47852bbc657f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 05:21:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/92efe4f3851e45c7a61b5771708ef4d1 : SUCCESS in 1h 01m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9b050117a8d4e58bd229b50165afd34 : SUCCESS in 1h 22m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ada87365caa948d6ae74f0c78028ee28 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad22225d75d40baa2707d8e47457ca7 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1e96e414d4f402792e37bb90bb86ea9 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fb2434ed1a34f2693c836f1ec0a06e3 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93758d2ad2bc4fad8fe98980ea220ac5 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ab7e17e25ee454c803408cc4ebbd8e2 : SUCCESS in 4m 37s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/8853903da3c4457899e0dd57a1f639a1 : SUCCESS in 1h 53m 25s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/4a1cbdcebe91465096bdc8947fd5bc91 : SUCCESS in 1h 22m 37s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/871f1d7c9de341d38ae1156bac2ef955 : SUCCESS in 1h 09m 06s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/231e4e8759a542a3868813350b5d22b4 : SUCCESS in 1h 30m 57s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/59ab3efc4ebb4d449ada5e60ac34ed14 : SUCCESS in 1h 17m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5480f7d9d71e448888253b19170a06af : SUCCESS in 1h 35m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c1b783bcad374a729bafeff9bf868716 : SUCCESS in 46m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87559199b3aa4cb1a2ed25ce6e7a3935 : SUCCESS in 59m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/996f3f95ffea4987aa257299cd2ff41e : SUCCESS in 1h 56m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7227051e38c440f6a7f823d2179359c2 : SUCCESS in 54m 24s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/075b36ca44b84a96984950b5b36e7c7b : SUCCESS in 1h 07m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a871b56ca674c7082d201b5bb425d55 : SUCCESS in 1h 01m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bb2a490c55984af8ad2008e60b499605 : SUCCESS in 1h 12m 37s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e6aee530810b4484925d63c97621a288 : SUCCESS in 2m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/299d7075c1d84d02b476ebf7a88c4bfe : SUCCESS in 1h 19m 49s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c0c04280ce684076a1ec07e51ddef346 : SUCCESS in 53m 38s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5ed00641c6734f80a01c9c349fb9b67f : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"d960c257ddc9bba85f907dbca79d9a4de66f57db","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-24 05:46:25.000000000","message":"Abandoned\n\nNo activities on this since long time, feel free to restore it if needed.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"521cf39e0ad435051e748b28b0f7c9fa9cb60b0b","revisions":{"9255002917c482d22ce3cd84827aac152b5f9baa":{"kind":"REWORK","_number":1,"created":"2020-02-26 07:39:05.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/61/709961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/709961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/709961/1"}}},"commit":{"parents":[{"commit":"0e52b485e9554d279adf30c5d906ec12df3bc327","subject":"Merge \"add unit test for delete_group_user method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e52b485e9554d279adf30c5d906ec12df3bc327"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-02-26 07:31:26.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-02-26 07:35:50.000000000","tz":480},"subject":"Add image task api","message":"Add image task api\n\nAdd image task api and unittest to image part.\n\nChange-Id: I486049a2994d2b16e7ffdcf9afb3b759e685465c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9255002917c482d22ce3cd84827aac152b5f9baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9255002917c482d22ce3cd84827aac152b5f9baa"}]},"branch":"refs/heads/master"},"25d1f300bc8e2b2a9fadf437e4f25082f36564f0":{"kind":"REWORK","_number":2,"created":"2020-02-27 01:19:27.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/61/709961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/709961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/709961/2"}}},"commit":{"parents":[{"commit":"0e52b485e9554d279adf30c5d906ec12df3bc327","subject":"Merge \"add unit test for delete_group_user method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e52b485e9554d279adf30c5d906ec12df3bc327"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-02-26 07:31:26.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-02-27 01:17:22.000000000","tz":480},"subject":"Add image task api","message":"Add image task api\n\nAdd image task api and unittest to image part.\n\nChange-Id: I486049a2994d2b16e7ffdcf9afb3b759e685465c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/25d1f300bc8e2b2a9fadf437e4f25082f36564f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/25d1f300bc8e2b2a9fadf437e4f25082f36564f0"}]},"branch":"refs/heads/master"},"bee83e68d9c0276fa83423642968f96c32761b3e":{"kind":"REWORK","_number":3,"created":"2020-03-19 06:46:20.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/61/709961/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/709961/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/709961/3"}}},"commit":{"parents":[{"commit":"0e52b485e9554d279adf30c5d906ec12df3bc327","subject":"Merge \"add unit test for delete_group_user method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e52b485e9554d279adf30c5d906ec12df3bc327"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-02-26 07:31:26.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-03-19 06:44:14.000000000","tz":480},"subject":"Add image task api","message":"Add image task api\n\nAdd image task api and unittest to image part.\n\nChange-Id: I486049a2994d2b16e7ffdcf9afb3b759e685465c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bee83e68d9c0276fa83423642968f96c32761b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bee83e68d9c0276fa83423642968f96c32761b3e"}]},"branch":"refs/heads/master"},"e39e3b1475cc8ac670560ffb7ed6520db1e72617":{"kind":"REWORK","_number":4,"created":"2020-03-19 09:23:18.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/61/709961/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/709961/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/709961/4"}}},"commit":{"parents":[{"commit":"0e52b485e9554d279adf30c5d906ec12df3bc327","subject":"Merge \"add unit test for delete_group_user method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e52b485e9554d279adf30c5d906ec12df3bc327"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-02-26 07:31:26.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-03-19 09:20:23.000000000","tz":480},"subject":"Add image task api","message":"Add image task api\n\nAdd image task api and unittest to image part.\n\nChange-Id: I486049a2994d2b16e7ffdcf9afb3b759e685465c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e39e3b1475cc8ac670560ffb7ed6520db1e72617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e39e3b1475cc8ac670560ffb7ed6520db1e72617"}]},"branch":"refs/heads/master"},"521cf39e0ad435051e748b28b0f7c9fa9cb60b0b":{"kind":"REWORK","_number":5,"created":"2020-10-12 03:18:00.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/61/709961/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/709961/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/709961/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/709961/5"}}},"commit":{"parents":[{"commit":"9ddee691f3f44f462a88253e20352403def53a60","subject":"Merge \"Add missing replication_status in group schema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ddee691f3f44f462a88253e20352403def53a60"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-02-26 07:31:26.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-10-12 03:16:11.000000000","tz":480},"subject":"Add image task api","message":"Add image task api\n\nAdd image task api and unittest to image part.\n\nChange-Id: I486049a2994d2b16e7ffdcf9afb3b759e685465c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/521cf39e0ad435051e748b28b0f7c9fa9cb60b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/521cf39e0ad435051e748b28b0f7c9fa9cb60b0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
