)]}'
{"id":"openstack%2Ftempest~713144","triplet_id":"openstack%2Ftempest~master~Ibec72b578424cb278006ac810bdd0359ec23de3f","project":"openstack/tempest","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Ibec72b578424cb278006ac810bdd0359ec23de3f","subject":"Make tempest-multinode-full py2 job on py2","status":"MERGED","created":"2020-03-15 19:46:37.000000000","updated":"2020-03-24 14:00:44.000000000","submitted":"2020-03-24 13:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713144-1585058117582-6b47a46b","meta_rev_id":"400a1be7678d5ccb7c52a7295be3e539f35783b0","_number":713144,"virtual_id_number":713144,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 13:55:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2020-03-24 00:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2020-03-24 11:50:59.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2020-03-24 11:50:59.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 00:46:55.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-03-24 11:50:59.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-03-24 13:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7891261d0da7b384824e8c61830cf95a6b7ccc9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-15 19:46:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3efc4fc632e9d5ef9905da1ea0418ad92e27e393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-15 22:03:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5e955ffaeb0941878c54a1485387bcc8 : SUCCESS in 1h 06m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bbb2b15d5114069aeff5fc433d0b019 : SUCCESS in 1h 15m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/014e7a32563e49e5a5cad00c352aeaa9 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/127d0be6ffc1417684cd6296dbf99f6f : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0812e7d8c784c3ea37bd5824ab7a294 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc37daa44c304700a81a12fe2ed1360a : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/872818a300d5449487bf27cbe471dbca : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f818d70c2b8f4192973d4bcef580d762 : SUCCESS in 8m 56s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/019d2ce8331d4b529d05878659e13913 : SUCCESS in 39m 34s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/1365dde37a2f4fec94fd5337aff4137f : SUCCESS in 40m 01s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a0f9e06bb5c048bc8cc3fd8c64a1970c : SUCCESS in 1h 52m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3127301297a3460f8f99627eecf8c5bb : SUCCESS in 1h 24m 46s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ab95a827b2c54564b461dbda9950024b : SUCCESS in 1h 18m 42s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9bf1a531e9d9430cba05c8ea1345e5c8 : SUCCESS in 1h 21m 36s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/d895e97e389540cb8665a2c5efa41808 : SUCCESS in 1h 17m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/611e7a22f5fc457488751c20788f2581 : SUCCESS in 1h 41m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ba66a8869a4141cabf28d4e7a59b22e7 : SUCCESS in 48m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9d4b1e0ab32445f99ebffec404a30d6 : SUCCESS in 1h 04m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e65e8f86d2f43fc95b4736268a0e8ed : SUCCESS in 2h 15m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b447ff122b8e422893f3e717ed1c2347 : SUCCESS in 53m 13s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d63872aabf02428294a733b989208213 : SUCCESS in 58m 43s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e599e74a3284e14af28071fe6a2468b : SUCCESS in 1h 10m 33s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/01a01e430a9a4eadba8fd56cd30e11ad : FAILURE in 25m 32s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0ec481aec4a748b088bdb52dac5dd616 : SUCCESS in 1h 14m 14s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3ae279d4ebf3462f82b5946326e8fcbe : SUCCESS in 3m 27s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/da39cc440e864de880550e697b620371 : SUCCESS in 1h 16m 00s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6864e0691e044a38985b4e96d6fef5c8 : SUCCESS in 58m 01s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/44344abac16541b49cd3319ded84937a : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"af964b25290995c1eb8f063e518626d149562f07","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-03-24 00:46:55.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks, make sense","accounts_in_message":[],"_revision_number":1},{"id":"72e21793484cbe998853416cbd863794b97eebb1","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2020-03-24 11:50:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10d4e8c9e1a2bc235fd6327c085a8cdf11c48e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 11:51:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85b6b34adf860643a29f875071bea0c63501eb20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 13:55:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5588e4b3976772c03c841c3fadb4667ff5506dff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 13:55:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2a4315d6d579492bb1e18631743b8ae3 : SUCCESS in 1h 14m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f55f94a400444c839a3a9f32b145f743 : SUCCESS in 1h 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7841169885d64a9ba59708c48edafa9d : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4badf052e0e4d218bb4afef1d6dde10 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d569b13819041d89582045b1d13d5dd : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbac2d5fb4248cc8acfa08a2fc20a34 : SUCCESS in 13m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/121b68851bad41cb8d92391070fc3c86 : SUCCESS in 1h 47m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/01e823ead5d7499e8fffc9ef38ec42fa : SUCCESS in 1h 02m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/624c10c331e04d7e8de8f64938d7fc54 : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1ae23f4066dcfc51805fd94ddaa3fdd04703e567","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:00:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d30fefc4db043baa1189ae94cb7b636 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fb306f2e0e48f054ffad30b058a4c99c0cf9371","revisions":{"0fb306f2e0e48f054ffad30b058a4c99c0cf9371":{"kind":"REWORK","_number":1,"created":"2020-03-15 19:46:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/713144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/44/713144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/44/713144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/44/713144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/44/713144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/44/713144/1"}}},"commit":{"parents":[{"commit":"714c607a3623b6693f843544449a640a8ae6d85a","subject":"Merge \"Prevent SSH client proxy loops\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/714c607a3623b6693f843544449a640a8ae6d85a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-15 19:43:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-15 19:43:44.000000000","tz":-300},"subject":"Make tempest-multinode-full py2 job on py2","message":"Make tempest-multinode-full py2 job on py2\n\ntempest-multinode-full is python2 version\nof mutlinode testing for stable branches and\nsupposed to run on python 2 but it was left during\n- I75868d5c9b6630fe78958ff89e58a0aced09a6b3\n\nChange-Id: Ibec72b578424cb278006ac810bdd0359ec23de3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0fb306f2e0e48f054ffad30b058a4c99c0cf9371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0fb306f2e0e48f054ffad30b058a4c99c0cf9371"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
