)]}'
{"id":"openstack%2Ftempest~713154","triplet_id":"openstack%2Ftempest~master~Id8d315d29d9931fd71c9b50d1644629f5f550169","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"Id8d315d29d9931fd71c9b50d1644629f5f550169","subject":"raise an empty string no output command on ssh exec","status":"ABANDONED","created":"2020-03-16 01:16:44.000000000","updated":"2020-09-21 19:10:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4be1ea79f6747d9d0b8246ff87e3a159b7f647f3","_number":713154,"virtual_id_number":713154,"owner":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-16 03:26:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"all":[{"value":-1,"date":"2020-04-13 15:30:55.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-16 03:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fb6cbae0400f8255aacbd02e960a9bf640b8703","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-03-16 01:16:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f98042d2b83e9c8985eea46d5f09941c16e409d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 03:26:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b17d349a239546d58d9fc1c97bf2aaec : FAILURE in 38m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6183a52d79b54684bb78633a97c78c67 : FAILURE in 1h 01m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f87e86b1dbc1434ebe712d9865a06ea1 : FAILURE in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e15d05f8f44f2ea0550c10e4b5b029 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27faa6baf99a4a3bad3070e4ebd989a1 : FAILURE in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a098bb22956c4097a15b13b69e0012df : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a2e3df1b9a54d319951a547571f229f : FAILURE in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db723d2531146a1ab57502edadf66d4 : SUCCESS in 13m 39s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/8fd5f88080df43db8a3fae458df31250 : FAILURE in 1h 44m 24s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c7be79bdce754b608f7c4f1300355580 : FAILURE in 1h 09m 51s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/486b6701c52a4efcb307c8708f6fa79b : FAILURE in 1h 03m 55s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/629ce70de30d43d19b909cccc56858e4 : FAILURE in 1h 03m 57s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/1ff05ef90930457db86668b033df1a15 : SUCCESS in 1h 32m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/51f6e29429474cbe9c471d34c87e9053 : FAILURE in 1h 23m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ab2d612500a6470d994ce520e901690e : SUCCESS in 47m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba7e7d77a6c14c13be612d27163fe471 : FAILURE in 1h 01m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbcf7d1cc7894e738e48d9bc8cbd0773 : FAILURE in 2h 03m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e413e201ca40488387902155229e9619 : SUCCESS in 51m 05s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3043e207c70b456db079a9cf5a688f4b : FAILURE in 50m 06s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d19b685e478467a910b3e3ca6ab7ef6 : FAILURE in 57m 52s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/faf1e73df69341e3b153e46593da37d7 : FAILURE in 24m 18s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/807002dbd1c84f51a6f3a34851abe6a1 : FAILURE in 1h 04m 55s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/47e7574161f54d8e8aaae73d50a6f51a : SUCCESS in 2m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/26242a59f395413098a57a72bd01cdd6 : FAILURE in 1h 12m 18s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/29b920ee501b42238ff70525c5e6e79a : FAILURE in 1h 00m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ae4f398fe190409e8445f5c6f7add091 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bf49564b29768a88c7ab38e6beb160656a2554b9","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-04-13 15:30:55.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7f3bf628a7db622d5caddf9ef05ad268859eea24","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-21 19:09:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"4be1ea79f6747d9d0b8246ff87e3a159b7f647f3","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-09-21 19:10:11.000000000","message":"Patch Set 1:\n\nthis seems like invalid request, so hence abandon it","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ad6f4a354c316b99638842b58c1bb4be947ba00","revisions":{"9ad6f4a354c316b99638842b58c1bb4be947ba00":{"kind":"REWORK","_number":1,"created":"2020-03-16 01:16:44.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/54/713154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/54/713154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/713154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/713154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/713154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/54/713154/1"}}},"commit":{"parents":[{"commit":"714c607a3623b6693f843544449a640a8ae6d85a","subject":"Merge \"Prevent SSH client proxy loops\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/714c607a3623b6693f843544449a640a8ae6d85a"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-03-16 01:12:27.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-03-16 01:12:27.000000000","tz":-240},"subject":"raise an empty string no output command on ssh exec","message":"raise an empty string no output command on ssh exec\n\nraise value error exception when ssh command return no\noutput i.e empty string\n\nChange-Id: Id8d315d29d9931fd71c9b50d1644629f5f550169\nCloses-Bug: #1865276\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ad6f4a354c316b99638842b58c1bb4be947ba00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ad6f4a354c316b99638842b58c1bb4be947ba00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
