)]}'
{"id":"openstack%2Ftempest~716311","triplet_id":"openstack%2Ftempest~master~I289a44c5604a12bc8849989954588896c4ff8bd3","project":"openstack/tempest","branch":"master","topic":"locked_reason","hashtags":[],"change_id":"I289a44c5604a12bc8849989954588896c4ff8bd3","subject":"Add compute API locked_reason response for 2.73","status":"ABANDONED","created":"2020-03-31 14:22:31.000000000","updated":"2020-04-24 14:49:26.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f486a3444edea8277c38a5a0279157f45f9296b6","_number":716311,"virtual_id_number":716311,"owner":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"date":"2020-04-06 17:01:17.000000000","_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-06 16:21:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"date":"2020-04-14 08:52:20.000000000","_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 21:13:16.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2020-04-05 00:05:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-05 04:08:20.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-04-06 16:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-14 08:52:20.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"}],"messages":[{"id":"dd45eb351bc7595836f5f6627d105d97e706df13","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-03-31 14:22:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb285aeec87da788b560ce6ef0187f1184fc6ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 17:06:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7550d047cbd84a87a6d817f634cf2d92 : FAILURE in 29m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8d2bf53af894a758de8ef057b0e15e5 : FAILURE in 32m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62fa575829d04f9bba6c49022cc4d2aa : FAILURE in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029749008e2e476c8013d1ffcc9d6acf : FAILURE in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c665db23979049f984a3af6f230b5b45 : FAILURE in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ce359893f3d4e77af636f3a9dc24ddf : FAILURE in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fc57b1c9d9b42c6b2aed25a1e5bf3cd : FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256aa594026e4636a12663332a50675d : FAILURE in 5m 04s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f560f55e11614d4fa9a37c5817ca8e15 : FAILURE in 32m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/2196c95e864e43ba8a7a054d87d7e5d5 : FAILURE in 31m 23s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e083e6aa8fbc4b8588ee55feae2348cf : FAILURE in 32m 09s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/10c3a55159134589824a01d7aaaf5a24 : FAILURE in 26m 35s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/162fa058bba14b07b0c63228a871bd25 : SUCCESS in 1h 43m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/24092cce41d14723a18a7e15a6c9a7a7 : FAILURE in 32m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ed46bf8d07864cb8a0cb01721f43a55a : FAILURE in 53m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1bd8952ef374362aa7a5461a31efcd0 : FAILURE in 28m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e219f59cd7244ef58058efb8967afff9 : FAILURE in 33m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/49f020154e564748858a980781431364 : FAILURE in 37m 20s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3761eb10dd3a47f282648b8c937d707e : FAILURE in 29m 56s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e78121bfbdd410ea3d3b8680c4228ef : FAILURE in 38m 40s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/fca001e0020b452fad8dba2b6013f14c : FAILURE in 38m 37s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7e353d808aa24279b959f3f5c6e884a0 : FAILURE in 29m 38s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/eeaca1f6cfb0407289918655cbf55712 : SUCCESS in 5m 27s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f331089b0e3f46029d618ea72b57c0f4 : FAILURE in 30m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/36e3ec4800a34efbb0669a344c202b54 : FAILURE in 24m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2ec5157d86c64284b63b33b04a534441 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dd78f193302e920fc76f72a0c45eada77c5f2da3","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-03-31 17:33:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d8f4f8d093964b25032b187563e989328f4d65d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-31 17:35:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nthanks for adding. version issue on schema decorator.","accounts_in_message":[],"_revision_number":2},{"id":"550a7e7312501a653d09bf7965908fcfe30fa2b7","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-03-31 17:59:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a773c8da5fbb86b893a34b6bccdb4cf108d899ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:56:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/08d9a4c84cfe45fa847d946b645bfa5c : SUCCESS in 1h 05m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e28009212494cf4a02af1a88c6a2d47 : SUCCESS in 1h 39m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/640482cdb11b4fd09c86c1ca1133fb78 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6cbe026fc843fb954a6dc1059e8f73 : FAILURE in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5af5a90a64d4b70942e74c402ee2c7b : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78b4414e1e9048d4b5ac560a851a22d9 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da656f5ddef241f09539a686328f6f0e : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4edae763ae114d8fb44f907554b234cf : SUCCESS in 11m 26s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b5c77946f92147bc965a8aa663fbbf0f : TIMED_OUT in 2h 07m 10s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/34b819c2593842028912eb9a6300b49e : SUCCESS in 1h 18m 40s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e7994d8783ee4bdbaeaf9cd620a1b849 : SUCCESS in 1h 35m 59s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/cc03fd42adc5404a8c35a13933285d62 : SUCCESS in 1h 34m 48s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/8c109f647588440b874afd3ef2722046 : SUCCESS in 1h 36m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a82af1a2500146aca44e7cc120aef62a : SUCCESS in 1h 39m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/dbca20f8a4114344a506ef5c84fe4d49 : SUCCESS in 49m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/029f91c952904840a5f5d9fb5881fc5c : SUCCESS in 1h 02m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b1fba93360f4dbdb215a1eb6a715e99 : SUCCESS in 2h 24m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2a2a94adc0a744b7acfe7ce0cce81cb4 : SUCCESS in 53m 21s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/442d023eabae4c4aa3be4dac78c429cb : SUCCESS in 1h 05m 08s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54177e4b35844d2483f8f97fda6d1f36 : SUCCESS in 1h 06m 27s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/56171b6b165944a69b153230f877ae8e : FAILURE in 38m 06s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/cff80b9dea7c4c6ea6e43a17e137e034 : SUCCESS in 1h 33m 51s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d331f6a35640408aa0bf310c975c2285 : SUCCESS in 5m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a8221339aa9d49678640e6dc5af7d441 : SUCCESS in 1h 08m 50s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c4de143be31847c8a0a81603cbe1010f : SUCCESS in 54m 43s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/de5e9857763f4372884852093e583b24 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"097a9bedbfcc4846259dc971abe02562660721be","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-04-01 01:51:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b6274f48b008883ec78e12385e7a2b27402c30ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 05:00:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/945f60d95a9b4e878b7ec8088bdcdb18 : SUCCESS in 1h 02m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a19ac6c6d4c4f0aafa977011e0a69d3 : SUCCESS in 1h 26m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cda0c557bd34584bbfcb26c56a07506 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a078e36c24ee44068c9cd9875eb31f06 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/407adc9d8d874557848f99d3f563b9f0 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15bd0d6695de4b028d871cec0b175638 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cc67bc9b70a462a80ca8068c0301c1f : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da81d45ce4ab4b94b091c6ea7ed479e3 : SUCCESS in 10m 18s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/09b68d11ec864753963ab56b19b18925 : TIMED_OUT in 2h 06m 05s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/8065ae3b6c984a31a823b09cbd592ca9 : SUCCESS in 1h 59m 17s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/38620f7898eb4be8802b2413e0cd7493 : TIMED_OUT in 2h 05m 03s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/bbf6bff899f341b9a402ba73f3add661 : SUCCESS in 1h 24m 08s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/3797e429f20a45d2b8bd9c50cf116d01 : SUCCESS in 1h 19m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a1cdb329a412438f9ad75c3dd0147446 : SUCCESS in 1h 40m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9b9f6308d0074776a2a25eb677819f18 : SUCCESS in 47m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e70d29bddb840b381e1f6c1d233221f : SUCCESS in 57m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d921610fa90447b9883097412e78424 : SUCCESS in 2h 53m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/874606218b2e4d928052aff328260ae9 : SUCCESS in 53m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0181ddd4747749cfa06cee3ff610558a : SUCCESS in 1h 09m 03s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/18a0b0ec27754212bc627e389b226cb4 : SUCCESS in 1h 33m 01s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8e355550fdc846e7be60e9401833ee45 : FAILURE in 26m 30s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bf2e61ee785540f79c57bb89bff844be : SUCCESS in 1h 11m 10s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b2ebc5d060024b3da7e5f70bc2b2441a : SUCCESS in 2m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c9da9d2ae8bf4f4eb207c06da010693e : SUCCESS in 1h 55m 57s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3dbb3eb1bbc045fda2ee596868ed8fd7 : SUCCESS in 1h 07m 14s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f9da581e416049ed96e428b0bdc3152e : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":4},{"id":"b7f264e66d29a3c3ea88428946e9919cc479ca7b","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-04-01 12:17:45.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1afe72f421e4e012ac9a1eac8fff95020c842bf8","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-04-03 10:06:41.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"86ef519fdfdeb15dbda7333139d1b9ec85fd689c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 13:13:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8cd3fc2ba5dd406cab17a7fb2d5f9f85 : SUCCESS in 1h 09m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e12e95fd6e484b1885c456fa74af9eeb : SUCCESS in 1h 32m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/685d0798eb9b45f0ab1cca3bb1a18324 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf33e3778d8455fa7e7808244023d18 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9d8038ec21040dfbed87fd75a9698f1 : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c20cd523a44d4d2d95c0ee62ba30182c : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e609ee162ce14e238553bfba4f892ae8 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad115461f5d84cf8970a35df790eb119 : SUCCESS in 14m 53s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f481f94a3e404b4d9a743a1af41271ad : SUCCESS in 1h 49m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c1f8d74885f14f4894431a3f470c4b84 : SUCCESS in 1h 19m 54s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/8dc5c510a28e42bbb46c460502b83d2e : SUCCESS in 1h 32m 26s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/644cbb1ca47e4a268a73bfd09beae14c : SUCCESS in 1h 17m 16s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/e2a7cbcde2fe41708d27ec2f9e7d1cd1 : SUCCESS in 1h 45m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a1612c2e25bc482882a1990c70aa9aa9 : SUCCESS in 2h 06m 08s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ac4dd47c7ce142ac9f86ce85125f0fb5 : SUCCESS in 53m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d5df62652a74df28bae41611b66ad20 : SUCCESS in 55m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88e84237d5444d5ea7ed8da4cacb57f1 : FAILURE in 2h 17m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7f2cf0bcd3c947f9ac631303677e8b2a : SUCCESS in 51m 04s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/44e106b7acba45fd88d3f7764c657ecb : SUCCESS in 1h 11m 11s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfb23f5c6be046a39268df4167ef51c4 : SUCCESS in 1h 12m 56s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2c2af7ed4dda4fb9b8310cb2e19d14f9 : FAILURE in 28m 21s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d8fb93a6e2ad4a8faaf61ed90c46cb42 : SUCCESS in 1h 18m 17s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/bb945373471a4e8d917b56680e31c3a2 : SUCCESS in 5m 18s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1ea68e4f9fc14c1c9d482b74eb06d188 : SUCCESS in 1h 25m 43s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bcba7fa13ae14341a697c90646db9dcd : SUCCESS in 1h 18m 21s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/50a6404c596a43839a6113fbd9d8e108 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":4},{"id":"e313b7fa0b97e63a56553b787f6d1a699cf6c6cf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-05 00:05:23.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"537de0bfad2928b9c418aec828c923a9574c1ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 02:37:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a0444d48a7a4f4da10848de88775449 : SUCCESS in 1h 01m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e65655f452fa4a3f8a8565db3e217a7d : SUCCESS in 1h 21m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb1d471717e44f06910d82151561071c : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f810fa897a4142c0b19b005239ea252f : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d5f42091f464b57bb233ccdaa3a3db6 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f99969ba83f453ea9ac63fe7cc7332a : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fab5c79b0ec54a42a7a9a289e122052c : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c932ebc4c9e44327b1b71259686c75af : SUCCESS in 7m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e27d87e895034d43918ae31eebf236e4 : SUCCESS in 1h 34m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7d8962a4badf4ddb8c09b8e467a0ce49 : SUCCESS in 1h 28m 24s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c14b5c1449954ed7a12d030cf7e84737 : SUCCESS in 1h 17m 48s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/b6b50d2542e243b4a4ea215565526c8e : SUCCESS in 1h 27m 48s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/84fdfbe8ae2f4b6db92d1daa09398ac0 : SUCCESS in 1h 16m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3838b6602e46449694e78cb6572ab8e2 : SUCCESS in 1h 35m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9b0a690d3a2a4239a271d7a581f758e0 : SUCCESS in 46m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7119b32c1bd410eb41892d9892fa704 : SUCCESS in 1h 05m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a643b4da5864e75a38419520f948c6c : SUCCESS in 2h 15m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f7c026d9309e4f0e84057d5ad03047e3 : SUCCESS in 46m 45s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/de570b49c1ad4cf18d72551ae0b39b39 : FAILURE in 57m 10s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71b04ab7560e401c8051ce8245059c40 : SUCCESS in 1h 06m 14s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/91bb293b39de42dd91db02fcac64b5b5 : FAILURE in 24m 05s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c1f2fad1208b48a2ac8c4c40b2fcf7ed : SUCCESS in 1h 14m 51s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1f3d6d4f868c4715b9180cbedd1d1e42 : SUCCESS in 3m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f25f5d8ea6e6462ba2e42d8ba5af36c1 : SUCCESS in 1h 08m 35s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5b13ef1fa35147e6b6c66c37bae8260c : SUCCESS in 1h 06m 57s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/21ff17530809436f9b52b3a34d6ede47 : SUCCESS in 4m 16s\n\nWarning:\n  Job tempest-tox-plugin-sanity-check: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .venv/lib/python3.6/site-packages/oslo_config/types.py in the commit\u0027\n  Job tempest-tox-plugin-sanity-check: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .venv/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Comments left for invalid file .venv/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file tempest/lib/services/clients.py\n  Comments left for invalid file .venv/lib/python3.6/site-packages/oslo_config/types.py\n  Comments left for invalid file tempest/tests/lib/services/object_storage/test_object_client.py\n  Comments left for invalid file tempest/lib/auth.py\n  Comments left for invalid file tempest/api/volume/test_volumes_snapshots_list.py","accounts_in_message":[],"_revision_number":4},{"id":"501af63f9061947bcf3a45328ff6bd4a84bf0fc9","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-04-05 04:08:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nI think the specified maxLength for the locked_reason is incorrect if I\u0027m not mistaken.","accounts_in_message":[],"_revision_number":4},{"id":"8eeaf16e451e2710731bf18223d8bf118aaee4db","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-04-05 20:10:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"752d072a5cb9bf926148cb449ad84599f06cd984","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-04-06 12:44:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"033843524ff20407b1915edfd932a41f7cdbea4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 16:21:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/73f3ae9b66af4e2dbfb361111f5c8b8f : SUCCESS in 1h 15m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e833020583d40d386c5e48da9398fb5 : SUCCESS in 1h 21m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79520b3a17d1466bb150da010ef23a96 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76334aa5a640405f9bfd6d90ebd098ad : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7f800f3226c42d0b7c205d1b37cbaf8 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7311b23bf6d34f678b75af27342e7def : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a20213351bf240809747a6c762525c91 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac8cbddc2f6448da15cd371d4dad0eb : SUCCESS in 9m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f4d3d8ff1c064f4d9a80f347f3b9a69a : SUCCESS in 2h 01m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a965c2102a2f4daeb110866cafcb1614 : SUCCESS in 1h 46m 06s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/5713027e11bf4b1cb2b190b86347aa3b : SUCCESS in 1h 46m 10s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9e974e007ae749489661f21a3fd4db3a : SUCCESS in 1h 26m 20s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/d541d9b38c3f46a3b9f4272c3135136d : SUCCESS in 1h 20m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0844a10a54d1497bb070e713b7d80bef : SUCCESS in 1h 39m 50s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/269fd362c7fb416eaeb1e3852b827a5d : SUCCESS in 50m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/826744ab0c02413c87c892b958958244 : SUCCESS in 1h 25m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ac309ae0e4948fcaa81aa11ac3de97a : FAILURE in 2h 19m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/daa867e45d11448c8533396213ab992a : SUCCESS in 49m 07s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3c83668506c64639af8c898c08ca1317 : SUCCESS in 1h 28m 02s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b150684d7f03496da38c034bb8bbff30 : SUCCESS in 1h 42m 57s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1371df164fed484fb12fd49e6b3e1be8 : FAILURE in 36m 52s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/68bcb6abcc564811b92e2d92a5c7ebf9 : SUCCESS in 1h 18m 33s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e56c7def18da4acfbd3b77d1afd500b0 : SUCCESS in 5m 17s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/06fba7f215f344988d87049707004ec4 : SUCCESS in 1h 32m 18s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4004f009801d42c6bdcf12aeef32edd5 : SUCCESS in 1h 10m 21s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9df4702679fe47dfa52f7f73dc161c01 : SUCCESS in 5m 09s\n\nWarning:\n  Job tempest-tox-plugin-sanity-check: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .venv/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Job tempest-tox-plugin-sanity-check: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .venv/lib/python3.6/site-packages/oslo_config/types.py in the commit\u0027\n  Comments left for invalid file .venv/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file tempest/lib/services/clients.py\n  Comments left for invalid file .venv/lib/python3.6/site-packages/oslo_config/types.py\n  Comments left for invalid file tempest/tests/lib/services/object_storage/test_object_client.py\n  Comments left for invalid file tempest/lib/auth.py\n  Comments left for invalid file tempest/api/volume/test_volumes_snapshots_list.py","accounts_in_message":[],"_revision_number":5},{"id":"d1df33277c5a46e7d6498a1ee0c946a676ee3f8e","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-04-06 17:01:17.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3647d9db31eae195dd1d2138efd0e1b11e7a0ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 20:05:26.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fe89a1d3e2584753a8d2d1820193738c : SUCCESS in 1h 03m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6d262866ecf4fb0bb6271c8a321881a : SUCCESS in 1h 16m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3793ef38ae1e4bd790bb44574b32ee13 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7841cc1bed24543be67ee203f41c5a4 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e15dd87e2c74653a8ddeadb5cefdb01 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9b8c50e4e524ddca99b6464cbb16134 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe95bfcd255849028da03486fb9b355d : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8166cad2e5a34064ba07891cd3baccef : SUCCESS in 10m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/be5d754dc5864ba4bab9e1715b126f20 : SUCCESS in 2h 00m 27s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/2c6e47d1d6e54409ba66bf9e1fb22818 : FAILURE in 1h 01m 16s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/52889f7c4c454d3aaa9680fae15487ef : SUCCESS in 1h 25m 43s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/7250751a4ac248959b7eeec93a6afd36 : TIMED_OUT in 2h 05m 43s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/b912761f00cc482a8315ef8f8d949437 : SUCCESS in 1h 26m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ab19bbf6011b478dada7caf1fbda9df4 : SUCCESS in 1h 47m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1d74331df51743a5adee1b4e1ff2e230 : SUCCESS in 48m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/947f8f6f5b2648b99eb7ca00431e82d5 : SUCCESS in 1h 02m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a2ab8e10d614527981789f61295ddc6 : SUCCESS in 2h 16m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dd2d423bc7fe421a9c5d189cb3c46f5d : SUCCESS in 52m 01s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3bb73db17a0f4588a6d1f6574ed47590 : SUCCESS in 1h 05m 51s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7366235cb3c4bd39f5e2305d8fbc91a : SUCCESS in 1h 21m 58s\n- puppet-openstack-integration-4-scenario004-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c1cc5b59db8f4087ab92eb58855747c8 : FAILURE in 25m 40s (non-voting)\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1c8dfa1d6a874fc781f01aaa03df4254 : FAILURE in 58m 05s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f587154205944b9696842cec2ca193ad : SUCCESS in 5m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7e3df3777b3c45cea3707aff491700d3 : SUCCESS in 1h 34m 36s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3b557c0eb55d40b88ac7deb2e82a837b : SUCCESS in 1h 05m 34s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5fc14f4f5b9c45a98f6a81951bc52421 : SUCCESS in 6m 32s\n\nWarning:\n  Job tempest-tox-plugin-sanity-check: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .venv/lib/python3.6/site-packages/oslo_config/types.py in the commit\u0027\n  Job tempest-tox-plugin-sanity-check: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .venv/lib/python3.6/site-packages/future/standard_library/__init__.py in the commit\u0027\n  Comments left for invalid file .venv/lib/python3.6/site-packages/future/standard_library/__init__.py\n  Comments left for invalid file tempest/lib/services/clients.py\n  Comments left for invalid file .venv/lib/python3.6/site-packages/oslo_config/types.py\n  Comments left for invalid file tempest/tests/lib/services/object_storage/test_object_client.py\n  Comments left for invalid file tempest/lib/auth.py\n  Comments left for invalid file tempest/api/volume/test_volumes_snapshots_list.py","accounts_in_message":[],"_revision_number":5},{"id":"fbb30c522a62f7a2c9f245926442bae72ac87f1a","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-04-14 08:52:20.000000000","message":"Patch Set 5:\n\nSmall -1: does this require a releasenote?","accounts_in_message":[],"_revision_number":5},{"id":"8ff13cef6f2f1afef934e3795bec2a8e8041d834","author":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"date":"2020-04-24 14:49:26.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/701433/10, currently addresses this commit.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"45de6681869dbbf149a55744a4973d98c8496be3","revisions":{"b655deb25e7d9db97c7cd351172182305b14c508":{"kind":"REWORK","_number":1,"created":"2020-03-31 14:22:31.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/11/716311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/716311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/716311/1"}}},"commit":{"parents":[{"commit":"e743e5d39401ad3f15288c20047f9b3f88251444","subject":"Blacklist x/kingbird project from sanity-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e743e5d39401ad3f15288c20047f9b3f88251444"}]}],"author":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 14:19:05.000000000","tz":-240},"committer":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 14:22:30.000000000","tz":-240},"subject":"Add compute API locked_reason response for 2.73","message":"Add compute API locked_reason response for 2.73\n\nThe 2.73 compute API added the \"locked_reason\" parameter [1], to allow\nfor providing a reason when locking a server.  The parameter will appear\nin the following APIs:\n\n- GET /servers/detail\n- GET /servers/{server_id}\n- PUT /servers/{server_id} (update)\n- POST /servers/{server_id}/action (rebuild)\n\n[1] https://docs.openstack.org/nova/latest/reference/api-microversion-history.html#id66\n\nChange-Id: I289a44c5604a12bc8849989954588896c4ff8bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b655deb25e7d9db97c7cd351172182305b14c508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b655deb25e7d9db97c7cd351172182305b14c508"}]},"branch":"refs/heads/master"},"0df8536e779f0c672e2d4dc5801989610d9b6814":{"kind":"REWORK","_number":2,"created":"2020-03-31 17:33:43.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/11/716311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/716311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/716311/2"}}},"commit":{"parents":[{"commit":"e743e5d39401ad3f15288c20047f9b3f88251444","subject":"Blacklist x/kingbird project from sanity-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e743e5d39401ad3f15288c20047f9b3f88251444"}]}],"author":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 14:19:05.000000000","tz":-240},"committer":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 17:33:20.000000000","tz":-240},"subject":"Add compute API locked_reason response for 2.73","message":"Add compute API locked_reason response for 2.73\n\nThe 2.73 compute API added the \"locked_reason\" parameter [1], to allow\nfor providing a reason when locking a server.  The parameter will appear\nin the following APIs:\n\n- GET /servers/detail\n- GET /servers/{server_id}\n- PUT /servers/{server_id} (update)\n- POST /servers/{server_id}/action (rebuild)\n\n[1] https://docs.openstack.org/nova/latest/reference/api-microversion-history.html#id66\n\nChange-Id: I289a44c5604a12bc8849989954588896c4ff8bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0df8536e779f0c672e2d4dc5801989610d9b6814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0df8536e779f0c672e2d4dc5801989610d9b6814"}]},"branch":"refs/heads/master"},"37a8b8847c1fe6d6c00c213b07f638a461151f10":{"kind":"REWORK","_number":3,"created":"2020-03-31 17:59:31.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/11/716311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/716311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/716311/3"}}},"commit":{"parents":[{"commit":"e743e5d39401ad3f15288c20047f9b3f88251444","subject":"Blacklist x/kingbird project from sanity-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e743e5d39401ad3f15288c20047f9b3f88251444"}]}],"author":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 14:19:05.000000000","tz":-240},"committer":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 17:59:05.000000000","tz":-240},"subject":"Add compute API locked_reason response for 2.73","message":"Add compute API locked_reason response for 2.73\n\nThe 2.73 compute API added the \"locked_reason\" parameter [1], to allow\nfor providing a reason when locking a server.  The parameter will appear\nin the following APIs:\n\n- GET /servers/detail\n- GET /servers/{server_id}\n- PUT /servers/{server_id} (update)\n- POST /servers/{server_id}/action (rebuild)\n\n[1] https://docs.openstack.org/nova/latest/reference/api-microversion-history.html#id66\n\nChange-Id: I289a44c5604a12bc8849989954588896c4ff8bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/37a8b8847c1fe6d6c00c213b07f638a461151f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/37a8b8847c1fe6d6c00c213b07f638a461151f10"}]},"branch":"refs/heads/master"},"4fd53053c77e8887cc293718394c07f30f26cf91":{"kind":"REWORK","_number":4,"created":"2020-04-01 01:51:59.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/11/716311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/716311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/716311/4"}}},"commit":{"parents":[{"commit":"e743e5d39401ad3f15288c20047f9b3f88251444","subject":"Blacklist x/kingbird project from sanity-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e743e5d39401ad3f15288c20047f9b3f88251444"}]}],"author":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 14:19:05.000000000","tz":-240},"committer":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-04-01 01:51:04.000000000","tz":-240},"subject":"Add compute API locked_reason response for 2.73","message":"Add compute API locked_reason response for 2.73\n\nThe 2.73 compute API added the \"locked_reason\" parameter [1], to allow\nfor providing a reason when locking a server.  The parameter will appear\nin the following APIs:\n\n- GET /servers/detail\n- GET /servers/{server_id}\n- PUT /servers/{server_id} (update)\n- POST /servers/{server_id}/action (rebuild)\n\n[1] https://docs.openstack.org/nova/latest/reference/api-microversion-history.html#id66\n\nChange-Id: I289a44c5604a12bc8849989954588896c4ff8bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4fd53053c77e8887cc293718394c07f30f26cf91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4fd53053c77e8887cc293718394c07f30f26cf91"}]},"branch":"refs/heads/master"},"45de6681869dbbf149a55744a4973d98c8496be3":{"kind":"REWORK","_number":5,"created":"2020-04-06 12:44:23.000000000","uploader":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"ref":"refs/changes/11/716311/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/716311/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/716311/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/716311/5"}}},"commit":{"parents":[{"commit":"e743e5d39401ad3f15288c20047f9b3f88251444","subject":"Blacklist x/kingbird project from sanity-check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e743e5d39401ad3f15288c20047f9b3f88251444"}]}],"author":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-03-31 14:19:05.000000000","tz":-240},"committer":{"name":"James Parker","email":"jparker@redhat.com","date":"2020-04-06 12:44:10.000000000","tz":-240},"subject":"Add compute API locked_reason response for 2.73","message":"Add compute API locked_reason response for 2.73\n\nThe 2.73 compute API added the \"locked_reason\" parameter [1], to allow\nfor providing a reason when locking a server.  The parameter will appear\nin the following APIs:\n\n- GET /servers/detail\n- GET /servers/{server_id}\n- PUT /servers/{server_id} (update)\n- POST /servers/{server_id}/action (rebuild)\n\n[1] https://docs.openstack.org/nova/latest/reference/api-microversion-history.html#id66\n\nChange-Id: I289a44c5604a12bc8849989954588896c4ff8bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45de6681869dbbf149a55744a4973d98c8496be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45de6681869dbbf149a55744a4973d98c8496be3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
