)]}'
{"id":"openstack%2Ftempest~719426","triplet_id":"openstack%2Ftempest~master~I0112182811ce8fa573e66f52cc20a597764a3634","project":"openstack/tempest","branch":"master","topic":"fix412","hashtags":[],"change_id":"I0112182811ce8fa573e66f52cc20a597764a3634","subject":"Change the post url according to the api reference","status":"ABANDONED","created":"2020-04-13 09:33:25.000000000","updated":"2020-07-20 09:32:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d8ce268fd3b3c2ca709be7851a3875a54f667633","_number":719426,"virtual_id_number":719426,"owner":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-09 06:21:44.000000000","_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"date":"2020-07-09 21:25:47.000000000","_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-05 12:40:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 08:41:31.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-05-11 08:58:40.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2020-06-05 12:40:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-09 21:25:47.000000000","updated_by":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"reviewer":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"state":"REVIEWER"}],"messages":[{"id":"2a5f83de82d7e5a6b546d047502b009bee33521e","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-04-13 09:33:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14094e2c7d9b888893a6f55862c34145fe58200d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 11:52:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ec9b3f8fd41f4cdd8169479767919342 : SUCCESS in 1h 08m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c08c73dacf80407f83a142d51fd94564 : FAILURE in 1h 00m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e3047f5690949179a37dba72727b747 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ae81da3cf04f5a99394ac47ab443dc : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/068364c5feef485aaf2adff9c1a81df1 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a524bd1a2819474f9e3eb5a46ddbb437 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7ced98110fd46a5b01415b27ac982fd : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3fd75834918425982a2096d92a3e242 : SUCCESS in 8m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d1ff28d18f0e48d49afd316156de2ece : FAILURE in 1h 40m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/426b131d76ff42f2a9b0b2d3d4654639 : FAILURE in 1h 05m 31s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7675bc3cf7cb4de1b1750d4c62c3f959 : FAILURE in 1h 07m 14s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a20f3bb7d57e4159848bc838aaf87038 : FAILURE in 1h 00m 51s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/417512cbf0a64957a550c8af6b612bf3 : SUCCESS in 1h 26m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0ec2ed3f193b4eccb778fbc0bd42fedd : FAILURE in 1h 18m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9ec5643c27b443de85b1e7af3612b1cc : SUCCESS in 47m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8880b8b6693344319522c6cb4c1c592c : SUCCESS in 1h 02m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b933a880f23495db40b607eb96dc320 : SUCCESS in 2h 13m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bcae6e60294641e2b8d05703a3b6c1ce : SUCCESS in 48m 12s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/47657c8c02624ff280581b14e41dfdce : SUCCESS in 1h 03m 03s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed11940eb0474bfb86d66bcd2a33cb52 : SUCCESS in 1h 41m 31s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0cabd6589cf7485292a027b3488de91d : SUCCESS in 1h 05m 56s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/072daa1e6a3141628322eb72389db32d : SUCCESS in 3m 14s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/cf090ed1f20f472a8a46f8a84d0a0baa : FAILURE in 1h 08m 16s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/58179130b3584099a37a81aa3c5402bd : SUCCESS in 1h 00m 51s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2ab77224e72741bc873e46a7ede2df28 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6eec4a5195a295750b8e6189fa60fa8591304d8b","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-04-14 08:41:31.000000000","message":"Patch Set 1: Code-Review-1\n\nThe change is fine, but I think it requires a releasenote, plus, the commit message doesn\u0027t have too much information (I had to check the code to get the api-ref and check the change","accounts_in_message":[],"_revision_number":1},{"id":"6e3ce0d8ef7a3d70171321707816570e83ba8a6d","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-05-09 07:28:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c74d39ea9f2b42957c3f8012cd1969bc81fef89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 09:55:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/913bc8bb29234a51abe657544b6817f8 : SUCCESS in 1h 02m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14553e52dcc843a785782751aa265e06 : FAILURE in 1h 23m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a0953bc880a4d6fbb4a28de3d636b41 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6432a80032724596b6834678c0ee5f18 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3a5e62dadbd46dd88fcbe46520191f6 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84733a2658f64af5b031480c31c25dbb : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e7abc8b80624cb6a8fdc171a4aa4554 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde3d22eeff84b74ae85607903e2764b : SUCCESS in 15m 22s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/7ee8073401b8432d861407a4333c40c3 : FAILURE in 1h 37m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3f579cc14cc34f30a3bc82328bcd577a : FAILURE in 1h 08m 06s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/985b51bdc14c41748ed4b1786e800e7b : FAILURE in 1h 29m 10s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f136d97df1f24c8fbcffe4ec78268f1d : FAILURE in 1h 09m 03s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/dc74ad253cd64aab98a3ef55cd777a55 : FAILURE in 1h 00m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/559bfdbde56b4b4c96f285f777a90359 : FAILURE in 1h 47m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e8409482fee94748a158e17d070b7928 : SUCCESS in 46m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72d09e9d1b354e82b83e977da292ce79 : SUCCESS in 58m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fc09e82b164404b873701e5bf0f6d23 : FAILURE in 2h 24m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c206c155220643d0a42d58f9b61081e9 : SUCCESS in 52m 23s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b17c3dede4da4af08ce6f1b1460d05af : SUCCESS in 1h 29m 46s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ec193538a134aa6956097e97001d985 : SUCCESS in 1h 04m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7b86256e629c4f6e81ee22c4e99efb63 : SUCCESS in 1h 10m 58s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c901a15d85314e86aa5a270dc7e72bfd : SUCCESS in 3m 35s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/71d560e223804ba796a9dc54c9406be4 : FAILURE in 1h 28m 19s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d627c5d6a0de4305a70588e7bd2eaf21 : FAILURE in 56m 12s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b9d1022cbcf64e3a89b03a745fe86c01 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b70f6e7bf14e067b6a6e3f4792d7201b7f9d3e79","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-05-11 08:58:40.000000000","message":"Patch Set 1: Code-Review-1\n\nI agree with arxcruz, commit message should be self-descriptive, the commit message here won\u0027t help much.","accounts_in_message":[],"_revision_number":1},{"id":"d5b470d1b22cbf353f5498b5bf240253d3dd7d80","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-06-01 09:19:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9dbd94c66e326c128b0184018ac5e1e6463f14e9","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-06-01 09:20:01.000000000","message":"Patch Set 1:\n\nI have change the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"0186011799960e9317b9145d6d9dc12ff3364e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:54:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/17d8831610ed447883a6b17796fd508d : SUCCESS in 56m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a359d9673c13472fad8b623c17da38c8 : FAILURE in 1h 00m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a4c1f93ba8a48ebbba2b624ea6d719f : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d99144189294c5ca1350b7aa354b30a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca5dc3843c904b79aa709aefdf532ab7 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86c9fe07dd594ec38ec3aec69813dfca : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb8c2d0ab4d1494995f0b1250c376533 : SUCCESS in 11m 16s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/95879f27d7684f48ba419989a99a98a4 : FAILURE in 1h 30m 39s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/261ce22a2499448bbe6b54b74385683b : FAILURE in 1h 05m 34s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/3a60ad4276ec425fad93c6f05cf3510d : FAILURE in 1h 01m 56s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c65cbd4527fe466fa0cb8b37d1b18adc : FAILURE in 1h 06m 02s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/ff39996c98074fecac6e7e664a5dcd55 : FAILURE in 1h 00m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4c368419dea4490595983c031de32e43 : FAILURE in 1h 14m 07s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e00018c383234911bb47411ea6a736ec : SUCCESS in 43m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a28c528a08c94e5095860767527d0592 : SUCCESS in 50m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d23b5dbfe094490b9b886d46ebf8ea3 : SUCCESS in 2h 32m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cb3aa303da0344b68aa8f59962390a83 : SUCCESS in 46m 43s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c82b2bed36234bffa07add51c482b47c : SUCCESS in 1h 24m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7feb5582f5b94719ba34bb2ca91fc1f7 : SUCCESS in 1h 34m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1586ce0dd3ae4330aa00fcf4dd47ce2b : SUCCESS in 1h 40m 33s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3556b247997645918a06d4f6b2f929fc : SUCCESS in 3m 18s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3d2eec2bf6bd470b9867d4a5aaa8fe4f : FAILURE in 1h 18m 48s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bb8645094c6444628e8999b9a941e4f2 : SUCCESS in 1h 08m 48s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4bac198b66af4a88b087f21ddeb121c4 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"34a8a8e0eb78551595cce8e7adca545da333e1a3","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-06-05 02:16:38.000000000","message":"Patch Set 2:\n\nmy change will cause zuul error， so rollback the code","accounts_in_message":[],"_revision_number":2},{"id":"28e671166cd4672eedee4b276aece7b08c8c1df6","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-06-05 09:31:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f7c9f96588f9619603421ecc847314aa4aff541f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 12:40:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5de254a8b8c94175a2316735e6ca25fb : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f4350d3ba5074b2698778be6957b1f86 : SUCCESS in 3m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/37143910f63a4737953192b34abd5da3 : SUCCESS in 1h 10m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ced3ca26e884b97ad630e91fd4ab27c : SUCCESS in 1h 36m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72555e8c0e4444a2bb2e0aabc2fa2fe5 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3469c565c16e47b39afbc6f3824ace56 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a91d52c830124ace8f8e662281ec14d9 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfe9d96dbe9741088fcb54a08a887819 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a49aae570854601b61d81862fc811cf : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bba9637b277d411ba22b46198ec78250 : SUCCESS in 6m 37s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/730fc780b9d64c189d2c2cacbfbdbed0 : SUCCESS in 36m 17s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/613e56ac233f4e33a2214f62d8ed773f : SUCCESS in 42m 56s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c7d97636e9b54d078ec7e37f5be70bee : SUCCESS in 1h 53m 02s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/318026b3351844c6a2a1605d31509349 : SUCCESS in 1h 17m 59s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/e85fba2d49a548dc9ec56e01e4844927 : FAILURE in 1h 00m 53s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/151fb7db531e42e49d7559b2e9eeeaee : SUCCESS in 1h 32m 43s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/f73503761dfc4509801bea69f4f50314 : SUCCESS in 1h 27m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/81d2a66b4abc4612ab1272a1634ca8be : SUCCESS in 1h 45m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f09f02b005654136affb85fa2b9c8aba : SUCCESS in 41m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0a1627366204952a2b7422aca5e2e8f : SUCCESS in 54m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0efa10682b7b44ffb9131b2ea9353614 : SUCCESS in 2h 05m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2497d2a1151445aab82431c1884cfe74 : SUCCESS in 51m 23s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/022fcd5a6e2d43dcb5b9ee24a784ca6c : SUCCESS in 57m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/beeceb17660f4dd39d0935f05f58856b : SUCCESS in 1h 19m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0eaa2e6dd87a4895b29074532f9ff62b : FAILURE in 43m 21s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f8c503dd9e32465ba616140642f1300d : SUCCESS in 3m 06s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ad93614f45f14a52a73c6e3af9557817 : SUCCESS in 1h 29m 32s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/666176f4bc834868827442b1ba4e96f9 : SUCCESS in 1h 03m 35s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/fc38a713bd414b978d559c08454c88b8 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"64150bd4ff8752dab8d05ed11ef43ccb1d3a6a03","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-07-09 06:21:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b91f3702889ae0653f4d0f776903b8f32dfe4738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:34:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5e7ebfc32984a8d90e00f019c3d25e0 : SUCCESS in 3m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e1010c91ccb476ca53feb95a598305a : SUCCESS in 3m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/bd36deb33ecf478ab988546618f4f0c7 : SUCCESS in 1h 00m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50b96c34c50d4dd6b25f354374740ef1 : SUCCESS in 1h 07m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c11881c09989404e8eb3f6a1121034ee : FAILURE in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa888d73df74a37a26cb76903de6514 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cf45ede58714fdfb411a00539ada026 : FAILURE in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14ae2d2e93074b0196af3db1f284a512 : FAILURE in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c711e0d0f3c47599fdd6deb084b4a6c : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/920008a3e07b4bf6a8bd01f8cae5985f : SUCCESS in 5m 59s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/ebd6477198214f9bb1745be598fd507d : SUCCESS in 29m 11s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/14f7c4688394435488531ae1bcd8a672 : SUCCESS in 30m 05s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e9b7d6861604417aa010b32ce4a0670a : SUCCESS in 2h 00m 24s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/f23f0f066b4143f8abbb350e718291d8 : SUCCESS in 1h 16m 03s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6a8c3ba3b0874410ad139c9523bad9c7 : SUCCESS in 1h 18m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a1a7e959ecb0452c8a12db8aa66d4a95 : SUCCESS in 1h 21m 14s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a831c998995c463ab70fc491868b732e : SUCCESS in 1h 14m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2a056dc376b642379e94c6e69f08ff00 : FAILURE in 1h 17m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/da513525297d4245b0cf9ab1941264ae : SUCCESS in 45m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5de763cf7bcd4143b3b8320c85cbe621 : SUCCESS in 39m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/029e75aa0d684f2cba02b9077fd88736 : SUCCESS in 2h 04m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/952b3cf13a4c429fa8960592537f5c69 : SUCCESS in 59m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/52e468b7a0984b9183a0d6eee7800e66 : SUCCESS in 1h 36m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fea68afb6fc43378919e8734b530ce8 : SUCCESS in 1h 26m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7d247ce264864e34888723f663d68d1c : SUCCESS in 1h 04m 36s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0d497ba253654f3b9d7e13ed46c96df8 : SUCCESS in 5m 49s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7a5346130e284a6da35dc309101ac584 : FAILURE in 1h 01m 16s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4317093db4c845ce891b39b2cf843f30 : SUCCESS in 56m 15s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c020dd522d4940efb7ee1115c19af295 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"98da754a8c74ff8a4c52e66a57bd90668760c1b6","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2020-07-09 21:25:47.000000000","message":"Patch Set 3:\n\nrecheck fixed in HEAD","accounts_in_message":[],"_revision_number":3},{"id":"d572ddd5c23f9c19a6fafee06fea35d1c185a8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 23:35:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de3ce0b46c884261a19fa67874906a49 : SUCCESS in 4m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a33f270aded54b9b9ce5dda3ad27d14a : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0da05065c4634178bb51710f0732c651 : SUCCESS in 1h 01m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df348aa41bfd46ae9f55e5022f752c60 : SUCCESS in 1h 57m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71961f916400459fb83b75c268411f54 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9656699509e64dda8f326f07389a3786 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17f27460743d47ad82d70ac0e14c0d93 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/820a4062182846ea9f1c07aebf40fe22 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fedd12ffd74248acd63e282481ddaa : SUCCESS in 19m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/112e549ad1ba435cb61d717f1882d9d9 : SUCCESS in 4m 57s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/470d42ff38b243caa9f0d04572608449 : SUCCESS in 42m 44s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/322b00c0f19c4315988c8bca3d2b5d66 : SUCCESS in 1h 09m 41s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/06f52840397948c494e103d28b5de366 : SUCCESS in 1h 35m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/0a54be99502c4ee0b03985db0889e245 : SUCCESS in 1h 49m 42s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c44b10596371486898b8c23be4b4f804 : SUCCESS in 1h 25m 44s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/11aa9ce24bae4bb8bd6f79bc6b15917e : SUCCESS in 1h 21m 04s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/1603c834831c4591adcbbaace081977f : SUCCESS in 1h 56m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86604e15fdbb4a598d49da2dda7c4ff3 : SUCCESS in 1h 39m 43s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2083bbf21d954526a9073b6c3ec73706 : SUCCESS in 45m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38eb3f7ff8954ff4a391a2f323ea9f0f : SUCCESS in 1h 30m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af9d247effcb467891ec61a6f81a4dd1 : SUCCESS in 2h 02m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/19efb3a81c704ba3a7f02c4f775cfd15 : SUCCESS in 47m 23s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a5ca429e4c1c4d4f861f71a4ba4ae679 : SUCCESS in 1h 50m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/61bcca6d23624be6b8bbb2944dea9987 : FAILURE in 1h 31m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/da22905fcdf848d28959245389f5c343 : SUCCESS in 1h 53m 47s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/34cacbd4772b4b1e8851a87797aa6c93 : SUCCESS in 3m 33s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f5f5de42dbd34815a8c4407f2c3f8e29 : SUCCESS in 1h 19m 16s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/95f827dbb4b1455c914c1e9fb4c9c9b8 : SUCCESS in 1h 40m 58s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9e773fa3796a410680deb6c28964177b : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c62688fbe69941768fb7eee97f085fadfe004a82","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-07-20 09:32:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bad6522bb0ccb5d715132394e291aadd0e121d41","revisions":{"553f95868b05dc0e26278db0012c9904038cb61b":{"kind":"REWORK","_number":1,"created":"2020-04-13 09:33:25.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/26/719426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/26/719426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/26/719426/1"}}},"commit":{"parents":[{"commit":"91f0e008821b3d7be21a62da5706d51e9a30e1e2","subject":"Merge \"Keystone - List user groups \u0027membership_expires_at\u0027 attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91f0e008821b3d7be21a62da5706d51e9a30e1e2"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-04-13 09:25:17.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-04-13 09:25:27.000000000","tz":480},"subject":"Change the post url according to the ref api","message":"Change the post url according to the ref api\n\nchange the volume manage url according to ref api\n\nChange-Id: I0112182811ce8fa573e66f52cc20a597764a3634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/553f95868b05dc0e26278db0012c9904038cb61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/553f95868b05dc0e26278db0012c9904038cb61b"}]},"branch":"refs/heads/master"},"02ddc0059e2fc730eeeb2fb7ae5b8dd8eb0c3668":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-01 09:19:41.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/26/719426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/26/719426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/26/719426/2"}}},"commit":{"parents":[{"commit":"91f0e008821b3d7be21a62da5706d51e9a30e1e2","subject":"Merge \"Keystone - List user groups \u0027membership_expires_at\u0027 attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91f0e008821b3d7be21a62da5706d51e9a30e1e2"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-04-13 09:25:17.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-06-01 09:16:43.000000000","tz":480},"subject":"Change the post url according to the api reference","message":"Change the post url according to the api reference\n\nchange the volume manage url according to api reference https://docs.openstack.org/api-ref/block-storage/v3/index.html#manage-an-existing-volume\n\nChange-Id: I0112182811ce8fa573e66f52cc20a597764a3634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/02ddc0059e2fc730eeeb2fb7ae5b8dd8eb0c3668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/02ddc0059e2fc730eeeb2fb7ae5b8dd8eb0c3668"}]},"branch":"refs/heads/master"},"bad6522bb0ccb5d715132394e291aadd0e121d41":{"kind":"REWORK","_number":3,"created":"2020-06-05 09:31:29.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/26/719426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/26/719426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/719426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/26/719426/3"}}},"commit":{"parents":[{"commit":"91f0e008821b3d7be21a62da5706d51e9a30e1e2","subject":"Merge \"Keystone - List user groups \u0027membership_expires_at\u0027 attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91f0e008821b3d7be21a62da5706d51e9a30e1e2"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-04-13 09:25:17.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-06-05 09:31:19.000000000","tz":0},"subject":"Change the post url according to the api reference","message":"Change the post url according to the api reference\n\nchange the volume manage url according to api reference https://docs.openstack.org/api-ref/block-storage/v3/index.html#manage-an-existing-volume\n\nChange-Id: I0112182811ce8fa573e66f52cc20a597764a3634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bad6522bb0ccb5d715132394e291aadd0e121d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bad6522bb0ccb5d715132394e291aadd0e121d41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
