)]}'
{"id":"openstack%2Ftempest~719654","triplet_id":"openstack%2Ftempest~master~I20dcab6466e479223e8043e1129473c69062ace3","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I20dcab6466e479223e8043e1129473c69062ace3","subject":"Add run_validation flag to skip_checks","status":"ABANDONED","created":"2020-04-13 18:06:26.000000000","updated":"2020-05-14 19:03:16.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f28822d73411331c739e4eaa75a9e264830c67cc","_number":719654,"virtual_id_number":719654,"owner":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-14 19:03:00.000000000","_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-27 10:43:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-05-14 19:00:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"value":-1,"date":"2020-05-14 00:34:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-28 09:06:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 20:28:39.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-04-13 20:28:53.000000000","updated_by":{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},"reviewer":{"_account_id":5690,"name":"Masayuki Igawa","inactive":true},"state":"REVIEWER"},{"updated":"2020-04-27 10:43:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-28 09:06:45.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-05-14 00:34:34.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-14 19:00:00.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"eef629f0b87d967bebd363404fae8894f6b58e7b","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-04-13 18:06:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c3762a0209d96868d42eccb165ee267506d6350","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-04-13 18:07:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"288a498662c2070f272eb7f885d42921a6310227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 20:43:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/37c6334c00e842f488809025bdee56f0 : SUCCESS in 1h 13m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dc781a71caf40d0a2a35a28bfe723ff : SUCCESS in 1h 21m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1e1f1fd48c648b7adb5fe268c2b404b : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1aa2392ce5c46b79509a6f4843c584d : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee6c7d202ee144528462bc71fc98d6b1 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/077f04fdf9b64f668ad8ee46d7f46060 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98f62ff261b74d25be0b99e8f77b2664 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5e36020c624bb6bc811e71e9ebaed8 : SUCCESS in 10m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e608775cdbcd47c3b16d9886e6666a42 : FAILURE in 1h 47m 22s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d10c48105fbb477bab8c26cca93e625d : SUCCESS in 1h 16m 53s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/cae57bb16de54a67b0677d22fc886f6f : SUCCESS in 1h 19m 46s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9683c374ed064a2a9cd89057fa135eb6 : SUCCESS in 1h 14m 29s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/6cb3d1d7e01145dfa74d588d137d0e0e : SUCCESS in 1h 18m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/84c78f1fd1144301a7a80bff966772bb : SUCCESS in 1h 35m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/41a56686e30649ef82304ea090a57bf9 : SUCCESS in 44m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ebf19341d9a4fe7a830caab2a3f810c : SUCCESS in 1h 06m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06ceb41cc08b485589909f6619fc84fd : SUCCESS in 2h 29m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/551bf1a8d52c463d9f230a7407ffd63a : SUCCESS in 49m 31s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/de4771193e704c5d82da1e5c0e0616a8 : SUCCESS in 1h 14m 51s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c50f4cd47364e73b9985de3d71b1814 : SUCCESS in 1h 19m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9b52cc13b3a248a8bbab9c9748fc880b : SUCCESS in 1h 16m 50s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1f991d0fc67d41058c666ff8ac7c5c22 : SUCCESS in 3m 52s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5553ba178fb5418bba223273eac8c121 : SUCCESS in 1h 03m 23s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b049893b7fde499f912c38ba3220f06e : SUCCESS in 1h 19m 56s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2137bbf13e4a4c0185e5a1f32f3da04b : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"989ad23fe8534d0c428697ea91216dcbb4dd512a","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-04-14 08:36:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1bf3b71b21ba3e9985b50625d079cd7fab1545ca","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-04-21 07:55:10.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nas the test uses CONF.validation.image_ssh_user value, therefore it makes total sense to add a skip unless CONF.validation.ru_validation is true\n\nsee one comment below, after that I\u0027ll +2","accounts_in_message":[],"_revision_number":2},{"id":"7ff41231591c44cc44f1024d329af8c90ea33c61","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-04-26 23:33:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a20c9ef1922bd6f12bfb21437df6dfdd30ec96e0","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-04-26 23:46:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6f9a5cb4e6fc33e432f0567373bcff2de4ae7e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 01:48:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/35c2df5a220247b4946fa83b5d694d5c : FAILURE in 28m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a124468207364953913bec5a93de32ad : SUCCESS in 1h 29m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a03a0941dbb24a0a80696008f9dfea5a : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6511fd3ef273489c971b66da7ded353b : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/249c886956194cdbae74874725ed285c : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/799b419c7d504d91b3d6222c5c7f4136 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7d6a896dc98439eb5d0e067e898f139 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661503c3fae94301b0afeb0bbb82b033 : SUCCESS in 8m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/663abaf28c5245349f1c9b5d7ecf785e : SUCCESS in 1h 43m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/039e337967d44237bd40d75dca1f4c9a : SUCCESS in 1h 19m 53s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/63bb7aa4d38044588aebacfb0e5f64fd : SUCCESS in 1h 11m 41s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/126d09b917e6487aa25b7d6d89fe132c : SUCCESS in 1h 36m 48s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/c7e1eea8a5a7410797ca18d4b138cf57 : SUCCESS in 1h 08m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3678a56286994bd8be0af426130ea67d : SUCCESS in 1h 47m 09s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f374d5706cad4c2abeaec04b07d45174 : SUCCESS in 38m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/945882b0722248b6972c0c681d4a1669 : SUCCESS in 1h 04m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e749e992e2a845f5acdca6a0e616e051 : SUCCESS in 2h 00m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5b87a6b296e741f496664fdca336e077 : SUCCESS in 48m 16s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0e981dfadea740b1b025658f6c9f1010 : SUCCESS in 1h 03m 11s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80433c92009e4b93816448458016c085 : SUCCESS in 1h 40m 56s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f5d5b44c23944faea2c0f7ed2815d180 : SUCCESS in 1h 19m 45s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a00c00cf5ce34267a8607b4d39720643 : SUCCESS in 3m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2fa98dfaef2649d5bb408b0e477ff55b : SUCCESS in 1h 13m 36s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f1205d7a21ce4daf9b3cca849f04d07a : SUCCESS in 1h 00m 35s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9cb801c076a54b37b873c9f4982caffe : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"7799731301ada85fa0802e41d54503787d6c6d6e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-04-27 08:29:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"11f7e425232ae90189c6b0ee1fb1f6ff0735772d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:43:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/40874752e36f4892a266436d413cb54b : SUCCESS in 1h 24m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8cc13bbd007140159cf1a502cb016439 : SUCCESS in 1h 29m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa7ccfe33c19491ea86262d529f98093 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b221c35c465e4f6bb965f65d6981350a : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc02524b83ab44ae8d5ecf0111c445a4 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/984110cc554c41ac8282391108c93998 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bb9a47a2a074d278c50cef1e1115e69 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35991edfc1a4434c808fc5348bfcd178 : SUCCESS in 10m 59s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/92e1cf034e44459eb40b2a6c399cc4f0 : TIMED_OUT in 2h 06m 54s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/68c702f0d87f4822b4c23f66fdc6f18b : SUCCESS in 1h 17m 57s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/8524fed841e54f5b90679e5dfaee593d : SUCCESS in 1h 25m 36s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/af3138af3f8847ee966eb0d38604d52a : SUCCESS in 1h 14m 06s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/06ed45bb4dae4c16abd2822e716742ae : SUCCESS in 1h 20m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cea49782dae2454481ac79f07acd0354 : SUCCESS in 1h 33m 00s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c843c377cabc477d9655cc6c55470db1 : SUCCESS in 55m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9393e0b5c54632a09e610ef52d5b90 : SUCCESS in 59m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60b3d4533c3f42e2aa44fc29a7fc434c : SUCCESS in 2h 10m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4410cef2417a4283a96b3e2a7ec56c8f : SUCCESS in 44m 02s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/16a7a0d59fe34c9ea68d793590371eb9 : SUCCESS in 1h 05m 55s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/327e1cc8141a4b56baa8734bd5086a5f : SUCCESS in 2h 04m 16s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/ee82bc445d3949c8ab81d2cdde014dbd : SUCCESS in 1h 21m 40s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/28d8897e9c5c458d811d9071d1914055 : SUCCESS in 5m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8da72ac0bd7f4ae3b36b4c10be286b38 : SUCCESS in 1h 25m 55s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a8c9c3f178984ef4be9cb65be64dd8c9 : SUCCESS in 1h 43m 43s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1e86b2a756824aa893eac84ae95ad01e : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":3},{"id":"91a1c5a6ca2bec381a20ecb60756095c6242c261","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-04-27 11:42:41.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":3},{"id":"da95c68ca4c2b55026b461cc3f93eb6a65c7639d","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-04-28 09:06:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"29d121e2ef5b0a23fe385f0eca7b4d091be92eb8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-14 00:34:34.000000000","message":"Patch Set 3: Code-Review-1\n\nscenario tests setup their own ssh things and not using validated resources created via CONF.run_validation.so we do not need to control scenario tests with CONF.run_validation.\n\nCONF.run_validation and so does validated resources are only for controlling the ssh in API tests","accounts_in_message":[],"_revision_number":3},{"id":"45d87c87f6c17c2ae8c35761ab72afeaa1820dc8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-05-14 06:57:40.000000000","message":"Patch Set 3:\n\n\u003e scenario tests setup their own ssh things and not using validated\n \u003e resources created via CONF.run_validation.so we do not need to\n \u003e control scenario tests with CONF.run_validation.\n \u003e \n \u003e CONF.run_validation and so does validated resources are only for\n \u003e controlling the ssh in API tests\n\nBut scenario tests use CONF.validation settings, for example test_server_basic_ops:\nhttps://opendev.org/openstack/tempest/src/commit/e67ebd83d935488f8b53bf4ae314fd8e40fb29a5/tempest/scenario/test_server_basic_ops.py#L47-L48\n\nand test_network_basic_ops scenario uses image_ssh_user from validation section, shouldn\u0027t be then also use run_validation option when it\u0027s from the same section?\nhttps://opendev.org/openstack/tempest/src/commit/e67ebd83d935488f8b53bf4ae314fd8e40fb29a5/tempest/scenario/test_network_basic_ops.py#L120","accounts_in_message":[],"_revision_number":3},{"id":"8da82474df3bfb768212b44a3fb001dc3b79c0fe","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-05-14 19:00:00.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"1e95b727f08d742b6d476d16a802335629980d7d","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-05-14 19:03:00.000000000","message":"Patch Set 3:\n\nAs per scenario test run_validation doesn\u0027t seem to be a right flag for skipppin/unskipping the test as we are not really creating any validation resources here in scenario test and then using that to ssh. So if we are using any other instance that doesn\u0027t support remote ssh then the best way is to blacklist the test as of now until we plan on moving the scenario test to be done remote login in the same way as we do in api tests","accounts_in_message":[],"_revision_number":3},{"id":"f28822d73411331c739e4eaa75a9e264830c67cc","author":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"date":"2020-05-14 19:03:16.000000000","message":"Abandoned\n\nAs per scenario test run_validation doesn\u0027t seem to be a right flag for skipppin/unskipping the test as we are not really creating any validation resources here in scenario test and then using that to ssh. So if we are using any other instance that doesn\u0027t support remote ssh then the best way is to blacklist the test as of now until we plan on moving the scenario test to be done remote login in the same way as we do in api tests","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ea56abbf7542780c4f508c02106864b05087e35","revisions":{"c79c08b972a01c0ba45bd1c1eab2a87f3c7ebce4":{"kind":"REWORK","_number":1,"created":"2020-04-13 18:06:26.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/54/719654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/54/719654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/54/719654/1"}}},"commit":{"parents":[{"commit":"3d13f07ed64c3f8c5ae1b6366600841366e713d2","subject":"Merge \"Fix deletion of network quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3d13f07ed64c3f8c5ae1b6366600841366e713d2"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-04-13 17:58:34.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-04-13 17:58:34.000000000","tz":-240},"subject":"Add run_validation flag to skip_checks","message":"Add run_validation flag to skip_checks\n\ntest_network_basic_ops was executing all the tests regardless\nof the desire which affect some of them to failed specially the one\nwhich uses Windows instances.\n\nChange-Id: I20dcab6466e479223e8043e1129473c69062ace3\nCloses-Bug:#1870983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c79c08b972a01c0ba45bd1c1eab2a87f3c7ebce4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c79c08b972a01c0ba45bd1c1eab2a87f3c7ebce4"}]},"branch":"refs/heads/master"},"113c57c73b6404afec398e005b54c5e58efc0edf":{"kind":"REWORK","_number":2,"created":"2020-04-13 18:07:44.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/54/719654/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/54/719654/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/54/719654/2"}}},"commit":{"parents":[{"commit":"3d13f07ed64c3f8c5ae1b6366600841366e713d2","subject":"Merge \"Fix deletion of network quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3d13f07ed64c3f8c5ae1b6366600841366e713d2"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-04-13 17:58:34.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-04-13 18:07:35.000000000","tz":-240},"subject":"Add run_validation flag to skip_checks","message":"Add run_validation flag to skip_checks\n\ntest_network_basic_ops was executing all the tests regardless\nof the desire which affect some of them to failed specially the one\nwhich uses Windows instances.\n\nChange-Id: I20dcab6466e479223e8043e1129473c69062ace3\nCloses-Bug:#1870983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/113c57c73b6404afec398e005b54c5e58efc0edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/113c57c73b6404afec398e005b54c5e58efc0edf"}]},"branch":"refs/heads/master"},"5ea56abbf7542780c4f508c02106864b05087e35":{"kind":"REWORK","_number":3,"created":"2020-04-26 23:46:43.000000000","uploader":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"ref":"refs/changes/54/719654/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/54/719654/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/719654/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/54/719654/3"}}},"commit":{"parents":[{"commit":"a188178416c1e7a317dbfe6dc98f9786ae99d38b","subject":"Merge \"Add release notes page for version 24.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a188178416c1e7a317dbfe6dc98f9786ae99d38b"}]}],"author":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-04-13 17:58:34.000000000","tz":-240},"committer":{"name":"Paras Babbar","email":"PBabbar@redhat.com","date":"2020-04-26 23:45:02.000000000","tz":-240},"subject":"Add run_validation flag to skip_checks","message":"Add run_validation flag to skip_checks\n\ntest_network_basic_ops was executing all the tests regardless\nof the desire like where we even explicitly mentioned to not\nrun validation this flag will skip them from remote login.\n\nChange-Id: I20dcab6466e479223e8043e1129473c69062ace3\nCloses-Bug:#1870983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5ea56abbf7542780c4f508c02106864b05087e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5ea56abbf7542780c4f508c02106864b05087e35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
