)]}'
{"id":"openstack%2Ftempest~719750","triplet_id":"openstack%2Ftempest~master~I355bc829e7c0f01613fc907a6bf9e03b785ba3ea","project":"openstack/tempest","branch":"master","topic":"missing_space","hashtags":[],"change_id":"I355bc829e7c0f01613fc907a6bf9e03b785ba3ea","subject":"[Trivial]Add space between words in log message","status":"MERGED","created":"2020-04-14 07:39:34.000000000","updated":"2020-04-18 02:44:48.000000000","submitted":"2020-04-18 02:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719750-1587177888185-e9ba32a3","meta_rev_id":"4c9c23cb9c5a53d07c0a5a2e077c0be1c355be6a","_number":719750,"virtual_id_number":719750,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-18 02:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2020-04-17 22:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-04-15 16:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-14 08:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2020-04-17 22:57:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 08:32:16.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-04-15 16:02:24.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-04-17 22:57:46.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-04-18 02:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fab3a55be13a68d09497c8bf17072e57eb491e76","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-04-14 07:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a75c18b678da7151cd2f24da445b4afad28cf8d6","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-04-14 08:32:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b6af5c7dc73a6dd60bc944b9d6cde86d4eccbc18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 10:14:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7732b8f40fad408b84fc7f1eb98e452a : SUCCESS in 1h 06m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e5df0658f6b473987e5191ca386dc5a : SUCCESS in 1h 10m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33530b7529154608b1768398a45f3a6e : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c945c02492d48fd9bd8c48d70bc34da : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc4b6c87ced3479cb77b8ee5716e8b43 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/045582e4cf574f2c9c1562e7c718037d : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e3a5f0246544312854023ead64d3bf7 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1d5a2db4f442d1bedd5b9612924970 : SUCCESS in 10m 05s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c0b88c2488e74406a717662fa0892b00 : SUCCESS in 1h 57m 39s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/60c83bc0ebfe4c119aa42242066cc8a0 : SUCCESS in 1h 44m 52s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e5bd379faf2a47beb98d0acc27dc36bf : SUCCESS in 1h 36m 05s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/dad261768d9c42de966eca2efbf4cdb0 : SUCCESS in 1h 40m 08s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/b76dfa27b3c542629f08a4afc18e20e6 : SUCCESS in 1h 36m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/291549252f524a5fae30134a3de0f155 : SUCCESS in 1h 54m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d554e6f663704571ba732b708238baed : SUCCESS in 50m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fd29c281e6e45e7950f71c876f73638 : SUCCESS in 1h 10m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b11ec55e71d74fa2933dbe57b6659a17 : SUCCESS in 2h 30m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/39b51aa6528b4a209c373f5a2d448d53 : SUCCESS in 54m 17s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d1ef9d3c3350438a822897700c511f9d : SUCCESS in 1h 07m 12s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fb2028ebc9443bfa9704d29212e9910 : SUCCESS in 1h 16m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/035d26c9ddcc467bad692267ba7246df : SUCCESS in 1h 21m 01s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/78bf70c28846461892349f5ea2cc7814 : SUCCESS in 3m 21s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f1181a435efc49df94f8512fba5f02b0 : SUCCESS in 1h 26m 40s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/82360fa19aa842088a2f25412190d062 : SUCCESS in 1h 08m 11s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/31b592d0cefd43c5bdb3a983e74120a0 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fe619e25974827d9dce06ed2548dd6efc0e48ef5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-15 16:02:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"984299beba3b4f8af389e860c23284c515dc7f2c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-04-17 22:57:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82907c299b7c456ecbce2bf752c515dcd9b89717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 22:58:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab510ebc9fea83b139e977058ee85d308c9ba237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 02:44:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/977f4a7480ca4a379fa02eade45fcd55 : SUCCESS in 1h 11m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/336950a5b3e64979951febc425fc20cd : SUCCESS in 1h 21m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b741ec4eb34564a3dbb783fb83f6e5 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b03908a0dc4d2ab7eba5e45be30b9c : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b72390608903496aa24d41d1550ca50e : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ece6f3e92345e0960f9d80b6e65f13 : SUCCESS in 10m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a7f1c4bc9a024d7f8fd86f56b78fae26 : SUCCESS in 2h 25m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/525d79ef31514407aa4a2726a12de5f3 : SUCCESS in 1h 15m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c93efa124db64cafb44aa628f744c0fc : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5889e92908d3d69a7b25dde20494b13d777621c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 02:44:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"747300b369afc3257222f3af6613bd9aeecfabeb","revisions":{"747300b369afc3257222f3af6613bd9aeecfabeb":{"kind":"REWORK","_number":1,"created":"2020-04-14 07:39:34.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/50/719750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/50/719750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/719750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/719750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/719750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/50/719750/1"}}},"commit":{"parents":[{"commit":"26062e4ac9dc6192aa053ba7d45d2a62d6f9475a","subject":"Merge \"Add glare and taas to blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26062e4ac9dc6192aa053ba7d45d2a62d6f9475a"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-04-14 06:23:47.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-04-14 06:23:47.000000000","tz":480},"subject":"[Trivial]Add space between words in log message","message":"[Trivial]Add space between words in log message\n\nThis is to add space between words in log message in\nwait_for_volume_attachment_remove.\n\nChange-Id: I355bc829e7c0f01613fc907a6bf9e03b785ba3ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/747300b369afc3257222f3af6613bd9aeecfabeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/747300b369afc3257222f3af6613bd9aeecfabeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
