)]}'
{"id":"openstack%2Ftempest~721154","triplet_id":"openstack%2Ftempest~master~I0058851ee5ae724be72a2edf65af09636078cba9","project":"openstack/tempest","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"I0058851ee5ae724be72a2edf65af09636078cba9","subject":"Use unittest.mock instead of third party mock","status":"ABANDONED","created":"2020-04-20 03:04:16.000000000","updated":"2020-04-20 07:34:04.000000000","total_comment_count":527,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af9da208036a2ad445b272c7f5d3050013f4e9a0","_number":721154,"virtual_id_number":721154,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"all":[{"date":"2020-04-20 07:21:05.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 05:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-20 07:21:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"ffb76f5cc62b36e91ea6a59d4e67810642a00105","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 03:04:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab648dcd9c8253532f3efc66bb429918e53218b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 05:19:42.000000000","message":"Patch Set 1: Verified-1\n\n(527 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2855644883e846289c2abda5c7481043 : SUCCESS in 2m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/788a0b504f4b4f8582424f6fbb5b8279 : SUCCESS in 1h 20m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3997110384c4a9abd54e025fef727a7 : SUCCESS in 1h 21m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1c46fde472e4dcc9f4944b9b425a455 : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87dcb3945b84b8b9332e003d183f077 : FAILURE in 9m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2adcf0dd492c4aa3a25b44f4c8ab9da1 : FAILURE in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2166984d2c0f4181b740c5b3472c72ce : FAILURE in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a0fe27e386243be98a78b02164635bc : FAILURE in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312aac0f6a20449f9cbb4b4f892ff31f : SUCCESS in 15m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fb2e81419ec459ca1fcc0fd46f8c729 : SUCCESS in 12m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3784c6a5702a4f2f9d4b2239c8fb7bbe : SUCCESS in 2h 03m 54s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/9dd9221aa9f6431aa2fbd8b8657c2269 : FAILURE in 1h 46m 45s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/16b01c27e7d94fdb9b0277b88df717e0 : SUCCESS in 1h 17m 55s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/023fb737bd2c4b9396f2a9570efc715a : SUCCESS in 1h 33m 59s\n- tempest-full-rocky-py3 https://zuul.opendev.org/t/openstack/build/255b2cede0204cedb1f22551db64d32a : SUCCESS in 1h 18m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/15796b0fe9c84927a6f632e9ee502a3b : FAILURE in 1h 43m 10s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4df1d1ec11884ace9b3e4a4887b9d392 : SUCCESS in 45m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ae76ee991e74c03991a24099e5b9600 : SUCCESS in 1h 00m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66dbe43d18224406ba6e8a7ce9b195f5 : POST_FAILURE in 2h 09m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/87886a6cf5c149abac78cd3705ad5ce8 : SUCCESS in 48m 43s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4dd74ab7db69448cbce50eafe372445e : SUCCESS in 1h 06m 26s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d5a5a579eea4b1faf5e0a4afcb83d4b : SUCCESS in 1h 18m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f68131c9cf3e48e8adea3cfe11288702 : SUCCESS in 1h 21m 48s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/933fccf20a7f43c7ae898ee74f1b1979 : SUCCESS in 2m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3f6623949015404bafd008007bc48781 : SUCCESS in 1h 14m 03s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/374753abe5ce4a9184fb1f492fe23202 : FAILURE in 1h 34m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ebd8f972215b43c8a0526b4b7e444120 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b14d2fffacd94def35ce4640b2423435d4f46d7b","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 07:19:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d30284529e625a562e0b0e75267132e578fe96fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 07:21:05.000000000","message":"Patch Set 2:\n\nConflicts with https://review.opendev.org/#/c/721021/1","accounts_in_message":[],"_revision_number":2},{"id":"b23c65e34259c93eaef54917eae4b28f71e7c2f2","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 07:34:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45c91c603612f8bc1b183fd5b084308bca7ca4d8","revisions":{"d9211c44dc26ec86f6aa4c40ba1f10d5e56c81a3":{"kind":"REWORK","_number":1,"created":"2020-04-20 03:04:16.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/54/721154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/54/721154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/721154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/721154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/721154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/54/721154/1"}}},"commit":{"parents":[{"commit":"414f2116a00e2c260191f7624a7273832fad212b","subject":"Merge \"compute: Add stable device rescue tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/414f2116a00e2c260191f7624a7273832fad212b"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 02:46:37.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 03:04:13.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I0058851ee5ae724be72a2edf65af09636078cba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d9211c44dc26ec86f6aa4c40ba1f10d5e56c81a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d9211c44dc26ec86f6aa4c40ba1f10d5e56c81a3"}]},"branch":"refs/heads/master"},"45c91c603612f8bc1b183fd5b084308bca7ca4d8":{"kind":"REWORK","_number":2,"created":"2020-04-20 07:19:48.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/54/721154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/54/721154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/721154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/721154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/54/721154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/54/721154/2"}}},"commit":{"parents":[{"commit":"414f2116a00e2c260191f7624a7273832fad212b","subject":"Merge \"compute: Add stable device rescue tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/414f2116a00e2c260191f7624a7273832fad212b"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 02:46:37.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 07:19:04.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I0058851ee5ae724be72a2edf65af09636078cba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45c91c603612f8bc1b183fd5b084308bca7ca4d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45c91c603612f8bc1b183fd5b084308bca7ca4d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
