)]}'
{"id":"openstack%2Ftempest~724631","triplet_id":"openstack%2Ftempest~master~I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1","project":"openstack/tempest","branch":"master","topic":"bp/tempest-scenario-manager-stable","hashtags":[],"change_id":"I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1","subject":"Need to have stable tempest scenario manager","status":"MERGED","created":"2020-04-30 10:35:18.000000000","updated":"2020-09-04 15:22:14.000000000","submitted":"2020-09-04 15:19:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":99,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724631-1599232784372-038e877c","meta_rev_id":"3150cb5c52fbce262f26870a3720392b54f5d877","_number":724631,"virtual_id_number":724631,"owner":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 15:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"date":"2020-09-04 10:48:22.000000000","_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-09-03 15:58:50.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-03 21:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2020-09-03 15:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-03 21:37:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 12:21:58.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-04-30 12:22:05.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-05-05 13:43:49.000000000","updated_by":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"reviewer":{"_account_id":31239,"name":"Paras Babbar","email":"pbabbar@redhat.com","username":"pbabbar"},"state":"REVIEWER"},{"updated":"2020-05-18 13:49:07.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-05-18 13:49:15.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-05-18 13:49:29.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"REVIEWER"},{"updated":"2020-05-18 23:14:10.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-08-17 08:52:14.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-09-03 15:10:55.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-09-03 15:58:50.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-09-03 21:37:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-04 15:19:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c0bee09c8d9ea7f82124bb3431bbf09a4711061","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-04-30 10:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba5ff5c6a6c6cd6a9d2f8b726e602180d7616cd8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2020-04-30 10:35:46.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"ubuntu-bionic\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"84bb80aaec1ab797e889c09cfa61ea6ded1e1ab7","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-04-30 10:57:41.000000000","message":"Patch Set 1:\n\nI don\u0027t understand, if \u0027common manager methods among plugins should be\n   defined in Tempest\u0027 then why you deleted it (tempest/manager.py)?","accounts_in_message":[],"_revision_number":1},{"id":"5480f1219c977ee4aeeeaffca47cb85987f4d584","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-04-30 11:50:05.000000000","message":"Patch Set 1:\n\nWe want to make tempest scenario manager stable, so I think we should define all common manager methods among plugins in tempest/scenario/manager.py and not in tempest/manager.py","accounts_in_message":[],"_revision_number":1},{"id":"ae7b7c9db937f12db1bbcbaf10b8b2280a62470d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 13:33:41.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9700c1b49b8c4e49af608ac05f4845aa : SUCCESS in 57m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfe2abaf3a3747b0b9e54f5e8f72f042 : SUCCESS in 1h 27m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e205e8930134f58bf32347309cffeea : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d8b26405c1e4f01a563bda5c4a7f22b : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72a2147ecd944cc591c56294bfed5e91 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3f5587175e846b2acce5ec546cc9293 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23625573305d43fa9934b62d513b0326 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf24499f8add430ba310c081e92653b7 : SUCCESS in 9m 10s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/9435c18a253f4ca980c345ef03ce6fa1 : TIMED_OUT in 2h 05m 57s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/de3de694624f4e269111bdd9ec7eece1 : SUCCESS in 1h 41m 47s (non-voting)\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2267758f6a3743dcb6c0397ca2bbe401 : SUCCESS in 1h 46m 59s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/ee32f33d05fa490c8c81e8178a042324 : SUCCESS in 1h 47m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c288c35dff66496aad10570217321ff3 : SUCCESS in 1h 43m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/fdcaba61f16e4f0ebc9822afbf751afd : SUCCESS in 45m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a08ccd06088741859f38c53c4385924f : SUCCESS in 57m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f95c49b9dbe4c859c67a212f5eaabdc : FAILURE in 2h 56m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4ead2e4ab4474d01a4582702f1bc3ead : SUCCESS in 49m 35s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/467b7cdab5e14a2ea22d1bf0b5be250f : FAILURE in 1h 12m 54s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7413970994e42b995e6f19b3373c77f : SUCCESS in 1h 07m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/bd68b7b45ca8473aa64bc6004109b8a8 : SUCCESS in 1h 08m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/67964e935d9d4ff8ac295627756724c6 : SUCCESS in 1h 21m 29s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/68bc1ae0c7ee4963b6d98083d280c5bd : SUCCESS in 3m 19s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3dfa36c9ab7e4ca8b69e3eb941301a1b : SUCCESS in 1h 31m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fe32de5c9b794f438ae15cd0b711397b : SUCCESS in 1h 51m 15s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0a35439c50444f8488992faa8dd62399 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"affab921aedd00c3d5293fb85f5f34323e4b0e49","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-04-30 14:25:55.000000000","message":"Patch Set 1:\n\n\u003e We want to make tempest scenario manager stable, so I think we\n \u003e should define all common manager methods among plugins in\n \u003e tempest/scenario/manager.py and not in tempest/manager.py\n\nI see, but are we sure no one is using tempest/manager.py? Personally I\u0027d let deleting methods at the end and first focus on implementing the methods we wanna declare stable for plugins. That way we decrease the chance of breaking something which depends on this.","accounts_in_message":[],"_revision_number":1},{"id":"07dfd1ffd9d9dca2d2bce361cc1548b8cbf193ed","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-05-05 13:15:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b0211f8aa40ee08125d36c2f6dfea91641d543d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 14:56:13.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2cb5e801b4bf4904a9df6ec92e851dac : FAILURE in 41m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00986f4337a64808940e37b9b5cb6944 : FAILURE in 1h 11m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77b0a0567f63422e9fa4b786b4ff5782 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766e02d339df4e90864c1406cfd786a0 : FAILURE in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8d47e64540c4e0eb08dee963f97eded : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cadf89acf7d34c5d96737fa5f2f6e8bb : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70d496312750448f9b68c5841f5fa6d1 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2fb863466749c1912c926733163a7b : SUCCESS in 8m 30s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/979a99939b084b82839f2fce98728021 : FAILURE in 1h 34m 02s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c6870df0b206465bbaa4dd440de27674 : FAILURE in 1h 03m 13s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2b03a1777d394624bfa1ec4289e74d98 : FAILURE in 1h 09m 05s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/72c5dca36b7b43358c064b25db4f4856 : FAILURE in 1h 03m 35s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/6a729900c3cf4af5aaa17a8d2c46bd07 : FAILURE in 1h 06m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b525ab1518f749ed8337bc404c899f84 : FAILURE in 1h 31m 25s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/55c820023a94411595e77d9f905f09e0 : SUCCESS in 46m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c9ebc81b82147efac9614d2c6f616e9 : FAILURE in 38m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16df62cc0a03409bbef4050f502044d3 : FAILURE in 1h 05m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/41c3825926cf4340b15f599a5f63304a : SUCCESS in 52m 40s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f88e016079934d328f08c7942c2bc665 : FAILURE in 53m 37s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d6931ecaa5648df9b9863de5dbcb8a0 : FAILURE in 45m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0d64a7e27bd246b3b560bd1788a55d4a : FAILURE in 1h 07m 05s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/36b46595a3694ebf8c200d618f452aee : SUCCESS in 3m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/00d22b8bd2004e3889ca99c0aa216bd9 : FAILURE in 1h 07m 41s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/92c356596bd64a3bab8b6787d1e6d893 : FAILURE in 56m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/77a66432743c41a49d2b68c210c2f4e4 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7da2b6c718e43fbc491c24aa1a7efcd742de4d1b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-05-11 07:21:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0738b385ad698cdf300d91e3625e738d4a87dee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 09:01:14.000000000","message":"Patch Set 3: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/eb80fe2688ac49a2bca36c83d8fc92cc : FAILURE in 47m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da30fef4d3594bdeaa01d4ef748924ea : FAILURE in 1h 24m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e40a0560c94d4152a066ea4f311c58c3 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f360d439184aa3b1de14033228848e : FAILURE in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be9b19fae34a433d974d10471cd59834 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e105ec579e8c4832b1b20902b314d082 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f6690d084e48979560471badc4c227 : SUCCESS in 11m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/5077cfcd21a44266848e72fba52a0c83 : FAILURE in 1h 23m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/68cdb1b233d4404e911758f0137a2da8 : FAILURE in 1h 22m 04s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/8ceeb2d89bef45dba3c3e2848465bc3f : FAILURE in 1h 18m 24s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/436ad22e14d447858462b5a45257bbad : FAILURE in 1h 22m 15s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/d7c3c5a50b8a40e1ac87f4a7da913724 : FAILURE in 1h 21m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/457dc9d8537441a0b0596eaf24cae9cd : FAILURE in 1h 37m 41s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a626798c7520410b8aa3a3d455a49570 : SUCCESS in 59m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83154e9f220443d0bff9d68a5a0099c8 : FAILURE in 53m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32cc5d151c224f4392d0fb28338015dd : FAILURE in 1h 16m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9fccc395a11340a7882bac03534e8b08 : SUCCESS in 1h 09m 49s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5348c7e463394aa29c2d96b1841d51c3 : FAILURE in 1h 21m 06s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe0d6ac725ba4a2c9d294f068b436d00 : FAILURE in 45m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/4f9b421847d44a7786b7c8ab89ab6172 : FAILURE in 1h 22m 26s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2ce17d1249e54ba38a162bf27d57f62b : SUCCESS in 6m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8e21a65b3bf34a26be35ddf8d28ff0e6 : FAILURE in 1h 00m 51s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e89ce5a4c480458192e6ded30c50f5f8 : FAILURE in 1h 22m 11s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b6fb8cf9a0994dc4a340a9c672ee6ead : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"022e2dedd8bb54ec05fd97646ca8bad9d26f64f0","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-05-18 13:48:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f6e859df5f23e48d70c494e9cf6e4570029e4734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:10:35.000000000","message":"Patch Set 4: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a967f15c1a1452cb71502b10c1aa02f : SUCCESS in 1h 08m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72c4ad85a7cd4f6e87a98d4529d1072d : SUCCESS in 1h 21m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5d98cd5f8d241ceb37eff4f3d79e1d6 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6294e22ddc54787bf711309418815e2 : FAILURE in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b9b241b1f4a45b0b2111dcaf1fb0511 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0d84d7277c74ddfa45aae0473120850 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a414fc552b6f49cdab1bdb9e9de42f1f : SUCCESS in 15m 44s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2b808dbf84bf42799c9a598e72d917ca : SUCCESS in 1h 40m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/06c6486b189f486bb502dbe23c32eed9 : SUCCESS in 1h 27m 33s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/105aba0b52894347a7ef313d3d1389c3 : SUCCESS in 1h 38m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6f3f0e4d98de496d9b775972debf5f7e : SUCCESS in 1h 19m 01s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/62b1a9073f114bc19afd00a63953342c : SUCCESS in 1h 35m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/13f4d6fcd69041308c3801523587e6f0 : SUCCESS in 1h 34m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e0915bbb8bbe417dab7fad1239774db3 : SUCCESS in 47m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73449f4e10d746c2bc8148199fdd6b45 : SUCCESS in 1h 11m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/382cd96703c24ee98ddc8a3712e24998 : SUCCESS in 2h 16m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9c5649cbb83b4a8095030847651484a1 : SUCCESS in 48m 28s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9d4025040618446c8c7989fc0387947a : SUCCESS in 1h 08m 13s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4414df7874547309b6e4877dc2c2c84 : SUCCESS in 1h 04m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/49bf29eb2cd94c11b9760cc1d8f3ebb0 : SUCCESS in 1h 16m 13s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ec6ae6c4bed244d8ac96428dbfb2e3a0 : FAILURE in 4m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/01fb451019c846eea018c07dab822ccc : SUCCESS in 1h 40m 57s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d694578b3f1c4e2e87913f91dab06d89 : SUCCESS in 1h 02m 44s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5f72f31e248c46e1a9cde718872e8be6 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":4},{"id":"ce249421fed179efe7467c26f929f18861fd04a0","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-05-18 23:14:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nSoft -1 because I put some comments and zuul pointed out some hacking violations.","accounts_in_message":[],"_revision_number":4},{"id":"9d535c5f183f2eb7905be7a257c305cbdc954c14","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-05-19 05:58:02.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nThanks masayuki","accounts_in_message":[],"_revision_number":4},{"id":"0967c23ab4c7b8dfeaa5cb4f864b74af1a839444","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-05-19 08:12:36.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nBesides some small nits, and the concerns from Masayuki, looks good to me","accounts_in_message":[],"_revision_number":4},{"id":"a97c21f3e53f8c3fe8c62277ccf43fe512c9926f","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-13 12:40:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7319940981fc071b3e7aaa1f5fb7fe03e0726e78","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-15 11:06:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae0e27a2eb9ec55e6e606c492c5d7b38ce9d70ea","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-15 13:42:35.000000000","message":"Topic changed from temp to bp/tempest-scenario-manager-stable","accounts_in_message":[],"_revision_number":5},{"id":"12337ee547e23f1188f7786f2d78686c302f0185","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-15 13:45:06.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"827c11160d037ac7ccbb9bf9b6333120f92d38c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 20:13:03.000000000","message":"Patch Set 6: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f358b79bc5754a2fb2ff974d51eacc99 : SUCCESS in 59m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c41b36bc27e4523bc61c2b6c5a5bc0a : FAILURE in 1h 19m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/001fc22106a645ed8f32ea9fac0a7b18 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c52e844f16e4005be94f9d86b158780 : FAILURE in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cefb4cb17c5948c5be26482ce4144812 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d924ac7bfc1445dc8d2da0bc45d1fa64 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca645872eb24be8b91a10f7117a015d : SUCCESS in 10m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a68b0cf6cbb042efadc14049d528ef92 : FAILURE in 1h 51m 31s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/56022e2e742640db8cff583fdfe7ae99 : FAILURE in 1h 16m 05s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/e62a20d8fd934445b4c1d2d30ddb33d1 : FAILURE in 1h 18m 45s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/8fa1fcd68ea7423787a9e1dd3f4ec00d : FAILURE in 54m 40s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/d30fcb3f9f31458c92df0a824e67713d : FAILURE in 1h 12m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7b9b3385840b45309d68322273a2881f : FAILURE in 1h 42m 17s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c70f2de61b1f4f46be19659b7450d07e : SUCCESS in 46m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdbe1fe2e5f14712850d8098b1abab2b : SUCCESS in 1h 00m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab581c53f7d04cc6a68077990f883d6d : FAILURE in 2h 07m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e86f0b9d858c4e3e85b58fe4484724fc : SUCCESS in 57m 51s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ae230ba77afd4436886327d64096a984 : FAILURE in 1h 05m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8a4b73bcda24c21993688cb0394c6fb : SUCCESS in 1h 15m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/eacb9a6d055d4b42b6cc4e8f3de99a6d : FAILURE in 1h 25m 15s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/18b744f72fbc411992b2cef2147c34b2 : SUCCESS in 5m 27s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c48b2ef8584d4a1c896c0a5f02a8cf69 : FAILURE in 51m 45s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f04c4e800fe1483fae2f245671ed0ddb : FAILURE in 57m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/08b5550f7c054f62bd0a6b9411c50fa9 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":6},{"id":"71e11cdb39997d24bf2989455bd277581c883a81","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-27 11:14:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"47b1a6a105e21adc666131c8456fcef565c4f40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 14:07:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1ffa96a435246c4aba2f17f32e07b99 : SUCCESS in 1h 05m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abcdf418c03244ec92f707f0eb2415b2 : FAILURE in 51m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee393242210545dbbd3c0e20ffe3dedb : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57cf1ef9bac49ed9aa26139989550cd : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1006f514b80d4e6c98ceb33db3d4e33c : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cdc3be1acc0431abff69d258a179345 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b47b52b85c8a4d75bd1d30e30d16a6a4 : SUCCESS in 10m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d1cb684ade5e476b886a1d6c5fd69304 : FAILURE in 1h 46m 13s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6978e80bf7984e80bdc83d28d6c44da5 : FAILURE in 1h 23m 28s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f81f7d30bc924dcdbfd68a90827119a4 : SUCCESS in 1h 27m 10s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/d9cfe9cd761c4ede9969f8030044977d : SUCCESS in 1h 28m 06s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/7e034141402843f2a29bc19407214a3e : SUCCESS in 1h 23m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f3e3999e5f8c45769fc810972be5bdbe : FAILURE in 1h 12m 32s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e8cf7b58e73a412282b2f9f09edde3f7 : SUCCESS in 47m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6d23395528546d9b846feb04d422602 : SUCCESS in 59m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d6f084f43564c588cf720b6c9b13bc3 : FAILURE in 2h 02m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/905cc455ba3b497796a110c3431b5bec : SUCCESS in 56m 24s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3452f24cdeb34392a5679c6e8df6451f : FAILURE in 55m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6995b203f2f14b9ea78a0ffc83327aed : SUCCESS in 59m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a127cf0706864f8fad64721fa517af07 : FAILURE in 1h 16m 33s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e888ae5156cf47c1a5c434207ae79ad0 : SUCCESS in 5m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2b5c456c6f9246daa131c89edfa9b972 : FAILURE in 1h 13m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ec83999ca19446ab9a5003affc0f592f : FAILURE in 1h 01m 29s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e4f4b4b6010e42e8a1361263562ebadc : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":7},{"id":"95d7e0a20fe839532b74532ca626aba87123209c","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-29 11:18:22.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"5b7f284c4141c7609e3812f125e15ee90c4ca9fb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-07-29 12:57:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"98ffd69a754d6e1160d0c923fa6ff4e022b86cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:09:31.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0573869be2c34d8ea59b28e5ea899955 : SUCCESS in 1h 01m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/309218729da94d0dbc14b8af35e02943 : FAILURE in 1h 18m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a986f107f348490cbf17f78d7c712a05 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adce9b2f1e0546d4b0ca3d64ad4a4992 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c5f83f234e64cc8b444f06450d5f721 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26bba2b60804414cb958d153e569d04b : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380097474f64487c8a907877361b2783 : SUCCESS in 9m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/baa0956779e24ab691ab2231716fdfe1 : FAILURE in 1h 38m 07s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c1847e7e179e4b239cf0edc645fc9165 : FAILURE in 1h 19m 14s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/ed855bb370b64a7e821fa60cb19fe463 : SUCCESS in 1h 15m 47s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/137018133d954d6daa1c0108cdbc831a : SUCCESS in 1h 27m 22s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/76f9bb83a9d1403abd7cd995ff4860b1 : SUCCESS in 1h 24m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7c384ccc4c314a47af7d00485672131f : FAILURE in 1h 35m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/caeb8257c7d947f28938e43e877b5c88 : SUCCESS in 42m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96eb32707ac945fb815d37437344beb1 : SUCCESS in 1h 04m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd82f63842534074922179950f65e7f7 : FAILURE in 2h 03m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f93406d26c0f463fa2a5f07df20d3e3d : SUCCESS in 41m 52s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c147479dcd9b432bad96936856f334d3 : FAILURE in 48m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d0365c8ddf04601949b2f3aafeb496e : SUCCESS in 1h 14m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6b4c1f2cf8cb49f4a559ae767bb1572a : FAILURE in 1h 21m 16s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/89ccece9cd5e4a5caf6738ebea687c14 : SUCCESS in 3m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4058c5ad52a949e298629edad174b8ef : FAILURE in 1h 13m 24s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8351cecac6fe4c19afc5f13b6b157c73 : FAILURE in 38m 41s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b4732e8938764c628008f6789b84ac11 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":7},{"id":"3a56c0b52fecb7c880a95abdb8e1d76d083b60fc","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-30 07:29:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fe50e9de895a063dc3218773285e62b8fe3017ee","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-07-30 13:27:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6f122cf0477f6b2a50f05bb8ad57d20566d9e746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 20:25:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5fd923c4138e4a979b0f9b3cc0fc6fe2 : SUCCESS in 1h 06m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7011a96102c345ecab41fb29b1d2a84b : SUCCESS in 1h 14m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb46c69490a549a287ecc790a3f46990 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65fa95c1896f4a15bd85acea258767fc : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833a519e868b4a71af0bda7f55e7e648 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/550d78b3462241d99ebf7773b2f8b1f7 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b3808762c8463690caa71e7b452ce2 : SUCCESS in 10m 06s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/6a040e70d1994d50a3fd989031cf021b : SUCCESS in 1h 39m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/541db8b27fa147f7900e7f3ace797555 : SUCCESS in 1h 51m 41s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/92791e9f736947f6ba6e2e886b13f824 : SUCCESS in 1h 31m 31s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3318a72ff10a46e9b03fc2d2450b60ce : SUCCESS in 1h 13m 18s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a6ca3fda82c44ec1a5fa1a019945b843 : SUCCESS in 1h 06m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5ffcbcb832d94b29818656b8ba3dca48 : SUCCESS in 2h 12m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/07f6da20cba04a3ea6e4a7c1a38238a1 : SUCCESS in 44m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a8a3b24510a462eb8ee79e7023de8ea : SUCCESS in 50m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fcb06a3cee324144bac327be7df6d459 : SUCCESS in 2h 26m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b3cefcba15f64757b440ba8b308817b2 : SUCCESS in 53m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bc2264f83c5b4f949354aa811238c1d2 : SUCCESS in 1h 00m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be2a73709b5e4c14a9ed6b6525d0bd33 : SUCCESS in 1h 09m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/299a3aa7ce914e68ad8117d2f69d4354 : SUCCESS in 1h 04m 55s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a3ac4be00939445bba501e776c992dc1 : SUCCESS in 2m 52s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5e9e4ee5630d415e842f97efd7f28dff : SUCCESS in 1h 37m 33s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/11084c3366044232bc61c0a2eea19e61 : SUCCESS in 1h 13m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ebbe11057b194127a20bc1c933b13bc0 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":8},{"id":"8ee1a04a0105fb9d7f467863ba975da440706c7e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-03 09:33:46.000000000","message":"Patch Set 8: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":8},{"id":"adfcc8a4766afd3577fef122b8d57e789baf48cc","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-04 12:24:47.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"91bb13ac84e5f3b6906a80d5f72b2dfc0477eb5b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-04 14:01:14.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1b4681ede3e5787a1e8fa07adcf1fb83af026949","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-07 10:46:39.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b1b588678b0e46e3582b71eef64271fde80c9a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 13:45:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/39385b41e3324867b9638150eaed3f52 : SUCCESS in 1h 10m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56a170dc161e472aa1d5bca7f7faa408 : SUCCESS in 1h 27m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/293e7da45f194c5aa820aacbd06703ac : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d83a200043e47bf996d13f836b24ec2 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c25a06e1e934f2cb84cc30d5e6f4df2 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dfe710096fc415ab3f282cf83df2f8c : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8fef7f2c7f4d08aa65dc7b33d78e8d : SUCCESS in 13m 40s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/56269837c8384a949b184388e6e010b5 : SUCCESS in 1h 44m 11s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/751c9f326f4046e69261436632f25e76 : SUCCESS in 1h 43m 18s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c5aed364f0fc44d7909a5d49d7f4b681 : SUCCESS in 1h 22m 40s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3eaac5feaa5e4185a4152ceea188e08c : SUCCESS in 1h 21m 50s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/8f951584ee5042c0b9d5fb4c17bbcc51 : SUCCESS in 1h 28m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dedf73341e5f470482b083686a483982 : FAILURE in 12m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/eab73240a2fb46a88fa0d0f9eb3aeacf : SUCCESS in 46m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffef5d6a30e04569aa5206b1fee46453 : SUCCESS in 1h 01m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/576b99e0cc874afdb981c717465e22b7 : SUCCESS in 2h 33m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/859dccd2fbf14d20ba903efd14ef0ef2 : SUCCESS in 53m 14s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/32363a816aeb41658cb0769746c394bb : SUCCESS in 1h 05m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0df6d0dc66164bf58785f0ce2f5939fe : SUCCESS in 1h 10m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bfa2b4ee022849eab4f0ff87f6df3058 : SUCCESS in 1h 14m 10s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0f163d777b56453394d678918bad1eb4 : SUCCESS in 3m 59s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3932143e24fd4f69839bae20c7b3f1a1 : SUCCESS in 52m 17s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a63fac46ca884af9a85de234cfff54c1 : SUCCESS in 53m 46s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/00dca5c36ba2490e992ba294468214c7 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":9},{"id":"9966c4dff5d314fb685b971a65d885cd358d760a","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-10 06:36:24.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"065dc8294ce83c53d7593777a7c89abab254661d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-11 14:16:12.000000000","message":"Patch Set 9:\n\n(14 comments)\n\nthanks for working on this, few comments inline","accounts_in_message":[],"_revision_number":9},{"id":"7ba44e4f5dff0f716ccf075b6b6cb8f40dfd8cd1","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-13 13:50:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c7cc808d89d0a985f32bfbef7aa018c21aaa33fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 16:01:15.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d50317c3b58248b695aae19252476658 : FAILURE in 36m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab90fd25d8ae497c8748ee13b17c17e5 : FAILURE in 1h 06m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a79d23bebee411499a4032e4312c82c : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349561198d9240e5ad0df3afcb8fa5a5 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69310d310235499c9463968715eb3e18 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/759fbf5fed43485e93d984a9f199fa9e : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9a4c44403c425c813fc775cbdd54df : SUCCESS in 10m 47s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/47537a75e0fe46b1930c8436262d5839 : FAILURE in 1h 07m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/8aaa1493a403498a83a7ed0735426dae : FAILURE in 59m 50s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/5b2e818442424ff991522ca396eff5a4 : FAILURE in 1h 01m 05s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f45a8ab04a744c92ad2fb0bb33d6a0fa : FAILURE in 1h 10m 57s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/b805e97648234c7eb76d013c763f87c7 : FAILURE in 1h 02m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f1f70b107b184c93aa2573bcf18b0fe7 : FAILURE in 1h 28m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/bdabf12326894cc0bf9bf7e18bf3a742 : SUCCESS in 46m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f582b3a78bd244babe7834d41f498a0b : FAILURE in 51m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a0480fe12f34f3da08ef3813bc66147 : FAILURE in 57m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b90c1b37bb444c5faafc04e311063548 : SUCCESS in 54m 35s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0adadc73a43a4b70a265eaa5b37b8028 : FAILURE in 1h 08m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa562a8ae4cd403aa7948f6b1329cf7a : FAILURE in 51m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2fa78f5707004302a771fedb4df5fbb3 : FAILURE in 1h 20m 37s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3ad8cd883f6b4f22b0012b6d1761a505 : SUCCESS in 4m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9d28b22c11674780b24038d3bb3d0996 : FAILURE in 1h 25m 46s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e5d3cfce62c84b8cbe9c3ed6fa52b884 : FAILURE in 1h 02m 37s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6b78db0b5ddb46629c806c9f2def3cf1 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":10},{"id":"d9c53fbc3bad8ff0723b1be2db16835ee9495778","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-14 08:05:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d8ac821721f6930a8b51222f5d61d58f1c69735e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 10:49:19.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f63d3cfc8b2b48faae6e61189ff92c1f : SUCCESS in 1h 02m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e849b3f33387491cbc9456d393dcc553 : SUCCESS in 1h 16m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/176331abddde497db62ebfd7d42a287c : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f8f6f126134f74b8094ad26de1b393 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0908823188cb45b08ce0305465006a67 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/facfbecf4ad14bb68c162c2bfcf3e163 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a44ac5cdc141129c5a3a51c37035ad : SUCCESS in 9m 57s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/32a68308c961493188bd546a6fa5f479 : SUCCESS in 1h 38m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/255f5f301913445ca6c0f0a4970124fb : SUCCESS in 1h 26m 36s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1653f982142e4a4dad9b91eb5521da6b : SUCCESS in 1h 31m 01s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3723ac0b3e574199a515eb4868a11299 : SUCCESS in 1h 15m 28s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/db617a9fecfc42cf875ff5d4aa034e2b : SUCCESS in 1h 17m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/976ea1e06e174dbabd69bef835e9293d : SUCCESS in 1h 05m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b23735da5bfa44fcbdb52f037659f584 : SUCCESS in 42m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0dad6b75e5b416890fb6d6421ab9c35 : SUCCESS in 43m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6df316d37557477ab2dda97763013822 : FAILURE in 2h 30m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ebfe759ac8264f4bbc653353560f91fa : SUCCESS in 35m 39s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4e0fe4f4f8d7433e9d10604fe0c25b46 : SUCCESS in 1h 02m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2574e1834184e6b9c7c7734f06fe8ca : SUCCESS in 50m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d154d6730e8f40cfad9c3735b752d567 : SUCCESS in 1h 05m 31s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2e588e1f6eea4d2488ffc9902bb8ef49 : SUCCESS in 4m 13s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6acc396c5b3244c184cb326d85aded1b : SUCCESS in 1h 04m 04s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/cd91b1866dce488d95d87d095099b913 : SUCCESS in 54m 56s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4f99286fbe164bb6ba144edc9a73b8ab : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":11},{"id":"b679e258077c9462e2e68f39093357b0c0096e29","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-17 06:55:55.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"903aadd0e91803f317e276359b848634185780d8","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-17 07:42:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"25e009542a13539c7f79ed8ddbab1f24f72c1474","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-08-17 08:52:14.000000000","message":"Patch Set 12: Code-Review+1\n\nLooks fine, let\u0027s see what god zuul says","accounts_in_message":[],"_revision_number":12},{"id":"d1d5ed64ac3c0a29bb3e69ddf02d06fd6fecaf7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 09:33:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/05975554a98e4391953f8dff7b439224 : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23636a9b7a004ae48708e44492470db6 : FAILURE in 1h 12m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79e9d2ba897344a88523b50adea6c6a0 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9bbaf37ee4a4c359c4e154b40c63e82 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e004b491168450a81d2d90c7ae5758b : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b73159876b144b7bb8d158335b8a001a : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d652fcd6c5eb424ab5df656da8e3a82e : SUCCESS in 14m 18s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/216b311f0ee540369d3796fb80f74ace : FAILURE in 1h 37m 44s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/664063ed29b84172bf4731061b36d19f : FAILURE in 1h 10m 58s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/39f03165acba4e6dabe7080e4e733efa : FAILURE in 1h 11m 55s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/18d7fc5098c845a89b7a315b4afbad15 : FAILURE in 1h 07m 13s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e5a0ead791924133a2914641223b73c1 : FAILURE in 20m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d5c6d57575254b4fa0915454cd359f01 : FAILURE in 1h 42m 36s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d0c9a433a46b43aba3bd3195bc6cbda5 : SUCCESS in 37m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1f2c71d31a34217a1e5e53990b3dd5e : SUCCESS in 52m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3e5ce26c20e4d7c82bd2f8a91271541 : FAILURE in 16m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ba8eac3ee5044867b57767f099d6db19 : SUCCESS in 50m 35s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6b1957496a7d46e6b18be46c1de9cbd6 : FAILURE in 1h 00m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d762f9fd6baa44aab2d6791b66fef787 : FAILURE in 26m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/9e85137560284543b31c315bd08b30f7 : FAILURE in 1h 07m 12s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c6caa18abe514134bb9106235dd04bed : SUCCESS in 3m 13s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/45fdc7206fcc48e188123fce8e5ad71d : FAILURE in 13m 57s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/13cde8f0d0b24b42941e3aac4e030cf4 : FAILURE in 55m 38s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b21eab3893a642d6ae402e53a0adc03f : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":12},{"id":"3f3bcb9228064bacc8ed0e560304f94daa04534f","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-17 13:17:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9ad58b84f423a4c93742d33012503e71c7997a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:41:20.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8fb7609934cb45ce968e6e806e293365 : SUCCESS in 1h 01m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/660677c6cbe94e538468dd76a9d5e3e7 : FAILURE in 1h 02m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8180217ecdf4e31a7ad66f298b754d2 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e003c1896806430d80c63b8f6a3fbdae : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8d5126bd7bc47729bd84a455ecb50d5 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18b470699d7b4e3baa93596d711d0f1f : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08da937fa6594c2a857a1c3bacf076c9 : SUCCESS in 14m 25s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/5c879b544514480c9eb204fd9c88ba71 : FAILURE in 1h 40m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/f40f88d172f24f9aabe11384151ecd7f : FAILURE in 1h 19m 55s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1552824f21184f21a11aea3c66b5799b : FAILURE in 1h 20m 48s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a63b7279b752450d986ac93069684a82 : FAILURE in 1h 23m 16s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/0b448a982d7c4b269646c47901ac102f : FAILURE in 1h 17m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fb6804dc57ed4586ad875647bfa7ebd1 : FAILURE in 1h 31m 13s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ed880bdfbd4741fca88a0cfecac9e28b : SUCCESS in 45m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1893206ef354085b085fec6ec867832 : SUCCESS in 1h 00m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/993583560c644fee94dae436e85e58c2 : FAILURE in 2h 13m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/767155b0b3a8416d8b93e167e546275f : SUCCESS in 51m 43s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/699194a75659444280e8811654152d66 : FAILURE in 1h 01m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d858e806631487b9ea461522269bc4c : SUCCESS in 1h 05m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d6ab15717af247f4873cdac26fc4bedb : FAILURE in 1h 13m 39s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/98a3d123dc944055a03988290a55c033 : SUCCESS in 3m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/78a8a61aa3594a75999f2b35004cdaf0 : FAILURE in 1h 17m 18s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/875bcf7789234913a02f5fa0743282a1 : FAILURE in 56m 04s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/da2958ee91d44a1f906745e40956aabd : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":13},{"id":"224aa7632b47ec294a2556c7e0a0b0ba10a05121","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-18 06:55:45.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"e3afc8e0026d0fdf4e8353ed200942115547d7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 09:29:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5991e76020804c0897b284cadd15ee76 : SUCCESS in 58m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5944de1e20c4038819f5b4762a75caf : FAILURE in 1h 20m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dffccc1a61a4afa907e66aaae57982e : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bab148d6148405c814e7b517e1f2bd2 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/626b9c66a57648eba17494b5e5eb96e5 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17dd7c6d17a9402386d616d69fdd77e0 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a2b2430574340248b299897060020a1 : SUCCESS in 11m 48s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/9d9eb74c7bc64b11973137631cc4251f : FAILURE in 1h 08m 03s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c9fcfcbbb3d74df28b5325a71bc84530 : FAILURE in 1h 21m 32s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2fce539cd54349ddb2d9f70dbb8928ea : FAILURE in 1h 31m 26s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/531169d4fefa48debd08d76ec212374f : FAILURE in 1h 16m 56s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/16a5c94aa3024de9840bb9964a184232 : FAILURE in 1h 11m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1c8e31f724fc4e359eae218d99b43124 : FAILURE in 1h 28m 34s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/36975581a72247978bbfd47dad9aae0c : SUCCESS in 42m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c256908bc7d4c9aadd6bfce740c2cd4 : FAILURE in 10m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/95d2b12b91a54d6fa8d0c165129912dd : FAILURE in 1h 55m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3f7d763b04854371a14d67fc957dbbc3 : SUCCESS in 49m 21s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0146c365188549f5a4789577128b91f6 : FAILURE in 1h 04m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e790b1686993437dbe4e4285729bc6e9 : SUCCESS in 1h 03m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/81d5f13569a64d12909a07769e75080d : FAILURE in 1h 20m 45s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9498cc4764b24fbc9dd3f5583d8e344c : SUCCESS in 3m 22s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4bba5e91828a46f48c7101e7a0ad42bf : FAILURE in 1h 03m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/790aacffe604498e8e4aca0d6a576040 : FAILURE in 53m 53s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7b13bda5aeb147cc846426654b34c4e9 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":14},{"id":"cd0f1977e953d5911da2aa1a54982345a65e0be7","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-18 19:39:10.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ed6e7e16a90c30ee1387640c0f265129c8fb2877","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-19 08:24:15.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cc536891c80ad420785e21b4d16a7b17b04e24e7","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-19 12:53:07.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d95a6ce91ced1ec3b9b34a4e4ec3a3f294e1cd0b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-24 14:26:58.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"9c911d294bd78e21395f10b2e9911338ca07d99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 18:27:44.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a62085def75a44a3b1e00163767eabfe : SUCCESS in 1h 08m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5642b06228f4cbe92e129ae2a6c0a9b : FAILURE in 1h 16m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fefef5be577409691de78323615040c : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd0a487e2f8546c582b8f9acc46f5b2c : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47e4e80233d248f5bb377cdc17de2e89 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de01e96ef9a4812a41f488acf34b40d : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb3ed915dba142e29a438a8880e4c41b : SUCCESS in 12m 11s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c77115ae53a0480da1016994c943924f : FAILURE in 1h 53m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/934f2b614f0345069ac9db04c5baad23 : FAILURE in 1h 11m 42s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6d64a0211b564191b294b1af6c091053 : FAILURE in 1h 15m 07s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e5481873c2ea4d91b357d7735f6e2d02 : FAILURE in 1h 26m 21s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/b373904a48a7455c931116bcf2b0814e : FAILURE in 1h 19m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0e1f40ae1d8f41c281a39e02aed5d331 : FAILURE in 1h 33m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e7ae85096ed2495c987698faf34a533a : SUCCESS in 41m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ff8898697394f67bf09c0b232dd51a7 : SUCCESS in 54m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f1507ed51fb405780529daf73b96d7e : FAILURE in 2h 02m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fbe6131a23eb422fa6062bc8369f2797 : SUCCESS in 52m 32s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e2bfdeda5a814498a18d4fc71f30ef34 : FAILURE in 1h 03m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edf8b074ee31476cb134a6eed70dde3a : SUCCESS in 1h 09m 26s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2fa398bb6e6a47d39571cf130eb60f78 : FAILURE in 1h 03m 06s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6b4b1f8ed8a44762bdb09a7ad836b173 : SUCCESS in 3m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ddf76647e8324db98941563bef85c901 : FAILURE in 1h 15m 36s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9f4d7a2a66fd46619cc202ed3220cc64 : FAILURE in 55m 59s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/44df54855ae1406ebb501b4e979c9137 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":15},{"id":"076c5c01ec0f3e4df8f61c9e09f42d052316933f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-24 20:38:44.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"15e551d9d52588a624832a5891c598c72b722a29","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-25 11:40:46.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"277ea93a9d465545130216106d1ff869d4b92505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 16:33:13.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7ee2729e8e074005828793a511209c5d : SUCCESS in 1h 07m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9be28ff427ec49b8934f2b98e5219338 : FAILURE in 1h 12m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a17223dfb7644c6593a8e69d8769c3a8 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b17933bca9049338dead6ed1cc2b74e : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14771dde4db046869e5ec112605974ab : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42e04c2fc0c84a958c7fc5e3847bab8c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8196b7b2294fb69a44c2dd549f5e0c : SUCCESS in 12m 36s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/bc31ca06543341caaae98c45e1d6d69b : FAILURE in 1h 45m 26s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/16c04e4a6f124a4db4c24fac0045c315 : FAILURE in 1h 24m 45s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f274d7fd100b4e36966395423a916671 : FAILURE in 1h 16m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/8a62560680ae486799df1131fd6888a4 : FAILURE in 1h 17m 45s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/08a91f035cd04f9686adcc4077e0dbee : FAILURE in 1h 15m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/625e8cfe082146919a483ad5e6b8e3a7 : FAILURE in 1h 35m 45s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/cf0d129389be43d98a494c1c1c186d1d : SUCCESS in 41m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00f3efa132a8417db655b717338354d7 : SUCCESS in 51m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/604bcdcab8f64e0cb507ca88cb824a8e : FAILURE in 1h 56m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8594741b1b7546059050ad8a6c0e8e04 : SUCCESS in 47m 16s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0b2f046aa2e0415985ddaa63a2c70ae5 : FAILURE in 1h 10m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/82adacb487f54d9b9147c6095b6300ab : SUCCESS in 48m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/143fa72253d047efa24c78382c60d6de : FAILURE in 1h 04m 43s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/28773c9cbd9c44efad7aeb5e76e984a3 : SUCCESS in 4m 22s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/686d19c9b6af4c49b29df0d8f4e07f05 : FAILURE in 1h 05m 50s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5d8558aeeac341cfa7cb9168d4a07486 : FAILURE in 55m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/25dd1560f5db4f3fb50b1cb38fd3ee1d : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":16},{"id":"368ac3ec267b97c5b934d5664434652f82f9b454","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-25 20:37:13.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"d2da8805b3a09d6e27a4466bab9cd1188d5151b2","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-26 12:47:23.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f3b6ae5eba7cc747073e3a50a091d1293d5fc79b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-26 13:33:14.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"3f8ce0c5b160c05e049493206d4f7a8c7f44b00e","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-26 13:48:17.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"80cd582778c2952c7275879a4d9f81b23ae1e60d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:37:45.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/51bb08af5ace4569bfc9f9d605191058 : SUCCESS in 59m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d04d1d13b041450cb3088f30eba795bd : SUCCESS in 57m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cfc9f8cb29f4ca790d3bc893b5f2001 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2818cd477f82451e808b226cc948e410 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6369322750043f6b953d5eee9b68cd6 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a236b4c1471a4d90a075ee29b6921fef : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f0149cdf9c429ab8228214230a5e4f : SUCCESS in 13m 13s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2066677b617a40b395cde2b1149617a1 : SUCCESS in 1h 59m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b21189f00511428194e0d97ca9307e7c : SUCCESS in 1h 17m 47s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/571da8a48c8b45319047ed1151fac8ce : SUCCESS in 1h 26m 51s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/12eb6f97acd54334b370b6016d6c66a4 : SUCCESS in 1h 31m 35s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/96c3d9874a8845bea5c7b55dce332e2d : SUCCESS in 1h 09m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e1b8ba7b24a8495fa1ea7d8160c9faa9 : SUCCESS in 1h 23m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/94f6183583ac4de58dd41e1e2e58bcd1 : SUCCESS in 40m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45b8085778ec411689e55e8720d662f3 : SUCCESS in 52m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5eaf241a82140ccaf4f22b9e89a43f0 : SUCCESS in 2h 06m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e2dbea793cd342dfac2aaaad68083b90 : SUCCESS in 1h 03m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8e9f9913b07c40c097ee94263f8510c6 : SUCCESS in 51m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51aee3776e4e4f56921fa4d5f1ff79dc : SUCCESS in 1h 15m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f4d7d9c914ef4ec48b41a60e11cf29d6 : SUCCESS in 1h 13m 36s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0f8502d39fdf4cc6882f43d13bee0621 : SUCCESS in 4m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d78c7e4d733f4af99cb1b2b8a9e796ad : SUCCESS in 1h 02m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8b9e36624429484a921770b499bab4cc : SUCCESS in 1h 29m 59s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f7ae319521ed41649c77fd072af5e484 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":17},{"id":"ea7ab0c452ff1c5ea08ec60fe277766656a6ecd3","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-27 12:11:56.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)\n\n-1 due to the \u0027thing\u0027 usage in scenario tests (see inline comment) .. apart from that it looks good to me, good job!","accounts_in_message":[],"_revision_number":17},{"id":"f5be78c0fbb485e3dabc6b3cdc07eff79867b97f","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-27 13:22:32.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"98c6aedf1b04ea3c050b0e6f247490da1fe79e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 16:40:33.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d1e4223c77d4f1da02fc506ba82d39e : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/670998e427964776bab94ce05f5a4ac9 : SUCCESS in 1h 13m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e55b871b554ab1a5bfd364e58f8a7a : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3907dcdd0c8f42c39309f15331c3ce04 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb893dc3a3cd48abbc2a2b9d7bf7fe1f : FAILURE in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b750471e99cd48ab927b6f3b92122d2e : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f138f706f9cc462791ccdc3a2463fbb4 : SUCCESS in 15m 42s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/92d348a12d6b46758377cdde75113aa2 : FAILURE in 12m 27s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ac2ba8dbf60b4e69b700e3f30de30b6e : SUCCESS in 1h 17m 29s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7a5b54f9f831475cb7a85e8e0e020048 : SUCCESS in 1h 29m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/843fcb1a57db43e09f1d77e2d3a9c514 : SUCCESS in 1h 22m 19s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/54a7e1dd470144d5bce8989da565b301 : SUCCESS in 1h 16m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/318b348006934221b7cfeb394c57b9ab : SUCCESS in 1h 49m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e89d75b2587a48a1991ccd7eae91903e : SUCCESS in 39m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfbf78e41ecf4cb7ad714ae5a4430b01 : FAILURE in 52m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/260652e863a84b2a84416de29a7ba63d : FAILURE in 1h 53m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/47a93b1582684891a9bf26d2c3d95ab5 : SUCCESS in 51m 28s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2517d14448504a448f7595f34fb1b46c : SUCCESS in 1h 04m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/18cd6a0bde184401bca466c159c5432f : FAILURE in 41m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/abcd6669a487440ca139ba22fee7f6a4 : FAILURE in 1h 01m 46s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1d5345d76e4c462294cd3091cf799f01 : SUCCESS in 3m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/24a778275c704d1cb7228b33d1e65d54 : SUCCESS in 1h 18m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/560ba63e32044947ac7a8d2da41641df : SUCCESS in 1h 00m 30s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0ad7fb847410428595c696d665718bbd : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":18},{"id":"b8e7f9f2dbc34a1a53a8e29546f1f0a799beec1c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-27 21:59:17.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7d7913b35e2ec519fbfb39f3e26c232ff7f69081","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-27 23:09:21.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"518454396c0420ad6ed49ba96422cbe2feb3e91a","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-08-28 07:23:54.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"73000b4bbf444020d4618215c7537e2260ba88d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:13:33.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce547aaaee5343c6911895931f40e83c : SUCCESS in 1h 06m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5240ae657dd4dc7a4ae28d86d87ec10 : SUCCESS in 1h 15m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abb1c33a311342e8accf046fab5abebe : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/440c4b74492d4b6d9e35476b7f14afe8 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d5fd6a4f13948b7a8aab154fcb05487 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6671e9b4a924a25badce0a9907954d8 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1315819e09df47b59486cea4cdd3e8cf : SUCCESS in 14m 11s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b67bd718e92c408e82e958b21fc9b2f0 : SUCCESS in 1h 41m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/90c62b65e6094adaba557d13ad165f3a : SUCCESS in 1h 13m 28s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/17d33185983744a4a413396c14457c75 : SUCCESS in 1h 14m 08s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/745f7ed34acf4c5fb812512c8fd758ec : SUCCESS in 1h 14m 54s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e538f12e7fab4f2fa2478c4cce509eac : SUCCESS in 1h 13m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fc2d9a19328b4bd49a04a4507a41f19a : SUCCESS in 1h 44m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6e5463411d8248bd82caa02e44aebcf4 : SUCCESS in 46m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26c73ee61b0246da9851875f5e76e4e9 : SUCCESS in 59m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8391e443468943fa8bf6437f7c41c706 : SUCCESS in 2h 02m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0dd280b5a05c4d5687a0cc0a58e4bb93 : FAILURE in 36m 47s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/658c0de769d44293abf57c92dd2736b7 : SUCCESS in 1h 20m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23e2892e40ae4810a195b0305ef6aeea : SUCCESS in 1h 21m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/2076d258b21a4caaa241109111477ac5 : FAILURE in 1h 17m 38s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b8607fe3c7f942089d605f30fff0602b : SUCCESS in 3m 18s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5189dc80b19e4abc93d0d46930158778 : SUCCESS in 1h 08m 35s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5d84b7ec732a421c8fd6bf724ed4aa7f : SUCCESS in 59m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a5082232ac0f475f8d44855f7ccfe3e1 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":19},{"id":"4508771fe99215f5f108a69493594f10c2bb55cb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-28 13:18:52.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"f35706e15f6b282650c540a984c882b9b0ccbb9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 16:22:09.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fd01385e987d4cb9b161f20c90527511 : SUCCESS in 1h 15m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2829d081a82e4a1fb105fd7e214347ba : SUCCESS in 53m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51dcc2166f98477f8c03812773202b22 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc5356a395d34f589de82db40307dbd8 : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32b835e341fa496fad1a107a870d2d59 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c28e0fdb32043af80b436722a2f75ba : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3d9a662c364d1499b9cb2488671aa1 : SUCCESS in 8m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3ec2c1ea76eb4c4197dc19fd95bf7f39 : SUCCESS in 1h 43m 01s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/fcb28e73c92742eb909a7a2d02ced8a7 : SUCCESS in 1h 28m 59s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2392574c074546448b89dcf355b18aa8 : SUCCESS in 1h 36m 59s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/5a1bf0a093c04f1fa15b15e8a77ef3af : SUCCESS in 1h 22m 26s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9d05302633df44078998946954a29376 : SUCCESS in 1h 04m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aac3607dfa934b6aab8ae7d939d3fbf4 : SUCCESS in 1h 34m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/762afb5c34d842e3b561dd49527e07c2 : SUCCESS in 46m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18a72e6e802441e58eb18777e46fd09f : SUCCESS in 56m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ce497e6cde341a4a685c2410e43f86d : SUCCESS in 1h 58m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3d204586ac9a4ce88168ea32c14efbd2 : SUCCESS in 45m 11s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/16551c909ee441759fd30cd72e950039 : SUCCESS in 1h 08m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68ab2aa742dc45e29c36e25a1a1e5701 : SUCCESS in 1h 20m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/265deef8e6fb47cb95a69d31c404cb3f : SUCCESS in 1h 33m 48s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/06412d0c48884dabafdf18bdbe649093 : SUCCESS in 2m 49s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ab48e384b5bf4ae7a56f17cfa4564af2 : SUCCESS in 54m 13s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/180e26ddbe1b474db85d37a0a2abbdee : SUCCESS in 1h 05m 37s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/17bdc910c99c449bb95bbd6b3763ae95 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":19},{"id":"3a919909c84aae8dab8333afd90a212885655bed","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-31 08:19:04.000000000","message":"Patch Set 19: Code-Review+1\n\n(10 comments)\n\ni have just a few doc improvements, apart from that it looks good to me","accounts_in_message":[],"_revision_number":19},{"id":"9b8715d1d6541bec231bbcdd0cca96b4d7232a96","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-09-03 10:35:51.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"8a44ef1c4df2be03e029a1cc9966f2e95d5cf7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 13:47:48.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4761281eb72f44a4879014cc80631953 : SUCCESS in 1h 01m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6958bfeb746d4e089d919e758b6bffe9 : SUCCESS in 1h 32m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/053097aa4caa4da8bf9fd92fbe10510a : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adde0822361b4fb3b9dd43ed9dc6107f : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2f2de26d8de4971847b756af1f5b13c : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5caed1410bb34a41ba32f79d088024a5 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab86dba1ff0e46ab88f382029fcdfb0a : SUCCESS in 13m 23s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/bea4e3a343324d5bb9ab7a7ba8d02f3c : SUCCESS in 1h 51m 34s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ed5d601516684fa2a0060b57328b5471 : SUCCESS in 54m 20s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/99d320c32fff4f83b376d439c0e5d23b : SUCCESS in 1h 21m 15s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3f5f4cad53934344833ca2b7428aa496 : SUCCESS in 1h 22m 00s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9bea6af89a544260998e07e86d5f5f25 : SUCCESS in 1h 17m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/524a072ca3d24a18a73dc6862d865c9a : SUCCESS in 1h 16m 10s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/91e67832537040fba5c59bf9cf7f4a39 : SUCCESS in 45m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4730ce843bad47e3944fc737836b8ad0 : SUCCESS in 56m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c923b85a1ef84585aa5a1643fe434edc : SUCCESS in 2h 09m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/428aed093c58473295907708ea02fa52 : SUCCESS in 51m 31s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2b7c4b16895f4a6fb8758f3812cdbcc0 : SUCCESS in 1h 06m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b94b449a03c74c2b9bb9eb0ef0b0f62a : SUCCESS in 1h 30m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bfcb5510c63d49b9b9877ef61644720e : FAILURE in 1h 01m 44s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a55c36546e954868b207e8d0d7f250d8 : SUCCESS in 3m 03s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1c1aff848e2f4bbfbcc1b689783fbd12 : SUCCESS in 1h 12m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/70099554d55247e29a5f794982a4b998 : SUCCESS in 57m 36s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c3d888bb54c74f60986aae0ebe1060d2 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":20},{"id":"b7e692f6cd2c77ad35499e232aef68d9ff933f2f","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-09-03 14:13:47.000000000","message":"Patch Set 20:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"114f2ad614551faff7ce46eefb8ca48dfd337b6f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-03 15:10:55.000000000","message":"Patch Set 20: Code-Review+2\n\nthe dvr job failure is unrelated , LGTM, thanks","accounts_in_message":[],"_revision_number":20},{"id":"aa8e488c679862c80bd808c84f553f293c918e88","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2020-09-03 15:58:50.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"a028ad6458d041ab8ee42f2df326f0725178b3e0","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-09-03 16:08:23.000000000","message":"Patch Set 20:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"fe761f8bc96efef02ea2ddd12b33a3c7cdbf905c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-03 21:37:28.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"89b2ce15db297fc6c398e86ba4348192b67e969c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 01:05:41.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/919064b6fe8d420c80b354bc0d981d87 : SUCCESS in 1h 00m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38e08f2b32274fadafbf674d22c182f9 : SUCCESS in 1h 19m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c32fc583a1f8479b85a51738f8e90d6d : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2611e5bb764b4f78bab1b3c8f954a697 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc9f22d2c8424e498b27d1b5fcf6cb2a : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/194504a592834f9495e1ed2d9ff8e7f1 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d419bb957ef443c6a3abf39620ea5631 : SUCCESS in 10m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/baf4699c0d6f423491d485654c8f4868 : SUCCESS in 2h 20m 13s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a1df8a2c1efe4321a83d900bc116af92 : SUCCESS in 1h 23m 20s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/0906e3207a5143b3b4ef180ce7b71e84 : SUCCESS in 1h 34m 18s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3afd53dcf81e4823b617d18e56e17ce4 : SUCCESS in 1h 31m 53s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/b91154187101473684e2d4ac0062ac7f : SUCCESS in 1h 43m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fcc3a76ae10b47768fbd83474c8d69d0 : SUCCESS in 1h 37m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b8c2995bc0e742ba8d6032cbddad9789 : SUCCESS in 44m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b65af727b19e4ba9a425965983e13511 : SUCCESS in 53m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dda3558330ac43b48c8cdb732191e7b8 : SUCCESS in 2h 22m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/580e9cf9eb6245aa8b909dc5b5152c8d : SUCCESS in 50m 12s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0ff7f78d8e1d46b4b3d7fccf71ebdf5e : SUCCESS in 46m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6dd7c459a29b4e8bbbd8e0d072bc2106 : SUCCESS in 1h 03m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d1efbc43675543729f04e2556e0b7cb6 : SUCCESS in 1h 12m 14s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/24ad91934dfe43feabf00e3b439da686 : SUCCESS in 3m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/73c9052f529f4eac8aad05cf7674d7ae : SUCCESS in 1h 12m 00s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/930c4fc33a844adca66ceb827bd8bcd0 : SUCCESS in 1h 05m 49s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/23dd65a80bbe42fab1005b83f9ef2e92 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":20},{"id":"f175967b4d40ba32f87a26470f86ea14188a07c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 01:06:24.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"5f8730e60bfdf9f4c32d35d3b3d2b6ab3a12119d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 07:40:45.000000000","message":"Patch Set 20: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/83bba91ffd524dc5a7116d19a7b64101 : SUCCESS in 1h 12m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/672aeec8e08546269d2776f833049819 : SUCCESS in 56m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116e731dc0e34259b5589254b9015550 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2293f72f9ef4022a3eaf393e02f6335 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc98aa0426ff4d438d2b88245d1842bd : FAILURE in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d600bcf584468facd38feb4d10207f : SUCCESS in 19m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4856907f32d64c70a19787542af67b45 : SUCCESS in 2h 05m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fad69818472944e8a473f52ea01da3bd : SUCCESS in 1h 16m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1d5a890dba040879b4ec95575ce5a0b : SUCCESS in 56m 32s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/13af4b9cea8b4ac2a891830fade0b040 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":20},{"id":"dfb870f21768931e085c5ecd3b53d217d55d2b26","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-09-04 10:48:22.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"19ec73b07e9f259916c0165c134bbf5e2aca8d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 13:01:59.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/704d862d5415419e86108fc6931b96ce : SUCCESS in 1h 00m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/587cb5dc81684b8ca810f04f3fb742fa : SUCCESS in 1h 20m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5faeb812018147088351e842f6a3ed65 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3482326ab71645b483cdf9c63d8d0807 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53b6a8d6e4854f9ba23b57ea019ea17e : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1589d0725e1b41d9b43f717ecbc372e0 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db0349cf9944985aa1cf145b3f61b1e : SUCCESS in 12m 52s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4246a536e3d84f5c80d3801e960b6c51 : SUCCESS in 2h 00m 47s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/29a2f3331fd74f4783796b1702f1b3c1 : SUCCESS in 1h 08m 34s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/3d52800f912848e3a0a83f79bedc4453 : SUCCESS in 1h 27m 15s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f487377b569c4b819bc471427a2d6c68 : SUCCESS in 1h 39m 20s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/01870df3ba454f2bb5d59aeedd260346 : SUCCESS in 1h 15m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f83f5de774034ab5a4040a6fb9712f75 : SUCCESS in 1h 45m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/01d2d2492fa64de190a2dd1581c9d899 : SUCCESS in 42m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60bdab12fc074f51a9e93701f36b97ef : SUCCESS in 53m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/633a044e03094aa0bb341c29f7ad0939 : SUCCESS in 2h 02m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/69137aaba9ce424a851e90144141adb4 : SUCCESS in 49m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ede3a4cced414b3fab9fefc13bc421de : SUCCESS in 1h 02m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98508fc38e754069a42f79784ad5470f : SUCCESS in 1h 05m 39s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/47b5a476d0d342b8936248ddf166a6d4 : SUCCESS in 1h 27m 53s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/57a387f7097b42a59ad82813b623b830 : SUCCESS in 2m 55s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b3fe027645594b7f9d916dd4c47c75e0 : SUCCESS in 1h 16m 17s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d8df61a934af4cc9b0c2229efd6451d2 : SUCCESS in 57m 53s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ad1bf385c4d845d596a8794f87380212 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":20},{"id":"4bb53bd5f9d56e16dbd364c77b8dc76fdb4d78c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 13:02:27.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"b74d87c5da92e8dc289e781a200e36baa4100e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:19:43.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/76e691bcd0bf4174918628f129421320 : SUCCESS in 1h 09m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43cc567060754aefbe3a8ed88eea869b : SUCCESS in 1h 25m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7369801683847fcb284ebbe48a31eab : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39331c182bce4908a689a0270b586f11 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8f8a65282d347c2ad0844eaf7f97fd3 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3431e66a4641388fed815058a4dffc : SUCCESS in 10m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26922f471a5f4d748ad7dd976b2ca5fa : SUCCESS in 2h 00m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af3b04b4dc544f59bbebbb3d27ac28f0 : SUCCESS in 1h 08m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/890c515156e24191948329156d85173d : SUCCESS in 55m 46s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d4c44c57e8f842c7ac20277e1e0b4370 : SUCCESS in 1h 09m 35s","accounts_in_message":[],"_revision_number":20},{"id":"1fd176046c4181bb7dc5179181528e5247285b08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:19:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"3150cb5c52fbce262f26870a3720392b54f5d877","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:22:14.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7505b96e1aa94d87867b7a806125befe : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"0c84f3ee2eec4216cfa659d5ac21225f08eaef09","revisions":{"47c6fc0042f0d44fddf8eb3bf53bf0202d6e4916":{"kind":"REWORK","_number":1,"created":"2020-04-30 10:35:18.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/1"}}},"commit":{"parents":[{"commit":"b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf","subject":"Merge \"add unittest for delete tag definition interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-04-30 09:32:25.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-04-30 09:54:31.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n4. Add docstring about method usage and parameter.\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/47c6fc0042f0d44fddf8eb3bf53bf0202d6e4916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/47c6fc0042f0d44fddf8eb3bf53bf0202d6e4916"}]},"branch":"refs/heads/master"},"164fdae22733da106a9aa421e72be7c02cdefbd5":{"kind":"REWORK","_number":2,"created":"2020-05-05 13:15:46.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/2"}}},"commit":{"parents":[{"commit":"b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf","subject":"Merge \"add unittest for delete tag definition interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-04-30 09:32:25.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-05-05 12:55:33.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n4. Add docstring about method usage and parameter.\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/164fdae22733da106a9aa421e72be7c02cdefbd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/164fdae22733da106a9aa421e72be7c02cdefbd5"}]},"branch":"refs/heads/master"},"f2a98740751e6afca0f782f17e6af33331a470dc":{"kind":"REWORK","_number":3,"created":"2020-05-11 07:21:54.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/3"}}},"commit":{"parents":[{"commit":"b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf","subject":"Merge \"add unittest for delete tag definition interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-04-30 09:32:25.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-05-11 07:19:41.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n4. Add docstring about method usage and parameter.\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2a98740751e6afca0f782f17e6af33331a470dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2a98740751e6afca0f782f17e6af33331a470dc"}]},"branch":"refs/heads/master"},"0c6b819f392a87ef4b06d6663cbb05dc7d78f123":{"kind":"REWORK","_number":4,"created":"2020-05-18 13:48:08.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/4"}}},"commit":{"parents":[{"commit":"b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf","subject":"Merge \"add unittest for delete tag definition interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b8d383cf20a40258bb3db8d8f3f2ddb90b923dcf"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-04-30 09:32:25.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-05-18 13:39:38.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n4. Add docstring about method usage and parameter.\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c6b819f392a87ef4b06d6663cbb05dc7d78f123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c6b819f392a87ef4b06d6663cbb05dc7d78f123"}]},"branch":"refs/heads/master"},"054e1925c2ee2373522fe4ffdc456d41dbbaed2a":{"kind":"REWORK","_number":5,"created":"2020-07-15 11:06:03.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/5"}}},"commit":{"parents":[{"commit":"90018eabe73a3d55e2690c4b7284cd8d88692f3d","subject":"Merge \"Remove checking of Accept-Range header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/90018eabe73a3d55e2690c4b7284cd8d88692f3d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 11:05:54.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/054e1925c2ee2373522fe4ffdc456d41dbbaed2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/054e1925c2ee2373522fe4ffdc456d41dbbaed2a"}]},"branch":"refs/heads/master"},"4184be3c180f33de79b4469579a8ec90ed479e22":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-07-15 13:45:06.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/6"}}},"commit":{"parents":[{"commit":"90018eabe73a3d55e2690c4b7284cd8d88692f3d","subject":"Merge \"Remove checking of Accept-Range header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/90018eabe73a3d55e2690c4b7284cd8d88692f3d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 11:06:58.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4184be3c180f33de79b4469579a8ec90ed479e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4184be3c180f33de79b4469579a8ec90ed479e22"}]},"branch":"refs/heads/master"},"6d389a325a749480b07a1f10549069cb710e395e":{"kind":"REWORK","_number":7,"created":"2020-07-27 11:14:23.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/7"}}},"commit":{"parents":[{"commit":"90018eabe73a3d55e2690c4b7284cd8d88692f3d","subject":"Merge \"Remove checking of Accept-Range header\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/90018eabe73a3d55e2690c4b7284cd8d88692f3d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-27 11:10:57.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d389a325a749480b07a1f10549069cb710e395e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d389a325a749480b07a1f10549069cb710e395e"}]},"branch":"refs/heads/master"},"2aa18d974ac30cd736284a54bd21be28925be335":{"kind":"REWORK","_number":8,"created":"2020-07-30 13:27:34.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/8"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-30 13:18:05.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2aa18d974ac30cd736284a54bd21be28925be335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2aa18d974ac30cd736284a54bd21be28925be335"}]},"branch":"refs/heads/master"},"20c6d924386f98baa5c2ccd7b25bbef3f6177a94":{"kind":"REWORK","_number":9,"created":"2020-08-07 10:46:39.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/9"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-07 10:44:56.000000000","tz":330},"subject":"[WIP] Need to have stable tempest scenario manager","message":"[WIP] Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/20c6d924386f98baa5c2ccd7b25bbef3f6177a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/20c6d924386f98baa5c2ccd7b25bbef3f6177a94"}]},"branch":"refs/heads/master"},"61cfe4e4d29575366869160c91469ac80a48078a":{"kind":"REWORK","_number":10,"created":"2020-08-13 13:50:03.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/10"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-13 13:46:16.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/61cfe4e4d29575366869160c91469ac80a48078a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/61cfe4e4d29575366869160c91469ac80a48078a"}]},"branch":"refs/heads/master"},"8da2d8fb6299558e5d10c958aebf8b1cfc2e41a1":{"kind":"REWORK","_number":11,"created":"2020-08-14 08:05:31.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/11"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-14 07:59:03.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8da2d8fb6299558e5d10c958aebf8b1cfc2e41a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8da2d8fb6299558e5d10c958aebf8b1cfc2e41a1"}]},"branch":"refs/heads/master"},"be4b958c07362a3f45a051617cf8d1e665f5062c":{"kind":"REWORK","_number":12,"created":"2020-08-17 07:42:08.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/12"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-17 07:38:45.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/be4b958c07362a3f45a051617cf8d1e665f5062c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/be4b958c07362a3f45a051617cf8d1e665f5062c"}]},"branch":"refs/heads/master"},"0cdbf55a150304f0fe6868db0898e3bc62adf38f":{"kind":"REWORK","_number":13,"created":"2020-08-17 13:17:26.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/13"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-17 13:11:31.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nDepends-On: I33e6abd416bbe5964a279f7969615ffd2974b4dd\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0cdbf55a150304f0fe6868db0898e3bc62adf38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0cdbf55a150304f0fe6868db0898e3bc62adf38f"}]},"branch":"refs/heads/master"},"b9102351016b50a15d5a1f55b44f0df1fb26e947":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2020-08-18 06:55:45.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/14"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-18 06:51:35.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b9102351016b50a15d5a1f55b44f0df1fb26e947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b9102351016b50a15d5a1f55b44f0df1fb26e947"}]},"branch":"refs/heads/master"},"2b77434c980cb547f56dd559b6df4d15079a826f":{"kind":"REWORK","_number":15,"created":"2020-08-24 14:26:58.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/15"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-24 14:23:50.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2b77434c980cb547f56dd559b6df4d15079a826f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2b77434c980cb547f56dd559b6df4d15079a826f"}]},"branch":"refs/heads/master"},"a8d7ca7e95dda83263e446eb56bc8fa694923174":{"kind":"REWORK","_number":16,"created":"2020-08-25 11:40:46.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/16"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-25 11:37:18.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a8d7ca7e95dda83263e446eb56bc8fa694923174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a8d7ca7e95dda83263e446eb56bc8fa694923174"}]},"branch":"refs/heads/master"},"ef0d95aa43cbd11abbaecc2ab193c306ea3bb6e5":{"kind":"REWORK","_number":17,"created":"2020-08-26 13:48:17.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/17"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-26 13:45:01.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ef0d95aa43cbd11abbaecc2ab193c306ea3bb6e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ef0d95aa43cbd11abbaecc2ab193c306ea3bb6e5"}]},"branch":"refs/heads/master"},"447260c115a97084a7722abc9b54d150ed4a7404":{"kind":"REWORK","_number":18,"created":"2020-08-27 13:22:32.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/18"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-27 13:18:37.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/447260c115a97084a7722abc9b54d150ed4a7404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/447260c115a97084a7722abc9b54d150ed4a7404"}]},"branch":"refs/heads/master"},"6cbccb9e664de7ef0bb702794ef48773d75980f7":{"kind":"REWORK","_number":19,"created":"2020-08-28 07:23:54.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/19"}}},"commit":{"parents":[{"commit":"c02aa28390ed48710f57ae18531b28d59d968f8d","subject":"Merge \"Adding description for testcases - compute part4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c02aa28390ed48710f57ae18531b28d59d968f8d"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-08-28 07:20:00.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6cbccb9e664de7ef0bb702794ef48773d75980f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6cbccb9e664de7ef0bb702794ef48773d75980f7"}]},"branch":"refs/heads/master"},"0c84f3ee2eec4216cfa659d5ac21225f08eaef09":{"kind":"REWORK","_number":20,"created":"2020-09-03 10:35:51.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/31/724631/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/724631/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/724631/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/724631/20"}}},"commit":{"parents":[{"commit":"e866da9a115e623433b4078ba368c1d623b9c71f","subject":"Wait until dependent resource is deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e866da9a115e623433b4078ba368c1d623b9c71f"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-07-15 09:50:59.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2020-09-03 10:19:53.000000000","tz":330},"subject":"Need to have stable tempest scenario manager","message":"Need to have stable tempest scenario manager\n\nThis patch addresses following facts:\n1. common manager methods among plugins should be\n   defined in Tempest.\n2. methods should be consistent with names and\n   parameters.\n3. Add docstring about method usage and parameter.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\nChange-Id: I191c8861c9ec29b78c9c8ad12ff2ddf3ffbf4db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c84f3ee2eec4216cfa659d5ac21225f08eaef09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c84f3ee2eec4216cfa659d5ac21225f08eaef09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
