)]}'
{"id":"openstack%2Ftempest~725139","triplet_id":"openstack%2Ftempest~master~I2c21237bc34826eea3033dac1bd83068bf3f97ba","project":"openstack/tempest","branch":"master","topic":"stackviz-py3","hashtags":[],"change_id":"I2c21237bc34826eea3033dac1bd83068bf3f97ba","subject":"Use python3 for stackviz","status":"ABANDONED","created":"2020-05-03 18:49:26.000000000","updated":"2020-05-05 20:02:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2d645a98dcb3dbfe702395dc1d50b05c9d8f2af","_number":725139,"virtual_id_number":725139,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-04 04:42:31.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-05 02:13:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-05 19:19:52.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2020-05-05 15:55:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 02:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-05 15:55:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-05 19:19:52.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"49ce8238aaa5ddab2070d9d82248227c2582dca6","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-03 18:49:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7755cead0f09f7bfa0996d8a56b188607b275959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 21:27:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2553a667238648ec8674f91afc2b3653 : SUCCESS in 1h 38m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ad7967785ae492090a85fb07e80a0b4 : SUCCESS in 1h 33m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9000f9b849ed447e9e204d5d10cfe764 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d01a565823ff4eaf8818314d43637142 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d10725450c9d4616abe8c8422dccfc55 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c99cc2eae1054ab3a75601af322a973e : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bdad1a7cd9c405b89850b5ea072aa7f : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef98cba63f3a47c699ff8ce0ef055236 : SUCCESS in 8m 29s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/ea2479070bee480ab781439d8d6c82e4 : SUCCESS in 42m 53s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/d6d296c53b854fcc8cf8d1157d581f19 : SUCCESS in 31m 50s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3d1b15465fb745f99629b64b728dd5f0 : SUCCESS in 1h 44m 44s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/16eee4ed3f2a451084ab42545836e0c0 : SUCCESS in 1h 42m 23s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/cdd8b23c9ff047ddab29c4eeae95cfeb : SUCCESS in 1h 44m 32s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/cf2fcf33cbeb4c08946fd079445f7b80 : SUCCESS in 1h 58m 58s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/6d2e4203e6ff4aaa8f46484c74e8d4fd : SUCCESS in 2h 03m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ac1c12cd261847a0846835671750e09d : SUCCESS in 2h 20m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/19f4802dd5154ca5a1606bc14a212ecf : SUCCESS in 43m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/650a28e39c824ce39b12089dda08a428 : SUCCESS in 1h 44m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89654e457cdd41f1b9a421cfabd40c39 : SUCCESS in 2h 27m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/edc0bc0b2fb44e2a95974b9c253bd7f1 : SUCCESS in 47m 20s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bc1d8a18a41347d2b4919ea578599a6f : SUCCESS in 1h 08m 17s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4fd9f206867421690856954d500e7ef : SUCCESS in 59m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/6fa7798b30a049e68172c095df5173e9 : FAILURE in 6m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/221598e860c14ab88561af02ec160f3f : SUCCESS in 1h 00m 39s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5b51d46c9f764e6fab113f2a10bdcef1 : SUCCESS in 2m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e25739ca511142238afdda4cbcf224dc : FAILURE in 1h 45m 26s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1cd97bcb71024f45bdec1e6b62636efe : SUCCESS in 1h 41m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/09621499b1604a89a199d033ede2a999 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d48a113679e4382684466c423e4adcacd4983267","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-04 04:42:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f47052b5a97d3007744d725d5e3ab84a51eb9d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 07:24:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8893ef983e9e4a1999fbdcfd630c62ad : FAILURE in 5m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5289e2ece6aa4ccab4df6ae83e3a94dd : FAILURE in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35d62a5411054d3ab681eb8dfe0fcbd2 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43cf6e548c7e4d288b2e541f3982bba4 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3bc8aa7aa3b40b98e7270aa37e03d88 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e688ede43834898a18e5127c03b5406 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86d5b74a703349a1813ee25c093c04ae : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751885c2babc4ce88a497e6ff0215e58 : SUCCESS in 9m 47s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/580a37db5f2047a1b0b26417e35212d5 : SUCCESS in 1h 24m 50s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c7495c74f48f4b01819e1593c684dd48 : FAILURE in 5m 14s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c2bca97b109543fd938c63368f0434c2 : TIMED_OUT in 2h 09m 37s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6bb8bcce35504cdea5f1e46c02d0398b : TIMED_OUT in 2h 08m 10s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/626eba1eaca54303a9d6833014c04bfb : FAILURE in 5m 59s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ec4ef91c65664484909b81eb74f7b77e : TIMED_OUT in 2h 12m 25s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9850f3bfb0e74145a9563f86f27c3892 : SUCCESS in 1h 26m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7de15482449f47b5a2e5e065b0c4b314 : FAILURE in 2h 38m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/8d39a5257d8c45b6adf92c9cf257d1d4 : SUCCESS in 44m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/340a9fff83a1461c8146845607fa88d8 : SUCCESS in 2h 13m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de36839a5e2e453cb60100572b4d9b46 : SUCCESS in 2h 02m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/28108e3df8c04da7ad32a72cfe3daf32 : SUCCESS in 1h 59m 57s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/06869da30cce4e69a7282d72da7a5a18 : FAILURE in 5m 15s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0740b44509d24591a583a901962c5dec : SUCCESS in 2h 21m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b7584342444540e4ad0d20d8a20aa181 : FAILURE in 5m 13s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d8a96c80cc2f4bb2afc498e8e0f78f50 : SUCCESS in 3m 06s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ecacb5d1822c4cb6a89e4520b8ca02f2 : TIMED_OUT in 2h 07m 57s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8a362087a9714d51861b654db7458ead : FAILURE in 5m 14s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5fd91ad8a99e489991aecc83f31cc414 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0f1ddfe2c4c25d74681a46e8c4e2f65dcc465aa2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 23:34:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4a53a462b46e4659136f0bfcfedfdf4227e2fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 02:13:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8e5d05f5c77443c6986654cefa5c19cc : SUCCESS in 1h 24m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34d70ff2be044e388cbac572db1d17d8 : SUCCESS in 1h 30m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d379103fd849459fb1b7afa8ba621f68 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44be7e794b6f4cfb8215ca5710472523 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1d4624f18964247af1d84c4b6efee02 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79d87098b96149989b76910bcb49a594 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ef668c5dae48758125f3c60e77d9cd : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa946dc858634b00ab188890e04e2218 : SUCCESS in 10m 05s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/5c2976beeec946a5be78f842c8860ec4 : SUCCESS in 38m 14s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/17bc6df96d7f488aac8d730fffcc60a0 : SUCCESS in 33m 45s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/5d3a50a4d55840c8864432fba2f2d08c : SUCCESS in 1h 54m 57s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/be2eb8cec9b2421a8df8291305b7ea05 : SUCCESS in 1h 16m 39s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/8b35243aea8f4e42a15bfc8115715b1c : SUCCESS in 1h 48m 34s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/b61a4ba9bd03432eaa597567f4332125 : SUCCESS in 1h 45m 04s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/daf24d8e50344daea76c5ea8a5b8ad44 : SUCCESS in 1h 12m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/37b146dac4ad469d8d5f6bdec7281f29 : SUCCESS in 1h 58m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d9402a0161d240ecafb6e5502f894f5a : SUCCESS in 46m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7456b454fc340549c8c024b2cbff7dc : SUCCESS in 1h 01m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/10c8a88aa624421ab2b7fcf281f36d30 : SUCCESS in 2h 17m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c2df42ad29d7487197885350189d2d4e : SUCCESS in 53m 23s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c0b024bef60948af9a59f3886df248fc : SUCCESS in 58m 55s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/59a4d862544445d990fb64f015daddf5 : SUCCESS in 1h 07m 42s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f44ae3517dd04cad980cbf87bb607990 : SUCCESS in 1h 35m 13s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/899b06848deb4171b0ec5b75c406f92d : SUCCESS in 3m 04s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9ceabf5de9fe4fa4b00efa48541367fd : FAILURE in 1h 04m 32s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/157869f2bd274dafac136bb6696b252a : SUCCESS in 1h 01m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/84563366de014d45a826cd0e916993a7 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6ec79b94da38c93c8bdb2099d46ac67777403bd1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 15:55:45.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm, for safer side it is ok to explicitly ask for py3 venv.","accounts_in_message":[],"_revision_number":1},{"id":"58b5329aa49b66928a20c8a923f9c54837778deb","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-05 19:19:52.000000000","message":"Patch Set 1:\n\nI have something similar in https://review.opendev.org/724198\n\nThat uses the ensure-pip role from zuul-jobs and installs stackviz under python3 using venv, so doesn\u0027t require a virtualenv dependency.\n\nOne other problem that fixes is that suse 15 still runs ansible under python2 (https://review.opendev.org/#/c/724777).  that tries to \"import pkg_resources\" so requires python2 setuptools to work, even if it\u0027s installing python3.  724198 works around that too.  in some cases something else may have fixed that up for you, but at least on keystone testing, this is the first place that tries to use pip: to install stuff.","accounts_in_message":[],"_revision_number":1},{"id":"03d26b2a9557e1388e7d24d9cc35166c7f95e80c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-05-05 20:02:11.000000000","message":"Abandoned\n\nIan\u0027s patches are a better fix for this","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab96b3c162ee4166bc3712ec09cf64738f1e7bcc","revisions":{"ab96b3c162ee4166bc3712ec09cf64738f1e7bcc":{"kind":"REWORK","_number":1,"created":"2020-05-03 18:49:26.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/39/725139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/39/725139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/725139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/725139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/725139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/39/725139/1"}}},"commit":{"parents":[{"commit":"ba95c36a2d4920688b12ce15b0c46d233e20eeee","subject":"Merge \"Integrated gate storage/swift: blacklist more tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ba95c36a2d4920688b12ce15b0c46d233e20eeee"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-05-03 18:48:43.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2020-05-03 18:48:43.000000000","tz":-420},"subject":"Use python3 for stackviz","message":"Use python3 for stackviz\n\nNo matter what interpreter is used to run ansible, ensure that stackviz\nis installed in a python3 virtualenv.\n\nChange-Id: I2c21237bc34826eea3033dac1bd83068bf3f97ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ab96b3c162ee4166bc3712ec09cf64738f1e7bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ab96b3c162ee4166bc3712ec09cf64738f1e7bcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
