)]}'
{"id":"openstack%2Ftempest~725373","triplet_id":"openstack%2Ftempest~master~If964909a73a52836e3adbb7b695f1c7322b81830","project":"openstack/tempest","branch":"master","topic":"switch-to-victoria-testing-runtime","hashtags":[],"change_id":"If964909a73a52836e3adbb7b695f1c7322b81830","subject":"Switch testing to Victoria testing runtime","status":"MERGED","created":"2020-05-04 19:17:53.000000000","updated":"2020-05-11 02:06:22.000000000","submitted":"2020-05-11 02:04:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725373-1589162663112-b84da92b","meta_rev_id":"3144286640694f202467aa2e1f7a65a421acdb91","_number":725373,"virtual_id_number":725373,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"date":"2020-05-06 06:47:12.000000000","_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"date":"2020-05-05 18:28:31.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 02:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":2,"date":"2020-05-10 23:51:14.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2020-05-05 14:36:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":2,"date":"2020-05-05 14:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"all":[{"value":1,"date":"2020-05-10 23:51:14.000000000","_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 14:36:44.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-05-05 14:40:57.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-05-06 06:47:12.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2020-05-10 23:51:14.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-05-11 02:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f0e3cf3a6fb5384b5ee343d7bb9d94d97bdaed7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 19:17:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d7ca20827841d988b3d1fcebbf21582879bfb82","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 19:31:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8f6485c17768071e346325608571296ec832826e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-04 19:40:12.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f92b56310aa443e4e9f356087c7ceefe2cea5367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 22:04:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd965452ac244aedb01a429305eac09e : SUCCESS in 3m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/b757b880396a401989e0ca245cb6361b : SUCCESS in 1h 03m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84965cde992b47458d67f2ab8187ae82 : SUCCESS in 1h 16m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/589d1f50c7a94bdfbabf56d6acef8a92 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c7ed35191a4b0da059e48767e3413d : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f97fcff6bf4e4485a38b5ecd763c78de : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e037bd0e12f947f3bce71960ff2a4146 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ba9463096b488cb30dc8acaa7af1bc : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bef31d624c9c467e80ff0a275e43fabb : SUCCESS in 12m 08s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/b4ea89c8393741c197a9c8193c51f8dd : SUCCESS in 41m 37s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e05e91afb74e46b0b3ec2782ec2faf92 : SUCCESS in 1h 07m 08s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/89ed735012b94436bbcbb3d72bf46a83 : TIMED_OUT in 2h 07m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/65bd6913effe400aa587cf792aeec7e8 : SUCCESS in 1h 54m 39s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1a52415c97fe45dbb1b87ed4acfaa210 : SUCCESS in 1h 37m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/638122eb594944a8bf13983d0917bd9f : SUCCESS in 1h 52m 22s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/1e067a1d169d4b7db2ff5bc1f47688d1 : SUCCESS in 1h 51m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0807ed2fc9f246c6b2326267a9ebdfc8 : SUCCESS in 2h 18m 28s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/74cd4e5ce1964f5488ae6483889adef9 : SUCCESS in 49m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8428764bd72b459aa94381525126112e : SUCCESS in 1h 01m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8fc1dfbabc76471788f590669eb5c6ba : SUCCESS in 2h 16m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/13e0498de6c949b6a33f1d04d8b2fe5f : SUCCESS in 52m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8e307b9842c6479d85acfa18f686af70 : SUCCESS in 1h 11m 07s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a65ecf01bb3f4b6b86907f3a22c2e9c3 : SUCCESS in 1h 11m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a71949b5533340118fd88faed9df0fdc : SUCCESS in 1h 27m 14s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d45136b1f6d54e84ace8b64ebadb1465 : SUCCESS in 3m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5d112986905f49c8be205c28ddb7636d : FAILURE in 1h 10m 12s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/82604c3127a04deba3df021f9d4cccb6 : SUCCESS in 1h 07m 10s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8ce3a78f55a64c61a424ca9c4b405cfe : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"dfddf5ebf6be0c49a52bb1d7166489c1a54fc22f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-05 07:54:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e8ad4e2147e27f502410ba90d238ca43b69edc5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-05 10:04:50.000000000","message":"Topic set to switch-to-victoria-testing-runtime","accounts_in_message":[],"_revision_number":3},{"id":"438a67953091b457c1e463f45f93eedafaf8934a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-05-05 14:32:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"95f3a020a4f529dbbdf3c5a319792164d4acd1f3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 14:35:14.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f7e32661a29a983c74a0c5adeef639366b3ee657","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 14:35:54.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"7d18c8dd65edf3fca034231f85b5a73b8756c6a7","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-05-05 14:36:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4900aa35d4b908e2bd61af66f1b5d15503f426dc","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-05-05 14:40:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5c137d0fec36126294c77b94415d6df341e6ff89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 17:51:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ca44d829811144a4a4d41049b320c8b5 : SUCCESS in 4m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/5b0302da721845c0859cd0c96209b600 : SUCCESS in 1h 04m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/439c513178bb463684724ca1113e34b8 : SUCCESS in 1h 21m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb8849a4317142ce967e144316cc7aa2 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967c5af1bb2d4d13b5742ae278206310 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80d55d0cc9614aac8aadea0ce4378f7f : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918f245f6f5f450b9331a8dc554cda4c : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca1f4495a634e9d9675b9e11326693f : SUCCESS in 16m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f25bec83a08483285cde386cd19b345 : SUCCESS in 12m 44s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/c28f434d7ed84e908aed6746e4f31b46 : SUCCESS in 45m 55s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/3604e939e46b430f94e5bb397559090d : SUCCESS in 44m 14s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0768f089049d4fccb37d3e9c7b27c0a8 : SUCCESS in 1h 53m 15s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/85641a1cd1af4c419a731a4c245c6865 : SUCCESS in 1h 40m 23s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/ee6ba68960b44da6a3eaf40b91e404cb : SUCCESS in 1h 33m 48s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/dd1a07d2dbaf4538ab48515e7a5cf390 : SUCCESS in 1h 51m 42s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a0429b04b8174f59b1a78500ac980663 : SUCCESS in 1h 26m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f5625a742204413b95e5f18b22e9284 : SUCCESS in 1h 38m 51s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/95b41230a8cb4b6eb628126cddcac822 : SUCCESS in 46m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b58a66e53a7640fca83261293e7ba1dc : SUCCESS in 1h 04m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/95dc3186556d4f19ab64b1a1a007d10a : SUCCESS in 2h 45m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f83d11b747ae4c7cb3f3ca9fa035669c : SUCCESS in 1h 00m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0293a18419174106898c48dda4a837ed : FAILURE in 1h 13m 56s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38b5ffe9a69b44f8a0ecd542b6c7b7ab : SUCCESS in 1h 26m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e103355739174c959bae308031e10470 : FAILURE in 58m 27s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/595bdb108a534ebfbf778a3b2011a869 : SUCCESS in 6m 35s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/de2c57b500f64a78a4bdb2a114e7f0a8 : SUCCESS in 1h 38m 40s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ae4259eb65b84125857f0edd426c6203 : SUCCESS in 1h 21m 18s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4176da870b414b1ca0de2d63318182a7 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":4},{"id":"6545940949b85da0669216e30836c48a39a1218d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-05 18:28:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5f2dfd58ed3ae00a498365a53ca7c0a7e5834bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 20:55:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f77434e1e714f55815d5b14c8e5e692 : SUCCESS in 2m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/e41dd8126b564346aeb6c756756f9ee2 : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f65ee653877f4a64826306331c9b9277 : SUCCESS in 1h 26m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70ad560443b749d68a007b78fd1aa3ab : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0698b8556364422087a42bed47254a3e : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cc3db7837014ceda1834e99212e2659 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c060f63ab354178a5e5f5c813a550b9 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d713fb53c42a4111a0ac5d6342212781 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a1182d9008647ed9c11b31ce394c23c : SUCCESS in 13m 34s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/ac23a2f572b74e00887fe4f0f406c698 : SUCCESS in 38m 40s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6f06eecfd8ef40ea822f0f8d12d01b8e : SUCCESS in 43m 41s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d277da886ab84d789e6ece2f428cd084 : TIMED_OUT in 2h 05m 11s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/48be795eabb0494899419ade9e6941c4 : SUCCESS in 1h 19m 16s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/0ad708a815714930a471285f0380723c : SUCCESS in 1h 35m 12s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/90b8621a3e1d4b8fa2d89a282475c20d : TIMED_OUT in 2h 05m 20s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/37874d46a8434be2994785d1176899e2 : SUCCESS in 1h 27m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/da378691faec4514834a7963bc4411db : SUCCESS in 1h 36m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ec93b9e72f0b4294a08dbdbe2097c215 : SUCCESS in 50m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6453399a1e6c42d883121d7f57552235 : SUCCESS in 1h 02m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f2c2051f8034daaac3d2f41a0387a21 : SUCCESS in 2h 26m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/50f311fecb3e42bd93507e3febb726c1 : SUCCESS in 1h 02m 01s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/680285eb3ab0418ab061e50bc184d710 : SUCCESS in 1h 12m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d350d7f635724187880b3310b55eb6ce : SUCCESS in 1h 11m 18s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/da41afd61fe846ec8667e2c46b95fe16 : SUCCESS in 1h 21m 07s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ffd826b3fa9241799db762b299afed1a : SUCCESS in 3m 13s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/cc10d9c463c1427893d16b3456abc0af : FAILURE in 56m 03s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b9ab8064de9b471ca280acf10d1d11b2 : SUCCESS in 1h 02m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4d029b16ac0f403eaa9cad063ac3b29f : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"6e40a6dfb0f4c6055aebebe2f65b6fdcd02cb36b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-05-06 06:47:12.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2d5d1c194d9a7aedc95800b6eef207d2f4ef8206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 09:27:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b199b22d1f8e409fa07341ce7c7fe3dd : SUCCESS in 3m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/e7775720442b475187c39efe62e883be : SUCCESS in 1h 07m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73e264d59b7643cbaf116f5a4ba2bc42 : SUCCESS in 1h 19m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dd7b7df9d9942849b72f73d3d9d48fe : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113f4699e6a842839af72bfef301bfb1 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e108c749458847f08251de5da263dd35 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9ba215b4c82402790205fe84725e78c : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb43969ed2cf4f8c945b0a30ee183a53 : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21483970457d4d98a65e8655e174b132 : SUCCESS in 14m 31s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/d702b3f9a4a546c0a4415645df470e48 : SUCCESS in 38m 41s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/42902580b5d54e08af3473a5a2d0a3cf : SUCCESS in 43m 50s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/05ba0f3d8f9b4f0197093d8857531a83 : TIMED_OUT in 2h 06m 03s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/efa6908b6f1c41db886684cc90924de7 : SUCCESS in 1h 10m 18s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/9d1e3a1b8f174f399add78ab5ff78040 : SUCCESS in 1h 56m 43s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/4d6dbc7ad99346349b8e08111556d91d : SUCCESS in 1h 50m 00s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a91f387fb2e1456ba906c3d2ec2892c9 : SUCCESS in 1h 44m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2af3de3f6ced49e3a150e4998ea84b09 : SUCCESS in 1h 33m 34s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1ba78c41b04e42b1b25a5b01b17ec6d9 : SUCCESS in 45m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cbb53df228d4ac8be5ce51c63d0274b : SUCCESS in 1h 06m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a34810ba7d244c3692597e70c054309d : SUCCESS in 2h 32m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6a3e66204fed4895bd42e9733f82ee38 : SUCCESS in 48m 47s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bff3bf46e1cb499dab07cc0267e0b978 : SUCCESS in 58m 16s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0040ac40ef08484ea5bfbaba01a8bd81 : SUCCESS in 1h 07m 55s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a27a4df788ff44acb9a71e05c4c83ceb : SUCCESS in 1h 57m 25s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9c04008bdb0a4320a9df9d97fbec8ff7 : SUCCESS in 5m 48s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/98c23481fa2d4236a824da9c15eccfb1 : SUCCESS in 1h 27m 03s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0a213073c2264e8dbed93cb553606ae0 : SUCCESS in 1h 04m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ee7e77f423174e88bd770af5d6aefd53 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":4},{"id":"eb0bd43f619e92e31483de2b21ce4174d4f983fa","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2020-05-10 23:51:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"988c388bda6b9cbfd2bbc91310fb46b957f071a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 23:51:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b503ba2f03fd0ce616e8cb533143414aa7bb886f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 02:04:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66dc265fed5b45dcbc9848308d55249d : SUCCESS in 2m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/4f621e99014c412e94622302de224ed2 : SUCCESS in 1h 00m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fa7d8de6eeb45b1961e6ad0c73a6eee : SUCCESS in 1h 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5352eb584747acbb774677283d0951 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/770774b784774faa956d969daf523748 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83e1938b4f9b4e38928f2b3095e7dba0 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760f2abbbed4486aad45ca47b7b01e18 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01df521f25da48bdbff56c9c61ef150e : SUCCESS in 10m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce0eed38af914e6d94c45eacb9ae8c1b : SUCCESS in 2h 11m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea550b8bb4734f718f9c7b29375555bb : SUCCESS in 1h 09m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2de2c3e02b4342c781ca01ee188987c1 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":4},{"id":"eb8ea376b31697b77461ee7832f9090f75f5bcf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 02:04:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3144286640694f202467aa2e1f7a65a421acdb91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 02:06:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25629051b97341919a1bf1b60788214e : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bebb2b4c52894d9588c03a0200c3ff3c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"de69845d75fdf6d53196e3eb9ba7e73b12a46e21","revisions":{"86ca9a613e15cb1581f7ed11452a82cffe48fc5c":{"kind":"REWORK","_number":1,"created":"2020-05-04 19:17:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/725373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/725373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/725373/1"}}},"commit":{"parents":[{"commit":"e8f1876aa6772077f85f380677b30251c2454505","subject":"Update supported version doc tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e8f1876aa6772077f85f380677b30251c2454505"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 19:16:54.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 19:16:54.000000000","tz":-300},"subject":"Moving tetsing to Victoria tetsing runtime","message":"Moving tetsing to Victoria tetsing runtime\n\nUpating the template from ussuri to victoria\ntesting runtime\n- https://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nChange-Id: If964909a73a52836e3adbb7b695f1c7322b81830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86ca9a613e15cb1581f7ed11452a82cffe48fc5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86ca9a613e15cb1581f7ed11452a82cffe48fc5c"}]},"branch":"refs/heads/master"},"bb6175eb911a4e14da1427181514c43a8d0622a9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-04 19:31:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/725373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/725373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/725373/2"}}},"commit":{"parents":[{"commit":"e8f1876aa6772077f85f380677b30251c2454505","subject":"Update supported version doc tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e8f1876aa6772077f85f380677b30251c2454505"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 19:16:54.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 19:31:50.000000000","tz":0},"subject":"Moving tetsing to Victoria testing runtime","message":"Moving tetsing to Victoria testing runtime\n\nUpating the template from ussuri to victoria\ntesting runtime\n- https://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nChange-Id: If964909a73a52836e3adbb7b695f1c7322b81830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bb6175eb911a4e14da1427181514c43a8d0622a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bb6175eb911a4e14da1427181514c43a8d0622a9"}]},"branch":"refs/heads/master"},"b682bfe7026fb06ed233e1748e7463919e049026":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-04 19:40:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/725373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/725373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/725373/3"}}},"commit":{"parents":[{"commit":"e8f1876aa6772077f85f380677b30251c2454505","subject":"Update supported version doc tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e8f1876aa6772077f85f380677b30251c2454505"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 19:16:54.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 19:40:10.000000000","tz":0},"subject":"Switch testing to Victoria testing runtime","message":"Switch testing to Victoria testing runtime\n\nUpating the tetsing template from ussuri to victoria\ntesting runtime\n- https://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nChange-Id: If964909a73a52836e3adbb7b695f1c7322b81830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b682bfe7026fb06ed233e1748e7463919e049026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b682bfe7026fb06ed233e1748e7463919e049026"}]},"branch":"refs/heads/master"},"de69845d75fdf6d53196e3eb9ba7e73b12a46e21":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-05 14:35:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/725373/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/725373/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/725373/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/725373/4"}}},"commit":{"parents":[{"commit":"e8f1876aa6772077f85f380677b30251c2454505","subject":"Update supported version doc tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e8f1876aa6772077f85f380677b30251c2454505"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-04 19:16:54.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-05 14:35:52.000000000","tz":0},"subject":"Switch testing to Victoria testing runtime","message":"Switch testing to Victoria testing runtime\n\nUpdating the testing template from ussuri to victoria\ntesting runtime\n- https://governance.openstack.org/tc/reference/runtimes/victoria.html\n\nChange-Id: If964909a73a52836e3adbb7b695f1c7322b81830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/de69845d75fdf6d53196e3eb9ba7e73b12a46e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/de69845d75fdf6d53196e3eb9ba7e73b12a46e21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
