)]}'
{"id":"openstack%2Ftempest~730978","triplet_id":"openstack%2Ftempest~master~Id2c662c0d6146423855714a0f240eed682e3c004","project":"openstack/tempest","branch":"master","topic":"cap_jsonschema","hashtags":[],"change_id":"Id2c662c0d6146423855714a0f240eed682e3c004","subject":"Cap jsonschema 3.2.0 as the minimal version","status":"MERGED","created":"2020-05-26 20:14:13.000000000","updated":"2020-07-14 16:18:46.000000000","submitted":"2020-07-14 16:16:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730978-1594743416869-a8fb295d","meta_rev_id":"3f7d0c9866fb8393491812c95765c7b458ed49dd","_number":730978,"virtual_id_number":730978,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"date":"2020-07-09 07:48:45.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 16:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2020-06-24 19:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-07-14 13:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-29 14:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-07-14 13:44:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 19:17:18.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-06-29 14:42:00.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-07-14 13:44:33.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-07-14 16:16:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f37c3cd96f2512a3b6833b014c5de0ac6c48f029","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 20:14:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e626c742dcd39f518e7045c87815d19ebc43557f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/64ae3ad07e9d454e9e0a9a69205318d7 : SUCCESS in 3m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/dc921a591a5245d2b3c15169f4171bb6 : SUCCESS in 59m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d899f4c8208412fad3ec4a37438c530 : SUCCESS in 1h 25m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffbbcde0934a487f90bd9389503a0201 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c506292e4f4748428d48d5cadf7c0aa0 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43077a4918be4b3291449caf6c03d4a6 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26db12bbf7fb441b9df3f54d820998a4 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37d2500b2cc4ac8a571d3c3cd885473 : SUCCESS in 20m 07s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4bd2c5dcce884bfb83b5039188e53fde : SUCCESS in 1h 57m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/36b5e270efe248e095d1d7fcf9b14a79 : SUCCESS in 1h 15m 11s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/da5c28fc18cf4f92a9248580fade0611 : SUCCESS in 1h 33m 22s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6a7a1130d1274055a1d44c837ce23c4c : SUCCESS in 1h 18m 49s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9fdf37a198df42baa70fd77daf3bab3a : SUCCESS in 1h 16m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/82976a50116341ac91fb12ebe7069312 : SUCCESS in 1h 34m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/200232818d774230ac4a02a092527b7e : SUCCESS in 46m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73f9f684e2fe4b65821d7b96d620d1aa : SUCCESS in 55m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ce1bee0c1ca94ed790cfef8cce6d5875 : SUCCESS in 1h 59m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c8629f59587f4d0d8fbee4dfeab88fb7 : SUCCESS in 49m 29s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3c7661d131634804a7a8a9f304bb9fad : SUCCESS in 56m 03s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/24fae0e1b49b4f35b08553b0da6e1182 : SUCCESS in 1h 12m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0b07fc2ec6294344856ac236162d0383 : SUCCESS in 1h 10m 00s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/444b69c1bf0242bfa525c1f7f4d58257 : SUCCESS in 7m 23s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9d20a79119654399bdfd4223f581c215 : FAILURE in 1h 08m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0505fa12815d458585895163281d3607 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/dca9ce25b2324f4a9e334e1865eadb12 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a9f8f59f5158b921afeab986ac45ab7ac7667ce3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-27 00:31:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0b0897134bbfb4281e5205d28385d0829e4900b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-29 08:26:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"321b73aace203077eb9bf43bed204aeb8859aed0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-23 11:41:23.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"f61bcd5ab5b86a8a4df2924b463d52403d8b1834","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-06-24 19:17:18.000000000","message":"Patch Set 1: Code-Review+2\n\nif we wanna fully support python3.8, we need to have jsonschema\u003e3.2.0 .. plus it\u0027s been released for more than 6 months, so there shouldn\u0027t be any hidden issues by now .. +2","accounts_in_message":[],"_revision_number":1},{"id":"5ff0acc3cf08cc7d33ef35c210c1c34405beb273","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-06-29 14:42:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b9bc13a7936dbed5febb03778697adb4c89f060e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 07:48:45.000000000","message":"Patch Set 1:\n\nHello, do you think is it possible to ask for someone to merge these changes to avoid possible future conflicts and rebases.","accounts_in_message":[],"_revision_number":1},{"id":"2b76869557d64fb0ab8d24268e417e2a6ee3eb42","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-14 13:44:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks, yeah version has issue till 3.1.0.","accounts_in_message":[],"_revision_number":1},{"id":"e272e91a1dedd2ff67970f3a2e995bee67d2ac74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 13:44:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ca2f78df312af8f233e3eb8580f72cd8a652722","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 16:16:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/92beb20d5df64db68c6582d8c4bf273d : SUCCESS in 3m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/7511d8562ef04b3ca17c48f3b5beb480 : SUCCESS in 1h 01m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc71140a379848c4ab41243c6d03989e : SUCCESS in 1h 07m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef086163b3c3437b9124cb37441091cb : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ded86b81f5a540d9a14eea394e4eb3dd : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3727278d9d1e4c68b0c2f556c3f3747c : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1e314642024da0ad6ec0d69429d4a1 : SUCCESS in 11m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d488504fcfc64c50aefcd4bb14614e7b : SUCCESS in 2h 06m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da6e927224b4492da196898edd48a324 : SUCCESS in 1h 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fb68b9cd9e04627870cf7fbe5e702e7 : SUCCESS in 1h 02m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7623706d64cc4b059bed373202217fcd : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0c42d917b8210d326910b6affa284516dc09b242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 16:16:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bbde04351ed71e0a32f72ecb46c43cf312c44e6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 16:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ff097b72e5412c97e961b18dc7b972 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eee87b504c180bd2b04110d5e4dce7d82819502b","revisions":{"eee87b504c180bd2b04110d5e4dce7d82819502b":{"kind":"REWORK","_number":1,"created":"2020-05-26 20:14:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/78/730978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/730978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/730978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/730978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/730978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/730978/1"}}},"commit":{"parents":[{"commit":"a9ef5dbc73e82a72049697eee07d8158b34529ed","subject":"Merge \"Add response schema validation for volume pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a9ef5dbc73e82a72049697eee07d8158b34529ed"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 20:14:02.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 20:14:02.000000000","tz":120},"subject":"Cap jsonschema 3.2.0 as the minimal version","message":"Cap jsonschema 3.2.0 as the minimal version\n\nPrevious versions of jsonschema (\u003c3.2.0) doesn\u0027t support python 3.8 [1].\nPython 3.8 is part of the victoria supported runtimes [2] so we now force\nto use jsonschema version 3.2.0 to avoid issues, remove ambiguity and ensure\nthat everything works with python 3 in general.\n\n[1] https://github.com/Julian/jsonschema/pull/627\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: Id2c662c0d6146423855714a0f240eed682e3c004\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eee87b504c180bd2b04110d5e4dce7d82819502b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eee87b504c180bd2b04110d5e4dce7d82819502b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
