)]}'
{"id":"openstack%2Ftempest~737453","triplet_id":"openstack%2Ftempest~master~I4c68caa693744e06de6be89a8f09d8f7e8ce8ba0","project":"openstack/tempest","branch":"master","topic":"unskip","hashtags":[],"change_id":"I4c68caa693744e06de6be89a8f09d8f7e8ce8ba0","subject":"Unstable test_server_connectivity_cold_migration_revert","status":"MERGED","created":"2020-06-23 07:51:32.000000000","updated":"2020-07-15 11:30:52.000000000","submitted":"2020-07-15 11:28:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737453-1594812536924-5cdf3d8f","meta_rev_id":"0714ca36b8965572c7969dd834ee7dc9711ca349","_number":737453,"virtual_id_number":737453,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-06-24 07:23:59.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-15 11:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-24 06:53:10.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-07-13 13:26:16.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":2,"date":"2020-06-23 16:59:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-13 16:03:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 16:59:17.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-06-24 06:53:10.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-06-24 07:23:59.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-13 13:26:16.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-07-15 11:28:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"173e96b1d3e2449af215f81976baf3407b29ae86","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-06-23 07:51:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b63f3d9e177c8a9c855613e86c7108d57715a866","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-23 08:10:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/53/737453/1/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0f3ab0aaeaf90a29e3411e13d5c4ccbad4a723f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 10:26:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ff3f1fa1aa249eabc184977c49531a6 : SUCCESS in 1h 01m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1cf00ec142e464192013d26da37126b : SUCCESS in 1h 24m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a1cc4c15760430d94ca234113ed607d : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a0aeb3df84452aab6e25be61b90ccd : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7784e06a341f4308b3b92f7cb8745d80 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91295992379d4cb7b5b218260ac928ed : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ef0d18a83c41c38152f27ecdeaf56a : SUCCESS in 10m 00s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/406fd923e49948f58b182c99b581ba43 : FAILURE in 1h 55m 46s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/483258ad94d743c19189b09eea191701 : FAILURE in 1h 19m 54s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d0162873c4fe4079b08c6278cb08b95e : SUCCESS in 1h 23m 09s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/659a0bd6e3814e789c1672a1bee0cf70 : SUCCESS in 1h 28m 09s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/8c06b102dc65426899d437a9a925e5ac : SUCCESS in 1h 12m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/981b5cc93845471eaddfcb123bff30fb : SUCCESS in 1h 32m 53s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/974811b75a4948389cb920019a730817 : SUCCESS in 43m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f08aab5ee1c940ef99212be2947aa904 : SUCCESS in 50m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/247362d1c8134cb08a2dbeb641efd3c8 : SUCCESS in 2h 06m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e36eb0a1e68741c18414b2cbfc33b401 : SUCCESS in 1h 05m 01s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b5c7cedb72834758b9ddad7898a803fe : FAILURE in 58m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/441cdc0a9dba4ee9a5f9457b2be6a999 : SUCCESS in 1h 07m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b07b73bfffd54fc09024ab73436c62fd : SUCCESS in 45m 16s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/abc9d37668a749bbba1ab6d647d7dde0 : SUCCESS in 3m 37s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/369025493ca5432ba8b2c61da7fcc837 : SUCCESS in 1h 14m 46s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/83182f2c809d498288f311ed69008115 : SUCCESS in 42m 18s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3bba2cc6c40b4b8d821ecc86439ddcef : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c1dc19327370f60c9492a54f82c40afb8f14b065","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-06-23 16:46:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4aa137826dc5a2bb682b3ff9129c891eb47a6d0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-23 16:48:26.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/53/737453/2/SRIOV-tempest-direct : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"2cae304be5c7c9e1b5f5d75c2f7ca18cd594e838","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-23 16:59:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9185f1ba64d818ddd2bd6dce3556a6a174175fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 22:32:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/98ecf092ba864db3b8187ebb15c6f263 : SUCCESS in 1h 02m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95fd14e4c9ce421485ec7baf0d610067 : SUCCESS in 1h 08m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ec91dfb11bf4a42bcd34fdb1bb646e0 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218fa2a6ea504079b1f76817b52fa724 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d00d3a756864eeaa140d84f1c8d2043 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e1f641ed7b244799408cc3e60ceb102 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/767638dd9ab3405facd1786845e96dde : SUCCESS in 9m 07s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b4b5d99f97614b628e3a09432f36f229 : SUCCESS in 1h 49m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c8e7c34b22fb4cc0bcb828ea31f35ce4 : SUCCESS in 1h 10m 51s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d474995b6ec2442d82ca062477981416 : SUCCESS in 1h 18m 09s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/1159f69db92949fc80b525b5f8404fa6 : SUCCESS in 1h 15m 18s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/7bb41715603142bc9dd537f5edd24395 : TIMED_OUT in 2h 05m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1c88c95b42b8429e8c45594635742ca6 : SUCCESS in 1h 56m 03s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/50d2a16c122a4f3f9b08f4fb1b104674 : SUCCESS in 42m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79ef6b712b464f259d5cdd19e94f32b7 : SUCCESS in 51m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/abc3910fa53541c69b16f23dab45dfbf : SUCCESS in 2h 02m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cc0e547cf5284f77acfee01d444d4375 : SUCCESS in 58m 11s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4ad81d6d72be440b86bc4e97dbe45e26 : SUCCESS in 1h 02m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da5dfd51df2b4b88bf47a586c4bb5913 : SUCCESS in 1h 04m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/95493874188c4be9b5a035abbdf04422 : SUCCESS in 1h 19m 11s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c6ce31bc01f94b5b80cc6d270317ebbf : SUCCESS in 5m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7e6ae9404a754e8dac7ad20051828978 : SUCCESS in 1h 19m 56s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/706828ba34184e3ea7cb8bf4e6ef6b24 : SUCCESS in 1h 02m 11s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/08daf526117a436bad1e40f738f79d0a : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4837a0e8b01ff049bf2214be62a5c6a01ef67512","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-06-24 06:53:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"43f05285f6a0a512948a49c63d6ca2d102650ed1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-06-24 07:23:59.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SRIOV-tempest-direct http://13.74.249.42/refs/changes/53/737453/2/SRIOV-tempest-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"536f40a7d48693848e04e436ca3953ad8ce62e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:15:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/68e10accf1f94c2397c465812d7083ce : SUCCESS in 1h 11m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/451316008c5d437b85abb1b9b8eb8b49 : SUCCESS in 1h 50m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f1b9657c61845e1b72142c69e1bc70d : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4520647fe3f44fd7b73171c69b887853 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a60e2f0d77bd4fd0b8dc7ccfd1d9c9d0 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/638553e34ab645e59d747d7314f9c1db : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40f69f02946474e81bebe9930dc9e03 : SUCCESS in 12m 36s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/20539e8dbd9d4ac9a34cfb3c360cfd05 : SUCCESS in 1h 38m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/75c53e945f68461f9a1f5d31d26b7801 : SUCCESS in 1h 06m 26s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/343d463b51244fdc8d2886e2e0789f2f : SUCCESS in 1h 21m 36s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/66544167bb714babb9b2c536dfb31967 : FAILURE in 1h 05m 53s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/bb5e84c1a4b247119fcf504bb5c4ba2d : SUCCESS in 1h 09m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/857404ce471a4907bf10518aadd215b9 : SUCCESS in 1h 31m 39s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a3609b389d454708b1ab0f854ea07870 : SUCCESS in 43m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4387c66da1b546aebc3b4fdefbc74fd3 : SUCCESS in 37m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9379026a90d54c67b6c6570f557af583 : SUCCESS in 1h 28m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/802f9cdc6dff485cb9ff0bf8784ded0d : SUCCESS in 48m 53s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a10c3fcf890c4c5cb74ed831917955ae : SUCCESS in 1h 05m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/52b917852eb04a4cb39d50e8f082e00c : SUCCESS in 1h 14m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1e74ad42b1c346dd820fbb06598ab1b3 : FAILURE in 54m 24s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0ba03e7a274440a8ab794372d8928f2e : SUCCESS in 4m 11s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f0840394f188404b8c7ada13a9aec540 : SUCCESS in 1h 14m 10s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b3c3fe8eaf8d4f3f9e20ae4a7278df73 : SUCCESS in 56m 33s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3b78d7bbef7b4e23a793de2e6a3ecdef : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f31a7c4db78905d60a87d634c0ef1d9d49b272c9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-06-24 19:05:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"88e47b0d6fe7591750dc4aa53f349bd3352db839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:54:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/df51c62488dc44628d50cd9e7852559d : SUCCESS in 1h 00m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1ff7730408c4a90af3ca5e2b40ea474 : SUCCESS in 1h 18m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b5814dc3384d27af676e2584f1d28c : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2883ae1371f448028aa6ee1ec1bd6beb : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c69395023f94e1da0948186bdb88d02 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7ae6a5083e746b892b09d9583d3a2f0 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbd9a6e59a748778789c150c258f74d : SUCCESS in 8m 58s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/bfa953b7c6bc49e0ad62a15b421bfbe2 : SUCCESS in 1h 38m 01s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7c5e0496d8ba46eaade8cd6e58fbc5fb : SUCCESS in 1h 19m 26s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d7e83cfadfc24b5094831d67fcbcce7c : SUCCESS in 1h 31m 41s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/13c3a1ddee0048348d9e977c71a2fe72 : SUCCESS in 1h 41m 07s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9489f6e989e344468fe2c406c02fdc0d : SUCCESS in 1h 33m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/afffe9089aa640d8a0c291d8b2d578d9 : SUCCESS in 1h 26m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ea1dfc6d699e4312bb3fc51371a0b3f4 : SUCCESS in 45m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1146d0ce135b4ac09bd640b30c28a0dc : SUCCESS in 1h 02m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0642327109824e6eb4939cc4ade62fc1 : SUCCESS in 2h 09m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/844fbe24ce6746b6816d5a9a74338538 : SUCCESS in 48m 15s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/84c6b5c722a245f5b0a0212dc2594220 : SUCCESS in 1h 16m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3dffd91c7604acbb455b9a85643b8be : SUCCESS in 1h 13m 01s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/46708b48f9df49be90d2318e89f7a818 : FAILURE in 1h 07m 28s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9fdf2c1a014d4f45800975a11b9fe880 : SUCCESS in 3m 46s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f1ef114d01bd428c95eaeb72c39e4d5d : SUCCESS in 1h 19m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b085c881d858448984aafeda7bc2a6eb : SUCCESS in 45m 01s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/72520ad9e7a44e82bce7ea23118c0d86 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"9dae25b03c3d23f13296ee399bc30ae80656ba64","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-06-25 05:51:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"98569961656fb648491fd97f30d6c125a354a99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:01:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a7550bc10e184bdb88c8054a9c081d89 : SUCCESS in 1h 15m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3dac43e2cbfe4ba693d5d2c7d3aa50ac : SUCCESS in 1h 27m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9ed36045474449fa7cbbde99e321c84 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1eb9ccb69f42ff9900da66ec5b0f1e : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e015f3ea9c3c4735a921a89df1a8dc92 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69c0f67f01aa4ffea1650e123e13bd0a : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9afc20a88ef4d6497c59089a62e2c1d : SUCCESS in 10m 41s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/33d7988301fd40ac8f9f7f670bf4cc4b : SUCCESS in 1h 58m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7a56c26e74f4484885b18453ea337da8 : SUCCESS in 1h 43m 27s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f0e664b651794d428c1b50f48b98844f : SUCCESS in 1h 28m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/acd628bd69f249ae93b729497a4e9f2c : SUCCESS in 1h 18m 04s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/dffba8d7f1d24d10ab36961f5105bfc5 : SUCCESS in 1h 24m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bf606064a23d480e8de7be916c3806f8 : SUCCESS in 1h 43m 18s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f736bce21ecf4fab9fc278c34a2e6993 : SUCCESS in 45m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/014f2462e2fd42d99ffc7b3f1334181c : SUCCESS in 1h 02m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1cecc0e264b44c839fad85a5c50bcb2d : SUCCESS in 2h 04m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5363fac00c6a4bd79ceb6913875a78e4 : SUCCESS in 58m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1d550731ff294f58a52d0e5dd2072430 : SUCCESS in 1h 20m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4f63050184043b5a2345217957f4328 : SUCCESS in 1h 13m 32s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0011854896bc4e38b5d0869c93e9117a : SUCCESS in 1h 13m 09s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/20903ecddab1451c971076d8a9a633a5 : SUCCESS in 4m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/95166c99264d4ab1abd21cd5332597ef : SUCCESS in 1h 25m 10s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/66e9ccf82cba4b9b92d3f3ae4ee57082 : SUCCESS in 1h 06m 33s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e91f3fdad4464b93906522e077622756 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"60f30d7431846fca00a84647c5e0906771e603e1","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2020-07-13 13:26:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aabd43c4a4ef75b5c65298f24115052c6a3745cb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-07-13 16:03:06.000000000","message":"Patch Set 2: Workflow+1\n\n+w as the patch already has 2 independent +2","accounts_in_message":[],"_revision_number":2},{"id":"e5e70739493727cf9ed2c5a828d07b5499e997bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 16:03:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0dee0cd9d6e8d07a715fd57722eb8a7d5abe6beb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 18:06:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8fdf3c97dbf44ae1a1e956f6dddcf339 : SUCCESS in 1h 03m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3355e5b5f05d4df4a28c74c23b3e8e3f : FAILURE in 1h 02m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/797a04e0058041b3bb2bff94ee43a703 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0f69693f2f456aa5481b75aeb7e542 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b77154e13ab4f74953e748837f133d3 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fbdc4855eea48e7b5c8f2617c88021e : SUCCESS in 13m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5da42654235343b0868b5ea3344455c2 : SUCCESS in 1h 59m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57e7c454e80045cd8fb66f12f430d9ea : SUCCESS in 1h 14m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73aeb0e182ba482a9d031092e3ace61c : SUCCESS in 42m 40s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f5e457610602445e8091829d038164bb : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"229dc2222057d736db8249b597ef5f247a565548","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-07-14 08:34:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6f7e01f71ec4bd8b78ee8cf302d2c31c3676bee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 11:20:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c07b353647c04674a2d04b85eded70f4 : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d540fa7503864f008976442bbc6bf127 : SUCCESS in 1h 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb4efd6b6965466bb03f0e01962fd73b : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a894b7108b8e480291ad795f75769231 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b53e613e9a049c7a027d28a0aa759a6 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b5658f536714c85b81c9efe2c8ae8fc : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730e89fb80e04c8792389f3fe6db72db : SUCCESS in 11m 14s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0aa7614651264d77bac85720732fe89c : SUCCESS in 1h 39m 58s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/61474189afaa4f20b5830a3883bdf543 : SUCCESS in 1h 26m 29s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/fc103feb690a4af9ae760aeca69f61b2 : SUCCESS in 1h 38m 40s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3055353e6e754d419ca9bec046246423 : SUCCESS in 1h 25m 31s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a5166675d2a74ba2ae1fad878f1797ca : SUCCESS in 1h 40m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a164348ec0fd410992ffc50f103060eb : SUCCESS in 1h 44m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/43a8f1eecdf94eaeafd27a77b1b1c051 : SUCCESS in 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0e8f89585944def949dd2e3b28dd41e : SUCCESS in 56m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/779c7662e07042f1948043100bf0d7f7 : FAILURE in 2h 28m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1f486e380efa4e2e9925ea1d18995fdf : SUCCESS in 47m 50s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/698c515e0a4142cdb4182f0444f80b30 : SUCCESS in 1h 02m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f527caad1bec422fa3b156925021e38d : SUCCESS in 1h 12m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f06b213c068b49b995b9ed011f1c1292 : SUCCESS in 1h 12m 49s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/dfbcc6e6f7d1453fa712f75d6b9c0229 : SUCCESS in 3m 37s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9830921d81d549d4b74e83d317a8bc72 : SUCCESS in 1h 03m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/784bcf06812b408a94ebdac58aea8063 : SUCCESS in 1h 22m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/dbbf660504ea4fe09bf652053c5a17a6 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5dc078a7b4269476a01ae9b87cbeddb95ac6a945","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-07-15 07:14:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0cd5a5349e3e8c77007fb1c315473d35686b7afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 09:16:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/897071fe114645d1abdb764900160fc8 : SUCCESS in 1h 00m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1782df904ac4cb88671bd1082eb7480 : SUCCESS in 1h 11m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daeeb891386f4fc6b61046178193dc8d : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dec8dc4e04c4ff0bd3b70a8e6445412 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef543ebe18df4d309cd15b9c4c0f9857 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e55b45ce95b04497bc170b27fed0dbc2 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58db6215d2d849c3b6707d492b203553 : SUCCESS in 14m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/583863a059004897a143785e1711f507 : SUCCESS in 1h 50m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/bb4a9078ea1d4f0eac02f4e809f59752 : SUCCESS in 1h 28m 27s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/237dd80088154bb385c00bbc5b8c881e : SUCCESS in 1h 29m 37s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/3c75550d0ccb42f880a7573d1c7ad47a : SUCCESS in 1h 30m 32s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/8e943d711eb34850b5c1d14f9d17e4ff : SUCCESS in 1h 13m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c85e16a49c7e443f8cc21a6e724483fc : SUCCESS in 1h 43m 20s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a53446e6685147b2bda9a215a7c91424 : SUCCESS in 40m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5d5bc56603a4cb098c14b8fd48b9d8b : SUCCESS in 1h 02m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49cf732b48dc4b1d824d476578305660 : SUCCESS in 1h 58m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8737a46f70424c4c9e1aca412ef55b85 : SUCCESS in 1h 01m 07s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d2023e62114c4ce893a41513eb94b1a3 : SUCCESS in 1h 17m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/92bba77e199647e2bb8d81c1000b9234 : SUCCESS in 1h 21m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b947e6d4d0e04004aa4e13c27f83d139 : SUCCESS in 1h 02m 46s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/4541b26796de4715b18eda46aa314362 : SUCCESS in 5m 01s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5945c4fe819e42199909a3eb086de373 : SUCCESS in 1h 17m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/68543576d41c41c7acc1a8aae9c6f6bb : SUCCESS in 1h 11m 17s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5ede07537a5c4eb3a2dd50a4bb755160 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f7a61546d9e1d81c2802b42101a9572cca3a876e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 09:16:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1da6e0ed371bb2a56a40564a15a0d5e66ccd2a0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 11:28:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b5797a6c054ae101872b6d1445f8e9fba7d83372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 11:28:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca30117cbdcb41439eaa4447d878952c : SUCCESS in 1h 15m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dacc1ec70a740c6bfdbcc381f4537f1 : SUCCESS in 1h 38m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24d6f279e8545c98cdb47b7559170c2 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca59b52050a64488b9c40aa64c52cf20 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f500ab490a040c08535e61474106fa2 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ce4efed2f14e2bbb3a9008017c0bdd : SUCCESS in 9m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/42e6c41ca7c144778c54b84a96f13ce2 : SUCCESS in 2h 05m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71a2e0f7f0214a4d8bd80dd70f619216 : SUCCESS in 53m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a298a39254104f299fffcbe43f5ff5b1 : SUCCESS in 1h 00m 22s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c1a2e4055f114740abd596c33dce6576 : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6b0afb82f2ce1e77fa33f8e6e5cf663341bb8d2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 11:30:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ddd6200a564f20a6b6d7604909b601 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fba44537c75e2d586055ff1467d77af073913b1d","revisions":{"1afd33932c5a2c5bae0c7c97a2b5c0e29c119d1b":{"kind":"REWORK","_number":1,"created":"2020-06-23 07:51:32.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/53/737453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/53/737453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/737453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/737453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/737453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/53/737453/1"}}},"commit":{"parents":[{"commit":"206a9abec5e317b0644072d1308f74ce4f6a326d","subject":"Merge \"Fix request body for compute unshelve action API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/206a9abec5e317b0644072d1308f74ce4f6a326d"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-06-23 07:49:41.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-06-23 07:49:41.000000000","tz":0},"subject":"Unskip test_server_connectivity_cold_migration_revert","message":"Unskip test_server_connectivity_cold_migration_revert\n\nThe bug which was the reason for skipping the bug has got expired\nwhich suggests that the test is no longer causing problems.\n\nChange-Id: I4c68caa693744e06de6be89a8f09d8f7e8ce8ba0\nRelated-Bug: #1836595\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1afd33932c5a2c5bae0c7c97a2b5c0e29c119d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1afd33932c5a2c5bae0c7c97a2b5c0e29c119d1b"}]},"branch":"refs/heads/master"},"fba44537c75e2d586055ff1467d77af073913b1d":{"kind":"REWORK","_number":2,"created":"2020-06-23 16:46:21.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/53/737453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/53/737453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/737453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/737453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/737453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/53/737453/2"}}},"commit":{"parents":[{"commit":"206a9abec5e317b0644072d1308f74ce4f6a326d","subject":"Merge \"Fix request body for compute unshelve action API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/206a9abec5e317b0644072d1308f74ce4f6a326d"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-06-23 07:49:41.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-06-23 16:46:15.000000000","tz":0},"subject":"Unstable test_server_connectivity_cold_migration_revert","message":"Unstable test_server_connectivity_cold_migration_revert\n\nThe bug which was the reason for skipping the bug has got expired\nwhich suggests that the test is no longer causing problems.\nJust to make sure the test works, let\u0027s mark the test as unstable\nfor a while.\n\nChange-Id: I4c68caa693744e06de6be89a8f09d8f7e8ce8ba0\nRelated-Bug: #1836595\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fba44537c75e2d586055ff1467d77af073913b1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fba44537c75e2d586055ff1467d77af073913b1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
