)]}'
{"id":"openstack%2Ftempest~737750","triplet_id":"openstack%2Ftempest~master~I59e2cb3bd0ee9a8b525b5904fd39eba446205efd","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I59e2cb3bd0ee9a8b525b5904fd39eba446205efd","subject":"Add unit tests for check_uuid.py","status":"MERGED","created":"2020-06-24 11:32:55.000000000","updated":"2020-06-30 21:27:22.000000000","submitted":"2020-06-30 21:25:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737750-1593552349642-6d9093bf","meta_rev_id":"6f63b2cf63c581b321a57f6453856fc0daf66848","_number":737750,"virtual_id_number":737750,"owner":{"_account_id":31983,"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","username":"somtom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 21:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2020-06-25 08:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-06-30 18:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"date":"2020-06-25 08:47:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-06-30 18:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 08:47:21.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-06-25 08:49:43.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-06-30 18:51:29.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-06-30 21:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f55a86d14dc03fe8020c0fb73cc6e223659e0fc","author":{"_account_id":31983,"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","username":"somtom"},"date":"2020-06-24 11:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"255e63dea51acf49b866a39030925895b246577a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-06-24 14:30:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLooks good, while we\u0027re waiting for the gates, i have one small comment:","accounts_in_message":[],"_revision_number":1},{"id":"1b166082da1ce71ccf3cc8538287066279f5e7fd","author":{"_account_id":31983,"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","username":"somtom"},"date":"2020-06-24 15:04:56.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ad6dcd38198e9d0a1c24c5df75d1165a533511ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 19:32:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a45a4a7f43334dd2a994de6e9590bf5a : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa69b5fd9354224b6c7af6a8d91646a : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5564353c0a844aaa78b96834d74bd86 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/977a40671bf04c33b4352be7ee0597fd : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047446499ba54cbebe01c1e404e74a00 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":2},{"id":"025f440f50bdbc7e50cb5c5076bd2e55f2c89a44","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-06-25 08:49:43.000000000","message":"Patch Set 2: Code-Review+2\n\nlooks ok, thanks .. it improves unit test coverage for the check_uuid.py script from 0% -\u003e 42%!","accounts_in_message":[],"_revision_number":2},{"id":"b1eea58b9434a56808cc0a88ff0ceb175fffd966","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-30 18:51:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlgtm thanks for starting this. As next step, we can add the more test to verify the missing/--fix option with run method.","accounts_in_message":[],"_revision_number":2},{"id":"d3ec91f9e8717e97f62761ed2b5cebe6ee879620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 18:51:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db1a4c7dde5e97f33205901b08049f57cb04e0cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:25:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85cde67836094d49921ba63f5c54e679 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9da0a2a915684280a6e110a259c5ee67 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/737942d974684e65b98d87f816d14a60 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec51d850a8f4641a763b10aed1ac706 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5b91a51fbde0098be8158146cbc6f433f3b56053","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:25:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6f63b2cf63c581b321a57f6453856fc0daf66848","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:27:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b474bc02c12c4168ad0b14f71143bbdf : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"536d8284f4f2d81799275e5042e21065329f8b35","revisions":{"9d8c86388172819c4b9ddfd2651391f0d1a6767d":{"kind":"REWORK","_number":1,"created":"2020-06-24 11:32:55.000000000","uploader":{"_account_id":31983,"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","username":"somtom"},"ref":"refs/changes/50/737750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/50/737750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/737750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/737750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/737750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/50/737750/1"}}},"commit":{"parents":[{"commit":"b3229878a33d1cf01e29368a76ef01bfb282edec","subject":"Merge \"Unstable test_container_synchronization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b3229878a33d1cf01e29368a76ef01bfb282edec"}]}],"author":{"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","date":"2020-06-24 10:41:45.000000000","tz":120},"committer":{"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","date":"2020-06-24 10:41:45.000000000","tz":120},"subject":"Add unit tests for check_uuid.py","message":"Add unit tests for check_uuid.py\n\nChange-Id: I59e2cb3bd0ee9a8b525b5904fd39eba446205efd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9d8c86388172819c4b9ddfd2651391f0d1a6767d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9d8c86388172819c4b9ddfd2651391f0d1a6767d"}]},"branch":"refs/heads/master"},"536d8284f4f2d81799275e5042e21065329f8b35":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-24 15:04:56.000000000","uploader":{"_account_id":31983,"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","username":"somtom"},"ref":"refs/changes/50/737750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/50/737750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/737750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/737750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/737750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/50/737750/2"}}},"commit":{"parents":[{"commit":"b3229878a33d1cf01e29368a76ef01bfb282edec","subject":"Merge \"Unstable test_container_synchronization\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b3229878a33d1cf01e29368a76ef01bfb282edec"}]}],"author":{"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","date":"2020-06-24 10:41:45.000000000","tz":120},"committer":{"name":"Tomas Krchnak","email":"tomaskrchnak186@gmail.com","date":"2020-06-24 15:03:08.000000000","tz":0},"subject":"Add unit tests for check_uuid.py","message":"Add unit tests for check_uuid.py\n\n* adding a patch\n* applying a patch\n* adding uuid to test - with and without decorator\n* adding import line to tests with uuid\n\nCloses-Bug: #1878171\n\nChange-Id: I59e2cb3bd0ee9a8b525b5904fd39eba446205efd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/536d8284f4f2d81799275e5042e21065329f8b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/536d8284f4f2d81799275e5042e21065329f8b35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
