)]}'
{"id":"openstack%2Ftempest~745812","triplet_id":"openstack%2Ftempest~master~Iffbaa7489d219ae4138874fe6c5702e858417573","project":"openstack/tempest","branch":"master","topic":"bp/testcase-description","hashtags":[],"change_id":"Iffbaa7489d219ae4138874fe6c5702e858417573","subject":"Adding description for testcases - compute part9","status":"MERGED","created":"2020-08-12 09:13:15.000000000","updated":"2020-09-06 13:24:00.000000000","submitted":"2020-09-06 13:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745812-1599398570662-33708db1","meta_rev_id":"6abf9efce57fd2d7ec7e1e9db1c89b632ed51eca","_number":745812,"virtual_id_number":745812,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-06 13:22:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-14 01:07:13.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2020-09-06 07:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-09-04 15:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2020-09-06 07:08:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 15:55:00.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-06 07:08:33.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-09-06 13:22:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31eff14c61e0e2e61a91d85a8824395aada459f3","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-12 09:13:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"327a6b0c38f1cbd3390c12f22bbab67d4dd5e135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 11:16:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5feb66708604d6d953fdfc49d61e4d2 : SUCCESS in 1h 08m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c990150ad52f4d5682fe241b7b241159 : SUCCESS in 1h 07m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c62b79350704b369c3d48b61167a401 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24729e9bd2c048e8b2aa050185be9696 : FAILURE in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8499de26c874181b2ab8ce480d64e90 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef9dec18c1f4476e856f34be1bb74b6f : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b493d06a92364f4aab653276aa5a0850 : SUCCESS in 15m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/55c3988c61204019b0c006d1dc84f1cf : SUCCESS in 1h 37m 04s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/90b3400d9947491ebecf92f7e08cabcc : SUCCESS in 1h 16m 44s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/fcc465aea05043fdba2b50e73bba9e33 : SUCCESS in 1h 17m 25s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/47469ad2b2294d93ad365d439ff89166 : FAILURE in 17m 11s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/92e077ec03954b2ea7a33853f667d523 : FAILURE in 1h 46m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/42f62603171342ab9ef48b763d098c38 : SUCCESS in 1h 34m 19s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/48d7f3875aac4259b9bc6a88b7a338cf : SUCCESS in 43m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf758d0e9d134f179e9f9d649a2e91db : FAILURE in 26m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c6874b7e3894520aea542db406151e8 : SUCCESS in 2h 01m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/59e8308e3ac4419c9371bfcf36e9228e : SUCCESS in 55m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b5b6e6d050f14322a0a2af300712f649 : FAILURE in 27m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2688b2741c04b3aabcf684f59e263f2 : FAILURE in 1h 03m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/51bd57f29537435380622f6ffe2402bf : FAILURE in 26m 31s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/92c7a2f5fcaf4c019f3113703a08fcb0 : SUCCESS in 3m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5f8b65a4f8164e8fb798253a5b3dda41 : SUCCESS in 1h 12m 01s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b7f6a828454c4bef80168a421ab9fca1 : FAILURE in 24m 53s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/fbc0aaa719c84c148c81ebb0d3b14e67 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1a6af23e468b6663c5b2c9a6c3e3e270b2ecfc53","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-13 02:47:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33972c686adcd06ecde3c03aca6f3420d4954d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 05:17:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f5ac067fd494980b65809ed1e00cdc9 : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/480d1f86a9fe4de7b8c8a1c6417944a4 : SUCCESS in 1h 34m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fab3be612c240a69f7b3d685ffffe63 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7701768dae8645f68cbd19430d4ca920 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6cd8303b3f54d6c8a2a93301d0f9ff5 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f80ac6babf438fb88d31f8486d5e46 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997a62859508473a882e2523cca862fe : SUCCESS in 10m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/48aa31ec017a45c1a6a977938fa7387f : SUCCESS in 1h 25m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6ec440363df94266931050d6304ec654 : SUCCESS in 1h 07m 17s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/26d24889f750475bbe38dca419e2c35e : SUCCESS in 1h 27m 51s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/137af9f94dfa47dcb4debd9244b64ee7 : SUCCESS in 1h 39m 34s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/3769776475d845a4a3d476ceedf7eb14 : SUCCESS in 1h 43m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a66ee8c8943d4530bc8bcb62abe323dd : SUCCESS in 1h 58m 31s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3b9ef9d5a9944de49ccea4b83d215d15 : SUCCESS in 38m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e0a2826fbea4a5aa1e02616b3e5e3a1 : SUCCESS in 59m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3755ecfe4c254fdd9fd03ab28a131066 : FAILURE in 2h 27m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/42ddb6738e764108a88e195039da41ac : SUCCESS in 49m 23s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fa4683ec48504a0bab663a3fcc579b0d : SUCCESS in 1h 24m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f0ae096a64a4235adb564fbc729d1a1 : SUCCESS in 1h 12m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/009333826a444f4087e5aac315d733cb : SUCCESS in 1h 40m 16s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9cc93f2eb6304d93b3b715fa01c280b3 : SUCCESS in 2m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f137c67052a646f0b7c8fcc0786895b9 : SUCCESS in 1h 09m 23s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bc4f8db41e5c41df99ff151338a05bef : SUCCESS in 1h 14m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2aa0c243c5a5424cb31f93657d8e367c : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e4e26a8a9bacb3f639a95c1de9bbd3910ebfdfff","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-14 01:07:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be226298411bdf196683f55c4dd75238e9213ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 03:16:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d974938385e84ea68f5ff8a698e58921 : SUCCESS in 1h 18m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed87e60f1aac4e3c92dd2cb13f8929cf : SUCCESS in 1h 15m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/febf8cdca661437192a982f23cdb4548 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7102b84e93f94f2984d7a4b4e4f5b587 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e353d391cea54ef288b08fcc30df23ba : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f22a4c4037b94c0c950dd8fba42750e2 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959609d6e29e4b4bbaef0bbceb1d1a86 : SUCCESS in 9m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b2c13cc8bc8043288c981ef74a90a2a1 : SUCCESS in 1h 34m 52s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/8524d28ada314687915f54d16218c41e : SUCCESS in 1h 36m 50s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/3ab2d98853e0499c82bd7cb5ae622e10 : SUCCESS in 1h 11m 29s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/829f22bdb02f42dcb3fd79ce65b60f12 : SUCCESS in 1h 07m 01s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/2246bc2dda2b44f480fd804feea3814b : SUCCESS in 1h 08m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ec4d4dae792b488e8e7b29f7b1951470 : SUCCESS in 1h 30m 05s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/512b5d002ecc4d2b8d08fe46ddc76404 : SUCCESS in 41m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/483db3449ddc437b8f2b45550ce91b86 : SUCCESS in 52m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/917d92cd47914c96ba34fa68c6be89af : SUCCESS in 2h 06m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ab92b41dad9045b0b74aa06336e57db4 : SUCCESS in 1h 15m 25s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a4af647efb3441fe862d66b6c32e0117 : SUCCESS in 58m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa0c7102078141c9b334ef0090a78072 : SUCCESS in 1h 04m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/13df899d47df4b649a0ecbe788d7820a : SUCCESS in 1h 30m 26s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/4638745aa3e247118e7fc97d630bae6d : SUCCESS in 3m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0196b2c725684b15a8791a15eb3187cf : SUCCESS in 1h 33m 32s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/23971ec779ad4634bda96de361206958 : SUCCESS in 1h 17m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2b6a2a20274e4e1d9ea163cdeccf0ea9 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"63e4232b342c0ac66e83f9ff3642ca5f320ebc86","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-04 15:55:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a5deb25f7b4296a7b4d1ad2afa55cd1e794fba3","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-06 07:08:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02af7bc2bad72c7007245a9477b8967089674c28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 07:08:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa80adf94bd9b740d549cd407f04ec17bee91570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 08:31:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/431a044b01f24766bbf8fafc577d4053 : SUCCESS in 39m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d344fda919474f488e6edc8b30144969 : SUCCESS in 50m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e667b0bdbfe447cb86e89dbf145fc2cf : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e86710e742e4c01979342a1259388ab : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ff81d53bd104b60b711aad7dda6c331 : FAILURE in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341f962ff7b94f79bb332ad5c281a0f1 : SUCCESS in 9m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0db646bb6fc74e62ba74e41e7f3898b7 : SUCCESS in 1h 21m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9227e512b90d417e9b142f32e7ca7bc3 : SUCCESS in 1h 00m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b29383d97fd2423fb311398b6157df30 : SUCCESS in 35m 45s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dca651b832d84e3382a9848c50a53b59 : SUCCESS in 45m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b8eab232ca069709a70f48aa7d13c25af4de97a8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-06 09:01:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"113b3cc839691903cd1caf2273397a0b4cb009f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 11:18:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c292614bf63c41c783b2d7949f7276fe : SUCCESS in 56m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8c6bdaa672d433087e1f8638547333f : SUCCESS in 1h 12m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24515db42c304fef81e333b10dd4f52d : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dbb84a391884f93ae707d0b78f4c1ff : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b39d4119f04ae38b56d79b8f8a08e7 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3903d2672fb4365aeec7a24ae6d9719 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634f85d24de349b08134873312082d1d : SUCCESS in 8m 09s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/dec0bbdc1f914337a26ec151efc4d86c : SUCCESS in 2h 10m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d5460ae1d138400a9e54afaa1965dd48 : SUCCESS in 1h 16m 44s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1d7b714cd14c43dc88f2d24411ba74c7 : SUCCESS in 1h 12m 22s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6e7f1c0a005646f9bbba218269947289 : SUCCESS in 1h 28m 15s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/c8846530b5824b1183ebd4bc4f21dc9b : SUCCESS in 1h 22m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7951d23a40e94a76a19387ff0eccc193 : SUCCESS in 1h 37m 28s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0a53ff37667e495eb56efc3614404ac0 : SUCCESS in 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/122c3b1320104e45b4fb2ad1d6a21ec8 : SUCCESS in 59m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff1802dc4135440aba619107ca26d219 : SUCCESS in 1h 59m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/646d03740f944aacba1df54b7cd73194 : SUCCESS in 46m 50s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b15290fa00e24520854f27253025c8f9 : SUCCESS in 1h 27m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73e1f6f3add74f9b9fe3c8a9d2c0a182 : SUCCESS in 1h 03m 46s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/fceaabce40f449e694ea864d4a0a1222 : SUCCESS in 48m 22s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9486375e04f34510907872aaea771be5 : SUCCESS in 2m 19s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5c143c22dd3b4cdab7b58cb9c9fd1ec9 : SUCCESS in 1h 01m 53s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/672362d45d55437e9f8ea4a95451aa10 : SUCCESS in 1h 17m 47s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/59cbc95052de48c698e5427e29285de2 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"63ff2b0b596c40632f71f8295dd63f1240d788c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 11:18:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59fb191ad81c2cb30b13919a97c4cff6a9f0eec9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 13:22:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e3694bfadb5b9310de1e894c5784e9d292d82f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 13:22:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9342d44be7414b399111063e50fc391a : SUCCESS in 42m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd8e6beba7bc4fe6b7b346328ee4f21f : SUCCESS in 1h 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834bd57bc6404a2ba1d65d16d2981384 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/725e242671954dc0ae691f94f57f63e3 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/557e9f4e67df4f1c86c6ba502040957e : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f984234eefb548f9aa4c326c226b6767 : SUCCESS in 9m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8dc2a070c9784d2f87935db6d56f8ecf : SUCCESS in 2h 02m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89fb84b1138f4f66a61bb5d7c566bbbb : SUCCESS in 44m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3843399037b44b7f9e4b41f00cb89c8d : SUCCESS in 53m 03s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/73fe19a180634978a7eccb5625aea613 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a76b771162265f9955dc7976d4b8b7e2411feb41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 13:24:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9e21a66b94407885518b6f49bbfa75 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e354fa7675a614e63f8430b86ab1539d33e1154b","revisions":{"e354fa7675a614e63f8430b86ab1539d33e1154b":{"kind":"REWORK","_number":1,"created":"2020-08-12 09:13:15.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/12/745812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/745812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/745812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/745812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/745812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/745812/1"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-12 09:07:50.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-12 09:07:50.000000000","tz":480},"subject":"Adding description for testcases - compute part9","message":"Adding description for testcases - compute part9\n\nWhen Tempest is used in customer site, often we are required to\nprovide a testcase list including testcase names and descriptions.\nNow no this kind of doc is available, so we can add descriptions\nwith the format of doc string for every testcase, so later we\ncan generata such a testcase description list.\n\nThere are hundreds of testcases missing descriptions, so we can\nadd them gradually, and limit the modified files in one patch\nfor the convenience of reviewing.\n\nChange-Id: Iffbaa7489d219ae4138874fe6c5702e858417573\npartially-implements: blueprint testcase-description\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e354fa7675a614e63f8430b86ab1539d33e1154b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e354fa7675a614e63f8430b86ab1539d33e1154b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
