)]}'
{"id":"openstack%2Ftempest~745814","triplet_id":"openstack%2Ftempest~master~I212cdce9713c80491153b0bbdb313e75c1f96f1d","project":"openstack/tempest","branch":"master","topic":"flag_fail","hashtags":[],"change_id":"I212cdce9713c80491153b0bbdb313e75c1f96f1d","subject":"Improve tempest-cleanup ansible role","status":"MERGED","created":"2020-08-12 09:20:13.000000000","updated":"2020-11-05 10:20:41.000000000","submitted":"2020-11-05 10:18:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745814-1604571503564-294064b8","meta_rev_id":"132db10729f2306199da8a34012114142e039a32","_number":745814,"virtual_id_number":745814,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 10:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2020-11-05 07:48:41.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-04 00:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2020-11-05 07:48:41.000000000","_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 13:08:13.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-09-09 03:44:06.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2020-11-04 00:50:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-11-05 07:48:41.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-11-05 10:18:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df232b9369192c875d949a5ccd65d5aed5695ecc","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-12 09:20:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2b02ade670489ba1214a74b75a3d3a2eefc8325","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-12 09:23:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8a37323babfad1daa54c634ac636e4bacb3b968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 11:30:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4f8b68ce8d4b46e9840733fc2a42341e : SUCCESS in 1h 03m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/004099ec3b5242c38cbf813fad53dd2e : SUCCESS in 1h 01m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/945d8544ebf043aa97f276d1a36e2f3c : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62efc6bb65014af998fe010ecb508b0e : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf9eca27c37e48418364fd316c1f542c : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c6f6b0e60004442947c9675ced2ed6b : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efb649a43ca406d8498951bd3d2b869 : SUCCESS in 9m 32s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/048546f234f24b1b8456de0e6aaac6d2 : SUCCESS in 36m 07s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5218bdb0a3cb48e28bf038ec22b8a8fb : SUCCESS in 41m 57s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e2c9e395ec47493684ba7be16646e4d4 : FAILURE in 1h 49m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/f98ae70f407847f9bd50a9882931820f : SUCCESS in 1h 30m 52s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2498eb9fe144493e94e04b6b749c5597 : SUCCESS in 1h 24m 36s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2fe25f30d6704f2ca0faa05e5cfc8197 : SUCCESS in 1h 30m 24s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/c1c87d89880e49e2b4b1f2d1fc7ea64e : SUCCESS in 1h 26m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/678d07854130462b873344915b831474 : SUCCESS in 1h 29m 43s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e1feb0bf133241f0b166e2bcb4546d18 : SUCCESS in 45m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db5b552711d64626be4ee96d0f74a29c : SUCCESS in 53m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa5d6b04883649d48d6adfd68e14e48a : SUCCESS in 2h 01m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7f02b226cb37462e81651bf4a392a524 : SUCCESS in 50m 33s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/aa86994e8ef849c7b71b57f9dc17c343 : POST_FAILURE in 1h 17m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b917f98d2a542e9b297455dadc5ba3e : FAILURE in 1h 07m 47s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bc5442e0433c4e069c601cc45724f24e : SUCCESS in 1h 16m 03s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/13e103d285fc4d09879e71b6a1846cac : SUCCESS in 2m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/310f76d900ad49f481807350d60d1d04 : SUCCESS in 1h 09m 26s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9c8c444b7ab94b24a91e03c1d063f8e0 : SUCCESS in 58m 11s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/cac9386d99674642ae26870d92d91fbc : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5c2660f3eb9fc2ac7c49fffa42fb0465722535e5","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-18 10:17:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea7086068f5f3574a0cb9ec38e9e0776baa5e2f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 12:06:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6cb3583de63d4d7dad96e55375048ccd : SUCCESS in 56m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07acdc2d75fc40e7be8e1d3863ba91ea : SUCCESS in 1h 13m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e101bc763443403eb5bce8bdc33df22d : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bb19c32b5c40898da0574dc8018078 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/748fd34d3dd44502af8f80f0f5ca1589 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa528eaf41384e23a56eb1c1e6c4bae9 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac076d9eb254e7bbc746c3881631e47 : SUCCESS in 8m 50s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/6f255b8959c5463e867056711759ff6d : SUCCESS in 34m 15s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8f8f327cce194ef4935ddd6370dfbd3f : SUCCESS in 34m 45s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c0f29de866574e35a97022d13573e2f6 : FAILURE in 1h 37m 44s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/108c0ff9cb60480ab87cfc6a8cfd5ae4 : FAILURE in 13m 20s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1624794016a742d9814ba6f4fb5537a5 : SUCCESS in 1h 23m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2920a02268dc493e98607e55b7af6130 : SUCCESS in 1h 18m 08s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/0652be259fc9487faf94a2e7e7307b03 : SUCCESS in 1h 18m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b1e8f72d54b64e9b95d098d98a84c066 : SUCCESS in 1h 34m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/56fad62008a54f94a68ba0d3effaf51d : SUCCESS in 44m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b91e71e488e94a388e974f64742ac31c : SUCCESS in 57m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ddc0daedb7f40eaa5472e42eda6f446 : SUCCESS in 1h 47m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/44c06548c10d42d899a97c57f088512d : SUCCESS in 48m 54s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f6806896a45d4fd880c7befcd169e646 : SUCCESS in 1h 11m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b3ff9f413a34458ab74ad42f405bdd9 : SUCCESS in 1h 03m 33s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e03fbddfe5d140d9b1390c9759b74af0 : FAILURE in 56m 47s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a82ae797b0a3484cb82d0675f565524e : SUCCESS in 2m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/742cff74c3944623a2510f8c67624a4b : SUCCESS in 1h 23m 58s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/005c8a6bfb0c472aae087b7f89e2450b : SUCCESS in 56m 29s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4f6a7d98766843de8044ef5cd5087e20 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"02b2903eed1777ea3f2810c2c6c5272a5de71850","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-18 19:17:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a16a349bc37cd0be646211960f3212238befcd3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 21:24:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/039043cd40b2477db2e7d5b42ec4ed96 : SUCCESS in 57m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7bf8ad5c5a947fea47076ef9378b4fc : SUCCESS in 1h 14m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a601fb2858504d8f8f9dc889b0062f14 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7437e5c1b14f868b2c241bfbc12531 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94321822fc5e4452a123a714f4e76d05 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb9666fab167494698adb0b4f5d72a73 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6fcc66a8c3f451bbbede0b5f2825322 : SUCCESS in 10m 44s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/d7e47edf9ec049e2a0fafdf6079e5433 : SUCCESS in 35m 03s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/fea9dcf112c141ccacbddb0edd258789 : SUCCESS in 37m 58s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f1b91f966b5a441fa63fd8bcfa2ff922 : SUCCESS in 1h 35m 19s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d16af3e4cac541f7ad252db4cf796494 : SUCCESS in 1h 13m 29s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d1e0f16db50f4cf980002ff163def07f : SUCCESS in 1h 08m 39s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ea19a8d3ca274c419065a7d6d84d2076 : SUCCESS in 1h 19m 06s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/de318847c61245b196f6e38df48de611 : SUCCESS in 1h 09m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/45d3a9a0f8064861a2066a66bd649e37 : SUCCESS in 1h 29m 25s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/933d5d3d88de41e5898d5f4c5cb395e3 : SUCCESS in 42m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aa419912b66443598a5a7beda6f7774 : SUCCESS in 51m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/636944f15f9b425681f9c467e61ab9b5 : SUCCESS in 2h 04m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c3268ffec9cd4eaba6bc480a99b1fa83 : SUCCESS in 49m 33s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c8255e4f082a457c9aafa959dec27448 : SUCCESS in 1h 02m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5396f3121a5243c8a89b2f2107c56805 : SUCCESS in 1h 06m 50s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e91622714f7840caa3c1d45dd790f834 : SUCCESS in 1h 07m 23s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e2125c2565f148279624190c1036bec8 : SUCCESS in 3m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d510187f47394ce48b979b7693318a2c : SUCCESS in 51m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/7c4f1a64669a4c51a4d766fab429100f : SUCCESS in 1h 01m 35s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/41febbbd1a0b445586161d4f080641a7 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4},{"id":"28817f58dade06ee2e5d086e9722bf7e190f6d4a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-19 13:44:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2c75da20d3d0810d15538e3c17d8c8ca4cec4bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 16:28:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/847f6d064925424b95241481850e3316 : SUCCESS in 1h 08m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4d16c2d57a34ab58e803dc4cc1a6f8b : FAILURE in 14m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1185a715f9643ab9eb9fb700b2da053 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9df0d9e08d43aea33ef2666dd95841 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ad967104af74754a6cf8aac6c4731d9 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee4850acc7564a95b2dc50f1c5ae71e2 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d33e3990c5430c80d58dde9e75b801 : SUCCESS in 13m 40s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/e8b4005686af4748a6357a20cf730506 : FAILURE in 32m 59s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/359e8cc425c74410ad5391401c274eee : SUCCESS in 40m 42s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eaa855a70e324675a010f003e8ef6262 : FAILURE in 33m 50s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/fb1edb269eee4cecbd4dfaf02189ef99 : SUCCESS in 1h 28m 12s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/75450d8963fa44a4993e9a4189c23827 : FAILURE in 25m 21s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c1e687e1d5d247f288de7ca8eb41cb28 : FAILURE in 32m 42s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/0cc1a2b82b00444da662d58dc1545916 : FAILURE in 34m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f7c9e3a17cb4452e8bc02e2caf84e271 : FAILURE in 43m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a1bcc885b925448f94904de46a3ee65f : SUCCESS in 40m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f77f742bd2344666802901a64c024596 : SUCCESS in 49m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/775ffc066f6b471daf8f5d3e1558aaf8 : FAILURE in 37m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a5bda22f03d842e8817c4b4c07fb5dab : SUCCESS in 54m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/95ea24aec06f4ef8bfb6b791f13a3497 : FAILURE in 31m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edffda0ce05142f0b6f76ea251b00573 : SUCCESS in 1h 12m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/a4d7a4f545664b56b226a359429b7409 : FAILURE in 29m 01s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/200d56744ac44a2fb0bea2febbae2f86 : SUCCESS in 3m 36s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8936ccf5122b490f9fb92e4c8c578f0d : FAILURE in 33m 02s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e34cf21fd34f471884e0705ec55342a1 : FAILURE in 32m 02s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c82d7ec530d7428ab9cb291e9bf78132 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"25a05b3c7ea4135e1153fbdfc26a4dcd50b2d42b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-19 16:40:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"226430f1bbf84cdfd3ba572c0079a74e0a190de1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 20:43:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2d080aa4bca94538afa9bb3919d9f674 : SUCCESS in 1h 00m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f8adc9298e74471a84ae7afdd19c223 : SUCCESS in 1h 34m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49390f3cefa343dcbf8f97354d5fd4a9 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb221a68d7004df5987b516669959661 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36690a8a70b041dca905dc0e4ad92aae : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/697c11ac176e4d54af609b4b66b74485 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfbd4d86b1541f28336a55f60b0922d : SUCCESS in 10m 10s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/899608c984904369af3d42f4c375c24d : SUCCESS in 36m 20s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/354745427b6346d29edb33cd0708ae7c : SUCCESS in 35m 04s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4ff670739cc344aeaa19ff34d9699750 : FAILURE in 1h 38m 45s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7860ceeb09b1425a8b8d13236b1664c9 : SUCCESS in 1h 10m 54s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7f9b738382b848129b3c7f79c8ba07b5 : SUCCESS in 1h 19m 20s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ec763f38d9bb4a4eae144dac43d8832a : SUCCESS in 1h 19m 22s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/491c12bd8bff4e3a918d467b44fc5a9a : SUCCESS in 1h 16m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dc7e546689174188b2970fff769c11e4 : SUCCESS in 1h 38m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/81ccd160b9a040a0aa411c8dd0b2c4d8 : SUCCESS in 42m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c719679ff4148f981028163e3134fec : SUCCESS in 58m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f05ed7bb23e44bcaace6c9de6f75ac6 : SUCCESS in 2h 05m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9d255a88ff454a50b115611f6f5521df : SUCCESS in 32m 16s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/26b9f6670414415cb9f3b21c25907673 : FAILURE in 10m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/45cc00d03c0a496bbfb0c3b080534d33 : SUCCESS in 1h 14m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e6c73814239b4905a79225b661b36528 : SUCCESS in 1h 18m 59s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/87126c30a8d34a6d8f0d67ae0b73b4a2 : SUCCESS in 3m 25s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7927a22d602e40d08d2c4464eb773f65 : SUCCESS in 1h 12m 50s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fa31179433cc4c10b52fb9b8b5003b86 : SUCCESS in 42m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b0a0b6a5b03b4af2b5c13cb6914ad8bb : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":6},{"id":"9c4c380b75f4889aaea4f808848e7c532af26fea","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-19 21:47:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dc9ca1058690a63cca02516a6c79c47a3d244710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 00:13:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fddb4929c1d84c698b90147ae2fe466e : SUCCESS in 57m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3c81099776046b38613a6fc13895ea8 : SUCCESS in 1h 18m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89d7dc63329e4abeae3a38e54e8ce539 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a822edb7dc8a42c5ac564daab613a38b : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5e71c03066e44739e9d4ff67ce0fc2b : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b594ff4fa144cd2a11dc5036dd25b18 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a415ac994b544009b413d249940f8d89 : SUCCESS in 11m 12s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/c4bf6f5c3cf2440f91688adbb6c2fbc1 : SUCCESS in 37m 23s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4d2c441d13824c229ed3d53a890b9712 : SUCCESS in 36m 10s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a7c9db36e9384be981785536d6fea2fb : SUCCESS in 1h 42m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/765ac0eac81c493d8384e2802c23076c : SUCCESS in 59m 11s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/b63b15cad6634e25b49c601bb7e0bb27 : SUCCESS in 55m 31s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f1f6c1633348448fad0ce75dac8047b4 : SUCCESS in 55m 58s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/60ea2e1353b148648445f30845c7fa2e : SUCCESS in 1h 20m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f5438d334c6b4a3396cd9b25863b0c85 : SUCCESS in 1h 34m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5d925e2705fb4a0c94b8ead6546421c7 : SUCCESS in 40m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca13273ce0774c6791662b2c2c3fb000 : SUCCESS in 55m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e406d25805a467a9b5fbad879e4317b : SUCCESS in 2h 05m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0976d82f41344e1fb3e2794afd2c5d23 : SUCCESS in 44m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6aa82843fc48483a9997648f2cf747eb : SUCCESS in 1h 26m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff0b875665a04e60819925da61e60a52 : SUCCESS in 1h 28m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d24631a1684f49f2937abdd2d80642ea : SUCCESS in 1h 02m 55s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5f724dd928a8496491c39cd590e5db15 : SUCCESS in 5m 36s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3aeea997dd3044a4b9f03747d25e200d : SUCCESS in 1h 11m 22s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1221185916364917b9c9e7191a3d4088 : SUCCESS in 58m 46s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/12a008d1d3584ae7a8a728ba266f4725 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":7},{"id":"b68cd4f43a6c9e22d532a00a4f4d6e6d45c08378","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-24 08:24:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d0b301cddcb0df851720e603f5b6cdd4415071a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 12:24:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/350b8023e4354e1386e9f48e8eafb970 : SUCCESS in 1h 07m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed5bc727f5d04a92aeb0509f89273699 : SUCCESS in 1h 20m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fe3ec657731439aa66e76a08a8de2a9 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e20d9023814321ba64e8b3dc4aa15c : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f225be975404f5e88c59881499c4460 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed171a441a2e4b2e916d00540581fe2b : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d095f12f9c4c4dc48a1c7dd497fb7cbd : SUCCESS in 16m 35s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/e480d2019d2c4c19a5501c24f54a4734 : SUCCESS in 37m 01s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/35b9bdd6abaf46d8b0a4bbfa6530b475 : SUCCESS in 37m 49s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/82a9cc472b73462aa7b0a97f1704b00d : SUCCESS in 1h 39m 23s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d4a9d2889da2489584081e157e45ea93 : SUCCESS in 1h 33m 09s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/530b19c4dbea4c9c84bf457cf4def341 : SUCCESS in 1h 21m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ca3577ebe46a41f791c1efb35c8e9224 : SUCCESS in 1h 17m 34s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9dbc26dba74249bcb4e0ac9df775c987 : SUCCESS in 1h 19m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4a881d30131941e2bef166f31f0b145c : SUCCESS in 1h 32m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6bba37ca082943388c62d3ee3e6d904c : SUCCESS in 38m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0df0caaed1074c2aa1c8205a15620607 : SUCCESS in 56m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4fbc91f114ea4387814a35bc2746570d : SUCCESS in 2h 01m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e1facc91d15b4df5b7dc626350310d7a : SUCCESS in 53m 37s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ae836a0adbad4f82b99fffa4e166bea3 : SUCCESS in 1h 12m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ea6348fecb0448bb5538a7c8b7ca89e : SUCCESS in 1h 01m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1b863aa8cc3c4518a91a7f7cc08df7e5 : SUCCESS in 1h 43m 52s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6b81eba161fe4ef98409218aafa67e38 : SUCCESS in 3m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2e6e2e7054cd463aa220fddbe79730cd : SUCCESS in 51m 01s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/cf9bf17154ce4fdfa38b52fb1a3dbf5e : SUCCESS in 58m 26s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8ecd5e239e1c44639b2f5d9c9686ca16 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":8},{"id":"64443f6589bc065081a1d662d401ef86c6f2c2fe","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-09 03:44:06.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"66d205af6a9ce3dbc83d44853267709af400ab14","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-09 13:31:04.000000000","message":"Patch Set 8:\n\nFYI, We have a similar resource leak issue but we need to do it while tenant resources are created.\nSo we do the following\n1) sed the prefix default value in the data_util.rand_name number to a short unique id\n2) after the test suite executes we run the --save-init-state to collect all of the resources in the environment.\n3) We wrote a custom json checker to find resources with the standard prefix.\n\nThis does not find all but if we find test cases that do not use rand_name we will change the tempest code.\n\nLet me know if we want to talk more about this direction.","accounts_in_message":[],"_revision_number":8},{"id":"0d463ff2172608b4b9b2d786a2413ff267daaeb4","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-16 18:23:26.000000000","message":"Patch Set 8:\n\n\u003e FYI, We have a similar resource leak issue but we need to do it\n \u003e while tenant resources are created.\n \u003e So we do the following\n \u003e 1) sed the prefix default value in the data_util.rand_name number\n \u003e to a short unique id\n \u003e 2) after the test suite executes we run the --save-init-state to\n \u003e collect all of the resources in the environment.\n \u003e 3) We wrote a custom json checker to find resources with the\n \u003e standard prefix.\nNot sure if I understand why you sed prefixes. Basically you do the same, but you run --save-init-state instead of --dry-run. \n\n--save-init-state and --dry-run works almost the same with only one difference, the output of the --dry-run is filtered so that the resources from saved_state.json are omitted.","accounts_in_message":[],"_revision_number":8},{"id":"37a9d5eb2b09908eacfb7f75f3ba2c940c408bba","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-13 16:06:40.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)\n\nThanks for adding this, comment inline","accounts_in_message":[],"_revision_number":8},{"id":"b33fe4b83dabe8ae7776c43c4d85f6ebc17cec75","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-23 09:23:00.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nI\u0027ll try to simplify the logic in the next patch set","accounts_in_message":[],"_revision_number":8},{"id":"d8302d6d811bf24705c02cd028bfe13def785b05","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-23 09:23:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a355c46b4603b0af54b1b829470eaaf7a8d50545","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-23 09:27:12.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"ffd544642022f707f7817b47f9c67c9995e5ab6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 12:39:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/14ef478c58944f809c04aeac74ae977e : FAILURE in 50m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce2e846289b14369a5b311859754716c : SUCCESS in 1h 17m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0cb12bb20ff4e6da7e9b0a1dabcd5f3 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d207a47712904002aa3ed57e846944cb : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d20c3a3765ad499dba1b402286c3ea15 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/331a9b946c3f4f85b20d35047ccfc276 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d996bd0ac9149e7a580a0723439ea8e : SUCCESS in 13m 31s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/de170a8fadc945cf98d8c00ce61ecb63 : SUCCESS in 48m 00s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/42fe48e26b6046da86eaa118019aedaa : SUCCESS in 47m 56s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3a37214d796449838a617db28010074b : FAILURE in 30m 11s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/22759af0d85149e38944fb2301cd4e6e : SUCCESS in 1h 25m 34s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/da9f3507f3064f2190ce97fb2ad7f7a8 : FAILURE in 1h 20m 57s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c91b309a0f1248d4a616b9775d9d66fc : SUCCESS in 1h 32m 09s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f484a23c185341589ae9de697833822b : SUCCESS in 1h 40m 34s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e8ca382a82634766b627cee2499e0224 : SUCCESS in 1h 18m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5cb0f97c9f784a96b7754f194d840771 : SUCCESS in 1h 52m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a6afa543c3784416886b1b650063fa9f : SUCCESS in 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11f20b02d5dc43ebb9129f208e6b00d4 : FAILURE in 5m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6dca2d4f43b4283bf567e0147a6fa49 : SUCCESS in 2h 15m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/daaaa8a242134574bd783c38f6bcaf76 : FAILURE in 40m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/545d29681cd74e549b0f3988e95933bc : SUCCESS in 1h 06m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1901cf225e954497b9e705cf20edc429 : FAILURE in 1h 04m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/92ab2c599ba0450cb2cb5cf8488001d5 : FAILURE in 1h 04m 48s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/fca8feff11b641648932c53cb3d1e743 : SUCCESS in 4m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a52eab7aac4d4b488dd974bb361e240f : FAILURE in 5m 30s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9c16c553e18c4370840e98b95f939b09 : FAILURE in 1h 02m 06s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f72176e277e0444cabee477e261ec57f : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":10},{"id":"11f1b1041f9ed6702605200eb18b5d7ae6b40e9e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-23 12:50:22.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"faa84b2e902333d64b5ce3d4903810902b47d560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 18:41:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/82dbad4651fb46d69d3f56a09f25dd25 : FAILURE in 54m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/632902c0f75944a7b5f58e36a0de17eb : SUCCESS in 1h 21m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0e9999a0b16495ca74a051c95b05d5e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e35c4521c94003ad19af624dc72be7 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85cadef93fad4f85955f28cab72d3068 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4533914177ed4cfcbf4935a82ae977c8 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558aa958279e4de48cdb62f6d1ae0c06 : SUCCESS in 22m 46s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/186789635ca54112ab1178895aec65e4 : SUCCESS in 1h 01m 47s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e8dd79fcb0b5427ab55f3297f937c6c2 : SUCCESS in 42m 35s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f2aa892ee00b4356b61617c696126737 : SUCCESS in 1h 35m 13s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3beb644eb90048a99c0004435a7d4b90 : SUCCESS in 1h 15m 46s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/50fb6fc9437e498a8f1d642a41778057 : SUCCESS in 1h 18m 21s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/3e3d7c0b1657407c83f3a1e678e2c683 : SUCCESS in 1h 16m 04s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/5ff7e00633c84b0cb3622e55db4baae6 : SUCCESS in 1h 25m 29s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/f362d0eb526f48c68663d1024a65d5e4 : SUCCESS in 1h 25m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0c68c5a8a0e14bcb9348eee936585814 : SUCCESS in 1h 35m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1600ea3b44244662981d7a27f436c103 : SUCCESS in 52m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09e6358a961f480bb5eb74087596a77f : SUCCESS in 1h 07m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a1660efe47547429809a9708776c0d8 : SUCCESS in 2h 13m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/373138fe2b5f45439c85b92846851a4e : FAILURE in 16m 13s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5e926045002848b59f22b4cc828203bc : SUCCESS in 1h 08m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79e520c2936149788cc94873fcd0ecc3 : FAILURE in 56m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b68cdaa1c50149af90613e97e9e6dd8d : SUCCESS in 1h 16m 02s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/cd3486dd362e414999769d162566d520 : SUCCESS in 4m 49s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8ca5aeeb8c42489eada82fbad2803266 : FAILURE in 1h 07m 13s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1fd59b78d8964c2fb3a3ccce536af8ed : FAILURE in 1h 01m 58s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c52c6354dc3442faa6bc1bd13a2afc2d : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":11},{"id":"d36dd1c3c75c1f5446b2fde76a8f31c4b35bf59b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-24 09:35:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4752acddfad1c21a45e69da62ac78728b2c0a51f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 11:39:52.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b0fa92fd2d7249a2ad972aa1d8d7e73a : FAILURE in 44m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2aa9b67859854a78868e9779bc6967a5 : SUCCESS in 1h 18m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80e43a0812ec45eebcf955b7a0178919 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012164d3d78f462989cf836365675818 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce5ad33b529b4df6b8aea08071720eb1 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3193a2c7750c469ea32d35439d449e03 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e0e8c19ad242519bf92ce7c02fa54e : SUCCESS in 12m 13s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/60da473e0a374b77901ecd7150ef615b : SUCCESS in 38m 27s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c6e3c4f113f34b85ac5566c72ff0029a : SUCCESS in 37m 28s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a97b5a55c5c2433e8d8e8c8ca81e181d : SUCCESS in 1h 42m 45s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/dc4bf98364cd449eba54e0a4a7dd1050 : SUCCESS in 1h 22m 11s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/f2bb190143214b6b91c8095b0c1fbc6c : SUCCESS in 1h 25m 28s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c5a24e1385664717bec383bd7058e834 : SUCCESS in 1h 24m 43s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7d8cd1f65d2b445989cadcc96574d261 : SUCCESS in 1h 16m 22s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9a0007291ebe454087ae445ee656e06e : SUCCESS in 1h 18m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8731f424e26340d6966a2fb1642b76db : SUCCESS in 1h 37m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5db34b24b5ff4554919c829063c79435 : SUCCESS in 44m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/409c8fd6dbe74d30a271fdf9ae0f0a8e : SUCCESS in 52m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0e86113f4234618a5b68682f4a2960c : SUCCESS in 2h 02m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/093a190ddb934199ab7d1256c4c5f511 : FAILURE in 30m 19s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a451661fefef4e33a56230ce186fad21 : SUCCESS in 56m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc97e2a07f74422c9272da76e2fdbb11 : FAILURE in 56m 41s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7be38b390ce6482aba18b4076e6ee113 : FAILURE in 1h 02m 46s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f222b814e6834373a53e349d46361b4e : SUCCESS in 2m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4a85892482e341c5b607143bd6bea343 : SUCCESS in 1h 13m 02s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fe22813d61004cc391b6bfa280dd0a5a : SUCCESS in 57m 52s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d6da679cd34e48cea13365b7ace68e96 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":12},{"id":"bf28144849bf68aa2a09cd5575bfaccb573fa556","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-25 21:41:33.000000000","message":"Patch Set 12:\n\nScenario 1:\nrun_tempest_cleanup: true\nrun_tempest_fail_if_leaked_resources: true\n-\u003e https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_109/746690/11/check/tempest-full-parallel/1096d68/job-output.txt\n\n^^ that practically tests if tempest-cleanup deleted all resources it\u0027s supposed to meaning, run_tempest_fail_if_leaked_resources here causes the job fail if tempest-cleanup fails to delete everything in its powers\n\nscenario 2:\nrun_tempest_cleanup: false\nrun_tempest_dry_cleanup: true\nrun_tempest_fail_if_leaked_resources: true\n-\u003e https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_122/746690/12/check/tempest-full-parallel/122f39d/job-output.txt\n\n^^ that practically tests if tempest tests left any resources behind meaning, run_tempest_fail_if_leaked_resources here causes the job fail if tempest tests not cleaned all resource they created","accounts_in_message":[],"_revision_number":12},{"id":"17aec0f08919a590a1064f3f70a96585502b6756","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-25 21:41:51.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a45035652f0670bf1483535e5cb27809ad0ffded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-25 23:34:35.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1fbdedf68b24660acd6785dad9aa0da : FAILURE in 39m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2127b2f378c4762bfd229f53a348640 : SUCCESS in 59m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b4f2db68c6a4058a0bbebe7ebc73c81 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9cd983756a4a19a59b7492d5336314 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d284516e96364e3a8a0330a77e711002 : FAILURE in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1358184032b14ed5a974f0ab1ce6e631 : FAILURE in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c569a0a985f14226925b0ac87891e180 : SUCCESS in 11m 08s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/00c3ae4fa5eb4a5593e4f94df76b3374 : SUCCESS in 30m 12s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a4e45828572449309de41b02ebfbad69 : SUCCESS in 51m 47s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/ba855d34272444408bc84affad972160 : SUCCESS in 1h 43m 40s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/4dcba181c0044cada455d32f203db7b9 : SUCCESS in 55m 28s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/e2780d421eea458f91d2771cc1d6b835 : SUCCESS in 1h 33m 00s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/8a876cbf20dd4623a5b1790b4e3b8b76 : SUCCESS in 1h 00m 03s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a18f5dac366b428aa7ee52856710c931 : SUCCESS in 57m 32s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/818b18ecaaad42b683ab704f5bd1546c : SUCCESS in 51m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7d9b25e1cbd6405d94cb47087f7a4b7b : SUCCESS in 1h 12m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/33080c2f6a494e078e9858963612cf8c : SUCCESS in 27m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bedeb896aa3247b08bae52cb6f4cc522 : SUCCESS in 45m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/92a283b1d7374695879417d363d51205 : FAILURE in 1h 48m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/652534b01b814a578c41548b301c913f : FAILURE in 26m 27s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0bc0e86136624e85a06a1302b299fada : SUCCESS in 45m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a91d0fa66ede4369a738594c27946f75 : FAILURE in 57m 11s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6932f82640684df596e05d309291db72 : SUCCESS in 49m 08s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7d76b6f567204b0b8d5cc3829fdaa982 : SUCCESS in 2m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d7fd42e5cc6c4f4899144a9af0de8530 : SUCCESS in 55m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/05e9f9f7f1364559912f4cd33516f8a7 : SUCCESS in 1h 19m 49s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/55f6a24a430940efb20b606abcfa0dca : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":12},{"id":"a587164af85c99e10e65b65d18639448c9376a69","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-29 08:48:00.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"d0631061bccd83d117961f8a92e427bdafaaff3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 19:22:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7627df3f0eb64cf6948e079bb77ec517 : SUCCESS in 1h 10m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e72c7f4710b147b1ba136e13f2acccad : SUCCESS in 1h 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bb3ba65caae48889a45ef738b9cb38b : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56b48e96f094c79903c001e54453bf9 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c1e993299242eab9c61dc60bbb2eac : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06af6df097e346179473c24267acb5d4 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25f219194fd4988b6324c0ddbe63b8b : SUCCESS in 11m 28s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/6dcdb739357b4a3c912445d229431e9f : SUCCESS in 47m 03s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/fee5f19bc4804725818ac83734bcd4d3 : TIMED_OUT in 2h 03m 17s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4e61f060fa10455e896a13ce75ad745e : SUCCESS in 2h 28m 07s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/18f61cb04302431fad6d3fcc3d2d5f70 : SUCCESS in 1h 19m 19s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/524f7d6e305e47858ffe8a239787f4d9 : SUCCESS in 1h 23m 40s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/9fdccf8362ec4b3e905ef11aec3a3e88 : SUCCESS in 1h 17m 08s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/0a8cfa1389be4b848d29df993bce2602 : SUCCESS in 1h 40m 41s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e894ee50ffc14e97a7e9464bf29e3b7b : SUCCESS in 1h 21m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5c2de61cc12b42d2883603975b523393 : SUCCESS in 1h 51m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/52a1fce5c66e44028be5103a04510178 : SUCCESS in 30m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1df568ab15a42fb84122c34fee8dc71 : SUCCESS in 1h 05m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76edf878d03a463395536187293222bd : SUCCESS in 2h 04m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/78759f87e475443eb7f719b5b4468086 : SUCCESS in 45m 06s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ceaf77cc9be14091a41e1b612da7ea96 : SUCCESS in 1h 18m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d95ba9ff23d40f99de7ef7c3555aad3 : SUCCESS in 1h 28m 20s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1d49becdb9994149975849884a4d50f6 : FAILURE in 1h 14m 35s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9d574a263adb40daaab16a4cb5b639d7 : SUCCESS in 3m 08s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d1297676664c4ac0ac922b964c6bc35f : SUCCESS in 1h 23m 36s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e7bf5a4951394ea19bff91e0d279b7c1 : SUCCESS in 1h 02m 56s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/19243d4c8ce74d238f3c6dc0358d5632 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":12},{"id":"b03e6978dc296e1a50e9faac46abe1a711f3b822","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-03 18:35:44.000000000","message":"Patch Set 12: Code-Review+1\n\n(3 comments)\n\nnice work, testing in https://review.opendev.org/#/c/746690 is really nice.\nlast comment about README","accounts_in_message":[],"_revision_number":12},{"id":"1075f07ccf8021bbdaa14161143cd8afbf904eae","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-03 21:47:22.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"189c1ba6b595d0f06d548ab80f6ae34d666640a7","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-03 21:47:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e60c1017ab9d570493c1ae866d75666cfcba7fbc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-04 00:50:48.000000000","message":"Patch Set 13: Code-Review+2\n\nthanks. lgtm","accounts_in_message":[],"_revision_number":13},{"id":"ff3f81d9beed60c76b5c3f8e70b50ade52e05157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 06:59:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4889e1adeefd4044ab669d97367d279e : SUCCESS in 1h 01m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f720af1a8f046d5b3d433655cc840c9 : SUCCESS in 1h 15m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e4d2797aafc44679547994f26a11a02 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe0f18a4ee3f436cb84a13e3c26e96fc : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9930027ecb8940aabeb9eb49dc09fb0b : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55be37f73e554207b0ac71d98eaa25e2 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc9d61f0cd34580ae0155003f2cb702 : SUCCESS in 20m 44s\n- devstack-tempest https://zuul.opendev.org/t/openstack/build/bf8810b7b98b4c30b4266396309ee8c1 : SUCCESS in 40m 27s\n- devstack-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4e9f51ce070d4f548853bfbd476a5548 : SUCCESS in 55m 37s (non-voting)\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/36d5d3f3e55d491fb096b95ca1647980 : SUCCESS in 1h 30m 35s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/1c5c4d332d21441c8cc399752afd9e98 : SUCCESS in 1h 19m 49s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/bacc6bc9f3f548c295cd0353dc33e688 : SUCCESS in 1h 30m 30s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/39ee5bf38c804047b386e02f0cdfeb6c : SUCCESS in 1h 15m 06s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/79c6f05fb3c84319842512cac9d628e6 : SUCCESS in 1h 13m 53s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/91930aaa2a6146e7ad85bac411f7595d : SUCCESS in 1h 12m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8b62e03143ab4d8abb7233c446b4fa58 : SUCCESS in 1h 35m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a615c3ae85484c309b3b4a9328bf568d : SUCCESS in 49m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69544db6f26245e9b492004862787b2c : SUCCESS in 1h 06m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4543fc67680245a3bb001d54b195ed52 : SUCCESS in 2h 07m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/95da54f18784435ca6f7f677b4bd56ce : SUCCESS in 48m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2f15a97c5ded4b4586244b312d8606cd : SUCCESS in 1h 00m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/781e2399b1ea48eb916c07baaada8970 : SUCCESS in 1h 13m 03s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/55262c4ca2d14985a8ef3bd8f8bfc3f7 : FAILURE in 1h 11m 22s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/215db56810db470396440cbd24892e1b : SUCCESS in 2m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2ab62706ace941f39922f365ee2e9317 : SUCCESS in 1h 16m 57s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0c2fc9fe23104376b5b28e7709f6720a : FAILURE in 49m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8415818c97b84913afdc41ae4dc1c1f6 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":13},{"id":"099350b9f0874b58679f162c91d638463dfc5de4","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2020-11-05 07:48:41.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"95f73fc623d4997a52f37af6d1f4171d4044493c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 07:48:59.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"db62025609fd622058d63b54e7ffd57740998dad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 10:18:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"327ae958d55f155e4dcb3a16953a2591007cacc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 10:18:23.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bcf66a56a92146098bf01ae19d6a3278 : SUCCESS in 1h 01m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bee0a25241049168438f3764e475da5 : SUCCESS in 1h 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ada9546f484a1e9ee5f29aaa88fef5 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9341e2ca91d144648ec19a14f137d7f4 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34a5871febca4d0a8ae5574a26a0b7bd : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfce23dbeec24bdaa7ef26169bbb0cb2 : SUCCESS in 20m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e41ea1e866841bd9f9a18d5492f551b : SUCCESS in 2h 20m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70d7130e1a8c4eadba4ba58c937da11b : SUCCESS in 1h 03m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8d3b4e6b1be4151a5a6eeeb3a1506cf : SUCCESS in 1h 06m 52s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/aa39765627e44552a6519a391098ac20 : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":13},{"id":"b9bcd42d0e0a9356af7bc6e7046513eaa02b4fa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 10:20:41.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f7ef918c844af092cdf07f81762196 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"05c35eb7932171e2f86a47fa72cad15c4d3dfebf","revisions":{"7e0fd104235ee76b2690c79a17cdaf8c8f7569da":{"kind":"REWORK","_number":1,"created":"2020-08-12 09:20:13.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/1"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after temepst cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7e0fd104235ee76b2690c79a17cdaf8c8f7569da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7e0fd104235ee76b2690c79a17cdaf8c8f7569da"}]},"branch":"refs/heads/master"},"804c2e324732de9f5e25c3774f723387d51fcb22":{"kind":"REWORK","_number":2,"created":"2020-08-12 09:23:10.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/2"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:22:58.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after temepst cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/804c2e324732de9f5e25c3774f723387d51fcb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/804c2e324732de9f5e25c3774f723387d51fcb22"}]},"branch":"refs/heads/master"},"f2c48c8b1691b095203b1636920d58ae77fa87ff":{"kind":"REWORK","_number":3,"created":"2020-08-18 10:17:33.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/3"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-18 10:17:22.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after temepst cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2c48c8b1691b095203b1636920d58ae77fa87ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2c48c8b1691b095203b1636920d58ae77fa87ff"}]},"branch":"refs/heads/master"},"f45cc093c2ee78308a48084c7919fcde2dcddd5d":{"kind":"REWORK","_number":4,"created":"2020-08-18 19:17:57.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/4"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-18 19:17:43.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after temepst cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f45cc093c2ee78308a48084c7919fcde2dcddd5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f45cc093c2ee78308a48084c7919fcde2dcddd5d"}]},"branch":"refs/heads/master"},"80c6655a61873cb3822d28e5af37aa1ab5970cb6":{"kind":"REWORK","_number":5,"created":"2020-08-19 13:44:22.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/5"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-19 13:44:03.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after tempest cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/80c6655a61873cb3822d28e5af37aa1ab5970cb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/80c6655a61873cb3822d28e5af37aa1ab5970cb6"}]},"branch":"refs/heads/master"},"a7dd398f8b23ddcc7fc489976474e9d2fa781099":{"kind":"REWORK","_number":6,"created":"2020-08-19 16:40:33.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/6"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-19 16:40:26.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after tempest cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a7dd398f8b23ddcc7fc489976474e9d2fa781099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a7dd398f8b23ddcc7fc489976474e9d2fa781099"}]},"branch":"refs/heads/master"},"cd5227f6e006c6be79af9a53b8949768abd063b4":{"kind":"REWORK","_number":7,"created":"2020-08-19 21:47:59.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/7"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-19 21:47:31.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after tempest cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cd5227f6e006c6be79af9a53b8949768abd063b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cd5227f6e006c6be79af9a53b8949768abd063b4"}]},"branch":"refs/heads/master"},"58c13cc3f690c8505b77c4e8b412c4efc7c1f0f2":{"kind":"REWORK","_number":8,"created":"2020-08-24 08:24:30.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/8"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-24 08:24:18.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nBesides that a new task is added after tempest cleanup one which\nverifies that \u0027tempest cleanup\u0027 command removed all (if any) leaked\nresources.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/58c13cc3f690c8505b77c4e8b412c4efc7c1f0f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/58c13cc3f690c8505b77c4e8b412c4efc7c1f0f2"}]},"branch":"refs/heads/master"},"655a99dedf12c0502fb5ad9af3fbe05e53160193":{"kind":"REWORK","_number":9,"created":"2020-10-23 09:23:06.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/9"}}},"commit":{"parents":[{"commit":"24961f6244fd3798b26c0d516d18ab03cd764df8","subject":"Merge \"Adding description for testcases - compute part3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/24961f6244fd3798b26c0d516d18ab03cd764df8"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-10-23 09:19:43.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/655a99dedf12c0502fb5ad9af3fbe05e53160193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/655a99dedf12c0502fb5ad9af3fbe05e53160193"}]},"branch":"refs/heads/master"},"359b73260864f5e4efa54039cac064f963a26803":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-10-23 09:27:12.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/10"}}},"commit":{"parents":[{"commit":"b439c974faab1d593ce14ecb3cbfd5467b0005cf","subject":"Merge \"Add Tempest gate job for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b439c974faab1d593ce14ecb3cbfd5467b0005cf"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-10-23 09:25:15.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/359b73260864f5e4efa54039cac064f963a26803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/359b73260864f5e4efa54039cac064f963a26803"}]},"branch":"refs/heads/master"},"9922bdc1faf8488963193195c46b159217604060":{"kind":"REWORK","_number":11,"created":"2020-10-23 12:50:22.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/11"}}},"commit":{"parents":[{"commit":"b439c974faab1d593ce14ecb3cbfd5467b0005cf","subject":"Merge \"Add Tempest gate job for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b439c974faab1d593ce14ecb3cbfd5467b0005cf"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-10-23 12:50:18.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9922bdc1faf8488963193195c46b159217604060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9922bdc1faf8488963193195c46b159217604060"}]},"branch":"refs/heads/master"},"4e1e07e240a5892bcf139edcca8d2f3c79ab83d7":{"kind":"REWORK","_number":12,"created":"2020-10-24 09:35:35.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/12"}}},"commit":{"parents":[{"commit":"b439c974faab1d593ce14ecb3cbfd5467b0005cf","subject":"Merge \"Add Tempest gate job for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b439c974faab1d593ce14ecb3cbfd5467b0005cf"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-10-24 09:35:29.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4e1e07e240a5892bcf139edcca8d2f3c79ab83d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4e1e07e240a5892bcf139edcca8d2f3c79ab83d7"}]},"branch":"refs/heads/master"},"05c35eb7932171e2f86a47fa72cad15c4d3dfebf":{"kind":"REWORK","_number":13,"created":"2020-11-03 21:47:30.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/14/745814/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/745814/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/745814/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/745814/13"}}},"commit":{"parents":[{"commit":"b439c974faab1d593ce14ecb3cbfd5467b0005cf","subject":"Merge \"Add Tempest gate job for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b439c974faab1d593ce14ecb3cbfd5467b0005cf"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-08-12 09:17:35.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-11-03 21:47:14.000000000","tz":0},"subject":"Improve tempest-cleanup ansible role","message":"Improve tempest-cleanup ansible role\n\nThe patch implements a new flag which will fail a job when any\nresources were leaked - that can be used for verification that\ntests are cleaning their resources after they are finished.\n\nChange-Id: I212cdce9713c80491153b0bbdb313e75c1f96f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/05c35eb7932171e2f86a47fa72cad15c4d3dfebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/05c35eb7932171e2f86a47fa72cad15c4d3dfebf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
