)]}'
{"id":"openstack%2Ftempest~746658","triplet_id":"openstack%2Ftempest~master~Id8f747c0eb018847bd9dc5bc3801ee973704da4b","project":"openstack/tempest","branch":"master","topic":"bp/testcase-description","hashtags":[],"change_id":"Id8f747c0eb018847bd9dc5bc3801ee973704da4b","subject":"Adding description for testcases - swift part2","status":"MERGED","created":"2020-08-18 08:27:42.000000000","updated":"2020-09-05 12:58:32.000000000","submitted":"2020-09-05 12:57:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746658-1599310637762-1189cc01","meta_rev_id":"016bfa35491d3d277119de81ef582195691516f0","_number":746658,"virtual_id_number":746658,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-05 12:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2020-08-24 08:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-04 15:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":1,"date":"2020-08-21 12:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-04 15:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 12:55:13.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2020-08-24 08:35:59.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-09-04 15:59:21.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-05 12:57:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47818e52fb545757f154efd52afbea1444282974","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-18 08:27:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb6a884229963a481e256f799509d1e3b2b395e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:34:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f1456d451c284583acca5fa5149cdc80 : SUCCESS in 1h 02m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/760362a678524605b5707452f4a0a77a : SUCCESS in 1h 13m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe9f3e84dca14e94986833dea259ecd6 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5a6725a1be416fb83e6fd8de00f03a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cde16172e7741d18d49fc0774d091fe : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/411be88d4c244a0ea09481f0225d3479 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d023a1ee5b6485a82cee56fdf935ee6 : SUCCESS in 12m 51s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b6685a83f4d2485da2499f07d4d38a4d : SUCCESS in 1h 44m 46s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d7a37611031c4a4a9d7c7c71e5a07960 : FAILURE in 16m 20s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/677cb8e7ecfd445cbce317c568399889 : SUCCESS in 1h 14m 26s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a7e66c69bc7e412997205ef67b549d60 : SUCCESS in 1h 12m 21s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9eb7f9292e28413fbbf62735ab01cc71 : SUCCESS in 1h 14m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/30b8321ca950469999db9c2a5ccf6b5a : SUCCESS in 1h 27m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/65e21dcc3c4841faacb9a4507ebeb429 : SUCCESS in 41m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdcf84a50d444d258b61ea5ab166efa2 : SUCCESS in 53m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/481565d3a6ed49b0852ebc8f6b9f4311 : SUCCESS in 2h 01m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/73f6b32d6d514b8490e69f72da91d1d2 : SUCCESS in 47m 05s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9ef0064fdba84c2881194836fd04c2ac : FAILURE in 1h 03m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/314a04403907442787941cce55e710f3 : SUCCESS in 1h 05m 23s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/91622bb3dc534a65b66f087be3c3453f : FAILURE in 58m 55s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/fb1d4647a5564da38b24b1d74e48c81a : SUCCESS in 3m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/875ef8c92b374c9ba234d8a92a5cb866 : FAILURE in 1h 01m 03s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6081dfef2b5a43db90074a005eb65cad : FAILURE in 57m 45s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0a62e343609648e38a1985af7ab35bd5 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"111b03597ae8e65dc719ee451249dfb0660fd9c7","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-19 01:14:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b7f07f947fac1a7228c6281fe889f98c90bb518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 03:33:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d16deeba75034771954cf9180044e2c6 : SUCCESS in 1h 19m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cc94c97ada44c61a7df6a23048fd6fc : SUCCESS in 1h 19m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ada42238a214afba0cc0904cfa5ffb7 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a645095c03a54c19a47cf9ee805d3a49 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c8ff4e17ef64b10925df3cb6cef7460 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3149864dbd2c4dae939e824e11804beb : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78b494dca2f497a9237c9084b423527 : SUCCESS in 8m 24s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/73de0af4021340e9a475231f51793759 : SUCCESS in 1h 44m 00s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6b798c6326b946c49eda95b07e8d7c80 : SUCCESS in 1h 16m 14s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/95941f00b9e94590a086712bb7927a8b : SUCCESS in 1h 29m 14s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6fda28a9d0434812b6b92f9c86656f1e : SUCCESS in 1h 27m 28s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/d5e26640d4394708801bec512106cf05 : SUCCESS in 1h 11m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/70112af08e004c2e879ef9e931ee241c : SUCCESS in 1h 32m 23s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/34d63e3fd3ce412aa019f62c5ef83556 : SUCCESS in 54m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a21fd7bda4a4f85b11f2c2531adb481 : SUCCESS in 1h 12m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/edf3c1e58e3b4db7a76c8981a13549d6 : SUCCESS in 2h 02m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a68a476e3e904283b728777feba6db80 : SUCCESS in 58m 48s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/148d52aaf759471494b6224b523c9bfd : SUCCESS in 1h 08m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2ce4fe5f0a740028121a9f841e1b369 : SUCCESS in 1h 15m 19s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/67de508c08c04a08ba5e9abb26c756b1 : SUCCESS in 1h 40m 37s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8df191e8882047bdaadd84f12820d688 : SUCCESS in 2m 32s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6aa6fe60be704a478419b08d8424285d : SUCCESS in 1h 20m 40s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ba40809d871448e6be73018c4d3a13f3 : TIMED_OUT in 2h 06m 25s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/cacaae92f5b34344adf35ed91e7c53b2 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3d1bd36d40f141e598081b7bde38d26dcee02871","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-20 02:23:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"440c1a36bb8fe26467da317c42f3bcc76e013b07","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-20 05:44:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e146d737175c34e2dabf1e257c969b31d908417e","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-20 05:45:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcb492d66f28893e032e16a1045cb25d9a49aa79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 07:49:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5cd196fc4e1e4570a5a144a921aaa078 : SUCCESS in 1h 02m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6cda2c3e54d848fda14c5ac070cbcd16 : SUCCESS in 1h 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57444fae2db8412bbfe1d466e1cbb9d6 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9f30a3616c429098d4af770c96dc71 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f376d26c1a0b4b43a79ef1885bab13f7 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80224e662a0a4d0999e102d4329e0a92 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1097529091cc44af9b16e7cb96a65b37 : SUCCESS in 11m 10s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/217b27ff34234ca588db0f99f149bd3b : SUCCESS in 1h 44m 06s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/29388da3c7404dbeaa38c65f4377d8f2 : SUCCESS in 1h 18m 25s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6eb98305591444f19bc61f7b6e932bd2 : SUCCESS in 1h 17m 06s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e20f9fe6b7944371a3cf0dc3f86bd4d4 : SUCCESS in 1h 13m 18s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/87a8a80b5bd94fea82ba6b3553da2133 : SUCCESS in 1h 12m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9562556260bc4b1d9692edf980de741b : SUCCESS in 1h 40m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f4c365a4255448e88d8f0bc6ded8734d : SUCCESS in 41m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b337a848151c45788a508162c8297c94 : SUCCESS in 1h 02m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8542e480afda43178b50ce9278e2c568 : SUCCESS in 1h 59m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b10a4b504f6d4deebe6a7745f5ecafff : SUCCESS in 55m 07s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1823b13c439f4861bbec913514175a0a : SUCCESS in 1h 09m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0469328e7a444a0f83e570213204e0b8 : SUCCESS in 1h 11m 16s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bfaff948ac1b4ca08542819029887f3b : SUCCESS in 1h 13m 09s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a12b062d5b9a421eb7fe33620e4ac12c : SUCCESS in 2m 46s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5841b20c77c242c1bf65c84ef32806dd : SUCCESS in 1h 12m 41s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e70b97687c48421995f1c0a4f529deaf : SUCCESS in 1h 04m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/fb5cca5030fa4be7acb70b8c5cbce134 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"47ce14db1bb00f5cfa60d60e699cd1ee45ed158d","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-08-21 12:55:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"60c5096394572c87127182398f2329d23ba06259","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-08-24 08:35:59.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"82bf9b29323ef0523158ac8d6c96a525edf10a11","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-04 15:59:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81aff0fcbeb176326588fbd66b5058f982fbc263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 15:59:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9eee553ca13498ea43150d050fffaeb12c4d145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:12:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/15b9ba42305d4418b9cde543c1293e92 : SUCCESS in 52m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/655b0fc509c84d0b8505b3449594bc6c : SUCCESS in 1h 06m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c90f2b5d6304a8eac9f9e498f9c1a6d : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/697cd8c8766b4784ab35cf7edb3057b5 : FAILURE in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ecfb634ca7a495eb40d2b5aa615ae79 : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c704a28ef0ab4b55a06a6f46fc7d4b23 : SUCCESS in 13m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f63f8b641e964e0a98e76f65994da29f : SUCCESS in 1h 48m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04fe0c7aab04473f960a515f96250ecc : SUCCESS in 1h 08m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/950c2ccaed4443e0a599ac567e5aed1e : SUCCESS in 42m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c56bc66622be46578c15743e8784b7f0 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":2},{"id":"fa5bb23caccb04a442a9a5c154d3ab19aac347de","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-04 21:34:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f98029e25fed21602f74d5a0b975fc9e2fb62610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:46:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/62398c83aa1642a6b0c9e1fda9f030b6 : SUCCESS in 1h 13m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6e1bb735f744d43b8497902b9ffe020 : SUCCESS in 1h 35m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50c4e828df754108a6e0a490d352dbe1 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b889ff0b08044fd3af49c2521c0a252f : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ca47da0bb5646e3a26aea11914cb38a : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7729bce6dd0b49daa5c03f007dbdbd38 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41afd18298ad4201958c4f294d654855 : SUCCESS in 14m 20s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a4d11d03273c437f8981f06f0c7d21aa : SUCCESS in 1h 39m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/fc60d133956a4f95bb35392bfcdd6647 : FAILURE in 1h 05m 50s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/a20391f80a0840f48ffc9821fa43f1a7 : SUCCESS in 1h 29m 35s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ea29884dcc9846a5a75156e37482e21d : SUCCESS in 1h 23m 08s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/32f3941d084947669ae1fbc249383712 : SUCCESS in 52m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bdce76318e5d49bebdfd2787418fb205 : SUCCESS in 1h 30m 13s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/60236a3a3fae4d8e9891bf212df80f43 : SUCCESS in 48m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46c22689b4874b09898b37882e46cbb6 : SUCCESS in 1h 06m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e87b4403f92a4f6cbdfbc357537573dd : SUCCESS in 2h 05m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/faf9a75efcb84542a8470a9f4c8072b2 : SUCCESS in 51m 14s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9c99e23083df43abbf1a638a54daed2b : SUCCESS in 1h 01m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a52de70841244dbe8712649ab8069531 : SUCCESS in 1h 16m 51s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/16afd89271df4f20b87487bcec4791b1 : FAILURE in 56m 47s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/34951dedf2de4ea28db39f5a38f510e9 : SUCCESS in 2m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0ac736ed9b6f4fabb87ee3472e5ebd33 : SUCCESS in 1h 05m 10s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0858f65708ab40b38705b35dc11490fc : SUCCESS in 58m 48s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f5b7ec3b94ff4bf6ab5ad31987d765b3 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"154b626c6188eefe6e0876f28594235b8b8c283b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-05 07:35:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ee6512b8b092262a67842d9efd58d38a19732303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 09:57:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/93776b10e3ef423cb32432e26197c5ee : SUCCESS in 1h 23m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0746de03debd45bea5414d27ad11d4cf : SUCCESS in 1h 52m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65dcc97f80c1493eb2e3a1ac6ff45a46 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88802c63e22e46b9975159ed909da253 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/370cccf769dc4b81aff8e4c6e2601a64 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76b3618f260d4a9ebbdd51b9d0e24e51 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609e6ed3813a4525b15e9f4b7b11e8ff : SUCCESS in 14m 06s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3c96103bb0264ded99f594f2fd874669 : SUCCESS in 2h 14m 26s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c59eb741431a4c84b7fd19a0cffecaa5 : SUCCESS in 1h 53m 01s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/87b60734f305409d943ccb493640b5de : SUCCESS in 1h 17m 29s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/06c2ae585e3644ce9d595dd548c818a4 : SUCCESS in 1h 12m 42s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/757ce88905fb4788a9bbd8bd42c99d10 : SUCCESS in 1h 17m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9c72efac5a354fa89e86705d30545ef9 : SUCCESS in 2h 00m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3a0475240e134275b082375aaa8a7a59 : SUCCESS in 51m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37896e8eb97b43529151fdbff2368b43 : SUCCESS in 1h 16m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfb437a0258844dda7744c48bffe7955 : SUCCESS in 2h 19m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e225966dd953407faf4b8ab287ec8a8b : SUCCESS in 1h 03m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0ce8e97be24d47888288f4de8251e0cf : SUCCESS in 59m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db55d8ec5e0d47978e5b30765c741d49 : SUCCESS in 1h 05m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/05e6d20ab6bb451d8a423d7a1b0c4275 : SUCCESS in 1h 33m 13s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e5b3088064c24d0bad5ee1f6a3cd06d7 : SUCCESS in 3m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/cff68fdf197c4945b200822de5e06455 : SUCCESS in 1h 15m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c99ad2a08f914f69974c1601e1e158e2 : FAILURE in 1h 24m 57s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e83769e9869e43cab42cf95ff9ec6ed2 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"043d599df4f67b8c54ac428b0b0805c03756c993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 09:58:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f4408cad88f925218a37bb629aba641d19be45e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 12:57:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e8dbbb508664303a1fe9855264d1e44 : SUCCESS in 1h 00m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/173ee356677c4ef291cab2ab2635fa5e : SUCCESS in 1h 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c07ab2bfa74ebfb5b0b543c9bc734f : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a006eff86ee4dbdbbf8500880777ef6 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d6041194e7944159c249b33d216c260 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1f24a9c0b742b8b26f83d87d38d28a : SUCCESS in 7m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa2217f2ccae4ea08d36a1efd0b7cb72 : SUCCESS in 1h 52m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2852a74fab144fd9a94013679f49b682 : SUCCESS in 1h 05m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00b5f6a129a74a1faf6eea7b7f103acd : SUCCESS in 52m 56s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dd8170b7d30941698758e5ca1a5f8ac7 : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ffd7106c4c3d59f25de22694cab36ef148bfb58f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 12:57:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7cf4fbf05e19821db3bf75ecce6955027f22ee71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 12:58:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162b45db8f3e46c6a13c03351e5c6a86 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0539a64d1b13adc10d9912676dcb21648add9f4e","revisions":{"6afa80303359214389e8db27aea716a8fbf9bb68":{"kind":"REWORK","_number":1,"created":"2020-08-18 08:27:42.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/58/746658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/58/746658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/746658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/746658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/746658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/58/746658/1"}}},"commit":{"parents":[{"commit":"25ec807af3f6be9efb6bda25c4657b42ab713596","subject":"Merge \"Adding description for testcases - identity part2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/25ec807af3f6be9efb6bda25c4657b42ab713596"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-18 08:10:03.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-18 08:10:03.000000000","tz":480},"subject":"Adding description for testcases - swift part2","message":"Adding description for testcases - swift part2\n\nWhen Tempest is used in customer site, often we are required to\nprovide a testcase list including testcase names and descriptions.\nNow no this kind of doc is available, so we can add descriptions\nwith the format of doc string for every testcase, so later we\ncan generata such a testcase description list.\n\nThere are hundreds of testcases missing descriptions, so we can\nadd them gradually, and limit the modified files in one patch\nfor the convenience of reviewing.\n\nChange-Id: Id8f747c0eb018847bd9dc5bc3801ee973704da4b\npartially-implements: blueprint testcase-description\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6afa80303359214389e8db27aea716a8fbf9bb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6afa80303359214389e8db27aea716a8fbf9bb68"}]},"branch":"refs/heads/master"},"0539a64d1b13adc10d9912676dcb21648add9f4e":{"kind":"REWORK","_number":2,"created":"2020-08-20 05:44:55.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/58/746658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/58/746658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/746658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/746658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/746658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/58/746658/2"}}},"commit":{"parents":[{"commit":"25ec807af3f6be9efb6bda25c4657b42ab713596","subject":"Merge \"Adding description for testcases - identity part2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/25ec807af3f6be9efb6bda25c4657b42ab713596"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-18 08:10:03.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-20 03:06:16.000000000","tz":480},"subject":"Adding description for testcases - swift part2","message":"Adding description for testcases - swift part2\n\nWhen Tempest is used in customer site, often we are required to\nprovide a testcase list including testcase names and descriptions.\nNow no this kind of doc is available, so we can add descriptions\nwith the format of doc string for every testcase, so later we\ncan generata such a testcase description list.\n\nThere are hundreds of testcases missing descriptions, so we can\nadd them gradually, and limit the modified files in one patch\nfor the convenience of reviewing.\n\nChange-Id: Id8f747c0eb018847bd9dc5bc3801ee973704da4b\npartially-implements: blueprint testcase-description\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0539a64d1b13adc10d9912676dcb21648add9f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0539a64d1b13adc10d9912676dcb21648add9f4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
