)]}'
{"id":"openstack%2Ftempest~749441","triplet_id":"openstack%2Ftempest~master~I5e49036fe7a2e17d9591cff1181280d1cc293061","project":"openstack/tempest","branch":"master","topic":"test-copy-image","hashtags":[],"change_id":"I5e49036fe7a2e17d9591cff1181280d1cc293061","subject":"Add additional glance image import copy-image tests","status":"ABANDONED","created":"2020-09-02 06:26:04.000000000","updated":"2022-12-24 05:51:03.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c5d63dfa9733e4c560f776e6136267690c9bba26","_number":749441,"virtual_id_number":749441,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-02 08:05:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":-1,"date":"2020-12-25 20:54:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 13:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-08 19:15:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"0eaf53effba6cfa5be52f6a763cd1e4a8bb30b39","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-09-02 06:26:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"337d67e18af5ddab24860af44b4d7de081aec257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 06:30:09.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"121d11c9fb47995085d2ed55f36ccbda1dfd59b1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-09-02 06:50:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88b32cb958d7b5c7475929bf938b940cfac420a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 10:38:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/49dbb0b06a8447b18e531392e2b3ae9b : SUCCESS in 58m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c864b216a874fbab0dfdd65a3f66e86 : SUCCESS in 1h 17m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/459d3163a9304485ac04986d6b6b6c45 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964cadf1f66d45bfa7847b176f563e6f : FAILURE in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1203479aadf4904b06d51902186f1ac : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fb1bf11e74b4304b1a1c84dd93ee8e6 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392a4e1922774486b1afe79493e03297 : SUCCESS in 12m 54s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/876d64e13ebc493e9c6cc2898956de52 : SUCCESS in 2h 02m 04s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/528a8548fc194a2daf449bb62d68eabf : SUCCESS in 1h 37m 58s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/15bc69a691284792b3a9234ba9f0a379 : SUCCESS in 1h 28m 50s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7626158301a0438f88d763d0a793db4b : SUCCESS in 1h 19m 34s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/c29a048196af4a3b997d4ff54cf08a11 : SUCCESS in 1h 26m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/58e70af920ac4bea9031bcccd09fe0eb : SUCCESS in 2h 53m 13s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5f89411e9d0c4fb39652249f36611c70 : SUCCESS in 47m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66f028f71dc14957adac2baebcd6249f : SUCCESS in 57m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2bad194e54b4dfebd7a024ef82f48e5 : SUCCESS in 2h 00m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b89b6b64f25f44769b2ce6d7a600c210 : SUCCESS in 1h 13m 20s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/88e5db4cd5e74284a56f3570065bf767 : SUCCESS in 1h 16m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9e75fc14390484ab236fd9211f95584 : SUCCESS in 1h 05m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d2fc3d95a20f4ac3bc24a58c57a79901 : SUCCESS in 1h 21m 01s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/32e15c8c999a47f1a26226a63b042084 : SUCCESS in 4m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/56e975e1f8164af8aabbadf7f180c210 : SUCCESS in 1h 23m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0ef9b809c85f452ba84303d6c835a485 : SUCCESS in 1h 01m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/17d174f4748c4c84ac6156bae6c5e5c1 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c17ab56b056cf9fecb906f6f2b4211b430bdd912","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-09-21 08:22:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ef74eccde14e4ba759562541af6b997d1ec81af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:02:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b057a4f6e4c04e0dbf3f8cbdee4cc0f8 : SUCCESS in 59m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba48b32d678544e2b5d3ae84ce1a258d : SUCCESS in 1h 10m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1812b18f2d748fca475110ffdc3403c : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b6ec4bae92f4e7a82ea2863f0f0a47e : FAILURE in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/329ede54da814ce884871fba57094222 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a12fa78973144cc811a89473fc059b4 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718aa46372f94764a39f8ba05dc9bf26 : SUCCESS in 23m 19s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d6e05b6ace42425c998d23d7d1c29a3c : SUCCESS in 1h 55m 46s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/513ada58c4af43bcbc06bd85aea75466 : SUCCESS in 1h 19m 48s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/abed1685ebc043e09386749c20dc5181 : SUCCESS in 1h 17m 16s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/02fd36e263074abb8db22b3853b120da : SUCCESS in 1h 15m 42s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/345c0673c8ca4b62a2c990dc90882c13 : SUCCESS in 1h 11m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6aad15028255439eb57c94e918278cc0 : SUCCESS in 1h 27m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/556a23d3eda94248859cf7e076a3114e : SUCCESS in 46m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/771ed827aa7b406dac2d378b07515576 : SUCCESS in 54m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5bfb1c62a9da49a5a9d65224ebf7d06b : SUCCESS in 2h 03m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/015613ff220f4fcfb9f86dd4b146221f : SUCCESS in 55m 13s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/129feb606c0943f2b8e6921e0cf24474 : SUCCESS in 58m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af7466fea86f42fcad04b54f554affeb : SUCCESS in 1h 05m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/09eefbf8d11f445d967a0c1b17af8efa : FAILURE in 1h 13m 21s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/66671eb3db264f528b11770cad737824 : SUCCESS in 3m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6644fd0dd20c41f3b0d3e261c35392e8 : SUCCESS in 1h 01m 31s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5a823ada11e244c4a4509157b8989113 : SUCCESS in 55m 24s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2c11c38883b2476aa021a4a0dc646ebb : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e8f173ca0a4e89640be83798d5c960b0b7a411b2","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-10-01 10:04:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5bcc355511487265e58b963d578b878cd26ed2ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:04:42.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"de9b7b7b818ca7f4121f059afc87f3ab5c16312f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-10-01 10:41:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"24ed2f0cb0081227b840404e60eff8345c5b8281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 13:07:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e057c222844e438f81e4537d1224c396 : SUCCESS in 1h 01m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/012b7991bb2f41feb5910f3b89bbfd36 : FAILURE in 1h 12m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8671f54d31c4899ab71df8b3ed78fab : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/128acb6d860042ad94bfa334d45e827f : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79cd7f4d869b42f09b2781a5bd9e7ee5 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8fefaddf8904430b3af3601020b60ff : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2dcb5765464e10a3bf3feee1d11a4d : SUCCESS in 11m 19s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/45220e4833ee46ff8ad1ccfcf5ed54a8 : FAILURE in 1h 37m 17s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a1991362130844b1895ebf5ec50d2c5d : SUCCESS in 1h 23m 11s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/4a4db101962447b2bcac2c72d9ebe31f : SUCCESS in 1h 23m 09s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/d90bdc185bd84514adc41a9b21cf4a96 : SUCCESS in 1h 14m 08s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9fae05df36344847a6a52a124af9bb1a : SUCCESS in 1h 19m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/431dc1a035c943c5a9a865b831d6e273 : SUCCESS in 1h 43m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/624926d0891e437c99a17020948ef190 : SUCCESS in 43m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38db059a4273469ab68a55c50cb1f7e1 : FAILURE in 28m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04ee5e1ee4a64ad8bc778741561d99f7 : SUCCESS in 2h 06m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dac9faed4ab242798051cfa7d9654fcf : SUCCESS in 55m 43s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e6f63581be08426688743f8bb3fc092d : SUCCESS in 59m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b7a67995597460fb01f8a1b89ca74a9 : SUCCESS in 2h 03m 10s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/22411ccc2d8f47fa99bfbd7789541d85 : FAILURE in 1h 15m 29s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b4f10a0c2d54443d83516cd01be59a1c : SUCCESS in 3m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/24e4cbb4107d4e889ea6651297bbdd4d : FAILURE in 1h 20m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/18b1e3db2ed146a192288fcfdf9a5627 : FAILURE in 58m 36s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6dcfe889e9824158bb82515497139cda : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":4},{"id":"d18b935f6006ebb80dfd53e71d7b64f42935f716","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-10-04 04:42:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"3b8e871c5ea316503135d186d8f71aa37b47da8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-04 06:48:09.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/59a4ad3d150c44d39f162f8c107ffbf3 : SUCCESS in 53m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69d4408dd8ec4f56bc3196dd21395846 : FAILURE in 58m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e52a19bf73b1476780acfcdcc6e99e24 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7a7a1d067d4a4392830562853fe45b : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b1c7fdb2bd94294a3ae75b3352d82f5 : RETRY_LIMIT in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca626d2fff4945a3abdd16374da0b6b5 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d507ec9e4fd41f9b352883d6dff6c05 : SUCCESS in 11m 53s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/31973134fa29435b94ca40ee0f2b6f94 : SUCCESS in 1h 31m 05s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c9b38846ccec4093a75b59279d83ecb2 : SUCCESS in 1h 12m 53s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7abad660e2254459bd88e7f606d12d4a : SUCCESS in 1h 06m 41s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e7210958e48c443fa36b47b4a1782a3d : SUCCESS in 1h 11m 25s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/62cdfa3ea9de44dcb733d4a9695745f1 : SUCCESS in 1h 05m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9429a934312a494c96745da7505179e7 : SUCCESS in 1h 34m 31s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5f4706e438d0459e93a3c18692da6c40 : SUCCESS in 43m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e9d1d8c69454c31b17ac3dded29beeb : SUCCESS in 56m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ec0db6623f3249738abb7a633db13536 : SUCCESS in 2h 03m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/267bbfb2e6414e4d9c0f65ff4a04c5eb : SUCCESS in 49m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d6ae997620584f2bbb1c4d6ad612e2d5 : SUCCESS in 1h 04m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b50f3b14909b4942bed71df84e08b333 : SUCCESS in 1h 01m 07s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/7226465aa46340a3990578e13c7ff7d3 : FAILURE in 55m 01s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/54be22eb1e724f31a2602193982cca70 : SUCCESS in 2m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f2ea715816924438bc97c7e0379a12c9 : SUCCESS in 1h 12m 18s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/46ca3df9e14547238e0ef39359559632 : FAILURE in 57m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1f4ae3df75134b30ac0444e201b9a993 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":5},{"id":"7640922b354a8bf6381f3750738b6504a795542f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-08 19:15:45.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2f2084c94d5b06875f0318bb5b1b6e1a3a317c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2020-12-02 05:51:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7e3ff88a0b0ffee6f2d9703b6a393137cc53b7ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 08:05:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e072010512f4d8a95d024488cc0e0fe : SUCCESS in 1h 05m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/107042e40f6b4a8b85b12dcf66a9e857 : SUCCESS in 1h 30m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fd7836d55c742b7a6c71626940e3f94 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe2914b428143c68d06550f48942708 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f55c8c909134b4191fb1900ed6b2a50 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0e80d58b98a4485af7dbcce4c25ff50 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2184a00f6a2436e99fcb447deb7771a : SUCCESS in 14m 08s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1177fc21b723407a81dbfecea313ce39 : SUCCESS in 1h 53m 57s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d14e1deb5dd54269936bafc4c1412ca5 : SUCCESS in 1h 12m 50s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/484ba063581741e9a6898abe6d8c9ea2 : FAILURE in 1h 02m 54s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/0b6870df1e5a440da31341b3f3dae277 : SUCCESS in 1h 21m 20s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1048d91124114844b726d50c050b382c : SUCCESS in 1h 16m 34s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a84812adf5924a758f0fbe9274e2c9d7 : SUCCESS in 1h 14m 42s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/675f32301ae14906b7ca213ca2022116 : SUCCESS in 1h 27m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6817dbb3c9ae40f184c8fad1d7c506e5 : SUCCESS in 1h 41m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ae3b50f6f0924f12b9ad8e54cef740bd : SUCCESS in 50m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2497a883e2841f7878cf2c2088129c8 : SUCCESS in 1h 02m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c17af8188aa4df9a90e361e3effcede : SUCCESS in 2h 08m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bf4e33830ad544679ac4c13cef235858 : SUCCESS in 57m 19s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/68aaeb64e9944f52b0aa2c31d2ce9ccd : SUCCESS in 1h 18m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/41b46a8a3a014c3cb2750211ed88f152 : SUCCESS in 59m 36s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/54485536ec1a47b18eb9089084b9205b : FAILURE in 1h 02m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f853d44986444825aacea37ad82fbae2 : SUCCESS in 2m 21s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/471a6a0dc52247028e1cc0b1b1574efa : SUCCESS in 1h 57m 58s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9bc6cd44f4c547cf948932c1563fe628 : SUCCESS in 1h 13m 45s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f59c29ef2f73496989ebcc843e6c76f2 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":6},{"id":"bfd13791869ac0a272bd7a5009336c126da810b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 20:54:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\n1 comment, most part is good. also I would like to see this test running in nova ceph job.","accounts_in_message":[],"_revision_number":6},{"id":"966fa670a3316a99fbff96c6dcba8f17cd55a4ab","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 20:57:50.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e 1 comment, most part is good. also I would like to see this test running in nova ceph job.\n\ntesting here -https://review.opendev.org/c/openstack/glance/+/768519","accounts_in_message":[],"_revision_number":6},{"id":"c5d63dfa9733e4c560f776e6136267690c9bba26","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-24 05:51:03.000000000","message":"Abandoned\n\nNo activities on this since long time, feel free to restore it if needed.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b76658488eec0861cf52e18537976f994d21f411","revisions":{"d7d8ba03325970210ad8cc38a4b33ff05b2d56eb":{"kind":"REWORK","_number":1,"created":"2020-09-02 06:26:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/749441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/41/749441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/41/749441/1"}}},"commit":{"parents":[{"commit":"3003ab2afd53460ac93cac5536a4579f9d4ced59","subject":"Add glance image import copy-image tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3003ab2afd53460ac93cac5536a4579f9d4ced59"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-08-31 04:58:43.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-09-02 06:05:49.000000000","tz":0},"subject":"Add additional glance image import copy-image tests","message":"Add additional glance image import copy-image tests\n\nAdding testcase for copy-image import with specified\nstore list.\n\nDepends-On: Ia7c967946304a8c7a79774b4116dd6169822f3ac\nChange-Id: I5e49036fe7a2e17d9591cff1181280d1cc293061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d7d8ba03325970210ad8cc38a4b33ff05b2d56eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d7d8ba03325970210ad8cc38a4b33ff05b2d56eb"}]},"branch":"refs/heads/master"},"1ad10942665a2c296d83bdbff39013684752eff8":{"kind":"REWORK","_number":2,"created":"2020-09-02 06:50:09.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/749441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/41/749441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/41/749441/2"}}},"commit":{"parents":[{"commit":"aee94663111befadbb570673ae7155f734876029","subject":"Add glance image import copy-image tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aee94663111befadbb570673ae7155f734876029"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-09-02 06:44:23.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-09-02 06:44:23.000000000","tz":0},"subject":"Add additional glance image import copy-image tests","message":"Add additional glance image import copy-image tests\n\nAdding testcase for copy-image import with specified\nstore list.\n\nDepends-On: Ia7c967946304a8c7a79774b4116dd6169822f3ac\nChange-Id: I5e49036fe7a2e17d9591cff1181280d1cc293061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1ad10942665a2c296d83bdbff39013684752eff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1ad10942665a2c296d83bdbff39013684752eff8"}]},"branch":"refs/heads/master"},"033bf5d8c878095ff9187e507acdaff905ff990f":{"kind":"REWORK","_number":3,"created":"2020-10-01 10:04:29.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/749441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/41/749441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/41/749441/3"}}},"commit":{"parents":[{"commit":"aee94663111befadbb570673ae7155f734876029","subject":"Add glance image import copy-image tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aee94663111befadbb570673ae7155f734876029"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-09-02 06:44:23.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-10-01 10:04:04.000000000","tz":0},"subject":"Add additional glance image import copy-image tests","message":"Add additional glance image import copy-image tests\n\nAdding testcase for copy-image import with specified\nstore list.\n\nDepends-On: Ia7c967946304a8c7a79774b4116dd6169822f3ac\nChange-Id: I5e49036fe7a2e17d9591cff1181280d1cc293061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/033bf5d8c878095ff9187e507acdaff905ff990f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/033bf5d8c878095ff9187e507acdaff905ff990f"}]},"branch":"refs/heads/master"},"4ed91ce4e9c4843fcd7d1d7d1868948a5e137f74":{"kind":"REWORK","_number":4,"created":"2020-10-01 10:41:01.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/749441/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/41/749441/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/41/749441/4"}}},"commit":{"parents":[{"commit":"5dbc5c81877ebb072cc27ab03f9b5f91b2ef194f","subject":"Add glance image import copy-image tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5dbc5c81877ebb072cc27ab03f9b5f91b2ef194f"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-09-02 06:44:23.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-10-01 10:35:41.000000000","tz":0},"subject":"Add additional glance image import copy-image tests","message":"Add additional glance image import copy-image tests\n\nAdding testcase for copy-image import with specified\nstore list.\n\nDepends-On: Ia7c967946304a8c7a79774b4116dd6169822f3ac\nChange-Id: I5e49036fe7a2e17d9591cff1181280d1cc293061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4ed91ce4e9c4843fcd7d1d7d1868948a5e137f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4ed91ce4e9c4843fcd7d1d7d1868948a5e137f74"}]},"branch":"refs/heads/master"},"26dfb4a73d4cb831a37eb367975a739fdc1d8c7e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-04 04:42:46.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/749441/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/41/749441/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/41/749441/5"}}},"commit":{"parents":[{"commit":"42f338412cd6857e040bc37830ed40fce20b7108","subject":"Add glance image import copy-image tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/42f338412cd6857e040bc37830ed40fce20b7108"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-09-02 06:44:23.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2020-10-04 04:42:46.000000000","tz":0},"subject":"Add additional glance image import copy-image tests","message":"Add additional glance image import copy-image tests\n\nAdding testcase for copy-image import with specified\nstore list.\n\nDepends-On: Ia7c967946304a8c7a79774b4116dd6169822f3ac\nChange-Id: I5e49036fe7a2e17d9591cff1181280d1cc293061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26dfb4a73d4cb831a37eb367975a739fdc1d8c7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26dfb4a73d4cb831a37eb367975a739fdc1d8c7e"}]},"branch":"refs/heads/master"},"b76658488eec0861cf52e18537976f994d21f411":{"kind":"REWORK","_number":6,"created":"2020-12-02 05:51:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/41/749441/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/41/749441/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/41/749441/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/41/749441/6"}}},"commit":{"parents":[{"commit":"e5287c6ee879ca609d4df10b7cced71ea6454ca6","subject":"Merge \"Decentralize the zuul jobs into zuul.d\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e5287c6ee879ca609d4df10b7cced71ea6454ca6"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-12-02 05:43:14.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2020-12-02 05:43:14.000000000","tz":0},"subject":"Add additional glance image import copy-image tests","message":"Add additional glance image import copy-image tests\n\nAdding testcase for copy-image import with specified\nstore list.\n\nChange-Id: I5e49036fe7a2e17d9591cff1181280d1cc293061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b76658488eec0861cf52e18537976f994d21f411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b76658488eec0861cf52e18537976f994d21f411"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
