)]}'
{"id":"openstack%2Ftempest~750269","triplet_id":"openstack%2Ftempest~master~I4be010dae9b83549b970b117b389626caa2806a0","project":"openstack/tempest","branch":"master","topic":"bp/fix-microversion-gap","hashtags":[],"change_id":"I4be010dae9b83549b970b117b389626caa2806a0","subject":"Add volume services schema for microversion 3.7","status":"MERGED","created":"2020-09-08 03:30:21.000000000","updated":"2021-06-15 01:10:12.000000000","submitted":"2021-06-15 01:08:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750269-bp/fix-microversion-gap","meta_rev_id":"375cfb6a84006cf5e5b13bcf213ceda6ef57ef40","_number":750269,"virtual_id_number":750269,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 01:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-14 22:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-14 22:18:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 09:14:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 01:42:23.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"cd9a731062f2869c5e1be13ee0c7cb965c0b2c38","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-09-08 03:30:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ba6621c1acef5590b855e1b20a2d7098b25d2c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 05:32:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0a5c23c692ed46b589d45c2776276014 : SUCCESS in 1h 02m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1745c7d0dbbd4d57acda1281ae6f930f : SUCCESS in 1h 11m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b68d1c5298e445cfa572d20b4c0f9af9 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fb6f0382a54de593bd0a0c5389d2b7 : FAILURE in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a083e65151144e26ba984fc6236c5963 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4972429360914ba4b979a10d989bf95e : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56bdd0d6c1e449b2aff3b02bfd249164 : SUCCESS in 8m 21s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/ca7b874848244e999ddd999f3a9ef4a1 : FAILURE in 1h 42m 37s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6250b813fca64610a95329c97c253a47 : SUCCESS in 1h 14m 43s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/dd1204935a6449e48d765d09d343f181 : SUCCESS in 1h 22m 36s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/eb10c7c0d889438393983ba8eb198bfc : SUCCESS in 1h 13m 53s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/ef2d0ca4907a450ab994273056e3122b : SUCCESS in 1h 15m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/73f984b2dfbd494b971cb2b952d1adad : SUCCESS in 1h 30m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e9f1cbc7947d4c5d844e1c75ae36d90d : SUCCESS in 41m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa997a3edce4417fa5e83a00b30a486c : SUCCESS in 52m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2b1e53a8adc04204b3b626d2737d9783 : SUCCESS in 1h 55m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/17756322fbcc4a8889c38e63c7a160f5 : SUCCESS in 52m 25s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e729a4f2abb04545aaecde2cc5d9b4ab : SUCCESS in 1h 03m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fc43c73f3b744c5b0d66359919d939d : SUCCESS in 44m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b11896addcff4caaba3eb1fcd1c53aa2 : FAILURE in 1h 02m 14s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8fa46bfc7d824a6b8ec0fdcfc118d9fb : SUCCESS in 2m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9c748eada0a9495e9e2a4f49ad709cc5 : SUCCESS in 1h 15m 27s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/49eff294b3044dbaa60ee46cfa995c31 : SUCCESS in 1h 01m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/413f8de6f3f44aa68220cc079faea068 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"95299a9451dbf9278087fb3fd37d6c9273d8cf64","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-09-08 06:18:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a4d2d834dd9ec96526e6c552dfdf497bb862db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 09:14:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1d365b6a65bf4574b287a3a398850114 : SUCCESS in 56m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df7515b0472b4c75b7452d3a26ac8266 : SUCCESS in 1h 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6569b2b9f23d4cd6b17a0c9ad261fa75 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e785c609bf354f628364b5e76367ae82 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed03213f211c4d54a4812619afd3a8f2 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a680596b89df478980114568448e4c92 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907174e4e2364d6183903ca8b743b5ca : SUCCESS in 14m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/43e4e5abd46148c7b61f6f50599a908d : SUCCESS in 1h 20m 09s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/fe0656f84ac749838c3c6ef6496019b9 : SUCCESS in 1h 09m 53s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c9614b83bc294d59a65608df2e223925 : SUCCESS in 1h 08m 48s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/daaa790e661a491a96f3115a2cee04ea : SUCCESS in 1h 20m 30s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/b161af7da119430dab2b281e72c30f42 : SUCCESS in 1h 04m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/58f0572f7d7e43f7954ec791f03b63ec : SUCCESS in 1h 30m 50s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/019d7ec9b58f4d54b4d17f4704b64bf1 : SUCCESS in 40m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c292cbbe48e453bb8030afd6abaa3f9 : SUCCESS in 55m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/448e1b6e4d184642a7732790bb84abe6 : SUCCESS in 2h 02m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d0f436dc5dfa47f88a40d1beb97e1f7d : SUCCESS in 53m 51s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0b583d2b81d74e4a8f606105d9412475 : SUCCESS in 59m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35da9d048231405a9f319c85cbb78065 : SUCCESS in 1h 06m 51s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/da5998f1ae4b468ca47da1e4f337b047 : SUCCESS in 1h 07m 09s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d3aaa05f14c04b228691df59a45f2698 : SUCCESS in 4m 17s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/95b594597f324fa8b436c0b295b89af9 : SUCCESS in 1h 01m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4c1749090d5b4e9e9154d7d66c0955ce : SUCCESS in 1h 21m 02s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9f9419106b91468a91aab020fc7c5feb : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"becdce722ace3cdf5bbf0a3ebc51bf517b9d8c12","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-06-14 22:18:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"b76d47382284fc1e96b7cc9c18cc05e11d0ed83c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 22:19:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"598660863ccf1ef985422ed0f8a9a2b604e0de06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:08:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c874651fbdb4c4bac7268894d29b4ce : SUCCESS in 1h 08m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65be91a10a284efabba19f37f3146191 : SUCCESS in 1h 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3cad78e8a646c19af544b4f4b67bd5 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb48b46183c04852924ab2393ea82c92 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7e127a20594d96917296d7e3a5852c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e7a7cffad24dbcb4f330f337dbbd0e : SUCCESS in 15m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b609dae23dec4528abaee4bcdd894d93 : SUCCESS in 1h 59m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/599ee6981da84b12b6288e6ff5193be5 : SUCCESS in 57m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29042e4ca7024a0aada88c9e47ebf321 : SUCCESS in 47m 51s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ca94faad156d44caa2c085ca245540a3 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":2},{"id":"84cab1ca017573f48119843eb070c380cc86950a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:08:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"375cfb6a84006cf5e5b13bcf213ceda6ef57ef40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 01:10:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf8077da81c42cdb5b15695bbadccfb : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e75f8ca33c302ef306512bb65ea84fd4200a25f3","revisions":{"ff925c0f9c9e479273289f7cf0eaac4f63c7f54d":{"kind":"REWORK","_number":1,"created":"2020-09-08 03:30:21.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/69/750269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/69/750269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/750269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/750269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/750269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/69/750269/1"}}},"commit":{"parents":[{"commit":"accd1e0d0c1b7448ff10b2d1df1a83677fe4de59","subject":"Merge \"Adding description for testcases - compute part9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/accd1e0d0c1b7448ff10b2d1df1a83677fe4de59"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-09-08 03:11:31.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-09-08 03:23:02.000000000","tz":480},"subject":"Add volume services schema for microversion 3.7","message":"Add volume services schema for microversion 3.7\n\nIn volume microversion 3.7, optional attribute cluster is added in the\nresponse body of list_services, so this is to add volume services\nschema for volume microversion 3.7.\n\nChange-Id: I4be010dae9b83549b970b117b389626caa2806a0\npartially-implements: blueprint fix-microversion-gap\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff925c0f9c9e479273289f7cf0eaac4f63c7f54d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff925c0f9c9e479273289f7cf0eaac4f63c7f54d"}]},"branch":"refs/heads/master"},"e75f8ca33c302ef306512bb65ea84fd4200a25f3":{"kind":"REWORK","_number":2,"created":"2020-09-08 06:18:23.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/69/750269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/69/750269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/750269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/750269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/750269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/69/750269/2"}}},"commit":{"parents":[{"commit":"accd1e0d0c1b7448ff10b2d1df1a83677fe4de59","subject":"Merge \"Adding description for testcases - compute part9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/accd1e0d0c1b7448ff10b2d1df1a83677fe4de59"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-09-08 03:11:31.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-09-08 06:08:45.000000000","tz":480},"subject":"Add volume services schema for microversion 3.7","message":"Add volume services schema for microversion 3.7\n\nIn volume microversion 3.7, optional attribute cluster is added in the\nresponse body of list_services, so this is to add volume services\nschema for volume microversion 3.7.\n\nChange-Id: I4be010dae9b83549b970b117b389626caa2806a0\npartially-implements: blueprint fix-microversion-gap\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e75f8ca33c302ef306512bb65ea84fd4200a25f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e75f8ca33c302ef306512bb65ea84fd4200a25f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
