)]}'
{"id":"openstack%2Ftempest~750329","triplet_id":"openstack%2Ftempest~master~I91f5ced7873755081f94117c6a43bffcb85faab6","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I91f5ced7873755081f94117c6a43bffcb85faab6","subject":"Fix test_allowed_address_pair test case cleanup","status":"MERGED","created":"2020-09-08 12:08:15.000000000","updated":"2020-09-17 12:48:31.000000000","submitted":"2020-09-17 12:46:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750329-1600346790588-515a9e8b","meta_rev_id":"374312b63f96d4598b8a4d99a7e3542a267c1e49","_number":750329,"virtual_id_number":750329,"owner":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 12:46:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"date":"2020-09-16 17:22:10.000000000","_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-08 19:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":1,"date":"2020-09-08 19:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},{"value":2,"date":"2020-09-11 16:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-09-15 13:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"date":"2020-09-16 20:10:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"value":1,"date":"2020-09-08 20:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-09-15 13:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"date":"2020-09-08 20:01:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"date":"2020-09-08 20:01:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"date":"2020-09-08 20:01:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-09-08 20:01:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"date":"2020-09-08 20:01:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"value":0,"date":"2020-09-08 20:01:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},{"value":1,"date":"2020-09-11 16:09:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"date":"2020-09-08 20:01:05.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"value":0,"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"value":0,"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 19:15:46.000000000","updated_by":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"reviewer":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"state":"REVIEWER"},{"updated":"2020-09-08 19:30:41.000000000","updated_by":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"reviewer":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"state":"REVIEWER"},{"updated":"2020-09-08 20:01:05.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2020-09-08 20:01:05.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-09-08 20:01:05.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2020-09-08 20:01:05.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2020-09-08 20:01:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-08 20:01:05.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2020-09-08 20:01:05.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2020-09-08 20:02:51.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"},{"updated":"2020-09-11 16:09:54.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-09-15 13:42:20.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-09-16 20:10:00.000000000","updated_by":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"reviewer":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"state":"REVIEWER"},{"updated":"2020-09-17 12:46:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecf5a07c8971d5d4e9002423144350decb86f091","author":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"date":"2020-09-08 12:08:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"955f2eb00b5d6b1e40b2c5d7b89eb6f064c77c80","author":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"date":"2020-09-08 12:20:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0706318c7b024bb612a536ac590481ebff940518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 14:22:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e80fca8eee040a69571171ee695d15b : SUCCESS in 1h 05m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/909e7fbf78d944a381a2c4f830deb23c : SUCCESS in 1h 15m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae6248902c504f4aa95bd5f2f918ae2a : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f22485dadf04628b087e139e699a8a1 : FAILURE in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c807c68664f64221a3526275e13cac4e : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c32e237042c42538109f6e3d3c667a4 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5751207d32034f18b27a4791b5090502 : SUCCESS in 11m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/fce6ebe8bc7f42abb1aee7b4b856696f : SUCCESS in 1h 37m 48s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d3ef6323fbc645308032c21e27722336 : SUCCESS in 1h 17m 04s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c7f9e1bb66964580a821e267e6082342 : SUCCESS in 1h 26m 41s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/fb4a51aaaf274de1814affa7fd5f20b6 : SUCCESS in 1h 16m 20s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/af945bd1132f422082628113190273dd : SUCCESS in 1h 16m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4ba04b204f774d9f9d2ede6c37109a01 : SUCCESS in 1h 54m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c514737d8c10496da5c524a6e0e7da78 : SUCCESS in 35m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3642813b2c64658b46e98739d81441e : SUCCESS in 54m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f69d68355f514448a25f4f223f97c02f : SUCCESS in 2h 00m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d73c42605def442da194518c408fe106 : SUCCESS in 44m 07s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/41b75d2034d942d0b59c3767f3f3c39d : SUCCESS in 1h 08m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf8e001511a54a52bb88e537b0b2b186 : SUCCESS in 1h 09m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d1d90de8cae044369d63d78f5ab4278d : SUCCESS in 1h 08m 33s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b66f262283e14b1bbd85e952b80211b0 : SUCCESS in 2m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/43503f72b66c48aebe8f05668bd4f7a4 : SUCCESS in 1h 03m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5c80599996424379b7027d7e0f7c6b26 : SUCCESS in 54m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/366d8a41f08e49299fee72dd70ef286e : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"fd5d34e160ea5e37753b89b75653e1493e497dab","author":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"date":"2020-09-08 14:33:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d4f19beccc03c621aed5788dd37db1f80de261ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 18:00:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5f45933837e471da65042b9fa05051e : SUCCESS in 1h 06m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f3ba8ca117f4dbea389fc06382fbaf8 : SUCCESS in 1h 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c22a91949a8544c589405f92f8337c0c : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39326b71fd3a48c8b80ecebfb546a9e6 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60a948e1b92a460aa946160b942cfbf0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbe70edaf11c4133b9fb13c1d1db4a6f : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c13a7f246a47f4bf747100d140264d : SUCCESS in 12m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/fe73c5c7aa614fc58af2a02ef6f20fd2 : SUCCESS in 1h 30m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/31cdc2e8f94d477382659a6180655f8e : SUCCESS in 1h 12m 11s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7593e472c8044084983ee9164fa5ab58 : SUCCESS in 1h 13m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/301975028787479cb73174a389f0372a : SUCCESS in 1h 22m 01s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/5cf0705813e143cea1429030e451a4ca : SUCCESS in 1h 14m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/24688fc0653c4b089e7065a0d36d23e1 : SUCCESS in 1h 35m 09s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b5c71252e4e044d0b82d039487ad9917 : SUCCESS in 45m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e5f0e85bd6e46bf98425278b6985942 : SUCCESS in 59m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0dfccfa60d59433a8f711a101859684c : SUCCESS in 2h 05m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0008ce2b83c94c49a104cbad46b2a8db : SUCCESS in 48m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1586dc47a9444c9cb784f59f3b890871 : SUCCESS in 1h 02m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fd3063a4df34ff88dd5426e44dae661 : SUCCESS in 1h 11m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/74cf403ec7a94119af795acf44d46260 : SUCCESS in 1h 18m 12s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6af2eda9f8a448fea8d3a617852a4af0 : SUCCESS in 5m 37s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/cf89b08c4bc047a4b5f87b8861623a34 : SUCCESS in 54m 39s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8bde533dd1964e2f9971dd9068ccef49 : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c37ca039db3c4b359b250b63dbb91c6c : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f5ec0a8d82699ef2babd76b440e581ff54c6be20","author":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"date":"2020-09-08 19:05:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b25f800796ff35c3241a1aea1b4aecab85a8aa69","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2020-09-08 19:15:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d2adbd56272167fddbf692f1aae31739d4a7dd2c","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2020-09-08 19:30:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"59a4c0fabeffe344e98dc2b4188e28e443cd6adb","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-08 20:02:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0cfc1efdcd8615a0731827e07536d52b976ca3a0","author":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"date":"2020-09-09 04:37:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef18b332e39d42e222a6d961e04d46707c147b5e","author":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"date":"2020-09-09 12:56:54.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\nAdded detailed description in the bug 1894835.","accounts_in_message":[],"_revision_number":3},{"id":"4749706f5c6201e18f4158462ec6e6a8c6539f62","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-11 16:09:54.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\nThe same problem with other network tests has been fixed recently by: https://review.opendev.org/#/c/748765","accounts_in_message":[],"_revision_number":3},{"id":"9ffa54041e77ae40a629ae1309c94b6ccac86574","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-09-15 13:42:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nAdding more commit message is ideal, but I don\u0027t think it\u0027s a blocker to approve","accounts_in_message":[],"_revision_number":3},{"id":"61759f75d10d7d9081a9a69dbc60e6b99ddfc168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 13:42:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d4fb0fcc535a416ff09fbb70aafb8aa8c7a3cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 16:04:22.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/17b62030f1c44db8861b44f2cec37733 : POST_FAILURE in 47m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37603ed3fb774d08a48ba1e24bf52017 : POST_FAILURE in 15m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e99270df1ea4a9b89c393c6af6633e5 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d43b3757cb54ad789b414ecbe292cea : FAILURE in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9782436d90244c83b3c1dacb840e16d2 : FAILURE in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ca1c82b8144f369ec70d4eea119497 : FAILURE in 11m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4097442b67f24c119fb7a7ca9148407f : SUCCESS in 2h 03m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c8e3381317a407e99351491ef7c7a87 : FAILURE in 58m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a8d70b129be41269d4fac5c01faf8ad : FAILURE in 13m 02s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b1ba5d10e44c45fe99990bab2f32e358 : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":3},{"id":"7bb49c0053a648fe4ea33bc94f708bfa1a64a578","author":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"date":"2020-09-16 03:06:42.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2ab6dfe9651652172cae4e827f58efcd3692821a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 05:25:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/49c6145889e04fbd9c7f38b07da95788 : SUCCESS in 1h 01m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4491c168ef3c47968469a81fed117cb5 : SUCCESS in 1h 25m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/012aac316a514a538fb7f26a21e5ead4 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/424c7626f6884c1dbe4b423db9b8c00f : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fdea9b6db584cd393f4e5bf26445758 : FAILURE in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bed23db0b054768b016d703ae90b7d9 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d82078da62148d4912d1bfb6508a888 : SUCCESS in 18m 16s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d549d512b04f43778f298266ed40b94e : SUCCESS in 1h 48m 25s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/36628dd54b9f4d4f8b272dfc128ee933 : SUCCESS in 1h 07m 26s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6aa657549dcb4c4c985babe1d3acf252 : SUCCESS in 1h 06m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/4ff4b5c5d33b42fa942df1e52e0f40cb : SUCCESS in 1h 27m 34s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/0ec38711a19d405ab92878105e8c1cbe : SUCCESS in 1h 07m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3bba281eced646e1b7663294bb9d810e : SUCCESS in 1h 14m 10s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7ef86d577d024a4eae0dc9985736e840 : SUCCESS in 41m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/460c7a6a38484ce1a8d089ffb7198ad0 : SUCCESS in 50m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a29620e18aa44d7ebc0db7962ef11918 : SUCCESS in 2h 15m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b0cc7b534ebf4cebbbca9040a322fef5 : SUCCESS in 41m 08s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ccba7f1f62c645029319863b4ecdd706 : SUCCESS in 1h 18m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fb33e48143a4cdab43f7b4774322581 : SUCCESS in 58m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bf8be62339e74a5e87ca25724a9e27e9 : SUCCESS in 1h 27m 54s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/aeb593976c244d63a5d6cba323676828 : SUCCESS in 2m 47s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/448f8534163a4e1fab28112c6281f3da : SUCCESS in 56m 00s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4d6890adb4074bac8c93b400093d40a4 : SUCCESS in 52m 06s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/176bf56de05a42ec9545f18c7e55b576 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"fc775224e5b33f0abf76270fbab8644149de1bbb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-16 06:47:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8e2f140df730a9d1276ebe9272501e317a341ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 10:03:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c200a2410b4a44e89a11a47f2486d0d3 : SUCCESS in 1h 05m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9e8c654f9ae4422a92dbe6c73b78723 : SUCCESS in 1h 29m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/356c63bdb1c04d56a992fd58ae550876 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92bf0d74ca943c3a99b6cd3dbb5d4d9 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aefe5cc11ac346ee9d7da296f2577f21 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b84f0b02fc2f48798cadc16a0cd2b8c0 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b929b972cd64214823b8c76aa8bdc8b : SUCCESS in 25m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/8131363872924f4ebf06e854a1a5d5f3 : SUCCESS in 1h 43m 51s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/eff5e919bbb44fc596429313fe113a77 : SUCCESS in 1h 35m 31s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/190c4a6b80e14cc8b78c92ec9749cd83 : SUCCESS in 1h 13m 11s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2a8f5f1882ff4749b766572f4a6abb8f : SUCCESS in 1h 12m 57s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/a2d27e9f70a64652bae93076e268a9f9 : SUCCESS in 1h 33m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6233cc90d72c4f99835e8d4f1c54c469 : FAILURE in 1h 51m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3d4db210a68344038158e2849ee4515e : SUCCESS in 43m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6582a476681a435b921c96d1e55c66b3 : SUCCESS in 1h 04m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eae70cb466024c08b5288a901df8ec84 : SUCCESS in 2h 15m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a8927412b0ca4c2f92f44dc1da1bd937 : SUCCESS in 53m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e4e6080f2a7348c5903d20ad614d6c8a : SUCCESS in 51m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e165463522b04c68bf33643e6c7e5a6c : SUCCESS in 1h 13m 48s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f058529cf4194d98a03041d492d859f1 : FAILURE in 59m 55s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/cd4124e5151b4f708c6f15717fea3354 : SUCCESS in 7m 19s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/54a48c1b0f8b41f9937d597a357c41df : SUCCESS in 1h 09m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9cf0ebbab5df4702bf9d3307bb3d8d61 : SUCCESS in 1h 01m 55s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/93d5df302489474ea5bbf0cddfffd1e3 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"b0ca2b2dd48fdbd007b3a1df9816cdb6bacd1a63","author":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"date":"2020-09-16 10:37:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2f4af39cabbc77ac5f091ee1c32069007bbc036b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 12:49:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f6cb344cba6491d88b63546c84388de : SUCCESS in 1h 05m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41b2b6eb6a9344389b1dde17d1b760bb : SUCCESS in 1h 14m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f49b4268ec94c9989366f80007ebc8a : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c59208754b964326ab9c7a9c0d7fb9bb : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/594bc20ce13341e49e56c43e88b4d18b : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f922d86e96f4dd19d19aa5d89aa4892 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/760ffd1a63924c9fac9241322964194f : SUCCESS in 19m 51s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/138bbc077c904038a2bc0abd8e48a926 : SUCCESS in 1h 38m 51s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/640f711297ae4388bfe75abf5bfbf474 : SUCCESS in 1h 22m 30s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/b290f8a012014e05839e92124d9d4991 : SUCCESS in 1h 16m 04s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/4ddd03535db74faea9edea4aca87d0d6 : SUCCESS in 1h 20m 49s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/2a9bc952ed634118a634ba21b8cb45c3 : SUCCESS in 1h 16m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/598ca8a8ba064258b8bc5a5c30afe89e : SUCCESS in 1h 48m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/62e7d4f81b214e83b8acc8e85a2a2634 : SUCCESS in 44m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49e5d2bd1fa6469d9ce94eac81b4a774 : SUCCESS in 1h 07m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/517ff30b398c4a95bf6724cccc9a5c79 : SUCCESS in 2h 04m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/84b84cc440654de1a0b24b5df02f5751 : SUCCESS in 50m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f42df33648f54cd789d37afb0f9547af : SUCCESS in 53m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/88b456f7d9594109927e80762c9d039f : SUCCESS in 1h 04m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/31530039267449c58a5249720f681cbd : SUCCESS in 1h 23m 03s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0f5faf184bc14d0e937b10ae5f8f5efa : SUCCESS in 4m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/bf1c1b688b8f462995fa4a875fedd661 : SUCCESS in 1h 19m 04s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bbaf03f057af4987886c3d21db3fcb1e : SUCCESS in 1h 01m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5b6a8efb256746649378121e002a728d : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"01d25d4eed98229c47585a39d9656fa354fbe4b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 12:49:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ed6b82304b3d579d73b2ce8aed36da100cc1f1c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 17:11:06.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/26ec7d4ebdda4fcc9f60b2cf007002ec : SUCCESS in 57m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49cd6bda489144b1883149e546dc73d0 : SUCCESS in 1h 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e66f3ab3424dbb8f32cd7d588f3738 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4364de59d204c46908c95d693474e93 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd009dc6fa9148db863576e0ef106c45 : FAILURE in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcc82adc0094b17a5303918efa6f9df : SUCCESS in 22m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df9436163cf44995acfc84e90983a929 : SUCCESS in 2h 10m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9114c40598ec4a198a3c8c9396c80e61 : SUCCESS in 1h 07m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14d843caef4c40d19c097484b18c022e : SUCCESS in 48m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f9bca5c14839460aba7031ee33c070e4 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":3},{"id":"d57a31a239cd07cbbe86b11d3109eae8f8c89769","author":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"date":"2020-09-16 17:22:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5fcf85986311ac8b85e0d3788151ca550aad7990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 20:04:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/61a416772ad948eca3755eef99ba03d9 : SUCCESS in 1h 08m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5a1a36243284e02ad58df39cb7e236c : SUCCESS in 1h 18m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bc182d584b542c6ba64e16f00860d3f : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5b87cdd58b4c2682cb50d6d8dfa59b : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b56e05ade135443e83653d27fbaeba02 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f7c3f75544c4786a7aab0ca92450248 : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c9cc3465584aaabac23f6f60a525ad : SUCCESS in 18m 46s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/92c484c368b4404ab79c3c727afcb2ec : SUCCESS in 1h 36m 56s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e9544311b7584cd68e53085b01f82c5b : SUCCESS in 1h 16m 30s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6c3eb8f3a9fb425684e5c862d376f126 : SUCCESS in 1h 20m 01s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/264b5f740ba347ef97af31795964a4db : SUCCESS in 1h 28m 23s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/815538324a354388b4cd53ff2f3784fc : TIMED_OUT in 2h 10m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e425ab4d12fc4dccac6593e682537260 : SUCCESS in 1h 37m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ab4d9987cbd14f10beb4b223484c027c : SUCCESS in 41m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ea26e4061da433185fec14da9133193 : SUCCESS in 1h 02m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91cf8bd04f674cd38ff784065525a57e : SUCCESS in 2h 13m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0fe89cc92f664dfe9d0fd8553092ff58 : SUCCESS in 50m 41s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f1ed360aec8a47dfab4050dcb86c3713 : SUCCESS in 1h 06m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b74621b466ef45c795e1d048153fddc0 : SUCCESS in 1h 08m 53s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e14725d49e1042ce926c2f517e97ee00 : SUCCESS in 1h 00m 21s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/25ab75782cf04d7e88123d4703e57e45 : SUCCESS in 3m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/485a61cef29941ed8e7cd111a6fcaec0 : SUCCESS in 1h 29m 39s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c16ee6c9465f4b5f9410d94c06f590eb : SUCCESS in 1h 01m 53s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/67d3242b289941efa07aa0595f9bbfec : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"69fe2aa3e7903d54cab5f1f14ac1d9cd4a8d6697","author":{"_account_id":11934,"name":"Scott Grasley","email":"sg0818@att.com","username":"grasley"},"date":"2020-09-16 20:10:00.000000000","message":"Patch Set 3: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"be5b0a624fd9d64e10e5b4bfae414006b49eea3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:30:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/315811daa8ca484abe5fe51e5cc14ce2 : SUCCESS in 1h 03m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7659982c2d04b4da27e797ebc9ef251 : SUCCESS in 1h 19m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3f3fa8120d44e9e8946fe8a3219f336 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc8762eee484f5c9cd64d643727d64a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/853ab1a83da5434a94d0234dc153a0ba : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/befb87722c004d90b45e17672d153ae2 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dffdfba0096f4003a69db52358863468 : SUCCESS in 21m 23s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f9f3fb2ed9e943d789b3707251663066 : SUCCESS in 2h 03m 03s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/984ee29f80f444a8a4d7488b40bbc1b6 : SUCCESS in 1h 52m 17s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/a2613a4001d04951a899cdb0fcf415f5 : SUCCESS in 1h 19m 02s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/35791f199fde43d6a8478915c8489ace : SUCCESS in 1h 35m 23s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/e42ccad4ab7b49958e2cf3dfc9b7a7a1 : SUCCESS in 1h 19m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c27279dad7814abf8ed13dceb30385c9 : SUCCESS in 1h 35m 00s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d64706f4cb85488099b2760072709a8e : SUCCESS in 35m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1718dafdff1647bd9c6cdbe698f1cc3a : SUCCESS in 45m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be588a64e41643a5bd635b4e49a76321 : SUCCESS in 2h 10m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cfce3dcdf3f64088b6fd739983fa321b : SUCCESS in 40m 28s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9ca07c53320c425aa74b966d9f2257fd : SUCCESS in 1h 09m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de772d96eae440938637dfc126f411da : SUCCESS in 1h 04m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/368a3a4724f64bffa09932f0a609a37f : SUCCESS in 1h 08m 11s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7fab01e8c70a4e929eeb48de62dcebab : SUCCESS in 3m 10s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/af9511887f354381ac6c2cfb06028556 : SUCCESS in 1h 23m 40s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a43da443c9c345209c314507e941624e : SUCCESS in 55m 42s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4835f25ed12e473492cbc34b6dde1175 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"0021658f8669dc24efe8aa02a831667fb047b84b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 22:30:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"99c35b7d1f790c205524d472f3d115311c356abe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 01:35:17.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dca22a9fb8cf4ce38a54fd8f97fe8113 : SUCCESS in 1h 03m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b57e6502cb444a5daa828f696d3ec79b : SUCCESS in 1h 06m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf0d013669d4f3cbfb5c1792122e1c2 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a5e524f2f01460fb33536a2d5139cc4 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1de0bb7a09a84c9ca927f79e229ad3b6 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4600bdc9b44e2c81dc8ebf699c8086 : SUCCESS in 18m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e1cebfbd99b544c29d017a5a5ea4e4ca : FAILURE in 2h 18m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0cf90dee33f7492fa46b668ac1e6b034 : SUCCESS in 1h 23m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bbf3d33fc794106b978d63ab3d7298f : SUCCESS in 56m 26s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dd17efe99c8e4e039a1d6fdb7a03c480 : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c608f659a2843c4d70e702b278c502e36d2d65bf","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-17 06:13:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"56c199430a59abe754c88b2b58ee99d63524ccd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:46:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d20750ea756f409b8e42c5dee42e73f7 : SUCCESS in 1h 00m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b82f7101956143f1a239d6a87799b98b : SUCCESS in 1h 16m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1de14cd49aa7463baa73a5177412342c : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772c29f060664336b1d3a1b2fc85da67 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b0d248abb1b41828a74ff63857e3bf3 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c36d3cf31c0b48b1b02b06c01170eae4 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bef1f8693a644c3893b594d50539cf1 : SUCCESS in 26m 21s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/8b64b599eb4d4c5b98a153f70996ab41 : SUCCESS in 1h 50m 22s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/530204e004fd4768b7860fbde6a8b181 : SUCCESS in 1h 19m 46s (non-voting)\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/89c5a41ccbcc4f36a1fe4cdc782c7ae8 : SUCCESS in 1h 23m 28s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/188b8b5b31204342b8a2f9b82989aa7e : SUCCESS in 1h 19m 52s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/c9911839dcff4cf7850167ebeb12f81f : SUCCESS in 1h 18m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee13457a130e4566be385095057ffbd1 : SUCCESS in 1h 35m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e6414cd7428d4031a681f7167d46f6e7 : SUCCESS in 46m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55214648eb464dc28d2f68d3f17b978a : SUCCESS in 45m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e46b58ad4b641189e5729107b403123 : SUCCESS in 1h 56m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c5125408ca5f4c5f8afd6ff1d47c70b8 : FAILURE in 17m 44s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/db03ab2834a843da8b051dd6f2603660 : SUCCESS in 1h 02m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca9e50ed33fd43caa1b4005bad79cf9a : SUCCESS in 1h 06m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/118c2d588919429f8c4dbd188836fd89 : SUCCESS in 1h 24m 30s\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2bd86b14332046bbb74f5e7e1c67bf58 : SUCCESS in 3m 53s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/56686f1d4cbc43998ed773bb1842689e : SUCCESS in 1h 37m 33s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fe8f72e26cc043f9b62852f01edc0f52 : SUCCESS in 58m 03s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b3eb93e0b89d4a94a492bb58e8a5ca97 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b37cafe3e62e6af6d464b480a95382d1b74354d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:47:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"928a898297e4b991a820e2144c48840111f66a76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:46:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/05318ae3de044b3b9efe8db059b418ef : SUCCESS in 1h 00m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/081b6e2abb624ea1add8dcd62686e3cc : SUCCESS in 1h 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923d3f3b173e48dead5ca2e3dc7a1388 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8036b96395ca4f5aa553d042b0ca1dc6 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13e9d5ad356e476a8bdfe72caac70ded : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105bf972e81d49198594424c9e545549 : SUCCESS in 22m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb3eafb6c98e4cbf85cab7cc003369db : SUCCESS in 2h 02m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d320aa5bedf48db99cbfa0e43d39d75 : SUCCESS in 1h 16m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/388433cf1b13457997cbb9074a5d1751 : SUCCESS in 55m 31s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dbee70701a524523b3af504370e581cd : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":3},{"id":"7d501866dcba831f1582989a9d9d44ee55a825e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:46:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"374312b63f96d4598b8a4d99a7e3542a267c1e49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 12:48:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86bd4b29a1cc45f7ad6776dd807aa211 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a09be7e018c9e3b59cb4707b9146f336552b060d","revisions":{"2efe87d41d360cc111cf6736add4b587d7174736":{"kind":"REWORK","_number":1,"created":"2020-09-08 12:08:15.000000000","uploader":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"ref":"refs/changes/29/750329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/750329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/750329/1"}}},"commit":{"parents":[{"commit":"3d761179c36b9de4f907e51161cb22c95f79dfaa","subject":"Merge \"Skip test_qos_min_bw_allocation_basic when not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3d761179c36b9de4f907e51161cb22c95f79dfaa"}]}],"author":{"name":"Gayathri Devi Kathiri","email":"gayathri.devi@imaginea.com","date":"2020-09-08 11:54:07.000000000","tz":0},"committer":{"name":"Gayathri Devi Kathiri","email":"gayathri.devi@imaginea.com","date":"2020-09-08 11:54:07.000000000","tz":0},"subject":"Fix test_allowed_address_pair test case cleanup","message":"Fix test_allowed_address_pair test case cleanup\n\nMake network cleanup to wait until its\ndependent ports are cleaned up\n\nChange-Id: I91f5ced7873755081f94117c6a43bffcb85faab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2efe87d41d360cc111cf6736add4b587d7174736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2efe87d41d360cc111cf6736add4b587d7174736"}]},"branch":"refs/heads/master"},"557741c2e2c76a8cad878c3ef615464f90f1fc22":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-08 12:20:07.000000000","uploader":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"ref":"refs/changes/29/750329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/750329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/750329/2"}}},"commit":{"parents":[{"commit":"3d761179c36b9de4f907e51161cb22c95f79dfaa","subject":"Merge \"Skip test_qos_min_bw_allocation_basic when not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3d761179c36b9de4f907e51161cb22c95f79dfaa"}]}],"author":{"name":"Gayathri Devi Kathiri","email":"gayathri.devi@imaginea.com","date":"2020-09-08 11:54:07.000000000","tz":0},"committer":{"name":"Gayathri Devi","email":"gayathri.devi@imaginea.com","date":"2020-09-08 12:19:57.000000000","tz":0},"subject":"Fix test_allowed_address_pair test case cleanup","message":"Fix test_allowed_address_pair test case cleanup\n\nMake network cleanup to wait until its\ndependent ports are cleaned up\n\nChange-Id: I91f5ced7873755081f94117c6a43bffcb85faab6\nCloses-bug: #1894835\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/557741c2e2c76a8cad878c3ef615464f90f1fc22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/557741c2e2c76a8cad878c3ef615464f90f1fc22"}]},"branch":"refs/heads/master"},"a09be7e018c9e3b59cb4707b9146f336552b060d":{"kind":"REWORK","_number":3,"created":"2020-09-08 14:33:51.000000000","uploader":{"_account_id":31479,"name":"Gayathri Devi Kathiri","email":"kathirigayathri48@gmail.com","username":"gayathri"},"ref":"refs/changes/29/750329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/750329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/750329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/750329/3"}}},"commit":{"parents":[{"commit":"3d761179c36b9de4f907e51161cb22c95f79dfaa","subject":"Merge \"Skip test_qos_min_bw_allocation_basic when not supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3d761179c36b9de4f907e51161cb22c95f79dfaa"}]}],"author":{"name":"Gayathri Devi Kathiri","email":"gayathri.devi@imaginea.com","date":"2020-09-08 11:54:07.000000000","tz":0},"committer":{"name":"Gayathri Devi","email":"gayathri.devi@imaginea.com","date":"2020-09-08 14:33:33.000000000","tz":0},"subject":"Fix test_allowed_address_pair test case cleanup","message":"Fix test_allowed_address_pair test case cleanup\n\nMake network cleanup to wait until its\ndependent ports are cleaned up\n\nChange-Id: I91f5ced7873755081f94117c6a43bffcb85faab6\nCloses-bug: #1894835\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a09be7e018c9e3b59cb4707b9146f336552b060d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a09be7e018c9e3b59cb4707b9146f336552b060d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
