)]}'
{"id":"openstack%2Ftempest~763132","triplet_id":"openstack%2Ftempest~master~Ib0b1e6ecb23481103791b065a622b4b89190412e","project":"openstack/tempest","branch":"master","topic":"unuse","hashtags":[],"change_id":"Ib0b1e6ecb23481103791b065a622b4b89190412e","subject":"[Trivial]Remove unused variables and methods","status":"MERGED","created":"2020-11-18 07:37:22.000000000","updated":"2020-11-19 18:36:24.000000000","submitted":"2020-11-19 18:34:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763132-1605810859641-61dee02a","meta_rev_id":"a08abfc5ea7a81f51cb7f55fc21828c771a3e7aa","_number":763132,"virtual_id_number":763132,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 18:34:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-11-19 01:23:39.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2020-11-19 07:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-19 15:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-19 15:51:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 07:34:41.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-11-19 15:51:40.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-11-19 18:34:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18a85bf2a27aa5574028244db79bd3fc85cf5972","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-11-18 07:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ba8fc5b89b0f75ffe93c5f2a592506d24842b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 10:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/175ccb18e43d45389326377e158f3262 : SUCCESS in 1h 03m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9974e2d50bc8452cb399708ac96173d0 : SUCCESS in 1h 35m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4adb4b4fa5546b8b129154fd05aecf8 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e837d5247d434a48be34933230321dea : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecd05f54d8ac4e62b391c493173e91e6 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ed9287ca17f4e5e9506d9de95319aea : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68db2a0e0af2477bbc159743e46686dd : SUCCESS in 22m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a215a34d48db4cf69e51eb1f9d8f540d : SUCCESS in 1h 47m 31s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/f409f27e4d15435f958ebe786e9a88ef : SUCCESS in 1h 30m 15s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cae9f44d3fb54495962ed4ff929b6201 : SUCCESS in 1h 28m 04s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/e3c243d2f76d42d3bb18f955c2d3ac4f : SUCCESS in 1h 22m 10s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/e94987ee26d24e38adcac43794edb5c2 : SUCCESS in 1h 50m 19s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/0fc8832212d74b19ba50ad5798a938a6 : SUCCESS in 1h 26m 17s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/364078c6ed1642168961bd59c852c1d5 : SUCCESS in 54m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c32c22ea7c834b999355f0ccb12ff4e8 : SUCCESS in 2h 12m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/46ee3f3015ae48c581a445f61d5f4ae9 : SUCCESS in 41m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ec0bed45fa24139b3747b4cdf4150f0 : SUCCESS in 57m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d4aa3393b864ebc8cf3f6aa79a2d149 : SUCCESS in 2h 04m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/06eae4c447934c64bc33314f8af08db9 : POST_FAILURE in 54m 43s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/47b02d98639f465285d22db0f325c2f3 : SUCCESS in 1h 27m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1f279b2d04f4545a027d27ac0fb5830 : SUCCESS in 1h 07m 52s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/316ad2f35f30486b80264475c77c9959 : FAILURE in 1h 04m 57s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/be028cf59a44437a9f10571c2d9237dc : SUCCESS in 2m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/101b2d05b3d741188e2c968e1cef8523 : SUCCESS in 1h 56m 30s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2836bb3acee442fd8dfbb5c1c5166be4 : FAILURE in 1h 07m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0c79570d1a474e248d1a29fabfaa1b6b : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cd0f622f0fcbe56fc312f298c26111558e65d182","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-18 14:33:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLGTM, I like the idea","accounts_in_message":[],"_revision_number":1},{"id":"c843ebf0eaa2bb868fcf8efe9548651cefe17d87","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-18 15:17:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nwaiting for update on one comment.","accounts_in_message":[],"_revision_number":1},{"id":"b59dcc42542f5ba6bb1aa50f7575d17281b0cb0c","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-11-19 01:19:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eaaaba51a62682d52c044cf85fef5dee8529dbb1","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-11-19 01:23:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4bbe88b5facab957d250ef8f418fd9c2f915213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 06:30:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/134ded9afed24a2c8bef9d609ffd4bb7 : SUCCESS in 1h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c379146999bb483db8aa7c7d0c035dee : SUCCESS in 1h 30m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3973e5d836a24450b9d8d31793c9ffd1 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76aca769b23245efa8d31c4c41f91c02 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ecac332d89439899054bdc74f8350d : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a455896d0edf47aebe38575d9bfee7ae : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37215c09d7324d4fba6d3066513d2f5e : SUCCESS in 23m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/74a0ab626798496dbbf7d399e55d6605 : SUCCESS in 1h 36m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/1314cb6df82948cdaa5ac88b54b8414a : SUCCESS in 1h 18m 34s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0a01f8dd52cd414e885a646734c7ad3b : SUCCESS in 1h 27m 29s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/98815344ec70480483a905d76f910c78 : SUCCESS in 1h 26m 55s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/eb1184dcf5b2452bbdee514d65841574 : SUCCESS in 1h 19m 30s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/cba85b9227994771bf4bbc13b917ed1e : SUCCESS in 1h 33m 26s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/5d0cbc09684545cd91326c847efd41c4 : SUCCESS in 1h 24m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9c303062f55d46eea234c0f85bd44aa6 : SUCCESS in 1h 48m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3ceafd5d2a104067aec50c36a789740e : SUCCESS in 53m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f35fd4964294a0ea0790355c611c90c : SUCCESS in 1h 03m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36eb1623db624cfaaeb9380f50d4a491 : SUCCESS in 2h 00m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/63949eccaf794c6b9ad96e7666d391be : SUCCESS in 56m 35s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/88f3296273de43278d9c1ac502f3fa78 : SUCCESS in 1h 07m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f7ed43d9dc74d66a0640dc8a72c2f98 : SUCCESS in 56m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c097dfbad73b414d9127a65283cb455b : SUCCESS in 1h 12m 33s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/98bdaf435a804dbcbf53a54b64bfcc30 : SUCCESS in 3m 36s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/97e2f743c87a4c2ba7a1ccc2c3650da2 : SUCCESS in 1h 17m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bd2649e2b7084bf8802211d3e61d1072 : SUCCESS in 55m 34s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a5425003cccf4bb98655098d2d32c562 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"b93cc0f9375203321ea3502eb8a5a24bedb918a9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-19 07:34:41.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"75296156985569a3deeb6a2054bb533bf2852004","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-19 15:51:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"e575324eac0cdefabc7235603a6caa9a20a4e2f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 15:51:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84ac6504bd604d9b33cc7a73fe4e70219b06b273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 18:34:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c1c07a9d0b8233e054252f1418367d3a24eb9c6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 18:34:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/31e09cd7b3dc44cb9782e89cc5929e82 : SUCCESS in 58m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9f4a4d191964900859f34bd827e91cf : SUCCESS in 1h 22m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43d62b2eacd422685114db258563636 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/948a2281b870417f8a3aec8e269218d8 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80cb702094d145be8e321927e923e674 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70482f2a3aa4424a9ab8d5d870abcd29 : SUCCESS in 21m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd4506813f8e4a72a6c53e36e9df0049 : SUCCESS in 2h 34m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36b225db442e41d4955ff0a771f4ca96 : SUCCESS in 1h 03m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02118e68c0414fa89391ea40809b31dd : SUCCESS in 1h 29m 47s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5a25e27311e943e09f7974744f43c75f : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ce20661c710c9473586177db75713cf1d97d276d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 18:36:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9256ce893ed42a0b7fac91128326d13 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf5ed0b676b634af156b313e333a8f37e396b98c","revisions":{"55cf3027d6b90e20e34b8038500d903e33f57e99":{"kind":"REWORK","_number":1,"created":"2020-11-18 07:37:22.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/32/763132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/32/763132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/763132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/763132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/763132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/32/763132/1"}}},"commit":{"parents":[{"commit":"805f2a070b7b2227d1646bf83b4b93ea414e1377","subject":"Merge \"Remove \u0027gabbi_tempest_path\u0027 from devstack-tempest job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/805f2a070b7b2227d1646bf83b4b93ea414e1377"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-11-18 07:34:25.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-11-18 07:34:25.000000000","tz":480},"subject":"[Trivial]Remove unused variables and methods","message":"[Trivial]Remove unused variables and methods\n\nThis is to remove unused variables and methods to keep code clean.\n\nChange-Id: Ib0b1e6ecb23481103791b065a622b4b89190412e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/55cf3027d6b90e20e34b8038500d903e33f57e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/55cf3027d6b90e20e34b8038500d903e33f57e99"}]},"branch":"refs/heads/master"},"bf5ed0b676b634af156b313e333a8f37e396b98c":{"kind":"REWORK","_number":2,"created":"2020-11-19 01:19:35.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/32/763132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/32/763132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/763132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/763132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/763132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/32/763132/2"}}},"commit":{"parents":[{"commit":"805f2a070b7b2227d1646bf83b4b93ea414e1377","subject":"Merge \"Remove \u0027gabbi_tempest_path\u0027 from devstack-tempest job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/805f2a070b7b2227d1646bf83b4b93ea414e1377"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-11-18 07:34:25.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-11-19 01:19:12.000000000","tz":0},"subject":"[Trivial]Remove unused variables and methods","message":"[Trivial]Remove unused variables and methods\n\nThis is to remove unused variables and methods to keep code clean.\n\nChange-Id: Ib0b1e6ecb23481103791b065a622b4b89190412e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bf5ed0b676b634af156b313e333a8f37e396b98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bf5ed0b676b634af156b313e333a8f37e396b98c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
