)]}'
{"id":"openstack%2Ftempest~767165","triplet_id":"openstack%2Ftempest~master~I72e68a5dfbef569bb5a039cc5c2052175d05378b","project":"openstack/tempest","branch":"master","topic":"bug/1908269","hashtags":[],"change_id":"I72e68a5dfbef569bb5a039cc5c2052175d05378b","subject":"compute: Wait for volume snapshots when creating image from server","status":"ABANDONED","created":"2020-12-15 14:42:28.000000000","updated":"2021-11-09 10:57:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9322cf968ff9ccf2d72f097d8ceda1417d5b9a9","_number":767165,"virtual_id_number":767165,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-16 01:38:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-15 15:01:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":-1,"date":"2020-12-16 09:06:33.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 15:01:21.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2020-12-15 15:01:26.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-12-16 01:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 12:24:37.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"}],"messages":[{"id":"2b0c8b2356d570fbf231f868c411f597ce9703b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-15 14:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1c88e55e996eb3a6744660b59bffc76d1bb8e9e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-15 14:43:23.000000000","message":"Topic set to bug/1908269","accounts_in_message":[],"_revision_number":1},{"id":"ac626fa115367cee7faba312063f06d9b59b7a37","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-15 15:01:21.000000000","message":"Patch Set 1:\n\nawesome. thanks Lee!","accounts_in_message":[],"_revision_number":1},{"id":"9e965954d4244f607795a2812905187699f6ee22","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-12-15 15:01:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5da35ad036387408a7a488032ebe04181529ba0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 01:38:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/da68fd5f9f2e4fa1b35f3798998ea722 : SUCCESS in 1h 06m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/241e3a3fd9bb4f52afb3e341573d2cfc : FAILURE in 1h 11m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6889a8580f84f49860e63f5017ae62d : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e86e25b74c44cc4b030de575a110563 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2b380538e174be4ac925630184548b3 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85cacea131b242059fc51086ce15df14 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7439582d9be431a86e4090ddab79294 : SUCCESS in 13m 10s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/236666ee25674e748a3fefdc4b10c903 : FAILURE in 1h 46m 00s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/41cd758c92f74749a9c1afd205c4f4d8 : FAILURE in 1h 11m 59s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/65c825a461574722b57fc97e54296d41 : FAILURE in 1h 32m 48s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/6904148be8514506b1b1659a35b66723 : FAILURE in 1h 09m 31s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/fe84b6be4ecf456db0a79e5bae36ae7e : FAILURE in 1h 07m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7a9d224490434425ab0d4eca55615e8d : FAILURE in 1h 10m 02s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/7227ca1048964d469467e90e6ea1ef9b : FAILURE in 1h 13m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80d2e34d5c8149b7b713acdb30ef475d : FAILURE in 1h 30m 06s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0206d6bebc394419a727162320e600b0 : SUCCESS in 45m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93702097c13f4193b060d794aa9e5014 : SUCCESS in 1h 03m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/56b91f73eb2e40d3b884f19206a6da25 : SUCCESS in 2h 11m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bf26d10f5eaa436fa70130b27ada99f3 : SUCCESS in 53m 40s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/508d412af3c849618e34271507326b4a : TIMED_OUT in 2h 05m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4caa24fc348a4b8088cf874f8c4021a1 : SUCCESS in 58m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c022498e19ff4a9ea84d34587a477777 : FAILURE in 51m 02s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0e832f51906d40f893c76f0ed8bc411a : SUCCESS in 2m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/56bae0915bd6448e8a3cbdd8e3dd68a9 : FAILURE in 1h 07m 53s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2e757f5db564494ca9346043553dd877 : FAILURE in 57m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/97f3f561503e4fefa8b1ff6160dfac0a : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a8a4a7867f12236b7ce89650ff31da3f41af8e2a","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-16 09:06:33.000000000","message":"Patch Set 1: Workflow-1\n\nLooks like this broke some tests that are somehow getting an image dict back when creating a snapshot. I\u0027ll look into this more today.","accounts_in_message":[],"_revision_number":1},{"id":"b340aef162d07e82f9625fc3c359a8d07c5564e6","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-29 12:24:37.000000000","message":"Patch Set 1:\n\nany updates?","accounts_in_message":[],"_revision_number":1},{"id":"b9322cf968ff9ccf2d72f097d8ceda1417d5b9a9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-09 10:57:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54b4720762e2f358149bde56700c23b67bff28e3","revisions":{"54b4720762e2f358149bde56700c23b67bff28e3":{"kind":"REWORK","_number":1,"created":"2020-12-15 14:42:28.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/65/767165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/65/767165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/65/767165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/65/767165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/65/767165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/65/767165/1"}}},"commit":{"parents":[{"commit":"2c3da54a82e8844ea575c93576324855793738e1","subject":"Add releasenote to tag the end of support for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2c3da54a82e8844ea575c93576324855793738e1"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-15 14:29:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-15 14:42:26.000000000","tz":0},"subject":"compute: Wait for volume snapshots when creating image from server","message":"compute: Wait for volume snapshots when creating image from server\n\nWhen Nova is asked to create an instance snapshot of an instance booting\nfrom a volume using the imageCreate API it will make a request to Cinder\nto snapshot the underlying volume. This request is asynchronus and\nreturns to Nova before the volume snapshot becomes available.\n\nThis can easily race other volume operations if callers did not wait for\nthe volume snapshot to become `available` and overall state of the\nvolume to return to `in-use`.\n\nBug #1908269 outlines an example of this race where\nAttachVolumeMultiAttachTest.test_snapshot_volume_backed_multiattach\nattempts to delete the overall volume while the earlier request to\nsnapshot the volume was still in progress leading to a failure.\n\nThis change simply introduces the use of the\nwait_for_volume_resource_status waiter into the create_image_from_server\nutility method when a volume snapshot is identified in the snapshot\nimage metadata for the instance. This should result in the volume\nsnapshot reaching `available` and volume returning to `in-use` before\nthe test attempts to cleanup both the instance and volume.\n\nCloses-Bug: #1908269\nChange-Id: I72e68a5dfbef569bb5a039cc5c2052175d05378b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/54b4720762e2f358149bde56700c23b67bff28e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/54b4720762e2f358149bde56700c23b67bff28e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
