)]}'
{"id":"openstack%2Ftempest~767184","triplet_id":"openstack%2Ftempest~master~I31433c15984e38e5651b2f3d10c8769f59e30cc1","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I31433c15984e38e5651b2f3d10c8769f59e30cc1","subject":"Use victoria stable constraint in tox to release 26.0.0","status":"MERGED","created":"2020-12-15 16:38:42.000000000","updated":"2020-12-16 16:10:15.000000000","submitted":"2020-12-16 14:15:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767184","meta_rev_id":"72aa8527d746795710388a42992b09453a4731e7","_number":767184,"virtual_id_number":767184,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 14:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2020-12-15 18:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-12-15 23:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-12-15 23:51:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 18:41:57.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-12-15 23:51:01.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-12-16 11:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ec9760c991f0cd6d9da1de786dbc95381c0f233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-15 16:38:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a68b768e59f93402a336b8ebddfcc2db0924689","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-12-15 18:41:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4e0b99bbf4bcfd7150e9bfa627a16efb5130690","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-12-15 23:51:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks. \nThis is for the release https://review.opendev.org/c/openstack/releases/+/767163/","accounts_in_message":[],"_revision_number":1},{"id":"7d16e2a45ae4dd438a8d25f993c7292e57f76c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:47:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7535f81f2ca7441daa30948300a853ef : SUCCESS in 1h 02m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e991053e4a104192a431203e1e7a8560 : SUCCESS in 1h 32m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/197a3c69b20048c895d9dda256ccc7b1 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce8c7ea0e14a481c9ab7b07adf2e7913 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7c7e7ae0b61427ea8531f2d471c231a : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b76fd20cb254b39811acb28d6164fd2 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6238daa4c2413bbb6a1dd374ad481b : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd615568acd346e9ac2b7250a147994e : SUCCESS in 5m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/33d5efed715a4ab7a25107621e3dc5b4 : FAILURE in 1h 36m 53s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/f28a292c4a6e4198843950eb437a45d5 : SUCCESS in 1h 18m 46s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27d33a5cdab64cda841186509e100463 : SUCCESS in 1h 30m 06s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/6787afa4edae46a4b6a612597ec02362 : SUCCESS in 1h 18m 42s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/a2ab491d17df4fe8a640e1c93078aca4 : SUCCESS in 1h 27m 10s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2ac8321d6c404906893f558e17270f37 : SUCCESS in 1h 44m 14s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/645b7d059d6248aaa9f0accb17380a6a : SUCCESS in 1h 17m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/acd56a4dee5a4f7ba8ae06586aa32566 : SUCCESS in 1h 42m 34s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/bb314efe568340f789e7f0e9db9ef5b6 : SUCCESS in 41m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f91fc6569876445b9cfad3e7e2ea862d : SUCCESS in 56m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ac5ca9ac916433b9729fcbe5caa31e0 : SUCCESS in 2h 13m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a7a8d59c71f3437a909da41fe566e9f8 : SUCCESS in 58m 28s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/82886a4fba2a416e826b56f859718ef7 : SUCCESS in 1h 05m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6487ea7b17c418c95e2e82a5fc223bf : SUCCESS in 1h 06m 45s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/c3fb051631f4494e9174ee9d4d24a676 : FAILURE in 57m 45s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1cb133e748b34652ab09c00d924af45a : SUCCESS in 4m 27s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f5004ecd63614fb7bb336239f12e54e7 : SUCCESS in 1h 18m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/954438df62e349b89081d68211e26df0 : FAILURE in 1h 08m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9db797dfa50a447d94e8150b8eea18cb : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9dc5d28c1f51df345cb264b69755f127b77fdd98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 11:48:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f86553a5a97b6466e13cf8cb322aaf9f0aa7d618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 14:15:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/79f21cf9ca8e48e3b41747d3d906758e : SUCCESS in 1h 16m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e46bd97b275440dcac98881a28586b21 : SUCCESS in 1h 41m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7778c6637d4ed195015ba17edfc1d1 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c204a802bb5a4f1d973e448caf92de51 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25d13eeb7f1b432386473ba18e76f9b0 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973f2cfd4b71407f8e3576ad189c1b54 : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/339d5a4cf1db4c3e8df3a0c1960ec0dc : SUCCESS in 5m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad4ce8a816a44d02adbbc93a13da41e4 : SUCCESS in 2h 20m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df256381ec644635914fad8cfb709427 : SUCCESS in 1h 05m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/339e78bb1aa64624a4ffbd4c2f97ad65 : SUCCESS in 59m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b1aa64dd265040c7b300f6c282208057 : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8b1d09a73b5415800b3a3350f4f6b32e875cdd34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 14:15:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8aea2714f2da15c606526da7c57da766cf16ada6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 14:17:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98f6d16bc5240c8a00d221b3c31f79c : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ce7f7090e6b4471b3b9d19c1edc57f4 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1},{"id":"72aa8527d746795710388a42992b09453a4731e7","tag":"autogenerated:gerrit:revert","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-16 16:10:15.000000000","message":"Created a revert of this change as I78780ef88c5f8394ec11b144f148f30b060ed657","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9d31f2c46204abbbef53d3c0ae0f80125878fc9","revisions":{"f9d31f2c46204abbbef53d3c0ae0f80125878fc9":{"kind":"REWORK","_number":1,"created":"2020-12-15 16:38:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/84/767184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/767184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/767184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/767184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/767184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/767184/1"}}},"commit":{"parents":[{"commit":"2c3da54a82e8844ea575c93576324855793738e1","subject":"Add releasenote to tag the end of support for Stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2c3da54a82e8844ea575c93576324855793738e1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-15 16:36:14.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-15 16:36:14.000000000","tz":-360},"subject":"Use victoria stable constraint in tox to release 26.0.0","message":"Use victoria stable constraint in tox to release 26.0.0\n\nWe are going to release Tempest new tag 26.0.0 to declare\nthe end of support for stable/stein. So that new tag use\nstable/victoria constraint in the tox env as victoria constraints\nfor sure will work with tag. It did not use wallaby constraints\nas wallaby dev cycle is in progress and so does their constraints.\n\nFurther information:\n https://docs.openstack.org/tempest/latest/requirement_upper_constraint_for_tempest.html\n\nChange-Id: I31433c15984e38e5651b2f3d10c8769f59e30cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f9d31f2c46204abbbef53d3c0ae0f80125878fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f9d31f2c46204abbbef53d3c0ae0f80125878fc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
