)]}'
{"id":"openstack%2Ftempest~767452","triplet_id":"openstack%2Ftempest~master~Id51eea1574e1ad9601b1a89d01f58e905363885a","project":"openstack/tempest","branch":"master","topic":"fix1217","hashtags":[],"change_id":"Id51eea1574e1ad9601b1a89d01f58e905363885a","subject":"edit the error response code","status":"ABANDONED","created":"2020-12-17 03:15:31.000000000","updated":"2020-12-17 07:10:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b39ddb73d93978fe6bdb9cbf062c367272b51eb5","_number":767452,"virtual_id_number":767452,"owner":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-17 07:05:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 05:56:38.000000000","updated_by":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-12-17 05:57:56.000000000","updated_by":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-12-17 07:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5051221e78ae4d1025f7c06a379339d68ca0577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-12-17 03:15:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9aa31d2af8bd190bbe9ead051ca711c3154a3b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 07:05:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f14b81c421143058ef955c588f91de8 : SUCCESS in 1h 01m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19c247e5a3c844978304b42deeab48c7 : SUCCESS in 1h 22m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/746345f874764228982826e3102cb78c : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57320cf130ce47b19986af8789378836 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d2217b6f4324e71b7b4d03946643269 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b626f71af797400b8ec6f648a87d8eb0 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c40ca1b7c6347f6b6ca30a520d4c474 : SUCCESS in 12m 24s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/72f0688d87df47e2afef2925472b91ab : SUCCESS in 2h 06m 57s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/59da3673692849218b1b5b5b4e5e5e93 : SUCCESS in 1h 21m 11s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2520ceabc02d47eea196962a465da168 : FAILURE in 1h 07m 30s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/9c4b0e7e278e45079374e446ff063759 : SUCCESS in 1h 40m 46s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/3e5ea6325c504d578e5bd3c643f8ebc7 : SUCCESS in 1h 19m 05s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7d8215de3c194db78b0c1f1eb553ff2b : SUCCESS in 1h 35m 20s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/5061816fa9f94ddcaae310c12565b40a : SUCCESS in 1h 17m 52s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/42cf79f1dec044a4960d11774771bb84 : FAILURE in 1h 31m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/46e142e2aa354db3bcd56c4a7a818cd5 : SUCCESS in 50m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbbc64a2413e497faa7f9bf445d245ee : SUCCESS in 1h 11m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8e8b732ba474593ac3e43c449e05142 : SUCCESS in 2h 13m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/930a7d2dd2024ad8a258be2c77d7169c : SUCCESS in 50m 28s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cab795cb2331492798fe9088da78ad26 : SUCCESS in 1h 08m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6106a063a35f4e08a0a349405d372c25 : SUCCESS in 59m 37s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b35d77687bd444f7badeaedee3270414 : SUCCESS in 1h 03m 37s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b1e93a18d0bc469188e3b4ab9a4fa4a8 : SUCCESS in 3m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/81a6e2418f62423c86aad0d9273f6087 : SUCCESS in 1h 30m 48s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/7f9844ea3a2844fc94a20a5983a14647 : SUCCESS in 1h 24m 35s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3712d0ab59ee4b5ab46c27b6603e0f31 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b39ddb73d93978fe6bdb9cbf062c367272b51eb5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-12-17 07:10:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"989f6a3374232c17fc86424f3d037c8c31d34d27","revisions":{"989f6a3374232c17fc86424f3d037c8c31d34d27":{"kind":"REWORK","_number":1,"created":"2020-12-17 03:15:31.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/52/767452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/52/767452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/767452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/767452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/767452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/52/767452/1"}}},"commit":{"parents":[{"commit":"0e0264636f318920e709b97c4ea6f4df1e2d27fb","subject":"Merge \"Remove slow tag from test_minbw test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e0264636f318920e709b97c4ea6f4df1e2d27fb"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-17 03:10:56.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-17 03:14:05.000000000","tz":480},"subject":"edit the error response code","message":"edit the error response code\n\nAccording to the api ref\nhttps://docs.openstack.org/api-ref/object-store/?expanded\u003d#get-object-content-and-metadata and\nhttps://docs.openstack.org/api-ref/object-store/?expanded\u003d#delete-object,\nchange the response code.\n\nChange-Id: Id51eea1574e1ad9601b1a89d01f58e905363885a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/989f6a3374232c17fc86424f3d037c8c31d34d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/989f6a3374232c17fc86424f3d037c8c31d34d27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
